)]}'
{"id":"openstack%2Fneutron~747066","triplet_id":"openstack%2Fneutron~stable%2Fstein~I2674bdc448d9a091b9fe8c68f0866fd19141c6be","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1890539-stable/stein","hashtags":[],"change_id":"I2674bdc448d9a091b9fe8c68f0866fd19141c6be","subject":"Fix port can not be created with the sg of other project","status":"MERGED","created":"2020-08-20 01:11:27.000000000","updated":"2020-08-28 11:15:12.000000000","submitted":"2020-08-28 11:12:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747066-1598613124440-8a9bf4a0","meta_rev_id":"1e447483afb4eb06d7b18d5b6a947f59d544d3df","_number":747066,"virtual_id_number":747066,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 11:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-22 07:02:46.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-08-27 22:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-08-27 21:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-08-27 22:43:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-22 07:02:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-08-27 21:42:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-27 22:43:46.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-08-28 11:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb6a7ce1c9c36de777819a53e5ce686d7f6ec9a3","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-08-20 01:11:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"50c98c37c35a095487930600b524c83e5386d170","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-20 01:20:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747066/1 : FAILURE in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d414b92a7f77ee833cec4ea52cef4750c1e0eb11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 03:25:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a61e8e5c0cf247b68f6092d3cdf2522e : SUCCESS in 59m 13s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/41d8187871d6446e86511ffdbd4c9bcc : SUCCESS in 33m 51s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d238d5d7488e4bec960acc8631a00e5f : FAILURE in 1h 08m 25s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/a40ec837f84143298ba0c7685bad364e : SUCCESS in 1h 01m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/234ee639edd54979b5b1aaca7fdc8aa1 : SUCCESS in 1h 05m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81eb170de3224038805bb827f7513ebb : FAILURE in 25m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22f014769bfd45c4ad523bb721980fb6 : FAILURE in 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44df1322808a4a9cb4ad9bb722daf0a5 : SUCCESS in 12m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab6bb5cf32a24b6f8800db94cef012f8 : FAILURE in 26m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/384e6296fb974677b13e6d560ccf33e7 : FAILURE in 24m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/233bea91b38b4db489a4e7f9cb3d3a89 : FAILURE in 20m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bb7c370498e41b6acc798a52a1f0029 : FAILURE in 21m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501b598634fa4fb8b002f308ab78eb09 : SUCCESS in 5m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bec3f980909c4bdaaa92d0c88ceef43a : SUCCESS in 1h 22m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/93e7129f29684c59968a87848f583d36 : FAILURE in 36m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/361c955a321d482b86ac1781939147d2 : SUCCESS in 1h 23m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/64811334fc534fa7a40f3596a3dd332f : SUCCESS in 42m 48s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3c7cf8e9fa88498784a8710e2242e1e7 : SUCCESS in 41m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3fd3cf80c3624cfcae0448600d67e985 : SUCCESS in 1h 06m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b9964185f5db427fa2cde5126091de96 : SUCCESS in 1h 46m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3e700f17b33d444fb4679e604fdf7bb4 : SUCCESS in 1h 40m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4e4a0ff3e268459f9f9e959e91a2abe1 : SUCCESS in 1h 16m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/936b96accb464e5b85d1770f85f51c07 : SUCCESS in 1h 15m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b122ddc51cbd40379e6292acd834358d : SUCCESS in 1h 16m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ce6b1fe566dc49bc922129c2d3c198dc : SUCCESS in 1h 30m 00s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5e6052a6eabd434f8ef3d236ab2fe9f0 : FAILURE in 1h 37m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8fe6d2f92b0448928727ad404d6591f4 : SUCCESS in 1h 12m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/931213240e344f839e20c0fb70ad819e : SUCCESS in 43m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd785a7c5fe94739b6785c6aeb593b2a : FAILURE in 2h 12m 59s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/963912a1a4da424482a11f571a62449d : SUCCESS in 1h 25m 41s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/29a5e7af71644a02968556afc9190df5 : SUCCESS in 1h 00m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/79e36c2bbcd04f07a8ed262f62b69d03 : SUCCESS in 1h 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ad722b399ae3531de9e9b0a91ecab6a8bcb75ab","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-20 07:41:31.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds same UT (or code) fix as train one","accounts_in_message":[],"_revision_number":1},{"id":"2f02fbb6500899c3dbb6ddbf60bbdc7828b3c85e","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-08-20 13:51:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cefce9f10de8dc2b62b64487c98dfa0d53df2c75","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-20 14:03:54.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747066/2 : FAILURE in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"09981e8d2c17f1c5598836dff4fc22e56af43531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:16:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9c78794380c44460a8bd87053615c904 : SUCCESS in 52m 21s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/347f4626a6e24afdb730313022a5d2dd : SUCCESS in 31m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/a92608fb02654337bba4e2fcf63b877d : SUCCESS in 1h 02m 02s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/051f74fba3274fcc9549e3c71e4e6e5c : SUCCESS in 1h 07m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/7fb77e6444d44a55945a985d905dc2f8 : SUCCESS in 1h 08m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afb66f97eb454c6bb51e9de30e2765c8 : SUCCESS in 27m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c5996e28acf42aeadff88391bd2c2cc : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c5fda18fa24c639ed23ad00a619374 : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c53ab3280ea45d8aed66d6448f2cac4 : SUCCESS in 22m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/efc15ccca1f640009101c8c673fb21da : SUCCESS in 26m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f6ce38eac3a41c99af5bcead5bfec5d : SUCCESS in 25m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a991cb70231470b9f85e11959dec6ac : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ed51c98a6345af8333afb3262897b3 : SUCCESS in 7m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a454f20659b44c38bfcb7734f2d560f : SUCCESS in 1h 20m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d2d5f19f96074a02a30406ffd8fe3900 : SUCCESS in 52m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b04383dc2624902a1c558db8cf8607f : SUCCESS in 1h 20m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c15e8fc82aea4f4a9aaacbed2bca64ec : SUCCESS in 47m 07s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3aa1e83ac3cb478f94ebc8d9f165515d : SUCCESS in 44m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/661c905b75dc4870b872a91749390085 : SUCCESS in 1h 12m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f7556a25cafb4679aa99c14eaff874d0 : SUCCESS in 1h 31m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bb9516e7c65b4792b3eee28d3cfb1aa0 : SUCCESS in 1h 21m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/df22f68615704ecba24685ff84fc511f : SUCCESS in 1h 36m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68fcaf3bb05e4425a533e28788c4ffb5 : SUCCESS in 1h 15m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dfc8278dc7a64181bc6ba450cc6a11f8 : SUCCESS in 1h 02m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cd95cc829b334ac5bf1e02aa1723458b : SUCCESS in 1h 34m 22s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e712a82af26b40c8bc5fefbcf2cfeab0 : FAILURE in 1h 30m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f936d63be1e41d5a018669c51edad1a : SUCCESS in 1h 08m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/01a12a4ed6804e4da7f6acab27583f3e : SUCCESS in 55m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92ec0e108a014d818ab36f13c3746f89 : SUCCESS in 2h 07m 00s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b630fb677130463ba108631b6abb998a : SUCCESS in 1h 08m 51s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/28466887e89f45f5824cc5bf033532d1 : SUCCESS in 1h 04m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/44107c691b3c49bdb392f4478a06bf28 : SUCCESS in 1h 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bbb4b661f80afe7f3fc724c0296ab553718b06a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-21 18:48:10.000000000","message":"Patch Set 2: Code-Review-1\n\nSame comment as the train version regarding the conflict section in the commit message.","accounts_in_message":[],"_revision_number":2},{"id":"782074e39adc8353b99e544f2254563322eaf38a","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-08-22 06:54:47.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e52f7bb4d7e74839395a1ac01898322dcb1f556f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-22 07:02:46.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747066/3 : FAILURE in 7m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d7fa4c584e7c741650e2dff97c85c757015b9bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 09:12:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9234cb1eae1244129ae012db1373084d : SUCCESS in 49m 03s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/1de41da08fae46a3b2454c509f79a593 : SUCCESS in 37m 22s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/5057ceea342647dba21c9369c049c415 : SUCCESS in 1h 13m 26s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/37df20e6f09a47c58e827b64242630af : SUCCESS in 1h 35m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/1dc81aa4286f42979b7af90065fa44c7 : SUCCESS in 1h 11m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bd5d2dfa6684811b2f8ae7df28bcf43 : SUCCESS in 24m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d9bdb3bc71b4238822af46a12e996ae : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf6e4bae70449aa8055c6f19fcca0b2 : SUCCESS in 11m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4217691b6dc34f9ead8d97808289e1b5 : SUCCESS in 22m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19886f7244eb400da81562c179748e91 : SUCCESS in 27m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8872a3c8b40244728b1aca4137f5f463 : SUCCESS in 19m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0d5170e6fb34c9a9b2688e3a624528e : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f79d424d0f14763abe1c4640f3ed7cd : SUCCESS in 5m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/911ea365dc8048b5ac6c16bf69b4d12f : SUCCESS in 1h 35m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fba68793cd974293af77edbe8f2678aa : SUCCESS in 50m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f44cf303f08b4c4cb1e41fb1e6b5d4cc : SUCCESS in 1h 29m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b0eaac77016c419683dba1d1e612136b : SUCCESS in 39m 19s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/561aecafff9c41979c8b65a4e681de92 : SUCCESS in 41m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b01166fe19ee4fe8906ddf2f68f4b060 : SUCCESS in 1h 00m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a06e90c832274c80b32545771bb96146 : SUCCESS in 1h 52m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4043c60c3ffd41bcb6795a066a87b70f : SUCCESS in 1h 34m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c2581c7a84684332aae6ade973a858ec : SUCCESS in 1h 05m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe681883530f488c9a9ba38d9ee3a380 : SUCCESS in 1h 05m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6e025adbd8e244b88883012eeab1f305 : SUCCESS in 1h 18m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b53c964f6e524a4591cf394be7e1c53a : SUCCESS in 1h 31m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e237ea31b0a43d7a0ecd6f867ca6648 : SUCCESS in 1h 42m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2509d27f5424e34ba8a82747a7a236d : SUCCESS in 1h 12m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/52799a039e58479a9aca0bfa36f6dded : SUCCESS in 47m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b0a4bd2f02642a6af492366ff892730 : SUCCESS in 2h 09m 59s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7ac2e204237e44798238d907b3fec517 : SUCCESS in 1h 18m 34s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a47525ae4104088a0b6500b9e52a6db : SUCCESS in 1h 00m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a3f1f71963c444bdaa299489b66617df : SUCCESS in 1h 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bc93a750432ea504932c1387aefd2801755a5f86","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-27 21:42:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"848158a5448b8c4ffcbff9255f03c22c6ec9fbe8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-27 22:43:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f117c0bff2c89bb98ba2779f88b961fae19dc55b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:36:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f0bb59aad7da44e5435356bcd8701723856792ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:12:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9e6b129c6102454c8d9f3f821cb6f71e : SUCCESS in 55m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6234f0471d9f4aefa54d3199f4ed45d2 : SUCCESS in 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a67ecfe24904807ad21754ca4495310 : SUCCESS in 13m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f94f2fc33ff49fcaa5173ceb3fb547d : SUCCESS in 27m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c776607677074dea88f8a8bde5e9d82b : SUCCESS in 26m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4883ad947efe4f789aac9019689726ea : SUCCESS in 21m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37ece1e10b524afbaf8d833276b74031 : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b61a8400a9b444a8c61b145332cd46b : SUCCESS in 5m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b50a1dc677fc4501ba5f2b88185a6d0d : SUCCESS in 1h 22m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/88769fce48d94bbeb94bea42e85a06a4 : SUCCESS in 1h 04m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efcf10c7763b412dafcbeba070e5357c : SUCCESS in 1h 19m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/88578790c9124bd9bfc6ad06031e4a77 : SUCCESS in 40m 45s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/df836d40d05146f39a1ee9a5122b783f : SUCCESS in 41m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b11eb29b25d741bbafe92522da9d3cf0 : SUCCESS in 1h 07m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3bf7ec28e9a74664ad039a5962138e04 : SUCCESS in 57m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d223ad43a0d446ffb6066ff0e83c3f21 : SUCCESS in 1h 12m 42s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50e05c38fd0848ebbcf558790a36b353 : SUCCESS in 1h 16m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/12b29a62c51a4ad8bad1018bb4334225 : SUCCESS in 1h 04m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4dcae63990d8482fa7e6be73fd3a400e : SUCCESS in 48m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efd50b6b575e4d43b0d7f785d384c773 : SUCCESS in 2h 01m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db65876a12ad48efa7fdaa74864b2338 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":3},{"id":"3c53a3e5e7d6dccfe6c00c00e43afc2bd1aad878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:12:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b94e6cd8d201da60bb48a5a3f15cc661b08b15ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:15:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e16734d12ed45628f3c35bfac23cd40 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"494f9751d3b8d85c91494a290a64c175126b6b57","revisions":{"f20e4f9ad86c76f3dd5b02cc85b0c278a26c7b8d":{"kind":"REWORK","_number":1,"created":"2020-08-20 01:11:27.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/66/747066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/747066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/747066/1"}}},"commit":{"parents":[{"commit":"4ceee76477399956c2ca2a563b15fc56e404f749","subject":"Merge \"Fix Traceback when running neutron-ipset-cleanup tool\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ceee76477399956c2ca2a563b15fc56e404f749"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-08-06 08:02:36.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-08-20 01:11:27.000000000","tz":0},"subject":"Fix port can not be created with the sg of other project","message":"Fix port can not be created with the sg of other project\n\nThis patch adds the verification of whether admin context when\nverifying the valid security groups of port.\n\nChange-Id: I2674bdc448d9a091b9fe8c68f0866fd19141c6be\nCloses-Bug: #1890539\n(cherry picked from commit cc54a1c38e0b537883de43fecda781034c80daf3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f20e4f9ad86c76f3dd5b02cc85b0c278a26c7b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f20e4f9ad86c76f3dd5b02cc85b0c278a26c7b8d"}]},"branch":"refs/heads/stable/stein"},"709ce459add68b00662bc144cf3638b9ee78aae0":{"kind":"REWORK","_number":2,"created":"2020-08-20 13:51:04.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/66/747066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/747066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/747066/2"}}},"commit":{"parents":[{"commit":"4ceee76477399956c2ca2a563b15fc56e404f749","subject":"Merge \"Fix Traceback when running neutron-ipset-cleanup tool\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ceee76477399956c2ca2a563b15fc56e404f749"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-08-06 08:02:36.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-08-21 03:49:10.000000000","tz":-240},"subject":"Fix port can not be created with the sg of other project","message":"Fix port can not be created with the sg of other project\n\nThis patch adds the verification of whether admin context when\nverifying the valid security groups of port.\n\nChange-Id: I2674bdc448d9a091b9fe8c68f0866fd19141c6be\nCloses-Bug: #1890539\n(cherry picked from commit cc54a1c38e0b537883de43fecda781034c80daf3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/709ce459add68b00662bc144cf3638b9ee78aae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/709ce459add68b00662bc144cf3638b9ee78aae0"}]},"branch":"refs/heads/stable/stein"},"494f9751d3b8d85c91494a290a64c175126b6b57":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-22 06:54:47.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/66/747066/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/747066/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/747066/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/747066/3"}}},"commit":{"parents":[{"commit":"4ceee76477399956c2ca2a563b15fc56e404f749","subject":"Merge \"Fix Traceback when running neutron-ipset-cleanup tool\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ceee76477399956c2ca2a563b15fc56e404f749"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-08-06 08:02:36.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-08-21 04:12:14.000000000","tz":-240},"subject":"Fix port can not be created with the sg of other project","message":"Fix port can not be created with the sg of other project\n\nThis patch adds the verification of whether admin context when\nverifying the valid security groups of port.\n\nConflicts:\n    neutron/tests/unit/extensions/test_securitygroup.py\n\nChange-Id: I2674bdc448d9a091b9fe8c68f0866fd19141c6be\nCloses-Bug: #1890539\n(cherry picked from commit cc54a1c38e0b537883de43fecda781034c80daf3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/494f9751d3b8d85c91494a290a64c175126b6b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/494f9751d3b8d85c91494a290a64c175126b6b57"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
