)]}'
{"id":"openstack%2Fneutron~747293","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Idd42491dcaa10ec36c963b477438eaa2336ef3a0","project":"openstack/neutron","branch":"stable/ussuri","topic":"remove_unnecessary_verify","hashtags":[],"change_id":"Idd42491dcaa10ec36c963b477438eaa2336ef3a0","subject":"Remove some unnecessary usages of verify()","status":"MERGED","created":"2020-08-20 21:16:42.000000000","updated":"2020-09-05 02:04:36.000000000","submitted":"2020-09-05 02:02:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747293-1599271364768-941d6ed0","meta_rev_id":"37d7df3c499e11dcf9153e50f5f671a2711f8c5d","_number":747293,"virtual_id_number":747293,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-04 21:12:16.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-09-04 21:22:01.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"date":"2020-09-04 02:13:45.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-05 02:02:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-04 23:00:22.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-09-03 21:58:48.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-09-03 21:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-09-04 13:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 21:17:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-20 21:17:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-08-21 15:30:58.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-09-04 02:13:45.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-09-04 13:12:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-04 21:22:01.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-04 23:00:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-05 02:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f2e7d25eba2781209548e67417ac8313be8927c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-20 21:16:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67620d8c9744a79e9cfd0d5336b320361dcae226","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-20 21:53:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"359a2e782a058d4ce6569a48381fac0457abd98a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-20 21:55:20.000000000","message":"Topic set to ovn/port_forwarding-stable/ussuri","accounts_in_message":[],"_revision_number":2},{"id":"5e13bac79651a1137867c20d37eb38a721354c03","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-20 21:56:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"93715b3911ac8caea49d4b63d18a235f68eaa1de","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-20 23:10:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"37a83d8772eedc4f0bddfbf5554b117af2adfcaf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-21 01:42:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747293/4 : SUCCESS in 2h 30m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6a5087fe91cb81b2576bcf7456a875325ae0f26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 02:06:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/567aa2b072874a2da6714874bc9ffb3d : SUCCESS in 1h 20m 44s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/5ab02a34c9d4459683c95a8cadebb261 : SUCCESS in 22m 52s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/49ec2aa6e9e341b389edd5447cb041bf : SUCCESS in 1h 07m 32s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/8442dd5b2ed3434da57af451628ece4f : SUCCESS in 1h 09m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/45bd171c217544fcafe8b4a34b07f311 : SUCCESS in 1h 01m 35s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/ce1fac2a1aac48d885dd0ead6643b8b0 : SUCCESS in 1h 06m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c563edae1f9741b3b6648945fbf6b2a1 : SUCCESS in 37m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a574b9cf1f6c4dd28a6d8446815fa6a7 : SUCCESS in 23m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b3d416a3214576b0a9ab75a75d0a63 : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82749818afda4fd69a74f2f3add9bb59 : SUCCESS in 26m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37e3ce9da4674615b414ad0b7130542e : SUCCESS in 35m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6762ff30b6b545c495053dc5ea2f62c1 : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657df694d24849bdbf4097e4770d2a79 : SUCCESS in 8m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/829a0ed77a114e5fb5a49ad23aaa4429 : SUCCESS in 50m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd5e5317299f48cf87b1b083c7abe57c : SUCCESS in 1h 24m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ac8fd4e75114218b14f1823ba6d58e2 : SUCCESS in 1h 10m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f1497f9f1584769997187d09f039d03 : SUCCESS in 2h 08m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ee4fe956d524eb7894982bdbb7e2f17 : SUCCESS in 1h 17m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ab081a692fa6437da4c08fae758ae392 : SUCCESS in 1h 16m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/50f7dfb53d5447e8a87f016a08cdf999 : SUCCESS in 1h 30m 43s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/720cba98da1b4cf2b349f8cd7a7cf836 : SUCCESS in 1h 26m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9061aa3c50674fce835d43b303f2c988 : SUCCESS in 1h 30m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8e0febbb9ca8426fbe661ee639ee2d71 : SUCCESS in 1h 29m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/572c9f2aa60041138536c11c1fbe00cb : SUCCESS in 1h 34m 01s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b7b2f5b94aa442997d1b61ac4ab016a : SUCCESS in 1h 09m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e53bbfc919e14aacb61a1cbd62c5d45e : SUCCESS in 1h 15m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/152a839e21234f5599b5782ddea61e94 : SUCCESS in 2h 06m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4c615159dce480aa75b9ac68c5620cf : SUCCESS in 1h 25m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5437f725a8bb447ca1843133f952c713 : SUCCESS in 1h 04m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/872d62404acf4fc199f641138d974ef0 : SUCCESS in 36m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/527a4ec923874fbeb63aaeb547520cc1 : SUCCESS in 1h 13m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6b6d2ccb01e47919ba80a152ce59801 : SUCCESS in 51m 02s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ce3592300598420d8ee2cf7b0c437c2b : SUCCESS in 2h 08m 27s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a12ad24d333418cadcd692df2372002 : SUCCESS in 1h 51m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b3c51cd1c8024b8a8171a0aff78492af : SUCCESS in 2h 18m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e02d3e57487149b09b42cd7ef2eb090d : FAILURE in 2h 09m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"43e618beee854a2e9672bd6a7f75f937144400b3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-21 09:02:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8a65a536c0ed392c512ecbf23393921c2a30bad2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-21 14:54:24.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3bfb675a0a2a3b92fa0452467bf9b67ffe858e04","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-21 15:16:42.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747293/5 : FAILURE in 21m 03s","accounts_in_message":[],"_revision_number":5},{"id":"cee12333c3734c6a7342e06d9aeb0125c6e40807","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-08-21 15:30:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f3c1ea148ab45edf38e187295646f178458f6a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 20:39:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/7c5ceca14586476884fb5507bccf23fe : SUCCESS in 46m 03s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/c3f69cddba58402e8ced4bc330c275e5 : FAILURE in 34m 16s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/7ed12a8a6166416fa0515a8cd9387fc0 : SUCCESS in 1h 13m 01s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/9ee5b913bfc345ffaaa465f1bbd2a640 : SUCCESS in 1h 06m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/bf2791ab331247fe9624ac0c8e34e2c1 : SUCCESS in 1h 02m 17s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/561d118bb036420eab72847f0d2c3f1d : SUCCESS in 1h 18m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d0083b417964e22bdbf6e805fc17cc3 : SUCCESS in 40m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aafc4c3eed643249c34bde3da9f8a90 : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f781a3085d444fbc90bf63aed8f0791f : SUCCESS in 16m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34cc96afa70c43d8b7d09852d73a6bb3 : SUCCESS in 26m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43eb66ca560a4c37bba35215f2a4e5e0 : SUCCESS in 29m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/676ddaa356844e678d73cfc116bf7f02 : SUCCESS in 26m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90e033600a24818afd90c245bf96f6a : SUCCESS in 9m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/34d76f1113114958a91acf7ec119db47 : SUCCESS in 52m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8d9e6582529e4721b02a505469aa3abb : SUCCESS in 1h 09m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/423ac8c3bdb34715870d5ef9b19319d6 : SUCCESS in 1h 13m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0915a6c5412449ada030c5f7dcd0da8b : SUCCESS in 1h 59m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccb1b9258c5b4281b016a48ca79b37e4 : SUCCESS in 1h 18m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f3e3e6c8c7474553b2b0e861cac00404 : SUCCESS in 1h 27m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/652d497c3f8c41c4913225ecbca8d8b4 : SUCCESS in 1h 07m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/24f3acd449f944d4ae734dc43c3ffdc4 : SUCCESS in 1h 07m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6db8af57b6004be1825112628eda9f4e : SUCCESS in 1h 06m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/36bc8f1974b4426f8d4052b8261366eb : SUCCESS in 1h 34m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/00118b0eaee8483e9d08f4f78877a87d : SUCCESS in 1h 38m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0091be22976e49a4ba25c3592b3c26b9 : SUCCESS in 1h 09m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b12903051ff642099bbfb21e93a19a4a : FAILURE in 1h 21m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/285ef5573842436083425d17993de565 : SUCCESS in 2h 01m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6553dd159c5742c2a72aaa374aa30c0f : SUCCESS in 56m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fff6385a5de3412fab49e1d8b2127580 : SUCCESS in 1h 16m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/946a7d2ad1ba429ab9a7b8362cc71ad1 : SUCCESS in 39m 17s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5a2006cee5474ef6a7d781b6c187829a : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fea4157fc0f14e4eab61deefdcdd1a06 : SUCCESS in 42m 39s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c646e52981244cd180b11ef73ce12c79 : SUCCESS in 1h 31m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/52b56749970c412fa413034e4eb39f90 : SUCCESS in 1h 35m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/82d0adb857a94988a062bd23ecc1e2de : SUCCESS in 2h 38m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8d8203b8183747b18d0e3e6d35308dc9 : FAILURE in 1h 45m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f0d127a4069adde5a0a45fd8c10243711854355e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-08-22 14:19:35.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4a7046c25787c0f19db24cd7ca04627efe0ad04a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-22 16:11:31.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747293/5 : FAILURE in 1h 45m 03s","accounts_in_message":[],"_revision_number":5},{"id":"47575e23725b68bbb1ecaebb05b91465848d1db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 16:57:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/967712d0760443c7b3df0e88ef8460cc : SUCCESS in 53m 40s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/dcfed60c7ac2480a965c7c8acb0a104b : SUCCESS in 32m 15s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/30ac592bfb3e4196bdc9d2603ba116a3 : SUCCESS in 1h 34m 37s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/e3ee0406bdd54b76a36afbe719ebc5d5 : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/0466a248f9304b3188ac6b06f84749f4 : SUCCESS in 1h 18m 17s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/a8574093157a44c1abf3b05d5054f9f4 : SUCCESS in 1h 25m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35eaac776dd14a0eada6e07ad4c6f79e : SUCCESS in 41m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16526599b5bb42c798fe1e78dc776833 : SUCCESS in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89b6804f6e849dd9bdccb20d8ecc275 : SUCCESS in 15m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6def851ff8db417ea5108c0e14ba167f : SUCCESS in 28m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd836e4ed4a04e8cae8088c6e3b29362 : SUCCESS in 30m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71972021f4a3498184c061215d587190 : SUCCESS in 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4550bee357a24d38a8e88a28e99761d7 : SUCCESS in 8m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/664f7d9988764139812629d7ef546ace : SUCCESS in 51m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0c12f26aeff146aaa83baff3df7f40c3 : SUCCESS in 1h 49m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/72e9e1bf35f64c9a9ee6286d986ef8b2 : SUCCESS in 1h 36m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9e277b099e7e41539989c0653ea9c38d : SUCCESS in 1h 42m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e89937e4c01640b888263716a099333e : SUCCESS in 1h 21m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f34f9a6c890c4af8b882a9e0d69145a8 : SUCCESS in 1h 22m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0b5d71f7afa74e2a9b969f2a5d1d9b10 : SUCCESS in 1h 39m 30s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c56e1f757454b3690f5f1f0bb5c8e05 : SUCCESS in 1h 10m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e155763eeb64d28b0e68530d0ae229a : SUCCESS in 1h 08m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/740f0e0133424328b763438bca4d94db : SUCCESS in 2h 08m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4287fd65c5e74862b096a20ddf333081 : FAILURE in 1h 21m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e04344b4e8324c58a18e9f0faecfad4b : SUCCESS in 1h 16m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9f82da1e23864673abb83a9ca145583b : FAILURE in 1h 24m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c1f37d1da23848e8bb2ec7b9ff4c492e : SUCCESS in 1h 24m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1749040dd0124aa5a94c372c0d278674 : SUCCESS in 34m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b3691c31d738451baaf3dd3575eae7e0 : SUCCESS in 1h 49m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/317c4dd19c1046f29a22b30b0f69d1f9 : SUCCESS in 39m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ce78702ed7be4dbe913872931f6b0f0c : SUCCESS in 1h 19m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b7785ac75844f808f461641d0dc43eb : SUCCESS in 51m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b6ec1659f98d4947b11f398eef8241fa : SUCCESS in 2h 01m 54s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/13413a8cd33b4a7fb923e537d576214b : SUCCESS in 1h 47m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a42a79a0085d405c8a196ea98818fc0c : SUCCESS in 2h 31m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6a3dd0bcd0f945028758e94161e27c1e : FAILURE in 1h 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7496bb1ea2ccaaa3b1aaebd2ececa091ffc659d9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-02 08:58:32.000000000","message":"Patch Set 5: Workflow-1\n\nTo be abandoned.\nDo not backport this to upstream/stable to avoid making it less stable.","accounts_in_message":[],"_revision_number":5},{"id":"d289bcf738fa5fd6f58c610e47a653b7f7cce2aa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-03 12:19:54.000000000","message":"Abandoned\n\nAs we discussed in the team and with stable-maint-team, we will not backport this patch to stable branches.","accounts_in_message":[],"_revision_number":5},{"id":"35bcd45e8ebcaeb396c495fc6aac46687ff00917","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-03 21:58:42.000000000","message":"Restored\n\nRestore, because this gerrit change does not need abandoned port forwarding, but the other way around.","accounts_in_message":[],"_revision_number":5},{"id":"a4f3f59b12d3f345094897127465f419c766caf0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-03 21:58:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"8612ad836dc15253f2d4fb468866f45073ffcc46","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-03 21:59:35.000000000","message":"Topic changed from ovn/port_forwarding-stable/ussuri to remove_unnecessary_verify","accounts_in_message":[],"_revision_number":6},{"id":"813bce4b06d1229c173274a4b550529514038622","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-03 22:05:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/93/747293/6/check/devstack-networking-ovs-dpdk/ed9fe20/ : RETRY_LIMIT in 1m 41s","accounts_in_message":[],"_revision_number":6},{"id":"9839d20c23bac930ab4a18cee7b5817b5fa8cfa0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-03 23:43:15.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747293/6 : FAILURE in 1h 43m 45s","accounts_in_message":[],"_revision_number":6},{"id":"a8f3d860f971fefb742d3f253df65c38d43ad3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 00:56:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/5aee2e5811654952bb87decb003e3fee : POST_FAILURE in 57m 11s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/75666d6160c24cae9b72e0034e0868c3 : FAILURE in 45m 01s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/ff5174f029ca4289a0ba090ec3299404 : SUCCESS in 1h 13m 13s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/4010df3df5c0441d8a2cc7bdb028eabf : SUCCESS in 1h 13m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/aa0b69f77f6148078a1ef8b4805606bd : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/5d329a0391274ff1842ac8b4af462743 : SUCCESS in 59m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3312d321df1480285f6b65d8cf44a1e : SUCCESS in 35m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57faec60736746189ff4f7349c2e4d23 : SUCCESS in 29m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936f0f4ec5624880906b3870c441edba : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52174f830f20488ca714c7f195d162d2 : SUCCESS in 27m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a28b1e8c2304e77bf4949befc50b597 : SUCCESS in 27m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a630c6589c64f4ca473d04381fc5931 : SUCCESS in 23m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b821bf9b1af04791921780500a2d1fa9 : SUCCESS in 8m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/947b60ce76c74a26951bd716f084e689 : SUCCESS in 1h 00m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3950dfe496ae475e8bc0a230de48fef7 : SUCCESS in 1h 16m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/51296c99b15e44ceac68541ca48917ef : SUCCESS in 1h 10m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/406dac2260f2419d85a6b5c91c802953 : SUCCESS in 2h 05m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf13d77e05b64c54a2b0ab2f902a1862 : SUCCESS in 1h 25m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/998bfa328f89405c817913af0f35ca66 : SUCCESS in 1h 10m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0812e6864d68411fb1f1d3da95df39b2 : SUCCESS in 1h 09m 46s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca8cd958596449808284f6b83b1c2ebf : SUCCESS in 1h 05m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3109587349e84d5f8ae951dcac137a4c : SUCCESS in 1h 12m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/456a0ba15f9a4dcba042f8c1697ea320 : SUCCESS in 1h 33m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1650f87626b541c3af44950464e0b138 : FAILURE in 1h 17m 26s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/485078e8073b41929c95ad4df48a87c2 : SUCCESS in 1h 34m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a514a20abfcf4aa281ec03f5ed6f692b : SUCCESS in 1h 16m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2d42175b31942b8958f13565e80611b : SUCCESS in 1h 25m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d3d4afc2c0f4d4dbbb808bbec48f997 : SUCCESS in 57m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/027318a504ac4fd3b380856cd62e7ee8 : SUCCESS in 1h 09m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c71dc51918564940a76036dcea57c5d8 : SUCCESS in 42m 21s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fe45150c3a234a1090715dcc60eceb9d : SUCCESS in 1h 20m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1460862af0254e398fa9db58560d425d : SUCCESS in 54m 46s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/039412c8e111432f9a209face392350d : SUCCESS in 2h 01m 06s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a1154253c7b44b29b4d0f6d2db2abb2 : SUCCESS in 1h 43m 42s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5a8be3443fd04b8cb9efac0722fcbec7 : SUCCESS in 2h 11m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/075ee55616a64da4b19db5dc9b22ceff : FAILURE in 1h 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"203e73f5766b70ea9779c399a2b2ad352c3c5e89","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-09-04 02:13:45.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7d91b0d4c4a9373794e38b1acde1cd5b1e40292f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 02:27:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/93/747293/6/check/devstack-networking-ovs-dpdk/30b3100/ : RETRY_LIMIT in 1m 39s","accounts_in_message":[],"_revision_number":6},{"id":"9a151b1af9a6d2cc969a6041ceefe88fa50705cc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 03:54:38.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747293/6 : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":6},{"id":"47c9ed331f5dc58e89fae134274137ce92e427f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 04:52:51.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f541d328e095451abfe14906f87d5cb2 : SUCCESS in 54m 12s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/56d750cef19c4447b730bda057dbc92e : SUCCESS in 31m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/5d7bbda3cce747bf8350693e1567f2b9 : SUCCESS in 2h 37m 46s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/1d8d141195e44af09d1215c0b3dbcc62 : SUCCESS in 1h 01m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/334a17ba2abd498d8b187b01a8b3103b : SUCCESS in 1h 10m 34s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/cda32a7fa2b046368480de63ca4f8454 : SUCCESS in 1h 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3e19e915fa54b5393a427a9e59a003e : SUCCESS in 46m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a58317084e64f5d92fa2c045d4d07a1 : SUCCESS in 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c715dd500e4530a9fe27398b537358 : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1d4f9bb9aab45309c5fefd13b86e3d3 : SUCCESS in 25m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7abe6192ab549c1a86a3385137e6d28 : SUCCESS in 27m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60b59fad5b64438899706aff5f291ab7 : SUCCESS in 26m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589e84e2bc6e4a5b9b4c0f447c0e6b17 : SUCCESS in 16m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/90d9c6e64587419aaf90c47057cafe79 : SUCCESS in 48m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5e93b9836c994737aa824abae2a38e07 : SUCCESS in 1h 07m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fc3ecabd5ee42afa63952fa783f1068 : SUCCESS in 1h 07m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/07df1e72dccd4398934b683e02a23249 : SUCCESS in 1h 36m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de33642afa314dd6a39b818251e39fac : SUCCESS in 1h 14m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/39d40f179f0f416cad6524d08ea7bfb6 : SUCCESS in 1h 18m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5e7d60a7db8347e6b23caed9039c7b27 : FAILURE in 49m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/47daa330ed664c2d82babdd33bd93474 : SUCCESS in 1h 02m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b39ec1a8627249bd8669320ec7deec47 : SUCCESS in 1h 01m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f377570578c04c2d96d0dd279441a322 : SUCCESS in 1h 25m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/837ba10cbfff49b282ac183eb1fc09f3 : SUCCESS in 1h 28m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a2eef2e6f5e748dea67d54cf52aa7023 : SUCCESS in 1h 35m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bcfe7bc1063e417a8982bf35302e5ee2 : TIMED_OUT in 1h 33m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f7ffeaa5ad14e03b4cf13ee84158ecc : SUCCESS in 2h 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daa130df4054478492825de8840a0a22 : SUCCESS in 50m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cff6019410d2464fb086d89438e68d19 : SUCCESS in 1h 21m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4b9c71674d38410fbaa731f6fe3b508d : SUCCESS in 58m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e5d8034e2daf4e72964aa148ddcbcd4a : SUCCESS in 1h 10m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f369bb008d234400b3936d6e9a055dd3 : SUCCESS in 1h 06m 29s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9eee6ff66c94df9b03d018fff2623c6 : SUCCESS in 1h 40m 25s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef6538f165ed494487d2282b91592ac0 : SUCCESS in 2h 09m 41s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/33bafd66096142cb8f74006b5dc4422b : SUCCESS in 2h 30m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bad6d925196341d393ad084922000407 : FAILURE in 2h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f537b57bf2164d30dc22a0fe090fd49e5fc04b00","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-04 13:12:09.000000000","message":"Patch Set 6: Workflow+1\n\nNo LP in original patch, but cleaner logic and no API change on the outside","accounts_in_message":[],"_revision_number":6},{"id":"883af4eff7fb815d002c5839ce811ce45c8b3cdf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-04 13:12:17.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c00dd6e0567512bd2de100b01cf0b599b19239ab","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 13:28:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/93/747293/6/check/devstack-networking-ovs-dpdk/266f344/ : RETRY_LIMIT in 1m 44s","accounts_in_message":[],"_revision_number":6},{"id":"49dae80c385256ad87d5286e00f27cc41cb25b41","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 14:55:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747293/6 : SUCCESS in 1h 42m 46s","accounts_in_message":[],"_revision_number":6},{"id":"adef9b2ea392b67d7dd205816e8e0c65acc40cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:20:28.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/5d434e5021114258bef50dbf18d71c3d : SUCCESS in 52m 58s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/4825e7ade48f4832bc5bfc54acfff7ff : FAILURE in 35m 06s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/4d898cdeb9d64c08809d5a1d765c50c3 : SUCCESS in 1h 43m 31s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/c3fe5733cbe1496e80adec4471585a01 : SUCCESS in 1h 17m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/f2acfe7b16874730898a3e34b40f5b66 : SUCCESS in 1h 01m 18s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/0ac8645eb45b4c17a60f5ca7fe0ab103 : SUCCESS in 1h 13m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a07eb1933b450680baa156c3c6d414 : SUCCESS in 42m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45a13698281d46b598dc8eb14428be71 : SUCCESS in 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa93d6f7eb974c569254d8f3b45fae3d : SUCCESS in 12m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92031d18333b496a849025acab1c8c3b : SUCCESS in 28m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6456e835126d408cb2b67f07d8f2d108 : SUCCESS in 31m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8b67525ebd443b49b7ca8301c1eb556 : SUCCESS in 27m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b16da8237b4e0198e17e708e5d347d : SUCCESS in 10m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d9dcda7c16b46e2be09aa8780ec4a2f : SUCCESS in 57m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8235caea6a6f4033b87099fc203da759 : SUCCESS in 1h 22m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cec23d3c17764e12bce0e98577a37ac7 : SUCCESS in 1h 18m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2b2f68c66b2b46ce84033d8f30696a60 : SUCCESS in 1h 42m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08ce81f46f114ec1a817a2d14aec313e : SUCCESS in 1h 21m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cc39abd5a84e4bbd9820cce4725bf915 : SUCCESS in 1h 16m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9173ae4d462a4e4db4a1a6834ee9c1bb : SUCCESS in 1h 13m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c967ee6c9a1437fa691d5fd8528117e : SUCCESS in 1h 18m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b225cde7932d428ca763ba6079429615 : SUCCESS in 1h 04m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ca493a17a3d423e9c8357696591af01 : FAILURE in 1h 54m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb095a13fe14485a9310ee99e97f9a47 : SUCCESS in 1h 07m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1fae60cbc1b14b08851d9aa82ccd75f5 : SUCCESS in 1h 01m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/31e048cbb22c4b1a98f7059b36ae0fee : SUCCESS in 1h 19m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97c0da8529474475b1060ee5c9654976 : SUCCESS in 1h 56m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/212b98ba3b9d441eb8e04c06faa7b021 : SUCCESS in 1h 05m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fe02af8503c04470a0f2a8e501180b44 : SUCCESS in 1h 12m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b2403a698d6a416a8034b54dcb90a049 : SUCCESS in 38m 05s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/86f79a8c56534f0e99740bd32648bf2a : SUCCESS in 1h 16m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/47eac53510b84ab894c9ffab5734ecf8 : SUCCESS in 50m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/520e579bb530498ca0a0a76208e0aee8 : SUCCESS in 1h 33m 11s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cde3e6e8a5ed4269bf3adaa59bf23cda : SUCCESS in 2h 02m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/965ee42a778849e0a020c2f7b5636322 : SUCCESS in 2h 29m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5bc934fe87574ebab9b4fc741450ab64 : FAILURE in 1h 51m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dcca9c865097721c008761013b5fd50896227e53","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-04 21:12:16.000000000","message":"Patch Set 6:\n\nrecheck DNSIntegrationTests.test_fip","accounts_in_message":[],"_revision_number":6},{"id":"fdcb56d1de98b71cf018c0475bf50b9cf3c50b39","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 21:22:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/93/747293/6/check/devstack-networking-ovs-dpdk/ac221f3/ : RETRY_LIMIT in 1m 46s","accounts_in_message":[],"_revision_number":6},{"id":"d1c29f0d2c1ddf47e67a9dd366c0c588ce6b9530","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 23:00:22.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747293/6 : FAILURE in 1h 47m 37s","accounts_in_message":[],"_revision_number":6},{"id":"25e1a2d9174729fd668974e059528b7b92dd74f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:54:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/acdd49338c464661ba2edff6c44bc72a : SUCCESS in 1h 00m 53s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/6f336b899fe1443abc032eda0ac10eb9 : SUCCESS in 37m 57s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/aa342483c7e84a7095fb3a965ac53f85 : SUCCESS in 1h 37m 04s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/314a2284dc44432f8e54e649548ae9e2 : SUCCESS in 1h 26m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/23cb77ef17324495a2d8e1f3ec85abfa : SUCCESS in 1h 02m 46s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/c6608ee4f75142d3833e799095290dd2 : SUCCESS in 1h 07m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb01c6509b6450f984f9a97a366c212 : SUCCESS in 37m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b7dd76b906947248d9b601b49be445a : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1add3e0bf940f79f068a6a2a3d7293 : SUCCESS in 13m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea8ec13d6815473cb099a74e1a3c9bec : SUCCESS in 26m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b0aaf859a9d4b06a7e334a20d5d95a5 : SUCCESS in 26m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/804bd7ee3bfe4ee48c20f2e06276bbd2 : SUCCESS in 25m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f038042234204af381c87cf12a67dadc : SUCCESS in 8m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/046c36dc3bc34898a79324cab3210a9a : SUCCESS in 51m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/42469fdf66064a2ea89d16024d1fc3f3 : SUCCESS in 1h 14m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bc486087f754265893e64edd739d5fb : SUCCESS in 1h 12m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/efa3896047d044508dd8b8a45d40a74c : SUCCESS in 2h 19m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8aaa5ae056ad4bb1a9db6d40472ed3a0 : SUCCESS in 1h 24m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e978a0b50ad4bb183c246b9fdde702a : SUCCESS in 1h 23m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/72bc838b98f24cc98e2d4ed35cee2d08 : SUCCESS in 1h 18m 42s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/30747c1798f24c97af9828721fdd3e1d : SUCCESS in 1h 04m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a1d8505d938746c2a1c4da7a94e484b2 : SUCCESS in 1h 04m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7d5b0f355a3640e4b2071fac7fa6d026 : SUCCESS in 1h 29m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9521809ab334315963d305a9c58cdf5 : FAILURE in 1h 18m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/54d20fd35c69475380eaa8bf0945b7ad : SUCCESS in 1h 17m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed5c29d227784c5fa0c449d42219022c : FAILURE in 1h 29m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/33811a283a6a417b9e47bc1fb0ca9712 : SUCCESS in 2h 00m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fd8c0e7a37b4add8adebb7f775ef559 : SUCCESS in 50m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1438ebc68ccd42868056bb758bcc2b97 : SUCCESS in 1h 17m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d8b5135e213c4bddac28d9c948e33b49 : SUCCESS in 42m 21s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ad6afe3e2df24026a58b221f423a8c97 : SUCCESS in 1h 09m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7b1618fa8b148609ecfda4beb5cd709 : SUCCESS in 48m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03bf47f6f1564b49933932bc875f5406 : SUCCESS in 1h 46m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/78eacb7d21ce4c2fac54e2ab41cb9811 : SUCCESS in 1h 46m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/274a29deed7c42318f72bd4d95c74bfb : SUCCESS in 2h 13m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6acf9a3b25a0458e9d0eb8a50c752de2 : FAILURE in 1h 53m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3890dc4a7d350934eaf68e3c6dad11ca6fa2ed0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:54:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2b9f6c36e052fdb128fc571a4d65881faf42f51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 02:02:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/361a23bfeeb54627a752392e0711c14d : SUCCESS in 48m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5c4298000b14263ba8690054c27e92d : SUCCESS in 29m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba2dea09b0b04179ab9afff00c6f6b2f : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a8b5fb32dd4f9093a225fda1025b35 : SUCCESS in 24m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/791ee6bdef7e44c298137ab8d83d806f : SUCCESS in 28m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d600d82c0c04a86a18f51c5c125b1d9 : SUCCESS in 8m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1e1877ecb3334e4eb978fafd0a7de69f : SUCCESS in 52m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/dbbd815a534b4ca5956eca1fcf9057a2 : SUCCESS in 1h 10m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c928905f441146688c4be9868bdec0a0 : SUCCESS in 1h 18m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bc6394c988d94259975e3a37ad06e4cc : SUCCESS in 1h 13m 27s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c1d3db94211b4730927bbf794a8ae6a5 : SUCCESS in 1h 11m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1788e59c51b9447ebb399c4d4470f99c : SUCCESS in 1h 28m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b806940bbad84b418cf7f7647030ac56 : SUCCESS in 1h 15m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/10d374a2ad624ca7bc2b1c2a7b55f5f5 : SUCCESS in 2h 02m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/094ae204997d42e788204221db53f94a : SUCCESS in 54m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b969b9a877745ce9a40f8ad9cb79d12 : SUCCESS in 1h 23m 25s","accounts_in_message":[],"_revision_number":6},{"id":"2a376b6d1bece1241b4b504b8ccc427fb7812ad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 02:02:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"37d7df3c499e11dcf9153e50f5f671a2711f8c5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 02:04:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8815a164952c44cfb249fef9affa982b : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cd8fbdc04ec2a0a4ee0570fba9f1cd452669d044","revisions":{"f282de0a7b8c1191e52d292108b5856edb7b854c":{"kind":"REWORK","_number":1,"created":"2020-08-20 21:16:42.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/93/747293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/747293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/747293/1"}}},"commit":{"parents":[{"commit":"74ce755df148ad4ac63347fa2d3a64701be34ebb","subject":"[ovn]: port forwarding -- Functional test for IDL changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74ce755df148ad4ac63347fa2d3a64701be34ebb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 18:52:18.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-20 21:16:15.000000000","tz":-240},"subject":"Remove some unnecessary usages of verify()","message":"Remove some unnecessary usages of verify()\n\nThe general usage of verify() is if you are doing something like:\n\n ports \u003d row.ports\n ports.append(\u0027newport\u0027)\n row.ports \u003d ports\n\nwhere you read a value from the db, modify it, then overwrite it.\nThe setkey/addvalue/delvalue mutate functions make it largely\nunnecessary to use verify(), so this patch removes those.\n\nChange-Id: Idd42491dcaa10ec36c963b477438eaa2336ef3a0\n(cherry picked from commit b11dd3836eba35d65561f582660fa081abd49c19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f282de0a7b8c1191e52d292108b5856edb7b854c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f282de0a7b8c1191e52d292108b5856edb7b854c"}]},"branch":"refs/heads/stable/ussuri"},"1affd1905f19ef11e047738dda30d0673a6f90e0":{"kind":"REWORK","_number":2,"created":"2020-08-20 21:53:28.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/93/747293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/747293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/747293/2"}}},"commit":{"parents":[{"commit":"74ce755df148ad4ac63347fa2d3a64701be34ebb","subject":"[ovn]: port forwarding -- Functional test for IDL changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74ce755df148ad4ac63347fa2d3a64701be34ebb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 18:52:18.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-20 21:52:40.000000000","tz":-240},"subject":"Remove some unnecessary usages of verify()","message":"Remove some unnecessary usages of verify()\n\nThe general usage of verify() is if you are doing something like:\n\n ports \u003d row.ports\n ports.append(\u0027newport\u0027)\n row.ports \u003d ports\n\nwhere you read a value from the db, modify it, then overwrite it.\nThe setkey/addvalue/delvalue mutate functions make it largely\nunnecessary to use verify(), so this patch removes those.\n\nThis cherry pick required a small change to addess the following\npep8 error:\n./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py:717:48:\nH703  Multiple positional placeholders\n              \"Can\u0027t update external IDs\") % (self.table, self.record)\n                                         ^\n\nChange-Id: Idd42491dcaa10ec36c963b477438eaa2336ef3a0\n(cherry picked from commit b11dd3836eba35d65561f582660fa081abd49c19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1affd1905f19ef11e047738dda30d0673a6f90e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1affd1905f19ef11e047738dda30d0673a6f90e0"}]},"branch":"refs/heads/stable/ussuri"},"bcd1b0c1f616a5db61377b7c957a3cb06db47c4a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-20 21:56:05.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/93/747293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/747293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/747293/3"}}},"commit":{"parents":[{"commit":"74ce755df148ad4ac63347fa2d3a64701be34ebb","subject":"[ovn]: port forwarding -- Functional test for IDL changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74ce755df148ad4ac63347fa2d3a64701be34ebb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 18:52:18.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-20 21:55:32.000000000","tz":-240},"subject":"Remove some unnecessary usages of verify()","message":"Remove some unnecessary usages of verify()\n\nThe general usage of verify() is if you are doing something like:\n\n ports \u003d row.ports\n ports.append(\u0027newport\u0027)\n row.ports \u003d ports\n\nwhere you read a value from the db, modify it, then overwrite it.\nThe setkey/addvalue/delvalue mutate functions make it largely\nunnecessary to use verify(), so this patch removes those.\n\nThis cherry pick required a small change to address the following\npep8 error:\n./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py:717:48:\nH703  Multiple positional placeholders\n              \"Can\u0027t update external IDs\") % (self.table, self.record)\n                                         ^\n\nChange-Id: Idd42491dcaa10ec36c963b477438eaa2336ef3a0\n(cherry picked from commit b11dd3836eba35d65561f582660fa081abd49c19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcd1b0c1f616a5db61377b7c957a3cb06db47c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcd1b0c1f616a5db61377b7c957a3cb06db47c4a"}]},"branch":"refs/heads/stable/ussuri"},"122d6935fba5f3ae3662af222b2ac8fd691d4b34":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-20 23:10:39.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/93/747293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/747293/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/747293/4"}}},"commit":{"parents":[{"commit":"c51e51c723b8fedae88375ac1d422222c1db4e6d","subject":"[ovn]: port forwarding -- extend support for OVN usage (cont.)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c51e51c723b8fedae88375ac1d422222c1db4e6d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 18:52:18.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-08-20 23:10:39.000000000","tz":0},"subject":"Remove some unnecessary usages of verify()","message":"Remove some unnecessary usages of verify()\n\nThe general usage of verify() is if you are doing something like:\n\n ports \u003d row.ports\n ports.append(\u0027newport\u0027)\n row.ports \u003d ports\n\nwhere you read a value from the db, modify it, then overwrite it.\nThe setkey/addvalue/delvalue mutate functions make it largely\nunnecessary to use verify(), so this patch removes those.\n\nThis cherry pick required a small change to address the following\npep8 error:\n./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py:717:48:\nH703  Multiple positional placeholders\n              \"Can\u0027t update external IDs\") % (self.table, self.record)\n                                         ^\n\nChange-Id: Idd42491dcaa10ec36c963b477438eaa2336ef3a0\n(cherry picked from commit b11dd3836eba35d65561f582660fa081abd49c19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/122d6935fba5f3ae3662af222b2ac8fd691d4b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/122d6935fba5f3ae3662af222b2ac8fd691d4b34"}]},"branch":"refs/heads/stable/ussuri"},"4d56d1624ff8a213d9c8c021ca0a68187628a3a2":{"kind":"NO_CHANGE","_number":5,"created":"2020-08-21 14:54:24.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/93/747293/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/747293/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/747293/5"}}},"commit":{"parents":[{"commit":"9d919953d23478ac7d78da9ce9233dcf561f2482","subject":"[ovn]: port forwarding -- extend support for OVN usage (cont.)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d919953d23478ac7d78da9ce9233dcf561f2482"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 18:52:18.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-08-21 14:54:24.000000000","tz":0},"subject":"Remove some unnecessary usages of verify()","message":"Remove some unnecessary usages of verify()\n\nThe general usage of verify() is if you are doing something like:\n\n ports \u003d row.ports\n ports.append(\u0027newport\u0027)\n row.ports \u003d ports\n\nwhere you read a value from the db, modify it, then overwrite it.\nThe setkey/addvalue/delvalue mutate functions make it largely\nunnecessary to use verify(), so this patch removes those.\n\nThis cherry pick required a small change to address the following\npep8 error:\n./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py:717:48:\nH703  Multiple positional placeholders\n              \"Can\u0027t update external IDs\") % (self.table, self.record)\n                                         ^\n\nChange-Id: Idd42491dcaa10ec36c963b477438eaa2336ef3a0\n(cherry picked from commit b11dd3836eba35d65561f582660fa081abd49c19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d56d1624ff8a213d9c8c021ca0a68187628a3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d56d1624ff8a213d9c8c021ca0a68187628a3a2"}]},"branch":"refs/heads/stable/ussuri"},"cd8fbdc04ec2a0a4ee0570fba9f1cd452669d044":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-03 21:58:48.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/93/747293/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/747293/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/747293/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/747293/6"}}},"commit":{"parents":[{"commit":"680383ad0d7c3dc1569e3fd89c28d07f48455ca7","subject":"Merge \"Don\u0027t raise FileNotFoundError during disabling keepalived\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/680383ad0d7c3dc1569e3fd89c28d07f48455ca7"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 18:52:18.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-03 21:56:19.000000000","tz":-240},"subject":"Remove some unnecessary usages of verify()","message":"Remove some unnecessary usages of verify()\n\nThe general usage of verify() is if you are doing something like:\n\n ports \u003d row.ports\n ports.append(\u0027newport\u0027)\n row.ports \u003d ports\n\nwhere you read a value from the db, modify it, then overwrite it.\nThe setkey/addvalue/delvalue mutate functions make it largely\nunnecessary to use verify(), so this patch removes those.\n\nThis cherry pick required a small change to address the following\npep8 error:\n./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py:717:48:\nH703  Multiple positional placeholders\n              \"Can\u0027t update external IDs\") % (self.table, self.record)\n                                         ^\n\nChange-Id: Idd42491dcaa10ec36c963b477438eaa2336ef3a0\n(cherry picked from commit b11dd3836eba35d65561f582660fa081abd49c19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd8fbdc04ec2a0a4ee0570fba9f1cd452669d044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd8fbdc04ec2a0a4ee0570fba9f1cd452669d044"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
