)]}'
{"id":"openstack%2Fneutron~747774","triplet_id":"openstack%2Fneutron~master~Iebdfd2b303f47ff9f049cf583ea754b35ca26f78","project":"openstack/neutron","branch":"master","topic":"bug/1882804","hashtags":[],"change_id":"Iebdfd2b303f47ff9f049cf583ea754b35ca26f78","subject":"Allow replacing the QoS policy of bound port","status":"MERGED","created":"2020-08-24 17:06:53.000000000","updated":"2020-09-30 04:33:09.000000000","submitted":"2020-09-30 02:21:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":67,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747774-1601432499565-84033bed","meta_rev_id":"e94a3651c2017ec680f3708ebd08c06508a1768d","_number":747774,"virtual_id_number":747774,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-09-24 07:22:00.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 02:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2020-09-24 07:44:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-28 08:01:10.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"date":"2020-09-29 18:48:10.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-24 06:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-24 06:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2020-09-24 06:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-09-24 06:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-09-24 06:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-24 06:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 09:38:54.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-09-03 21:58:58.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-15 18:19:48.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-09-16 09:43:52.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-17 11:45:57.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 03:53:03.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-09-22 07:40:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-09-22 14:16:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-24 06:18:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-24 07:22:00.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 07:44:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-29 18:48:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-30 02:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4fd21652c2e499e4f35c8bf7213fca01ef38d3a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 17:06:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"432c27430e4cd2e77d00980ebb07d4c5a35836e5","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-08-24 17:38:49.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/747774/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8cd887c972c465f9b80c064bdef2cb636442541e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-08-24 17:53:44.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/1 : FAILURE in 46m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b869f1b5133c9051449380c5db7557ed4e7fe453","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-08-24 18:28:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2020/Arista_Tempest_Testrefs/changes/74/747774/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9fff89834c42cf5436fa7b4a67ff098fbf84d815","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 20:08:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/74/747774/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/74/747774/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"455029001bb0e2760fb7e58753ed74edd944236f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 22:20:45.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/212a11cf81d64801b37e33fb5dcab7db : SUCCESS in 54m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e2d57679cf044e98459f41cad3b5cb5 : SUCCESS in 34m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1933cb1ef86840cb97a70080084e3b34 : FAILURE in 1h 02m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/30b51404515e47689179778b2ae3073d : FAILURE in 1h 04m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ea065c123654380b806955ffb84229c : FAILURE in 1h 02m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b0f985c350b54383be3543e8549e6f97 : FAILURE in 1h 03m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a472a2271d945b0b867c2b97728c37b : SUCCESS in 38m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/802647aba91a42508ddb920182e0ad40 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d6d7affc02463fa939d714e99983b1 : FAILURE in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dca3b364b3134c069d74bfceacf6baa6 : SUCCESS in 26m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/861b7e9fa331485cb60a2c520b989738 : SUCCESS in 23m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb49ff1aba8146a89ae36ee88e7a7d69 : SUCCESS in 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a20bc7d6c48c40f7968ea49292847d37 : SUCCESS in 49m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0855fc1310e940259277683333608aca : SUCCESS in 46m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/49d571464eaa410b972f245db6abaa23 : SUCCESS in 1h 08m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64ebbb6cc1c7492faeaed8c3faabac51 : SUCCESS in 1h 09m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c3e23556726944189fc6db7e78b53c9f : SUCCESS in 1h 41m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce597678f7f6428684354928810ea85f : SUCCESS in 1h 12m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cfe328a28c75499d8f60da9d5ddedd78 : SUCCESS in 1h 06m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7957b7cce9b4436f8bf2b2bbefe23f25 : SUCCESS in 1h 29m 05s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2782d7fc7ab45498675af1ffe557d3f : SUCCESS in 1h 05m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3a9ad560464944ef9d9d16feb4f4d2b9 : SUCCESS in 1h 22m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7de814b56c0c47789b218e380670c31d : FAILURE in 1h 17m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d8a7ad5580549848f59679b2e2dd585 : SUCCESS in 1h 19m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/da5339f135eb4d6c98beb312ad3faf44 : SUCCESS in 1h 15m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed8b70c14ebc49af8db71ae116af2447 : SUCCESS in 2h 00m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6978e4f1f7244fc8411e1c02d96d9f1 : SUCCESS in 55m 19s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/89dd502e2e9448698e4f758a5ded90e6 : SUCCESS in 1h 23m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/409f16243b1543eb8dfef01d027a8f09 : SUCCESS in 1h 25m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f992dbceb6244b519c60b4a90a28b730 : SUCCESS in 45m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/555c083fdfe84ecca45c34c8ed938294 : SUCCESS in 1h 20m 19s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e02585922243473fb114e0b6791e6f47 : TIMED_OUT in 3h 19m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/230e869e435047318af27e00c089d8ba : SUCCESS in 2h 11m 30s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2e20bf54b7af426fbb949f976ec75fcd : SUCCESS in 2h 35m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a1fc9db174074e26842cb06e78734525 : FAILURE in 1h 47m 00s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/421610d7a7794e8ca8f53a800eb8f6ca : FAILURE in 1h 57m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c4af29f9ac3a0c3b4cd2548c1ebd68bbf0d225a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-25 09:38:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c31c10bd11478888a8f405cc1204721312e7599","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-26 16:37:20.000000000","message":"Patch Set 1: Workflow-1\n\nBased on last qos meeting discussion let\u0027s go back to the approach where placement call (allocation update) happens after db transaction (AFTER_UPDATE event).\n\nPros\u0026cons of this approach (with BEFORE_UPDATE event):\npros:\n * placement REST call is befor db transaction\n * if placement update fails port update fails, no db rollback is needed.\n\ncons:\n * if port update fails in later phase (like db error, etc..) no way to rollback the changes in placement.","accounts_in_message":[],"_revision_number":1},{"id":"ea99e9eaddc9370fb93eb79efdfafaf273b69c91","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-03 12:59:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a89293c605a54a1bbae6003111fff7323977633e","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-03 13:31:58.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/747774/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bba96a2d401cd29eb45f124abea9f3569fecc916","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-03 13:41:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/2/check/devstack-networking-ovs-dpdk/4547210/ : RETRY_LIMIT in 1m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0343b4a9b705ba096d3dcb5b3d858710c9285a2b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-03 13:51:00.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/2 : FAILURE in 50m 27s","accounts_in_message":[],"_revision_number":2},{"id":"48a4e5b9fb9c403cb8132361bfd0bf82d78eb932","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-03 14:12:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1fffaa6e15a622dbd9c571d8a39aa50ebe0f27d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:17:59.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0c9c7bd8213e4ccd83f276338c50965c : SUCCESS in 58m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f505bed1b730491bab58c08f1ad0381b : SUCCESS in 46m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/613f70b38c6542028d70356add27252c : FAILURE in 1h 35m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/99daece09d124611bb9540fcfedc04a5 : FAILURE in 1h 23m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5fc36081882e498987945ffe9e23b58e : FAILURE in 58m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6b61175af6b94a2dad9578bc6cbc169d : FAILURE in 57m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d165e3de697344e1bc0bc53c7174c4fe : SUCCESS in 28m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d317ec53583546968eea64b49a713eab : SUCCESS in 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80247bfdfeff4179bd131748a299846c : FAILURE in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed6a15e7d9f4d1c83845ddbffb0db97 : SUCCESS in 26m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46dc99dd90374512abb0458850000ac6 : SUCCESS in 26m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a9085714a541b5b4e4b895a282ee7a : SUCCESS in 19m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3846b6c1996c4444a282a152b256f5f5 : SUCCESS in 48m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba0b73f3af054dec80045bf34756c08f : SUCCESS in 1h 13m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/16db52ecdd594389bb43ea36a72cb33c : SUCCESS in 1h 46m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b254c470adbe4dea9b64e7f0b5eafe11 : SUCCESS in 57m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/51da5388de4e45c8a476482f5725d6b1 : SUCCESS in 1h 15m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4a559c47873948eba9a235b34ab50b78 : SUCCESS in 1h 25m 19s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cc5f2cda9174f2abfcea5964ca620ac : SUCCESS in 1h 13m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c40136bb7e4345b585b8cd0f84700d17 : SUCCESS in 1h 17m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e76a64c254940779eeb6bb66f44d8dd : FAILURE in 1h 36m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/52a35e40b38343a1bfe6b4ba50de4a71 : SUCCESS in 1h 27m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/40b856e13dd342efa6a32434bab13558 : SUCCESS in 1h 17m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76c53dc1b7a9442299a869278796ee4b : SUCCESS in 2h 08m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13bcba3b731e4746817c7f1f94765006 : SUCCESS in 54m 46s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2f2cd9fd0aa348d288638cc92ca12113 : SUCCESS in 1h 06m 38s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38d8fa3768954f62b1bdc6076a81bdf9 : SUCCESS in 1h 28m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b47dff4ec69b41acb1e3491971e287ce : SUCCESS in 50m 01s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4c7460e057fc4609b93a4c9f30daf433 : SUCCESS in 1h 24m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eb39f59884d34a419bec17d4f2770bbe : FAILURE in 1h 21m 30s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cbbf84ef3f9348fca60aa8baac5a625a : FAILURE in 1h 41m 12s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8c29f3e4f5114d4d9aa9d60528ff297b : FAILURE in 1h 06m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/79c59e41061a46f7bf09bd78aee4e38a : FAILURE in 2h 09m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/84d72861d81046228c955fce6764d27b : FAILURE in 1h 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24c05340419f1776e8371e4363cd2c20a157a100","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-03 21:58:58.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/74/747774/2/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/74/747774/2/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"11c13a575d01ea1562bf33213589353b2082a4d7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-04 15:28:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2278a26aa7d72bfe3a68a4d93f153664130eb04b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 16:03:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/3/check/devstack-networking-ovs-dpdk/caad848/ : RETRY_LIMIT in 1m 44s","accounts_in_message":[],"_revision_number":3},{"id":"18eb61a5eeeb3e77e9ae4c4eed32257c0c4e5731","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-04 16:07:27.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b421e3c4c1e4f4b77aa471b94e6185653fa2536a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 16:13:07.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/3 : FAILURE in 43m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7b339c37d8ba3a0acdf8fe46f4f88eb5d7f66074","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-04 16:14:18.000000000","message":"Patch Set 3:\n\n* sriov-test http://52.27.155.124/sriov/747774/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6fb787d71b63ebcc6f58e43eef78d57d6cc9fbcc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-04 16:18:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"773de96366bda51b32d4bbd1b93f60e0e19dd93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:01:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/365255e2038749fbb74cadab8ba24c21 : SUCCESS in 1h 08m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbd9eeb649124da4803c5b60a12440ab : SUCCESS in 39m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/61983255e9d844139119524d72be69f5 : FAILURE in 1h 17m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0d24489c2efc421581c83cf5d99bad14 : FAILURE in 1h 04m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fc97e7f223814bf7b8cf25873e4b68a7 : FAILURE in 1h 14m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/276b87bd380e422bbd1395b624f08092 : FAILURE in 1h 27m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bca79297ce9144aca5fb13fdf3ba58f5 : FAILURE in 42m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2f4ea40c611477abb7498a7c572870e : FAILURE in 27m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6893ef0dabcf45a399ee99eb5d870332 : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b351c7951b04a05a9417f1f67bab7d0 : FAILURE in 35m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aecafc92afee4feb8570a497b719a1fe : FAILURE in 25m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ecf60710984c9fa3144d3a261874f2 : SUCCESS in 8m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eaf7f69f8b794e5ba85dcbce3cbf2ce0 : SUCCESS in 56m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e10690c427384ef797271418b166311f : SUCCESS in 1h 08m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/760b6e11eefa4135b02e47594261efac : SUCCESS in 1h 52m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f41b766428ea43fea56b132d74098765 : SUCCESS in 1h 14m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52f151f16984445fabcbd4c05c49a755 : SUCCESS in 1h 12m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8c782b0bde654bf4a00f115b04a2b744 : SUCCESS in 1h 42m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdb25fd8764f41c8b3a661212e0695d1 : SUCCESS in 1h 39m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8129b563065b4e2e92ef43c3f6035243 : SUCCESS in 2h 10m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b9e37d4c160d40e888608753e2427d73 : FAILURE in 1h 46m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29b4781649da41dfa63983853b49be0f : SUCCESS in 45m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8c4b2a8953694c4c8aba78a257c4be60 : SUCCESS in 55m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e034b488dc049998bbb848aff528932 : SUCCESS in 1h 24m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22d5560384d3449e901541ec7130d3f9 : SUCCESS in 50m 56s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b3e4a5260a9e4eaf89d35a2666a2a851 : SUCCESS in 1h 02m 30s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07c6e86916fd419e97d8b4f1dbe0a992 : SUCCESS in 1h 17m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e9ececc355a54968b6bf510a5e62fc7d : FAILURE in 38m 36s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/37dee32765684e2895eaf56114c90916 : SUCCESS in 1h 16m 06s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/23fea02fb9db4f98bd4ff2a4c3db092e : FAILURE in 1h 10m 27s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/506c633ae5be42efbe6a28596b1766c8 : FAILURE in 1h 59m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0294bf82b7284ad5b16a0604d087e736 : FAILURE in 1h 09m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2a346a1ba6114039b63dc390a7660b60 : FAILURE in 1h 38m 14s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f4759d50b12248768fc937c81a1131c6 : FAILURE in 2h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0bc67efffff933851204f4384d42518f6b8d2d42","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-07 04:09:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bdf40a6a368824db2d40f0a682f9a21cdfbe7884","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-07 13:05:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"530171601dbf8d0c053458d9ff8ac9306cd1bc52","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-07 13:10:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b26dc3c162fe778fb0f80704049e0f0c808df20a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-07 13:44:26.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"17520a48b796551b169aeaeb9ce8fd53e46dff7d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-07 13:46:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/4/check/devstack-networking-ovs-dpdk/1fc7ee5/ : RETRY_LIMIT in 1m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4be9ad0698ae27ddb8bf76a91e09d915bf4f8b0a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-07 14:01:03.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/4 : FAILURE in 55m 26s","accounts_in_message":[],"_revision_number":4},{"id":"0a985ea4205f42a18efe3298b8a40438cd1446ac","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-07 14:22:53.000000000","message":"Patch Set 4:\n\n* sriov-test http://52.27.155.124/sriov/747774/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"abd1629bc8ba4c46a4d268b56c16a7d4e94afe81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-07 15:47:26.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"086f97414a84aa078392ff5ae494ab676227d300","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-07 15:47:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dacc6dcd8341acd19e5ba636b0ae0cd992115c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 16:11:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4c5f1715ba7f4929a99b78269a61ece7 : SUCCESS in 45m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3fe37c2057e74fb6ace9c515071d3895 : SUCCESS in 29m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2f89062cd532430087a52c121d7f48a5 : FAILURE in 1h 06m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e30bde155c14465befacf6c0c7388f3 : TIMED_OUT in 2h 53m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b18a17c0ffe0477c8a38e9fc49ce58fd : FAILURE in 57m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/da14650b51494c51a37fc2a7e989ded7 : FAILURE in 1h 13m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff5d50f15cc1407b9cabc1866e093853 : SUCCESS in 32m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b649a7dedbb407b83b255a9a529650b : TIMED_OUT in 1h 01m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee98c41c2084cba8b73cf6dff5e580a : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b0054748b794ffbbde6b61bf3dd90e3 : SUCCESS in 24m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d1ecf26e1c147abab597df30ddbc4f7 : SUCCESS in 26m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b58c7b6ac064abfb6642d2c47989ea4 : SUCCESS in 13m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ca2606940254fac96aeeccf3322502c : SUCCESS in 45m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c39269513d04546a8408030b7057580 : SUCCESS in 1h 13m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6458c637964f46388221f118fefb148c : SUCCESS in 1h 43m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/700683c55ff442a581a4bba956a2c439 : SUCCESS in 1h 05m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/90b3781252ed471a97926368d08d79fb : SUCCESS in 1h 07m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/675b8c3a53244a91bd2c6906bf1b0de6 : SUCCESS in 1h 06m 24s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9650139818f242eaabbef63240c0f210 : SUCCESS in 1h 12m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/141d4a0c82e1423f9fac8eed57bb57bc : SUCCESS in 1h 17m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b4ed09b716b045a58ce049bfb13d13cc : SUCCESS in 1h 23m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ffa02878aa424c959927896e61e6d789 : SUCCESS in 1h 21m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d0ab65297cfe493c8dececb0aaac1eab : SUCCESS in 1h 13m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/829b5d6b1da04364b338cef245a9c9aa : SUCCESS in 2h 00m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d99dfe04ea4c457199d3303eea68fa48 : SUCCESS in 54m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ffe17515e1fe4ae0b4ebd0c56176f32e : SUCCESS in 1h 11m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/98ad3a093b764923b3f77308d7eaa983 : SUCCESS in 1h 12m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dbc7a1b74c16494e86e201608e1debe8 : FAILURE in 33m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9929476a69154d1e91d589004ffad607 : SUCCESS in 1h 23m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b27611509b844427b795069c0fe7791a : SUCCESS in 1h 29m 43s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af42aa8c7af54d728ca1356fa1f34e46 : FAILURE in 1h 45m 11s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d8efa992549249979224f6070ae28172 : SUCCESS in 2h 00m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/22dcd284d598425ab1538dc119921a4c : FAILURE in 1h 46m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c17dcc46357f446db8fb9ec520914208 : FAILURE in 1h 52m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"41ba2c3b36ff05e92d639b2f1fc41def84be676b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-08 10:38:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a214949af2109d564ad9e0d971d360bcc3051f8a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-08 10:51:16.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0bf9676cb9b7d915f2ce79625b010b05f41f0dfa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-08 11:25:11.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/5 : FAILURE in 45m 39s","accounts_in_message":[],"_revision_number":5},{"id":"6c37bd14892127b26fbf577fdbca145905dbbd64","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-08 11:32:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/5/check/devstack-networking-ovs-dpdk/ea8b3d3/ : RETRY_LIMIT in 1m 55s","accounts_in_message":[],"_revision_number":5},{"id":"d998beb938c1b4406a92ac67c74e4e2eb13965fd","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-08 11:49:13.000000000","message":"Patch Set 5:\n\n* sriov-test http://52.27.155.124/sriov/747774/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"3eea502411dc61a6d866eebf6930b5be01f655f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-08 12:55:23.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"036ea2d7e8516513c987dbf079854159a59c833d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:11:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02961a94657b4d65a2addebf0ba6244c : FAILURE in 46m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11e1947038b54230b8e61cdcf89150ce : FAILURE in 25m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6aeff80f24d447879a7a76eea7764ec2 : FAILURE in 51m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6e62fbe0237b46f19399430e2074dadc : FAILURE in 51m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6a0c12f182674a5d822d688c844cec54 : FAILURE in 51m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0e2c0ba8c3314d2387795207db94c272 : FAILURE in 24m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4f58e5d53334a3481989c9ec1ab6c8a : FAILURE in 35m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d6e64d577a64e18b685de6f939f9144 : FAILURE in 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34479670dbe24d729665486a07273c09 : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf71af9cbdd47c98a3c10178bcf3b97 : FAILURE in 31m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22976cc033074668a533c3810cfaa117 : FAILURE in 31m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7393e538d2254274acb7c16681c71ef7 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7122b094ae504c79a4bb622ef7279b18 : FAILURE in 49m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6c0d89111014b99ba2235650116d69d : FAILURE in 53m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/870308fd49a64f90936c5ac2427236a0 : SUCCESS in 1h 43m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7f8707238c34235b52d7732c2a64a0b : SUCCESS in 1h 06m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/222b4524c18a4cd990722f3480e03e35 : SUCCESS in 1h 06m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/450d5c7e72164c9e91179f27cfa40a88 : SUCCESS in 1h 09m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c7e8646ed1641c78bc91966bbf52418 : SUCCESS in 1h 09m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a6282bc3dd7142c68c1b4b920e355576 : SUCCESS in 1h 22m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c6b3dfa20ca41f1a644909cc8d274c1 : SUCCESS in 1h 27m 26s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9597414a72cc4e08929f49ad75293f64 : FAILURE in 28m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c9eeba1c0a584ba5beeac1ca76065996 : SUCCESS in 1h 24m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8da3eae0744e4e9084d8d6365d732a9e : FAILURE in 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2332bc450004349967ab9e4c09f5919 : SUCCESS in 54m 08s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4a595cfcb8db46048a3098555e4eb92b : SUCCESS in 1h 16m 07s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6c331ad6eaf943d18296e0110b1673fe : FAILURE in 27m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cfbce1aa037b42f8bce8dc5ccb38dfd0 : FAILURE in 27m 02s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/81662bad1a414da9ac12f121ee8c3ea7 : POST_FAILURE in 25m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e0d17593e15d4e169c6f5d2a1c2d6d5e : FAILURE in 1h 26m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/48327090462348a98b53d130b251919b : FAILURE in 1h 43m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b1b578bc8e3d4c6783d71c917ac0a384 : FAILURE in 2h 25m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b40bec73b04e4df9861abb8f12947ab9 : FAILURE in 36m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e80f0eb58a374e5ca1e1cdadd60c5a7e : FAILURE in 30m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ae7f7bba0c1e5119ddd8226a130316a606869d03","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-08 14:21:27.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"6bb438d817e53dae747707e140ef6ad803c1dc6b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 07:52:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"11712a120008a4510c92736da00f55bd3d3f8e55","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 07:54:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6f4dec357f12d0a60cd6b10cfbd08acf34a7571f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 08:19:08.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"37d5601dd900f20d10e9790b72e5daad184a4b38","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-09 09:04:42.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/7 : FAILURE in 44m 58s","accounts_in_message":[],"_revision_number":7},{"id":"fde5a949a2e106ef1473c5929c1f51d2ebdc7bfd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 09:06:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a78ca2b0bd4993ec6027e0d011ff7a6ac458c2e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 09:10:45.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"0814d426c9af3ad8a919a55f9a89dd861e00796b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 09:31:00.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d7a2eb7c3663043bc2adadecf7e86ef2d64b2371","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-09 09:57:13.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/8 : FAILURE in 47m 54s","accounts_in_message":[],"_revision_number":8},{"id":"42a7a8c20bc5891165bced155ef0d6ff66ff69eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 10:44:49.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"563d37d85848e9be1da26e8f9b7f984d7d888057","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 11:48:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ce8215d7da58a211ddb7c18a721cd60fbe19a60f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-09 12:33:14.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/9 : FAILURE in 43m 47s","accounts_in_message":[],"_revision_number":9},{"id":"eee80209acdb16b2b4f46b461c76e8748343466a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 14:23:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8df7baaa6d1240bb50060b7229ffeaab596d94d8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 14:25:36.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0602bd28831568229cbfaacba64921ef44618fe7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 15:08:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0a7b990dd800d427cdaf267d45ee416f03dc8e38","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 15:38:52.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"59a31339b1821dd0962d6a93b99ca406473e43aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-09 15:49:25.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9b92c62b49d9773260e0e1e000555abb880e75f2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-09 15:54:24.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/11 : FAILURE in 45m 31s","accounts_in_message":[],"_revision_number":11},{"id":"b7aa5a6e2eb8a21baf6a32cb596739204b8213d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 16:14:21.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"6eb3646d2846bad5bd6bcd480516f7b1e0ccc6f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 17:02:25.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"de72423b06521656ef92b8a5d9c235ee0c054097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:18:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d8037a02099484bb30155fb806758d5 : SUCCESS in 49m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/74c555c7d3154556bc7678f8e1d3936a : FAILURE in 34m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5669db1f4f534256be1a6bcee9b57216 : FAILURE in 1h 30m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/687c1ee40e744a76b35b7945eab2fc84 : FAILURE in 1h 08m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c54ac6adc2c9455483543b38343dfcaf : FAILURE in 1h 06m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/63ab0a4e8a394ae9a70e052dc4806387 : FAILURE in 55m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44ebdf5123c74089a68f6700e0df9d05 : SUCCESS in 33m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54bbff09c1b048ea93fb9082bdc8fb6f : FAILURE in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd09e045e7d648e0b075b86602f466d5 : SUCCESS in 9m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/561bcb555263481fa96e326d9b97acc0 : SUCCESS in 18m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57756a0b4cc64105883ee0935cdd7005 : SUCCESS in 16m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2fb0a10b314e919ff6c2163fd212fd : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33e61b834d794658a72c5fa1fd08c5dd : SUCCESS in 5m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b1cf538fb04426d84f24e7652919ffa : SUCCESS in 37m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1111dfcb03ef4eef9fafadaaa1a9e149 : SUCCESS in 58m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1d89e61d8cbb43ef9eac527eb6c995bb : SUCCESS in 1h 14m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/212e9a7b082640dea2cd9ed83a2df6b4 : SUCCESS in 1h 03m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/69fd6b8abdce42768770b598efaa6133 : SUCCESS in 47m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/677848e42c9045e8a86c0939e1c19d87 : SUCCESS in 1h 12m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac2218cf570a407b81656780582b6759 : SUCCESS in 50m 51s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f4b9e33bb9954a999f3ad5c91d6cc4fe : SUCCESS in 1h 29m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3c0b9e9efba49f9aed6443f841ee2d1 : FAILURE in 57m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/796802bff6f040fa90798dfd570f5fa9 : SUCCESS in 1h 08m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e899c99a083b4e04b2dc1ac6e271e3cf : SUCCESS in 1h 15m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de9b2e4e3ed9415bb69894202b4f5427 : SUCCESS in 2h 02m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a99597e11eca44769426c5d13f1fb4a9 : SUCCESS in 36m 35s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/dbde62cc8546481d9173bc24624094bb : SUCCESS in 1h 07m 15s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b24c97eb7c5c4be69092f6e21abaf3d6 : SUCCESS in 50m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c2b708805c984f48a60e43b5c96c82f4 : FAILURE in 37m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1ac7ca28d2e34f2c8dc294fcfc64f602 : SUCCESS in 1h 13m 22s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e2124e7875bd45a492325a797834cfe7 : SUCCESS in 1h 08m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/828487a45e954e329f08f215cb1c5200 : TIMED_OUT in 2h 14m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9be21fea61524ed4a974b7c760c2d51a : SUCCESS in 2h 13m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/747bb163ffe94f7091d105630b9e87e5 : FAILURE in 1h 15m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a2c93d8bd18a439087225232bbf80762 : FAILURE in 2h 03m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9e4d906c37a1dab78554fd0851f2a91ff5cfa85a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 20:31:00.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6de17e5a21ab8df4bc13162029277d9cfca8a917","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 20:37:01.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"64efbad5c07553969d67fa6cf0a1da1ada730c50","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 20:38:38.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5205593f57aa5d68b30609a5147275f6c26dcbbf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-09 21:22:24.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/13 : FAILURE in 44m 34s","accounts_in_message":[],"_revision_number":13},{"id":"9725b2b711adbeead2e7d581ca4e170f5b3d054b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 21:48:11.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"4606b4a4a58f98d13a9e83fba062de9c9e13464e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-09 23:11:01.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"a8c3c72270054195d0b9fc6922fcc3b40e7e10eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 02:17:46.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf35eca7c96a458e9ebf75bc58f0c6f4 : SUCCESS in 50m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76c684c28d424758afcd96cc9c03579e : SUCCESS in 31m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/514cf5d73b0e485eb38b609046c5e3cd : SUCCESS in 1h 07m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5aa1fe6d590143f28e63b032741dbd9d : SUCCESS in 1h 08m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81c01d046bd047b89bbfea8b5a8a9a6a : SUCCESS in 1h 06m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f5d28d7b182b4356a520c2c910c4bddf : SUCCESS in 1h 00m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a4a148fa9dc4352855fc6faa5f4fea9 : SUCCESS in 31m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/130a0fb76df2435181709e7771701f8c : FAILURE in 26m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea7357e950541df9ca8cac48715d2a6 : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f8feb055a6b419ab9a11fa34c057d41 : SUCCESS in 23m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af2729d216d14c22953cfc656f257fca : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd46bc8b2574a22b5d216b58960455e : SUCCESS in 17m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b256f37e7fb6479d93b3a79a6bb1e37d : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f015a9fdb18403ba096aa55a1208336 : SUCCESS in 46m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28a59ed669fa40a48ad982210bdadfd1 : SUCCESS in 1h 20m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/063d45cbf61a4340a98a0ce2c256cd37 : SUCCESS in 1h 42m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b43e4bb132e4af8a578442a107b0136 : SUCCESS in 1h 03m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aa64243647cd4cfc9d593c533f542d37 : SUCCESS in 1h 06m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/99cb4c44babd4820a9c16edde543ad07 : SUCCESS in 1h 08m 41s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5218709a978d4db7b60fc9e77f1cf92a : SUCCESS in 1h 07m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/81f216778de148a492bdfb4471d05fc4 : SUCCESS in 1h 19m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/80a966b5261944ff907154d7dc16ae4d : FAILURE in 58m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df9f3237afca465dbdca86043e8e173c : SUCCESS in 1h 07m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/778aacf4956249e6ac1de2bf6d913acc : SUCCESS in 54m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e52511928c74057815cbd44aaca347c : SUCCESS in 1h 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/632ee6d759dd4503a3c2a84b0f16c7c7 : SUCCESS in 37m 28s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/556c43f86b4946baa5f76ad1baa252e8 : SUCCESS in 47m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/91a5adad5afc432c83fe7607c29ea31f : SUCCESS in 48m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a6be5234b03c4ced84712b2f2dfdf419 : FAILURE in 29m 16s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a3b916a8a965422cb8d04d90ae7cacd4 : SUCCESS in 1h 18m 15s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/79f9eb3121a34d43a39b43ee6963b15b : SUCCESS in 1h 35m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb4f5b31332b4b8d99b6c544fde50e5b : SUCCESS in 1h 54m 01s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/96ad2113c50b4915a6449c7eb652f2b7 : SUCCESS in 2h 33m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9bd8059393424ceab2a779c98a4b561d : FAILURE in 1h 56m 08s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a80497b6ba1443bf9317f1f034703b61 : TIMED_OUT in 3h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b031da08fc84587678707fed44c7b365a551e535","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-10 05:05:47.000000000","message":"Patch Set 13: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"fe98e4c4d5e3f1d4ba48aae65a9a6b10df5a2271","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-10 08:23:07.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"14f3525c52b054b00b3f2fe906b5e22797459d99","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-10 08:27:33.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d163ab2247ac5317b65f3c4074914cc958100394","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-10 08:28:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f99f2190bbd794ee47d07aa73b1f45b2bcdd32c3","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-10 08:30:41.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/14/check/devstack-networking-ovs-dpdk/c10bf27/ : RETRY_LIMIT in 1m 55s","accounts_in_message":[],"_revision_number":14},{"id":"c80cec599d9ac945c696a2843b418c4132607013","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-10 08:46:51.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/14 : FAILURE in 23m 08s","accounts_in_message":[],"_revision_number":14},{"id":"af73da543407f0d31197f45041b0549cf0ae126b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-10 09:37:42.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"89dad04826ca6e35fc4f8d6564c26cfc1cccf94b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-10 10:07:23.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"42e13d365acac7a7744aad7d747f85b5b3f8dadb","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-10 10:39:58.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/15/check/devstack-networking-ovs-dpdk/daa5718/ : RETRY_LIMIT in 1m 58s","accounts_in_message":[],"_revision_number":15},{"id":"5909db0fc2041103cdb0a978003e10440e8d1870","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-10 10:57:23.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/15 : FAILURE in 49m 24s","accounts_in_message":[],"_revision_number":15},{"id":"da0afc2ef9749a1d9bfe40dfce0cc27ca9c6d796","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-10 11:23:16.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"a93f34ce8075b8979952851f8110dc967aef81e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 13:23:23.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0277ab79d8f14c86a1b5d0582ed62d86 : SUCCESS in 48m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e136c3179a524af18497cba3a83af3e0 : SUCCESS in 27m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f0647e7c8f0947eb9424defdfc11941e : SUCCESS in 1h 08m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d270fe585b564d038e7b5495cd8de78b : SUCCESS in 1h 06m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5cdbfec856942fe88775097a7de6142 : SUCCESS in 58m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dcf81b16a30d4b3d87c4597898c177ca : SUCCESS in 59m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f9645c1913846939737626e900202e5 : SUCCESS in 42m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf242b15d36e4bd8baa2d058695a6bc2 : FAILURE in 32m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567306a3e08b41da86069e0f85ff2cc8 : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf0eb29df5814b71afd0c2261a87c53a : SUCCESS in 32m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64f083e3e64b48f4b7f3142b8a426ffd : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b56002a43b34ab38443ba92dd845ec3 : SUCCESS in 19m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/288cf42c9aca4be390a3d519329f1b78 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45e87a8bcd76407eb0cd51f237231086 : FAILURE in 49m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae0d94d926f54e769001f3ba7f35f6ae : SUCCESS in 1h 13m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/695b9a4678b444f0a038dbcd37afc764 : SUCCESS in 1h 46m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6799ef78f1764bd1bee2568f5a328a13 : SUCCESS in 52m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cd268b3ec7d45fe803525b3b741047f : SUCCESS in 1h 09m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/878852ccfc44460b8a3efe29811112f3 : SUCCESS in 1h 18m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e68c2268be746c98c0cf6964e4eebf6 : SUCCESS in 1h 10m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c07381b6bc824cc1a56e4aa400fc15e8 : SUCCESS in 1h 22m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bef726a0a89e44988194f2b1cf0706d6 : FAILURE in 1h 13m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b987d2caefda455cb730dfc1220b48a2 : SUCCESS in 44m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/90b75e2d4a2f40609ff99dc894aa5c99 : SUCCESS in 1h 15m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1271b63964aa48448e975f6000abae36 : SUCCESS in 1h 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09facff8848e405e8f54e0246cbf71a5 : SUCCESS in 37m 14s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/56e9dbbb25094de6bedfe59d0364b0b1 : SUCCESS in 47m 50s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6a4606a41f4542b3b9c532a1d008b981 : SUCCESS in 1h 12m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/31f0c3b9c56d460a8fffa33de5b6279f : FAILURE in 36m 21s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8552d8d91406469aaf0a720ef7538ce1 : SUCCESS in 1h 22m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aa1b41c8d6a14221a209be048c27b907 : SUCCESS in 1h 28m 11s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b030002973994d08bb6bf84efcdb757b : SUCCESS in 1h 47m 19s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/09ddf63221a746ed878ca7939f1d6774 : FAILURE in 2h 13m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/40ec99f42a1e4af4af70dd9d45191347 : SUCCESS in 1h 46m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e7ac0acb00344259985987f2247210fc : TIMED_OUT in 3h 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"242137ac4bbd7c27631fe1cb255b5d23d54db7a8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-11 04:33:52.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"fa60759ab7ab296ea136d1ff1034ee923d80ec4f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 04:42:03.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/16/check/devstack-networking-ovs-dpdk/0162ab1/ : RETRY_LIMIT in 1m 58s","accounts_in_message":[],"_revision_number":16},{"id":"002d6056cd4cf1cc5ad92eda673de964bbc410d4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 05:20:41.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/16 : FAILURE in 45m 46s","accounts_in_message":[],"_revision_number":16},{"id":"092fcd4b2d13a6ec35702c0f5a4c11a44c6bf82c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-11 05:51:20.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"29c3c2c62d62d3e14475f6645251c9d1a0c3c652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:44:47.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b32c6993b3f2426ebe85afbf38408e57 : SUCCESS in 51m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8413be35e7114212b0605ffdc612812e : TIMED_OUT in 1h 03m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2090dea71aa54069867f7b319f22b348 : SUCCESS in 1h 23m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/413413b553314e3181fb6789c2a3d02e : SUCCESS in 1h 11m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c4258ca90a8485c9632954f6e25f512 : SUCCESS in 58m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/641a0e5406bb45a6b468b64a4c1dfb9c : SUCCESS in 39m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042d0850e99b4ab096f73fa2a2efb0e1 : SUCCESS in 29m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9969466163a74c109b7634eba478d915 : SUCCESS in 25m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6caea751b9744973a35406050d868b18 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15970e5689dc4c4fbc780f7c8867afe6 : SUCCESS in 28m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/149595af86b4499c9251117083363e2a : SUCCESS in 22m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be9a9e709784d0a9bb999d16198544d : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06e5b7643beb4042a7023dff747b29a1 : SUCCESS in 7m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1835f1d554d842ae8bc45cc5f6d675e0 : SUCCESS in 43m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b807aa18ed04cff9f475bfd58b5d082 : SUCCESS in 1h 13m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/59111888dc48467e95455e8433775d34 : SUCCESS in 1h 11m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be036e8c159341e58253d238d06226d3 : SUCCESS in 1h 03m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/beed5f141a644a54a9b852ecc5a48769 : SUCCESS in 1h 09m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ed7c919cb7de4faea3839c5407b97cdc : SUCCESS in 47m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/767604fb910144f3bbb5b1240f8b5837 : SUCCESS in 1h 09m 22s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e9120f110a5c40edba31c2d89e6fdcf5 : SUCCESS in 59m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc73a5ab999d456590efddd9ddc028a7 : FAILURE in 54m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fdd22bee0d14166aca648f358072203 : SUCCESS in 1h 06m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/207cc8a349f34756890dd0cef73fe73d : SUCCESS in 1h 23m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79dbef13a1e84464b731669ce6f05fa0 : SUCCESS in 1h 56m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/824882a4576043b183d3f16b6828de4a : SUCCESS in 55m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4f5d25d4675940d9a7e22fd6a9e44fd8 : SUCCESS in 1h 01m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ef44131e029c44109356e1285e53e354 : SUCCESS in 1h 35m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/13e8bd459baf4c54baad41e7e8daa74e : FAILURE in 36m 47s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6121cfd5c3204412a46aa0454fb46af3 : SUCCESS in 1h 20m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1de22d403c8a4993a13613ba45c5aa51 : SUCCESS in 1h 09m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8def4133e0ed4b6dbd4260c3aa878324 : SUCCESS in 1h 55m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/efe8820565174e708869cd0c7da535e2 : SUCCESS in 2h 40m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/436827cdffed4725a12a88a2f0668481 : FAILURE in 1h 11m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f166afeb1ba84376baaa40b2621867f4 : TIMED_OUT in 3h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ba0b7324c1b78c11ecb4267a17b3eec9f7adbf20","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-11 07:48:53.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"a54745d3dcfcdff9822b55021889a13e46f550f3","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 07:57:05.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/16/check/devstack-networking-ovs-dpdk/33b2295/ : RETRY_LIMIT in 2m 01s","accounts_in_message":[],"_revision_number":16},{"id":"99487a9184334617820d5dbe603a01980a85aeaa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 08:33:48.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/16 : FAILURE in 44m 19s","accounts_in_message":[],"_revision_number":16},{"id":"b6ea03c1d1f66965fe5acaa497e0c43c4ea5cc35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 11:01:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/89cd726aad3843a395c64f96833b2891 : SUCCESS in 34m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/39fcad6cc66e4591b30e66d280466a4d : SUCCESS in 24m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/199538ac3de34003801435796070cda3 : SUCCESS in 48m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d87d31516e7e4c17b4b1986028effa44 : SUCCESS in 43m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b862ade7e339492e944bf80c145e574e : SUCCESS in 49m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4d6ee1e6d8b448ff806cdacb73216330 : SUCCESS in 42m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e48477f2a4284d99833e9639f1455317 : SUCCESS in 21m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/881a7d53715343f890d99d0f99bd8a68 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e71dc4bc07d4dfaaaf6f9ce66df71a5 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc0f28e6303a48c9901b01a4f2cbd43f : SUCCESS in 16m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea77c83b8fb64856bd5d5d41ef6dadce : SUCCESS in 15m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dece67af564459aa9f3774e8259f62b : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45796de048f64a8eb7451157d07b9a42 : SUCCESS in 6m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80f02c3e0d4d4350a29f2010f002133c : SUCCESS in 38m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09a3171f2af74cf0b430f9572d47b135 : SUCCESS in 1h 05m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ef884c0a8a5843268ee112c67b9ec477 : SUCCESS in 1h 14m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cf123f4ad2d416481b49366cbbfb53e : SUCCESS in 1h 07m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0d4c6ef75d5e4893905ee85b417aedd8 : SUCCESS in 47m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1f4d47cbf9e742b9907e0d3564fd1c73 : SUCCESS in 1h 09m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4de514f5af34dc1825b40d76951ae9f : SUCCESS in 49m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/44c2a79fc816415f9e5c133673683313 : SUCCESS in 1h 22m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/341303f852bb4abc8c063bac7c988180 : FAILURE in 57m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91282515c4b6419d9c2558e427144380 : SUCCESS in 1h 11m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/678b693fdaf94acdab1b9338193d11a7 : SUCCESS in 1h 16m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efc1e0d2288143e5b60c86fcb67451d8 : SUCCESS in 1h 56m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb0b3ec51b5e4a2399439107adc6f8a9 : SUCCESS in 56m 31s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/dd296c0736d44506b7a51e04d83ccbc3 : SUCCESS in 1h 16m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1c8e0801266b49fbbd715d73728ff5a3 : SUCCESS in 1h 16m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/90993d4c5c214ffa8d20da70145518c3 : FAILURE in 38m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ec024d5ebfae4c099b7927b019ebf804 : SUCCESS in 1h 14m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80002593dcf14756b3ad976e5e4e1063 : SUCCESS in 1h 36m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f96387dcb5cc44f38a5f871c257bec7f : SUCCESS in 1h 48m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ebda5a27002d43628d2264d076c943a5 : FAILURE in 1h 42m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2817fee7e8cc4199aa3050be80c1c269 : FAILURE in 1h 51m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8e1f23507d3d43559929a2138dfcb8f7 : TIMED_OUT in 3h 05m 05s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bafca8d39ea6a4e5973ae25ad2c512665b3a52e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 11:01:27.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/89cd726aad3843a395c64f96833b2891 : SUCCESS in 34m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/39fcad6cc66e4591b30e66d280466a4d : SUCCESS in 24m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/199538ac3de34003801435796070cda3 : SUCCESS in 48m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d87d31516e7e4c17b4b1986028effa44 : SUCCESS in 43m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b862ade7e339492e944bf80c145e574e : SUCCESS in 49m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4d6ee1e6d8b448ff806cdacb73216330 : SUCCESS in 42m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e48477f2a4284d99833e9639f1455317 : SUCCESS in 21m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/881a7d53715343f890d99d0f99bd8a68 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e71dc4bc07d4dfaaaf6f9ce66df71a5 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc0f28e6303a48c9901b01a4f2cbd43f : SUCCESS in 16m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea77c83b8fb64856bd5d5d41ef6dadce : SUCCESS in 15m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dece67af564459aa9f3774e8259f62b : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45796de048f64a8eb7451157d07b9a42 : SUCCESS in 6m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80f02c3e0d4d4350a29f2010f002133c : SUCCESS in 38m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09a3171f2af74cf0b430f9572d47b135 : SUCCESS in 1h 05m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ef884c0a8a5843268ee112c67b9ec477 : SUCCESS in 1h 14m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cf123f4ad2d416481b49366cbbfb53e : SUCCESS in 1h 07m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0d4c6ef75d5e4893905ee85b417aedd8 : SUCCESS in 47m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1f4d47cbf9e742b9907e0d3564fd1c73 : SUCCESS in 1h 09m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4de514f5af34dc1825b40d76951ae9f : SUCCESS in 49m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/44c2a79fc816415f9e5c133673683313 : SUCCESS in 1h 22m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/341303f852bb4abc8c063bac7c988180 : FAILURE in 57m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91282515c4b6419d9c2558e427144380 : SUCCESS in 1h 11m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/678b693fdaf94acdab1b9338193d11a7 : SUCCESS in 1h 16m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efc1e0d2288143e5b60c86fcb67451d8 : SUCCESS in 1h 56m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb0b3ec51b5e4a2399439107adc6f8a9 : SUCCESS in 56m 31s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/dd296c0736d44506b7a51e04d83ccbc3 : SUCCESS in 1h 16m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1c8e0801266b49fbbd715d73728ff5a3 : SUCCESS in 1h 16m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/90993d4c5c214ffa8d20da70145518c3 : FAILURE in 38m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ec024d5ebfae4c099b7927b019ebf804 : SUCCESS in 1h 14m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80002593dcf14756b3ad976e5e4e1063 : SUCCESS in 1h 36m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f96387dcb5cc44f38a5f871c257bec7f : SUCCESS in 1h 48m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ebda5a27002d43628d2264d076c943a5 : FAILURE in 1h 42m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2817fee7e8cc4199aa3050be80c1c269 : FAILURE in 1h 51m 57s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8e1f23507d3d43559929a2138dfcb8f7 : TIMED_OUT in 3h 05m 05s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7e39525ba1e91d854730cb624d4b84f0d39adecf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-11 11:09:47.000000000","message":"Patch Set 16:\n\nAs asked in IRC, what happens if the minBW is enforced in the driver but not in the scheduler? Placement won\u0027t have this RP but maybe the admin wants to locally enforce this QoS in one physical interface. With this change, because we don\u0027t have a RP, any possible change will fail.","accounts_in_message":[],"_revision_number":16},{"id":"053b5927a4e16414e2d0faa5d253ababf3d3a0bb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-11 12:04:32.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a28fc2f2eaf8d8ed586d47bd8a3d483f3044f393","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 12:19:18.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/17 : FAILURE in 14m 11s","accounts_in_message":[],"_revision_number":17},{"id":"b81ef0b0bf23a1ddc00a461421ebcf5e568ad293","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-11 12:39:19.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"a9c62a8b9ed52badf8dc98c33a91670ba9eb0c53","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-11 12:44:28.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"cace28ef9c41fe4878bc2c85ee28b524bae4b92b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 12:52:26.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/18/check/devstack-networking-ovs-dpdk/013d828/ : RETRY_LIMIT in 2m 04s","accounts_in_message":[],"_revision_number":18},{"id":"f84825be6f6094907218cc29aab9caeec79c3e9f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 13:25:03.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/18 : FAILURE in 45m 10s","accounts_in_message":[],"_revision_number":18},{"id":"5affd989d42fc3ecc5dc14e5d45fa36b25d080cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-11 13:38:46.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"427d8d14e06572e645beef4f3dc518d5c8cdd651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:20:04.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/012fda0cc28743609dce1631d3d6f797 : SUCCESS in 47m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49cfdf068aab45cba5429564b73dd28f : SUCCESS in 36m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c72ea2a6c3544da3bba22504ecec351e : SUCCESS in 1h 27m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33b7f59d2fbc4067aab4d01ed3ce8989 : SUCCESS in 1h 25m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/797b2a3a354748bb8ddd8df1b673fe70 : SUCCESS in 1h 21m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/37d772bd625646e995890997638a9fbd : SUCCESS in 1h 08m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a7cd22f5853473fb5005b6e7f9f0b94 : SUCCESS in 21m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bca0e989a4ee4864bb4bad721f5dc7d7 : SUCCESS in 32m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f286c5dc1ed47e08fb51e443dd1ea57 : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6f5df1a6de249ffb3b5f688a69b6fd8 : SUCCESS in 29m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7853f2e93534917b5f04eb6a9062ad0 : SUCCESS in 25m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c658ba9d7445ed9db1e4cb92920b7f : SUCCESS in 18m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee6be4af127743728d81fcfe8da0f747 : SUCCESS in 11m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bbfbf024073455e95715ecb20c7eeff : SUCCESS in 51m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbdd3cf0054e41c09ffd7c07f8c6b457 : SUCCESS in 1h 14m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4c0a2f97318d4327a8239454d6102fe3 : SUCCESS in 1h 12m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c087b60a09834a2ab71aaf0c60d18369 : SUCCESS in 1h 13m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/87e535a86e874b35979c8996be5f1bbd : SUCCESS in 1h 12m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d37fa11713c5475890fcbc40de40d3ea : SUCCESS in 1h 09m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ab2e5e0b9494aee8b3dfc274d26978e : SUCCESS in 1h 15m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1e318b9fa2424c76af2c08499c74648f : SUCCESS in 1h 23m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f04c4b5a9fc846658fc6e6151a7fb59e : FAILURE in 1h 28m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/237bd1cba3054c14915676b9614f4474 : SUCCESS in 1h 06m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec2cf6110b3b47318812596bac398338 : SUCCESS in 1h 11m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0371d1ba1c3e4aacaee6b86dec815625 : SUCCESS in 2h 32m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f911f53a0a3497e930adba5196c6f15 : SUCCESS in 47m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bb5697ebcd044f2c9ae58de1dcab2b8e : SUCCESS in 1h 02m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82afc921c9c14144b16e19e6739e0fd5 : SUCCESS in 1h 05m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c5cb6312723b4b17b9ee735f5d71e5dd : FAILURE in 29m 46s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/56f7794388234da5945beec3c892192e : SUCCESS in 1h 15m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/831160dca4b04e57a48e0f0f4bb24dc9 : SUCCESS in 1h 35m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62a6aed554734872a4c32b7a6f568f29 : TIMED_OUT in 2h 15m 30s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b7cd19c197f047c09164996d52d8833a : SUCCESS in 2h 26m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb6a045948e34633a17996ed6ae4d5e6 : FAILURE in 1h 14m 25s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/db128002d17642449ed878cbcf8d23f7 : TIMED_OUT in 3h 07m 03s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"131a0b3e82c36aaed70c7a0f45adf5871684f2b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-11 16:36:08.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"0a335c62ca343ad8a3740edcb719b6e94997b066","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-09-15 18:19:48.000000000","message":"Patch Set 18: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"eb64471ad9e508fb0946f59bbf22440975df7304","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-16 09:00:45.000000000","message":"Patch Set 18:\n\n(5 comments)\n\nthanks for the review","accounts_in_message":[],"_revision_number":18},{"id":"2832c292ce64b3e3747e006a5655df0ce76da914","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-16 09:26:42.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a01c998c4cdc494af50e4bdac60196c6fe2f7a5d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-16 09:27:36.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"0a58465bb4b2d47503ed8319909d4f44bc9d66b1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 09:33:24.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-16/747774/19/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/55b5e2b : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"45088ec8a4e2607fa018a56c802e763ccb197994","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-16 09:43:52.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/74/747774/19/check/devstack-networking-ovs-dpdk/a717158/ : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":19},{"id":"b1df816fa80a5d562657a4324c5067838e6fada1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-16 09:51:35.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/19 : FAILURE in 24m 19s","accounts_in_message":[],"_revision_number":19},{"id":"894b238c06904a84c2b5a5aed746e8d8bf86a79b","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-16 09:57:52.000000000","message":"Patch Set 19:\n\n* sriov-test http://52.27.155.124/sriov/747774/19 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"479e8669a3eaaf2758b23b6a5dd446637d4ebbb7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-16 10:42:22.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"e2af7d5f34b229b369d2de2e8118676477d4003f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:17:52.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eeed70e9a9c44541ab61972e74829b13 : SUCCESS in 49m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2edaa4b42a3498a8993e41c64f1b440 : SUCCESS in 39m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3997925302eb4c988316b5eae8867553 : SUCCESS in 1h 05m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d21e2f9bc2de469f87c2ae53373001cc : SUCCESS in 1h 12m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a889f78e0ebc42a29682b5a855ed7c63 : SUCCESS in 1h 20m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bee79fb97dfa496d958908bf8b6b6431 : SUCCESS in 59m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b5b89606b524d55875ab4e1d8f4ed47 : SUCCESS in 33m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e5391b0f67647c3bef549441e165f9e : SUCCESS in 26m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b836e128f2344521b3b585ad01cfd0f0 : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d865b9f0865e4d95957da2082f01f639 : SUCCESS in 26m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71c7660dfe5040788413d52d61c294cb : SUCCESS in 26m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bca02c68b2e4cc983db888c4e0dffc5 : SUCCESS in 18m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2484ffb097946e19f48b4639cc2fcdd : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93bf1bb385c6444ca375de6b6e1b3ee6 : SUCCESS in 49m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/397fab0671f14bd384bda5790c763f10 : SUCCESS in 1h 12m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0e4e7a20c2cb449486f3611d6d465e7d : SUCCESS in 1h 41m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d53368b36d1d4d0aa563f561fe506af3 : FAILURE in 53m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b0fc1f5f7254bfe9b617fe9f75c1778 : SUCCESS in 1h 05m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3a61e09232ae46e88ff1bfba2ff9a7d8 : SUCCESS in 1h 08m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c0598dd43be4fcfb27a65ad6c784125 : SUCCESS in 1h 13m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a708e0518a904f45af74838896946c0a : SUCCESS in 1h 33m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a700577a0e8946e7abeafc1990683d6c : FAILURE in 1h 21m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/243deee166ab4248a64f700521ffdbae : SUCCESS in 1h 21m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/761a37c6b7cf4eff8005910a48d1512f : SUCCESS in 1h 24m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9bed1006d8de4534afaa7766978afbfa : SUCCESS in 2h 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d6b53c3e59144338c98601efd4c66e1 : SUCCESS in 1h 00m 15s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1591064b57f14077b4b763da6c859346 : TIMED_OUT in 3h 14m 51s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e849eedd20524bf7ba5089bebf2e975e : SUCCESS in 1h 16m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5ca9c4adc45f456496b3a2945f967395 : FAILURE in 40m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/267b547003a8408fb537d545fc9b7e29 : SUCCESS in 1h 10m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e3bd3451f9fa4d6ab00139fd9739a7de : FAILURE in 1h 24m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bcd7040f752f4229886a20cfc0c79728 : SUCCESS in 1h 47m 56s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6bcea037565647f58e7144f585af37f2 : SUCCESS in 2h 30m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf24663c67274bd7914ffea34277bf2d : FAILURE in 1h 48m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/67073d2338c74b0b9067be634c49f746 : FAILURE in 2h 04m 39s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b99d9e329507f531fcbd04eb631f808b30c146b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-16 14:25:45.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"b17bf6f2a741a249b39f449972fdf4b9472e6862","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 14:33:20.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-16/747774/19/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c4cf8de : FAILURE in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"f36ec702adc1f6a7b81aca43a580f156072fdc89","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-16 15:13:02.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/19 : FAILURE in 46m 11s","accounts_in_message":[],"_revision_number":19},{"id":"83ebc5d1ce69c1749f9eca67bf2e1ad5670a71a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:48:38.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a833121eaae4bacae4365e047d570fd : SUCCESS in 54m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/091028f6cda546a09aa02c4caa06e760 : SUCCESS in 36m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f8c71f6beac142608b075c4be536b4bd : SUCCESS in 1h 21m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/658b494149c24b69b9a3c67300d0a017 : SUCCESS in 1h 08m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf2888061b4b4e86b0a7bb17e2ac2c14 : SUCCESS in 59m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/66eb21ae0a8f4538b3763da4c581e209 : SUCCESS in 1h 12m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c426f8dffa4048ec964e1b0e9cbd92df : SUCCESS in 37m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5341d0c8f634e1186db7c02d5ee1ccb : SUCCESS in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1d9d66f26a43218ab1acb8f1a75602 : SUCCESS in 15m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a08a320af6b4f61ae3976e1686378f5 : SUCCESS in 26m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72110c178ccc4ba6a072d6d45d2a1ec9 : SUCCESS in 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4cc881282f410bb12bc873c08cb0f6 : SUCCESS in 16m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80e8bb83df404e76ae71b1d3ed3ba353 : SUCCESS in 8m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b85c762ee74d423f95ccb4f10ab2936d : SUCCESS in 58m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6171a562e7474c89b28df39472fd3fb2 : SUCCESS in 1h 10m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1c6145c4a7ea4c898e3b08a17a566d21 : SUCCESS in 1h 40m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3a05f9fa353456fb688ae9c8385d6c2 : SUCCESS in 1h 00m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c9cbdc6e9044cb48d4ea51e75d87a90 : SUCCESS in 1h 06m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7437bc11d2224d59b17818f9a7ac6c55 : SUCCESS in 1h 11m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b555c45924ff4188baa0f3737dc0ebf5 : SUCCESS in 1h 02m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/62e69b72f9694251b005af7a8f034de1 : SUCCESS in 1h 34m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1151a01b02af41f0b12776f425da2f13 : SUCCESS in 1h 32m 16s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89b405f5b2b54fdc90349438f4528cb3 : SUCCESS in 1h 17m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d0de3a6af9e4a189597e828a6329837 : SUCCESS in 1h 01m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c2c869e8c01430cb7d777f3f312396b : SUCCESS in 1h 32m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7029541c11aa4ff9abaf4f3f36ecc6f5 : SUCCESS in 58m 45s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/44070876e4c34a39840f73c4fb6f90ff : SUCCESS in 1h 10m 52s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/70538426883c4dd8a915988c2ade4296 : SUCCESS in 1h 07m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3c8d56268ede4ebfbda06629c0a9e11f : FAILURE in 33m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e687140cada44114bab977eeda0795b0 : SUCCESS in 1h 13m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ab2640a076d84810954f55c22554b797 : SUCCESS in 1h 43m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22e645c927e44a82b10475cc28ba49c2 : SUCCESS in 1h 50m 50s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/97f2fb10010142028c3bc85d63476571 : SUCCESS in 2h 23m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/25829bad9852456e995e7be31e39581d : FAILURE in 1h 47m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bd2dccfcb5574fb0a841e23d5f7c0a17 : FAILURE in 1h 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"43a248e1a2450d61be35c24715f6404b9091dd0e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-17 05:10:11.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"31e8b4db5b9996b084f2ebda4d4f1a5428435c2d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-17 08:44:45.000000000","message":"Patch Set 19:\n\n(2 comments)\n\nThanks for the review","accounts_in_message":[],"_revision_number":19},{"id":"fd6791c0e3b30b72bd6611f017309cf617ad7bde","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-17 09:58:46.000000000","message":"Patch Set 19: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"ff91f184c31f0202b97af2a7db9fabbe4d05a776","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-17 11:39:00.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a90db4faba413d7cfc87b5e2c1640321f68ad77b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-17 11:39:34.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"00e9f859a40ae3588d2f1a6609142d5fba2e775b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-17 11:45:57.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-17/747774/20/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c40b7f5 : FAILURE in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"54e722ce1742431de3d57208e282b647b8c2dce3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-17 12:03:40.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/20 : FAILURE in 24m 02s","accounts_in_message":[],"_revision_number":20},{"id":"104641be12a390c4503856d8d41230e8b61c2853","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-17 12:09:58.000000000","message":"Patch Set 20:\n\n* sriov-test http://52.27.155.124/sriov/747774/20 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"5d2d44dbf94441f4f4f9fe11de632949a9eeef16","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-17 13:00:37.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"dd5c5fd05442dba5d1cce84aaf47dcae6701348d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:05:12.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0db59db02dc54c1eb4ba76f8b431eb80 : SUCCESS in 53m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e26c8956e31468585b559269620a28b : SUCCESS in 36m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8c0b296b53b34938929914e7b1637a4a : SUCCESS in 1h 17m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/82af0b937b514ca1a1e597894eb6509e : SUCCESS in 1h 10m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b002907b896742f5b2bec0c3b1bb6f37 : SUCCESS in 1h 13m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/45998a34a30d4f259eb8ce5cd88b443e : SUCCESS in 55m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e27b21d1cf2e4e9f89b264f9b750a164 : SUCCESS in 34m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93171fd380824195afcaa34f23ca8f87 : SUCCESS in 25m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513f076b25a44a4ea96a7de16c594a16 : SUCCESS in 10m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f9f8c6fe8547dd9ac4c6c1f47d2c99 : SUCCESS in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5551b26251e643e29edec4820a7229f6 : SUCCESS in 24m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668ca611cd6a45c19b641030426d3f52 : SUCCESS in 20m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68ae95e20b66436b9013c68a84c0328e : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d76b1112b2d439cba395d96f32bfdc4 : SUCCESS in 45m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c3b5148932d4f91b9566a706e75153b : SUCCESS in 1h 14m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/44d6b0d3646a451eb1368061883dbc35 : SUCCESS in 2h 09m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3fc208d733d435a87b8c9754c572c74 : SUCCESS in 1h 08m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dafb750f05f8496db9d2d36b9e3c3972 : SUCCESS in 1h 04m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d1b210a7e4704c1391d350ae06c91dc6 : SUCCESS in 1h 17m 24s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1895d1f801a746a1aab8fa7cf3d1e758 : SUCCESS in 1h 13m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a7a2b6ed477496d94a9594cc4be4774 : SUCCESS in 1h 31m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90c6a17cd0a646f1854d1287b9ea3c47 : FAILURE in 1h 31m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d787d37ff1794e5bb7943d42b01fe584 : SUCCESS in 1h 08m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de8019d92cb64f4eb95b4f30aa30c9be : SUCCESS in 1h 22m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5bb8abd13dca45d7a602f58ddd72e67c : SUCCESS in 1h 48m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fb6889867aa4249bc7cb2c3b9417e76 : SUCCESS in 1h 01m 38s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/3132fc68535547459a2cd8d9e44132fa : SUCCESS in 1h 06m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c44a4d5f21a64c31b33b37b262d8a5f0 : SUCCESS in 1h 09m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ea981d3cfc4945b08cac62e8dc9d2abf : FAILURE in 34m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/10c1ed5478124aa8b6315efcf6ba7888 : SUCCESS in 1h 32m 50s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8186ccc9416a480a807f23b14ad973d3 : SUCCESS in 1h 29m 21s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af34db7226ef49b69afd72159c565de6 : SUCCESS in 1h 45m 31s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f00a059df9644a02b2eb1b553abad854 : SUCCESS in 2h 38m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9aa4a4b74b2c4694b6885e7d7ccdb58e : FAILURE in 1h 57m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1341f8ad46a5401f8bae3e1a0bec3c2c : FAILURE in 2h 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e2a31802ea1e636198332624b8b7a1e797794f37","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-21 04:25:34.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"f0432925a05d74e473be3673cc41b6719044c931","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-21 12:56:02.000000000","message":"Patch Set 20: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":20},{"id":"b95447476241a34073eeaf8a8d9672d29bd5dcae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-21 14:36:10.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"d8d67bca628f06ecf8b18eb8b4b57d8025006b5e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-21 14:43:31.000000000","message":"Patch Set 21:\n\n(7 comments)","accounts_in_message":[],"_revision_number":21},{"id":"75d964ad36ae5e67912bcd8b332e1603acf3ee6c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-21 14:43:57.000000000","message":"Patch Set 21: Review-Priority+2","accounts_in_message":[],"_revision_number":21},{"id":"1f5668901e10e0cb69622ff41241956a501e1c93","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-21 15:04:22.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"91f09e6b3530ba3305d6b91a83cf9ec17ae5dd32","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-21 15:21:28.000000000","message":"Patch Set 21:\n\n* sriov-test http://52.27.155.124/sriov/747774/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"c34414a16ed67ae27ebdfa7e303654dbdbc5b281","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-21 15:47:03.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"9ff4f65b68189b40d0428c2b22ec698f48ca9b8f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-21 16:18:27.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/21 : SUCCESS in 1h 41m 39s","accounts_in_message":[],"_revision_number":21},{"id":"7b117dd9afb38ec3a1a8106dbc2f79bbe6e0e232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 23:23:25.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb19245c9fb34e56aab4efbe884fc670 : SUCCESS in 50m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/419775a49fa44bd18f2b2c935af2a57e : SUCCESS in 34m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a847899a49fd4fa1a0e1df2454954201 : SUCCESS in 1h 13m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d712b3c2184c40a99ad5bf3c4f47469e : SUCCESS in 1h 08m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fa057be4ddd4811a2326c1b7eb1edb2 : SUCCESS in 1h 07m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fa4b820509224e8f8c5f27d09d31ceb9 : SUCCESS in 1h 19m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/866fd7b432464d608ad576ce61cad99d : SUCCESS in 38m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cb0af7a01914876812c858bf9ac5410 : SUCCESS in 25m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4044e8b2b65a4260a176fb1593ac0db7 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/864ff53cbdc643c79b338f3ed8a3ca25 : SUCCESS in 36m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c2049e39cdb44d98b3f086ff83b67a4 : SUCCESS in 35m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d6731dc34294af69b55e51f66c607fb : SUCCESS in 17m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a3700adab5d4d8ead525fb4d32a3b6a : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e2b28c262bd4a56bd33dea2514813c4 : SUCCESS in 57m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a5ffe5a20f5492b9e1d5f620a7b78c0 : SUCCESS in 1h 15m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/57d03cca88c94a279c1fe95c1b6bd4c2 : SUCCESS in 1h 45m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc56b8544b254680957082140b4c9bda : SUCCESS in 1h 22m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e640d10c2f4b47f68f690d1a7d012aa5 : SUCCESS in 58m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/37321c11597441ec8a489a2134cb2411 : SUCCESS in 1h 00m 02s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e15dcdde211e4dfe8289c12908457777 : SUCCESS in 1h 07m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1761c87684b543cb87099bb87a7877d0 : SUCCESS in 1h 57m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/adcf8339203f4d8cb51769cb92ab6833 : FAILURE in 1h 23m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2e24a017ba44a43b7e979642a8ac5f6 : SUCCESS in 1h 10m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/827555a7124242edab5554e4c76a2766 : SUCCESS in 1h 12m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c48438386b044d2bb86205f1e33a1fce : SUCCESS in 1h 48m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ead19649fd3f4fc7a6c5b14f96bb34cd : SUCCESS in 56m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d6b71dd33ea9439280e57c0ff5d55c28 : SUCCESS in 1h 16m 28s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/49c8925969064f6f808d096d89c73a70 : SUCCESS in 1h 12m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1522d44586bb44f0af8c7e733cabe225 : FAILURE in 36m 03s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/12c5cdce2c1d43eaacc10f57fee7affb : SUCCESS in 1h 14m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/00eeef8e4491433a9ce4c257377c3e1f : SUCCESS in 1h 29m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/333535939ba64272a9be4f1eb39b3bb2 : SUCCESS in 1h 44m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/647f33b193f6400e9af7e57f5a956f1e : SUCCESS in 2h 22m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/78a4327f7044455d98d9ecc447f34a3e : TIMED_OUT in 3h 03m 55s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1f2d41106c674ffd954a2502db7312fd : TIMED_OUT in 3h 07m 17s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"45676a0325d39a13888d1bd25077055d1dabe65c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-22 03:53:03.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"30deb84f728c144bb54a5ce0ed4a480267a17651","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-22 07:40:44.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"ad469036682b0e38796fc3f401d2367a1049e50b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-22 08:04:23.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"dad78d83cf21ebad2ffdcf28f32d7e1b856eac54","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-22 14:16:09.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"a75605cb19ef26eaaef0a150b8443d2060e98343","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-22 15:08:06.000000000","message":"Patch Set 21: Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"bfa6a7ffdb3ceb0bbac9c55d060b099f47d0e8c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 15:08:38.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"9599e472d3e462154ecb63083cf40a80a08d303a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 18:37:13.000000000","message":"Patch Set 21: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef176538a82847b280bb68e629897bfc : SUCCESS in 45m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4739b666d36549afa54e4b947ad42f74 : SUCCESS in 1h 05m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/22284260b72243c09a1fdf3886d4ac5c : SUCCESS in 1h 05m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f65d0c2a9f8d4a6daa5d4ce965d0c11d : FAILURE in 19m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36023d7494984a25bdbfd0b5f4fab930 : SUCCESS in 48m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95f4198564ab43f68d9ae2e120e79e1b : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed96b055ee7436e9adf567184267197 : SUCCESS in 11m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75cc8899a79f4bd7bc2f5bdfd3e5d626 : SUCCESS in 24m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/083dd38b6f924d3d8ed2b7a03091fc87 : SUCCESS in 21m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c82e5423b6248959f5b140d365a8936 : SUCCESS in 17m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e7cb80a6a6142abb0f572f037c39eec : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/70842af9a445464fbd7475151ad4c871 : FAILURE in 42m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbef2835286c4fb3b703a5b0e49bdfd8 : SUCCESS in 1h 11m 26s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bedfd177e21b4207b75cd244fd79aaa4 : SUCCESS in 1h 08m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8f35360d77444cafbf78bb9096b20471 : SUCCESS in 1h 13m 00s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a856976bef004b22a39d5850da57d67b : SUCCESS in 1h 19m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/adf58d58469d461ea18a76baaf4db787 : SUCCESS in 1h 20m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4714f57d140e44e4a37d288d4e415577 : SUCCESS in 1h 09m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ddd89145cbf5412ab77a4a15f887ea6d : SUCCESS in 1h 08m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e8e14a1990dd467ba154782ad41e79ec : SUCCESS in 1h 48m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bf8e16595784add823fed671889b19d : SUCCESS in 42m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fec1b1e8abbd45109ca624ac21b62fae : FAILURE in 16m 35s","accounts_in_message":[],"_revision_number":21},{"id":"2fb576cfead9b2a0044f35fb98277f88252ad8e7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-09-23 04:13:19.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"0a53b548784eac066a47eadfca2b43ea22b12e2c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 06:19:58.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/21 : SUCCESS in 2h 06m 10s","accounts_in_message":[],"_revision_number":21},{"id":"ed4f42a1125149686febcbe167d84aff25cbb60e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 07:24:26.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/677e0f15b4264c71991cdb5fa42a4bec : POST_FAILURE in 47m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e5847b263db4270bf4584671f0841b9 : POST_FAILURE in 29m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/33e8e52c0cc14ba1b676b02dd0f0eb95 : SUCCESS in 58m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/123615622e60430e990e096edebc7714 : POST_FAILURE in 55m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a3e68d8091c3415e8e7f02600dd3f116 : SUCCESS in 54m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9e4a0fac342c4367980d77e9222c7cf8 : FAILURE in 11m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10a729a5e9fb40598b6eccabe7344883 : SUCCESS in 42m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c175eba7f0774ca78ffcdba8e4f92961 : SUCCESS in 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb85331a4e374c5896fe70bbbc4ce517 : POST_FAILURE in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb1c8d2cea3144b5bf5f0ced91ebef58 : SUCCESS in 25m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1413480283d24ef3bf202bdb2cb83092 : SUCCESS in 22m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40798784bd645d9af5c10024997efe8 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12ede89a3bc040ad9a78a64f7d190136 : SUCCESS in 8m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/444cea830ec94206887c646491f189bb : SUCCESS in 48m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbf4673066684b0ea70cdb866b1180f5 : POST_FAILURE in 1h 05m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/73a1bc4545674594b4ce044f91dcdf75 : SUCCESS in 1h 40m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/032fd576f0bb4de4a9e16942a8c3f8dd : SUCCESS in 59m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f72ec083c2674bc09d9c5d0cc8c2165c : POST_FAILURE in 1h 04m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ae26fe4e73d84c6b90a4da431746a778 : SUCCESS in 1h 02m 09s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/251089e6b515422cb083207f514e2035 : SUCCESS in 1h 03m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ce152f4b755246a589403ae4b5eb1842 : SUCCESS in 1h 32m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bba2a19f1e9a42faba553059210a35ee : FAILURE in 1h 15m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c412111810844b49aadc6708a04b204e : SUCCESS in 1h 03m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c73318c1773c45d6a5a31407430c3ffe : SUCCESS in 1h 14m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b90c1e7987824062b9d34bbe078e797a : SUCCESS in 1h 44m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5177b26812c4e54b73ef54acc30b721 : SUCCESS in 56m 10s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0ee2da9f50164ae09166fccd34dc1c57 : SUCCESS in 1h 00m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d8a7b31f8c404711b8e2ee40fecccbf8 : FAILURE in 10m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/eeb6cf1751234d39966de0ead574339c : FAILURE in 16m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/02516a035be544f19a17e970dae1f3d2 : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/655e39fd7b694eff8577cb52f650c528 : POST_FAILURE in 1h 49m 52s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91f7269b610f4263ac5e2d42a98822cb : SUCCESS in 1h 45m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/510544fb95ed410fb74f67fb215225e8 : SUCCESS in 2h 01m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/24d0768ca9de4e25b2c2244b4853fde4 : FAILURE in 23m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d6345c66807747da9f677b4594b1888b : TIMED_OUT in 3h 04m 54s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c0c2511d2312fb15393f2770ed47ec2782ac3a8a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-23 07:56:52.000000000","message":"Patch Set 21:\n\nrecheck\nPOST_FAILURES","accounts_in_message":[],"_revision_number":21},{"id":"c75d8cfc9154479b1cb478c0a16805d612d02e09","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-23 08:24:26.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"19e0bca384fb59fa347a3dc72df5c2c1eade54c8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 10:00:03.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/21 : FAILURE in 2h 02m 43s","accounts_in_message":[],"_revision_number":21},{"id":"f62ffe5aea3e0944a1e689d95ec4ea9b7682fc35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 12:00:48.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/937093d59e6e46d788e5ef62f70a5b21 : SUCCESS in 37m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f189e399f034f9abbdfa6dee0d9ad00 : SUCCESS in 53m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eb0370e168784078a20205fcde85f684 : SUCCESS in 1h 19m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0d1016ad373247e1bf467169ddf4beaa : SUCCESS in 1h 05m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9298ccc552514621a38506a2f203e65c : SUCCESS in 59m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3b88f2b80ddd47e9b08095962405201b : FAILURE in 14m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86b2535e0ea34e71ad3c5d3cc1361071 : SUCCESS in 38m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e66e90d33cc743e5931d3d62b09708ad : SUCCESS in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee32114d61a4bbcbd9e4995d8f8c43c : SUCCESS in 11m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cfccc9094ab4bb1aa6c72dfe362d870 : SUCCESS in 24m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ef17a1f6950474ea1d4a62f0706565c : SUCCESS in 25m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f6a581af8d4c3d8fe414f19549bdf8 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd291f50ee0448a581f9e1bf80b9d8c6 : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bef5c39956c141a892f030b6e6239796 : SUCCESS in 49m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1d3ba8b658640c1a5df7276cf4c5e0e : SUCCESS in 1h 10m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d74245239f0d493c959ce20776ece5f0 : SUCCESS in 1h 42m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/075a1f5bb419468ca354e1acedfa90f9 : SUCCESS in 1h 01m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf08940237d4469e8cc31a1bcdfd52e7 : SUCCESS in 1h 11m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2bfe688c74f9428ead4394955bf7c121 : SUCCESS in 1h 02m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d900342dfa44f7189cbb48f665f2fc3 : SUCCESS in 1h 01m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/26d33a385dfd41d0bb4643030787fa5d : SUCCESS in 1h 18m 06s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3939a4a5911d4f028969f9e9b6cbf643 : FAILURE in 1h 20m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12f0834e1c4a456fb6b048522261fa56 : SUCCESS in 1h 02m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dea28cd60fba4a09ac578b99325dea25 : SUCCESS in 1h 06m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf664c61beba44f5af12be74d8852d0b : SUCCESS in 1h 47m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdcd379ee8a94e9a824244244cfca9c6 : SUCCESS in 47m 16s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ec0a513540164ca3bd7e14348f87dca0 : SUCCESS in 1h 08m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9688ae498ec42e582c6fc296df9d370 : FAILURE in 15m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c72dfaa756924176b98408c260718652 : FAILURE in 11m 16s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f27d9bdb7f244bc7912f39fcda173386 : SUCCESS in 1h 12m 38s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d69ec5993efe4fc59fcac3db077a411e : SUCCESS in 1h 29m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dcd1396813fe44229eebac5ad3b48fb3 : SUCCESS in 2h 03m 55s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/049e1fb66c514db78d6b9cf939ecb056 : SUCCESS in 2h 22m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0b84dd5da34a4ce18b0024ae1e7021a2 : FAILURE in 21m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/adf34c0591a940ebbe9f0625a478154f : TIMED_OUT in 3h 05m 23s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"286268db80815aa42a451f1167dea874f2ad01ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-23 12:28:02.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"71174f0cef6118f9b3b46cab152c557dfada4bbc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 14:39:44.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/21 : FAILURE in 2h 11m 12s","accounts_in_message":[],"_revision_number":21},{"id":"099adc99ede0bd43f33f56698e7e72f93cc317c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:59:53.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/53294d4a368544558eebc1700ad39867 : SUCCESS in 47m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96528719f84c4e2da1ea23d781030a0e : SUCCESS in 33m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fda8a3c56a114c608433b4275b8cde71 : SUCCESS in 1h 00m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/04acc32610954e9889532f5b49079fcc : SUCCESS in 1h 01m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/56f5658b9e804e12a08c65ab359788d2 : SUCCESS in 58m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5e4a2318576a4c1c9234852665d32904 : FAILURE in 11m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda7b473c5af452f941676d04e1c8b62 : SUCCESS in 32m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e13951bbc4af4d488e031cb0cbbf80b1 : SUCCESS in 29m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2600fb4ae1b4e10b642010609d4ad57 : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79e9b3507bf740be8306ebfea2817ae2 : SUCCESS in 26m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/055385b64e1d4d86a4dd470401df4f42 : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14deeaea728b413f8b1e9ff005cfbfac : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db0ebef70516466d83fa19b9f39f3d22 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf64550792534a1590a9ce4aabde93a6 : SUCCESS in 50m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f160069f13847629317b8943cf94837 : SUCCESS in 1h 09m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/651e39e003144871930beb13e88ef8c0 : SUCCESS in 1h 46m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a3f8f1ff07a4574b6034c7939f685fd : SUCCESS in 1h 08m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52bbbd00dc164106af0b5b58623b19a0 : SUCCESS in 1h 09m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/286ef2ff0f6a4f829ed6615e4b6ecf72 : SUCCESS in 1h 14m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c284f426a6b48a99e7abd715ed98a20 : SUCCESS in 1h 08m 40s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5c0fe6bb72504931a27bf5331b1c6308 : SUCCESS in 1h 16m 50s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/402eee9992974199aa6f8826800887e8 : FAILURE in 1h 17m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69fe271aca944d4abd98c7533980d6a9 : SUCCESS in 1h 02m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f3521f78dd8a4ac7b031bd0943eeed7e : SUCCESS in 1h 13m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6053dfc51c8a4ab896fe87afb7c3268f : SUCCESS in 1h 59m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58a319308ee54df28ccc9fa61e4622fb : SUCCESS in 50m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1fca6e59d20a46408551bb3456dac230 : SUCCESS in 1h 10m 13s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2bfeeb35ec5f4288865d9cf10dd8c8aa : FAILURE in 13m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/98eab041950f465cb7e4a15b0ef6d087 : FAILURE in 16m 41s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/be89381c06a3447e977f7985d34aeb30 : SUCCESS in 1h 16m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2d6f9dd0b43c4702979ece88130e723a : SUCCESS in 1h 22m 41s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9ae712b4d6e04e49bd8b4416d12ffdab : SUCCESS in 1h 39m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/77215752fe684b8e9bf4d6ba9f2a078e : SUCCESS in 2h 12m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/86839bc599ba453ba91eade6d4a0a0f5 : FAILURE in 25m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6cf5417b5dea40ec8860b19b1a4d97e9 : FAILURE in 2h 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"00b0075b0b746b499eb45bde4f989a02dd602a4a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 06:18:38.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"716ed9baed1de4803c0e1ad037009aa9d3ef3cb0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 06:18:43.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"fa89a303b2a7b4f9580546f2544b93654c331c87","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-24 07:22:00.000000000","message":"Patch Set 22:\n\n* sriov-test http://52.27.155.124/sriov/747774/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"a352fbee202b71728462e221ae12fadef33dcfa6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-24 07:44:32.000000000","message":"Patch Set 22:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/74/747774/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"da48d44f25d92a14a04a8bdd72bd9c5c48bbabb7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-24 08:23:42.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/22 : SUCCESS in 2h 04m 11s","accounts_in_message":[],"_revision_number":22},{"id":"7fa81cbd4b845ba91f76635cce3b871d7849525f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:08:16.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/20263795d1f34c1fac19762dfd2866db : SUCCESS in 52m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dadbdf4a0b24483da170a041ebf06e46 : SUCCESS in 31m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6869bb708849436699a42e3f070af4d8 : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ca8bedb42364e9fbae8b46cc6f45104 : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c4a64d49aed4de29a9e34edd0d9c4b5 : SUCCESS in 1h 11m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8a90314561814bf08ae65b9b2f239ccb : FAILURE in 22m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/678f138d77e6485292422ca4faa4e458 : SUCCESS in 33m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db103f255c01431fbbf50d18e6ef1853 : SUCCESS in 26m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ac732d3a4a4f719d27dab82a568408 : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dee3b2a10d84d148ff53887d4b7b5eb : SUCCESS in 25m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea1867ae9e2a4c069c1ed527d500ae59 : SUCCESS in 25m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19dc19be0d93475db17605b0402a12a3 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5339116966e74d588c21aa706171bef7 : SUCCESS in 8m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64f92b1a6bcb43d9a6ee0f8f1f240610 : SUCCESS in 1h 11m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1718c3924fa4d33a13d4ea31540563a : DISK_FULL in 1h 21m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2b304890242447f7b07821fa90832e10 : SUCCESS in 1h 41m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8444a94dfbca4fe5ac661632b0638a9c : SUCCESS in 1h 11m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/073de83f9088488c87b5f06fcef0f775 : SUCCESS in 1h 13m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d43db39e80be46fca0234a2fe4902963 : SUCCESS in 1h 10m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff974b598a6f45919f2bee6a04a95f3a : SUCCESS in 1h 09m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b3db517dcfd4193982f3a65b09b112f : SUCCESS in 1h 21m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de7fa117da4c4ddf9a3ecda59151cfea : FAILURE in 1h 26m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dca7c812e5524e9ca8f6ad7736f5096e : SUCCESS in 1h 05m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a6d335bac3104572af4dfd9cf2a5dea8 : SUCCESS in 1h 13m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c784b250272046bf9112416210c94565 : SUCCESS in 2h 01m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec51f0b0e5cf4b6a97b0b0903b21ad28 : SUCCESS in 55m 11s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c496dd753db9441995ba0d79e8e33ac3 : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7f5919ac8489474d8a51a8fea6f0843f : FAILURE in 15m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ceef527d8cd846ad957a4f8afef08545 : FAILURE in 14m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0710c830ea414eba8a6210b60ff78ed0 : SUCCESS in 1h 26m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d534de425272469aab474a9e671556d1 : FAILURE in 1h 50m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f33e1877f114bfb8ccd5d55b5efd971 : SUCCESS in 1h 41m 14s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/08d3d9bbd16b4edaae7ab52dc76cf2cb : FAILURE in 28m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3f4e44f4ea9443fe823e2f3b27504f31 : TIMED_OUT in 3h 03m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/700222b115ba434aa41278488fc74883 : FAILURE in 2h 53m 12s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"adebf033ff69434e4bf1c385a88aeea7fef6766c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-25 05:17:14.000000000","message":"Patch Set 22:\n\nrecheck\nhttps://review.opendev.org/751882 was merged to solve ovs build issue","accounts_in_message":[],"_revision_number":22},{"id":"f4c963ca771816af91838230e2c1273b1edd917c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-25 07:04:57.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/22 : SUCCESS in 1h 46m 45s","accounts_in_message":[],"_revision_number":22},{"id":"aa73c5de552bf47a8b2706f2bd2b97cc10def38e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 08:28:56.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e254de6f611240849f8ecf18e49d9209 : SUCCESS in 57m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7e1cc602f5f44d92ab95414867b06788 : SUCCESS in 47m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4821f4909764401c9b9a778d911b74e1 : FAILURE in 1h 37m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1ec676e3189d4ea58d8ba24bad4ef745 : SUCCESS in 1h 14m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/336ca14e061a4d2d8281cde2ba13fdb7 : SUCCESS in 1h 24m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e2f5ea78220a4a468fb60f464d218f77 : FAILURE in 24m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b832869c87a54a67912529a9558fb18f : SUCCESS in 35m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/388edcabfde44f7db3a1660a028add79 : SUCCESS in 24m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c739ada35b34b93b2e430246d3bc479 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03cf5845ca164bfe9fe9f4288cae63ba : SUCCESS in 22m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/031e0ec0f089405e8a26ac43a7efac01 : SUCCESS in 25m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6c7ff26d29479e91523c3b0f8643c3 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d08566cc560845128a1543e60cb82c61 : SUCCESS in 15m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e980cb9757a46d1ab361a574217d3b6 : POST_FAILURE in 1h 36m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d451c762ccd84a3a83262ad6e70b03a3 : SUCCESS in 1h 09m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/82cd7cb489f74f48a26d65dbe9fbebf7 : SUCCESS in 2h 03m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bd3c68504b84807b7696f05bc5cd9f3 : SUCCESS in 1h 00m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8d83e54c8b4b4d5ab5432bd354169bcf : SUCCESS in 1h 30m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/eac749f407a84ed790ea8cd94ea7d6e1 : SUCCESS in 1h 05m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6963882e66ba4a0bbc9be47558ba4f64 : SUCCESS in 1h 40m 43s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab97fbbb6df343b1bf98ea517b083904 : SUCCESS in 1h 26m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f43a484f82fe416896b5c8972664ad1f : SUCCESS in 2h 00m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b73ca5d7e4a24a67bcea46619a784175 : SUCCESS in 1h 42m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f3edc4d84c44c13a2c347cb22fb2716 : SUCCESS in 1h 13m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd03b4cfddde4cfa9770f0f3aa0dc995 : SUCCESS in 2h 09m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1072917b7ce4eeba866ddcf3f5f5fa0 : SUCCESS in 1h 15m 43s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6484003cfa864a07bff454484fb0a938 : SUCCESS in 1h 11m 28s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/962c873b60734845978de646f4151eb0 : FAILURE in 20m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/44c1b5252e2943499d62029a5db360dc : FAILURE in 18m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4e51b2fdf0124b3c8e684bb40a729bbc : SUCCESS in 1h 17m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/252fa549ca1a4f1986bcd40a29fb7981 : POST_FAILURE in 2h 23m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a4e30fe338584c3b83b9295de48b072b : SUCCESS in 1h 50m 49s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c6dac475498a428c8d5798c130d049fb : SUCCESS in 2h 35m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ed3e6e94f7a046ec8830b05d487991f4 : FAILURE in 20m 20s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6be0950f41d2438eb2afdcbe4fae77a2 : TIMED_OUT in 3h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"8346a4cd831d9e333f9b8e5bd1aef04d3e7b9281","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-27 05:04:00.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"61dd461cff4ad9e1c223e918831d2b042b152b1e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-27 06:50:00.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/22 : SUCCESS in 1h 45m 31s","accounts_in_message":[],"_revision_number":22},{"id":"636b1b09af158cfb75075b943977b43fe3d750d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 08:14:56.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3e70e5670b5f479389919a042730f778 : SUCCESS in 41m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c543f1f971eb4ac9ab5fec2d7d2e816c : SUCCESS in 25m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ba927c1cad9740b19459ca6c3dce88de : FAILURE in 57m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/60a3cedcc1b242a5b1e2ace689facd66 : FAILURE in 2h 26m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/634b64c0859849da9d28d4061be581a2 : FAILURE in 47m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d471c01316ea483093c87f067c418686 : FAILURE in 26m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cd5db2fbdb64ac0b864c9eee8dda6b9 : FAILURE in 1h 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2ba141bd769429cb7f0611591ac4150 : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a659366d8ab4c75bd3abc4dade22649 : SUCCESS in 14m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/852e26fc94fc4598a22f03c5c05cb98e : TIMED_OUT in 1h 01m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de8a9426356f499689206621e33f1dc3 : TIMED_OUT in 1h 01m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd146d7230434e578afe26dba5017fce : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cad979eb8b84234b13f6d9496366d22 : SUCCESS in 6m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d1d62e39f5d4d9581932cf1f35a011c : TIMED_OUT in 2h 17m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/24c2079fd10644419bbad09e06b9a2dc : DISK_FULL in 1h 05m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b13ea92125c14f0d87151450df0d406f : FAILURE in 2h 54m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80aa4b21f5bf40629bc6d2c0702cbc00 : FAILURE in 1h 39m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c57fb444f85245febf222507ab81f824 : SUCCESS in 1h 12m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/63455e919ad048e4a3613f840a8d6451 : FAILURE in 2h 19m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6223c178caae45c99ac0d06d62f34b72 : FAILURE in 2h 22m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cfed5723a4c44be9b85559f7fcfe8728 : SUCCESS in 2h 45m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1a5ba1336b54da493ee19a623d7b8d5 : SUCCESS in 1h 30m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fec15fc82a44ec9bf65cad46d4ea136 : FAILURE in 2h 20m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cb4b971ea8a4415fb5d6cabff4d530ac : SUCCESS in 59m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c70da84e41c54c5d9f47b4571e5be2d3 : FAILURE in 2h 02m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/563989d7df6f4aa0bf5663dbbec1771f : SUCCESS in 46m 45s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/33e07350315941739997e4115f10424c : SUCCESS in 2h 32m 52s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e74d8c000d004287a2dc91a44978da9d : FAILURE in 19m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ef4f062d4fb5493581c72816a1ffba29 : FAILURE in 18m 58s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/66e3f9232760434ebc7b3758634a5ec2 : SUCCESS in 1h 22m 38s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d3f2d2018b3c4967a63b7d9de707604c : SUCCESS in 2h 34m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/496c8d141b154f24844b58975bffa24a : SUCCESS in 2h 07m 11s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8c8a9c86affe481f8ce9e6603ee90090 : SUCCESS in 2h 19m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f22a0cecac8041cf8473b557ff896c78 : FAILURE in 32m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/991a7ac25a4c49a5a600144059c45f60 : TIMED_OUT in 3h 04m 33s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b14a0856995aa43ca55ea92b852aaa9e0873f263","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-28 04:35:14.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"d3a1ffd1a372331070ddd1697b1063a75fc62d69","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-28 06:37:24.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/22 : SUCCESS in 2h 01m 42s","accounts_in_message":[],"_revision_number":22},{"id":"ab05dffa4e2d3eed51098a9fa48f89e39339373f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 07:46:20.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/daa9c96191854e549382d9abf3875d2f : SUCCESS in 59m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7bc4d4d56ad4491890011670cec098d : SUCCESS in 40m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c70e130f0d894a99b49bfcbad38681ed : SUCCESS in 1h 26m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/77dbf5ddff8547c9811ad5ba7f1e729e : SUCCESS in 1h 27m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21fdc761ca824c9384aa2629f81ee4ba : SUCCESS in 1h 01m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/351a87c4e3cd444c979fab84beec6c63 : SUCCESS in 1h 09m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57e1ca7b3b774f89b81017f698767837 : SUCCESS in 40m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68acf8e903f24c42af8b6fbc50f684f1 : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3892f446dc249149d9d6b13766a4a4d : SUCCESS in 10m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f98cebd2365d4e8899e79cf2569e3787 : SUCCESS in 26m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4a2e84fba249bdb1d570adc2262cf8 : SUCCESS in 28m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07759ec3ed62478db275e6176542f958 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe1947d2237044e3b91368eb950dd892 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e0d24996e1f4163b8e043c45bb65824 : SUCCESS in 1h 08m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfec8212c59d43688fc8801a3e0e0394 : SUCCESS in 1h 07m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/88cb535367c4459b9075d267d42e6658 : SUCCESS in 1h 49m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f8d1b8297534b988cd52e79e5c13cb2 : SUCCESS in 1h 07m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/968e3e1433a94342bc8c025c9b7389b3 : SUCCESS in 1h 06m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/15fec7043b8e40b886ba10fe7ed0efad : SUCCESS in 1h 17m 41s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/03af9272fa844bf09ce226208cb6a91c : SUCCESS in 1h 11m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ead59df6742546c784ce1d42b91bc38b : SUCCESS in 1h 35m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1041c0c045834355b9662ff503fa3090 : SUCCESS in 1h 29m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/130a20ca0c3046a5bcee5f9d383b29f4 : SUCCESS in 1h 27m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c6ae4750a4dc4552b257958382c90a4b : SUCCESS in 1h 15m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb3fcaa211a349609bf82ed51b5f9f6a : SUCCESS in 2h 04m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddacc91b8f3f4690a075c26edc175f8f : SUCCESS in 1h 03m 55s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0bd37de81f4c492182d98d9ac7ebe495 : SUCCESS in 1h 27m 45s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d8ba62f23352428781aff9858aa803c1 : FAILURE in 20m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5a793e5dd2e241c3b3a9bc95db93a74d : FAILURE in 17m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/71a03c771a5c4571b4b94e384f47d2c7 : SUCCESS in 1h 28m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71f196114a8d413fa3d2fd2036188879 : SUCCESS in 1h 35m 48s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab22c390b401460fac51f035dce6252c : SUCCESS in 1h 53m 56s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e7bb37e15b1d402596b30c22f633b639 : SUCCESS in 2h 40m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b418ba5efb7d487b9279ccb32055eb08 : FAILURE in 25m 31s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/be7ebdfdbc924d13b3e8fbe553422b5b : TIMED_OUT in 3h 04m 12s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"515cf68b28caa3323d8d5284c8f696e20dba24d4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-28 08:01:10.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"e2b6c1830649315e651f98bc3690536399060338","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-28 09:47:06.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/747774/22 : FAILURE in 1h 45m 28s","accounts_in_message":[],"_revision_number":22},{"id":"2b62c192b825dd33e4230781110e2742d2732910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 11:09:01.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5165f94d6fc4b34baebb4829fa9e008 : SUCCESS in 52m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26335e48754c4e68aca59df26d7a7564 : FAILURE in 32m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/21c4e45a22274d9a85f25e5b6d28fbd6 : SUCCESS in 1h 01m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/811b20b8783445f3bbccb2e2ced7f5dd : FAILURE in 1h 32m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1b37eb14bf84a339ae90a6df44d2eae : SUCCESS in 1h 02m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/331efc14dc584bebafb1bc6554765fcd : SUCCESS in 1h 00m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8334d6bf5924445988765f26abcdea2d : SUCCESS in 37m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6bfee2feac74852afb16c34b4a2eda0 : SUCCESS in 26m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b6641668174282b3addd1377a11d55 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/258068cbcbe9443b8b1065a98bf0ed6d : SUCCESS in 31m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/723e6da3affe42d7ad8bf3507cf8f7c1 : SUCCESS in 27m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4099f4b53dd34a9a8db40cec40bc691e : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5577db8e55284452bb46c7e3ffdb3589 : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7238bd9fee4343ef8aeea4890f81a5b6 : SUCCESS in 1h 14m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ff2c4a68a9d496cafe17f76897beef3 : SUCCESS in 1h 10m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6303d29d9527416dbf9e5e3a72f78eb5 : SUCCESS in 1h 42m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3eae00befe4645cd82a99001452fd3a4 : SUCCESS in 1h 10m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2a75ca19d8c645c4a53608e21a7eefe3 : SUCCESS in 1h 18m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a9ee06fe104747c7babc71a72bd6f8d9 : SUCCESS in 1h 05m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c34d0d5408f04512b7b292867e582a82 : SUCCESS in 1h 08m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f428d525655a4feb8150b21a633db330 : SUCCESS in 1h 29m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12357d35748149b2af5bdee2c522a327 : SUCCESS in 1h 23m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eed1bb1a176548dfadb4cee14bddb7ef : SUCCESS in 1h 11m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/59a6c387c4da4c49b07a0458f16108c7 : SUCCESS in 1h 14m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a870f91e7e2d4b68b2d13b88d5c4123d : SUCCESS in 2h 16m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/751b7ac33bb6485dbbb05d94be48b71d : SUCCESS in 55m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c0588f5767f5409ea65372d675c13ef1 : SUCCESS in 59m 47s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6df08da4070c4aba981d7a5373dd0c94 : FAILURE in 19m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4d122ae43eb6443ea6494c0d82d8a643 : FAILURE in 26m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4522124db5984dd49d4580b4f0f59b92 : SUCCESS in 1h 22m 11s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93ba91749c02412fae8da7607e05e675 : SUCCESS in 1h 23m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dab412b9a5ae40b3a2d074ca82285eea : SUCCESS in 1h 33m 56s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1a7061b98963412da356d01c01c07d25 : SUCCESS in 2h 19m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5a4a249feafd41cf8b757698caf94b4e : FAILURE in 27m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e5af95c0f828494198af3afee2320e62 : TIMED_OUT in 3h 05m 07s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"aff10942e61eb9f3e987d034e395f88224ba7b81","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-29 16:59:56.000000000","message":"Patch Set 22:\n\nrecheck https://review.opendev.org/#/c/738163/ merged","accounts_in_message":[],"_revision_number":22},{"id":"abf9ad61400dce14b0bc3f7af05eb67b5b10ffed","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-29 18:48:10.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/747774/22 : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":22},{"id":"505bd5809413755d5fb94df25521bb1d50fb8e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 20:08:05.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d4dbfd9254a744908a4b2507603de50b : SUCCESS in 56m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0f40f09e36da49a7af7135596100afa1 : SUCCESS in 37m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9acf241df3e14fd5a2cdf904d526ad1a : SUCCESS in 1h 58m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ada62dc747ee4296bcbd1044c625f8f6 : SUCCESS in 1h 19m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15de7250dd2142d99147df4f5e9ab669 : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1fc99d6f2848442f97b77620d2d0a2e3 : SUCCESS in 1h 15m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2a5b22d4eca40a48aacb024c96be507 : SUCCESS in 33m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b980bc755eae459eb70526baabaa0b4a : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb959a36ce243e086930db23370f5a9 : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c580b923f46b414eaf4ebd210b052b50 : SUCCESS in 25m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c451ad7805543c890cfa1062e8bd615 : SUCCESS in 29m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74714d2742554ba093ac1fff9855fbd5 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb146dc543d48538ea718cebe9b54da : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea5ef6636d074ef7b38bcda804c6bb56 : SUCCESS in 1h 10m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/295db19fcef14cca90c758dc111bb914 : SUCCESS in 1h 10m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dba3e9c23a7649a0974ffe8a75308951 : SUCCESS in 1h 48m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2598edd6c0544139df3ea9e25486e16 : SUCCESS in 1h 01m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e28b0073321b4aa686bdd6a254e43460 : SUCCESS in 1h 19m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/93c12ddb5db24825b2e197efc63b17a3 : SUCCESS in 1h 08m 18s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/91372644bd244240ab5466af1ad68182 : SUCCESS in 1h 08m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ca84e9b2258456fb75235533cdd3790 : SUCCESS in 1h 27m 48s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6fc1863e01914082a101cd1289a432e6 : FAILURE in 1h 22m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4c7a5d3e48d14b41b053a44edf66e3b5 : SUCCESS in 1h 11m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d5e3ebc507e24cc68ee91e4f3518559e : SUCCESS in 1h 17m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/921c26d2e3394f1590b75d69df9d2699 : SUCCESS in 2h 29m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd72c8f7d4f40909d7817e51b0f1686 : SUCCESS in 54m 16s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c9e94e14ac0b4ad788902fbf7591a96a : SUCCESS in 1h 07m 32s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7f0f487929aa4c9bafe9b6fe7831ee2b : SUCCESS in 1h 11m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3cab9c08a1c44d0b86a916901d83cbb0 : FAILURE in 42m 38s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0c5f607e894f4738b74d626df301192c : SUCCESS in 1h 18m 08s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2e036b92922b4b4e9590ae4fb59e7090 : SUCCESS in 1h 44m 55s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e24b567434b74f1d801062a5fd79876d : SUCCESS in 1h 40m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bfa55cd3943e487b8b5fa391ef7f8709 : SUCCESS in 2h 24m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ce4601f5f19d4e17aecc588050b7907e : FAILURE in 42m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/aec11bc9d39f4f87937de60cb9f17f88 : FAILURE in 39m 16s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5c8a4f981e309fe9654760902d6a92274f46a02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 20:08:34.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"81f542c9e35a83dfa96a084ef4295b9d922a5ac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 23:45:10.000000000","message":"Patch Set 22:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"076d7ed9e365ffd5c74453a369cfd48548353696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 02:21:38.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a9cb0e860268489695e54cfed8027d67 : SUCCESS in 56m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/63ee0daf2df64931b7d5740c3e4534bd : SUCCESS in 1h 29m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79e6b750b7644e64b484a93c80ba563f : SUCCESS in 1h 04m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/084e9a1730c84d5b872d6fba16eddf43 : SUCCESS in 1h 13m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6efd5cf19ac4b128f68c5ca4cd7bd0e : SUCCESS in 1h 22m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7a464776a464cbabd7a95b2b245c014 : SUCCESS in 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ed2689780d40b1896ea64d0c647713 : SUCCESS in 10m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52111c74d04a43d7a79414f250d7d68f : SUCCESS in 24m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77787361153346629696e6bdc47216fb : SUCCESS in 22m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66da4b5b65714a6ebd3dd390d9bbcfda : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd6cfb049734c2ea0f00a0bac8be42a : SUCCESS in 8m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a8697f0d0734280a6b560aa573f814a : SUCCESS in 1h 15m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af2c42cd5c8d4fee9dc9f25f5511db9a : SUCCESS in 1h 55m 14s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/02aa7648bd5d49268a5abc2548ffd62f : SUCCESS in 1h 06m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ea7679524484df580505ce117b288cf : SUCCESS in 1h 16m 23s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe7bad16e61f43bf8742ac29ab0ac1f3 : SUCCESS in 1h 06m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f9da8df1e32c41c493b48c830b30e9d9 : SUCCESS in 1h 15m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e16458b56e564d1e9b26d4345650ad6c : SUCCESS in 1h 07m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e24b17a8232e46f88a82dfcecbd3132b : SUCCESS in 1h 36m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4fe6f8945c340a69d56d698775da6ac : SUCCESS in 1h 57m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cc1d94e490f4bb4924414e6e3b9b391 : SUCCESS in 56m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bd3afbf56f834c66a262e336874ae830 : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":22},{"id":"e3db34ab4fdf784c74b7b2f8eb04ae312417575b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 02:21:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"22dd84649a648c7d1a9ebc13be3dcb1ec52d3e97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 02:23:05.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2f00e9bd594d7780e8ed365f2f43e7 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5cd0943ffdf42a494b046c6197a5684 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":22},{"id":"f1a536393b452747a0c697e34a1c512d8bdc72b8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-30 04:33:09.000000000","message":"Patch Set 22: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit d44f33e1861fe222f9c4bea135edd031c45ce3f2","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"87e5131432f181459bde78b31b59204c8e5077e2","revisions":{"f3bfffc0cbfd551baa3b6b0d48eb5523bbe9a1ee":{"kind":"REWORK","_number":1,"created":"2020-08-24 17:06:53.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/1"}}},"commit":{"parents":[{"commit":"856cae4cf8e33c05b308d880df78b7be02ae90ad","subject":"Merge \"Fix get_ipv6_llas method in the interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/856cae4cf8e33c05b308d880df78b7be02ae90ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-24 17:05:44.000000000","tz":120},"subject":"WIP: Allow replacing the QoS policy of bound port","message":"WIP: Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3bfffc0cbfd551baa3b6b0d48eb5523bbe9a1ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3bfffc0cbfd551baa3b6b0d48eb5523bbe9a1ee"}]},"branch":"refs/heads/master"},"2c90966da09028f8eff8360a58e53947b4b66e60":{"kind":"REWORK","_number":2,"created":"2020-09-03 12:59:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/2"}}},"commit":{"parents":[{"commit":"856cae4cf8e33c05b308d880df78b7be02ae90ad","subject":"Merge \"Fix get_ipv6_llas method in the interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/856cae4cf8e33c05b308d880df78b7be02ae90ad"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-03 12:58:17.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c90966da09028f8eff8360a58e53947b4b66e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c90966da09028f8eff8360a58e53947b4b66e60"}]},"branch":"refs/heads/master"},"e61ca721cb018e9851c1cc615f72cb6ce98f7d55":{"kind":"REWORK","_number":3,"created":"2020-09-04 15:28:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/3"}}},"commit":{"parents":[{"commit":"be8f8d23330752d88de3eebf4fac12a03747c86f","subject":"Merge \"metadata-ipv6: Accept link local address in X-Forwarded-For\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be8f8d23330752d88de3eebf4fac12a03747c86f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-04 15:27:52.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e61ca721cb018e9851c1cc615f72cb6ce98f7d55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e61ca721cb018e9851c1cc615f72cb6ce98f7d55"}]},"branch":"refs/heads/master"},"1034dfc86b58b8493551f03c528ec677d2bc5498":{"kind":"REWORK","_number":4,"created":"2020-09-07 13:05:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/4"}}},"commit":{"parents":[{"commit":"be8f8d23330752d88de3eebf4fac12a03747c86f","subject":"Merge \"metadata-ipv6: Accept link local address in X-Forwarded-For\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be8f8d23330752d88de3eebf4fac12a03747c86f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-07 13:04:16.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1034dfc86b58b8493551f03c528ec677d2bc5498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1034dfc86b58b8493551f03c528ec677d2bc5498"}]},"branch":"refs/heads/master"},"dfea84911e1d7cc1e0adbb610d0fd95db166f04d":{"kind":"REWORK","_number":5,"created":"2020-09-08 10:38:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/5"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-08 10:37:51.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfea84911e1d7cc1e0adbb610d0fd95db166f04d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfea84911e1d7cc1e0adbb610d0fd95db166f04d"}]},"branch":"refs/heads/master"},"b216d62ee0c002b0b4c7e1fa121886c2d49e6287":{"kind":"REWORK","_number":6,"created":"2020-09-09 07:52:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/6"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 07:51:48.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b216d62ee0c002b0b4c7e1fa121886c2d49e6287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b216d62ee0c002b0b4c7e1fa121886c2d49e6287"}]},"branch":"refs/heads/master"},"f2ca48b79a0ef11dd8060f67b4feb686e12a2a93":{"kind":"REWORK","_number":7,"created":"2020-09-09 08:19:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/7"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-09-09 08:19:05.000000000","tz":0},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2ca48b79a0ef11dd8060f67b4feb686e12a2a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2ca48b79a0ef11dd8060f67b4feb686e12a2a93"}]},"branch":"refs/heads/master"},"af8396225127726597359ddc3a9de33a7af4fa93":{"kind":"REWORK","_number":8,"created":"2020-09-09 09:06:35.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/8"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 09:05:20.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8396225127726597359ddc3a9de33a7af4fa93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af8396225127726597359ddc3a9de33a7af4fa93"}]},"branch":"refs/heads/master"},"24aec0eeb4982804166fbf6c2009f330b45ceb29":{"kind":"REWORK","_number":9,"created":"2020-09-09 11:48:53.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/9"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 11:47:52.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24aec0eeb4982804166fbf6c2009f330b45ceb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24aec0eeb4982804166fbf6c2009f330b45ceb29"}]},"branch":"refs/heads/master"},"40126ebf4d8f3bebd3773f1e03572aae86a6e4ee":{"kind":"REWORK","_number":10,"created":"2020-09-09 14:23:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/10"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 14:22:45.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40126ebf4d8f3bebd3773f1e03572aae86a6e4ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40126ebf4d8f3bebd3773f1e03572aae86a6e4ee"}]},"branch":"refs/heads/master"},"d466cba590cad6151302df8a7e04b6cfd6c1350a":{"kind":"REWORK","_number":11,"created":"2020-09-09 15:08:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/11"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 15:07:31.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d466cba590cad6151302df8a7e04b6cfd6c1350a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d466cba590cad6151302df8a7e04b6cfd6c1350a"}]},"branch":"refs/heads/master"},"30d7f52b657308eee91596ae55d66dfb97a967a7":{"kind":"REWORK","_number":12,"created":"2020-09-09 20:31:00.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/12"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 20:24:54.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30d7f52b657308eee91596ae55d66dfb97a967a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30d7f52b657308eee91596ae55d66dfb97a967a7"}]},"branch":"refs/heads/master"},"e9e7986e6a1445f5e8663dc154af2b923814d1bf":{"kind":"REWORK","_number":13,"created":"2020-09-09 20:37:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/13"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-09 20:36:18.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9e7986e6a1445f5e8663dc154af2b923814d1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9e7986e6a1445f5e8663dc154af2b923814d1bf"}]},"branch":"refs/heads/master"},"2bef6d9b1610353c51a9650e90c2af5fc0146ce8":{"kind":"REWORK","_number":14,"created":"2020-09-10 08:23:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/14"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-10 08:22:20.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bef6d9b1610353c51a9650e90c2af5fc0146ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bef6d9b1610353c51a9650e90c2af5fc0146ce8"}]},"branch":"refs/heads/master"},"8092ee1b38d9d1404f5cc917cf0611c36c29eee9":{"kind":"REWORK","_number":15,"created":"2020-09-10 10:07:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/15"}}},"commit":{"parents":[{"commit":"d572463e7506a67fa209c3a598b242664c113a4b","subject":"Merge \"Use metadata constants from n-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d572463e7506a67fa209c3a598b242664c113a4b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-10 10:06:34.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8092ee1b38d9d1404f5cc917cf0611c36c29eee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8092ee1b38d9d1404f5cc917cf0611c36c29eee9"}]},"branch":"refs/heads/master"},"0ead589ad97d4f9cb5d7eb51110f5836eae2f652":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-09-11 04:33:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/16"}}},"commit":{"parents":[{"commit":"53625d0bcc11b4b716b021543cf4e264c2839082","subject":"Bump neutron-lib version to 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53625d0bcc11b4b716b021543cf4e264c2839082"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-09-11 04:33:52.000000000","tz":0},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ead589ad97d4f9cb5d7eb51110f5836eae2f652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ead589ad97d4f9cb5d7eb51110f5836eae2f652"}]},"branch":"refs/heads/master"},"7823dbaca1730c6d58e6705cd26a6fa1d3af8d11":{"kind":"REWORK","_number":17,"created":"2020-09-11 12:04:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/17"}}},"commit":{"parents":[{"commit":"53625d0bcc11b4b716b021543cf4e264c2839082","subject":"Bump neutron-lib version to 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53625d0bcc11b4b716b021543cf4e264c2839082"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-11 11:56:35.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7823dbaca1730c6d58e6705cd26a6fa1d3af8d11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7823dbaca1730c6d58e6705cd26a6fa1d3af8d11"}]},"branch":"refs/heads/master"},"862ac67c875e1cb6fd3f812364adc92e38540090":{"kind":"REWORK","_number":18,"created":"2020-09-11 12:39:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/18"}}},"commit":{"parents":[{"commit":"53625d0bcc11b4b716b021543cf4e264c2839082","subject":"Bump neutron-lib version to 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53625d0bcc11b4b716b021543cf4e264c2839082"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-11 12:32:46.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/862ac67c875e1cb6fd3f812364adc92e38540090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/862ac67c875e1cb6fd3f812364adc92e38540090"}]},"branch":"refs/heads/master"},"8ea44cededd93d90a1c37f146f7c0967df8ba206":{"kind":"REWORK","_number":19,"created":"2020-09-16 09:26:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/19"}}},"commit":{"parents":[{"commit":"53625d0bcc11b4b716b021543cf4e264c2839082","subject":"Bump neutron-lib version to 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53625d0bcc11b4b716b021543cf4e264c2839082"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-16 09:23:40.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ea44cededd93d90a1c37f146f7c0967df8ba206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ea44cededd93d90a1c37f146f7c0967df8ba206"}]},"branch":"refs/heads/master"},"2ef7445504e6eb5510e021c04ffab7d4be58da72":{"kind":"REWORK","_number":20,"created":"2020-09-17 11:39:00.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/20"}}},"commit":{"parents":[{"commit":"53625d0bcc11b4b716b021543cf4e264c2839082","subject":"Bump neutron-lib version to 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53625d0bcc11b4b716b021543cf4e264c2839082"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-17 11:38:14.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ef7445504e6eb5510e021c04ffab7d4be58da72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ef7445504e6eb5510e021c04ffab7d4be58da72"}]},"branch":"refs/heads/master"},"add9ee01c9c896534d3aeaf178e0fa23f496b813":{"kind":"REWORK","_number":21,"created":"2020-09-21 14:36:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/747774/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/21"}}},"commit":{"parents":[{"commit":"53625d0bcc11b4b716b021543cf4e264c2839082","subject":"Bump neutron-lib version to 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53625d0bcc11b4b716b021543cf4e264c2839082"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-21 14:35:23.000000000","tz":120},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/add9ee01c9c896534d3aeaf178e0fa23f496b813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/add9ee01c9c896534d3aeaf178e0fa23f496b813"}]},"branch":"refs/heads/master"},"87e5131432f181459bde78b31b59204c8e5077e2":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2020-09-24 06:18:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/74/747774/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/747774/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/747774/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/747774/22"}}},"commit":{"parents":[{"commit":"69bedc9beb1262afa195de24a1b6ed9a1e1f4207","subject":"Merge \"ovn: Use new OVS commit hash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69bedc9beb1262afa195de24a1b6ed9a1e1f4207"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-07 14:00:21.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 06:18:38.000000000","tz":0},"subject":"Allow replacing the QoS policy of bound port","message":"Allow replacing the QoS policy of bound port\n\nChange-Id: Iebdfd2b303f47ff9f049cf583ea754b35ca26f78\nRelated-Bug: #1882804\nDepends-On: https://review.opendev.org/748279\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87e5131432f181459bde78b31b59204c8e5077e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87e5131432f181459bde78b31b59204c8e5077e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
