)]}'
{"id":"openstack%2Fneutron~749945","triplet_id":"openstack%2Fneutron~stable%2Fstein~I645610febde446b78ed6edd868e699673648a4de","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1868724-stable/train-stable/stein","hashtags":[],"change_id":"I645610febde446b78ed6edd868e699673648a4de","subject":"Use dict .get() to avoid a KeyError in the segment plugin","status":"MERGED","created":"2020-09-04 13:09:13.000000000","updated":"2020-09-06 22:53:50.000000000","submitted":"2020-09-06 22:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749945-1599432746360-941d6ed0","meta_rev_id":"ca8489f006c1f07886bd0921731a81d6cc99032c","_number":749945,"virtual_id_number":749945,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-09-06 19:01:22.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-06 22:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-06 18:57:40.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-09-05 08:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-04 15:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-09-05 08:37:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 15:05:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-05 08:37:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-06 18:57:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-06 19:01:22.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-06 22:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96b8f979e9857c593c35b3392013a74435dfdb03","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 13:09:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"181b3b6fedfb6f3c7fb640e2783d6af610458b63","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 13:09:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit ce8d891c79ed18aa0f7cd353e4121b22a6395a76","accounts_in_message":[],"_revision_number":1},{"id":"e4108b6e48ab41870c48485e3d05fa6bb82f9b68","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 13:17:53.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749945/1 : FAILURE in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e618fd115aeaec56f06539bd29ee917f0f9da377","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 13:25:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/45/749945/1/check/devstack-networking-ovs-dpdk/8ff1fd3/ : RETRY_LIMIT in 1m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6cea04bbfaa10acb717e2a9605bf6c0f200836cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 15:05:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1b2e66de3d1e48bc8b0684162e93180e2e43d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:45:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/d59b6c847f784c4c97925499e21239fc : SUCCESS in 54m 08s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/d02414a70eb34588809ddef3a1b91ec0 : SUCCESS in 42m 32s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/1f95f965db0a49a0834e2180dcc273ed : SUCCESS in 1h 12m 06s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/1120f0f9c443424aa8890d6c81df8c19 : SUCCESS in 1h 15m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ba69d71c2a7d4567ac9c1c042ca46d99 : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0208d7b3325743f09f4be80c52c08bdb : SUCCESS in 29m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f72d6d84e1664d47adebd5e3c7016915 : SUCCESS in 25m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281be621fe4747ab950e21e22e324616 : SUCCESS in 11m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec6753defbd4411782b8defcf4f7df4b : SUCCESS in 22m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b051ca032c634a178c9af0f60611dcbe : SUCCESS in 21m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3f9035be40f417380c532ff6de7b2ae : SUCCESS in 20m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bc53ffeeb37434a95178fb35c26bd42 : SUCCESS in 23m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/250936f4177c42fc909046839aac76c9 : SUCCESS in 5m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b9b3189727a94fbc83b86c4d121c3e68 : FAILURE in 3h 00m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1e112c7375d9455a87942c37ad473f60 : SUCCESS in 1h 03m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74ab5ed0201c405795e2f0fb44cf9f4e : SUCCESS in 1h 26m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a8ee0aa418a74695a1fb53a121105ca8 : SUCCESS in 41m 54s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/dd443adbc6ab46cbb334f64e4c8c19a2 : SUCCESS in 45m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ba0a4f28616a4d72a0bff8344d3a1005 : SUCCESS in 1h 05m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/53cd23d4e81643d38e69ee2eb3aab76c : SUCCESS in 1h 35m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dc63418755b845dcbe6949e0b6e82fb1 : SUCCESS in 1h 19m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fc1a1a905f2a437d8f25086dac93d3db : SUCCESS in 1h 18m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/908c7a01499842149c82a94df4b0aeec : SUCCESS in 1h 16m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1aebb9f7f9b54c58b10e94b7ad9fd9ea : SUCCESS in 1h 21m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80dff61072a8478793a954db25c050cf : SUCCESS in 1h 41m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1979f00f27f34abb85249f6e258d1b9a : FAILURE in 1h 32m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e9f0ab0fb4e4bc592debad0312a8eb5 : SUCCESS in 51m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a013e333edb240d3bc23cd9ed24545f3 : SUCCESS in 47m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6527be652f74f319c323611f241b5a7 : SUCCESS in 2h 09m 32s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/04815bf2662347749303b0a967e86896 : SUCCESS in 1h 11m 13s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c41862a732d346a1b9e1ce2ac4e45c52 : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/abd24e6553d84d578a0a21cf8d4711b8 : SUCCESS in 1h 34m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76d0ee8d1c24a89504eb82808f470f15029786c1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-04 23:27:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8852da32f0f450ee0bdd4ca860cab17ca5412764","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 23:36:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749945/1 : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c62fc2d5899df62d3d399ffaa51179584a0b378d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 23:54:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/45/749945/1/check/devstack-networking-ovs-dpdk/4a49a6e/ : RETRY_LIMIT in 1m 40s","accounts_in_message":[],"_revision_number":1},{"id":"80ddda84f53a53c7949bee04099a4daa3382afef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 01:45:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/d840079adc1a4585bfba5a61c609102f : SUCCESS in 49m 07s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/803b5788d32744719c41b813fa0955e4 : SUCCESS in 33m 59s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/0fc6890e23e346768d960b9aab4d91c0 : SUCCESS in 1h 13m 03s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/2d8be849bbab466e992761530d753f95 : SUCCESS in 1h 24m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/a3bd386efaf5483fa04e6f61457e0a79 : SUCCESS in 40m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85fc05ffdd094cbe8e77b1ab54720a6c : SUCCESS in 31m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/feb9e868fe25402cb99ad95747756b72 : SUCCESS in 20m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f67713c9ee3c4a29989fa52ef800aa00 : SUCCESS in 11m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb1ea27b571c48399563c9b6a317021c : SUCCESS in 22m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/70ab9da105f949b8b97af9c73cfac7b6 : SUCCESS in 24m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/135f83c7af314d319ff05a5bdf368bec : SUCCESS in 13m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10a2463a3963481aaac3abbeb79cc12a : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f598e1d2f760425397d2902bb146fd6e : SUCCESS in 5m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5cc5682f43c24a79bdd6f95c2710d96a : SUCCESS in 1h 29m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c0a50df37e254be993e1fd20290886e6 : SUCCESS in 1h 15m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e9a1b4cac71470ab94a4a90d68f7ba8 : SUCCESS in 1h 20m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2445445d01194acb8a532ed6780bcad2 : SUCCESS in 36m 11s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/599b909d5de04c10af55c908d04c629c : SUCCESS in 35m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5f67cfc85e65493cbb5a3e12d8c57f28 : SUCCESS in 59m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fe0e16116fce4a2eb6f82c6ff860708c : SUCCESS in 1h 31m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b9cbe60521334b41b32665a906b86a3a : SUCCESS in 1h 13m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a003c32fc9ad47a797569303cfa045bc : SUCCESS in 1h 15m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdd08b335e1e41188118d968ab39d26f : SUCCESS in 1h 07m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e2233d796944b10ae6c12177ff6f15b : SUCCESS in 45m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cb635d0996ca4786b37ae4079d528e82 : SUCCESS in 1h 26m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99c146afa20c4f69a37de9c04626e7a0 : FAILURE in 1h 23m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/203decd558ea46a284346283dea867e6 : SUCCESS in 1h 36m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed7574270fc246709f5377c03c48f6d5 : SUCCESS in 45m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ff527fce5714e5a88ffd6ae42f41c2c : SUCCESS in 2h 15m 59s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/bd81a26d0dbd4d34b23e87dbb253609d : SUCCESS in 1h 10m 48s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/92c38381e7f04057b581010cb1b2d1a6 : SUCCESS in 37m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b4964d07d8654a578a7ed062c39b61dc : SUCCESS in 1h 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b44f9e79a86dbe46724e07f35e0c421ddd93579d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-05 08:37:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d28debfbb04df7e1fffddcd02c13083231e0d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 08:37:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b71c8818afc0dddc6f7121995027a249b215ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 10:42:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/ab83f2e9335b4049bdd6ac4cc640464e : SUCCESS in 53m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b6e2505042e42e6a8d8ab7f83e9e721 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/254351e44826448a945626c269ce3227 : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e008052473c2442ba762ff307350d44a : SUCCESS in 22m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9836a3895b1c4c0b9ed5b9f19be9a53d : SUCCESS in 31m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e535b9af104b1b8b7ad5dcec5babc5 : SUCCESS in 28m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caadefcabc004d92af5e31d0f92df4c3 : SUCCESS in 22m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2984e427ee044872839f05a7038cc8dc : SUCCESS in 6m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6cfbe5b2e8634ada8507ad86e712cb4a : FAILURE in 1h 52m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2641cb0dc4c14b5db8c5fea8ecf7d0b6 : SUCCESS in 1h 25m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9638a5e95baf47bd94dbf52ef19f42cf : SUCCESS in 1h 49m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6d06c39a94004c298f63c4cb368e57b9 : SUCCESS in 56m 13s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/ec499f625313448ba2693df69ca50715 : SUCCESS in 1h 01m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/53f39df5eb1e4e4588b0a7848c688050 : SUCCESS in 1h 22m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/865abe1aef424b49a908727a914b5906 : SUCCESS in 1h 40m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c7af2fb81eb04dd6907e7fa772c33f90 : SUCCESS in 1h 17m 03s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1d3ba42089742a998e3ebd6f1d7aa2c : SUCCESS in 1h 38m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bdf5d6106fb44f4b632a76a505630d8 : SUCCESS in 1h 09m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7709628ac63b45c899b93a6089cbac0d : SUCCESS in 1h 31m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/641c4539f89646f98d14254b5314cb1f : SUCCESS in 2h 03m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/55e57cb2e387493db1db17587a430924 : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"849da94b6f941c1d81e80c883e4842f950c091b3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-06 08:47:35.000000000","message":"Patch Set 1:\n\nrecheck tempest-full","accounts_in_message":[],"_revision_number":1},{"id":"2eab5320225970244449696c15a2a40510306d27","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-06 08:55:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749945/1 : FAILURE in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5674ac432383c0b1f0cfb5b2e4f4916360696461","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-06 09:00:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/45/749945/1/check/devstack-networking-ovs-dpdk/d891760/ : RETRY_LIMIT in 1m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ce3acef81433c6757f01dfff908dc87bec99081d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 10:50:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/cf3a17f7fe5a4e1da4e55f10458a55a7 : SUCCESS in 53m 32s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/dab55678ae6549f098a3698eeca7298e : SUCCESS in 31m 31s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/29257e98b03d4f8ea51706804e66a7e3 : SUCCESS in 1h 15m 24s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/d6e1817f8e5d40faafc238811c362bf2 : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ff7a6bb76c984f318b8bc1211cccadc0 : FAILURE in 1h 19m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f2ae090aea4de5ac84c856a26767b0 : SUCCESS in 26m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d87de5f08ee491eba015cba5b4056bc : SUCCESS in 23m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda15b08eb1a4f918b80a60b246634c4 : SUCCESS in 9m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/912c80cc98df42f486f79961ef8bead2 : SUCCESS in 14m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fc9a326265074cf49e4a8e37ce2a9de0 : SUCCESS in 22m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7afc4279af4ed1981a3c5095476c7b : SUCCESS in 16m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbc744dad45a46f89fc231035aabfef6 : SUCCESS in 20m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58eddc75ab104613824c62ffaf81dc68 : SUCCESS in 5m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/681a056bd82443aa8eccd716cf3688c6 : SUCCESS in 1h 21m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b32e856cf4bf4312a791451c70b1e398 : SUCCESS in 55m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2eadaf2b32a47dea16a5f292dd43e72 : SUCCESS in 1h 29m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9cf7d2c5cf844ab6b0b85850f757b15c : SUCCESS in 40m 46s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b8bced69fe684c39920ba6955739d137 : SUCCESS in 44m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b6191b3d321a4da8b02cd58ded2013fe : SUCCESS in 1h 04m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1f6563cd941a4b82a54680f5ca98d712 : SUCCESS in 1h 31m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3a85c79e2501419e8559b2ee386775b2 : SUCCESS in 1h 48m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/53ee2ac435bc4cb69a1740a580ae380d : SUCCESS in 1h 15m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce40ccd90ab6421cbce831e556596f40 : SUCCESS in 1h 32m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d83916c449c8468c8693559e2439a8e9 : SUCCESS in 1h 11m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e367c86c4ebd4ed8bfdec67858ad8c20 : SUCCESS in 1h 46m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98dadb7324164b309124a1f9516f7237 : SUCCESS in 1h 44m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1031983f8f664a4b9cff57e37e66eed1 : SUCCESS in 1h 24m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c0280e0c9f3a47e1be258dda6a903f32 : SUCCESS in 53m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/534188b21952419984c67fb740ddb9a2 : SUCCESS in 1h 57m 34s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6acefe2ff039486e9229d7bed45afd7d : SUCCESS in 1h 28m 23s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/70b964076ed543a891c1c140dc0dfe0c : SUCCESS in 1h 05m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/911e3c5f2bad47b6908cf89160272e8f : SUCCESS in 1h 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6db535c434beee3f4e4ee4cb3ae327508681db40","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-06 18:47:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22ec18a8c329dc2f76056e086d103e8c56277675","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-06 18:57:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749945/1 : FAILURE in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f8a71b2371a155385152a6f7b7be58666e72874c","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-06 19:01:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/45/749945/1/check/devstack-networking-ovs-dpdk/c6b87f1/ : RETRY_LIMIT in 1m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1c2950bf30b1252406f6381dfef8ddc611121609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 20:56:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/95bccc64ecd94505b46e55585fd50797 : SUCCESS in 54m 48s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/56f45dcf48544b8b85775da412f0289e : SUCCESS in 36m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/5fb911163c644daa8d02407e18b893ad : SUCCESS in 1h 27m 17s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/1a17d2e29e444e98bf81d43b6f6495be : SUCCESS in 1h 12m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e85c844e5cf84e36a379a9abfb0028e9 : SUCCESS in 1h 03m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caf527bb9c074daf809dc475f1472f22 : SUCCESS in 29m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cc6c3bdf1504e658fd3933ac612a27e : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6583a455495348d185273525e0743963 : SUCCESS in 11m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70fe8bcea13e461cb4d6ebed26f50c03 : SUCCESS in 20m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b41f304f7754a9ba5853cb4a5ede4c3 : SUCCESS in 19m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5578c9922ba45fab93f09e40282a36b : SUCCESS in 21m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5338e5b67adc4a5a9fd16bac28a9cc41 : SUCCESS in 20m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d0d26a144a405e9fd5642706f05112 : SUCCESS in 4m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/90cc8586b9954a79b79d12c482ad0ce0 : SUCCESS in 56m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/de91b7b976214025a5d8bacc40d8ea7d : SUCCESS in 1h 10m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35ffbfc960af4095bd8be002e3aa3d80 : SUCCESS in 1h 14m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/821fcff650014b1caeb37054f8c813b4 : SUCCESS in 34m 47s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/342619e961f74875a5ef3fb85b8cee2c : SUCCESS in 33m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d70772df9134418197ae40cb35e745f3 : SUCCESS in 1h 04m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1fdf3b60a22b400393b69617154b6fd0 : SUCCESS in 1h 38m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f381a58f5ea0475cafa0c757a708107e : SUCCESS in 1h 19m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/03497420b89c4ed2af06f2f793120a15 : SUCCESS in 1h 32m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58bf04aa26d34cdb9d4441ba514bffe5 : SUCCESS in 1h 18m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b7a6c746a0d4ef2a081199582de45a9 : SUCCESS in 1h 04m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0c52fc64006742c18c44954a2c8f6f9f : SUCCESS in 1h 24m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/365885646e174cde9814e904588a302e : SUCCESS in 1h 45m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b814716a5fd4e7c978c5733567b6bce : SUCCESS in 1h 15m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f32e882771b543f5b3622fb86f3854cb : SUCCESS in 57m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6958b583fcc84d6dae6929b2946a8909 : SUCCESS in 2h 02m 33s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4adcb8cedd5349bab3d06aefa46a1796 : SUCCESS in 1h 13m 00s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d793ffa7821e44f2abb1b73bf0d3571b : SUCCESS in 1h 01m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/10ae623134854c6ea49af40640e1b5aa : SUCCESS in 1h 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c46acd65b8def0993efdaf43f49014efad45cc36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 20:56:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d815dca2e3da20825c0c4cdc7c30440ca9fe0b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 22:52:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/689009258ec24d97a504d162987d6d6c : SUCCESS in 1h 00m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/679280bca9ad4ef091d8e99c1fb04abc : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f7e5fad0134e28958cc46b9f33a708 : SUCCESS in 12m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2514f43a1c2c43b68c0ffadb1ebed677 : SUCCESS in 14m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/81096f2c97954bdab25ed3418467a5b2 : SUCCESS in 18m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3014b1441bfe419c962714d17a7512ec : SUCCESS in 36m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcf90c45e2af4dceb569b63397f6c14c : SUCCESS in 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f924a4b79004c92810c7af41254d650 : SUCCESS in 4m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/92b434f1123442afbebaaf1e839dead3 : SUCCESS in 1h 00m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f94911e31a1b4764882e6804f4cc24ce : SUCCESS in 36m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4b9df8ef4814beebfc5fd888bd6871e : SUCCESS in 1h 07m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb61597ac03d4e71a5fcfdb10b8e8ec3 : SUCCESS in 45m 13s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/157767c99b6e485a843b80136a8e740a : SUCCESS in 30m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/408c02ea4a3a4ce2ab073757cd18dcda : SUCCESS in 1h 24m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/021ebc7dfffd448fb65fc2bf0888bf0e : SUCCESS in 53m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/90a17bdfac3c4d2e9daa017b90d07ac0 : SUCCESS in 1h 23m 40s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/704b8dd9f93c429fbb1b8d3d6b9720a4 : SUCCESS in 1h 23m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2ac8fa7133b41dbab19675df517bb5a : SUCCESS in 44m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1670278ae9f140b18885e9498191f3e9 : SUCCESS in 45m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f080f5d8400347bba5abdc3964abe4d4 : SUCCESS in 1h 49m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/68db0e4bdebf4fe1a6278438128bb52d : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7d09652a339d9e66dd7c3beb7ca32f4ca7578b1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 22:52:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"533576de786cf71d03064012866af6f7f3aa1172","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 22:53:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa26e71849f401b96f0843dfc5fdf30 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97f6e2569ee128f8831eba0b9272f30be39133f2","revisions":{"97f6e2569ee128f8831eba0b9272f30be39133f2":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:09:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/749945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/749945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/749945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/749945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/749945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/749945/1"}}},"commit":{"parents":[{"commit":"5656f3cc89b7bfb76dccfe234aa82c08fbb0fc2c","subject":"Don\u0027t raise FileNotFoundError during disabling keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5656f3cc89b7bfb76dccfe234aa82c08fbb0fc2c"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-24 18:10:38.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-04 13:09:13.000000000","tz":0},"subject":"Use dict .get() to avoid a KeyError in the segment plugin","message":"Use dict .get() to avoid a KeyError in the segment plugin\n\nOn subnet delete, the supplied subnet in the post hook\ncould contain a subnet without certain items, leading to\na KeyError in the segment plugin.  Fix a number of these\noccurences so this cannot happen.\n\nAlso fixed similar code in the segment tests.\n\nChange-Id: I645610febde446b78ed6edd868e699673648a4de\nCloses-bug: #1868724\n(cherry picked from commit 07b015d7897d6cea48eec1fdd02c9e19d4990973)\n(cherry picked from commit b019821abd9304f4eeaea4ed3ea37de8e903eb9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97f6e2569ee128f8831eba0b9272f30be39133f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97f6e2569ee128f8831eba0b9272f30be39133f2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
