)]}'
{"id":"openstack%2Fneutron~749946","triplet_id":"openstack%2Fneutron~stable%2Frocky~I645610febde446b78ed6edd868e699673648a4de","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1868724-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"I645610febde446b78ed6edd868e699673648a4de","subject":"Use dict .get() to avoid a KeyError in the segment plugin","status":"MERGED","created":"2020-09-04 13:09:49.000000000","updated":"2020-09-05 12:57:10.000000000","submitted":"2020-09-05 12:57:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749946-1599310630006-d5824c7c","meta_rev_id":"36213ae5ca2ea3dd6817139304e35848f4028b66","_number":749946,"virtual_id_number":749946,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-09-04 23:30:56.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-05 12:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-04 23:35:56.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-09-05 08:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-04 15:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-09-05 08:37:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 15:05:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-04 23:30:56.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-04 23:35:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-05 08:37:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-05 12:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1e56cb8a1564a1af727c4e6a60b0e49f58691bb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 13:09:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"8db1481d1bccbd44abefa747b047a576708edc4f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 13:13:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpjsx38_3u/44fddfdc88b746248252c4b40c01ed4a/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"e0b352dee4ae4015e0baed4dbc9b78d78bb98cbb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 13:18:51.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749946/1 : FAILURE in 8m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e3f3dba902600062c45e39672cbdd948ca24bbc2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 15:05:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85328f5aa647371298314955f0471038ba45a95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:27:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/020191b9ac8d4419a1d11f9decd71a5a : SUCCESS in 52m 24s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/9eae6b12a4bc45e2939f67e65058c77d : SUCCESS in 44m 09s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/13b7c2cea1da4e75a815b973b53bfcd5 : SUCCESS in 1h 01m 28s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/43391a0030d4410cb77f7991af802f6d : SUCCESS in 1h 38m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/7a32f523d2c3435f93d64c8f0afbe1a2 : FAILURE in 1h 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8241530d66294aada302a4b2478f3dcb : SUCCESS in 12m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/310b48ead66546a5a1214804c47b72be : SUCCESS in 23m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a9defb6f2dbd47d9b08b3011275e7732 : SUCCESS in 28m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/64d0a8164791407cb401c29f01b18a49 : SUCCESS in 8m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/911d31533e2740148e7391b3e62260e0 : SUCCESS in 1h 26m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9acf9688441e4deda34011239590cb96 : SUCCESS in 53m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea486ce5e7164200b5189644e0fec2bc : SUCCESS in 56m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7a3baebd122047d084f40fae8b685ca2 : SUCCESS in 33m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/594ccd4ca4724bbab05501cc943004bd : SUCCESS in 1h 43m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d42f2faec2254994ad63ed44e696ed8b : SUCCESS in 1h 31m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ed4e03b6d19d48e185e476c18381fd63 : SUCCESS in 1h 14m 47s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9b2e7d05f3b4e2a943eb0682bac51ae : SUCCESS in 1h 24m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59b75bbe0b0e424cb9430133186610ba : SUCCESS in 29m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4984add57d7b471b9d436fdf180e5a81 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"12f1584ffcf0a800278545a7309b731ae2da29f7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-04 23:27:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d892530806f87f10ec365d978a8deb10c2292b2","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 23:30:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpjsx38_3u/1ddba3ae87fb4cc686f9e070eebb9c56/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"5d9d77eba63b6ca8135db9a07ee146b9ddab446d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 23:35:56.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749946/1 : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"324cae5336251218b12d58c56153df607071efe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 01:19:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/857013ffc71e4bfe82a2c37df535cdb5 : SUCCESS in 53m 43s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/76d5e66dbc06422880cf05b23f5e28ad : SUCCESS in 41m 38s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/63bcca40587f4845929ab98d1fc9359e : SUCCESS in 38m 04s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/c893909cd5954507b8e3c0574aef2d68 : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/6be495811e724a728184036267f22e6f : SUCCESS in 1h 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b40b67ea289443795b22f1e09c65dc2 : SUCCESS in 13m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea9b393b38fb4c8e856a257d0871994a : SUCCESS in 14m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/91f2aa066d534fa08846d3fcfecfd000 : SUCCESS in 24m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6053b06c615149579e474eb8fe8ff342 : SUCCESS in 8m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/be606bfb404a4580825833b4276a6391 : SUCCESS in 1h 21m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b5730775bce14204b5c784bd6ad0dc66 : SUCCESS in 48m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a4e20994ac4437cb542ecb03daa3bac : SUCCESS in 1h 23m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1a1b2d23745445f9b81f572ccc2a431a : SUCCESS in 42m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/11512cca20da42bab1dbbfb318e1a799 : SUCCESS in 1h 50m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/19d8c358b00846589752221a59a6c0b5 : SUCCESS in 55m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8a02b03495594d8cbce227c261a8aeda : SUCCESS in 1h 19m 28s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d886a69da5e4302bb0b0b0b6613c487 : SUCCESS in 1h 45m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/919735dceaf14006bba50dfb671ca11d : SUCCESS in 24m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88749b6949614dfaae0f77e1a63c0566 : SUCCESS in 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d40ad6e32efe73a5d24ce7ced4f6dfe574b59075","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-05 08:37:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a63b2c7e53ec5916ba92db2e53c5169841bd095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 08:37:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3720c103847c6daba42bd727370ca85f3c58572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:57:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/9cab37215f5d49a5aa75f6b2b75dcb0c : SUCCESS in 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69391fc950af44c9a8fa9834f651a922 : SUCCESS in 12m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c349fc95d16e45efb233ab34270b4a3b : SUCCESS in 19m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c21efc32707b4194afb7a4f7a82678dc : SUCCESS in 14m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/76561305932441ef85f99458c663f4e9 : SUCCESS in 6m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f9f2d7a8534547f8a64614951403fed1 : SUCCESS in 52m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/13b687d26f80446ab2a0176c7ff29c9e : SUCCESS in 32m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b47b700885746a49d660b6cef0468e0 : SUCCESS in 1h 18m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba578c27468b4b38acb2421c9c1b7e02 : SUCCESS in 36m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6bed0ccf5ee34960b02b3159ac18b74a : SUCCESS in 1h 00m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a95763b5464e4b9a9cd3466f8a93cd09 : SUCCESS in 2h 17m 21s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7887d5c6d2104536b7e021210fa15d32 : SUCCESS in 1h 20m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cc0ac0ec7434760915f2d8798d9a1e3 : SUCCESS in 15m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/151c7b4abfb24f7a9dca00ee06229716 : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0377e7469411dcf31592c7e9b291159b3363b806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:57:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce8d891c79ed18aa0f7cd353e4121b22a6395a76","revisions":{"ce8d891c79ed18aa0f7cd353e4121b22a6395a76":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:09:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/46/749946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/749946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/749946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/749946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/749946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/749946/1"}}},"commit":{"parents":[{"commit":"a1a50a473f6c2bd00c72c017983c553af60172e5","subject":"Merge \"Handle properly existing LLA address during l3 agent restart\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1a50a473f6c2bd00c72c017983c553af60172e5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-24 18:10:38.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-04 13:09:49.000000000","tz":0},"subject":"Use dict .get() to avoid a KeyError in the segment plugin","message":"Use dict .get() to avoid a KeyError in the segment plugin\n\nOn subnet delete, the supplied subnet in the post hook\ncould contain a subnet without certain items, leading to\na KeyError in the segment plugin.  Fix a number of these\noccurences so this cannot happen.\n\nAlso fixed similar code in the segment tests.\n\nChange-Id: I645610febde446b78ed6edd868e699673648a4de\nCloses-bug: #1868724\n(cherry picked from commit 07b015d7897d6cea48eec1fdd02c9e19d4990973)\n(cherry picked from commit b019821abd9304f4eeaea4ed3ea37de8e903eb9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce8d891c79ed18aa0f7cd353e4121b22a6395a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce8d891c79ed18aa0f7cd353e4121b22a6395a76"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
