)]}'
{"id":"openstack%2Fneutron~749949","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I645610febde446b78ed6edd868e699673648a4de","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1868724","hashtags":[],"change_id":"I645610febde446b78ed6edd868e699673648a4de","subject":"Use dict .get() to avoid a KeyError in the segment plugin","status":"MERGED","created":"2020-09-04 13:15:29.000000000","updated":"2020-09-05 12:57:12.000000000","submitted":"2020-09-05 12:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749949-1599310632867-b29784c5","meta_rev_id":"0d87e33bc05322c14ebf46fcc288e5cd983bf2a9","_number":749949,"virtual_id_number":749949,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-09-04 13:21:06.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-05 12:57:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-04 13:23:42.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-09-04 23:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":2,"date":"2020-09-04 15:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-09-05 08:37:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 13:21:06.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-04 13:23:42.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-04 15:07:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-05 08:37:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-05 12:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db774cb54ee6a1920bb3ffa44ece7c7df130456f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 13:15:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72d1e0c5d4347a9df211dc09ed8ab255b4e5d291","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 13:21:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpjsx38_3u/bc554c8f0c0f4e8f8e6df3fa67be4a08/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"cbceb688c21064c4ec1c6935e1d379ffb712e746","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 13:23:42.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749949/1 : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"438795e86bb5331d8f2b38008c55a976dfdb1d6f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-04 15:07:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6347d9501f8acdd3d283b453d84055751aaec27e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:02:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/0bc8b14b30e5490a8cf78cd0a781e5f0 : SUCCESS in 49m 10s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a3914c3b482b4b9d92aef27f52e79eae : SUCCESS in 53m 28s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/8f59ca9b0d6c4ac58c1956b4b23157c7 : SUCCESS in 55m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a633fca50004377bba535854cb8bc70 : SUCCESS in 12m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c1ac6b618314bc882e18377c591adb3 : SUCCESS in 23m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b98735e81b434841977a68f2fd8ec598 : SUCCESS in 27m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8d086caf323447cd91b781669e2c3b73 : SUCCESS in 9m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/86ca3b01830240af9d92af3fc6ea7d3f : SUCCESS in 1h 30m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aba695a238cf47c6be70c1afee068e15 : SUCCESS in 54m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60fac3f8d10748d9bb47ddda381c800f : SUCCESS in 1h 18m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc8206f77178448b87cd7cb47d2dd20b : SUCCESS in 41m 01s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/6e61f89251e64454b141384768e26709 : SUCCESS in 1h 50m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/41cb7a6aec0a4054b0c11266242f1474 : SUCCESS in 1h 27m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/308b330e11ad4d73bae20f564d84e3a6 : SUCCESS in 1h 22m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1dff908d44045068de2c486dcf595cb : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4cd941e7e758cc5c7bf78a5554c994c9fb5df39e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-04 23:28:33.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, waiting on previous branches to pass CI","accounts_in_message":[],"_revision_number":1},{"id":"4d7a2dd942caa530a2f10ffc1f217b51af682b1a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-05 08:37:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37471176c0a0d4f2d010eabdeac5071e2fe157b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 08:38:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3b2385d0c74a0c0c9959147e2a338adcffeead9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:57:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00d1a6d1e1a931ecf2abac7cd4bf38eace53c717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:57:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/2bd8890031b749928c69ae9b2e158902 : SUCCESS in 46m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa7c6db5bbf45c69fe59ba0ebe2559d : SUCCESS in 10m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70c8f8c9c7914b32ab64c9879bed88cc : SUCCESS in 20m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/801b6c09396b485cb0f17090f65a0eb2 : SUCCESS in 23m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/346e4c4a1abf4aa2aa0cbad791ee711e : SUCCESS in 6m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a35ecfe52b7945fa8395715004334101 : SUCCESS in 1h 18m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3291e32d54c145529b68ea245b2de475 : SUCCESS in 52m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56f7697a4c054608b76b37320a1ce9ed : SUCCESS in 1h 21m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c2cb5e468d864ffbac3bb2fec588409f : SUCCESS in 41m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/96dec570e2b04745bdc2b55da4e85cf7 : SUCCESS in 1h 19m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fa5c1c4b428d47a38148e20cf2c743bc : SUCCESS in 1h 17m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4166f55473c34ff5a55a4c76344b5a17 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45f56356065a962c242e187c5ee6cacfa0bd89f8","revisions":{"45f56356065a962c242e187c5ee6cacfa0bd89f8":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:15:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/749949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/749949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/749949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/749949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/749949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/749949/1"}}},"commit":{"parents":[{"commit":"84562fc8921bbb2677d10978800945c29c25ad62","subject":"Merge \"Pass context in disassociate_floatingip notification\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84562fc8921bbb2677d10978800945c29c25ad62"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-24 18:10:38.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-04 13:14:40.000000000","tz":120},"subject":"Use dict .get() to avoid a KeyError in the segment plugin","message":"Use dict .get() to avoid a KeyError in the segment plugin\n\nOn subnet delete, the supplied subnet in the post hook\ncould contain a subnet without certain items, leading to\na KeyError in the segment plugin.  Fix a number of these\noccurences so this cannot happen.\n\nAlso fixed similar code in the segment tests.\n\nConflicts:\n    neutron/services/segments/plugin.py\n\nChange-Id: I645610febde446b78ed6edd868e699673648a4de\nCloses-bug: #1868724\n(cherry picked from commit 07b015d7897d6cea48eec1fdd02c9e19d4990973)\n(cherry picked from commit b019821abd9304f4eeaea4ed3ea37de8e903eb9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45f56356065a962c242e187c5ee6cacfa0bd89f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45f56356065a962c242e187c5ee6cacfa0bd89f8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
