)]}'
{"id":"openstack%2Fneutron~749960","triplet_id":"openstack%2Fneutron~stable%2Frocky~Ie5cbc0dcc76672b26cd2605f08cfd17a30b4c905","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1863006","hashtags":[],"change_id":"Ie5cbc0dcc76672b26cd2605f08cfd17a30b4c905","subject":"\"ping\"/\"ping6\" command support in rootwrap filters","status":"MERGED","created":"2020-09-04 13:27:57.000000000","updated":"2020-11-15 11:24:20.000000000","submitted":"2020-11-15 11:24:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749960-1605439460851-1ed5524d","meta_rev_id":"d26244da61de8d5dc6f10542ee054e1f96fff0a8","_number":749960,"virtual_id_number":749960,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-11-13 08:17:32.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-15 11:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-15 08:27:41.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-11-10 08:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-11-10 09:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-11-10 09:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 13:33:06.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-11-10 08:01:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-10 09:09:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-15 08:27:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-15 11:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa1d535506524b924f29bcd0ed2548dc57791424","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 13:27:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e62c28da2c99ab077b229f84ef90311a4f15f8e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-04 13:28:18.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 4d9d77fd2e48ad68d1affd6ecca2acf42766a855","accounts_in_message":[],"_revision_number":1},{"id":"a3f06e28d5831fb33264e89abc0d17067b6625e3","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-04 13:33:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpjsx38_3u/b81c4e7980824da29a8d5ca2acee3b65/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"42a7d0428a5ee1202b1a7bfb6d6f8edc15727a79","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-04 13:36:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/1 : FAILURE in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"11a3987f0d5af85ce8a79adcd051b7eaef6d31b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:42:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/59a7f5cc22324d2781438436731056f8 : SUCCESS in 53m 10s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/b264f5fd52584f7c955a61a189f84498 : SUCCESS in 37m 43s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/861bf7254f154bb89ff3bc0b273dd777 : SUCCESS in 1h 02m 40s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/ba78faebe25d4a1d89a531e03455c6ba : SUCCESS in 1h 09m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/315f2b1385c34a41bdaf1f3247077a95 : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c8db538bc3496b93afa65217110b49 : SUCCESS in 12m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ad2ede222b249d4bc9c2691e6b8e3f7 : SUCCESS in 32m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f3500d308b140f1a4a9aa11605cb402 : SUCCESS in 29m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/626642e4446c4e4b921f7b40670ef97a : SUCCESS in 8m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a80fa960036f498ebb717217e5f3bfd4 : SUCCESS in 1h 28m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cffd6ca98a774e6c84c97e3f82aef595 : SUCCESS in 57m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e07b1fbedd2347fa98e02cc72d006cbc : SUCCESS in 1h 20m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c29881f83dea4cafbdd85b24b9997be7 : SUCCESS in 31m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ed9edd4aa45a4e63ba0d5f98a17c15b2 : SUCCESS in 1h 29m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f3aa06f859184e87b025ef459d3cfa54 : SUCCESS in 1h 33m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ace03de012884f4fac694f76c085abce : SUCCESS in 1h 18m 33s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8fe24a888e994f47a0c6e3fa09aba50f : SUCCESS in 1h 00m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c79cc73ef787449aa72761a7165c7102 : SUCCESS in 27m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc196dd480e14d80a4ee8a22c4dbe70e : SUCCESS in 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c3bf458b4ed427e0e9410cefe84ce0e4ab3d2e2e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-05 08:43:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5fa7f45a455b95f347277ff1785037b5626e1560","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-07 09:06:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20014b375330aa6ef225ed1858cb6463dedff24b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-01 15:26:36.000000000","message":"Patch Set 1: Code-Review-1\n\nThis should combine https://review.opendev.org/#/c/709100","accounts_in_message":[],"_revision_number":1},{"id":"8fe76c74ab4a9f3181fcb659581e9ae906e902e6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-01 16:43:44.000000000","message":"Removed Code-Review+2 by Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"9b309b607130e2eecac3db4befc374222cd685a0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-08 15:31:06.000000000","message":"Patch Set 1:\n\n\u003e This should combine https://review.opendev.org/#/c/709100\n\nAre You sure? Because there is no rules which this patch removes in stable/rocky: https://github.com/openstack/neutron/blob/stable/rocky/etc/neutron/rootwrap.d/debug.filters","accounts_in_message":[],"_revision_number":1},{"id":"e315505557c803a1647ea94802f964e2e79d172d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-08 15:48:31.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e \u003e This should combine https://review.opendev.org/#/c/709100\n \u003e \n \u003e Are You sure? Because there is no rules which this patch removes in\n \u003e stable/rocky: https://github.com/openstack/neutron/blob/stable/rocky/etc/neutron/rootwrap.d/debug.filters\n\nWe need to remove from this patch \"ping_exec\" and \"ping6_exec\".","accounts_in_message":[],"_revision_number":1},{"id":"9a629a1f65e76c837cef09193008c14297a4cc9b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 13:52:01.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e \u003e \u003e This should combine https://review.opendev.org/#/c/709100\n \u003e \u003e\n \u003e \u003e Are You sure? Because there is no rules which this patch removes\n \u003e in\n \u003e \u003e stable/rocky: https://github.com/openstack/neutron/blob/stable/rocky/etc/neutron/rootwrap.d/debug.filters\n \u003e \n \u003e We need to remove from this patch \"ping_exec\" and \"ping6_exec\".\n\nNow I got it :) Thx","accounts_in_message":[],"_revision_number":1},{"id":"8157da42d4f219bbfd128f36b38e076b0d804eac","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 13:54:19.000000000","message":"Patch Set 1:\n\n\u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e \u003e \u003e This should combine https://review.opendev.org/#/c/709100\n \u003e \u003e \u003e\n \u003e \u003e \u003e Are You sure? Because there is no rules which this patch\n \u003e removes\n \u003e \u003e in\n \u003e \u003e \u003e stable/rocky: https://github.com/openstack/neutron/blob/stable/rocky/etc/neutron/rootwrap.d/debug.filters\n \u003e \u003e\n \u003e \u003e We need to remove from this patch \"ping_exec\" and \"ping6_exec\".\n \u003e \n \u003e Now I got it :) Thx\n\nDone in https://review.opendev.org/757105","accounts_in_message":[],"_revision_number":1},{"id":"697832749c7fe456057eeba6f0a57f2640c317a9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-03 08:00:19.000000000","message":"Patch Set 1:\n\nWaiting for 757105 to get +1 now that https://review.opendev.org/#/c/758377/ is merged","accounts_in_message":[],"_revision_number":1},{"id":"2a43929333074b2d2735a1540c6a5750fb22b44b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-10 07:56:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cf845c1badfb4f01cad280abca813fa2f9a3e483","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-10 08:01:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f1c9d6395433199bb4e4ae1fe00e837b0c21f831","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-10 08:06:36.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 9m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3c0a4ff8c49e6557341e9696c52908321e1df586","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-10 09:09:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1f35ecc97b7a61c1acf911670924e3c772a971ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 09:45:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/6583ac1b305f4357b20f9b7e6ea82149 : SUCCESS in 45m 40s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ebd971b7c87842aa9af3288aeef9b3b2 : SUCCESS in 27m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/da1871550ba145698b0005f6f12d54dc : SUCCESS in 1h 00m 57s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/f28e15409cc04e358b45fed9c185c8d1 : SUCCESS in 1h 27m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/bbda93e4f9194c70a3367cf202c162f7 : SUCCESS in 1h 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214e35e225324e5f8c37c0832fa20a02 : SUCCESS in 12m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de87a2f617494fb8b53103afdb03840e : SUCCESS in 25m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e4e74379baf4b6ca0949e933e7287af : SUCCESS in 28m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6afde2bf61024922939d55088a833057 : SUCCESS in 7m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c2456439911c4c549dca7a09fffcc19a : SUCCESS in 1h 28m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b5de452da2c6470995f151fb8196fa52 : SUCCESS in 47m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8df6626b672489a861d548aa82700c4 : SUCCESS in 1h 24m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/15a1523885544a82a5717b8d63d44c5b : SUCCESS in 45m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7dde9afa2f1d4c659ad03a12757a45ff : SUCCESS in 1h 26m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6929824d06cf4f3ea319e8d237b1a79c : SUCCESS in 1h 17m 31s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/561ec6d4a4ee47988a542f557ca7475b : SUCCESS in 1h 22m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49cb0a51e12848aa9856f4a8b9fa1377 : SUCCESS in 29m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6663012bc454740ac647ee943f976d8 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0720fab19881a32229d444cc0aedcad5faeb9aac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 09:46:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bef273582b4dd52ef6609537440b5e2cb2da79e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 14:26:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/663a36f8a08b4e298b4130b583a21f7e : SUCCESS in 49m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c05b0a5d6e41a7a1a9d3f86c8c6a83 : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/346858a7da454decb05175f1aede5ccc : SUCCESS in 31m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/33f3a2e3e6dc45bab09460f3d7af6834 : TIMED_OUT in 41m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7b19a5b0264647f1bb51f6ce6d1187b6 : SUCCESS in 8m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/81c29da099c142908aba587f77e59d6c : SUCCESS in 1h 43m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/34c5ffd0068f43dbbe5e5e4b80d6a62c : SUCCESS in 53m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/795607b17f1a4a11bbdb06d2da472b54 : SUCCESS in 1h 29m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/34b521f9e0e6435e821cae25370bc2d3 : SUCCESS in 44m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ac536167a70a48fba4795f05da750367 : SUCCESS in 1h 33m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7554bbe1eebc4271bd431a4707de8e6b : SUCCESS in 1h 53m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8c8077d221a4d0297a95e7b5753b394 : SUCCESS in 1h 47m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9a9bf3eba6f4574a4e797678f009073 : SUCCESS in 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/092f0e0b5027476583098e81fc0ee662 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"73441bd08a9d99dcb4af5213f4d108180b1d4d06","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-10 19:52:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5dc660c5ebdbff16af368d2d2f31201fa8adcc7f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-10 20:03:05.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"aa66534df213c1ebcc23b49c858b47395e877654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 06:36:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/f254b74b3dbd480092d551fb7b4994cf : SUCCESS in 45m 36s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/297e574372dd4c45809610ce1ec491aa : SUCCESS in 33m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e113ec337fba43e387674e1d62c2e114 : SUCCESS in 1h 09m 23s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/be94763994554f81a03b50e72e1405d9 : FAILURE in 1h 22m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/fbabbec1b6ac4317ae44265cc516bb86 : SUCCESS in 1h 06m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62f0838f51546a993b0f66bf792d960 : SUCCESS in 11m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48fd47dfc96e44938408fdad373546e5 : SUCCESS in 24m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7b14ac29b8e048da8f705f333505140c : SUCCESS in 25m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/34f255fcce904e67a0d5e9f1313e876c : SUCCESS in 7m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/836ca4c9a7cc4415866915096aaf646e : SUCCESS in 1h 29m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b82505302d0a4456bdc6a14e1f157807 : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53d04c188fc04dd09e7f71b2fb78ce35 : SUCCESS in 1h 21m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ef43edb777f747708eb55bb6557dd0b5 : SUCCESS in 37m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f1700e1b4138408880c4dff80ececd98 : SUCCESS in 1h 35m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/22380c9c6a4746f687e90e44f0699092 : SUCCESS in 1h 21m 45s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec2168b51a974c2c9a29fa98b2976d01 : SUCCESS in 1h 23m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84545dddd5f44080ae513e56097af57c : SUCCESS in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6e8f9aed354895adcb353d1dffc220 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":2},{"id":"6dee08a02b1056309041c276e71c9a056f7c71b8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-11 22:26:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d999cd71d277d7d83da4e1d8b016fdd0204da0e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-11 22:36:40.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 9m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a1d057704470fc5ce8e3577dd640a12e6b1c69c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 00:41:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/d0f94eee1cc34ea88f7a5ccb96cc4448 : SUCCESS in 56m 22s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/554299a9f17745fc9665c7ffa55d7e7f : SUCCESS in 35m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/408398c71ef147cda0397138040407d0 : SUCCESS in 1h 13m 46s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/4fda26260b3f4c16b1c121fa6e1a245d : SUCCESS in 1h 10m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/b89d46c4068e4f979dd5863c7221df89 : SUCCESS in 1h 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac6f1bb736346eca0dcc54481755471 : SUCCESS in 11m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a69e6b733cd4df686ede76b7ed3a345 : SUCCESS in 21m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a88112db805a4ea0a92db88f8183a5df : SUCCESS in 29m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d4d9ca5e7bdf40dd96659b73d7732bbd : SUCCESS in 5m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ec8214952ea4091bb3b8abcb4af7d93 : SUCCESS in 1h 32m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/be5a541009fb40e5be429d43ca641a09 : FAILURE in 54m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f696bc735cb4633b7b1218742b09e74 : SUCCESS in 1h 45m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/87a5f07c1a2945af984e5102fea22f8a : SUCCESS in 43m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3793cddf8c014b8fa553e787cc88e1d7 : SUCCESS in 1h 29m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2719f82291e04ba0aa0a7eb34141fce1 : SUCCESS in 1h 25m 52s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66e659c2ca60439e9be7d9c709772b7f : SUCCESS in 1h 27m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bea5af66ddc9454a9f515790e996bc66 : SUCCESS in 27m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89ae242e54324a98bb945dc2c907e12d : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9c1390ae69268687cf5d7b7572c961340405103e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-12 08:39:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b964f48ab90652a2dde4187b74df6d1b9c44aec","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-12 08:49:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 9m 10s","accounts_in_message":[],"_revision_number":2},{"id":"c6cde8e34a3899d2e76eab0c9d82d3b8d7b1c140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 10:19:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/dca8305acf7640cb9263685b4e6d6238 : SUCCESS in 47m 15s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ced0289912224d20a182ffa2e90e3485 : SUCCESS in 28m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e6cc29a83ec3484f85c74cedd9584f31 : SUCCESS in 1h 11m 53s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/eea9ed35a28848b29ec64581e3f653f1 : SUCCESS in 1h 22m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/576abdbac3d94f31acf16bdd9d28a3b0 : SUCCESS in 55m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a38a4fc9a68b448c9154fd8551caccb4 : SUCCESS in 10m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78cfe9af2e1b48b9a6118a545e8498f6 : SUCCESS in 26m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/85f5d6cd67b045a58600fac6ed30f8ed : SUCCESS in 31m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b6c69970b4c545cba6099b4926170230 : SUCCESS in 7m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3676cb6df75844769e6463e064cf32d7 : SUCCESS in 1h 25m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e808be5b8aae42078c73d0839c024270 : SUCCESS in 53m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79199bd6d13142388653a9406c017e89 : SUCCESS in 1h 22m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc99434cae89452285fde0572b4a35ca : SUCCESS in 40m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5a05864916f74f8fab4d1eeb65c6a5e6 : SUCCESS in 1h 27m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/086eab5095f0481babdd58e633d7088b : SUCCESS in 1h 17m 26s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81c99717541346a08fadd0f19db26266 : SUCCESS in 1h 20m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caf2515c6fff42779016108f361a5607 : SUCCESS in 32m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/861ee36093dd4076aee434994dbf16c0 : SUCCESS in 25m 13s","accounts_in_message":[],"_revision_number":2},{"id":"936f0d3b5d27cc4e600ad280bc9760b213db3173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 10:19:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9eaa6ceddcefbae2c80b5b37f0742b4e04c3fa1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:09:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/c980ab6e207a443dbee5cff9f1fdb9ee : SUCCESS in 59m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1542749a1d34e3ebc527232215072a7 : SUCCESS in 11m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f172f85dad74fb98b947ef85b57bedf : SUCCESS in 24m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/602658b1d0504d959c96b935760727ea : SUCCESS in 28m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0c7fffe3c933494b970a245e43429616 : SUCCESS in 6m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/30b8300cbe4e41d4aac7da9d84fadbc6 : SUCCESS in 1h 28m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/db34976136614a5bb5f6d988fc5ce179 : SUCCESS in 41m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43641268c25b487bb9bc2b136cdfa730 : SUCCESS in 1h 26m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d76bbf0f6c144642891cf5adb5043072 : SUCCESS in 45m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e5beb3684f2d4565b1e0d177c5624e82 : FAILURE in 1h 20m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c89973f04ab749a097642b11c6470f6f : SUCCESS in 1h 21m 04s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26154e73039c47f78ea89cb3b8c24fd5 : SUCCESS in 1h 40m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18eb326586634144874394e2de0f990b : SUCCESS in 33m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a02093097bff48e19c76d11a66162e24 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4cf2d7593f7a7073f4e387df5ea3eea75cafbbff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-13 08:17:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"63455b722745ab85eb30400dd53680b83e783633","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-13 08:26:23.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ab411ffcd37048706faf2c2391ede13d827a3e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 09:50:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/a39211dfd9dc4aebbf09633b4bccb445 : SUCCESS in 46m 21s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/8e7b49508ed04ee8b7d419330bdb8e95 : SUCCESS in 25m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/d4349658f8784c929c57de42bc8fee20 : SUCCESS in 1h 11m 15s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/ed0036cc764045609b2d43bff8c2603d : SUCCESS in 1h 06m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c7acfcfcd6fc437fa8123add02c8d960 : SUCCESS in 1h 07m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6751ef4f4e841608babe22dd7125a3e : SUCCESS in 12m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c4a5ce39f774daca6c8881a2168a946 : SUCCESS in 21m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c1e86f62fb4f47f4b6e960ee183321c6 : SUCCESS in 26m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c6d04bd2ff3a4552986cf9feddb42ade : SUCCESS in 6m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/821e7750b71c4faaa7cf63b6c0549912 : SUCCESS in 1h 25m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83731f90cdd84ab094d9d7da12592b39 : SUCCESS in 56m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01adc30736d94e289e46cc40947d8734 : SUCCESS in 1h 24m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21e5c198ca60422cbaf8fd46624aa658 : SUCCESS in 43m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ca725a20e06045698135240debb3bd11 : SUCCESS in 1h 20m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ced102f071984c8890d67ac3a0ae67b3 : SUCCESS in 1h 19m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d962863dcd2c4f0ca33c58db1ebbddee : SUCCESS in 1h 25m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/741f2f1b97db4c2bb13065af2626c9fd : SUCCESS in 25m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79ab4f80471947d9bf44f53032fb8cbf : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b519d9f150f86c127c09ff48e1902b9fbe29d339","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 09:50:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3cf8cc11ed264ab12540b41b8fd72b7a0bd87e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 12:26:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/f9cf2c9309b744409912032cd4ecd582 : SUCCESS in 56m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7041860e8942d2bcea6f18af619760 : SUCCESS in 11m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9133acf252c947a39046d8f2c52b3cdf : SUCCESS in 24m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0e71d72331b421f842e70dd4e33b693 : SUCCESS in 22m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0f6b7345b5584bdaaccfe5c2b5047f3e : SUCCESS in 6m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ac603c260be740049413dad82e258317 : SUCCESS in 1h 23m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c3cf810869e54b10bcb870a0994221e4 : SUCCESS in 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e00c4437a0b643779794effb08c0bf59 : SUCCESS in 1h 15m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9bf3f1319ce348578bab80ab9fb203c8 : SUCCESS in 40m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3cb31dac69764ff580d0e97bcc0ebe9b : SUCCESS in 1h 26m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/787e30b6b5c34e24a2cde06aa8859a9e : SUCCESS in 1h 35m 52s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b2097af8c374a7e8d696e08c7254791 : SUCCESS in 1h 28m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849ee8780d094e18b0d34d0c86eb70ce : FAILURE in 25m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43928f5a339a4173bf68aa4e6b62a8bb : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3a4ad95c77d7dd9741789cbd15b0037e9396f670","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-13 17:17:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3ab8f29a2a97bcc077a6bbec4f2f1894144ec119","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-13 17:48:54.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b1415495e29f2b2492c3b0be8e4e808d3ce36002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 20:39:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/1c0866ab1e334874aa3aecc60bba1e1e : SUCCESS in 51m 11s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e3cd700817014e9981077dbf963ade04 : SUCCESS in 35m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/d727191004d84c5e834c1ed673c1919b : SUCCESS in 1h 13m 56s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/bf93808e17604421b3b45bf90cc11e3f : SUCCESS in 1h 23m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/b716ca81baec40899b51e2772c182227 : SUCCESS in 56m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3cb0d880944a0d8d834ef02e6fc3de : SUCCESS in 12m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/013db1b515ef43e7b94f5b3ab910633d : SUCCESS in 28m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/edd8b77bfcfd4f408f1115515f78e78e : SUCCESS in 32m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/73466633da29419a84951564ffb9a6ce : SUCCESS in 7m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/720e3b058ce24f3e979d0cdd6d76deb4 : SUCCESS in 1h 22m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e224b0f2638e48daa87b28c02af4f66c : SUCCESS in 52m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f898ec585934cd885fd2735cd7dfe6e : SUCCESS in 1h 26m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cdbd719ca6e3470084be621740d536c3 : SUCCESS in 41m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6b5f269b46114133912ce7ff62b5339b : SUCCESS in 1h 19m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8583d4f111094f8cb7e6f0841a42bdeb : FAILURE in 1h 10m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c6b154e7e3b4ea2af367f6bcb4b5885 : SUCCESS in 1h 37m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/730459c67f3a4eee9102431b36ad4575 : SUCCESS in 30m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21a8e8767498451888eab1101035d0cb : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":2},{"id":"498cf4f1c43ed6eab698e8e07715f87d61e020d6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-14 08:15:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46e29c2c6a6c40e945c4e97fa75d54ac8dd5c603","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-14 10:15:04.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 24s","accounts_in_message":[],"_revision_number":2},{"id":"5186827dedc504bff528cb4b61e018c84f38da22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 11:24:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/c33e9e1d23ee47809f31fdf7ac25e583 : SUCCESS in 56m 06s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ad58c368d3344d2293caef2270f06aaf : SUCCESS in 26m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/37697129f09349c5b4a1e0e852e32954 : SUCCESS in 1h 03m 59s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/abe8198c450f42779448325f5a1499ed : SUCCESS in 58m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/db7ff51b8a4840d681aa710409ec2a31 : SUCCESS in 1h 00m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb73f887b0644b888918b97ac0b7cf36 : SUCCESS in 9m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3895395f09144bc48c16063a2ac98d5d : SUCCESS in 26m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d9dedf8776241cdb7b425c8ebae8b89 : SUCCESS in 27m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/eb96f2b8806b4d7788da1b4a86191e15 : SUCCESS in 28m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c658575a537c4c10aa7f3b0dfed53089 : SUCCESS in 1h 33m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/046fc87fa2ac4435a3d6c01f17eea388 : SUCCESS in 37m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ed9e3f403e947c1b8d2c6d2cfa191aa : SUCCESS in 1h 19m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cf4e9d593a034267805e93a14c84c254 : FAILURE in 34m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b498706f24be489f9737a41e5370f697 : SUCCESS in 1h 27m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/798b36b173364dbbbcbe80da6478763c : SUCCESS in 1h 33m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6dcc98881c3a4943a13e02cc3339fd12 : FAILURE in 1h 11m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d99c13406ee4c11a790a4f5a58a8c4a : SUCCESS in 29m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7d7e397198d4cd68743c57793b9350a : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"42276df9bafcb200f97b2e0da694f1c6b388e070","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-15 08:16:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"80e55dd6f3e2af6460764eb8c6efac98afcf3822","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-15 08:27:41.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/749960/2 : FAILURE in 10m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f7fd793961d30aec94e4e02a594d56b3afe0f580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 10:07:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/2507fabb831745878ef106ac795974fb : SUCCESS in 57m 08s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/c22b447d96c2440fa460c54ef0c45d22 : FAILURE in 41m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/0af05ba4bd644baaa8f65f5428edfcca : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/3943006047174c37973972c5648f95b3 : SUCCESS in 1h 03m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/18efa28c567942dabdf4fd7a9c4d920f : SUCCESS in 1h 01m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7063e21e42b43418566117e326fdcc6 : SUCCESS in 12m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a9f61b5a99140afbc1f6c8a0baf54f2 : SUCCESS in 24m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/06f738b212374a10ad84e2dcd7dfa664 : SUCCESS in 29m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/81183fa6b7cc4edaa4deed266324f230 : SUCCESS in 8m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a7ed109fccf74a82b5cd13463211e106 : SUCCESS in 1h 41m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe44ad7f74e9483f9e3716d8905703dd : SUCCESS in 1h 11m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/044f7fb2edd7466ba861cadf4c73d4ba : SUCCESS in 1h 27m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/724bcaa1cfbe46a198c1df38aa1e64d2 : SUCCESS in 51m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/370655fa5fe44cf78dc815776733acbb : SUCCESS in 1h 39m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/34108534944b496c8afc4f8f3a077c6d : SUCCESS in 1h 34m 44s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/afdf4d6b8d974830ac1b0e89a2975c07 : SUCCESS in 1h 31m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1687d11f961406494ad4c6caea8dbe8 : SUCCESS in 27m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e48dd25f415e485384423af2728fa3c7 : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":2},{"id":"254975554e61054261c904dd00ae0a846cd4f163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 10:07:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d0e0057118899dbb5bdd46a58752aeb92c24b13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 11:24:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/1225a87b5c3a4f75b60e13c557a5f3da : SUCCESS in 55m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edec31aa2d73439dafcc77bb952ccb75 : SUCCESS in 10m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2921144e519c43cfa091d989983aa36d : SUCCESS in 23m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/65a7f3423b084850b09d5849adebe979 : SUCCESS in 29m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0219c802d18d4a038dc354508630d186 : SUCCESS in 5m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/269defa3fb9e445e90cb01a2bb93a41a : SUCCESS in 1h 11m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a5ce7b3ea7c4861a6faf36e53546ed3 : SUCCESS in 46m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3384370bbc084f93b06fd3945ba47294 : SUCCESS in 1h 09m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e8daeffc8ab84f06b50589a0eac1153f : SUCCESS in 35m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c0fef2bbc6443c7a33aad9272f9721a : SUCCESS in 1h 13m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e9449098bedc44d1aca5cdf439fc6542 : SUCCESS in 1h 06m 26s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02c6f39ecba04b8d902cda2f45fe7321 : SUCCESS in 1h 12m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deba66aad3e646a29b8be9ffd0f7be60 : SUCCESS in 25m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca568d009c614ca9a2176b565f690ebd : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"778f14df37352757193a8acc32ff00798a5628cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 11:24:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"055484a3ab2fd6433487eac294b1bc3e00be3030","revisions":{"33a33763d98a0895df2fa528210a2ea402fd82b3":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:27:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/60/749960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/749960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/749960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/749960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/749960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/749960/1"}}},"commit":{"parents":[{"commit":"a1a50a473f6c2bd00c72c017983c553af60172e5","subject":"Merge \"Handle properly existing LLA address during l3 agent restart\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1a50a473f6c2bd00c72c017983c553af60172e5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-12 18:56:13.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-04 13:27:19.000000000","tz":120},"subject":"\"ping\"/\"ping6\" command support in rootwrap filters","message":"\"ping\"/\"ping6\" command support in rootwrap filters\n\nTo have correct support in rootwrap, \"ping\"/\"ping6\" command should\nhave the correct filters in rootwrap.\n\nBecause \"ping\" command is harmless, \"CommandFilter\" is used to allow\nany binary call, regardless of the parameters used and the order.\n\nNevertheless, this patch also proposes to use \"ping\"/\"ping6\" with\nthe same parameters and a specific order, to help in the debug\nprocess:\n- ping[6] -W \u003ctimeout\u003e \u003caddress\u003e\n- ping[6] -W \u003ctimeout\u003e -c \u003ccount\u003e \u003caddress\u003e\n- ping[6] -W \u003ctimeout\u003e -c \u003ccount\u003e -i \u003cinterval\u003e \u003caddress\u003e\n\nThose commands could be called from inside a namespace. The needed\nfilter is also added in this patch.\n\nConflicts:\n    etc/neutron/rootwrap.d/debug.filters\n    neutron/tests/common/net_helpers.py\n\nChange-Id: Ie5cbc0dcc76672b26cd2605f08cfd17a30b4c905\nCloses-Bug: #1863006\n(cherry picked from commit cc3b9df4268ac339d0b7316595a4148ba4c69836)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33a33763d98a0895df2fa528210a2ea402fd82b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33a33763d98a0895df2fa528210a2ea402fd82b3"}]},"branch":"refs/heads/stable/rocky"},"055484a3ab2fd6433487eac294b1bc3e00be3030":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-10 07:56:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/60/749960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/60/749960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/749960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/749960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/60/749960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/60/749960/2"}}},"commit":{"parents":[{"commit":"5936e4cea9f38d88004a9e5a019615dd20b50100","subject":"Merge \"Load the glibc library only once for Pyroute2\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5936e4cea9f38d88004a9e5a019615dd20b50100"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-12 18:56:13.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-10 07:56:43.000000000","tz":0},"subject":"\"ping\"/\"ping6\" command support in rootwrap filters","message":"\"ping\"/\"ping6\" command support in rootwrap filters\n\nTo have correct support in rootwrap, \"ping\"/\"ping6\" command should\nhave the correct filters in rootwrap.\n\nBecause \"ping\" command is harmless, \"CommandFilter\" is used to allow\nany binary call, regardless of the parameters used and the order.\n\nNevertheless, this patch also proposes to use \"ping\"/\"ping6\" with\nthe same parameters and a specific order, to help in the debug\nprocess:\n- ping[6] -W \u003ctimeout\u003e \u003caddress\u003e\n- ping[6] -W \u003ctimeout\u003e -c \u003ccount\u003e \u003caddress\u003e\n- ping[6] -W \u003ctimeout\u003e -c \u003ccount\u003e -i \u003cinterval\u003e \u003caddress\u003e\n\nThose commands could be called from inside a namespace. The needed\nfilter is also added in this patch.\n\nConflicts:\n    etc/neutron/rootwrap.d/debug.filters\n    neutron/tests/common/net_helpers.py\n\nChange-Id: Ie5cbc0dcc76672b26cd2605f08cfd17a30b4c905\nCloses-Bug: #1863006\n(cherry picked from commit cc3b9df4268ac339d0b7316595a4148ba4c69836)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/055484a3ab2fd6433487eac294b1bc3e00be3030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/055484a3ab2fd6433487eac294b1bc3e00be3030"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
