)]}'
{"id":"openstack%2Fneutron~751201","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I75ae14c64db076434ca9897ba9a6d97702e233ad","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1864225","hashtags":[],"change_id":"I75ae14c64db076434ca9897ba9a6d97702e233ad","subject":"subnet create - segment aware auto-addr allocation","status":"ABANDONED","created":"2020-09-11 09:20:28.000000000","updated":"2020-09-21 19:48:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cba4dd3541ec4f8e013598910dc41c4b1bade026","_number":751201,"virtual_id_number":751201,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"date":"2020-09-11 09:29:46.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-11 10:36:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-11 10:15:14.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"date":"2020-09-11 09:20:28.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 09:20:28.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2020-09-11 09:29:46.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-11 10:15:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-11 10:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33cd2d2687bcc31f93657c8ba18e3873e4ed2d32","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-11 09:20:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e62d19fc71a05f300f6ec5a0992c6b620d1346ec","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 09:29:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpcx6c7nhw/1d2c5b1d42344a07ab5e91fafacc9a54/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"a36a6b18504fe7f664c828adb5e1a4f203634130","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 10:15:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/751201/1 : FAILURE in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"33da53abca1ede87ae897d5ec121da8599b98608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:36:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/06640799af53410cbaed4f144e65dadf : FAILURE in 31m 23s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a7b6596ebf0b4098aed2cd1ae4908656 : FAILURE in 30m 28s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/a67af9cd1a9947ba82309667bd44a994 : FAILURE in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3a3cdee7f57470287760f5cfffce10e : SUCCESS in 12m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f8dc5c4433b4b398c7d28ea3b17124c : FAILURE in 21m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b598c2ae0fa84fd19b2f7f881cd66b59 : FAILURE in 27m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/292f1b0c63c549e0ad180d05caba355c : SUCCESS in 7m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b3ca0dbd775b4e408399dd927d92b172 : FAILURE in 21m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/22f46c38453f426ba6d7ac7e952199d8 : SUCCESS in 1h 05m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/701102a1f0cc4d4e8fc1f3a2fc9c3e08 : FAILURE in 28m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d982e59795d24ad8b9dc41f03f987615 : SUCCESS in 39m 05s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/cb2a1a852db74088a7289e815a738330 : FAILURE in 29m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/290f636b77ca46979f933d7fd39e41ac : FAILURE in 25m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c937b8744dff4101863880613462f2b9 : FAILURE in 27m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dc4b6f80c6e40d2a18e93b368526894 : FAILURE in 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4fb723e4d1612132a0eeff34b8ec10b6099f5ccd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-21 19:48:21.000000000","message":"Abandoned\n\nIt seems like there is more dependencies for that patch and I don\u0027t think we really need to backport it.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38b98af63af1ade2358f902bd32811ce36a019ee","revisions":{"38b98af63af1ade2358f902bd32811ce36a019ee":{"kind":"REWORK","_number":1,"created":"2020-09-11 09:20:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/751201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/751201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/751201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/751201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/751201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/751201/1"}}},"commit":{"parents":[{"commit":"45f56356065a962c242e187c5ee6cacfa0bd89f8","subject":"Use dict .get() to avoid a KeyError in the segment plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45f56356065a962c242e187c5ee6cacfa0bd89f8"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-02-28 02:09:05.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-11 09:12:21.000000000","tz":120},"subject":"subnet create - segment aware auto-addr allocation","message":"subnet create - segment aware auto-addr allocation\n\nWhen creating additional subnets with ipv6 auto-addressing\nip allocation was added to existing ports without filtering\non current allocation\u0027s segment.\n\nThis adds fitering to only add auto-address allocation when\nthe new subnet is on the same segment as the ports current\nipam allocations.\n\nConflicts:\n    neutron/tests/unit/extensions/test_segment.py\n\nRelated: rhbz#1803989\nRelated-Bug: #1864225\nRelated-Bug: #1865138\nCloses-Bug: #1864333\nChange-Id: I75ae14c64db076434ca9897ba9a6d97702e233ad\n(cherry picked from commit 3d3dc60408148cf16bc19cccb76b8652f980fa1c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38b98af63af1ade2358f902bd32811ce36a019ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38b98af63af1ade2358f902bd32811ce36a019ee"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
