)]}'
{"id":"openstack%2Fneutron~751202","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I9fb0f05ede42afa1a349635b1936028edf540a1f","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1864225","hashtags":[],"change_id":"I9fb0f05ede42afa1a349635b1936028edf540a1f","subject":"Deny delete last slaac subnet with allocation on segment","status":"ABANDONED","created":"2020-09-11 09:20:28.000000000","updated":"2020-09-21 19:48:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4c54056e52b44d0865938e71686cb64f5115f95","_number":751202,"virtual_id_number":751202,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"date":"2020-09-11 09:30:19.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-11 10:22:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-11 10:16:23.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"date":"2020-09-11 09:20:28.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 09:20:28.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2020-09-11 09:30:19.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-11 10:16:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-11 10:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76927828895b829957f81cdf44674df34e15f10e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-11 09:20:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a20ad16c6e6cac7624fd0da5ae26fd3658c0356e","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 09:30:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpcx6c7nhw/7c5ea1d9c7c743f685759caa90e89e2b/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"e64d14f03bdd74c623ccf049a4876dcac89169c3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-11 10:16:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/751202/1 : FAILURE in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5635bc1ab05a258e224219384755226a80671892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:22:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/9ba7372a5ce74944b32080e475d92b4b : FAILURE in 28m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/e464fadcc25642c2a07a1246ff7852b9 : FAILURE in 32m 17s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/afa3a2db102747ecbc718c203325bd5e : FAILURE in 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/149c3f5edce248b28ad235857148a02e : SUCCESS in 12m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ebc6f39bf1541aea8cae5f9581cf19a : FAILURE in 27m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab326a232ae84504a32fd7c225df79eb : FAILURE in 14m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/915725448e744285a7287018b662609b : SUCCESS in 7m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/15c1a694a2414d9d8afe8aee8e34a12d : FAILURE in 31m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ced7981ab173469fbdedd13966241033 : SUCCESS in 51m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/944352434fb34b00baba69394f8d401f : FAILURE in 32m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/08fe6c1d121b4b9ea1c143badc664d80 : SUCCESS in 43m 55s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/acd926dc32ee46b2bf625e0ecf2265b9 : FAILURE in 25m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/41ff2eb807d8456c830fa110f50e05f3 : FAILURE in 27m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0d2aa4e42da74b3a9c782e6fb48e190f : FAILURE in 28m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd1d66e11ba343ee99ed24a9009cc831 : FAILURE in 27m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b5b7c97bc9b2d6c4e0e67d844bb6f2b2a249d427","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-21 19:48:26.000000000","message":"Abandoned\n\nIt seems like there is more dependencies for that patch and I don\u0027t think we really need to backport it.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15de4d3c287938abbb5772bf89ca670ac27f0d80","revisions":{"15de4d3c287938abbb5772bf89ca670ac27f0d80":{"kind":"REWORK","_number":1,"created":"2020-09-11 09:20:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/02/751202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/751202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/751202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/751202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/751202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/751202/1"}}},"commit":{"parents":[{"commit":"38b98af63af1ade2358f902bd32811ce36a019ee","subject":"subnet create - segment aware auto-addr allocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38b98af63af1ade2358f902bd32811ce36a019ee"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-02-28 21:55:13.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-11 09:20:00.000000000","tz":120},"subject":"Deny delete last slaac subnet with allocation on segment","message":"Deny delete last slaac subnet with allocation on segment\n\nWhen a port has only one IP allocation on auto-allocation\nsubnet which is associated with a segment, do not allow\nthe delete of the subnet. Raise SubnetInUse exception instead.\n\nConflicts:\n    neutron/services/segments/plugin.py\n\nRelated: rhbz#1803989\nRelated-Bug: #1864225\nRelated-Bug: #1864333\nCloses-Bug: #1865138\nChange-Id: I9fb0f05ede42afa1a349635b1936028edf540a1f\n(cherry picked from commit f987486febb9fbe88232bb9139508981b92147f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15de4d3c287938abbb5772bf89ca670ac27f0d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15de4d3c287938abbb5772bf89ca670ac27f0d80"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
