)]}'
{"id":"openstack%2Fneutron~752114","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I74c3a4d3e9f833752571a75f196560cd45529385","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1881297-stable/ussuri","hashtags":[],"change_id":"I74c3a4d3e9f833752571a75f196560cd45529385","subject":"Set process name for agents","status":"ABANDONED","created":"2020-09-15 21:13:18.000000000","updated":"2021-04-14 13:37:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"45e86c7373cfb4e54e30ba8957e240a34f49ba99","_number":752114,"virtual_id_number":752114,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-04 05:31:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-11-03 20:32:20.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2020-11-03 16:03:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-15 21:43:22.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-16 06:53:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-09-21 11:10:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-03 20:32:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-04 05:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 13:37:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"2a2312423d067610d258e038c8af9ad8bac95380","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2020-09-15 21:13:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2a802f553526cc5c927e9674970a8800046fe910","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-15 21:43:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/14/752114/1/check/devstack-networking-ovs-dpdk/c5d8656/ : FAILURE in 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e80d5783f49fbc51ede4de0326df971ed9f82f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 00:09:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/0900326ce9634b2bb2367410e4cafe1c : FAILURE in 33m 19s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/33ca1b91993240b284d6402aa4208bdb : FAILURE in 30m 40s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/9809237de08f400598c1e8164438624c : FAILURE in 35m 21s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/7d5d7c324b9040e6a5fb5bcbeee0bda9 : FAILURE in 27m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/f211f4b731fb42dcab352bb48848a122 : FAILURE in 27m 38s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/7c78d9caba4b48ecaf5f8269f31b65ff : FAILURE in 28m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2b1160446f54ad59652be930e26ed00 : FAILURE in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3afa96eb6b984cbcbf318a91bca9ac77 : FAILURE in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ff7fd8510c4eac946db8278273feaa : SUCCESS in 16m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667b9732e80d4509a1a71efa16bc0c1d : FAILURE in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fba0073b13b438187773c9c99de5347 : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514f37c3087b45e3a82611dcc79d16fc : FAILURE in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9348cbf1a2ec4a54b73129110b7612aa : FAILURE in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/948763ed698648d088f20c880e88251c : SUCCESS in 8m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f2912a9337f4015a6bbde2bf164a667 : FAILURE in 13m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/48115208129a4e8f8a129d8fb5c1a0c8 : FAILURE in 19m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1b2b4c8785243a6a9e3c81d6b0dc9d0 : FAILURE in 19m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d66073f55d714c9b8dd0c828fa88a494 : FAILURE in 32m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c1e8926bfd44dd483c32bd08e8ecc2b : FAILURE in 1h 09m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dba60e99deb64fd284ee1e656d233aaf : FAILURE in 1h 09m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/953a743078414dd1bd8f61eb62bead8c : FAILURE in 23m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d2ff6ecdd654aadbb82217653c34ec1 : FAILURE in 27m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84964047fc6d486eb3a2a3899c2dbf12 : FAILURE in 23m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/339843ef7fc3410984f4a1091170f68f : FAILURE in 29m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d60fecba885346ae88e278ae969ca2b1 : FAILURE in 29m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be8338764e3846dfae74d573f5ab9e83 : FAILURE in 23m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e48a3ce004f542299441874acf2a01c8 : FAILURE in 29m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3178253c7dfe4cb091f96a8dc78f8c11 : FAILURE in 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af6edaeebb724a24be6764ecdf313e13 : FAILURE in 28m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6d3b9cc553ce43eabffb0e3ad82e6241 : FAILURE in 26m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/81c8b9960ed44c92901b04dbd6e11ec4 : FAILURE in 34m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b0062aaab07045d4bec20c15e8b5d919 : FAILURE in 1h 20m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/32b492dfdecb4314954480f910356bfc : FAILURE in 21m 36s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f0e2e03c8f4e4e62a047201e7d2c10db : FAILURE in 1h 29m 45s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0c4c390cffb4a16b0fce261fbe5807f : FAILURE in 33m 04s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0fc0a996000349ebbe339715b36bb061 : FAILURE in 2h 50m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0d5985e1a3f84086b993d57ee8bae17b : FAILURE in 31m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d84c87c9fb77c10860c863a363439cc28ff35a87","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-16 01:40:17.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752114/1 : FAILURE in 4h 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e2d76b48e3f8f41cf5b0698b8e816b0ad96cf83c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-16 06:53:27.000000000","message":"Patch Set 1: Code-Review-1\n\nneutron-lib 2.3.0 (Ussuri) does not have the needed constants","accounts_in_message":[],"_revision_number":1},{"id":"9bfca82368b77584fd25e7b301e358a91f309b47","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-21 11:10:28.000000000","message":"Patch Set 1:\n\nFor a potential backport, we can use patchset 2:\nhttps://review.opendev.org/#/c/735125/2\n\nIt is also a smaller change and does not depend on neutron-lib change, and is backwards-compatible, so it should be good to go","accounts_in_message":[],"_revision_number":1},{"id":"bf7ce5c698d1c8607fb003143e2203566ee6ec62","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-03 16:03:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3d1ae58d4efecc5de5c737a1d2c1bba98801b630","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-03 16:03:19.000000000","message":"Removed Code-Review-1 by Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"f7e2ec90c16482c70788e0df1cb7e8f7dfe763d5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-03 20:32:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752114/2 : FAILURE in 4h 28m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f3f35c148fb4bba91d97f2686a70bd7dab334a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 05:31:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/fd47c547c35d42979f36b8f44404d160 : FAILURE in 30m 25s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/182447836c9647a3a620a9f161ccca69 : FAILURE in 22m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/70462f1b91284a96866d5c85dbabcd6d : FAILURE in 31m 33s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/5dd66810c3004c00988092c22ecd5d02 : FAILURE in 27m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/eca006c7f39c49c08abfefb9033e1fc3 : FAILURE in 22m 15s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/371ccbf8eeb64847a5d7f84ccba37827 : FAILURE in 29m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0e287eae6d6464ea4505f032688368a : FAILURE in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95661065c57a49e795db74d641c8dc0d : FAILURE in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80d2664355f4e4f8b06394dd48b2f51 : SUCCESS in 12m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03a13f7ba064453e8939d82ca9ab5f20 : FAILURE in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d5061d4b0f14e3491d71d39c06f1931 : FAILURE in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee97d6ac8ca74a3bb37c5bf557d20c8d : FAILURE in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48972721938d48b1a95a862bcd0f8c72 : FAILURE in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d2fe1994a149c6b946bb89286dfe5d : SUCCESS in 10m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2430519dd69f40e29a74da5e0ef90299 : FAILURE in 19m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a057b998edff41edae98ee0931c48e81 : FAILURE in 15m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec6637b93b77450fa45c5f1fc5ac3a24 : FAILURE in 15m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a10c1e27070b46a9b4c768c4771d0587 : FAILURE in 26m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9723416298584fb68d64f300ac2107a2 : FAILURE in 1h 06m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/857f93cc1ab54100b382addca529d03b : FAILURE in 1h 16m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b4846c48c6e6400fb338da3a91269a73 : FAILURE in 29m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/93c81d8fa196464da9531bf2dd5f6ed8 : FAILURE in 30m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5e1dcffb5ba543e388d34b593af6866e : FAILURE in 29m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0faa06f52c884902a756d4f6f0c713d5 : FAILURE in 37m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9783f02f0b684fa89bbf3d94631adbad : FAILURE in 35m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5fbd60a78974ba895179819240523de : FAILURE in 25m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5ea8965dde14474995c56bc806d4cec1 : FAILURE in 25m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53c1bbc9cd01400093f475f13ad3da03 : FAILURE in 31m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d82239aa5f954a3285e3a3b27db96374 : FAILURE in 35m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9853ed8aee214c5a91564d7b8b108aa3 : FAILURE in 22m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2f717f09d3e6401fb4b63521fa1a2a9b : FAILURE in 35m 51s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/396bab1ca904448da55fb52230e71724 : FAILURE in 1h 15m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eeceb5798a5c4f75bd300660438f7f50 : FAILURE in 16m 03s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9fa7e18f050f494cb15daca4c78dd92e : FAILURE in 1h 27m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c413fdac2c74733b1e42ce6cfd1d465 : FAILURE in 34m 35s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6e04670f3e2541e6a1e2e5ffaea9f61a : FAILURE in 3h 08m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cabb2eeb3eb14721923d01ece08f11d0 : FAILURE in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45e86c7373cfb4e54e30ba8957e240a34f49ba99","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:37:41.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1da393df40aaaa188fcf0679d25797ea080add8","revisions":{"4575c841bad6089b7642d8095bbf3ecc291ec7f2":{"kind":"REWORK","_number":1,"created":"2020-09-15 21:13:18.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/14/752114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/752114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/752114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/752114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/752114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/752114/1"}}},"commit":{"parents":[{"commit":"cd8fbdc04ec2a0a4ee0570fba9f1cd452669d044","subject":"Remove some unnecessary usages of verify()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd8fbdc04ec2a0a4ee0570fba9f1cd452669d044"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-06-11 12:20:49.000000000","tz":120},"committer":{"name":"David Hill","email":"davidchill@hotmail.com","date":"2020-09-15 21:13:18.000000000","tz":0},"subject":"Set process name for agents","message":"Set process name for agents\n\nNow that we use setproctitle for neutron-server workers (and\nneutron-keepalived-state-change), this has the side effect of changing\nthe process name for agents, impacting some monitoring systems. More\ndetails in launchpad bug.\n\nThis patch fixes it by setting the name with setproctitle to:\nagent name (original process name).\n\nAlso use the newly introduced name constants to replace existing\nhardcoded uses.\n\nChange-Id: I74c3a4d3e9f833752571a75f196560cd45529385\nCloses-Bug: #1881297\n(cherry picked from commit 5ce0595803e72157a82f086b520f454786100872)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4575c841bad6089b7642d8095bbf3ecc291ec7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4575c841bad6089b7642d8095bbf3ecc291ec7f2"}]},"branch":"refs/heads/stable/ussuri"},"a1da393df40aaaa188fcf0679d25797ea080add8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-03 16:03:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/14/752114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/752114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/752114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/752114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/752114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/752114/2"}}},"commit":{"parents":[{"commit":"55e7a11b241d64cac057f4fcd142bf05dab85e56","subject":"Merge \"Local mac direct flow for non-openflow firewall\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e7a11b241d64cac057f4fcd142bf05dab85e56"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-06-11 12:20:49.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-03 16:03:04.000000000","tz":0},"subject":"Set process name for agents","message":"Set process name for agents\n\nNow that we use setproctitle for neutron-server workers (and\nneutron-keepalived-state-change), this has the side effect of changing\nthe process name for agents, impacting some monitoring systems. More\ndetails in launchpad bug.\n\nThis patch fixes it by setting the name with setproctitle to:\nagent name (original process name).\n\nAlso use the newly introduced name constants to replace existing\nhardcoded uses.\n\nChange-Id: I74c3a4d3e9f833752571a75f196560cd45529385\nCloses-Bug: #1881297\n(cherry picked from commit 5ce0595803e72157a82f086b520f454786100872)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1da393df40aaaa188fcf0679d25797ea080add8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1da393df40aaaa188fcf0679d25797ea080add8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
