)]}'
{"id":"openstack%2Fneutron~752248","triplet_id":"openstack%2Fneutron~master~Iadc1671c862f8c01e5761e92b82a04849d4bb411","project":"openstack/neutron","branch":"master","topic":"bug/1892405","hashtags":[],"change_id":"Iadc1671c862f8c01e5761e92b82a04849d4bb411","subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","status":"MERGED","created":"2020-09-16 14:16:24.000000000","updated":"2020-12-17 01:18:12.000000000","submitted":"2020-12-17 01:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":49,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"752248-bug/1892405","meta_rev_id":"1b36f84e38db8ac62adc437981fbd277a2a27d0a","_number":752248,"virtual_id_number":752248,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-17 01:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2020-12-16 22:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":2,"date":"2020-12-16 08:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2020-12-16 22:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-09-17 09:49:59.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-09-17 09:55:36.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-17 18:34:37.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-01 07:53:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-04 06:46:40.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-04 06:54:07.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-04 07:40:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-04 11:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-04 15:27:56.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"},{"updated":"2020-11-09 21:28:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-30 05:50:22.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"3381bc3b871764e2d3903ddfdb521098b0ebd6ea","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-09-16 14:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7cb6e99d02b90e97573dcc5e727634f2590fb08","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 14:23:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-16/752248/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/aa0162e : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e5c21bf3c5c87dd9cd66b4603777f1a10f917d1","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-16 14:47:57.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/752248/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"abed71ffc2186cb2581131459ad5287876ea04d2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-16 15:06:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752248/1 : FAILURE in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d131d302281e68ee9b797d03439f997dfa80900b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-16 15:06:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/752248/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/752248/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9d58b405c366f9e747a7f7c750d04f643bb72918","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-16 15:29:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/48/752248/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d47b95847e12cabf2d294e47d17d099024467f30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-16 16:04:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/752248/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/752248/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c6266c2b188a798e1dc785b87b1daf4c770be0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:38:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8aee977d94794ed3ac3adf055d556bf3 : FAILURE in 10m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc4152ee91f2473c85bfa0be99b40fb7 : FAILURE in 11m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b4508373261a450dac9a6af180ec70cf : SUCCESS in 1h 31m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6600959e29d4510a743a0e4da472b10 : SUCCESS in 1h 13m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a39b9ce6a71343cc88bf77e539c3f431 : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/149bc3907177486e9c4a4032332f382e : SUCCESS in 1h 09m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b5d47410aa4d90b23bc970313d514a : SUCCESS in 33m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0304a3a70957488d80ca845740d424f3 : SUCCESS in 26m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d65691234edd4940ad97e78d92913ad6 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb56df9dee54c219c2f1627337c5919 : SUCCESS in 25m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa1167d83cf84abeb4860360cf648566 : SUCCESS in 27m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee322aca97c447495c1ab012bb7b1e2 : SUCCESS in 17m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/665ca2a9c0e943c2a639d62a46def1e5 : SUCCESS in 47m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb451ed36a4644618a3b91bef6977741 : SUCCESS in 1h 08m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5e4eba772ab54382ab9a8abded7b0acf : SUCCESS in 1h 51m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edc8d55dcbd343899fccd564ebf91c02 : SUCCESS in 1h 13m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1e0c00f8bb9d43448923a407a859c2ac : SUCCESS in 1h 10m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8d7c80cf3fb044be901a7c6060ad9600 : SUCCESS in 1h 09m 24s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f520a224e95d4d608f0f6fdfcfee7955 : SUCCESS in 1h 01m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/289f6d042e534fb98f22b6ed30e9e39d : SUCCESS in 1h 18m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6c3a5f59cf8a4ece872030a9e69e5997 : SUCCESS in 1h 36m 01s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3768594d59004ce3b9a39cebf9705fd9 : SUCCESS in 1h 08m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/550669653ba440baaf0240b95c64fc60 : SUCCESS in 1h 13m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52ca0d329a1f4756bcf83e89867f1e4f : SUCCESS in 1h 59m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64b9fdf5c006400c92a6fdeeb93c00e0 : SUCCESS in 56m 17s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/16a79e4cdf3d43d39f68a725026ba6b8 : SUCCESS in 1h 04m 09s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aa28097e3452433eb4ba6f25f1388e98 : SUCCESS in 1h 15m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/abe1343bdb6c49df8e1eda2256d376f1 : FAILURE in 39m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fc7134ccea27433a91970ff7ec3ca5cd : SUCCESS in 1h 13m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/990e7bf5d8d74215a6da7e92427b5f3e : SUCCESS in 1h 31m 07s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a952e49c66a1424da9fc561e7c1ecd22 : TIMED_OUT in 2h 13m 55s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a2de863c4d274cc38a2a324fc1e591f5 : SUCCESS in 2h 27m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7c782d91a53549968655b5b7fb89e0e4 : FAILURE in 1h 44m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1a18cb9a440740d692895d17648a100d : FAILURE in 2h 55m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f37b999e55b705f693a4af00d0cdf4346e8a4722","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-09-17 01:35:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d985972c8f10c02d72482b134923e8c858824524","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-17 01:42:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-17/752248/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ffb1b6b : FAILURE in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08784e1a374f68f8f967a929a22eba7b9e809c43","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-17 02:06:48.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/752248/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f5a531cdf8fdf460c00e9ce0c8cb784c5788592b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-17 02:54:38.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/48/752248/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"42b65e70bcbad64d4753814a953fb301d770be80","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-17 03:35:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/752248/2 : SUCCESS in 1h 59m 18s","accounts_in_message":[],"_revision_number":2},{"id":"80a230fb10f5c18bdfd5014781b176de8d23dbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 04:42:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/757a3ba8827f41a082e9892821d43720 : SUCCESS in 46m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0880294a875e456dbebaf55f1efaf553 : SUCCESS in 31m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2abba19c4ba444fca57dbb10c4ae37df : SUCCESS in 1h 11m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4543becd75e346768ab15ef8b3d5178e : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cc6bacca5fe4325abaa7adc597735f0 : SUCCESS in 56m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8a7a5458c4fe48abb9fc29d85d21fde1 : SUCCESS in 1h 01m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acf0fbbd625c44b988e0f3335597ffd8 : SUCCESS in 31m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8944d162ae141f7af9b9282be9822f9 : SUCCESS in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba63f7429b224edbb210190d14abbae2 : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a82c996d426b40398b286abc01280c3e : SUCCESS in 24m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f446d33ca5b47cca458c091da53f728 : SUCCESS in 28m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33561145fa4d49ec996828d1314b2c34 : SUCCESS in 18m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba38cabfe9c84b46bc8e725ab7e5cde2 : SUCCESS in 41m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2df2d3d8bdc84b8a860b918a15134261 : SUCCESS in 1h 05m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f81653ec8bc247ff90b38c149bdb880d : SUCCESS in 1h 44m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77348fb98d2d49f8a1d90cda0293e3a3 : SUCCESS in 59m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0534721283564b5abdf2b0f8e77694e4 : SUCCESS in 59m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/125dc720f8484b12ace8399f035ff5e1 : SUCCESS in 1h 04m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/dffa8c01b0304374b5b068e9a687f267 : SUCCESS in 1h 09m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4d9231d6c9464e21aae5f1cb7779d64e : SUCCESS in 1h 29m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1f26f0d6c06d4b42aa532aa8ab12ec5b : FAILURE in 1h 17m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e4cbc15233a484187791fa1624f5410 : SUCCESS in 1h 06m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/63d44c5ce8524f8ea4d76aa03bd60246 : SUCCESS in 1h 04m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/785fc6c76d7546a4b175dc5b54295e6d : SUCCESS in 2h 02m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcfda09fa5084f5290a90cf8f36c423f : SUCCESS in 54m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bcff8472c5834d6d9caf0f34a3a48483 : SUCCESS in 1h 05m 41s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b8a62f931dc5432e80cb742f25c86546 : SUCCESS in 1h 03m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/390637616de24f3da333dda306b51fc0 : FAILURE in 32m 41s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ffd5524c906345f0831e249deee19607 : SUCCESS in 1h 06m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7f0329f77c754cd6bde6a4601a11ab9c : SUCCESS in 1h 33m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b425f641f0c4a99983e09db4271c264 : SUCCESS in 1h 48m 35s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fd5d141d60ad4d8da3482eca7691cbd8 : SUCCESS in 1h 57m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5ea17b1be2e94f0494eddd0b273013f1 : FAILURE in 1h 41m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4126750e9b614c42bd06cea4febf9154 : TIMED_OUT in 3h 04m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ead5ec2d57730e6798dc951c5e8b6f22ac52576f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-17 05:35:00.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/752248/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/752248/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"53a5a0d613d872979852ca34ba12e7bea3e38756","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-09-17 09:48:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3bbe176c801586b4c4c514e1e512f304bc13e4d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-09-17 09:55:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2020-09-17/752248/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9025045 : FAILURE in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6bec8875f9a4c236df1f9e363f8386678eb8a002","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-17 10:21:19.000000000","message":"Patch Set 3:\n\n* sriov-test http://52.27.155.124/sriov/752248/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"0ef7e1d9a9c72b18c48b364f3cff2d1de740d789","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-17 11:02:44.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/48/752248/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"674d2dde9b1eede85134f1c5615e2ab2e4455792","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-17 11:40:52.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752248/3 : FAILURE in 1h 51m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ca8c7d2d12e4f3502e8d866e7843099c2c6c69f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:24:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f9a72ead6bd43c8af8a299059ae3466 : SUCCESS in 56m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d1ec798e33144218d02b807ac23da05 : SUCCESS in 30m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/59239e090f0b4bceb0d476e65eb9549c : SUCCESS in 52m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/73fd3f386f8a417299443fcd2c0b6e16 : SUCCESS in 1h 10m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/90949a0fc44643daa86ae9d37fa4db9c : SUCCESS in 1h 08m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e1242da949d84c4c994cedf771b72849 : SUCCESS in 1h 11m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23d42e6e69bb4428a448c2f6f24d3da4 : SUCCESS in 32m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73c570a13945408a9a4a4272710d379d : SUCCESS in 25m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dde1d657a01487f882e4dfcaaab7857 : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5feea21d61934553ac1645825afaa86f : SUCCESS in 25m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01e09a0c53ac4ac6a0c9a685435de260 : SUCCESS in 30m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30b535dee1940038b805b4f4807ee98 : SUCCESS in 21m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19b8a88559494c8a832a3f1cb35419ba : SUCCESS in 47m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/21733f24e4d4432ab03b436bf1e4ff23 : SUCCESS in 1h 10m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d902987084744b5bbfe12378cf526379 : SUCCESS in 1h 45m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/67d037462edc406f9dd022cfe12fcdf7 : SUCCESS in 1h 06m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a178b4a458ae4deaa8ddfd7e9c2ef304 : SUCCESS in 1h 13m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/eda52c9d972442769087e477c9263971 : SUCCESS in 1h 12m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/93a44988c5da4a9e9b1e104939320eb6 : SUCCESS in 1h 17m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80fffb657be44b8ba94397673f665392 : SUCCESS in 1h 16m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/24777ca005344da0be6176319dc1ef92 : FAILURE in 1h 24m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86e6803ab8e849ef9dd223c9f1bdad67 : SUCCESS in 1h 07m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/904c8c42964f40a8ab23f32756e54fcd : SUCCESS in 1h 13m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bef718dc44a146e29f1db16cd8f1cdf5 : SUCCESS in 2h 01m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd48cd162c184e67b732b21c77b58d73 : SUCCESS in 1h 00m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a81f068bfda046418ac92f65db7ceb83 : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5511dc5142aa43f790178852a151c7bf : SUCCESS in 1h 13m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/baee4a3e21604ddabdc43d9e3918809d : FAILURE in 38m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cbe1494bd55849f5a544685778c3e03c : SUCCESS in 1h 17m 50s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2686f15b743146c19a514c38d44dd19c : SUCCESS in 1h 46m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/34e3bfbd81a44ed6a3d952fe2fce8aa3 : SUCCESS in 1h 57m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d14dd8e241d245d8879594a2e5f0d65c : SUCCESS in 2h 19m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e1c80fadc673461a8b9e4bbf85118e1d : FAILURE in 1h 51m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/910f1ef41fe24e68a1340cbfee340e8a : FAILURE in 2h 07m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"81bacb4d563aa0b7c11e70c43c0bd744544b24a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-17 18:34:37.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/48/752248/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/48/752248/3/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c6975a4f7941027396e5eec6b3124297e6c4050a","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2020-09-18 11:16:51.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nCouple of inline comments + I think this could do with some unit tests.","accounts_in_message":[],"_revision_number":3},{"id":"86eb83cf78cf11ea307187a8f3300c4a2527d2fd","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-09-22 11:21:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fd4822624cfac047687c5ccda8080af66606d9c8","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-09-22 11:23:40.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n\u003e (2 comments)\n \u003e \n \u003e Couple of inline comments + I think this could do with some unit\n \u003e tests.\n\nInline comments are fixed in PS4. Updated the PS with the unit tests as well.","accounts_in_message":[],"_revision_number":3},{"id":"263a2056030ee4902048421ffd5eb4edc4af2749","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-22 13:14:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/752248/4 : SUCCESS in 1h 51m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a2a9dc66a89ea1bf71a166e39eca085711b6d8cb","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-22 13:38:03.000000000","message":"Patch Set 4:\n\n* sriov-test http://52.27.155.124/sriov/752248/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"bb75683fac7b24adc2580d1bb7c76deac4ea9fac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-22 14:25:10.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/48/752248/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"c913b6f032b3c8f4e9a8dc7b049a6344673434a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 20:50:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/53cc0829e5a4480db6fc130309df311b : SUCCESS in 53m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3821521a21604be19d04ec71ae0d74a4 : SUCCESS in 38m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5b4ba915e9db46a58f77d0a34ccf7503 : SUCCESS in 1h 07m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c23d05321c5b4abea87afbeeeda1d519 : SUCCESS in 1h 02m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9da57181ab0452898d985b36e524527 : SUCCESS in 1h 01m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/55e4eff6c5a94ce7b4e347268a718277 : FAILURE in 19m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff96a25a6df4430aa4ddce59dcab7ae : SUCCESS in 47m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a2f8b283b7042e99738f86d5455fdd0 : SUCCESS in 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2696726bc8aa495fb5370bf94a80b1e0 : SUCCESS in 12m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48044c204dd54b039de4a2c505d3825d : SUCCESS in 26m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38157caf3b0f462ab962e1e5974e022b : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40cdc3a7d4f4bffa0919a30587e4d92 : SUCCESS in 16m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/05f21dceaf3845d8b49f2704bc6ac5fa : SUCCESS in 51m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12672ef2192f4ae7a49f9ee5a2794d36 : TIMED_OUT in 2h 14m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cb6d86a9a0744113b1457910f09c8980 : SUCCESS in 2h 30m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d82b3b65ff1940bf864c234421a333ae : SUCCESS in 1h 05m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ed26b8456d614951896a847e4c46f8d4 : SUCCESS in 1h 04m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6a9c7996e5e54406b83001b16cc3bafd : SUCCESS in 1h 07m 14s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9b5ab1367e342d793be24090110e846 : SUCCESS in 1h 03m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5a4db1b0d2514b4aaeb55f1bfa628296 : SUCCESS in 1h 28m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8a3033aab2942c9a160fe4ae4ffe442 : FAILURE in 1h 16m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f366c384eb740aa96ba4b0e0c2eba70 : SUCCESS in 1h 03m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/732f75170cdf4891a10c8277be0f396c : SUCCESS in 1h 02m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f9e2e2c97e347bda32e9a17e59f25ce : SUCCESS in 1h 53m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c135febcdd6447eb185c0ca4de1574e : SUCCESS in 56m 38s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b25208979b19489d8541b0a9d6b526ef : SUCCESS in 1h 07m 38s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/627687373cd542d3b64a714338d18763 : FAILURE in 17m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/cdbd5989d78c47178ed09d6dc505fd56 : FAILURE in 11m 03s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3a3ed7b568514bccb131608f5cc28d02 : SUCCESS in 1h 23m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e819ffc7740b41609a816daf44be4f0c : SUCCESS in 1h 45m 36s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a4a88f8b9f14c989d33b117cf12d455 : SUCCESS in 1h 35m 41s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0ed86a193f3e44faba7da9860ead6c85 : SUCCESS in 2h 33m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d8f131b35ae249e889b3ae9eb483ad53 : FAILURE in 21m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6e8e81f057954b6b9450a8c64b93c63e : TIMED_OUT in 3h 04m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8b1757113c45e3dcdc1191aa7afa1c7ef47da8e","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-09-30 12:48:18.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9c978b1b2311b24d1390667a5a8f9e0015769826","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-09-30 14:06:15.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2020/Arista_Tempest_Testrefs/changes/48/752248/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"c995911710645cbacffa725fcdf6561afd6dccc7","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-09-30 14:10:35.000000000","message":"Patch Set 5:\n\n* sriov-test http://52.27.155.124/sriov/752248/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"556e137044d0d9cd7b94e5baa8996febd5980626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 17:38:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7341f703fe7c41fca46b5ff953955dc1 : SUCCESS in 1h 01m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2bedb00bc7144f73a5944d8b35bbbb74 : SUCCESS in 40m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/806e91238b0d4a41960ac415a63b4f41 : SUCCESS in 1h 48m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4e70501591a94e79a5cbff6f0bb70d16 : SUCCESS in 2h 48m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf41cd5a63274a43946cdb40d00c2a70 : SUCCESS in 1h 33m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d3bfb87d71a1446da575521a592b5149 : SUCCESS in 1h 24m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d28680e15a441fb83ac039d6e61cc5b : SUCCESS in 34m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a08d4043315941fd8ab1450f83c94ca2 : SUCCESS in 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc71090e5a744e3a38610a67273ed14 : SUCCESS in 10m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a00dd91b554462487b38bd30c860cb6 : SUCCESS in 29m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87e5783f05cc4f8790030066d68468fa : SUCCESS in 25m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7083877a247546a3b39a675f6bd95ed9 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c293b1b6ae5f470ea596fad84965255b : FAILURE in 1h 11m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4817b990d255403780feb131e9a5c7ff : SUCCESS in 1h 14m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/eff7bf7e88d541b8b013395662d94b09 : SUCCESS in 1h 51m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f73c03d41a94eea99cfff55f3ce6610 : SUCCESS in 1h 09m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/759a7df4627544efbcd899896706b92a : SUCCESS in 1h 18m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9bc2d7469185404e995b4ac012ae16f2 : SUCCESS in 1h 09m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a1718431fc541abadcd7e76e28efe0f : SUCCESS in 1h 07m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ffcbda581610470a96122c07603a6133 : SUCCESS in 1h 45m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c39b98a33e71447b8c41b00a2f67bab6 : FAILURE in 1h 25m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a8df36257e94a2695ffa8444a065eff : SUCCESS in 1h 14m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d1775efa4564a6b96550efd9a653a92 : SUCCESS in 1h 15m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c85daa6edf7e4d2eba5f9f5dd8ac9004 : SUCCESS in 2h 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb34a6d2f6964d1a9a58a5ace3623bf8 : SUCCESS in 1h 03m 07s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9c9ba25b8a5c4a739e9be968a8b7516b : SUCCESS in 1h 11m 33s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/031046d7623746aca2e0641689e96527 : SUCCESS in 1h 05m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/91caec6f05da48a2b13319d4e37f59d9 : FAILURE in 38m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b4a09be8cd7c4bef8db302bcdb5cedab : SUCCESS in 1h 16m 26s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f67a3ad7375476abdb4c332cc63d521 : SUCCESS in 1h 33m 28s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49471f41ffee412dafe786b11e689f78 : SUCCESS in 1h 39m 29s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/46d9bc7681d049c7a0f8e95595c9add7 : SUCCESS in 2h 25m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ab7de7b1d5e644d7aa8aa44b956e6204 : FAILURE in 41m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f06bfb4c236a4a358ff139e6fcf4e3db : FAILURE in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f002e27fe4e1f06153709ea379885942ac644d52","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-10-01 07:09:32.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5eaede17265f8ac8ed66b54cd286db8302177862","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-01 07:41:25.000000000","message":"Patch Set 6:\n\n* sriov-test http://52.27.155.124/sriov/752248/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"9384a22b2a925292b8602250dfebb105e3d50cba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-01 07:53:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f8dc44be8a2ab636c29b2d0c40f7a6b3385898d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:54:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de1655182c5345368c60d0eb49cbec5d : SUCCESS in 58m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f43d4f5442734fc5956bd125e00a8678 : SUCCESS in 35m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/14626165a9544c0bb58122f5834dd28b : SUCCESS in 1h 33m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9b3f41a1437745388406c88f9662cfb3 : SUCCESS in 1h 07m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d2653898eb04fc89dc2b0b109a96f00 : FAILURE in 2h 12m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/843eee8867134b529f596f3b7ad53451 : SUCCESS in 1h 06m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48abd5ce96a848c284e7a75ac483acb3 : SUCCESS in 27m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2063f818d904724a1bb702612fc5fab : SUCCESS in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb535909f084e8bb9b931dcd1d74521 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95fb265f2554e81bca8195539cb69ac : SUCCESS in 25m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6a0ce144a8b4ba1aad7fd98e3ff32fd : SUCCESS in 15m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6dd873dcff40a58106832e08a4f201 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d0a36995cbe498191eacf9c5a0d544b : SUCCESS in 1h 07m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9652b00204614450b5b28f82afa5da64 : SUCCESS in 1h 17m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fc7017261af44b5ba6fe89a8e0b492db : SUCCESS in 1h 53m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d616fd9df5a4ac88558543f9ba3c170 : SUCCESS in 1h 10m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d39d538bb87944959005a14ffe424ec8 : SUCCESS in 1h 07m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ae70ef1767fb4535955ab23f305ba69f : SUCCESS in 1h 13m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7afab48603d34ec3a647263060e3dd92 : SUCCESS in 1h 17m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7fc50f43f59e418fb02c935e7e113cde : SUCCESS in 1h 33m 40s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ff70680e5b841509fcc7d31e8117e68 : SUCCESS in 1h 31m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66b72a1a079242d38378bb424d0d3942 : SUCCESS in 53m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0b2cbb5346c74243932e4da010b31552 : SUCCESS in 1h 08m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20a135f5ce55461a96f948046856c354 : SUCCESS in 2h 12m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e796c1ea5974451b94bcf9aea32b5027 : SUCCESS in 1h 14m 04s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/97f388e4307a42f49718c70095cabdd2 : SUCCESS in 55m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de066fbf565741adb90d458f800ea2ec : SUCCESS in 1h 19m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/eca63759c3544094ae92f57ea07d42d5 : FAILURE in 42m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/58f351abe29a43bc96aac9d232896477 : SUCCESS in 1h 17m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2d56fe0b45d9493fbe3311e25cfb4134 : SUCCESS in 1h 43m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efb48f304a894f24bfd0aeaa8ed343a6 : SUCCESS in 1h 55m 24s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/626d2370e31b4562a22d29026b844b0d : SUCCESS in 2h 41m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c74737815d9844259cbdd92fd1ee8306 : TIMED_OUT in 3h 03m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b78d18a881df4571b4d96f7b7e189ac3 : FAILURE in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"edd5371386f013b53a9a0fb662b36542ec92dad0","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-10-01 12:15:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e6a15748f12e0c5abecd51e9d1885ac8282d79ab","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-10-01 12:19:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"981940998aa74472f13d9fb1c16144788850d2c0","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-10-01 12:20:33.000000000","message":"Patch Set 7: Workflow-1\n\nNeed to add unit test case for the issue identified by Rodolfo","accounts_in_message":[],"_revision_number":7},{"id":"6e5bd15bf51fd97e04a2cfffc14201770fa8fcde","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-01 13:02:36.000000000","message":"Patch Set 7:\n\n* sriov-test http://52.27.155.124/sriov/752248/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"7e3a68dfe426c5273c3bb25a8be8ea8a08644dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 15:46:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f5f37b4175e4f4c8fdfc98dc77c29a7 : SUCCESS in 54m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6130c3d90be94324840841b29838cbba : TIMED_OUT in 1h 03m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5f319a74743547bc8ff4ad9c7327740d : FAILURE in 1h 27m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/75f298ac1669453d8f5eb999f7d012ad : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e04f2289596345a1b1689cbab8963241 : SUCCESS in 1h 15m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7dd279cc32a54365bfa2c23a2421dcb6 : SUCCESS in 1h 19m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85bac96d788f4350863ff5dde4db45bb : SUCCESS in 45m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7e873d8b6c84a4c954e031a6b302183 : SUCCESS in 26m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ab2aab6de54844b9dc03a8d674c299 : SUCCESS in 14m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f2057e007245fa969f946e7bb7bc4f : SUCCESS in 24m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deabe132b73e4b0586c7b728e1eee920 : SUCCESS in 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee273e0816741a38edc82af978903ba : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60453f1a67fb4feeb6d7859256e1ace4 : SUCCESS in 1h 09m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b59555a66504de3b5a74d4254eb9d04 : SUCCESS in 1h 09m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/36c433cae2c245de9ed09b9fcfd8bcf0 : SUCCESS in 1h 43m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3399e54bb3d84ff4977b6545e946f1c5 : SUCCESS in 1h 05m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72a166d763384d62b910fb845ecbda63 : SUCCESS in 1h 03m 59s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0d68b54fd0649e882116d56140095c1 : SUCCESS in 1h 10m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b2e403ddfc6465b9d10f3a769efc059 : SUCCESS in 1h 16m 30s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ca219516dae0449a972cd9dfcfeaba2e : SUCCESS in 1h 30m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da03ef5b336e46acb2717e7cc6a172e8 : FAILURE in 1h 24m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d361c89a379a4e3c901fcfe64830c994 : SUCCESS in 1h 27m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/734a71e58c1a4457a58ffaf15b760960 : SUCCESS in 1h 13m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f62f585a620f402b8c40019d7279bfc2 : SUCCESS in 2h 03m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e35bd8d1f05c4ed8a94ae7c23b2e47cc : SUCCESS in 1h 03m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8345dcb12e0446e0ae5069675414eabc : SUCCESS in 1h 08m 51s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8781785e09ef4f8d86dd70c78616cc47 : SUCCESS in 1h 10m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/aa86e9e537e04ad980002fbaebc8d6fc : FAILURE in 36m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fb44a0e2feed44e79aea5bb056111666 : SUCCESS in 1h 20m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec30a89a08054ee3a538401bffd6f90e : SUCCESS in 1h 46m 45s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ffe914e6defe49b2b357acc701634f0a : SUCCESS in 1h 47m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a73d92322aaa49a390fa0e9818a24331 : SUCCESS in 2h 27m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/42aa2cac07554d9a8f21c362d9dfe09b : FAILURE in 35m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ea46cdef6f9d4b089b01b3b453634732 : FAILURE in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7535bbf315ce794778f38e7e978e24829a9200c7","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-10-05 04:02:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5a2360ac0bf562b14e5157c4281d3f6c0e90283a","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-05 04:36:10.000000000","message":"Patch Set 8:\n\n* sriov-test http://52.27.155.124/sriov/752248/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"c5cbf1dcc758efa55642cd28e05654bc67f7c976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 07:11:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6a52a966b9064253af0a364063cb3001 : SUCCESS in 1h 01m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b8fd1b7cb974e369ebbf888a24fdf72 : TIMED_OUT in 1h 04m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5e2bf819a9994b63a3e7cf6ad7b36b79 : SUCCESS in 2h 20m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/17a1cbcb5a8f42ec853537f44b99bfec : SUCCESS in 1h 26m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb70bba587364f97bfc63990ea6779f3 : SUCCESS in 1h 12m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/83b8151a11714c86b302a403ffa2293f : SUCCESS in 1h 14m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53dcc27af6cb4585b8e3fed09b06afc6 : SUCCESS in 34m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82b808a48220426092f21c279d564f33 : SUCCESS in 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe24159399d4bdc936b1ef2875085c3 : SUCCESS in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1a31986d7bb4e838d01f78c739f9fa9 : SUCCESS in 28m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51c9d20e06264769933dd2cde8bbebaa : SUCCESS in 25m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a14e205ce84584900d853a50c9b9bc : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26c5d859ca0e4796bf2b5002625201b9 : SUCCESS in 1h 06m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee3ca68e6f934dfca5c404c8ea8c94db : SUCCESS in 1h 14m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f08dde5b0d145faac1cf87900700e5a : SUCCESS in 1h 19m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a1dc50388cc4b5597d04539036f7535 : SUCCESS in 52m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/797f3e5720084a27b0a9754de72e49dd : SUCCESS in 1h 07m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6088ad04eb1342c4a8988e4351fca57d : SUCCESS in 1h 18m 26s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4831ce6825343fc97f0e4ab6c5b850c : SUCCESS in 1h 01m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6bccb28814d6430ab299999a98bcc6e6 : SUCCESS in 1h 21m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65c3ba202c9c4db59d992b270b78a5f3 : FAILURE in 56m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d67e0d6aa2f44e0b86ec527fca3aeeb : SUCCESS in 1h 18m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cfc8b061adf44caebcf5063f427d6a65 : SUCCESS in 1h 14m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb58e045287a4d84bbfe2947a0cd51ef : SUCCESS in 2h 10m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d74ca0ddf21a40adae00206734a2c39d : SUCCESS in 42m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/94522b25928f43c9b0ffa6e6445c8927 : SUCCESS in 1h 00m 40s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3f66782292444e9fb64d9c2f01e71690 : SUCCESS in 1h 04m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/30b28cf6c6af4553a77e8e6f210e004b : FAILURE in 40m 15s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a38b202044414452b6b413fca40750fc : SUCCESS in 1h 12m 47s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/58c0d2a03f8a4cc69107f1153edf3295 : SUCCESS in 1h 15m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5fadf27c5c4481ab831c541ac014d89 : SUCCESS in 1h 45m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/38b5051c8dae470d87f572be27797d2c : SUCCESS in 2h 22m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/356861faf8564db6b08cc77a331a372e : TIMED_OUT in 3h 03m 32s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cad407501ab64390949bbe0949fc0a32 : FAILURE in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4ce0391273583916a329e793516a7d759ee566b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-06 20:17:28.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/48/752248/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"8727c9450f4fcd609c3868372ac15a42662ded52","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-07 01:14:59.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/48/752248/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"62b75d3d6f45d0822380a92bd384b055a2b10992","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-07 14:27:10.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/48/752248/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"93a126cf4751a3b934708bbc33897fe5c7af69a2","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2020-10-24 15:01:13.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)\n\nlgtm, once comment though","accounts_in_message":[],"_revision_number":8},{"id":"e57bd18abb19e5488ef85b50f7bb0f7957ddac62","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2020-10-28 12:33:48.000000000","message":"Patch Set 8: -Code-Review\n\nHemanth, I\u0027ve had several more goes at trying to reproduce this bug with Stein packages (14.3.1) and am no longer able to reproduce it. I\u0027m not actually sure why this is since I can\u0027t see anything that has landed since 14.0.1 (where we first hit the issue) but based on that I\u0027m going to remove my +1 for now.","accounts_in_message":[],"_revision_number":8},{"id":"19d8f1d29858b76044b330aba0a966fcc4ca8a09","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-03 07:15:20.000000000","message":"Patch Set 8:\n\n\u003e Hemanth, I\u0027ve had several more goes at trying to reproduce this bug\n \u003e with Stein packages (14.3.1) and am no longer able to reproduce it.\n \u003e I\u0027m not actually sure why this is since I can\u0027t see anything that\n \u003e has landed since 14.0.1 (where we first hit the issue) but based on\n \u003e that I\u0027m going to remove my +1 for now.\n\nEd, I am able to reproduce the problem with Stein 14.3.1 packages. There is minor change in the reproducer script which is updated in the LP bug.","accounts_in_message":[],"_revision_number":8},{"id":"ec370d18e670fecee3cd9122d2a7ecebb49b3361","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-04 06:03:41.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fb75ed7a10ffd0f00a73df558e5ee1f27883aa96","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-04 06:05:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a6d4a4bac87a40767015bca74982da60092e2d90","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-04 06:46:40.000000000","message":"Patch Set 9:\n\n* sriov-test http://52.27.155.124/sriov/752248/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"7a5bf7895bbc3e15c99b21bf9d8fa909032e95ef","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-04 06:54:07.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752248/9 : FAILURE in 49m 50s","accounts_in_message":[],"_revision_number":9},{"id":"502445c4368d0172ce631a6ab2a76cca67179c64","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-04 07:40:43.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/48/752248/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"c92800a8838c67313ff7ab68a7a6f3884669dca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:07:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b8bef0442d445e2bb8d0f340618cee2 : SUCCESS in 50m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/46e41dc3c62e4f3394eb3eba027f8db7 : FAILURE in 37m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/90edaa724228459795e01bb14a914284 : SUCCESS in 1h 42m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3aa04338690b450da88e6f8d0a489b38 : SUCCESS in 1h 37m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a1fac2ca0c724385b42fc36144e95b90 : SUCCESS in 1h 27m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7e00570947f64587937dc2c90ab1f97b : SUCCESS in 1h 20m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa738969731a40ada428ce0f1e351808 : SUCCESS in 34m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb6bca77db049899fd41eecafaeecf2 : SUCCESS in 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5182bba9ab84d538b48ad44046c8193 : SUCCESS in 10m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa4a6561f41e400fbc09a1677965375d : SUCCESS in 25m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bee34bc0ae6346a0964582e29385fe4b : SUCCESS in 23m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5860fcaabcd2492381c07e95a4b3d517 : SUCCESS in 9m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1872856ddd9e44cbb4d3325fc723567a : SUCCESS in 1h 08m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e14449842b5e4c76af964c450e71cb42 : SUCCESS in 1h 08m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d4105651fe0b4a0c970f4a2eb66a3546 : SUCCESS in 1h 52m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd85de73f7d946509b87b00dcbdb383a : SUCCESS in 1h 05m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7df7fb7610594547af4f6c162385c0fe : SUCCESS in 1h 10m 02s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/53cc7de5098e434dac562e9c3a1986df : FAILURE in 1h 00m 34s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f0cb27daa6604db7ae12078d9a85b2bf : SUCCESS in 1h 14m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/144d4a96db614b8287b19496a375c3be : SUCCESS in 1h 14m 58s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4214b065c5f6450a854af64d01817755 : SUCCESS in 1h 31m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/27dc3e2b86d446f2a5dbef85bb1d6a42 : FAILURE in 1h 25m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b193b576ba1b4fcc9faeda817b6e9ce4 : SUCCESS in 1h 03m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8873f6ab2114491a95e5d0fcbde0c43a : SUCCESS in 1h 10m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c40e619f682400aab85fef200452d4b : SUCCESS in 1h 58m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c2eb20bf7104752affe8a6888c53115 : SUCCESS in 1h 07m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/cf8d304db3be45bcaa8c3793b98fb352 : SUCCESS in 1h 11m 00s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cc9d0a03bdad448988201bf1350fe99f : SUCCESS in 1h 16m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8ff768b840ae4b0783370380dd0b4f38 : FAILURE in 40m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/07ed420694f94b9ea2b2dcec7aa699e0 : SUCCESS in 1h 21m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b66b8e63bf8a4dbdbe9c914ce5967b14 : FAILURE in 49m 52s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ee2667c2500142efa66c0eec39d7cad8 : SUCCESS in 1h 50m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bdbc25e1d0954a46868addef65255568 : FAILURE in 52m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7d8bb34bf93a4c8c80bcf4891b5c2972 : TIMED_OUT in 3h 03m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/db919efec40f459b8552d7916c1e919e : TIMED_OUT in 3h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c970d80eeea821585a059f6887e1555f31072f08","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2020-11-04 13:02:13.000000000","message":"Patch Set 9:\n\nthanks to hemanth\u0027s fix of the repro script I am now once again able to consistently reproduce the problem - https://bugs.launchpad.net/neutron/+bug/1892405/comments/12 - i was unable to do so before because I was removing n3 from r1 as opposed to n1 from r3. I am in the process if building a stein package with the latest patch and will +1 once I confirm it fixes the problem.","accounts_in_message":[],"_revision_number":9},{"id":"2405cecef1c2d349141dbd1687d87d3f81d1d97f","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2020-11-04 15:27:56.000000000","message":"Patch Set 9: Code-Review+1\n\nConfirming that I have deployed with this patch applied and it did resolve the problem i.e. I can still ping from n1 to n2 after n1 has been removed from r3 (see repro in lp).","accounts_in_message":[],"_revision_number":9},{"id":"e39ef8f44963e55c0569ea555e5cb16ed6b7de35","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-09 21:28:19.000000000","message":"Patch Set 9: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":9},{"id":"468d3c2a57a47f50f27fadce633f802e902c85a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-24 06:19:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"79f777872c2bd287c245298a521a2ce42d56afcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-24 06:30:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c3d45630e5a6c83de86ef59233ae2e9d3333f048","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-24 06:38:21.000000000","message":"Patch Set 11:\n\n(10 comments)\n\nComments in Patch set 9 are resolved in PS11 except for one which needs clarification from the reviewer","accounts_in_message":[],"_revision_number":11},{"id":"7a55ffa10ffd0ce72f0a21d9597ebe24102ae1f3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 07:47:27.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/48/752248/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c0e51cb1b6dc88b34aed2e05af0b08fee298f88c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 08:00:41.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/48/752248/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"c66c3f840c17c00d3b5e8b9e184da2b4d8d02cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:44:14.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/481c49af88dd41b1b541a59c61f55b8a : SUCCESS in 49m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d04bbeabc0943a6a1312212ffd5def5 : SUCCESS in 35m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/84753066b2944027890a756bcef131a5 : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a14bc9ae005f4f349b20e8e52507fe24 : SUCCESS in 1h 23m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c3f281a751149b1b59744f1f346c4fc : SUCCESS in 1h 34m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ac71cf00e0b2461d8345009c81492f0b : SUCCESS in 1h 26m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/763dab6af7514af1bf115e935f627d1f : SUCCESS in 40m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34ea60d9d82b4ff39b6d7fa798892221 : SUCCESS in 23m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c63447dfa2b45618388f5bb270c340f : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df49b74999454e0885c09d9535a64ace : SUCCESS in 28m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56719b0fe48549df943007e938f68f09 : SUCCESS in 22m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e293ab836f4a4a8cc29464f47cde0f : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7365eed33a34df5b3a243c05cc648a1 : SUCCESS in 1h 04m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/563043614d1d47b19bef1d3942271855 : SUCCESS in 1h 06m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bf8a06f3ba84429698fe5d5a6cca5141 : SUCCESS in 1h 42m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2ea894b9000444b9e512d9a0bf4e964 : SUCCESS in 58m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7001799613b0457c8ed09b643879f87d : FAILURE in 1h 03m 12s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/368e11ac6a0c4f3e952188643456b95e : FAILURE in 59m 34s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/380299a14f2e4132b66f07381d0108dd : SUCCESS in 1h 16m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c34464678084e86b1e4925f9dcf6d24 : SUCCESS in 1h 27m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cdf12e17cf01428f99105dda6cb6e38a : FAILURE in 1h 30m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/91414782fed4404b9a3b7e9bd2c41040 : SUCCESS in 2h 43m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/936b184533db4e3c9cebcb119510f922 : TIMED_OUT in 3h 03m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05a8ff50eae94a5e91a53677c92f731f : SUCCESS in 1h 00m 21s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b82ff65bc6354297a69ac5293f44270e : SUCCESS in 1h 07m 57s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3ac2d0967254bb8b7be016a20f4ef45 : SUCCESS in 1h 15m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/16bbb2ba579848c897e93cd791158a26 : SUCCESS in 45m 46s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/22a1f48cf8ec419788d3898b94320681 : FAILURE in 1h 33m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf72ee6af5a74a67baeb6cef74e4065e : SUCCESS in 1h 40m 36s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/647fa6389f7c45e8b6bc7bb24d7b6947 : SUCCESS in 3h 21m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/849cf153e6514a5b8b8666b9e3519660 : SUCCESS in 1h 29m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11206d6d6eda457f9069ad1d01922e3d : SUCCESS in 1h 35m 24s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f08e7d493c824f6f9fa00864dca5509a : FAILURE in 2h 13m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7ac9f50ec3044fb1a4800221bac19ceb : TIMED_OUT in 3h 03m 08s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a86ed886e6344438945973c8a81683ab : FAILURE in 3h 00m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e9bff0e4191a0f8f3cb08a4a976a311c6f46ec08","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-24 11:12:24.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ddad4b3bb8cac4f325ab100e05b2d5348cfafca1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-11-24 11:28:53.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6042334135e1584574112dadd395ce59dbd46529","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-24 12:31:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cf623f1725edc5c19b614855b0d96edab98b577d","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-11-24 12:48:34.000000000","message":"Patch Set 12:\n\n(3 comments)\n\nliuyulong, thanks for the review. Please see my inline comments for the commit message.","accounts_in_message":[],"_revision_number":12},{"id":"bfce5e968a30986890e43e2a9f1335c5cf640848","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 16:48:31.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/48/752248/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"94de322cd65370142269be5e4ea9e36345d1acdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 18:40:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b649b54e34124a67b9c4a6d48c64f5ef : SUCCESS in 55m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6d981315ebe4015990e397a1f652d24 : SUCCESS in 43m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2500984d0e494b40b5a2198638665669 : SUCCESS in 1h 58m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79de4269757b49268bc0127524ad9dfa : SUCCESS in 1h 49m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b867cba68e2b44c394ada4f8e129802b : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3c391fa708f94c829433489877ae6f22 : SUCCESS in 1h 42m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bae9aae66b274002881081412c82f0a8 : SUCCESS in 34m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb36a427802443a685fd7af5fc38df02 : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27cd08013f5a4e798da687858823f301 : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79381fb011d448c48e584f964a77d5ab : SUCCESS in 26m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa256ef088b48efa7413404c2022657 : SUCCESS in 22m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e51313c6abd45e593c21988a10cb1c1 : SUCCESS in 12m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8711460824f74377b575c140e5bf474f : SUCCESS in 1h 13m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea00bcfa7aed459dbae43fbf9cb8488d : SUCCESS in 1h 15m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dcb4cb7e57894ca29c4a9021b6ebf571 : SUCCESS in 1h 59m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af6af3ea38114a4a91b1c650189ded0d : SUCCESS in 1h 06m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a0a782ae25924b2e8f73e283646e44ca : SUCCESS in 1h 29m 32s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/84bd1e657ff24643a9b5351305bec972 : FAILURE in 52m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0a4c023828749d98b57b540259b73c4 : SUCCESS in 1h 13m 52s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bd4430d9f34344678ec7f0df3534f4a2 : SUCCESS in 1h 48m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f005366aec6c4789b7539c5895b0ce36 : FAILURE in 1h 54m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/74811eeedcde4f80a3bcc496c1712711 : SUCCESS in 1h 49m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c069abdb8aa34ed891d7fafb3365172c : SUCCESS in 2h 28m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7574ddbc164c4d89a8f9f1ad4080cf21 : SUCCESS in 1h 10m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6cf6b8a3157b4f0b8651414f6e1f304a : SUCCESS in 1h 13m 32s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8a0a4dd9b56e4bd7a45c8928b9abfc3e : SUCCESS in 1h 23m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/effcd020dcd04fde84abfeda74f34f33 : SUCCESS in 57m 28s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/01e6d398a03f4aedbb53b372702efbf5 : FAILURE in 1h 20m 15s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/241f34535f1e4346bf7415a96c20a1ce : SUCCESS in 1h 16m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/69da17079ad34f4190b7db2cc0eb5635 : SUCCESS in 3h 17m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a5581aa1d55a4eacbf900d6e8472a5a2 : SUCCESS in 1h 11m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/404710b5bc424fcdbe0ca4f8522bde60 : SUCCESS in 1h 39m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8eac3833cf9845f18152850d465b6a30 : SUCCESS in 1h 58m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4d4714db22704c08adb6acb583adb1a1 : TIMED_OUT in 3h 03m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f6c20374d0d2414a91803d539da6a20f : TIMED_OUT in 3h 05m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"98328ac3a05d6cfc0fe4d6556827d211feaa4e63","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-27 09:09:41.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7d60e9dc52609c2f1cc6b46c7890cee0e1467bf3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-30 05:50:22.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"bf4db05ebbeacab14d238253e977181d5749745d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-12-14 06:18:57.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"39df68086637cb6b88aa083a7a2d4082537aa7ef","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-14 08:09:04.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752248/13 : FAILURE in 1h 49m 17s","accounts_in_message":[],"_revision_number":13},{"id":"16d58498bfb6a45e4cd23bf6b00951e47ea4ca97","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-14 08:48:26.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/48/752248/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"c67b108ea4ede42aa4ab2a8422ba9d263eaeb5b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 10:09:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fc5bca45a9a843bab57a52d2ca064069 : SUCCESS in 48m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d3fb28241474e6dafdf1e24f1b1793b : SUCCESS in 41m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/244bc01ca814406bb201b509d7ca61c6 : SUCCESS in 1h 37m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fe4b831d06b042b498cbffd44e178566 : SUCCESS in 1h 59m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8269536ef084fa9ae40e7dd75c940c8 : SUCCESS in 1h 11m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f8c55e2634224ca4b07b433540b6eb5a : SUCCESS in 1h 26m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfe04e7b76a44f1d875f05cae7b9f68a : SUCCESS in 40m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32c539e18e664eeca362f66e9c2b411d : SUCCESS in 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f45931dd51043ca942ec2dfc45a79c3 : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ba90e957b1d4597b25abc7b5c92bfb6 : SUCCESS in 25m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21e19c7fae0841f49d2a1bbffdfb7f96 : SUCCESS in 29m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9eb12e53520b4279820b8b5b60f955a9 : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2481935c60948dcacd5605598c8d4bb : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fbb8b94d20b47f9bbde13c30e2bd62f : SUCCESS in 1h 06m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f5bcd9785de4e13aa5b6d60ee7afdac : SUCCESS in 1h 14m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/66c5842a178c46a9b6fa4038cfc1b390 : SUCCESS in 1h 33m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82ae41e2b53c4ab1a0146d48c4a04d92 : SUCCESS in 1h 08m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6c23a6a3d9124885b6072d35d3599013 : SUCCESS in 1h 09m 51s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/d81f60da730d466ebd022b6489a441cc : FAILURE in 53m 07s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/afe3e4a2ce35497596780a70d6b5829f : SUCCESS in 1h 27m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/66971dbd9fad407493bed5a8efe23107 : FAILURE in 1h 10m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/362c4e6a0f844168a2bf04927b957e8a : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0d98c3358384c9e8cbfb1d67a1c21be : SUCCESS in 2h 24m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5d60ccc6330488b87b40d91220bef62 : SUCCESS in 1h 09m 22s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/de3a484513184a0589bc1d8377dc101c : SUCCESS in 1h 02m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/614fc27100ee44109a61f9b540776c48 : SUCCESS in 1h 12m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5b900d638381490a93243ac19b4e453b : SUCCESS in 36m 49s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/2b0771e6bf7947adb6bb84a2933d5be7 : FAILURE in 1h 13m 58s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5d4612b46b1f4f5ab15aef6d78b9dd8d : SUCCESS in 1h 18m 30s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/65c9225e87d446379dd0794ebf429ca0 : SUCCESS in 3h 09m 31s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7aec363561ac4873a6747d1910edef55 : SUCCESS in 1h 13m 32s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5e36563086694292a33f524f5328617c : SUCCESS in 1h 38m 32s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6346728f91c947a4a6f0035c85858cc3 : SUCCESS in 1h 52m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/83672555a2454b5daf88ff75ecb0d6cf : FAILURE in 2h 18m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0bc11b929e214835b54387aa7a35ae9a : FAILURE in 1h 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"db95ae53b8edf5657c235a0deaeacfc7eede1f1d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-15 15:31:19.000000000","message":"Patch Set 13: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"3d2ca7155dc7dc3e121fc724af750e20bc490e30","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-15 17:15:26.000000000","message":"Patch Set 13:\n\n(3 comments)\n\nI\u0027m fine with the patch, just minor nits and L712 recommendation.","accounts_in_message":[],"_revision_number":13},{"id":"d2e5490147d12542c3358a643b2e9506d2c2744f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-12-16 05:39:29.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0741e5d3cc0be1b486405683fb2f5a6d5c5471fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-16 05:44:53.000000000","message":"Patch Set 14:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/48/752248/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"d43762c0d0e287b421b8afe3f19ef6c1e047c997","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2020-12-16 06:00:40.000000000","message":"Patch Set 14:\n\n(5 comments)\n\nRodolfo,\nsubnet ip is made as list based on review comment https://review.opendev.org/c/openstack/neutron/+/752248/3/neutron/plugins/ml2/drivers/openvswitch/agent/ovs_dvr_neutron_agent.py#118\n\nOther comments are resolved.","accounts_in_message":[],"_revision_number":14},{"id":"f9ff420b29226367f22fe348cce340382c160421","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-16 07:26:53.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/752248/14 : FAILURE in 1h 46m 28s","accounts_in_message":[],"_revision_number":14},{"id":"482230b1be7a178f4db32c71b370fbb8db68aa70","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-16 08:21:18.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"dc89a80e66180083f1b79cc5e08fc300887afdae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:25:39.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/feb853514119470794c944bb8799e6b1 : SUCCESS in 53m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f872e29379842f5a6db43dd21ac8f4d : FAILURE in 34m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0274b29df73a41d99d5e829de1acd2c6 : SUCCESS in 1h 49m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6be93d52848e414a97bfec1fe658ff97 : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62a170b00a634225b207691c2a0c6f70 : SUCCESS in 1h 45m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8c040dfee2314598a8e223fd0ccaaa45 : SUCCESS in 1h 43m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d277c1ce7784a1192d316cac93568da : SUCCESS in 22m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8c26a854c9a42718dd32a0db56857d9 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed0e2202d7e4b37b5ce467d59b2ad6c : SUCCESS in 11m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caca52f05b6d44bdb1ae741842d4645e : SUCCESS in 35m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1173b9ebc74fe19f6a1bcd40ba383e : SUCCESS in 26m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68bd1e0fa72743fda751caa66d922a5e : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd36e7b1ec849caafd1efe76cef3a2d : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c5ceac745254a9688e72811a54f8017 : SUCCESS in 1h 14m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69ff002d73bd4d4896c37b94889bec45 : SUCCESS in 1h 08m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8ec2db9645f9441995b3df75b59062d4 : SUCCESS in 1h 39m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/52e44f72cb684d0885861f339e7bf630 : SUCCESS in 1h 01m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ff8f3c1c895438d9a91c091f8745991 : SUCCESS in 1h 00m 48s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/c39edbcc634d43a2b0ceac3230a0377e : FAILURE in 58m 23s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c0d08ffdba1241f48cb491dfc105a6f1 : SUCCESS in 1h 34m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4006feedaebc49cc8cfae24c6f399cad : FAILURE in 1h 31m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0c631dbfefae48af87a7f8c1d67cf7c2 : SUCCESS in 1h 23m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59d6170319884bd3b2f9d32c65314612 : SUCCESS in 2h 40m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee3fefad63c5486bb6ce5cc884667707 : SUCCESS in 58m 25s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/fa5be231201d4710ac8acf6a5d77294d : SUCCESS in 1h 05m 04s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/832fe2dde5524da8b682637a8dad8b1a : SUCCESS in 1h 21m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/310b606ca27c463f905b6708cbaf86ae : SUCCESS in 37m 32s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/a4af6b26ad434c2eac45ee4d6bacdc68 : FAILURE in 1h 05m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/453923dc99c94d9eb66b2d8c1acc6bbe : SUCCESS in 1h 13m 32s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/46b7d27a29ae47e29a119feec3028e1e : SUCCESS in 3h 18m 55s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d2442a6a621e4d56926d64c8d721fef5 : SUCCESS in 1h 39m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/820aca35ee58443ca85c10c73c35688e : SUCCESS in 1h 33m 00s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f6409fe3cd4f4d88a84eca53ce776e14 : SUCCESS in 2h 00m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/73d7750d38b7445e8799d87689961bfa : FAILURE in 2h 54m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/758a3b3b3f69485aae4271cb2c273efc : FAILURE in 1h 55m 52s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"491435c4ec662741782435a0d6d06916691b650f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-16 22:44:01.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"86b279bed8ef9fcc0485a49c032838048aa8dc97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:44:19.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"05b469e6816a48814af248b6c4642a52803e39a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 01:16:09.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d5b9c3880a5f432381b4ec98ceff3b08 : SUCCESS in 48m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/97b2b085e848429ba0d7b64adcb8837e : SUCCESS in 1h 36m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ed866039304a492ba648d38c92327bf9 : SUCCESS in 1h 33m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb718659c390408c9dc04bade49e7f2f : SUCCESS in 1h 43m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/002583d022b4402e8903294a40412068 : SUCCESS in 1h 21m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37bfccbf7da241ba935264aca6231c24 : SUCCESS in 32m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d6609cb0b246fdb5dc1b29a0154c16 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8075d882147f40d18c2722e11ba38b2a : SUCCESS in 27m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6f6f7fdd6264f9ea0bc60f6020cb235 : SUCCESS in 26m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22a3a47f69042af8fd744d0914aad59 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50c26148eaff4c1885c3e8611c80d1ec : SUCCESS in 1h 18m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5812fa6bb6c4d4899be00ba7f968ec0 : SUCCESS in 1h 16m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9809c7168572430abe85fec7d46c23e2 : SUCCESS in 1h 15m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/33793a5771c640f994501bb3327a130c : SUCCESS in 54m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce974b3d498f48c3b68bc4ac2f29c979 : SUCCESS in 1h 02m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6fa2e1a5c6b348948231f3cf7ecb67ad : SUCCESS in 2h 22m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58f92cbac8d4473abfb0953f9009d30f : SUCCESS in 50m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a0eee5a4752b4c9eacfd496a443faf88 : SUCCESS in 1h 08m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/883fd8e4d3b54b51966d002db7c5321a : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":14},{"id":"492d0c6228049c05edac3927bb111603c5ed828f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 01:16:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"1b36f84e38db8ac62adc437981fbd277a2a27d0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 01:18:12.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892f75d93d014c999ffabdfb749e1443 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"329ea19f8b881421b9b5ae2fcc855f96af72a9f5","revisions":{"e05f99f8586ab83883410f0395f730da8ae606eb":{"kind":"REWORK","_number":1,"created":"2020-09-16 14:16:24.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/1"}}},"commit":{"parents":[{"commit":"a1d89e6f0eb27400984bf3e4f37704147f592a9d","subject":"Remove install unnecessary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1d89e6f0eb27400984bf3e4f37704147f592a9d"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:16:20.000000000","tz":330},"subject":"[WIP] OVS DVR non-gateway port delete removes dvr-src mac flows","message":"[WIP] OVS DVR non-gateway port delete removes dvr-src mac flows\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances on that compute node. The instances\nare not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway or not. And\ndeletes the DVR-SRC mac flows only if it is gateway port.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e05f99f8586ab83883410f0395f730da8ae606eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e05f99f8586ab83883410f0395f730da8ae606eb"}]},"branch":"refs/heads/master"},"d1920fd24b5e176ecaa0321d3758d43e17bf4e0e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-17 01:35:20.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/2"}}},"commit":{"parents":[{"commit":"c2d8de10e451e15cd0a86cb4bd0f19d4effbfcae","subject":"Merge \"Remove OVN floating IP deletion of associated LSP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2d8de10e451e15cd0a86cb4bd0f19d4effbfcae"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-17 01:35:20.000000000","tz":0},"subject":"[WIP] OVS DVR non-gateway port delete removes dvr-src mac flows","message":"[WIP] OVS DVR non-gateway port delete removes dvr-src mac flows\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances on that compute node. The instances\nare not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway or not. And\ndeletes the DVR-SRC mac flows only if it is gateway port.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1920fd24b5e176ecaa0321d3758d43e17bf4e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1920fd24b5e176ecaa0321d3758d43e17bf4e0e"}]},"branch":"refs/heads/master"},"bf1dc622a26b4cf8cd8d441a6e64b093eb3fdbe0":{"kind":"REWORK","_number":3,"created":"2020-09-17 09:48:38.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/3"}}},"commit":{"parents":[{"commit":"a1d89e6f0eb27400984bf3e4f37704147f592a9d","subject":"Remove install unnecessary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1d89e6f0eb27400984bf3e4f37704147f592a9d"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-17 09:39:23.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances on that compute node. The instances\nare not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf1dc622a26b4cf8cd8d441a6e64b093eb3fdbe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf1dc622a26b4cf8cd8d441a6e64b093eb3fdbe0"}]},"branch":"refs/heads/master"},"cdb96310b12791b45f17cb55f0380019f31f1c41":{"kind":"REWORK","_number":4,"created":"2020-09-22 11:21:39.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/4"}}},"commit":{"parents":[{"commit":"a1d89e6f0eb27400984bf3e4f37704147f592a9d","subject":"Remove install unnecessary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1d89e6f0eb27400984bf3e4f37704147f592a9d"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-22 11:19:49.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdb96310b12791b45f17cb55f0380019f31f1c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdb96310b12791b45f17cb55f0380019f31f1c41"}]},"branch":"refs/heads/master"},"3f4f00f2b7337c6fd2c0cf1eb466712929afbc34":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-30 12:48:18.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/5"}}},"commit":{"parents":[{"commit":"bcc530bd26fdafced181e23803ccb5e0d8b3ede8","subject":"Merge \"Add source_ip_prefix and destination_ip_prefix to metering label rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcc530bd26fdafced181e23803ccb5e0d8b3ede8"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-30 12:48:18.000000000","tz":0},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f4f00f2b7337c6fd2c0cf1eb466712929afbc34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f4f00f2b7337c6fd2c0cf1eb466712929afbc34"}]},"branch":"refs/heads/master"},"a2137fde85365024e5d20ee48da275e01afba9f7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-01 07:09:32.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/6"}}},"commit":{"parents":[{"commit":"6ccfc34227b46451ba915b48189f208434481989","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ccfc34227b46451ba915b48189f208434481989"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-10-01 07:09:32.000000000","tz":0},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2137fde85365024e5d20ee48da275e01afba9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2137fde85365024e5d20ee48da275e01afba9f7"}]},"branch":"refs/heads/master"},"f98bff379c5dc77a66d7485da7ca07de1c6bb761":{"kind":"REWORK","_number":7,"created":"2020-10-01 12:15:51.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/7"}}},"commit":{"parents":[{"commit":"6ccfc34227b46451ba915b48189f208434481989","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ccfc34227b46451ba915b48189f208434481989"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-10-01 12:15:15.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f98bff379c5dc77a66d7485da7ca07de1c6bb761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f98bff379c5dc77a66d7485da7ca07de1c6bb761"}]},"branch":"refs/heads/master"},"554f9e5e4a7c7a0f22ce5d36a66fbaf98bd0a6f1":{"kind":"REWORK","_number":8,"created":"2020-10-05 04:02:14.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/8"}}},"commit":{"parents":[{"commit":"6ccfc34227b46451ba915b48189f208434481989","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ccfc34227b46451ba915b48189f208434481989"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-10-05 04:01:55.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/554f9e5e4a7c7a0f22ce5d36a66fbaf98bd0a6f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/554f9e5e4a7c7a0f22ce5d36a66fbaf98bd0a6f1"}]},"branch":"refs/heads/master"},"a7385fde3a9b6c5b8de70899c10472a68f2eeaf5":{"kind":"REWORK","_number":9,"created":"2020-11-04 06:03:41.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/9"}}},"commit":{"parents":[{"commit":"dc10558c0b95b5ae6805b3aab4eeaf4e204a2e2d","subject":"Merge \"Fix recent QoS minimum_bandwidth release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc10558c0b95b5ae6805b3aab4eeaf4e204a2e2d"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-11-04 06:03:37.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7385fde3a9b6c5b8de70899c10472a68f2eeaf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7385fde3a9b6c5b8de70899c10472a68f2eeaf5"}]},"branch":"refs/heads/master"},"15ec075fc7b0ca2bf167f24adba4e73b7fdca2b5":{"kind":"REWORK","_number":10,"created":"2020-11-24 06:19:57.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/10"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-11-24 06:19:49.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15ec075fc7b0ca2bf167f24adba4e73b7fdca2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15ec075fc7b0ca2bf167f24adba4e73b7fdca2b5"}]},"branch":"refs/heads/master"},"3180d400593839e492e63c2ac86801a0468ee85e":{"kind":"REWORK","_number":11,"created":"2020-11-24 06:30:50.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/11"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-11-24 06:26:35.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3180d400593839e492e63c2ac86801a0468ee85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3180d400593839e492e63c2ac86801a0468ee85e"}]},"branch":"refs/heads/master"},"dceafabc8aa135d84bbd47a3d35717d3efd50b09":{"kind":"REWORK","_number":12,"created":"2020-11-24 12:31:35.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/12"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-11-24 12:31:02.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dceafabc8aa135d84bbd47a3d35717d3efd50b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dceafabc8aa135d84bbd47a3d35717d3efd50b09"}]},"branch":"refs/heads/master"},"0baa4e0d2d18cfc454a8cf3e74733559fb67a6da":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-12-14 06:18:57.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/13"}}},"commit":{"parents":[{"commit":"dc2c279c108e2d9eb15b86fe68f85b70f714bdfa","subject":"Merge \"[OVN] Ensure metadata checksum\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc2c279c108e2d9eb15b86fe68f85b70f714bdfa"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2020-12-14 06:18:57.000000000","tz":0},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0baa4e0d2d18cfc454a8cf3e74733559fb67a6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0baa4e0d2d18cfc454a8cf3e74733559fb67a6da"}]},"branch":"refs/heads/master","description":"Rebase"},"329ea19f8b881421b9b5ae2fcc855f96af72a9f5":{"kind":"REWORK","_number":14,"created":"2020-12-16 05:39:29.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/48/752248/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/752248/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/752248/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/752248/14"}}},"commit":{"parents":[{"commit":"64637e1468e0d3f02226bde410dd61a2a7b4bd26","subject":"Merge \"[OVN] Updates to tools/migrate_names.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64637e1468e0d3f02226bde410dd61a2a7b4bd26"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-09-16 14:08:55.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2020-12-16 05:39:22.000000000","tz":330},"subject":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted","message":"Fix removal of dvr-src mac flows when non-gateway port on router is deleted\n\nRemoval of non-gateway port on DVR router deletes all the DVR to\nSRC mac flows for the instances of same subnet on that compute node.\nThe instances are not reachable from any other network.\n\nThis patch checks if the DVR router port is gateway for the subnet\nor not. And deletes the DVR-SRC mac flows only if it is gateway port.\nThe DVR-SRC mac flows are deleted if the gateway is not set for the subnet.\n\nChange-Id: Iadc1671c862f8c01e5761e92b82a04849d4bb411\nCloses-Bug: #1892405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/329ea19f8b881421b9b5ae2fcc855f96af72a9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/329ea19f8b881421b9b5ae2fcc855f96af72a9f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
