)]}'
{"id":"openstack%2Fneutron~753255","triplet_id":"openstack%2Fneutron~stable%2Frocky~I5415cb3a88ff2640a19598a1fcb2278388815343","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1795548-stable/rocky","hashtags":[],"change_id":"I5415cb3a88ff2640a19598a1fcb2278388815343","subject":"Do not fail deleting namespace if it does not exist","status":"MERGED","created":"2020-09-22 08:48:20.000000000","updated":"2020-09-27 11:57:16.000000000","submitted":"2020-09-27 11:57:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753255-1601207836968-289a0dff","meta_rev_id":"5c8caa0f84c39779af63982d2b8dc4fb1b67eaab","_number":753255,"virtual_id_number":753255,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-09-24 08:00:41.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-27 11:57:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-27 08:42:04.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-09-25 14:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-09-25 13:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-09-25 14:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 13:57:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-25 14:36:57.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-27 08:42:04.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-27 11:57:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b350c0b415004022fecfbad381be895c3f2b5e0d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-22 08:48:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ada10976470e371b0e7be01e78d188af9fdc5a2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-22 08:49:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 5ee6e5bb31179acf6e86357b0b0b34ef051af161","accounts_in_message":[],"_revision_number":1},{"id":"c41b05976aa264983dca94e5156fe223a15bf067","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-22 08:59:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753255/1 : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"84b933351262be641e1a58a034b4f2994e1cdf79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:40:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/846a6772fcfa4607a22f0d2f89b86010 : SUCCESS in 44m 07s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/7c0e8fac72ea4b8f94d5128434e43ab0 : SUCCESS in 29m 16s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/c8fa71611afa4a52aa1e2a6e83fd30c8 : SUCCESS in 1h 22m 24s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/03fd65724fd24354ac8e1016281af021 : SUCCESS in 1h 00m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/5575ecd274ab4214a930ff5d83aa2540 : SUCCESS in 49m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171ba3e7bf424963aefcb9c64ba24382 : SUCCESS in 11m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a38fa6ebeb3f43b2ab6c32bb40e3b7a9 : SUCCESS in 22m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1284846781414e06877066a9aaecf92f : SUCCESS in 26m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6e90097f6d52497a8417c39bc921e4f0 : SUCCESS in 8m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f0a7716b5c44e91b823b8a97a34c7ff : SUCCESS in 1h 30m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4e29aa3f28114bc58b160de3b8a0ab67 : SUCCESS in 57m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97f34e0140c04ac7bbe9e07edb647553 : SUCCESS in 1h 16m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3dd3a8a2397246c0abcbe900e03360a5 : FAILURE in 38m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7b83e80ce1b444bfa1c226353f2a29a8 : SUCCESS in 1h 22m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a555f729e6bc43cc8c289538cf6bd192 : SUCCESS in 1h 36m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8b1767c5d6f34ec68e1311b56e03178d : SUCCESS in 1h 13m 41s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c756529df1be4e16bd6e54c4ba0ad22b : SUCCESS in 1h 46m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4be1d765578340d49fdd9817b8b79644 : SUCCESS in 28m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20c0c1bbe205416eab9f12cb15df5dda : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c7eabe0caaeb666ca60daed9875559cd170e606d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-23 15:31:26.000000000","message":"Patch Set 1:\n\nrecheck may not have been enough","accounts_in_message":[],"_revision_number":1},{"id":"0377ee456280d9310e500284dd2080317a7e4010","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-23 15:33:10.000000000","message":"Patch Set 1:\n\n(though it did pass in queens)","accounts_in_message":[],"_revision_number":1},{"id":"b77f45cefa352968ccf93510c7ac189e69d3a634","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 15:39:47.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753255/1 : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0b54587f0d1c62d9b188974447e3bb31411e1232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 22:14:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/f67fb184af6e4880b72c0614280adbee : POST_FAILURE in 50m 46s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e7828145a1764c6a9ec7e89162932434 : SUCCESS in 37m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/18be163c26514bd08e8c2be8a735f3f5 : SUCCESS in 1h 08m 22s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/0efadd930e334d5ab31406964fce1a1a : SUCCESS in 59m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/15c6e7f29a134e4992b8ceb1b5892a28 : SUCCESS in 1h 00m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c833421fa6734372986dcd00b0e88875 : SUCCESS in 12m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5ee279c49064f8f91ee3a8c90de84d3 : SUCCESS in 24m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/805a2036f5a340fca0fc612f74722cf3 : SUCCESS in 27m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e3c28cb14e1f4d2e9a87c98ad052f431 : SUCCESS in 7m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f8453e327cf54424b4a61476d2f23194 : SUCCESS in 1h 23m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ca595bf65f742249c37e070f97799c5 : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e8e5bfd0d00457eaf69e02c4fa57b9c : SUCCESS in 1h 25m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a94dda9294b843618a0944ced699ad70 : SUCCESS in 46m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c9ec71d23ee145799851db17e9a36aea : SUCCESS in 1h 31m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c9228d6ca9f246539bab07dc0814d5b8 : SUCCESS in 1h 33m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8c69bf5100714031a3627312a7985535 : SUCCESS in 1h 26m 59s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3251be4303d145c7b5f869da10f22128 : SUCCESS in 1h 23m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e0b57491ac34f7fa3aa60445093aaaa : SUCCESS in 36m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc7756a6c674d6fbf6f883d8bc3a269 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0d782a8da75ea4cd0a0883f98d085d0e546e81f4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 08:00:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f8b6f1ce740d43f4690ed419fc38f207a02a4685","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-24 08:09:04.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753255/1 : FAILURE in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9bd1ad7e05378647319dbd26ca7bd69a283e942e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 14:26:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/d8c249cb543d45a9bd2839d236aa9411 : SUCCESS in 43m 17s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/a4306432ada7460fa9d42154ac00f222 : SUCCESS in 27m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f875e680f44b472488fa0ea85bfd7d79 : SUCCESS in 1h 07m 33s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/2542985bb1d54936b514464c30bb61f3 : SUCCESS in 1h 06m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/75be6a2c3e2044b0bfd63f839a74b95d : FAILURE in 2h 52m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd883bb66c3d4b10824e0ea3d48b9b91 : SUCCESS in 12m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eecde322ada408c811b77571d0b7cbc : SUCCESS in 27m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/286e7c6d59784768afcd5abdda553eec : SUCCESS in 24m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/67c14466c3d544f795d880b31cca711a : SUCCESS in 6m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c25edd3422e4a2eaad9247352064e59 : SUCCESS in 1h 37m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a1c17063ae345d382e9f7aae64752f5 : SUCCESS in 1h 00m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab1bf393c4ab40959b522f324de2d729 : SUCCESS in 1h 20m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/942e7b5addd449e5aa1b518dfa6c33d6 : SUCCESS in 47m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5f3cda9d41d408abea38e1a45a993e3 : SUCCESS in 1h 35m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/096d5142c5914eee8eb5b1db0348d4fa : SUCCESS in 1h 22m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ffc91682927d4146936c107360a6281d : SUCCESS in 1h 18m 53s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd9c53fffb784a8095c2cde1dbe751d3 : SUCCESS in 1h 23m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/998e7c8bb12c4c65a095ff984dc98622 : SUCCESS in 30m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/712c2aa72ff740c18c130959bfe5b6be : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d734579f2700534f1b865df7c6d86998ff484035","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-25 09:19:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"945c22f4f52412f88c99f416e37740645c42f61e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-25 09:27:59.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753255/1 : FAILURE in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c40e190f2ab0067f84efebd69dfd8e26a587558f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 13:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/9fdfa06d8d324439b07a9c4172d1e840 : SUCCESS in 47m 46s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/d5e1d86e79fb43848414f844250297d2 : SUCCESS in 33m 10s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/7bf187a433d049dfa070c25f2d5c202f : SUCCESS in 1h 08m 15s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/82d8fa4ec35d41959ab29f1b8d792103 : SUCCESS in 1h 02m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/26fdcd815bde4d3587ed6b0a76a17dae : SUCCESS in 1h 08m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee583d69f424cc79a95496d673b22ef : SUCCESS in 12m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c919b4b60fd4333960bf045cbccc0be : SUCCESS in 28m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ec21ec2148d46ca9e6922e6cccefb6a : SUCCESS in 26m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9ca2b66ed9944cbaa99aff64b4fcbac4 : SUCCESS in 7m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/46f2d99df14f4b99bd375c007876e934 : SUCCESS in 1h 31m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/969fd2ff1ae14380a001d54b339e7b94 : SUCCESS in 56m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6c63fd2ed5a40c19efc36f599eb7f67 : SUCCESS in 1h 25m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ea43fa143a7e41d193d5b4e892e8da3e : SUCCESS in 33m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/692521e9d5244963866beccd7a9a4e4b : SUCCESS in 1h 35m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/10793937ea56457887acf9ab9aec8f4d : SUCCESS in 1h 15m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d98d55a042614d1088ee767904de668d : SUCCESS in 1h 16m 30s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/56c7ddf2994a472fa63da61b12801128 : SUCCESS in 1h 16m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/069de259c2ee433690d89eeada4cc15d : SUCCESS in 26m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b9f1baa7a534a41b8ac6271e1d7cb8c : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1c7413c8c613279b1add8062f69307891cd8244c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-25 13:57:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c84ca3be5443ef654c95ce31b1f8db08b5266f9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-25 14:36:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f5dc71b287700cc91e563b95f70e1a38aa1d6fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:38:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1c6db61584ed899dbedda2419ee0dea50511969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 17:00:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/c792e0e9347a45beb4f84a0c1322c6fa : SUCCESS in 54m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/960f470ca8d6441baae5109dfe022aae : SUCCESS in 12m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2ad4c4d77334833a40929c6121eefc7 : SUCCESS in 29m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7b955901740f41e9a785ee9cfde6aa19 : SUCCESS in 24m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b92e5ad7d952494a8504f7f72dde596e : SUCCESS in 7m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b185029ade3143d889fd7216f5dd42a4 : SUCCESS in 1h 16m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f33f15e8725242c39d680fa7bb38fba5 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7062eec9c1dc4ce4ab71ed674fd2859f : SUCCESS in 1h 12m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f89915bdd1ad4b23945e9696dff89da9 : FAILURE in 1h 01m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e5b3ffef8321447c8fff8aa5f42965d0 : SUCCESS in 1h 35m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dc680e1c4482435b9b7c3d8d49488a61 : SUCCESS in 1h 32m 55s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/553e6ed87dcc4371a783e5dca73a760a : SUCCESS in 1h 28m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb2220b3e06c4abcbd19168fbfd3986f : SUCCESS in 34m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e0ed6b8b29c4552a7174a2a9c5f2e66 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a56996226e041dbcedf2f3d0923a99683d778ff4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-25 18:49:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b582dffe4f2ae81093b55b9c8227ca0925fec24","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-25 18:57:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753255/1 : FAILURE in 7m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5c9446063711a20ec6233387279f24fb17a5c521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 00:58:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/a84ab484de1a48109294295af808adbd : SUCCESS in 55m 39s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/3474468aac314f1f9dbb8023308e785f : FAILURE in 34m 53s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/6c2e22b922fe44c2adbd055d736c72ab : SUCCESS in 1h 06m 50s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/359193e8527f4ab4bfcd6b82f7a99262 : SUCCESS in 54m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c6298e9a4437483ca0a33123a0108db2 : SUCCESS in 56m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e1edfd921a4140a811fc2dfb58665d : SUCCESS in 11m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0f63a7e4eb147b28c1ac03af6f922fa : SUCCESS in 20m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/041ba685228e49e2b0393b05c0fe91f6 : SUCCESS in 21m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2844b3f4e34a40eb835db6fe77e59d72 : SUCCESS in 6m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ec7a383317c480989eebd3e9f955944 : SUCCESS in 1h 18m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d18f39a293b040289467e4b8df1288ef : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52585a8feddf4e779d0e517b7aab0a4b : SUCCESS in 1h 22m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc6acdab713042fdbe0f883d5ba8aa2b : SUCCESS in 44m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fa51a587fdc94ca281273a2f2784f587 : SUCCESS in 1h 23m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/850e5db770ff417ebac0272a59b8c325 : SUCCESS in 1h 13m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0cacd4d81c75497b8c08f21438a82f08 : SUCCESS in 1h 14m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9ca9b33577014034b0f2c8ad0366bf66 : SUCCESS in 1h 19m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a36617fb8b4d4ba581f256447477ddb2 : SUCCESS in 27m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f481c24bfad9487e98f97e576a24bf7b : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"eba491c0fbac807a71e5eb0d147b23201e0a7360","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-27 08:33:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee2d14b459f75fdf3ea1f82fc369a79346bd6252","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-27 08:42:04.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753255/1 : FAILURE in 8m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b048a35da97eb60809a04c355424584320ad9eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 10:08:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/28884656c0a44ecb8609715344acd2d4 : SUCCESS in 1h 01m 40s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/056d351786f14c98a4cc310537fe2d06 : SUCCESS in 35m 02s\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f9d0e2f2985b48bcb502ea3cea500403 : SUCCESS in 1h 15m 47s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/5f8b3484d69b4293b86ead8925af677f : SUCCESS in 52m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/1b40a770a4764d4d8493157b30d8175e : SUCCESS in 1h 06m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb48a2d879e40d588507ac59d41ee0e : SUCCESS in 10m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f231b9ccb9344b168f67f5bb5289bb5b : SUCCESS in 28m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6256bdd9f8654c08b5a2a61fbeb77679 : SUCCESS in 26m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/59f5e4d583394392ac7582335ea049c2 : SUCCESS in 5m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/73c2e305c7b44a1594988cdf043a81d7 : SUCCESS in 1h 33m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6b22c90e2b5242ec9b19833edfaf59fc : SUCCESS in 46m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c8431622ee7433d977464c830fcef54 : SUCCESS in 1h 27m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/66c1d856f7f840bca06ff0202231aecf : SUCCESS in 39m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8becc68323db4c908423ddbef27858c4 : SUCCESS in 1h 27m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e23326ec09684edb9e80543ee524d0bd : SUCCESS in 58m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7689eb39f2344eaea2fa1ec97b66c6d5 : SUCCESS in 1h 06m 56s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f233ea80a71f437e9fe3c369b7060468 : SUCCESS in 1h 16m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19cc7987c401425caace1f47241050ea : SUCCESS in 26m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d016b3f793b43309582bd072945576c : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":1},{"id":"11b51e1a482b3b50525637a4b695fbe06b0fc79b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 10:08:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"406150868a4c40ec3767f5df560943b64af5c66b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 11:57:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18ccb7e4525f9909fc17fdb4c46e27e1c9207b6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 11:57:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/11bc9bdc6e124e9e8af038337b379d21 : SUCCESS in 59m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57218ab4c3e340a683221bed1bf1231e : SUCCESS in 11m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2d0754905c04895ba9e1766c8e892ac : SUCCESS in 14m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d765c3622424b46af250653e7a9bb29 : SUCCESS in 31m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4f0be8421af74f3eb185fa22f2f5c439 : SUCCESS in 6m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f8f30ad2543a49faab3695c0e4bf918b : SUCCESS in 1h 39m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/61873be6cc26465b9f2aacf5c1ff03a7 : SUCCESS in 57m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/372f5009e0b4454e9724d675eeb4ea43 : SUCCESS in 1h 26m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ca3bc24aaba141b591fdea00b6f03c91 : SUCCESS in 40m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0919535211f44a9f90faa559fe7cf990 : SUCCESS in 1h 31m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6db74186f44142c7b4e8f3827c3123dc : SUCCESS in 1h 26m 13s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9cec1e816374984b15f976b371a26c1 : SUCCESS in 1h 41m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/956c8aaa7fa3475a8a8ddcee5b45c711 : SUCCESS in 29m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d1c37e4bbb9422794a4eea52e83c2c8 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e90c5fb73612d73b393a7aa0517fb17d81faf77","revisions":{"7e90c5fb73612d73b393a7aa0517fb17d81faf77":{"kind":"REWORK","_number":1,"created":"2020-09-22 08:48:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/55/753255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/753255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/753255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/753255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/753255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/753255/1"}}},"commit":{"parents":[{"commit":"2bcbd947b373e3571b02a774d5323476b7114292","subject":"Merge \"Use dict .get() to avoid a KeyError in the segment plugin\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bcbd947b373e3571b02a774d5323476b7114292"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2018-10-01 17:49:32.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-22 08:46:32.000000000","tz":120},"subject":"Do not fail deleting namespace if it does not exist","message":"Do not fail deleting namespace if it does not exist\n\nNote: this is a squash of two changes since they are\ndependent on each other, and are currently blocking\nthe gate queue.\n\nSometimes cleanup methods are failing in the check and\ngate queues trying to delete non-existing namespaces.\nSince they could have been deleted asynchronously, don\u0027t\nraise if the failure is \"No such file or directory\" since\nthe system is in the intended state.\n\nCleaned-up the DHCP agent to longer check for existence\nfirst, and the tests to longer mock-out the namespace\nexists check.\n\nFix test_legacy_router_lifecycle failures\n\nMulti-path routes returned via the pyroute2 library have\ntheir outgoing interfaces in the \u0027multipath\u0027 dictionary\nelement, not in the route dictionary.  In that case return\nall the multipath routes correctly.\n\nChange-Id: I5415cb3a88ff2640a19598a1fcb2278388815343\nCloses-bug: #1795482\nCloses-bug: #1795548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e90c5fb73612d73b393a7aa0517fb17d81faf77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e90c5fb73612d73b393a7aa0517fb17d81faf77"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
