)]}'
{"id":"openstack%2Fneutron~753256","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I5415cb3a88ff2640a19598a1fcb2278388815343","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1795548-stable/rocky-stable/queens","hashtags":[],"change_id":"I5415cb3a88ff2640a19598a1fcb2278388815343","subject":"Do not fail deleting namespace if it does not exist","status":"MERGED","created":"2020-09-22 08:49:33.000000000","updated":"2020-09-25 18:15:37.000000000","submitted":"2020-09-25 18:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753256-1601057737544-6ef20142","meta_rev_id":"a11642e784a075c86b7c202f89524ed3b85a5b42","_number":753256,"virtual_id_number":753256,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 18:15:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-22 09:01:45.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-09-25 14:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-25 13:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-09-25 14:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 09:01:45.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-25 13:57:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-25 14:37:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-25 18:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49082d7cb7e938d8876427f2e924d62fc93bb546","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-22 08:49:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"2381406a4cede0dfbc16ceae467dd77331aec250","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-22 09:01:45.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/753256/1 : FAILURE in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0fca28d29dc9cfdc214206a6f56fad8e53d8a43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 12:51:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/3ae2c72ff48345a4a1d361870b0906ce : SUCCESS in 43m 46s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/56ddf121ba634892bf241c7c9d2e135d : SUCCESS in 48m 27s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/2525b62ca5d24e31bd855437805bebca : SUCCESS in 52m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c70a01163d94b0c93aa4660e0d0a0e5 : SUCCESS in 11m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e166289ba8dd493a936b19c494fd0413 : SUCCESS in 14m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b1b6c3b29504299bb3bd3852861b08e : SUCCESS in 25m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6dbc33f5173c44a798d50802ad77a35e : SUCCESS in 8m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2fb64703e035425fb1d887ef04f9dfde : SUCCESS in 1h 28m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/238293433e054023abafe47e4fb0f3e0 : SUCCESS in 57m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be39324dd67e4c56a9db7d5b7f029145 : SUCCESS in 1h 16m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e67a5405c3934cf2b0fc6170002a7bc5 : SUCCESS in 42m 05s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/9a733f378ed74539a509f4cc619d56be : SUCCESS in 1h 52m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/005d0a259a92485989cc4a9363f1ce1d : SUCCESS in 1h 21m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/59d2c0dd86864592b023af4b4bc6be43 : SUCCESS in 1h 12m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc223519f2ed4b45afeca4b482f3eac1 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b62c459560499215e91d0a6321eafc8981308e49","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-25 13:57:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79ed1df62ac04ee0148ded348884d13fdfbebb86","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-25 14:37:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d57e85ab58033c94f016eb3c145fd6d9a95536a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:38:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2f72a729ae8f9123166fadfb521d756aeeb3a18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 18:15:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c7dac465fc34a46da25ad63f05bdfba97b2f49f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 18:15:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e83b9249e2024af1965aeba0b4d9f50e : SUCCESS in 41m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb51e94ccfa47b3a1f63d624653256b : SUCCESS in 11m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5750bb9e4f5845f6b2b72befda3bb71a : SUCCESS in 28m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f553120eccfe4fdda166311429a499f1 : SUCCESS in 24m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4eee09dcb7f646f79e23b04a4c730625 : SUCCESS in 7m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8e803b02d6814afa81946ee31124aa15 : SUCCESS in 1h 20m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/025e62dd6b7d499b8e8d897dc7c19c12 : SUCCESS in 50m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7843c42e41a476e825f5364806cb099 : SUCCESS in 1h 18m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ca5713bd4ef4d05a47f2c42b0e8858d : SUCCESS in 46m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/19b9a93044c145e38b7e92217df14529 : SUCCESS in 1h 21m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ceb20d1172394c688b774036ff4865e2 : SUCCESS in 1h 34m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a63f42110fc44b53843c76670fc0a078 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ee6e5bb31179acf6e86357b0b0b34ef051af161","revisions":{"5ee6e5bb31179acf6e86357b0b0b34ef051af161":{"kind":"REWORK","_number":1,"created":"2020-09-22 08:49:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/753256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/753256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/753256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/753256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/753256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/753256/1"}}},"commit":{"parents":[{"commit":"45f56356065a962c242e187c5ee6cacfa0bd89f8","subject":"Use dict .get() to avoid a KeyError in the segment plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45f56356065a962c242e187c5ee6cacfa0bd89f8"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2018-10-01 17:49:32.000000000","tz":-240},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-22 08:49:33.000000000","tz":0},"subject":"Do not fail deleting namespace if it does not exist","message":"Do not fail deleting namespace if it does not exist\n\nNote: this is a squash of two changes since they are\ndependent on each other, and are currently blocking\nthe gate queue.\n\nSometimes cleanup methods are failing in the check and\ngate queues trying to delete non-existing namespaces.\nSince they could have been deleted asynchronously, don\u0027t\nraise if the failure is \"No such file or directory\" since\nthe system is in the intended state.\n\nCleaned-up the DHCP agent to longer check for existence\nfirst, and the tests to longer mock-out the namespace\nexists check.\n\nFix test_legacy_router_lifecycle failures\n\nMulti-path routes returned via the pyroute2 library have\ntheir outgoing interfaces in the \u0027multipath\u0027 dictionary\nelement, not in the route dictionary.  In that case return\nall the multipath routes correctly.\n\nChange-Id: I5415cb3a88ff2640a19598a1fcb2278388815343\nCloses-bug: #1795482\nCloses-bug: #1795548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee6e5bb31179acf6e86357b0b0b34ef051af161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee6e5bb31179acf6e86357b0b0b34ef051af161"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
