)]}'
{"id":"openstack%2Fneutron~753602","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Iaa03624c5e16dfa284f7947a6ae81143a867c364","project":"openstack/neutron","branch":"stable/ussuri","topic":"ussuri","hashtags":[],"change_id":"Iaa03624c5e16dfa284f7947a6ae81143a867c364","subject":"Test gate. DNM","status":"ABANDONED","created":"2020-09-23 12:40:55.000000000","updated":"2020-09-24 03:52:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ad1e9441fab9cf26f4ed4369fe1b42458b824cc1","_number":753602,"virtual_id_number":753602,"owner":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-09-23 13:08:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-23 19:48:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-23 14:35:08.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 13:08:14.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 14:35:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-23 19:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf6027d29ba6bf32f8c29705813e963771d290c3","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2020-09-23 12:40:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89cdb88ec2a41a129c0074527a65bc8e65f2a487","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-23 13:08:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/02/753602/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/02/753602/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"884816109cbb9fd35a4a19787bcb636bfdfaafb2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-23 14:35:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/753602/1 : SUCCESS in 1h 53m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4f6977276d1c92bb32c77bb036964efa89373f26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:48:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/95777ca582f54f06ab6bc93098f7244f : SUCCESS in 50m 16s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/d36ce72c19094d20bf0e0b931253c79c : SUCCESS in 36m 48s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/39ddcfbcf5af405082b9b5c0600aa60a : SUCCESS in 1h 24m 28s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/bfcddf5d4dcf434f9ef1935841676701 : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/ed54612c658e4ba2a746cba374fafa5f : SUCCESS in 1h 31m 07s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/e57143adf871469696374c6279f6910c : FAILURE in 15m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97942be9c18f4b8bbdd25c14fc311cf6 : SUCCESS in 44m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb31d495f7784fc3b996f7f073a13bc4 : SUCCESS in 25m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/461ff1397b694208802a5c232f6723ce : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49201a2201bc42c3822ae9ae3c60b69d : SUCCESS in 26m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccccb16b0d104441b1614e756e2309fa : SUCCESS in 18m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2b706d7a5aa4bd58b29d720a5cf019c : SUCCESS in 27m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389010c1fefe42caa5396ad522bd05a8 : SUCCESS in 10m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a8babb16a52142e99a7c6807a31bf7a0 : SUCCESS in 1h 01m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f865dd55b8c640109f19849e8998550e : SUCCESS in 1h 22m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e26f869bdda047e7963dc1c58307bd43 : SUCCESS in 1h 25m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cda3cecdb67440beb00d163e373653ce : SUCCESS in 2h 00m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51b324da63e1402b88c50975bf866af3 : SUCCESS in 1h 31m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9471d12e7ae7450da67de3e10cf99e94 : SUCCESS in 1h 14m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/02a957bed9b9490da4721db8ae4128a0 : SUCCESS in 1h 08m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/276de3b265054fa89cef8afdbd1983b4 : SUCCESS in 1h 06m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f99738a29e5445b9d13ad8c26ff512f : SUCCESS in 1h 08m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc627905efc341558271d78dc7f66c3e : SUCCESS in 1h 30m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/17de93087c1f44ea8d9b4f9f1500eb0b : SUCCESS in 1h 33m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc7a2e210f4144219c4f5959b1643e9e : SUCCESS in 58m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e214331ad7624eefa100ee9e8cc78563 : SUCCESS in 1h 12m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/155b3eeba7b2444485cc17754452a181 : SUCCESS in 2h 10m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18882f59eebd40319c6bab31051994ba : SUCCESS in 1h 02m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/310058a273364be0a18fbf4d08b163e8 : FAILURE in 21m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0685f2c4b31b496d8f96f411042e7d66 : FAILURE in 19m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/98b150215bad49b59380f69b1acc87a1 : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd1d4f5ebec648a9820212f3f00f89a7 : FAILURE in 52m 28s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/637af7e867e54593910b0cb4e13640d2 : SUCCESS in 1h 36m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77a8bd3b12f5470580fae2c2997cc733 : SUCCESS in 1h 55m 04s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/51352e6f54fd4fcb86ecfc92a23ac4d3 : SUCCESS in 2h 18m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43778e6c0d36403dab03935205fc83c5 : FAILURE in 26m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"061122c708ab0cef951d5dd99844ad6e23e7722a","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2020-09-24 03:52:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64fcae3a148fb297bb5beb835f5b0faba448664e","revisions":{"64fcae3a148fb297bb5beb835f5b0faba448664e":{"kind":"REWORK","_number":1,"created":"2020-09-23 12:40:55.000000000","uploader":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"ref":"refs/changes/02/753602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/753602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/753602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/753602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/753602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/753602/1"}}},"commit":{"parents":[{"commit":"1432654a7856e410e9fe31fb4bf725cfb2eec990","subject":"Merge \"Ensure fip ip rules deleted when fip removed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1432654a7856e410e9fe31fb4bf725cfb2eec990"}]}],"author":{"name":"asarfaty","email":"asarfaty@vmware.com","date":"2020-09-23 12:40:29.000000000","tz":120},"committer":{"name":"asarfaty","email":"asarfaty@vmware.com","date":"2020-09-23 12:40:29.000000000","tz":120},"subject":"Test gate. DNM","message":"Test gate. DNM\n\nChange-Id: Iaa03624c5e16dfa284f7947a6ae81143a867c364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64fcae3a148fb297bb5beb835f5b0faba448664e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64fcae3a148fb297bb5beb835f5b0faba448664e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
