)]}'
{"id":"openstack%2Fneutron~754475","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I79158baafbb99bee99a1d687039313eb454d3a9b","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1734320-stable/ussuri","hashtags":[],"change_id":"I79158baafbb99bee99a1d687039313eb454d3a9b","subject":"Do not skip ports with ofport unset or invalid","status":"ABANDONED","created":"2020-09-25 19:29:14.000000000","updated":"2023-02-28 15:17:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"72207a1519665bed71bec1b03513b0187b935a92","_number":754475,"virtual_id_number":754475,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-26 02:16:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"date":"2020-09-25 21:47:57.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2020-09-27 10:16:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2020-09-30 13:03:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"date":"2020-09-28 04:57:22.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-12 11:14:50.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 21:47:57.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-26 02:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-27 10:16:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-09-28 04:57:22.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-09-30 13:03:38.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-10-12 11:14:50.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"}],"messages":[{"id":"b193dfaa9babaded6b62585aa3a34e608c30a826","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-09-25 19:29:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df2ad732c83d1cc0dcea285be4237ae0b640f582","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-25 21:47:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/754475/1 : SUCCESS in 2h 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8026db1fe8e7114702a67cb1a828f8260f9158a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 02:16:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f9dafd362eb441148e13bb883f75bc8b : SUCCESS in 48m 56s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/3b0cb680b245464cb1d6e15ce2d595f8 : SUCCESS in 29m 44s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/171ec688d4e3464e95c68d2e24703587 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/a475908ba0624e1aae0d2531e9e17293 : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/f41aa954da2944e6a1982e7839a76d42 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/c8e82e59d45f4ed2ada81c1f6e593b6e : SUCCESS in 1h 21m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af8353e2f88148a78bb573fbf856c3c2 : SUCCESS in 24m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fbf934f15184a27b10be6b6afe3eaf3 : SUCCESS in 26m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2bc386720c46ca93595aba6f72aacc : SUCCESS in 13m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f87052295264f73a5072813ea94fd74 : SUCCESS in 26m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81ad0db4d76b4c8c8edee0a86d8902f2 : SUCCESS in 27m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee6a4fbeb8ce4c829d46f3d18a389eed : SUCCESS in 23m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924e3a4f66fe413a939d39a1e544fe4c : SUCCESS in 9m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d75d792fdd904dc5a9eaa69905cbcb3e : SUCCESS in 53m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5c9cd5c5e68d43f6a46bacc2822241ed : SUCCESS in 1h 08m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e53e45a060f4572810d4fc47b8d16e1 : SUCCESS in 53m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/de289aad49c24074995b9ad847392c4d : SUCCESS in 1h 35m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e613f53773e4d37a214f1453e55b986 : SUCCESS in 1h 23m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6fdede4c80d54025a543d880cb85b087 : SUCCESS in 1h 21m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/20e608a69c394eb695aa27796cd3af69 : SUCCESS in 1h 12m 08s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/47f55a5c23c8410c89c5cb3a99563ae9 : SUCCESS in 1h 06m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c523c4371c384df8b35dc0f13daefb44 : SUCCESS in 1h 10m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d0c68710bc146b58dbc2e5bf80a57cd : SUCCESS in 1h 24m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cf1a34970efa4fc5af57071a2511df03 : SUCCESS in 1h 31m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04cc1f03c5f5459697d913cff36ee9b1 : SUCCESS in 59m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9eb4750e320e4f408df08902cdd021ef : FAILURE in 1h 21m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35f1f7cbd01440fdbda12ca73ce7b149 : SUCCESS in 2h 01m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faee276529f343c89366ebf2468a6f01 : SUCCESS in 1h 00m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3448a362b9c84e0b920232ff2f019db9 : SUCCESS in 57m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bd43769e54c043dd926b4ce2f6f2761c : SUCCESS in 33m 03s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6ca9b07b10b547549210959ba6f89fd7 : SUCCESS in 1h 14m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/462c38a831b847fa893b90dff552d50c : SUCCESS in 46m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/431eb63e887e47ea83b2ccfedf6954e6 : SUCCESS in 1h 25m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c2409db43b224975bcc23a1b84a641c1 : SUCCESS in 1h 38m 42s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b5a25d37fa554e38a475e3ac99347ec1 : SUCCESS in 2h 24m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e2a787032d5b4a0db4a2301b2bfc4e3f : FAILURE in 1h 53m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73c1b8397ea985daeceaf8ac94bb17d0e656ca9e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-09-27 10:16:30.000000000","message":"Patch Set 1: Code-Review-1\n\nthis patch should be cherry-picked with Id12486b3127ab4ac8ad9ef2b3641da1b79a25a50, to avoid the error reported in the last one","accounts_in_message":[],"_revision_number":1},{"id":"1d40844249e8e8b630b8b7d3cb5d4fa3f598922c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-09-30 13:03:38.000000000","message":"Patch Set 1: Code-Review-1\n\nyes we need https://review.opendev.org/#/c/753314/ too and we said we would not start backporing the neutron patches until after the nova patches that use the code are merged on master.\n\nthis is just dead code without the nova patches so there is no point in backporting this untill the nova code is merged.\n\nonce the nova code is merged on master we can backport the nuetorn patch to all supported branches then start the backport of the nova patches","accounts_in_message":[],"_revision_number":1},{"id":"72207a1519665bed71bec1b03513b0187b935a92","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 15:17:25.000000000","message":"Abandoned\n\nIf needed, please restore the patch and address the comments.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9078685df95b3cd7120b96da5373224462b092ae","revisions":{"9078685df95b3cd7120b96da5373224462b092ae":{"kind":"REWORK","_number":1,"created":"2020-09-25 19:29:14.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/75/754475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/754475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/754475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/754475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/754475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/754475/1"}}},"commit":{"parents":[{"commit":"e2831cc752a28d3fb5712ba9c8fd5f626be3cbaf","subject":"Merge \"ovn migration: Fix neutron server container name\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2831cc752a28d3fb5712ba9c8fd5f626be3cbaf"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-03-01 04:43:20.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-09-25 19:29:14.000000000","tz":0},"subject":"Do not skip ports with ofport unset or invalid","message":"Do not skip ports with ofport unset or invalid\n\nThis change removes the \"_check_ofport\" function and its use form\nthe ovs_lib.py file.\n\nBy skipping ports without a unique ofport in the \"get_vifs_by_ids\"\nand \"get_vifs_by_id\" functions, the OVS agent incorrectly treated\nnewly added port with an ofport of -1 as removed ports in the\n\"treat_devices_added_or_updated\" function.\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I79158baafbb99bee99a1d687039313eb454d3a9b\nPartial-Bug: #1734320\nPartial-Bug: #1815989\n(cherry picked from commit 7fd2725cb14b81f442eb57a38755829270ff2c43)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9078685df95b3cd7120b96da5373224462b092ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9078685df95b3cd7120b96da5373224462b092ae"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
