)]}'
{"id":"openstack%2Fneutron~755082","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ie580534e4d91f1ca2e1dc8331632d49d4720e7ba","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1896827","hashtags":[],"change_id":"Ie580534e4d91f1ca2e1dc8331632d49d4720e7ba","subject":"[OVN] update_port should not remove values from external_ids","status":"MERGED","created":"2020-09-29 16:20:33.000000000","updated":"2020-10-06 06:38:45.000000000","submitted":"2020-10-06 06:36:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755082-1601966208642-d979b326","meta_rev_id":"733831a7047d41159d3ece9f5406936e4654f5b8","_number":755082,"virtual_id_number":755082,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 06:36:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-10-01 09:45:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-10-01 09:50:18.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-05 21:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-10-05 21:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 17:10:25.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-09-30 07:04:29.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-30 09:02:15.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-10-01 09:45:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-01 09:50:18.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-10-05 21:14:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-06 06:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0b991d3dcdb8eab27e42fe6ce52eaeeca21edc2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-29 16:20:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29ac692dda7f376d46ffd36d8c5c2d640b31307f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-29 16:21:25.000000000","message":"Topic set to bug/1896827","accounts_in_message":[],"_revision_number":1},{"id":"9a3ce32146c86664e82f4fc4b2724fc55ee172b5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-29 18:31:32.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/755082/1 : FAILURE in 2h 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6fd02bfd0e8635022216efd3007c8daf8d298620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 19:15:25.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/7ddf3200a5bf4261903e4392a91b16d5 : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/5d1e922847624d0e973f85214fe45ea3 : SUCCESS in 28m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/27a98b9c9f87405a8d76b044ab4f68fc : SUCCESS in 1h 26m 08s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/fc9fb2df1f6e4bb2a780cea78249ada6 : SUCCESS in 1h 18m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/802826845cf748e98b068accebeb0a91 : SUCCESS in 1h 14m 57s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/0bc0cd1acf6242dbb14a2c45b4f2d245 : SUCCESS in 1h 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/858bbeecc459489b9a6530217d285af1 : SUCCESS in 43m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fd02c200237409e9ef876a7444a489e : SUCCESS in 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f062c0a516d468fb73c71b6ff15515b : FAILURE in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9966a4b7bb2445bf8baa83a4939796a9 : SUCCESS in 26m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96f5254c66f446aea2f03759da3888db : SUCCESS in 25m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d59e2b41d16546f895947cacbe151e35 : SUCCESS in 26m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9a423a83034df1b66fae1a015d9ad2 : SUCCESS in 9m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/63318e590446417e90c4750c0db46783 : SUCCESS in 52m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d26a7e9ec3a943589592373c06a2f785 : SUCCESS in 1h 15m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5449e414bf4749be8bc6e20332338799 : SUCCESS in 1h 16m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8d40514463524d64b9f8bf5c584ed1e7 : SUCCESS in 1h 45m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd049c2b2d3f4e43a46a965bd33a7e22 : SUCCESS in 1h 12m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/953f7a1fd6b544f592d5d7427be99942 : SUCCESS in 1h 16m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dcccc731ce384c2fbdb6428c9ff728f8 : SUCCESS in 1h 12m 19s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/51c4e301fb494d87b6654dd8b7345fa4 : SUCCESS in 1h 05m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/052079557967437fa8e700f7668dc465 : SUCCESS in 1h 25m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9e5f7e758fc4223bec7e210786ba1f9 : SUCCESS in 1h 38m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5783c446283345829ae6167d3feffd4e : FAILURE in 1h 16m 26s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1cef9761c3104f9594d89964adefe571 : SUCCESS in 1h 10m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2d647a95a48a4a41af66ff286a3dc7fd : FAILURE in 1h 25m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da6e5ab978b04b7a98e0184019e88f23 : SUCCESS in 1h 56m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04dffbbcb4af4507a8b3c3938f357116 : SUCCESS in 49m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/74d017c49f0a4998a2eba1a3e19420ca : SUCCESS in 1h 06m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a6ce94af9a064cb7aa3e0b27bd87b335 : SUCCESS in 33m 31s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/090f8b970a5142ac90445ae29cbdecd5 : FAILURE in 1h 13m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f66ec0bacdc14687a8ff3b4affb2fa90 : SUCCESS in 50m 09s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/efff97dbaf15498c91501c68603be654 : SUCCESS in 1h 22m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b74613b040894382885dbb86eaeeae76 : SUCCESS in 1h 45m 02s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a0785df5deee4e108f64fa22d2001cf7 : SUCCESS in 2h 07m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9ca9685804c846e89dd03662693229a3 : FAILURE in 1h 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28216aff176159f285a0a6c49f185854825a8ea6","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-29 22:09:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30afeed65230a1c4a76f0177d5d34afff29ee546","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-29 23:51:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/755082/2 : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ef399b7ea2266146bdb2070610132484f3f5cf87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 03:11:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f107aeb091bf400084fc039f308a1574 : SUCCESS in 1h 01m 51s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/7c8e8a32dc5b44bc8069e2ce2c068c07 : SUCCESS in 33m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/1d69ca3c228345d78a0ea8d67ad1e360 : SUCCESS in 1h 11m 27s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/9227571d88714d46bfb42c7749cf3879 : SUCCESS in 1h 11m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/aa7c0d7f7f634a6fa9c371e6d2d4ee3e : SUCCESS in 2h 08m 41s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/1b401cc0fb68413a8b04763758677c1e : SUCCESS in 1h 53m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbac6d38130b4e388f59bdda953694c0 : FAILURE in 1h 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d53392023646cfb380cf01518f5c74 : SUCCESS in 50m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6312027f8414f89a3b2a967147ab7d2 : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e979ca3b54249d186dcf46c52e78cb0 : SUCCESS in 27m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1078651010624a848dbbd31d86f5308b : TIMED_OUT in 1h 00m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a04d08842904afba837d32b82539cc8 : TIMED_OUT in 1h 01m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fad91e6dee434ad3bcc34284de4a89a8 : SUCCESS in 9m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f3656d7295224b1dbab753b54a078502 : SUCCESS in 1h 39m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0a87885e55f144c2a11f511360be7b4d : SUCCESS in 1h 12m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2aa0f64a50a74b8ebba8eb7781ea0450 : SUCCESS in 1h 14m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b9e40551c6dd4cd6966c2ac4b2b56261 : SUCCESS in 1h 27m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3fb845315f94830826b937c54300e81 : SUCCESS in 1h 15m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7f0bb21279804a0bb9cac9bb0ea431d4 : SUCCESS in 1h 37m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b3e97b6c3d084a42b3a4d72fe0d8de98 : SUCCESS in 1h 10m 20s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b8109a86d6a4c8f8ea4fb395e41f8f9 : SUCCESS in 1h 06m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e138a4066704a0391b7d729affcd788 : SUCCESS in 52m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/24a0a62672ef4954a3280acaf1d01222 : SUCCESS in 1h 34m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f84b40defc740b7b717184b577ec05e : FAILURE in 1h 34m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00a616222c724e0ab727d4b40d36608c : SUCCESS in 1h 02m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2726e30c91284d1685409f0bdf091de8 : TIMED_OUT in 1h 33m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ee376b329784f2bae28b5842d1a4966 : SUCCESS in 1h 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad1ef76886e3450e924af5bfebdfbae7 : SUCCESS in 47m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a38a618e470a422babd2ae941f449d69 : SUCCESS in 1h 15m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ae3d856f77fb4d79a1af71acc8560fe1 : SUCCESS in 34m 39s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ad3ce2f2a5c14c078eed957ba89b4812 : FAILURE in 1h 15m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27d8d4139c6d44478cb3cf94eb92c3a7 : SUCCESS in 48m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6dcbd60ec2644603ac5a3fd2547d3bac : SUCCESS in 1h 37m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c236eabf0e7a4727953690acc67f5c20 : SUCCESS in 1h 41m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/71997098b62340e79ee5899b989d17dc : TIMED_OUT in 3h 25m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1f38c6a4004a4742976c9c11227ad5bc : FAILURE in 1h 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d85ec9d7a506b8d92e48a6bfac01590253c6b0b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-30 07:04:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5129d7e794ff69f67cd7c12fb5bee0fb86991e5f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-09-30 09:02:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/755082/2 : SUCCESS in 1h 57m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4b1c218bf8093e6ea291e504d36255f3cd479c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 11:42:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/95e7534d2cb04b4d964097346cf9adb0 : SUCCESS in 53m 39s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/7441f18d8ef449a48b4a86a167d6c1f4 : SUCCESS in 37m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/ba1c246278c841d9be038a43949700b9 : SUCCESS in 1h 18m 58s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/1eb6b98af0674afb84f1b26497df7f5d : SUCCESS in 1h 11m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/68c93fb47ab64841a057f4cad1c6d065 : SUCCESS in 1h 21m 06s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/3a81fc2b9763447caee7cd0c10ad0202 : SUCCESS in 1h 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f23b606c56074523ac11b2dd4304c98f : SUCCESS in 31m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aeecd2b778f4a9e839d22d0cb9c4ba4 : SUCCESS in 23m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a80d4c73fb4cb2b7eae85fa49808ca : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4c83b64f464cd48ae98583f39c1d76 : SUCCESS in 27m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50ac3b8de928439b99fd183e80bdfe32 : SUCCESS in 30m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d892784e48ba499ca591f16948e3c16a : SUCCESS in 16m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6543f450354e6bbcc19b319cf6a273 : SUCCESS in 9m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f4f0a23e31c45f2ac2443027623a1f1 : SUCCESS in 54m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/07a423ee17854bf8938070c18c1beb88 : SUCCESS in 1h 22m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/39f0d36aa5e944d5ac61d9d8b0ba4a4d : SUCCESS in 1h 11m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d553b878762644848b2d78f65d92bd2e : SUCCESS in 1h 57m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f144a90e85e1463bafa99f53226438cb : SUCCESS in 1h 24m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9fa8af9b84704f30af6a00bd0f7bb328 : SUCCESS in 1h 16m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e4791981a352488e9c3013da471a393f : SUCCESS in 1h 12m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/80314695b24d4b669559738d3992f45d : SUCCESS in 1h 13m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e1aa22828f574c789747b5b3077844d9 : SUCCESS in 1h 09m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a53c10e4e06344c2b6f172a287b865ba : FAILURE in 1h 18m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1d9016d43674d288aa07b076d227af2 : FAILURE in 1h 20m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9be33d241a634a87a5d72a27b7dcb5a8 : SUCCESS in 58m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/36f97021f43e47d09b154a3aa4b1230b : FAILURE in 1h 20m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b89eb184e6945b79c5600c85da5af01 : SUCCESS in 1h 59m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70a184e50af147e39ffc885dba79cc00 : SUCCESS in 49m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5ec2704ebf9546c7b7ce4ab4be1bccba : SUCCESS in 1h 15m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9c5981e72a36481c85b33fb08496020e : SUCCESS in 42m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ef17c1ce66724a359324c7b3bfa44f62 : FAILURE in 1h 17m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38a8e3de8e1f435b85414d0c844e550f : SUCCESS in 54m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9f557444f264cdc90f0bd39a5476292 : SUCCESS in 2h 05m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f95d797aff284ca28d030a6f52da76ff : SUCCESS in 1h 43m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/da39e6282f554ec4af17c4f90cc0edc7 : SUCCESS in 2h 16m 12s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bd1af6d799024d9da726d9f356530235 : FAILURE in 1h 50m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e09abea44e6dd55917d10793f310725b4e9755d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-09-30 18:56:46.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"828d172d35416e7aa199d410bea13b81d5dc1b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 00:29:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/2ad39f0dc710468db2b3d6b11485d8ef : SUCCESS in 48m 43s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/5420972db3a34c979cb68c98d71b4534 : SUCCESS in 34m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/9f71ff793b0243be8bc6413af3fccbe4 : SUCCESS in 1h 36m 26s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/a285d61bee7d4cb5893477e2c6a897d7 : SUCCESS in 1h 24m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/cb5d7d55cced483c86ea829b18b73114 : SUCCESS in 1h 12m 38s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/be022bc635254ae088a9955db38e778d : SUCCESS in 1h 14m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a6007decab74964bd1818a2c1f43f1c : SUCCESS in 34m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c11e4476be8b48e78be9f0a459e396a3 : SUCCESS in 32m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8772b85dd7c54960af159f93d1afc288 : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f87aa1b6dc42368629643166b6fc76 : SUCCESS in 27m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa1b2981de524fcaaaa5c24598ec2ffa : SUCCESS in 27m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d476e816fde54c93a1245d660131e14b : SUCCESS in 27m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871cb811119b4da8b44d35293c873322 : SUCCESS in 8m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/00e2e56e969144479fb8c4295237e275 : SUCCESS in 56m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0e658523e7e141bdab08f090f052fb45 : SUCCESS in 1h 13m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c96f9b1a67745c882285d3496ad511c : SUCCESS in 1h 14m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3844c2aa637e41adbc6b0dcef2043be0 : SUCCESS in 1h 48m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35431d4bb30642fba70bac90c7ef1117 : SUCCESS in 1h 18m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/788760625173474ea594174c4ab80869 : SUCCESS in 1h 20m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0f47ecc085434e1f962e9d46599916f5 : SUCCESS in 1h 08m 26s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b468263eea24d348b5a78bdaefa0c19 : SUCCESS in 1h 10m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa280bd1b711455a803f214ff87bfa92 : SUCCESS in 1h 10m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5928f60954ac481fb420a8f8f435edcb : SUCCESS in 1h 23m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ab7a9b3266af4db6ad68f431e1da7459 : SUCCESS in 1h 42m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cacd8c867620401e94d24f5c3e7fb1a3 : SUCCESS in 1h 03m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7812193928fd448b83ff1786e7995fbe : TIMED_OUT in 1h 32m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90a7b28dcd2c4e03b019665e62c4d7b9 : SUCCESS in 1h 35m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96fe053e7bb7468eacad6a54e0eb76d3 : SUCCESS in 55m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9851f1b170894500b78a14f4726dc984 : SUCCESS in 1h 17m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7cfd819a55ee400ba075b8b99492f1e1 : SUCCESS in 41m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/70b2b16fd1444f35994cbe27e4734992 : FAILURE in 1h 17m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e8aca64711744a1a5b081f8a47b64c3 : SUCCESS in 48m 16s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ea13e53e0854e848203d42055b446c1 : SUCCESS in 1h 38m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e97d10294c614cbf9ce6063cbdeed1b8 : SUCCESS in 2h 05m 26s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/75109f883a2646dd82b64424575d83ad : SUCCESS in 2h 38m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7950191466434e1b8344d6a18b8e69bf : FAILURE in 1h 54m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a0afcfced085312cc79942539a1a5c908586063","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-01 09:45:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b566d0bc618715a28be07f569b2706da7524d6e3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-10-01 09:50:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0ecb66ab3d198064ef5c98e5b3c4f42400bfa8a6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-05 21:14:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cac9e66ddf555e39b2ded43ada8f72fdf33b1aed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 00:31:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1cfe3578fc95ce012bc07dd94274b4735130d34c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:36:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/eac0727a5d194af08a42a3c5c710ae5a : SUCCESS in 47m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ffe16ee6d83451a9b981449f9043a87 : SUCCESS in 36m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe7d3afb6a74965b18157abfdebb2ac : SUCCESS in 15m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4801236fd2c94e839f7dd4c47600c584 : SUCCESS in 35m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9335cd1c1145493a9cbd0ba5e4d9c3af : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1821788271124e988e35946aa3781bd0 : SUCCESS in 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7e80b9cc07944b7b831407c1de543311 : SUCCESS in 53m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5da5810197824cfaa374758042187cca : SUCCESS in 1h 04m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/389d680cffaa418e8e5530abfc464144 : SUCCESS in 1h 12m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3d26d1680b6e462e98f5228aa2b8904f : SUCCESS in 1h 16m 35s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cfd032bfabab4b179e4acb678d2a0c11 : SUCCESS in 1h 18m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/156b5a46cfdb496c84ce927c9423ed36 : SUCCESS in 1h 09m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f2ed584c2db413391e944d1752e63d0 : SUCCESS in 1h 05m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/896101c860a6488eae99d7e8ee77c8d5 : SUCCESS in 1h 53m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c788b88ffcf94ff39022bdc17c374edc : SUCCESS in 41m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/957b8402068f49ee956c53ebe839c792 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a6ed1a9870d5768844ee87d9c1c1bcf3fe44cbf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:36:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"733831a7047d41159d3ece9f5406936e4654f5b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:38:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1ccc41fb664cf7bdaa4480196249c8 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"58e2def05f5e444e900a5dc03b6ef6b3cb3f5325","revisions":{"a8ceabbd5399c8d74e15bc6163e0c5c8e0b779fe":{"kind":"REWORK","_number":1,"created":"2020-09-29 16:20:33.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/82/755082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/755082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/755082/1"}}},"commit":{"parents":[{"commit":"70d0a7e769a88330118dde7073bdcb4bab04ba7c","subject":"Merge \"Remove \"vf_management\" and \"vf_extended_management\" checks\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70d0a7e769a88330118dde7073bdcb4bab04ba7c"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-23 19:28:55.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-29 16:20:13.000000000","tz":-240},"subject":"[OVN] update_port should not remove values from external_ids","message":"[OVN] update_port should not remove values from external_ids\n\nPrior to this patch, the OVNClient implementation for neutron\u0027s\nupdate_port was setting the external_ids of the affected logical\nswitch port to a hard-coded dictionary. This meant that any key\nvalue pairs that were not listed there would simply get removed.\nThis would make it impossible for any other users of the\nexternal_id to have a reliable way of storing its data. One of\nsuch users could be the ovn-octavia-provider.\n\nCloses-Bug: #1896827\nChange-Id: Ie580534e4d91f1ca2e1dc8331632d49d4720e7ba\n(cherry picked from commit be3669258cb6f8a6c43b2f016a344185075fa544)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8ceabbd5399c8d74e15bc6163e0c5c8e0b779fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8ceabbd5399c8d74e15bc6163e0c5c8e0b779fe"}]},"branch":"refs/heads/stable/ussuri"},"91114f79784819ad4edba0a4138a9fe2b6af7ed3":{"kind":"REWORK","_number":2,"created":"2020-09-29 22:09:42.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/82/755082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/755082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/755082/2"}}},"commit":{"parents":[{"commit":"70d0a7e769a88330118dde7073bdcb4bab04ba7c","subject":"Merge \"Remove \"vf_management\" and \"vf_extended_management\" checks\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70d0a7e769a88330118dde7073bdcb4bab04ba7c"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-23 19:28:55.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-29 22:07:09.000000000","tz":-240},"subject":"[OVN] update_port should not remove values from external_ids","message":"[OVN] update_port should not remove values from external_ids\n\nPrior to this patch, the OVNClient implementation for neutron\u0027s\nupdate_port was setting the external_ids of the affected logical\nswitch port to a hard-coded dictionary. This meant that any key\nvalue pairs that were not listed there would simply get removed.\nThis would make it impossible for any other users of the\nexternal_id to have a reliable way of storing its data. One of\nsuch users could be the ovn-octavia-provider.\n\nThis cherry pick had to have a small modification in the import\nsection of test_ovn_db_resources.py to resolve the following pep8 issue:\n ./neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_resources.py:18:1:\n I100 Import statements are in the wrong order. import copy should be before import mock\n import copy\n ^\n ./neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_resources.py:19:1:\n I201 Missing newline before sections or imports.\n import netaddr\n ^\n\nCloses-Bug: #1896827\nChange-Id: Ie580534e4d91f1ca2e1dc8331632d49d4720e7ba\n(cherry picked from commit be3669258cb6f8a6c43b2f016a344185075fa544)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91114f79784819ad4edba0a4138a9fe2b6af7ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91114f79784819ad4edba0a4138a9fe2b6af7ed3"}]},"branch":"refs/heads/stable/ussuri"},"58e2def05f5e444e900a5dc03b6ef6b3cb3f5325":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-30 18:56:46.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/82/755082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/755082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/755082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/755082/3"}}},"commit":{"parents":[{"commit":"70d0a7e769a88330118dde7073bdcb4bab04ba7c","subject":"Merge \"Remove \"vf_management\" and \"vf_extended_management\" checks\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70d0a7e769a88330118dde7073bdcb4bab04ba7c"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-23 19:28:55.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-09-30 18:56:23.000000000","tz":-240},"subject":"[OVN] update_port should not remove values from external_ids","message":"[OVN] update_port should not remove values from external_ids\n\nPrior to this patch, the OVNClient implementation for neutron\u0027s\nupdate_port was setting the external_ids of the affected logical\nswitch port to a hard-coded dictionary. This meant that any key\nvalue pairs that were not listed there would simply get removed.\nThis would make it impossible for any other users of the\nexternal_id to have a reliable way of storing its data. One of\nsuch users could be the ovn-octavia-provider.\n\nNote: mock module does not exist in stable/victoria and master,\nthus the conflict below.\n\nConflicts:\n        neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_db_resources.py\n\nCloses-Bug: #1896827\nChange-Id: Ie580534e4d91f1ca2e1dc8331632d49d4720e7ba\n(cherry picked from commit be3669258cb6f8a6c43b2f016a344185075fa544)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58e2def05f5e444e900a5dc03b6ef6b3cb3f5325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58e2def05f5e444e900a5dc03b6ef6b3cb3f5325"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
