)]}'
{"id":"openstack%2Fneutron~756483","triplet_id":"openstack%2Fneutron~master~Iea5ff76fe652ab1c04e23850b9259547c1d54365","project":"openstack/neutron","branch":"master","topic":"bug/1898842","hashtags":[],"change_id":"Iea5ff76fe652ab1c04e23850b9259547c1d54365","subject":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded","status":"MERGED","created":"2020-10-07 10:28:01.000000000","updated":"2020-10-10 14:41:53.000000000","submitted":"2020-10-10 14:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756483-1602340814065-91f5063d","meta_rev_id":"47bcbed283e8013020e9f04288001df3f9480585","_number":756483,"virtual_id_number":756483,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-09 22:17:54.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-10-08 14:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-10-08 13:44:24.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-10 14:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-10-08 18:34:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-08 13:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-09 12:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-09 12:53:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 08:15:20.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-10-08 13:17:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-08 13:44:24.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 14:49:51.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 18:34:00.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-10-09 12:53:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-10 14:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f01c39190aeb2b77f4ac69ef8eaf73a697a7037b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-07 10:28:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e3e1951d4d90ccfaca57abf6351da27cb999e2f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-07 10:33:47.000000000","message":"Patch Set 1:\n\nI talked with Rodolfo on IRC. The qos extension driver cannot work without the service plugin and that makes me wonder why it\u0027s even configurable. Perhaps we can automatically load the qos extension driver when the qos service plugin is loaded because there is no reason to have qos extension driver without the service plugin. Any thoughts?","accounts_in_message":[],"_revision_number":1},{"id":"a2530aadf30b48afb07a9066cb3bc761fbe79293","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-07 11:00:14.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/756483/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f51dbf9a59aa58dc97935fb9e41c2eb7f243bbb4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-07 13:09:52.000000000","message":"Patch Set 1:\n\n\u003e I talked with Rodolfo on IRC. The qos extension driver cannot work\n \u003e without the service plugin and that makes me wonder why it\u0027s even\n \u003e configurable. Perhaps we can automatically load the qos extension\n \u003e driver when the qos service plugin is loaded because there is no\n \u003e reason to have qos extension driver without the service plugin. Any\n \u003e thoughts?\n\nAgree with this: the agent won\u0027t do anything with the \"qos\" extension driver loaded while the QoS plugin is not. I\u0027ll create a LP bug to make public this possible improvement.\n\nIn any case, the change done in this patch should work fine always; now we check if the plugin is loaded, not the extension driver.","accounts_in_message":[],"_revision_number":1},{"id":"c0badcd5b0ff060664a1fded5bacd8612865b056","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-07 13:13:43.000000000","message":"Patch Set 1:\n\n\u003e \u003e I talked with Rodolfo on IRC. The qos extension driver cannot\n \u003e work\n \u003e \u003e without the service plugin and that makes me wonder why it\u0027s even\n \u003e \u003e configurable. Perhaps we can automatically load the qos extension\n \u003e \u003e driver when the qos service plugin is loaded because there is no\n \u003e \u003e reason to have qos extension driver without the service plugin.\n \u003e Any\n \u003e \u003e thoughts?\n \u003e \n \u003e Agree with this: the agent won\u0027t do anything with the \"qos\"\n \u003e extension driver loaded while the QoS plugin is not. I\u0027ll create a\n \u003e LP bug to make public this possible improvement.\n \u003e \n \u003e In any case, the change done in this patch should work fine always;\n \u003e now we check if the plugin is loaded, not the extension driver.\n\nOk, I see a possible use case: the driver extension could be loaded selectively in some agents of some compute nodes, not all of them. You can have (if needed) some agents enforcing the QoS rules in the backend and not others.\n\nIf the QoS plugin is not loaded, the QoS extension driver is silently not loaded.","accounts_in_message":[],"_revision_number":1},{"id":"bcac015054c1d81de29f1286d17de7ecf2a203ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 14:43:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b6ef17a2dbd24f7cb6091407054a1384 : SUCCESS in 49m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06b2855a6c724843892b1b44c68afad4 : SUCCESS in 35m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11ac6524048348c892c4c5386d861a52 : SUCCESS in 1h 46m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/62864582eeea44bca8b58c7258f96111 : SUCCESS in 1h 16m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/70669c0c3a4b48518257882f5e954b40 : FAILURE in 1h 15m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9725c8399f2d4205bebe1703aa84a925 : SUCCESS in 1h 03m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd6f4211181a4da8a690b8cb32575851 : SUCCESS in 36m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc685c62c3d44129abc8d687aada34b6 : SUCCESS in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a120664cea414a5eb4ba82f723aadf37 : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/135654257a3b421abfd0affcacab2b87 : SUCCESS in 23m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52449cdd8154401a8c4c98cf17ed8ed8 : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbfeec93dfb462c996a9bf9be75aed8 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95c23b0e5dfd47909fa8309892a960e8 : FAILURE in 1h 09m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cdda046d81d43e6aba46b4e4f9b60bd : FAILURE in 53m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/67c21242c3d64b0aa3917affba3727d8 : SUCCESS in 1h 41m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69663668ed3048dbb4c308287819e598 : POST_FAILURE in 1h 02m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2950d0dfad834f77a376bdf1bba68c1a : SUCCESS in 1h 05m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/76b508d6355b41a6bd5e9604897571bd : SUCCESS in 1h 13m 05s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9badc69acae454092c7fc5f791d6cc6 : SUCCESS in 1h 14m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7339b699af63424fb5a16498b5e98e65 : SUCCESS in 1h 42m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa3d27ebdba04c1cbfc863a3c48922f9 : FAILURE in 1h 25m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd95c0e4da144ae4a72d27d187c12b26 : SUCCESS in 1h 10m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6754a7fc38014654b3723a8babe308c9 : SUCCESS in 1h 22m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3cc8ca4efa447aaac055ee5a7aae6db : SUCCESS in 2h 03m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/925583b5ee0b45968c504e625127a205 : SUCCESS in 47m 18s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f0de6232fea24e90b4c3c0812050a8e2 : SUCCESS in 1h 16m 53s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/75502b853a1b4d598c6ff767c76469ad : SUCCESS in 1h 07m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e26f2538848f4ff0955cc02ff1eec138 : FAILURE in 33m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/42c2e954342c4e578c4f984621650605 : SUCCESS in 1h 24m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f6110dd425c4423c80bdc7430228ef6f : SUCCESS in 1h 41m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9b53270bae604928bc05a7c62778dca3 : SUCCESS in 1h 53m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/179655e93c7843dba2559b52acd8915e : SUCCESS in 2h 26m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2db5dec2c7ac4befaa5a637d9d9d59e2 : FAILURE in 37m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cb91f2d0ece64e5ba614f94327a0c8a0 : FAILURE in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecf1cf62c60cb2b00d5a7b23504e1f7b1ace8938","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-10-07 14:45:20.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fd7a64d247c8cc33efbe8d94613f4683dfcfb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 20:45:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f8ff71949c84bce8c9c582830078cd4 : SUCCESS in 53m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11e8267d7f83430f90bae82d9d60165a : FAILURE in 37m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/67df1c9df81e4c528bbce798ace934fd : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b48654c3570a4cd993ab1f59ff329120 : SUCCESS in 1h 11m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38c93ef7de87429a84d03207f30169da : FAILURE in 2h 16m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c0be0428b68e4a16b27568bfcf8ca9af : SUCCESS in 58m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca71d0ddb60a4df4b40093405191c027 : SUCCESS in 19m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91b4ad358c84eba9ee669e381f46f1a : SUCCESS in 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af7b0e130cad45a7bd7c41c2593fbae8 : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a1edeb8c444485eb524cfbd6ffe305f : SUCCESS in 25m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d373200414c641e68cb7af551794f94f : SUCCESS in 27m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099ef25a47974f2c9da31b0052ddbce9 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f77a0f19422641fabceb679d60638521 : SUCCESS in 1h 08m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b98e19c7ae2406595e342a06351cd3c : SUCCESS in 1h 07m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0febab9ff0d34019b0dd214134c22a10 : SUCCESS in 1h 55m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50a0e6f17658449a9b8ed9ce36673f7e : SUCCESS in 1h 06m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ba111891ce1249f48967da997b0b7e2c : SUCCESS in 1h 08m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3c297fd7eac14943b45a1db0fc4eaab7 : SUCCESS in 1h 13m 54s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ed5731acd584cfaa08e4e71516f58b1 : SUCCESS in 1h 08m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/48ea42e1db4a4702bce504474498be81 : SUCCESS in 1h 55m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f7914b2fe01a44b7bb7169af9727bcbc : SUCCESS in 1h 37m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4765871e85cb452e86d503f50c1abf47 : SUCCESS in 1h 04m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c2715ad4ad9b4bbfa3a7ef88026c796a : SUCCESS in 1h 14m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fe30c2a310642638807db6e0741363f : SUCCESS in 2h 13m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df435e6da04f4904a7c607eb072b1090 : SUCCESS in 1h 05m 45s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/db2a20de9cc449b9a114b09f18082be0 : SUCCESS in 1h 14m 50s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cb314e6663ce46cba0f20f9419ef272c : SUCCESS in 1h 20m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b6b79c67864b4ffd8444b52792fba5b6 : FAILURE in 41m 27s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/da72fe52ed37419ca786f53e3fb6cd9b : SUCCESS in 1h 22m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/57fe4ad40e1c4c6d80b877f48f13a131 : SUCCESS in 1h 46m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/40bae0d92bc44592b7426096d4ce2d21 : SUCCESS in 1h 53m 37s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/170ee6fea16342088e85af4b18cd2494 : SUCCESS in 2h 22m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d56f25bfc427446b985fcc513c2e27d8 : TIMED_OUT in 3h 02m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c54916f7819949ada735667ef58c8fd6 : FAILURE in 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7b0b9f6bdb5fe3544d9560ad9e594f8a7b7e2ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-07 23:36:39.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/83/756483/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"cd07cf650b1b2d43f3939c8835f374f2aec633bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-08 07:50:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"23a350620166aef38a83f5aed513e3144f8a19d3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-10-08 08:15:20.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e I talked with Rodolfo on IRC. The qos extension driver cannot\n \u003e \u003e work\n \u003e \u003e \u003e without the service plugin and that makes me wonder why it\u0027s\n \u003e even\n \u003e \u003e \u003e configurable. Perhaps we can automatically load the qos\n \u003e extension\n \u003e \u003e \u003e driver when the qos service plugin is loaded because there is\n \u003e no\n \u003e \u003e \u003e reason to have qos extension driver without the service plugin.\n \u003e \u003e Any\n \u003e \u003e \u003e thoughts?\n \u003e \u003e\n \u003e \u003e Agree with this: the agent won\u0027t do anything with the \"qos\"\n \u003e \u003e extension driver loaded while the QoS plugin is not. I\u0027ll create\n \u003e a\n \u003e \u003e LP bug to make public this possible improvement.\n \u003e \u003e\n \u003e \u003e In any case, the change done in this patch should work fine\n \u003e always;\n \u003e \u003e now we check if the plugin is loaded, not the extension driver.\n \u003e \n \u003e Ok, I see a possible use case: the driver extension could be loaded\n \u003e selectively in some agents of some compute nodes, not all of them.\n \u003e You can have (if needed) some agents enforcing the QoS rules in the\n \u003e backend and not others.\n \u003e \n \u003e If the QoS plugin is not loaded, the QoS extension driver is\n \u003e silently not loaded.\n\nI thought the extension_drivers is a neutron-server config option only. I don\u0027t think it can be picked by the agents, they have a different agent extensions. I might be wrong tho.","accounts_in_message":[],"_revision_number":2},{"id":"2e0f121ae9d34e26e19ce8d4f63fbb9ace1f542f","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-08 08:22:12.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/756483/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8ed63ca65eba894d72ec4ec29bcbd593b78ca51e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-08 09:04:59.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/83/756483/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"154534570e0e8721daa1870106d2fe484c1b8d70","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-08 11:02:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nJust a question, if that don\u0027t apply here, I will be happy to change my vote to +2 :)","accounts_in_message":[],"_revision_number":2},{"id":"880af52200f7cb7024dfb59a22980d15ac60c6bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-08 12:56:51.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e \u003e I talked with Rodolfo on IRC. The qos extension driver cannot\n \u003e \u003e \u003e work\n \u003e \u003e \u003e \u003e without the service plugin and that makes me wonder why it\u0027s\n \u003e \u003e even\n \u003e \u003e \u003e \u003e configurable. Perhaps we can automatically load the qos\n \u003e \u003e extension\n \u003e \u003e \u003e \u003e driver when the qos service plugin is loaded because there is\n \u003e \u003e no\n \u003e \u003e \u003e \u003e reason to have qos extension driver without the service\n \u003e plugin.\n \u003e \u003e \u003e Any\n \u003e \u003e \u003e \u003e thoughts?\n \u003e \u003e \u003e\n \u003e \u003e \u003e Agree with this: the agent won\u0027t do anything with the \"qos\"\n \u003e \u003e \u003e extension driver loaded while the QoS plugin is not. I\u0027ll\n \u003e create\n \u003e \u003e a\n \u003e \u003e \u003e LP bug to make public this possible improvement.\n \u003e \u003e \u003e\n \u003e \u003e \u003e In any case, the change done in this patch should work fine\n \u003e \u003e always;\n \u003e \u003e \u003e now we check if the plugin is loaded, not the extension driver.\n \u003e \u003e\n \u003e \u003e Ok, I see a possible use case: the driver extension could be\n \u003e loaded\n \u003e \u003e selectively in some agents of some compute nodes, not all of\n \u003e them.\n \u003e \u003e You can have (if needed) some agents enforcing the QoS rules in\n \u003e the\n \u003e \u003e backend and not others.\n \u003e \u003e\n \u003e \u003e If the QoS plugin is not loaded, the QoS extension driver is\n \u003e \u003e silently not loaded.\n \u003e \n \u003e I thought the extension_drivers is a neutron-server config option\n \u003e only. I don\u0027t think it can be picked by the agents, they have a\n \u003e different agent extensions. I might be wrong tho.\n\nRight, the ml2.extension_drivers is to configure the the ML2 plugin extension manager; the agent.extensions is the agent extension driver section. Very similar but not the same at all.\n\nI\u0027ll raise this question in LP. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"6c9ece8f29b4e72e4764c1b181a837ad6dc671f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 12:58:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ae4268105440425a9ef6e8ca59cdcca8 : SUCCESS in 52m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe638113a78d40b9814a337ff8460673 : SUCCESS in 32m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c669f9adaca34543b8afd19baf82ba6d : SUCCESS in 1h 20m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8cd821b12f3c4c79a82a75408095d378 : SUCCESS in 1h 12m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7d080c5bcc84ee8afdbe0d4d117171e : SUCCESS in 1h 21m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1c983cbaeca748d89f1203b1ac08de03 : SUCCESS in 1h 17m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2b4170e1d414be6bbe00a4bc09dcac3 : SUCCESS in 35m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13dae30852f0434d803d90135dd26563 : SUCCESS in 28m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223ada7103ce40c680d2355da6a6238a : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21680d5db86c4711a2e226e9d70be570 : SUCCESS in 22m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782ba500561e45d996702cdacccc69e5 : SUCCESS in 29m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715d442fdd9f4b18b5320d0aa9ae4405 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f80a883317b4111a36378a006a3fe41 : SUCCESS in 1h 10m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ece1239fee14ad1a7dd88abb7839e77 : SUCCESS in 1h 13m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/045c9493b53f4b50892f3e0630a77635 : SUCCESS in 1h 41m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f49e9b40f3aa46929adb1e4d1c6bbeba : SUCCESS in 1h 15m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0032eae35174425b9b4c302503e6c229 : SUCCESS in 1h 12m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/746a1f9639f944b1a542deea60f56a08 : SUCCESS in 1h 19m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7aa5a8bcaa6a40b4a3a3f77fb707fd61 : SUCCESS in 1h 10m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a9f1181d354f4264a84867feea40df9c : SUCCESS in 1h 32m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ee96989d4dc495baa455db253b16b17 : FAILURE in 1h 25m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d27618d9484a40b298e5c7df22fba9fe : SUCCESS in 1h 16m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/74ba6726ca674fcc88fa72ac52157682 : SUCCESS in 1h 18m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e5ae610e4ca4f39988c8d53343cd8bc : SUCCESS in 2h 07m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f6518d33ec1455dba3d2631192ea3c7 : SUCCESS in 1h 06m 57s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a476bace38654a968cf4af250f893a05 : SUCCESS in 1h 17m 49s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1f5288e7b0a046f9806ca4418bdd6d0c : SUCCESS in 1h 11m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/944278d72f5648078134083a3c494caf : FAILURE in 42m 36s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e4ede0854e2a4e5e89639205d5e90da5 : SUCCESS in 1h 21m 52s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ccd3258e466495683555dfc5bc7b03f : FAILURE in 1h 32m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5fd8116bc7b748e086d3c6e704abe3ec : SUCCESS in 1h 43m 30s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d94db700b04a4b6e8cdce76822c15d38 : SUCCESS in 2h 14m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2713134934f14ace977f45fd46b7df10 : FAILURE in 32m 38s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/045f57f112b14415b8df70dabec90e51 : FAILURE in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7dd654f69b1b8f105de722bc33f2fc86123f168","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-08 13:08:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58622b28cf8a166945ebda7916bb07e509ef60df","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-08 13:12:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31fb2f24e25d4b71cfe39fd5a8a1bd98d20602e8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-08 13:17:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"53b3f8caa70efa8ea4a152852f08a0b225866046","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-08 13:44:24.000000000","message":"Patch Set 3:\n\n* sriov-test http://52.27.155.124/sriov/756483/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ac048d5b1ccd5323ba0c410c12abdc4637ff3044","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-08 14:49:51.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/83/756483/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2b7cbeaa71580a08b7c563027d9e67dd4fd1487c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-10-08 18:34:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0045f01bfb50ad7b126997f160acdebc3553d943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 01:10:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b1eddbb90b544a8cad201b90cde93014 : SUCCESS in 58m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/509103cd389d488f85d7628384edcb5e : SUCCESS in 34m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/290d0705929741b7a529154e7ae9917d : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/994efd9256e54fa3a45aef697420a2f5 : SUCCESS in 1h 07m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d117b8dff0484d6393a5c9bc77398098 : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ed499b66c09647c9b02cbba6237bc808 : SUCCESS in 1h 13m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9776c0759b6d420ab9358959ea7560ba : SUCCESS in 33m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd66d0b47a9e41428eb03db9d6eb2c19 : SUCCESS in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb868f2b8bf4a83bc45680f24a5f8cd : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac60cb59f3643ab9c08fb25768cc20c : SUCCESS in 23m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e181b8e935654db0a5e1869e27edb339 : SUCCESS in 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc5e426b48a48918c7e94ad0a460800 : SUCCESS in 8m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8da065f47894255bc679b77c7854185 : SUCCESS in 1h 12m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93b8a154b6a04632bda24fd01b40d9b2 : SUCCESS in 1h 20m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/51bc7ead4666476cb32444b09ac008c2 : SUCCESS in 1h 44m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eae94e5e087543969115f411a69be2a7 : SUCCESS in 1h 05m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1a3cb9ff5ef84a6a905fd0a0bf190ba5 : SUCCESS in 1h 02m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/916b35e4e578429faf4ace8407c8c23a : SUCCESS in 1h 06m 28s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/01449c4574e94634abedd432b9e92260 : SUCCESS in 1h 07m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a01c8659384d4f19bc622396688ccc78 : SUCCESS in 1h 25m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4115687aedf84a04b9f1d3ee4113f49e : SUCCESS in 1h 54m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a4424135ae146529c399e5f1dbcf835 : SUCCESS in 1h 12m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/929c4cf934d24e289a721824bea6a793 : SUCCESS in 1h 26m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c912385a1f7041af85a9ed716b7d1ee3 : SUCCESS in 2h 07m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6c5eb242ad743128e87aef34e6c427a : SUCCESS in 59m 57s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7995af53a974462982a4e0cd3446aa0b : SUCCESS in 1h 10m 44s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ab71136dcb3847b4bf7dde55706a0982 : SUCCESS in 1h 13m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f0bbe2b8d00745dab35fdb9a15a76d03 : FAILURE in 36m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/53fe11f0e0ec4b16ad933a687583cdbd : SUCCESS in 1h 24m 36s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/657da3721d2d44f2a7b7e8bc6b08cc3b : SUCCESS in 1h 32m 25s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/82575a8e16a24c45879de9521d9b70d7 : SUCCESS in 1h 57m 50s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3aea0e44c3e948b88cf6e52c9a1e6cbe : SUCCESS in 2h 29m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6c8a875cab4341a285ed4574361051d1 : FAILURE in 45m 29s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1d31ee2e005e458e9be24d20411153b2 : FAILURE in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"209cb5e74ec5137467565162daeb8a8413e967ac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-09 12:53:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"174287002f7e8ae82b54eee20a1b6100813d5334","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 12:54:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e868a70e7cf970690abf7ab46df7cdc0198cca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 16:18:08.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a00150600a3740da92123f0bbfb4f03a : SUCCESS in 55m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4ca8ba7d7ecd4978b9b3c16b10477c94 : SUCCESS in 1h 33m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/280e0494878443a49d4aa7751b4ef7c3 : FAILURE in 59m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/94e7b702e88741239bcec089d487baf5 : FAILURE in 1h 00m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d94c173c27524f378d6f0c74d412632c : FAILURE in 1h 01m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46de80a6f4794fbf954a0780ae8974ad : SUCCESS in 42m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840464882af04514b4edc6fe17fd2495 : SUCCESS in 16m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fce7f388ecc34966ac672b41f4b422c0 : SUCCESS in 42m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00c666999b3c444a8cca2223989e108b : SUCCESS in 39m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210658b6a06d41289351232571b2ff1b : SUCCESS in 14m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/088c6f28317e410d9fb2006e16c33880 : SUCCESS in 1h 27m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/242902dee38243a587fa2ecabef81f49 : SUCCESS in 1h 34m 20s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f725e9f95bff4d0ba19031b973f2de9a : FAILURE in 1h 17m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d3dc5dc0acff429393733f33815f9b68 : SUCCESS in 1h 17m 02s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7602c2f34b864a8bb6d131dc0767b067 : SUCCESS in 1h 08m 19s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4a2f76ac8cdc4a579defb28b7e1ae9ee : SUCCESS in 1h 24m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/75231e6de60c4dc6b92115a78ab3178f : FAILURE in 1h 20m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ddb4a7deae04b00938dee750243ab9e : FAILURE in 1h 27m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d924fc11bf8469a8907d3e5ee6d62b9 : SUCCESS in 2h 15m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f656afb45eb422591a024324a30da74 : SUCCESS in 59m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/02a964cb90504579ba203c10f32aa4d4 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4293895fd63848a9db3bc67531ae10ac235971a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-09 22:17:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"57b43d5594c8cc77baf478bd80133935f8b2a47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 12:26:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6fa5003f7e8e40c88e3374cadff72735 : SUCCESS in 37m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e97560ee36684510bf255431e9f4069a : SUCCESS in 40m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b0094fd1615b4d3fa2c98cb31051dbc8 : SUCCESS in 1h 17m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ecac6a0a08b425d911a24b8aee40a21 : SUCCESS in 1h 37m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dda6ea9856034e5cafc6dce8db503ce1 : SUCCESS in 1h 08m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bfd49144396c4e1e80199de46b54e065 : SUCCESS in 1h 10m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deb3a7b05bcd43cc82ee7fadfa1029f3 : SUCCESS in 42m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/112442d667354adf95861c11f47c90f8 : SUCCESS in 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff573c6a547345038b1fa86529433266 : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63b63e0f08004f999cb5d08630536fb8 : SUCCESS in 26m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e131fa24c9994698a790477708a7432b : SUCCESS in 25m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb70b19a69b64e2da7e64f3283662269 : SUCCESS in 8m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1fae91656d24ddebb19c4eaee17dce2 : SUCCESS in 1h 14m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fd51b539aa14d01984426411b149da5 : SUCCESS in 1h 18m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8405aea71f5f499a97adc654639e3413 : SUCCESS in 1h 35m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b133d006955241309773b4fa05da63ba : SUCCESS in 1h 05m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/69c5351d2b934b159d083254714ce78f : SUCCESS in 1h 06m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/21e1b5a1b0cb498facd9bbf8419e74e3 : SUCCESS in 1h 06m 30s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad12f6c928714ed592f00dcea63ec5d9 : SUCCESS in 1h 13m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c232c9733c1042f1acec3f5e93b58f1f : SUCCESS in 1h 23m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2c22c12131e4bc58bd040a77075d89e : POST_FAILURE in 2h 34m 54s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9549ed41e41f4b98a2ffea966cb49f4f : SUCCESS in 1h 01m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/25edff4d12124acc9b33747aa99883ec : SUCCESS in 1h 07m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71be8ad4be7c4d109e07d51405e473f5 : SUCCESS in 2h 14m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16810b7163f44739b79cba492c26ffd2 : SUCCESS in 55m 03s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/de1e01342dae43c394e531274da519fa : SUCCESS in 1h 16m 59s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1d17285d11d94d4a8c7e2e8801c352a9 : SUCCESS in 1h 19m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9056ccbb5ea4433f9c34f2d26af0c83f : FAILURE in 40m 03s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e87a94dc859449e49fa37aae9e83bfc6 : SUCCESS in 1h 17m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b3b9348d517749e683e7dc3aac8addf1 : SUCCESS in 1h 38m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/34aaf7f7dc3e44ad9471818af192cb3b : SUCCESS in 1h 51m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/75d3707ee82e46b4b7b7606069a76d3e : SUCCESS in 2h 20m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b3ccabdc3222455aad3887c40de14d90 : FAILURE in 41m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3f71dd8b7dba43c881d4b9df2290f802 : FAILURE in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13c29c55313fa54781dcf6b1b6c2f64c99ec1be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 12:27:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"478e502cd322346dbf3a45a4cc7b9ac743916925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 14:40:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a4d13afbc8204df783e8937ce2ac7dc8 : SUCCESS in 59m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2a56ed630e384af48f00133976d7b58a : SUCCESS in 1h 22m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/420c7f80a2074c4ca25e2cffffc15719 : SUCCESS in 1h 24m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/545b4d8032bb4f839e56debdd6a428d9 : SUCCESS in 1h 05m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/46537422ac8448d3b20911d9838276c6 : SUCCESS in 1h 11m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89eea1714ead400d8e9e5b0b4d52ec35 : SUCCESS in 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41acc63bec794899af28d5606a6c3f9b : SUCCESS in 11m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40618180d1224320a4fc7f64e3b37f06 : SUCCESS in 30m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e9dc76482e34694b9c50e7f005aec1b : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6a5ffa4cdf4cfbb6c46e118dfee10d : SUCCESS in 8m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4615a9d7733c41cbb6dc4a1abb5a9741 : SUCCESS in 1h 15m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1bcf77ef779405085e94c330108d5ce : SUCCESS in 1h 09m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/373afef819ec4c7bb86bc6658f4a01bc : SUCCESS in 1h 13m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/55ef1aad8a8441abbe23db7ac8be4d82 : SUCCESS in 1h 09m 02s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7fd2b345992b4743ba2fcf2af0b9c4fc : SUCCESS in 1h 16m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/937687be0b5e4850bcdf80ce579b3ac8 : SUCCESS in 1h 28m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d96f464b4359467b94fa7a8b77516cff : SUCCESS in 1h 08m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2859d70808a94caba1849814973107e2 : SUCCESS in 1h 09m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f33dc33d22a040c28dd348c6b020b914 : SUCCESS in 1h 40m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce0c30586f6d4e929f87a5f8e9282501 : SUCCESS in 58m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0b68b3d6bf5845189978288b06424879 : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f6aff5fe174b5c2a861fff042a6e5cf6e352b7c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 14:40:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"34eb9a1a7204cdb837569b903893e8e5736f33a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 14:41:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24bbbf4f02414190b67af109c98357a3 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e31f2ae41b4512afd2b3dd4fb72fcd16ef0a373","revisions":{"6bebfc78816571404cd201d8f636acf018c171ba":{"kind":"REWORK","_number":1,"created":"2020-10-07 10:28:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/756483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/756483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/756483/1"}}},"commit":{"parents":[{"commit":"492c7a3b3339d78121f18ba35fa9896b0e4b8727","subject":"Merge \"Local mac direct flow for non-openflow firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/492c7a3b3339d78121f18ba35fa9896b0e4b8727"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-07 10:20:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-07 10:27:26.000000000","tz":0},"subject":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded","message":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded\n\nIf QoS service plugin is not loaded, the L3 QoS extension in the OVN\nL3 plugin should not be loaded neither.\n\nPrior to this patch, the extension drivers were checked to find the\nQoS extension. Although it is a misconfiguration to have the QoS\nextension without loading the QoS driver, that is handled correctly\nby the Neutron server, disabling the extension silently.\n\nCloses-Bug: #1898842\nRelated-Bug: #1877408\n\nChange-Id: Iea5ff76fe652ab1c04e23850b9259547c1d54365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bebfc78816571404cd201d8f636acf018c171ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bebfc78816571404cd201d8f636acf018c171ba"}]},"branch":"refs/heads/master"},"234707c2e70b5492d3bb333db46d090266796558":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-08 07:50:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/756483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/756483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/756483/2"}}},"commit":{"parents":[{"commit":"8cfc9a412bef8ebdd772d417bcd5234fe7fac793","subject":"Merge \"Remove more empty lines in ipv6 ra conf.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cfc9a412bef8ebdd772d417bcd5234fe7fac793"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-07 10:20:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-08 07:50:22.000000000","tz":0},"subject":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded","message":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded\n\nIf QoS service plugin is not loaded, the L3 QoS extension in the OVN\nL3 plugin should not be loaded neither.\n\nPrior to this patch, the extension drivers were checked to find the\nQoS extension. Although it is a misconfiguration to have the QoS\nextension without loading the QoS driver, that is handled correctly\nby the Neutron server, disabling the extension silently.\n\nCloses-Bug: #1898842\nRelated-Bug: #1877408\n\nChange-Id: Iea5ff76fe652ab1c04e23850b9259547c1d54365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/234707c2e70b5492d3bb333db46d090266796558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/234707c2e70b5492d3bb333db46d090266796558"}]},"branch":"refs/heads/master"},"7e31f2ae41b4512afd2b3dd4fb72fcd16ef0a373":{"kind":"REWORK","_number":3,"created":"2020-10-08 13:12:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/756483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/756483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/756483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/756483/3"}}},"commit":{"parents":[{"commit":"8cfc9a412bef8ebdd772d417bcd5234fe7fac793","subject":"Merge \"Remove more empty lines in ipv6 ra conf.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cfc9a412bef8ebdd772d417bcd5234fe7fac793"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-07 10:20:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-08 13:12:12.000000000","tz":0},"subject":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded","message":"Do not load \"qos-fip\" if \"qos\" plugin is not loaded\n\nIf QoS service plugin is not loaded, the L3 QoS extension in the OVN\nL3 plugin should not be loaded neither.\n\nPrior to this patch, the extension drivers were checked to find the\nQoS extension. Although it is a misconfiguration to have the QoS\nextension without loading the QoS driver, that is handled correctly\nby the Neutron server, disabling the extension silently.\n\nCloses-Bug: #1898842\nRelated-Bug: #1877408\n\nChange-Id: Iea5ff76fe652ab1c04e23850b9259547c1d54365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e31f2ae41b4512afd2b3dd4fb72fcd16ef0a373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e31f2ae41b4512afd2b3dd4fb72fcd16ef0a373"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
