)]}'
{"id":"openstack%2Fneutron~756922","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I22570a44f84d14a792633747c04d7426ab231009","project":"openstack/neutron","branch":"stable/train","topic":"bug/1896945-stable/train","hashtags":[],"change_id":"I22570a44f84d14a792633747c04d7426ab231009","subject":"Default dnsmasq --conf-file to /dev/null","status":"MERGED","created":"2020-10-08 19:29:25.000000000","updated":"2020-10-10 13:25:45.000000000","submitted":"2020-10-10 13:24:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756922-1602336241560-75899b5c","meta_rev_id":"8deb665fd2c65709d6a51cbfff09575a1c88f0b9","_number":756922,"virtual_id_number":756922,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-10 13:24:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-10-09 14:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-09 08:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-10-09 14:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-10-09 21:07:28.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 08:43:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-09 14:31:53.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-09 21:07:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-10 13:24:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab1a8159ca6ece20886330d4154682ab50bfc4fa","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-10-08 19:29:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2de42a8f3be0f69e4b90d087d079ceb17cb68d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 06:28:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/9f3276abbce54d70b9872f134891c452 : SUCCESS in 58m 41s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/3d72dd3c90444be2add9795d7078069d : SUCCESS in 37m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/35fad51195b14fc98df4d2462193c1dd : SUCCESS in 1h 22m 47s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3e696d46f7684aae9fb5624e6d76fe0a : SUCCESS in 1h 15m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/986fcf839ac546b39d3156532d520c14 : SUCCESS in 1h 29m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9acca9faefb74599a456d39898131756 : SUCCESS in 35m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ec0513ecd9a40ecbf06356be51ea36d : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe4352f3aed4fa99004e7715081ac3c : SUCCESS in 11m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/747ed1fc3bc642b39401d32a5089207a : SUCCESS in 21m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb97bd91066c4a969e0f6709fd1a124e : SUCCESS in 20m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed1889ab0f024f6baf649865fbb6d66e : TIMED_OUT in 41m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f4ed1711844c36af6c160527edb95a : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/01d19d5779c943dc9ff18396b3873475 : SUCCESS in 42m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2f9aa72566a54516882b20cba4b328d7 : SUCCESS in 44m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9b36a8d2a02b434882b69bc931c2a761 : SUCCESS in 48m 14s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/aa84a80bea83460baa09af90df4e2ef7 : SUCCESS in 42m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1a2d8d3b89744a2eb3db9e1383061998 : SUCCESS in 1h 15m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b5a55a7315e045268301100709acb8d2 : SUCCESS in 1h 49m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4172a0f0b70c4cf8bbd77888641ef41c : SUCCESS in 1h 14m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cf1254bf6aae43a9b0427b6000f82ea4 : SUCCESS in 1h 10m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b44ec17211e04444bfebe9865fe5ab01 : SUCCESS in 1h 07m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ec9d57601d6442e84b6cbf58678d87b : SUCCESS in 1h 29m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b2b0af4814594d7b94b03710a43c7fc4 : SUCCESS in 1h 46m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f8a6721ee73740699f4fb8f6b2c18eec : FAILURE in 1h 37m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac683a1b756c40618c9112bdfd1dd880 : SUCCESS in 1h 11m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c7df934ddf7b4eaa8bc9ed8316b900f7 : SUCCESS in 58m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dec6ae65c0f446048ac0eca4de4108b6 : FAILURE in 59m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/943f9c47fa56493c9e39b01774b104b7 : SUCCESS in 2h 01m 27s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9beeb7ffaae241129e895e93b41bc75a : SUCCESS in 1h 39m 26s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/47f229433a684e81b41f535fa1bceaf2 : FAILURE in 23m 44s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f021e370b5fa4f98a200666c35b10d12 : SUCCESS in 59m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/253b60ab37544ac6bfbaca22265b14b4 : SUCCESS in 57m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/617e0ce6e551479f9ab7a8fe7d163f74 : FAILURE in 42m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bafd42b3769b42a0b8be725403300d8c : SUCCESS in 1h 09m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c162e152dd394a41a05ec2ba57b4dc61 : SUCCESS in 59m 22s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/66e489103cee4b16941a37b03e50942b : SUCCESS in 1h 01m 20s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a5dfd680f8fa4e4c9278932ed7143dd9 : FAILURE in 1h 18m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4a55d61894a83879725adb8d4cd679bfd6398b2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 08:43:05.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74ffcbb027b71a6798911d4551b5ec71338d89fc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-09 14:31:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nUnrelated tempest-slow-py3 error in first CI run","accounts_in_message":[],"_revision_number":1},{"id":"f9db98a4aabf6a3a0cd0834723f1a9d2efdc09da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 14:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a36ad5ddd66e48038afb8d31d9bf2cac : SUCCESS in 53m 05s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/3b054d6862ee46dc81e723fae109f4f9 : SUCCESS in 37m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9fdd6aeba0fd4346b0d17b2cdec9a468 : SUCCESS in 1h 21m 23s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3fac2d80fb484f23b7bb466c90a92aaa : SUCCESS in 1h 30m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/924ac4fc170f4247b61a5a0e641e26d4 : SUCCESS in 1h 29m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cf482f2f6ee493b8e314eb5b2d3dd27 : SUCCESS in 28m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e215212dd2747fda56dadccc93bde41 : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8349e59bd0c645529a7260e9b0c2b592 : SUCCESS in 12m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33478684ef5b4c2fb96afdb9d75be789 : SUCCESS in 23m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4f9599c28ee4edd98301f386d1d9921 : SUCCESS in 20m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d4e0ed5f4414e418fa085bcc14fbd72 : SUCCESS in 21m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488cc8e8b47e4488a12ee142bfc206b3 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2ac526b7abab40a1801d218437627148 : SUCCESS in 1h 02m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/799eb9026ae346e7ac0f1b2fab2f0984 : SUCCESS in 1h 13m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e9ce681ba42c44c7b5c093b33671cd9e : SUCCESS in 43m 24s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f928106e60164774807439a681d843d3 : SUCCESS in 36m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/94ac3c8a65874a2e9e4265373adf6678 : SUCCESS in 57m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c8788e7984264855b308b5c72a079564 : SUCCESS in 1h 33m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/919284ffd5814d96b941b32125f0c239 : SUCCESS in 56m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cae9e43faa4f48948ea1eb0bce606d23 : SUCCESS in 1h 09m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79413be330974b77a4eb210c191dddeb : SUCCESS in 1h 16m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/15cb8c95e8aa420c833ce5777eaff31f : SUCCESS in 1h 12m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dad4b45f81c44c3684e893960f086e68 : SUCCESS in 1h 45m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/17f9d36b762145f1a70ee2d430b3a008 : SUCCESS in 1h 42m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a46c7cb9e544455cb98d654c12498740 : SUCCESS in 1h 11m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b6a06aeada944cebbb99ae9fd2f92c2b : SUCCESS in 58m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9079ed49d842407a8e99201175c55ba6 : FAILURE in 51m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21a0b70a95204ba0915642e58341b328 : SUCCESS in 2h 01m 57s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e0349de235ef40ce9e2e2edeaba83c55 : SUCCESS in 1h 45m 36s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/19837abc556849e7a604c592cc1f27b4 : FAILURE in 22m 01s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4bfee59a2f3b4e02a6c60f97b8f3d9d9 : SUCCESS in 1h 03m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/645c7d2cf9b047d3a300b226f4858c05 : SUCCESS in 55m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0d6b714a19d54fad8d98d1358b458cda : FAILURE in 55m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9a15855cd354e43a8b743e399031f4f : SUCCESS in 37m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcf1130c2450455fafb82e28e7400c7b : SUCCESS in 59m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b233ad2de8fa4c17a0cd82d60d4c7aa4 : SUCCESS in 1h 14m 13s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6a6a6471062c430b95275d5a179d76fa : FAILURE in 1h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0991518509ff719b4624252143af498770bdfd53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 14:40:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30a9ed5c0e9b15ce9d35b136cae4206efaf9429d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 19:26:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/e7156a52e9464b7eb85786992e5f0134 : SUCCESS in 1h 05m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/492ceae1e69c4ed19ed4c6b8b9aadd17 : SUCCESS in 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a10a53865c14620873e3407a5ba39cc : RETRY_LIMIT in 11m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3b7cb8d1385444ba7b601131e74acf2 : SUCCESS in 39m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8941644c377a4c9897a6e55d29ba50ea : RETRY_LIMIT in 16m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/503465b5bab34b47b9568d2c3109a250 : SUCCESS in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800c9b40368347778dd3d80a1e8e04bf : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/90938e609a0a405cae15c9350b18933a : SUCCESS in 1h 24m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/28439cdedcd843c89de4d1256d0dd3a5 : SUCCESS in 1h 06m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3aba9c4b001b4d66a76df6867176da1f : SUCCESS in 45m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b5b4e6672fe149ef8660ea1b1610d270 : SUCCESS in 45m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7aa0bd372e4e4b648ce8150a9dd49c59 : SUCCESS in 1h 07m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51277de45d694e03b4eb31a63c486c96 : SUCCESS in 1h 18m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7849e161dfa14600bedacf57bc497b51 : SUCCESS in 1h 23m 50s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00e1173b758a4ceeb9740c4f5afe9931 : SUCCESS in 1h 18m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce8561b23afd43dc88470afbe326b0a0 : SUCCESS in 1h 25m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/15f5311256d543c28c57f08a61e4f13d : SUCCESS in 1h 15m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3523d958545e472b8cc32e9bd826b457 : SUCCESS in 56m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d828d0707ac84d1cb85df318d75ebee5 : SUCCESS in 2h 03m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/94a3907e6fa1446481501c4453a3140e : SUCCESS in 1h 05m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b703054248594dbebee688c7cdde20d1 : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"00858f348cf311ac8568be50d4b107cafc5e85b2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-09 21:07:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"044354e122fae7c775f59ac8d304cab96c379999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 11:51:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/5e2269fc5909474092908b4769b5bf2c : SUCCESS in 50m 01s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/f2dca5d931884c31a23b7bf6042935e5 : SUCCESS in 35m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/370a9ae74ffe4f8ca8e5bc7ba6117004 : SUCCESS in 1h 16m 08s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/6a01e78f41ba450d903d555d8e2aebb6 : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/67d9d51b4d354924a2e1a2da8d471350 : SUCCESS in 1h 22m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53549bfd52c44a8ea1f9bb5c3e4eb197 : SUCCESS in 27m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae6efb627eac49b6ad525b23afcb14e3 : SUCCESS in 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dae13f8df6747758a6de900ad3bbc44 : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0cf9cb756ae4acd9f01a9d272e18e4d : SUCCESS in 23m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7aff34421234538bb2aed44913163a9 : SUCCESS in 22m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/277c82de39604c6fbc3d1c2992d8c89e : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5ef9df48ac485eae9761b2c076e89b : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/3ea6a7724c3e4530afddb3fd2412ae5d : SUCCESS in 56m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/224b0ec53a1941b2acebc2d3b98ab2c7 : SUCCESS in 1h 09m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0eb6de31a60945baacba64f5667dc301 : SUCCESS in 41m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b02125eee5d24e24abe54f02b8dbc7b5 : SUCCESS in 42m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/72f8d6e2c0c6494487371aa6d3c8db22 : SUCCESS in 1h 02m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/55665bc792774700969783a3f7f12e04 : SUCCESS in 1h 39m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f6daa790b7854e2ca1b05ec56ce458ec : SUCCESS in 1h 12m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a571e7b8505c4c90b71275cfd824b343 : SUCCESS in 1h 00m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19def8a13d7041ea9f83275034f1f3d8 : SUCCESS in 1h 19m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ab88052de26d4ebb8542babdc023e095 : SUCCESS in 1h 09m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc73f06442384ec38085d6c3dcae9e49 : SUCCESS in 2h 01m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f43425ba991f45dbaf8387f8357498e7 : FAILURE in 1h 36m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d0c09097836a4ca1894adff9dd285a6e : SUCCESS in 1h 08m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4b167e47907545cfa2bc9dba0b3c6bfe : SUCCESS in 59m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ff196a28caab4d8bbf2c9f0518679fd7 : FAILURE in 59m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3daf77a0b87407fb4ea2e7e212937d6 : SUCCESS in 2h 08m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/030b85c0f0df4fe59cec56d5d2b84945 : SUCCESS in 1h 41m 01s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ea5ef4414fc84b3497f8e8ce3d96cbb0 : FAILURE in 16m 03s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eb337e819d4b435d81006cf7120691e4 : SUCCESS in 57m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2dc81b270624536b07b0a30c180bdd3 : SUCCESS in 1h 14m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/18c4d4c97fe04842bb241f1d22f8acf5 : FAILURE in 42m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/745809137d1d4730a6a5bd5681ad171a : SUCCESS in 38m 18s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7280effd26bf4bbb88cc19057c519b0e : SUCCESS in 1h 08m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9702336dc9148bd87018f32f6a768f6 : SUCCESS in 1h 05m 54s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/caab42132809484f9cf64ac830de8fb1 : FAILURE in 1h 57m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62a954e65d0a6adf0113d96196cc2d4646fed193","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 11:52:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b84cfe8f9d881b37ecff175c69d93c2e7f967a16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 13:24:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/db5993e9d6954214829a62c187c8b467 : SUCCESS in 37m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91328a8315834ad8a5b848ddc66b7380 : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d143f8af7914f388e36538ef963fb94 : SUCCESS in 11m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d55f1bfe3a74b14aaf78830e7ab482f : SUCCESS in 15m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc3ae89b67e6442a9846831ac67a2a23 : SUCCESS in 21m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c4ae4a185e746a7bbf834e5a0f9542e : SUCCESS in 15m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1882400bce0b4be28e3fd815a30bd328 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/37924a00fb9e433fa0ba0b16be20ba07 : SUCCESS in 48m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/567297461c2145e8b957947df85ae148 : SUCCESS in 1h 04m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/371273d6b6a444cd8b2474fa4973d81e : SUCCESS in 46m 09s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/42bf182b1bb446fd9ae2636ccaca1dd1 : SUCCESS in 35m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a41f3defa80b4e3db3e9f5eea12e558c : SUCCESS in 1h 06m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bc185f383e324417abe6798485c22e8b : SUCCESS in 54m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/39aedbd9e033449c9f819968b9ef6795 : SUCCESS in 1h 07m 44s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/17635f62dae544ceae43f2bac4ad0f65 : SUCCESS in 1h 04m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/63d2fbc916fb42348e1d688c6dc9f8cf : SUCCESS in 1h 06m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/890881dea9df43e88cd9239b0aecc30d : SUCCESS in 1h 18m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da33a057a70f401fa42ef243317f03c7 : SUCCESS in 41m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fe8cc239dcb4a048434d4c1d0b9bf7f : SUCCESS in 1h 28m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7b0a8e8342194ecb80e82ab4a966aa98 : SUCCESS in 1h 01m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb6ba92d5cf458b8a70fb336ecde841 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"07c1ea648f52a43ab5c74ea3349d5af784d73d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 13:24:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4f5adfe4cebab780a52616caf9242664b4da7fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 13:25:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f870624ab442308ac407d9efe4e4b3 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8654eb2d183cf13699e50b8b967a6c52412602ac","revisions":{"8654eb2d183cf13699e50b8b967a6c52412602ac":{"kind":"REWORK","_number":1,"created":"2020-10-08 19:29:25.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/22/756922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/756922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/756922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/756922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/756922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/756922/1"}}},"commit":{"parents":[{"commit":"fa5967012d201a7bc3a407280fd7481f5dc81574","subject":"Drop invalid rootwrap filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa5967012d201a7bc3a407280fd7481f5dc81574"}]}],"author":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-09-30 15:00:07.000000000","tz":-240},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-10-08 19:29:25.000000000","tz":0},"subject":"Default dnsmasq --conf-file to /dev/null","message":"Default dnsmasq --conf-file to /dev/null\n\nPassing --conf-file\u003d with no value has no effect on the dnsmasq\nprocess. Intended effect here is for the default system dnsmasq.conf\nfile not to be read and included in configuring the process. For\nthat to happen some value has to be passed to --conf-file. Passing\n/dev/null will invoke the desired outcome to skip the system\ndefault conf file.\n\nCloses-Bug: #1896945\nChange-Id: I22570a44f84d14a792633747c04d7426ab231009\n(cherry picked from commit 704576e54e041340ed9c2964b110815e074239a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8654eb2d183cf13699e50b8b967a6c52412602ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8654eb2d183cf13699e50b8b967a6c52412602ac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
