)]}'
{"id":"openstack%2Fneutron~757049","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I69356e926b15de7f1f99540e7cb98671c634e8a9","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1861674-stable/rocky-stable/queens","hashtags":[],"change_id":"I69356e926b15de7f1f99540e7cb98671c634e8a9","subject":"Support gateway which is not in subnet CIDR in ha_router","status":"MERGED","created":"2020-10-09 09:39:08.000000000","updated":"2020-11-12 02:31:41.000000000","submitted":"2020-11-12 02:31:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757049-1605148301747-96dc3111","meta_rev_id":"104b863e555822aec3bba61dd3e93d5935e5d37d","_number":757049,"virtual_id_number":757049,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2020-11-11 22:27:33.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 02:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-11 22:44:27.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},{"value":2,"date":"2020-11-10 11:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-11-11 00:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-11-11 00:21:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-10-09 09:39:08.000000000","_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 09:39:08.000000000","updated_by":{"_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},"reviewer":{"_account_id":10518,"name":"Jesse","email":"jesse@easystack.cn","username":"jesse"},"state":"REVIEWER"},{"updated":"2020-11-10 11:13:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-11 00:21:24.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-11-11 22:44:27.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-12 02:31:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a21b8e4615ec335b9fd4482ed7484e58660644ca","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 09:39:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"327fb5fcc3ef8c477ddcfdc99133d9effe01f6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 00:06:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/6e11fc3727284f138101786379086c15 : SUCCESS in 50m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/3933f7d71d8f4ea9842f746a3d138176 : ERROR Unable to find role in /var/lib/zuul/builds/3933f7d71d8f4ea9842f746a3d138176/ansible/pre_playbook_3/role_0/neutron in 48s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/3c0f631cfaa649439b92a7299adeca81 : FAILURE in 2h 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff84f8aaed9a450b96063651f1f3571e : RETRY_LIMIT in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97a1503222ef4de1bd1a77239293942c : SUCCESS in 25m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e81d0aa8b074c65b418a0ec8d357a4f : SUCCESS in 25m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5d8cace5592740469005484019618fa2 : SUCCESS in 6m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb701e1d04ae45cead5ad4290c5ce0e1 : SUCCESS in 1h 25m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e578a90578534d6c9da81d43c32a7e9b : RETRY_LIMIT in 8m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f6f1f5cdb214c94ba6e40bc4d57e1f8 : SUCCESS in 1h 25m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/32a6b4541b4441b395126a0f553366a5 : RETRY_LIMIT in 9m 12s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/9ac353a5bbc84c86ae589cd5780d5765 : SUCCESS in 1h 38m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51a748aabc754fdcbff6d2f882b1a994 : RETRY_LIMIT in 9m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e3e44e2f16dc480493a733fb711b4849 : SUCCESS in 1h 21m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ab68b0569ce447b8cb4ce10c4ce1359 : RETRY_LIMIT in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e96d371b9a661f1403987e09a5d61f04583de0b1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-11 16:15:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cad016c8806986206b5dbe9b4f683a613e4ecd75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 19:11:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/8f018e1a31ac48a3819be6c5810fed7c : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/ae21eb868730473f84143b86bcaca987 : ERROR Unable to find role in /var/lib/zuul/builds/ae21eb868730473f84143b86bcaca987/ansible/pre_playbook_3/role_0/neutron in 1m 11s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/426db63c164f466580be1bd4a85f209f : TIMED_OUT in 2h 49m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/977f0837c40d45cbacf037694c7d1f55 : SUCCESS in 10m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09fa01e2216a417eb7669affc9ac097b : SUCCESS in 22m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/49c0e48295d141428d483535c0b8feb0 : SUCCESS in 24m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/67ae7509d9724f95b882acd88e64551c : SUCCESS in 6m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b480cff1b0849d08d2ddc4f5173fb6f : SUCCESS in 1h 17m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/521798fafeba468781f8b73d2609d913 : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f224681bfdc43d9bd5179235577420f : SUCCESS in 1h 16m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6f3f9b94f7c4a29b5deb1ba7284a012 : SUCCESS in 39m 29s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/663ed062af8e497f8d6d0922fa8d2b46 : SUCCESS in 1h 51m 25s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2eccad39294b41c4b6eca5a1b9fe1509 : SUCCESS in 1h 22m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7ea209cbe00547c9b160ef0209291856 : SUCCESS in 1h 15m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b92689eda84c98946dc6b5644c5b1d : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"cb525a9960e387e9c68dbf7c5587f1d923433974","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-12 08:22:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf7fb5014f23022ae4ddd0a8f1ed697668369efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 18:19:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d238155945654961ad2e144df89bc885 : SUCCESS in 54m 41s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/52a5f85dc63d440ca188b9cf42e5268c : ERROR Unable to find role in /var/lib/zuul/builds/52a5f85dc63d440ca188b9cf42e5268c/ansible/pre_playbook_3/role_0/neutron in 1m 04s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/142c0bf0a5aa4b52861c8ed5c82f4abd : TIMED_OUT in 2h 51m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd338498bfb74bf287a87652a9d67c66 : SUCCESS in 11m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ead80a0989c4d60aabff222e4078b4c : SUCCESS in 21m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a406572755147c7808150483f983e67 : SUCCESS in 23m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/793eaeafa74a4383bc1726dc73155f47 : SUCCESS in 6m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d240eec59084655a25f1825c9085566 : SUCCESS in 1h 10m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/be942a96efff4f11b2d21b73f5a76a6c : SUCCESS in 51m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5744120558094ccf9231d93f08057763 : SUCCESS in 1h 09m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d1cb32c05e6406c9d0dc72966387453 : SUCCESS in 50m 00s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/be017422dd1f4e1abf7b74780594945d : SUCCESS in 1h 42m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bdb89e22e84f4c71a73b35779c6a6993 : SUCCESS in 1h 20m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4bb8feb4c4ac4f108c2f13ec92d1c056 : SUCCESS in 1h 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d925ef08e474c35af3a6d4df700802e : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"768499814b9bbdd4d8c942acb82cdcc6bfdb8494","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-03 07:58:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"023bd9d1a161ddfbd03657d6217a145791c68b55","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-03 08:02:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757049/1 : FAILURE in 46s","accounts_in_message":[],"_revision_number":1},{"id":"161e8e21c6d76ba85b651a93a84d034375349743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 18:46:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/9a9afd6b3cd84b5d951a9409bf35deb6 : SUCCESS in 44m 29s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f0ade2cbca38446192028a2a3c7271e6 : SUCCESS in 57m 19s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/b96e6868064c4478addcafa59601e593 : SUCCESS in 1h 02m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8837495111455084d875c29b1fcdff : SUCCESS in 15m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ed9e1e13d6048f5bdf8433bf62530ee : SUCCESS in 22m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c546712c0ca449bcba3b333a3618a801 : SUCCESS in 25m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3eeca25358d94f76b4e7786353360c0a : SUCCESS in 7m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c587807d7f804975b70d43c1370ca6b9 : SUCCESS in 1h 19m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6943bcd25c64fc1bf87eac0eaf824c5 : SUCCESS in 1h 01m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76d403a7309d48ccbb4de47a28a8d6f3 : SUCCESS in 1h 17m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c0808b1a0bf445b89b404ffc9d36ba68 : SUCCESS in 37m 19s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/22f6d165b1344c4fbef71de360b128ee : SUCCESS in 1h 29m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2086ebcefc534ba1a32108390f0e0180 : SUCCESS in 1h 34m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c103f4cc597d45c0867cb3aac84d3712 : SUCCESS in 1h 29m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c64d8015dc74bc3982a231ad3789be7 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d639739a4e15fade816121c94272ab8236cec8fb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-10 11:13:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f212aea1af19878cc3b43856d134a9ceff7e1a15","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-11-11 00:21:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f89f9e76af193d8dc448862351a1b490ce6d1b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 00:21:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"226d2f3f8366c2b77f3a3936861bc33d364d3b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 06:29:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/702e1225d0d84831b3142f3051f86e88 : SUCCESS in 1h 01m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1738e1b34147a4a2a08ee07b30ff2b : SUCCESS in 12m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c14e5f3c208343619c81bc253a950902 : SUCCESS in 22m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c943c4351bc1493b9d43a1a3f3136673 : SUCCESS in 22m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dcd26d4445da4abd966ae3a02c9c924a : SUCCESS in 6m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e08b41980074700ac4e2ef483de7058 : SUCCESS in 1h 26m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4822c8abe4664acbb7f09157fd3ef61f : SUCCESS in 56m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c11917553f29442c91726d6e81f112b4 : SUCCESS in 1h 27m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4cc126f990ed4f378156ed7a8acd7cc0 : SUCCESS in 35m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e3a87d49fef249a1a839399235a58198 : FAILURE in 1h 18m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9a1b19a7a62444b1af9c7e68edb428fc : SUCCESS in 1h 22m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1eb0e3a016e475cb79a66ad9ae52290 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"27b4b7e66d2288b7049102278dddffebcea76198","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-11 22:27:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8acaf8ab63fa72e7e0eace25a59fb60508bbd72f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-11 22:44:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757049/1 : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"467dac7027a86eaf9a6f03c07fb88410bf1c1113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 00:50:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/10a9168544b5482ea038552a29775425 : SUCCESS in 52m 49s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/195a4ffa39cc4c76ae02e99aef1f72af : SUCCESS in 1h 05m 38s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/defc2aa613374cd897b003b09b01a91f : SUCCESS in 53m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14fe92b3730494a818902b4e384b212 : SUCCESS in 10m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15271c926b564c318858d513cf750f8e : SUCCESS in 13m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fc7ddeed701445f0ae989f3466eba400 : SUCCESS in 29m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/54f217d8df584a5e904aa672bbf168ea : SUCCESS in 7m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a08593e1ffd145cdb5090f87cc847920 : SUCCESS in 1h 08m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b1f64b273e9b4a19ba26d0f00ff7218a : SUCCESS in 1h 03m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37abe3a8b07342de96f61f70558ef921 : SUCCESS in 1h 28m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c8ce99f11f94f6186456612b8ca80dc : SUCCESS in 43m 13s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/553791a65d134969b9489ba9a5feaf21 : SUCCESS in 1h 10m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b1248ecc225d4a3389880f7edd8e4e5e : SUCCESS in 1h 23m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/099dd228272e472bb664cd5d747cfecf : SUCCESS in 1h 16m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abe6ddfc7b354610be45c95ae838f5b1 : SUCCESS in 37m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e575bc3ab1cc350d82c274a85bf239f646d8f54a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 00:50:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a6718471cc9d19691b16291688d14ae4f0ac84e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 02:31:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"76d0ea2790422232c8133749a465883f3ae60190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 02:31:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/dd0d90b3d7df4ab08362dce32f13b1bd : SUCCESS in 48m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b73e583bc5114ef28ef1a82613d4efd7 : SUCCESS in 11m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac50cd67fd9a4df1bc8d2cf4060c7752 : SUCCESS in 20m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ed2ed54cd5444a9697de34b2ff3966fd : SUCCESS in 24m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/96fb6e8cf4dd4adfa68c5ad5e0d4fa84 : SUCCESS in 8m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b3ac1e5c18134ba09d59e8d5081d6806 : SUCCESS in 1h 31m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e6e98585755540698a4e4b771b88b721 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64fde24de6794a059398493c72ab2a6d : SUCCESS in 1h 13m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/381c3176792c47638d1da280be280893 : SUCCESS in 41m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ff24f0447d9149c8a1528d2f0059812a : SUCCESS in 1h 17m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/587c35a3e2a244719deb190ed2b23d3f : SUCCESS in 1h 32m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd3d5380c6b3477abb81121399265320 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2acb249c804cefc411ed6586b694e21e3271cc2","revisions":{"c2acb249c804cefc411ed6586b694e21e3271cc2":{"kind":"REWORK","_number":1,"created":"2020-10-09 09:39:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/757049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/757049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/757049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/757049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/757049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/757049/1"}}},"commit":{"parents":[{"commit":"5ee6e5bb31179acf6e86357b0b0b34ef051af161","subject":"Do not fail deleting namespace if it does not exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee6e5bb31179acf6e86357b0b0b34ef051af161"}]}],"author":{"name":"jufeng","email":"jesse@easystack.cn","date":"2020-02-03 06:54:12.000000000","tz":480},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-09 09:39:08.000000000","tz":0},"subject":"Support gateway which is not in subnet CIDR in ha_router","message":"Support gateway which is not in subnet CIDR in ha_router\n\nThere is case that gateway is not in subnet CIDR.\nWe can set 2 routes as follows to support this:\nip route add 172.16.0.1/32 dev eth0\nip route add default via 172.16.0.1 dev eth0\n\nConflicts:\n    neutron/agent/l3/ha_router.py\n\nCloses-bug: #1861674\nChange-Id: I69356e926b15de7f1f99540e7cb98671c634e8a9\n(cherry picked from commit 554b5c226750ec75e99ebce241df8b7a700ff016)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2acb249c804cefc411ed6586b694e21e3271cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2acb249c804cefc411ed6586b694e21e3271cc2"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
