)]}'
{"id":"openstack%2Fneutron~757055","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I00c810fe541ac5e1e9923155fe90eb07a0b4b3dd","project":"openstack/neutron","branch":"stable/queens","topic":"terminate-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"I00c810fe541ac5e1e9923155fe90eb07a0b4b3dd","subject":"windows: fix terminating processes","status":"MERGED","created":"2020-10-09 09:41:34.000000000","updated":"2020-11-10 02:21:34.000000000","submitted":"2020-11-10 02:21:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757055-1604974894834-40a8ff94","meta_rev_id":"1089bab5aba5c11f1e0a7cc948268a0679cb16c8","_number":757055,"virtual_id_number":757055,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2020-11-03 07:58:40.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 02:21:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-03 19:29:38.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-11-04 08:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-11-10 00:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-11-10 00:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"value":0,"date":"2020-10-09 09:41:34.000000000","_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 09:41:34.000000000","updated_by":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"reviewer":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"state":"REVIEWER"},{"updated":"2020-11-03 19:29:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-04 08:10:30.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-10 00:08:41.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-11-10 02:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"420746bd56b37a9ee8233ad7df30b75788ff4c88","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-09 09:41:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"85f7ad8c41322394c91ad65dee132f1f47943090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 01:00:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/983dbc8ef7a7432c87d158892bf6a8e6 : SUCCESS in 45m 37s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/5b37eedfe08e4e04a9f47ceaca7f63b1 : ERROR Unable to find role in /var/lib/zuul/builds/5b37eedfe08e4e04a9f47ceaca7f63b1/ansible/pre_playbook_3/role_0/neutron in 40s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/1dce69519eb34503a517391c3649b241 : POST_FAILURE in 2h 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c32edcae574fbe99a014eb9ff2f67c : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e5f5443559d4e99b65f30a385f7a101 : SUCCESS in 38m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2472f60a97bd4d84b52b6537521e5d90 : SUCCESS in 41m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7219249f7ef424caec28b3e1076e2f1 : SUCCESS in 5m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d9839fc681d42479e10c7410c6019b9 : SUCCESS in 1h 22m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c8c14e11604c43f9b993489b86b28948 : SUCCESS in 41m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1b87a4b67b040cd9a1008e23ace7b56 : SUCCESS in 1h 29m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7eb2e10822aa4473aeaa7b7aba26a23b : SUCCESS in 40m 48s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/1a22cc4071dd4dd68eafbf19ced34191 : SUCCESS in 1h 42m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cc6a9ebe8038459cb5a9350b2c38f701 : SUCCESS in 1h 26m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/20cc941852b745a19a39cc6dcd60733d : SUCCESS in 1h 19m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebed9077d19c4162927ec1380f7e7a62 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6f4ea50bdfb0bf10771be7cd748a4527a60400bd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-11 16:13:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85d44e5fd5377bb8fc480dbad54ceb86418eabfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 18:14:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/8c15deb4ddfa4ff99be0fe3991922119 : SUCCESS in 44m 10s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/b6b059d323394cb2a6a61a33a81a9d85 : ERROR Unable to find role in /var/lib/zuul/builds/b6b059d323394cb2a6a61a33a81a9d85/ansible/pre_playbook_3/role_0/neutron in 53s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/968886950f7a467bb70d32de7b58460a : FAILURE in 1h 58m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4757465c0e04e1d86780f0e2a044195 : SUCCESS in 11m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e30326e9c7342c3898b74339694deb9 : SUCCESS in 21m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c4b6946996164936a40efb2b0fd264a2 : SUCCESS in 23m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d90058c4354f428e94f4e99be09afe9c : SUCCESS in 6m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/21f01437328a4046a0e19689d65496a0 : SUCCESS in 1h 35m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b37055e9adb47dea1481e017f1ee6ee : SUCCESS in 55m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cba1ed4c8d97402b8c2a12374ef7e22e : SUCCESS in 1h 02m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8661f6323add4324b2799138456393a9 : SUCCESS in 33m 44s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/c66a201462d340079152301815a0dfa0 : SUCCESS in 1h 39m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4b9abccab8b04fffa527d7fe049c6a1f : SUCCESS in 1h 27m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/13851ca5890e4ae1a3d471fae018fffb : SUCCESS in 1h 24m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e49888b378734ee8bbd568d21e3c6614 : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":1},{"id":"efb00009845abaa0a045f5a7e7be05f40c7b3790","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-12 08:23:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"288c1a621b1724b4abbf1f74b9775a29ae68d800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 20:40:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/55028d903b9b4258b7d7008088f3a20b : SUCCESS in 55m 39s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f98bf366237a4338918a66698a9d45ec : ERROR Unable to find role in /var/lib/zuul/builds/f98bf366237a4338918a66698a9d45ec/ansible/pre_playbook_3/role_0/neutron in 1m 34s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/0cab8db169cb420ab36fa0c0745cb9f3 : TIMED_OUT in 2h 51m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09bb133db5204a768e42db0f379a44fc : SUCCESS in 11m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3faaee41aad476194c86b4d80cb79ea : SUCCESS in 22m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/251d874310d044ee93f4e2a2d7640516 : SUCCESS in 16m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b1a848fca65a4805975ca2337efcd405 : SUCCESS in 7m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4530e4b224b6457a8391860435350ce7 : SUCCESS in 1h 15m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27f3161d03334493a98b5a699949cfa4 : SUCCESS in 58m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/973fb4df3b5c4b999af4d36be5da988a : SUCCESS in 1h 28m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/41cea80d92714939afa7d64f8b9c4d68 : SUCCESS in 41m 46s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/bba8c656b4ff4f1089029c0ba257cbe2 : SUCCESS in 1h 42m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/670a08a4789a44608bbe2c9e8b51f323 : SUCCESS in 1h 44m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0b10f03900f04848b7e56a14db090f73 : SUCCESS in 1h 23m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/865d155b455b4c1c950a652e50192dba : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"07462c49a1c06a5ad15a522a6ae95768249963f1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-03 07:58:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7d39b58ffcf03848d3f7b4d3d18727c2b1624e52","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-03 08:02:20.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757055/1 : FAILURE in 52s","accounts_in_message":[],"_revision_number":1},{"id":"fa2e01ee968c1f4ca6c8e0fb38469ce17e8b2adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 18:27:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/ad0f14c6b3c244eab275eed74ddc1b6a : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/9ca02c15b8934ad08da778ac20c6d60e : SUCCESS in 53m 02s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/c02014279e2f4aac9117649d6bab7e59 : SUCCESS in 49m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1f1de65ddc43daa193524da03ffa0a : SUCCESS in 10m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa1a8569411b4ce69c696c9629e7754e : SUCCESS in 24m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b79d5ffb6adf44e79693d32a7e869c17 : SUCCESS in 38m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/57daafc2e82e439c92239f99c5d185f6 : SUCCESS in 7m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4364a781488d4323bb5d387fcce4de6b : FAILURE in 1h 43m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/14474928788b44919fb002ccee0329ea : SUCCESS in 1h 03m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7117159537241a19b3e1382f328d46a : SUCCESS in 1h 15m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/30d1c5e0ba064b7db615252b94768535 : SUCCESS in 38m 21s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/bfc72f4708fc458d81f9d04ef97b5809 : SUCCESS in 1h 34m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/90667f5498944a2188da56a5f402866c : SUCCESS in 1h 22m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5b4d0359b4dc4efbb1379e63db002995 : SUCCESS in 1h 22m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ee2b87c15de448f8ff124ca55e6afc1 : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a40628feb20cb6153916f98f98bb231268596393","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-03 19:15:09.000000000","message":"Patch Set 1:\n\nrecheck tempest-full","accounts_in_message":[],"_revision_number":1},{"id":"20bd2be70002c6647a212499a3440c47af59450f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-03 19:29:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757055/1 : FAILURE in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e2803140acd7d2f25612bbd9519774b1be030e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 05:50:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/a5268bc6d8484bf58ca64cb2ed2be44e : SUCCESS in 49m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/159a73a4d4e146c7a83a4f2204621e57 : SUCCESS in 50m 56s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/196fbb6741e748eaa966ef140f062973 : SUCCESS in 51m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8853b2799346d5be528ae23849064b : SUCCESS in 10m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4bbf71c2c6444748d203fc15fa9d535 : SUCCESS in 24m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f1cb736dd714fbd976566b597bc7795 : SUCCESS in 25m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/25918486685d40f18da2b241e44bc53c : SUCCESS in 6m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6fb9202d8b8d4583abc496275e9bf024 : SUCCESS in 1h 25m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6c7e1126585e445986dc265aa4112de2 : SUCCESS in 54m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ff29f93d67140ad835128dd543a2f67 : SUCCESS in 1h 19m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db9688bdefbd42b69a594ab8fbc7aa3a : SUCCESS in 39m 10s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/9026645c21a44e6ab87e3115e47cda7f : SUCCESS in 1h 30m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e6b1e392273b471c9dbdb09b704cda4a : SUCCESS in 1h 18m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dff4f374f22f45b789e34c8f04beec07 : SUCCESS in 1h 17m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/544089ceadee4789b1e35b989b9e4bc9 : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a87651606944bfdc59a5214e229058ca263b99f3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-04 08:10:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a5c280e67db9a6a04a72657e458bca022313457","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-11-10 00:08:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"049a08fc57cb000702e6b194bb5991f3ddb48d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 00:09:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b55e7c12663b8a6c60e3eb86de403814744153a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 02:21:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"973dfe30f4e05fb92ad1bb528b3e4cd8a1c324db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 02:21:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/82fae56570784004a4606072920f71c9 : SUCCESS in 30m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a82dfd229f7541a5a6189a9d33f88438 : SUCCESS in 10m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14c2bb0563d7430fbb01fce9f5c6dcfb : SUCCESS in 19m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/548ddd5fa45f417894f266d83409d9b1 : SUCCESS in 28m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7084063fd404a949eaf4f018360fcc1 : SUCCESS in 5m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/74dd3e0069df4dc0ab211bdc91f7a2f7 : SUCCESS in 1h 19m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b33abe14177a4f05a8de13c717bc8b7c : SUCCESS in 1h 03m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddc47df2e70b4d05b85a8eb58d268572 : SUCCESS in 1h 24m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/66507b886e61425880a3888294977b93 : SUCCESS in 41m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/779c874eb3df449c80cadf08351df5ae : SUCCESS in 1h 24m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5c4e55d8f2684ae48284966b97ea531c : SUCCESS in 1h 18m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c95821ce55fa43aa81d0110832c710ee : SUCCESS in 27m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a5cb2437b7e55f2c4652ca8c6e1fc88133894d0","revisions":{"1a5cb2437b7e55f2c4652ca8c6e1fc88133894d0":{"kind":"REWORK","_number":1,"created":"2020-10-09 09:41:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/55/757055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/757055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/757055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/757055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/757055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/757055/1"}}},"commit":{"parents":[{"commit":"5ee6e5bb31179acf6e86357b0b0b34ef051af161","subject":"Do not fail deleting namespace if it does not exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee6e5bb31179acf6e86357b0b0b34ef051af161"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-04-14 08:25:18.000000000","tz":180},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-10-09 09:41:34.000000000","tz":0},"subject":"windows: fix terminating processes","message":"windows: fix terminating processes\n\nThe neutron Windows exec call helper doesn\u0027t properly handle the\nsituation in which the process that it\u0027s trying to kill was already\nterminated, in which case using wmi to fetch the process can raise\nan exception.\n\nThis change ensures that those WMI exceptions will be properly\nhandled.\n\nCloses-Bug: 1872663\n\nChange-Id: I00c810fe541ac5e1e9923155fe90eb07a0b4b3dd\n(cherry picked from commit 8fd3e884c7953645862663400af181ba233ec28f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5cb2437b7e55f2c4652ca8c6e1fc88133894d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5cb2437b7e55f2c4652ca8c6e1fc88133894d0"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
