)]}'
{"id":"openstack%2Fneutron~757569","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ia313b9c4028f0242cb0da8764b9921c5c30c465a","project":"openstack/neutron","branch":"stable/ussuri","hashtags":[],"change_id":"Ia313b9c4028f0242cb0da8764b9921c5c30c465a","subject":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec","status":"MERGED","created":"2020-10-12 13:43:49.000000000","updated":"2020-10-23 13:14:09.000000000","submitted":"2020-10-23 13:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757569-1603458743316-7901bf70","meta_rev_id":"fa51b667a4177bcae6922cc4b2f2d4b20edd9719","_number":757569,"virtual_id_number":757569,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 13:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-10-23 11:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-23 07:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-10-23 11:19:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-10-12 13:43:49.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 13:43:49.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-10-23 07:34:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-23 11:19:21.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-23 13:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0517e767b37353ca549a91509ca53acb0043468f","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-10-12 13:43:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"656da25d64e8c574d812a1812fa07d245d8bba85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 01:00:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d1779e9aa8a41a882a83fe82dcfeeec : SUCCESS in 33m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f13d3ddc3158442c988b692b4b119b86 : SUCCESS in 24m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29045de29b9f43129a93a87128945f0f : SUCCESS in 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5ff555831848588baacff12065b896 : SUCCESS in 26m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee348267365842ab95e49bee46e1b6ee : SUCCESS in 25m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b53b5b531ef46d7b2b3f7648b84051f : SUCCESS in 22m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a189bb5e23ef4d5cbc0c13fb516c39b5 : SUCCESS in 8m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cbce717b044e4be9a430142f1429f51c : SUCCESS in 41m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/87b7cc5d85144e59ab187d709c04b20c : SUCCESS in 1h 06m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3eb2b0d866534cd0a605d20be7881ff6 : SUCCESS in 1h 24m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8a0206923fa2489e8ddcaf35b09fed59 : SUCCESS in 1h 23m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/558c7ee4dbe74b128cc2a324477eb4c6 : SUCCESS in 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4352707ca071b679979eb5bc045ac2e589a9e53","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-23 07:34:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23fd95de9503016dc1b5e87700e263df7f12f9a1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-23 11:19:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6005a4edee1dca669a58fe08ffc5000ad949e628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 11:19:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d012f091674ef75e050f85ef8155b54a875eec25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:12:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da5b33ac62e14e4a9d21c0dfac1fa144 : SUCCESS in 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d97427028aa4af7920fd2b4f1f1c32b : SUCCESS in 15m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f43eaad3fdd416294e75abcf43f06c6 : SUCCESS in 30m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c998f1e72a3a4e38a3e64b9d5c317214 : SUCCESS in 32m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247c57ed93e44cd5a499ba065a9ba988 : SUCCESS in 11m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b9ada9bc109b4f67a8fe169d88d146b1 : SUCCESS in 55m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/be7c865f9e364b948dfa1785f248fcec : SUCCESS in 1h 50m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d277b87bee9e4e71942184fcc3e29275 : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f1c568c84eb5f70c3d16ee1b1b5aa3655ac52c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:12:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa51b667a4177bcae6922cc4b2f2d4b20edd9719","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:14:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce1b49b342f4c9f953c460e04c4195c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"010d8f74a92f471d9c45bc3d4451b4be0985c9bf","revisions":{"010d8f74a92f471d9c45bc3d4451b4be0985c9bf":{"kind":"REWORK","_number":1,"created":"2020-10-12 13:43:49.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/69/757569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/757569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/757569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/757569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/757569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/757569/1"}}},"commit":{"parents":[{"commit":"3447613efae7e291e43404ee7e8a9770faef4118","subject":"Default dnsmasq --conf-file to /dev/null","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3447613efae7e291e43404ee7e8a9770faef4118"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-08-13 14:56:23.000000000","tz":-240},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-10-12 13:43:49.000000000","tz":0},"subject":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec","message":"\"round\" may bump 1 second to 2 if sleep takes more than 1.49 sec\n\nUse less than 2 as a way to account for test on slower systems.\nThis is a follow up to: https://review.opendev.org/#/c/691883/\n\nTrivial-Fix\n\nChange-Id: Ia313b9c4028f0242cb0da8764b9921c5c30c465a\nFixes-Bug: #1891517\n(cherry picked from commit 6a9fce39d6cf0051c6ea0f3bf5d57b9a7a03c3b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/010d8f74a92f471d9c45bc3d4451b4be0985c9bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/010d8f74a92f471d9c45bc3d4451b4be0985c9bf"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
