)]}'
{"id":"openstack%2Fneutron~757573","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ib610b8513b4ea888a540873b26c3f205ac575b17","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1886956-stable/ussuri","hashtags":[],"change_id":"Ib610b8513b4ea888a540873b26c3f205ac575b17","subject":"[Functional] Add logging to the check test file function","status":"MERGED","created":"2020-10-12 13:55:25.000000000","updated":"2020-10-19 23:50:37.000000000","submitted":"2020-10-19 23:49:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757573-1603151347478-cd30e8dc","meta_rev_id":"2103629ec70c23614caa52e77d1df51a60270e04","_number":757573,"virtual_id_number":757573,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 23:49:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-10-19 15:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-10-19 20:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-10-19 20:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-10-12 13:55:25.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 13:55:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-19 15:58:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-19 20:35:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-19 23:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1da497c1532eeb48ad8985f339e07f022ef3d5b","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-10-12 13:55:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d67a239c244170655fe802d204548fd110da0f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 02:24:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/64476358bf584856ad33c2161517bf19 : SUCCESS in 1h 01m 37s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/651cd9276a514863a1a89dc0772a7aab : SUCCESS in 38m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/946e23b4503748c6b671651fd08b8b60 : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/8a5f62ff660d492ba09c43215b21132a : SUCCESS in 1h 11m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/c60291251f57472cb2fd2ad691c5ad67 : SUCCESS in 1h 19m 11s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/c17bd924b0964841bd4a7f6848228b8c : SUCCESS in 1h 28m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ba8b22b966e43e18e2cb26835fd0c4c : SUCCESS in 36m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14c58b8ed5504ecc80a2e1ae910d4b7b : SUCCESS in 27m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a46d99079ea42f7a619c9e34bda38ee : SUCCESS in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e52ff59e16c14d31984b3028d13e93fa : SUCCESS in 29m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/722974b0358c43029fb045d3b39e0551 : SUCCESS in 31m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f643b68c230462dae0fae22af1f915f : SUCCESS in 27m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3c52654ef74911ace62b578ad6305f : SUCCESS in 8m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb31c43452ac4fd3aa3cc12d3f4cb71a : SUCCESS in 50m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/01db0ee17043483783c218d49ef9e1df : SUCCESS in 1h 10m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7ab28a44e5b49b7ab438da80c76c2f9 : SUCCESS in 1h 12m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2e7eff26d8a646de918fc37d96253b5d : SUCCESS in 1h 49m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/85448da58abb45fcbed52b2747f815d1 : SUCCESS in 1h 23m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b5e5c9dee12e458ca9244437551b2f43 : SUCCESS in 1h 23m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2806ce2ef1f143ae9e075dce65450f71 : SUCCESS in 1h 10m 47s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e39a6f64762349faad335e12171e2cd5 : SUCCESS in 1h 11m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e18bc5b016634db09627cb7175054541 : SUCCESS in 1h 08m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2104a3f2806d4a589117cfc809f7b70a : SUCCESS in 1h 32m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3fac2ad652604b008da45b9bcddec363 : FAILURE in 1h 25m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dddcaaa4b97c40da9122194a60462aa1 : SUCCESS in 1h 09m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a26152ad40a64c0b94e9a8a55e817af4 : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3eab9d4b6a9a4076b19ddf411c6b6b17 : SUCCESS in 2h 08m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45e36764698740a0b2b59d532d20e6f7 : SUCCESS in 46m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/020036f6abf24aba996c1b3e00e2062e : SUCCESS in 1h 14m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ce1feedf247e4355bbd24d57d98e47e5 : SUCCESS in 42m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/15f7a0f866da47959619bde07401cbd9 : SUCCESS in 1h 11m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9701195fab7847d2a5ab7f5f9785acae : SUCCESS in 47m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b3b09b74ead343f79a77e0de3e2b95d6 : SUCCESS in 2h 07m 55s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/388067f2887a49c296eb1f9e11fed100 : SUCCESS in 1h 48m 41s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f3feaa0291984443b6c2666aaaa6702d : SUCCESS in 2h 25m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/34830d960d1d4c5580f4996416dcab0d : FAILURE in 1h 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa1d840230291461fb5eb9498bd48a8e353384b4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-19 15:58:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"706c8bb6d4b7d3f65c96d7b108370e810272596b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-19 20:35:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7dabd23d905e2764463bb72c0893de10426645f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:36:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a12c32d9464c7a02908b33d0f67ebf3bf5d903d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 23:49:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/bc76681723214a70b99663c112c2e4cd : SUCCESS in 48m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/009eda9f53bd467eb0ad977542a70722 : SUCCESS in 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f18915807e4430facce9a4070eb2b4e : SUCCESS in 15m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/895c9f931b3340bbaeb5865a94a68260 : SUCCESS in 35m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b938008e799646c2a34f92783a17c08b : SUCCESS in 37m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a297dab33f104472b6959d66aa505baf : SUCCESS in 11m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a2df8871f10546edb1d513c780216e04 : SUCCESS in 51m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/079708ac81f14e1795acf496961a4a8b : SUCCESS in 1h 11m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84d7e606d23f497586369e31b77bfa61 : SUCCESS in 1h 27m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/18185d24c58542c5913096de3c1a6423 : SUCCESS in 55m 19s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63375af394e349428aef8d99b928f046 : SUCCESS in 1h 01m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6febee630644203bfc070c9d253c4f4 : SUCCESS in 1h 10m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3ff80a5f5e9b4a80b9d189eed2af041b : SUCCESS in 1h 16m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/427201f0cc27412da908c732f220530b : SUCCESS in 1h 56m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af322662dfbe4f5f86b0d61cb8e29f1c : SUCCESS in 57m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7c13939fc66145368c58f51b82efae96 : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eae4fc6cb5e4dcef691677f76bcb0dd2499d10a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 23:49:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"49a2912b229de8ce92bab1d1b2a2ac78cb89f952","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 23:50:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0715d157354430b8d6f22c03523add : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0892ec3f76d4f22364eaad9dc353dd5b3e406dc","revisions":{"b0892ec3f76d4f22364eaad9dc353dd5b3e406dc":{"kind":"REWORK","_number":1,"created":"2020-10-12 13:55:25.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/73/757573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/757573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/757573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/757573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/757573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/757573/1"}}},"commit":{"parents":[{"commit":"3447613efae7e291e43404ee7e8a9770faef4118","subject":"Default dnsmasq --conf-file to /dev/null","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3447613efae7e291e43404ee7e8a9770faef4118"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-07-09 13:19:03.000000000","tz":120},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-10-12 13:55:25.000000000","tz":0},"subject":"[Functional] Add logging to the check test file function","message":"[Functional] Add logging to the check test file function\n\nIn the functional tests module which is testing WSGI Server, there\nis used function to assert that file created by the WSGI service is\ncreated and has got correct size.\nThis function wasn\u0027t logging anything so in case of test failure\nthere was no way to check why it really failed.\n\nThis patch adds some log messages to the check so it will be clear\nif file wasn\u0027t created at all or if size was not as expected.\n\nChange-Id: Ib610b8513b4ea888a540873b26c3f205ac575b17\nRelated-Bug: #1886956\n(cherry picked from commit 0c55ab9c78df028e9e321a3f8dea59ede217fef8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0892ec3f76d4f22364eaad9dc353dd5b3e406dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0892ec3f76d4f22364eaad9dc353dd5b3e406dc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
