)]}'
{"id":"openstack%2Fneutron~757620","triplet_id":"openstack%2Fneutron~master~I16dce5b3faf62ce02535170a1b0e1ad22ee8ac30","project":"openstack/neutron","branch":"master","topic":"bug/1896506","hashtags":[],"change_id":"I16dce5b3faf62ce02535170a1b0e1ad22ee8ac30","subject":"Keepalived version check","status":"MERGED","created":"2020-10-12 16:36:38.000000000","updated":"2020-11-05 18:54:39.000000000","submitted":"2020-11-05 18:52:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757620-1604602337971-3f776333","meta_rev_id":"92f2e783a7bb703d6f7d570cc07cf4885f0812ee","_number":757620,"virtual_id_number":757620,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-15 19:35:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-10-15 18:33:50.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-05 09:57:38.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 18:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"date":"2020-11-05 10:47:52.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-22 19:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":2,"date":"2020-10-16 01:01:49.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-10-22 19:55:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 05:20:58.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-10-13 06:40:26.000000000","updated_by":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"reviewer":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"state":"REVIEWER"},{"updated":"2020-10-13 11:20:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-13 20:56:13.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 16:02:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-15 18:33:50.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 19:35:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-16 01:01:49.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-10-22 19:55:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-05 10:47:52.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-05 18:52:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3311fb7e17b5366e12565d850790aaf7bb4d663","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 16:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"662319cbaac3d3de2552fdfacfdf6fd2a5b55a74","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-12 17:08:10.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/757620/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"527b2f9f4899b4e7c69db188ce16dedfc99b505b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-12 17:50:44.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/20/757620/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7d313e4de8ac8bc077fa49a2eb364b58b761b54b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-13 01:35:06.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/20/757620/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/20/757620/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"284791a54bdcad3f2408395b79e50b821ba86930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 05:19:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/599939d0568f43aa94d425b1bd67dd38 : SUCCESS in 56m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89aa59265b034fe0b826a72b4cf24352 : SUCCESS in 39m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7f36a2cc55e44e7f9e6b959197b4e40a : SUCCESS in 1h 36m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7233bf25c2cc4206a463eb356a544426 : SUCCESS in 1h 14m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3b93397f21547ac83bedb5b8b9331b6 : SUCCESS in 1h 21m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/77d63020f9c74012936074a5014e1c90 : SUCCESS in 1h 01m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50e677cff685431db9e257da4c097f31 : FAILURE in 20m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6007e4061085464aac3e77a69ee1035a : FAILURE in 26m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a789af8ea14b21b294ac1291749667 : SUCCESS in 11m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ee42a3183f44f26a2aed3bf44f7ca2c : FAILURE in 25m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a96b87a088ac46279ffd592042531a09 : FAILURE in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c54a1f1ece945df979dded14df19ccc : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bce284e168ce4f36bf557866d7601100 : SUCCESS in 1h 17m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71953cfb63874df6a14f305a4ece9eeb : SUCCESS in 1h 16m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/860c50fd83234d60998f886de5e8e012 : SUCCESS in 1h 46m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0af28fd8a67e47bca21efca9d585d0cb : SUCCESS in 1h 06m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c3723097ac0447ca1d0358ea5a861cc : SUCCESS in 1h 13m 02s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/306433864d0141d98bb9c3d2053317dc : FAILURE in 1h 06m 29s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b62db971d58c491d8c4a6f7552d4c3bf : SUCCESS in 1h 11m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef1a1776feb1485ea2b353d6e76894fc : SUCCESS in 1h 19m 48s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/53881f459c814a39857ca4c977fb5a96 : SUCCESS in 1h 24m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8c72efd94a694c4c86b2383821881e48 : FAILURE in 1h 21m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6573f565f3f640daa8eaf9571a128490 : SUCCESS in 1h 09m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/249f0902b3884c72891c3b45d38a8054 : SUCCESS in 1h 10m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bcfe01c658c348f391f124f0755c752c : SUCCESS in 2h 09m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dca3b4d95a3948669be960d980f0e030 : SUCCESS in 1h 03m 10s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/28653d56a2db4ce1b96fca6d7cfaad61 : SUCCESS in 1h 11m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2c7dad4caece4a8a911ef844f9172a79 : SUCCESS in 1h 14m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/24481c5b9fee403fa2bdd84e38619cad : FAILURE in 43m 34s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0b78651d8898416ea3647e71801ba953 : SUCCESS in 1h 15m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0a51a4ffda98470da96943e0b0703c7a : SUCCESS in 1h 45m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/93fb1212da384f978e3a6037ee85a249 : SUCCESS in 1h 49m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e25498a3db804098a49a3095325218d0 : SUCCESS in 2h 19m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/80c685252f4c4926aa30f2ea946a5bd8 : FAILURE in 37m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9de0fbeb7a914ed78cb5618b3f41ba4c : FAILURE in 48m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c91af78b10a15b12002c36a942f2cf8c9008cd2a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-10-13 05:20:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nDo we need config option if have version check?","accounts_in_message":[],"_revision_number":1},{"id":"77cc0fef1b43ae030193007da82445ede8f3fa00","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2020-10-13 06:40:26.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI agree with Oleg that the config option should be removed at the same time as adding the version check","accounts_in_message":[],"_revision_number":1},{"id":"388bcc6742d8465f9cf8aa0ad06af42f801bbc30","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 09:34:32.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e I agree with Oleg that the config option should be removed at the\n \u003e same time as adding the version check\n\nFor that we need deprecation period and similar, or as this is a short lived cfg option we can skip that?","accounts_in_message":[],"_revision_number":1},{"id":"40edbe340c0e3976fce5ea57e75ad17d7c7ef4ce","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-10-13 10:02:00.000000000","message":"Patch Set 1:\n\nSince it wasn\u0027t released I think we can just remove it, but let\u0027s wait for Slawek\u0027s opinion","accounts_in_message":[],"_revision_number":1},{"id":"8ce1c4672c718975dba88d2581143825f65f5ea2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-13 11:20:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"82531bf2f2d27b6185f76b935e4bd634c4223879","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 16:26:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05ac9921cca86fb1fa02035cbb11e64dcb2b6534","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 16:37:59.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nFor the config removal/deprecation let\u0027s wait other\u0027s / slaweq/ bcafarel\u0027s opinion","accounts_in_message":[],"_revision_number":2},{"id":"533fd42b659fdf98f7bc0cabe915418473f71d2d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 16:39:01.000000000","message":"Patch Set 2:\n\n\u003e Since it wasn\u0027t released I think we can just remove it, but let\u0027s\n \u003e wait for Slawek\u0027s opinion\n\nI am nut sure, it was backported even to stein (https://review.opendev.org/#/q/I2dfdb9f56de28d56ca0f240ff34fa7c3a12e339b ) so I would be careful","accounts_in_message":[],"_revision_number":2},{"id":"9c8d868311cbec77641d91158c94562ec7f3de42","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-13 16:57:40.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/757620/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8f052183d31f1df1ad02655635f8c23cfd0c49bf","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-10-13 17:12:35.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757620/2 : FAILURE in 45m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a9fbdfbde206a87741dacfe20810efef743a5989","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-13 17:40:11.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/20/757620/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"44f213a70b2e036a7b9ed2b01ef79c31f929262b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-13 20:56:13.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/20/757620/2/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/20/757620/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fa1317422986bc7aa23bfb7ecccf114ffe213cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:55:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/59a33921e10b42059a100e923ccf4616 : SUCCESS in 51m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d56d4849ab44dbab64d4e5b6b128bb7 : SUCCESS in 32m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e22b17c54989416c9cdc8c669e5db363 : SUCCESS in 1h 19m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ffc63c0bfc3f4d7c94d239d4fdfc5d12 : SUCCESS in 1h 05m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ff18f0e658504bdbb3b504f542ac8441 : SUCCESS in 1h 12m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2bdd669f086a4f12b1187393be2f9002 : SUCCESS in 1h 26m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffe43dd95d83485fb5a33d6cbc2c5116 : FAILURE in 31m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84dce691183140deb197388b56f7ab26 : FAILURE in 26m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d271f245872b442f86f3714dd583e1e4 : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10ba1b8265ce4446bd766a2753757f02 : FAILURE in 28m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc82267f32814220b8d44d37b707567d : FAILURE in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3d004c37a14d14acddb0a109e53075 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e07ea3b723fc4f7bb33a57f2404c0550 : FAILURE in 1h 20m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de3446761119458e9d43caa6e07e617a : SUCCESS in 1h 16m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/892f01e30bcf44dcad7829a36f773203 : SUCCESS in 1h 36m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/719390f1916c4565a65fef18642a64fb : SUCCESS in 1h 11m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f304b68e6e24d638d9c294f0c584926 : SUCCESS in 1h 08m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/4358e3136d664884a508929cfba9de58 : FAILURE in 1h 06m 46s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/64857e537a0c4974a988a7e07e7d041e : TIMED_OUT in 3h 07m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac2812b834e14c02bb1037c4cee4f5e2 : SUCCESS in 2h 02m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1f9dd22ea35346489557d8b395e75438 : TIMED_OUT in 3h 07m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d2d2786498244e38f56af8e27217ebb : FAILURE in 1h 24m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5aca5b1ea87e4d3aa5ae186538cb0577 : SUCCESS in 1h 15m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/af2081f3dc5540f8a5986939907cd3d2 : SUCCESS in 1h 21m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed8cd3978963425b85fbe7640c50caa2 : SUCCESS in 2h 10m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/209e89ad5853482ebe9c0288da246f7d : SUCCESS in 1h 00m 18s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8d97a69bba8549b99593aae89210084e : SUCCESS in 1h 09m 45s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b8d36f7bb3342bfadb67dd8038c69eb : SUCCESS in 1h 12m 16s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ce5deadc3fd141cd825d17bec66e2884 : FAILURE in 41m 46s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/712f4b53cd154a1cb54ef7915261180b : SUCCESS in 1h 17m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/369b9e46e7e547a2a52472ca2ea760e4 : SUCCESS in 1h 31m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a359525955154b4896d6bc65fd00b668 : SUCCESS in 1h 46m 14s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1c2eca4386264a0fbf3117ae43372ffb : SUCCESS in 2h 23m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0ec70d0684014df69e744637569b7fc3 : TIMED_OUT in 3h 05m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e67384441c444dc1b602f1cb1e85ce0b : FAILURE in 41m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6594d27ea18861edee358212175bcd84d08e26e3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-10-14 00:44:55.000000000","message":"Patch Set 2: Code-Review-1\n\nIt\u0027s better to add some test cases like the reference patch:\nhttps://review.opendev.org/#/c/745641/2/neutron/tests/unit/agent/linux/test_keepalived.py","accounts_in_message":[],"_revision_number":2},{"id":"3d19fd989a444b61716f9d6bb4e8e2c927f7d0f5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-14 04:23:45.000000000","message":"Patch Set 2:\n\n\u003e It\u0027s better to add some test cases like the reference patch:\n \u003e https://review.opendev.org/#/c/745641/2/neutron/tests/unit/agent/linux/test_keepalived.py\n\nThanks, I would wait with it till it is clear how to move this forward, for example to keep config option","accounts_in_message":[],"_revision_number":2},{"id":"60b1696c16d165cdba68f107ffdbcb6a1ac80d40","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-14 16:02:09.000000000","message":"Patch Set 2:\n\nI *think* keepalived version checking is probably safer than some other binaries (looking at you dnsmasq), where an older indicated version may have support for newer features.\nChecking no_track on 2.0.0 threshold seems OK overall I think as this was a major version","accounts_in_message":[],"_revision_number":2},{"id":"fdce2a0c832fc28675d182100143d3a57483fb58","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-15 17:57:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c650290fe8baa2f0f1498dd1d90f51fe7cf1ad02","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-15 18:33:50.000000000","message":"Patch Set 3:\n\n* sriov-test http://52.27.155.124/sriov/757620/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"470263af799559cc762ddb3bc9ca189b34f66237","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-10-15 18:46:56.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757620/3 : FAILURE in 48m 03s","accounts_in_message":[],"_revision_number":3},{"id":"c00966ceb8505bea26b68400891c6e1a74a1a336","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-15 19:35:55.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/20/757620/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2bdad25df7863bc2560e7a6824c934d4c3b9ec0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 23:06:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0167b3adcfd9442380d1106fd1c2bf7f : SUCCESS in 55m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6bbd2afb382b40989f5f60ae264727c2 : SUCCESS in 40m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/973894e71356498580a2f97953425360 : SUCCESS in 1h 35m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/810efb9316fa4d25a25a0f481b8102e0 : SUCCESS in 1h 04m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cf0164b595084598b3f196964d01265d : SUCCESS in 1h 13m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e5d6cf9b9c584c21aab3efb1298ae63e : SUCCESS in 1h 03m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e22de008a5604f0e8ee165ae82e6747e : SUCCESS in 36m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44df6241d6b04e51857b6cbab9b83cd5 : SUCCESS in 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02d4a7aec22c4602bee85d654fa5066e : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaa53593b35646b59a4060d9076d47b9 : SUCCESS in 30m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f51b48fb8b0d4036ad56d81df84b718e : SUCCESS in 26m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f730092034a246b2b841b06ff3b49cc6 : SUCCESS in 14m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9238e3f86b0b49f0a5a3de0b97e8d0d9 : SUCCESS in 1h 40m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bda5899e7494475b9ec31382c22bd178 : SUCCESS in 1h 05m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bfc9eae1bb534df1840a8f9366c6ba9d : SUCCESS in 2h 00m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1dd0abc1e604a1ab143e9ef8dca075f : SUCCESS in 1h 16m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eff9a12d8311429da644cfba1b36e6dc : SUCCESS in 1h 23m 14s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a50c6e98bebb41ddac01554a9e18beb5 : FAILURE in 58m 04s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c1ba36b620ef4216b4298402cc7513f0 : SUCCESS in 1h 28m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce5da897c0dc4ea29abb5e465755e31b : SUCCESS in 1h 28m 00s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/25b23e5a8aa3486f9bc7e8d0443bd1b0 : SUCCESS in 1h 45m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f3800027f37456cab743c6a0ee05a0f : FAILURE in 1h 44m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e439eafce83f4e9ea7071b9114894d27 : SUCCESS in 1h 09m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/18c5f7fd9fff4b76bacb4459caa0ba0a : FAILURE in 29m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/394a4e285e7848948707f5946df700d5 : SUCCESS in 2h 08m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70f70ec112294abe899ebe257c93934d : SUCCESS in 1h 05m 50s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a78a2aa9b4c049528c1a1bece56c17fc : SUCCESS in 1h 15m 40s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c04b9eee6c7401c80a8f63d8ccacee1 : SUCCESS in 1h 29m 44s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/be0824d0c02a435cb502d91b3463c4a8 : FAILURE in 40m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1b162d7b26ff40a389002fb88b8cc40c : SUCCESS in 1h 19m 13s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/39df0b1af26c48389bdfca0ff2c87a24 : SUCCESS in 1h 36m 27s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74fa9efd8af54135820d518c14ae44f5 : SUCCESS in 1h 46m 40s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7bcc43babf27488d941f98f77e9e72e6 : SUCCESS in 2h 21m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1f27394436d9442889285a3c2f39402f : FAILURE in 41m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3c8de8be8fa94cee925fcff24e609c0f : FAILURE in 43m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a46681ea74212e790a80015c7f499830578c96d","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-10-16 01:01:49.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"3359c44140a47185a879bdfdcfbdf988dc45652b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-22 19:55:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e65cdacdf83053465206c39f3d1cb7eeccfe4a53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 19:56:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f6cd0478629ad75f7e8b440aa355c9b14f0da37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 00:40:10.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d952e72714da43ccb877e3ea61e462e5 : SUCCESS in 57m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/78a6ff3ff5d045d3af35e9fd77e0d933 : SUCCESS in 1h 37m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd2a72f97c1848b485623d875690096d : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b8c390a8867b4c3fb27326003f706a96 : FAILURE in 23m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f850609db3142559d17966c4954e0c2 : SUCCESS in 1h 08m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0775587c443a495d873cbd04ac5e7ea9 : SUCCESS in 34m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b94f0739e9433b9a711e3f89fcb072 : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/305140e56f8a4a67b744e445a214f8ae : SUCCESS in 35m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f0c94b1c02241ba809c187038a87792 : SUCCESS in 29m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4488606da0f7463ab8b5c6c5f1fb1eab : SUCCESS in 12m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d8ce0e51857046b1b99e6cc62b971798 : SUCCESS in 1h 19m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93e97cf7a00c44419d4555cd6bf13719 : DISK_FULL in 1h 13m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/901dd4810d3a4a18b8169415ba72f075 : SUCCESS in 1h 22m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/112528e3a3fe4882bd2d7433134213fa : SUCCESS in 1h 01m 19s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/330d18124744480c8537431aebd76478 : SUCCESS in 1h 06m 11s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e1091752bf984c2ba58196b2e4def24c : SUCCESS in 2h 09m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d29903f22f904fbc9696c3f5305a267d : SUCCESS in 1h 11m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/87c594c6327e4cf78716e845a8a7262a : SUCCESS in 1h 06m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2e3299337d4424da12b807533048180 : SUCCESS in 2h 02m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b623e67e478146dd936867e02f154399 : SUCCESS in 1h 05m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a19f7573cf1430f89ca3524897e281e : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":3},{"id":"9b1c934b30b159b72daffc0296519a7527e177ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-26 10:11:25.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1a796b4f954df317da27c32023292a66ed634557","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-03 13:39:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a2c72eb13118f34208273df5afb9eb74425a98e4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-03 14:30:08.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757620/3 : FAILURE in 49m 33s","accounts_in_message":[],"_revision_number":3},{"id":"519e2993b6e8294d6c7d6720a473c269160ffc41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 02:02:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c90f656f0e564e2283355750350d9d42 : SUCCESS in 55m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8beb2baa42bf4f13a23fb5549d5f37d5 : SUCCESS in 33m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/52c1a08f474340bbbf1d21717939b86f : SUCCESS in 1h 22m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/325761a6e0ff4241bb7f6e93f896a78f : SUCCESS in 1h 21m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9838f72dd5944061b447de46aa67ff7d : SUCCESS in 1h 38m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ed89d3c55fa1438e8bf83e9ade684b48 : SUCCESS in 1h 20m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82a15b495a56454887e8f678826572c1 : SUCCESS in 32m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7925c9500a8343e990c500c3a44e16bb : SUCCESS in 29m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c919e13e294334b7b3774471131803 : SUCCESS in 10m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dedd5bfdba284fa7ae7d404172fc0b56 : SUCCESS in 24m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/156785b9489947638d93d23a8cc46ff8 : SUCCESS in 28m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6be7bc51614969935dae13f301a634 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5254bf87c7584b01bd078bc547fa29f0 : SUCCESS in 1h 00m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd2682ba9d3c46c0ac2db8efbcc2bf1f : SUCCESS in 1h 16m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/73c77d9ceccc4e8aa5f69cd1b2f78fd2 : SUCCESS in 1h 45m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdb1f0a0b8544c21a4c0d9696d00b606 : SUCCESS in 1h 31m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/02f2b92736cf47d8854bedd11f4c5c2e : SUCCESS in 1h 09m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8176166a86f94819b40b04b54e03a18d : FAILURE in 51m 27s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/228cacf716a64b3f858ff937bd365657 : SUCCESS in 1h 12m 16s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e83f0a25d9d4b4581ef4f3e4deb9480 : SUCCESS in 1h 13m 12s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c63e9168d19b448792f976b8f0587ca7 : SUCCESS in 1h 22m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bcecc75360ac433fb924dee5a551e441 : FAILURE in 1h 21m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8acdbbd8d0d43f788f3ce39971ee913 : SUCCESS in 1h 07m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2af1e3d365124dd88beeb892c48f0ef7 : SUCCESS in 1h 17m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6cdbde2152af4a0ab4192581999cf69b : SUCCESS in 2h 13m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/151188560a674df1b08e6d1c55bbae30 : SUCCESS in 52m 07s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5ce5d77ed2aa4d21b96fb7dcb076cc12 : SUCCESS in 1h 11m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7114fb2be1fe4267951a6d3f572469bc : SUCCESS in 1h 14m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6dd057da250d4d9f893787dd2ad5d9bf : FAILURE in 42m 25s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/700dd7744d5e499695ccaa04ea6ebf03 : SUCCESS in 1h 25m 43s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2590d656ec749ffacec28d4ba452820 : SUCCESS in 1h 41m 07s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/603f728272c547a0b097e01bf8f93ad1 : SUCCESS in 1h 56m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/12c7fe4ca29d4cf4b844796fc6ecaec8 : FAILURE in 1h 01m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/70b7355f9382419183fa9f930a4b7cc3 : TIMED_OUT in 3h 03m 48s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b1ec944698e9419885742a954d6c441a : FAILURE in 1h 42m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e3e2a6525420163f2f7bdc0926da31912d81e4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 02:03:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"feb1f86f3a1a6139ffeea23cf350385051fe69df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 05:44:07.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b3f8200977543a3a5186d8197b57919 : SUCCESS in 46m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8d2409c0b87449c6b6fe24f573431284 : SUCCESS in 1h 31m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9d9b5d4554f54fd9a03fbc88ddd7ec6a : SUCCESS in 1h 19m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c7289cdccdba4121a8d9b30147dd8161 : SUCCESS in 1h 10m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8713cea860174dd78da4485c3290f21f : SUCCESS in 1h 18m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be33dc7de5604617ad9e3310f297cb6f : SUCCESS in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ecd3b33dc474359a7ed82c27085bcc6 : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6805a12be57433d8b639ad400d5b884 : SUCCESS in 33m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef094897d064da199cf40a1eb279460 : SUCCESS in 23m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186af465e2114b26a246ae7d26ca8474 : SUCCESS in 8m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8416bb8ed3b54836a1ca3b21c63a828f : FAILURE in 1h 16m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8250767666045caacbe584cdb1132ac : SUCCESS in 1h 12m 43s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/82d03bfbde364129b87f478e52cc4586 : SUCCESS in 1h 09m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7a5ece2166844c6aa05bfea4f07d457b : SUCCESS in 1h 21m 07s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/884e610fd86b48c583c1fa0a0d8d50cc : SUCCESS in 1h 06m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f4fbe388fba24af0b07d8b5e806cbe79 : SUCCESS in 1h 32m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/42f7e5c0db4a437a85889261f2bbadd0 : SUCCESS in 1h 05m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/369a9d39f7e546a99337905ad6ee297a : SUCCESS in 1h 08m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ba7a801f9814067805dad137a2733c0 : SUCCESS in 2h 13m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfad9e59594644a2b494b8c42d39830c : SUCCESS in 1h 07m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f753620a64ab42429ae5d464ded9cce2 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f6d6199a191a47ebe2b49dbf55f5a999e07999d7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-05 09:57:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9acfe0b4023676c950fae54a5bbcc071ce2c6fdc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-05 10:47:52.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/757620/3 : FAILURE in 49m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4dcf3569699550223dc3c61d2edcbaccc7783aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 15:40:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f7cd2f9c3ec344b9943e9d72745191d3 : SUCCESS in 51m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1dfb7ee588a442dea0282c9928f604bb : SUCCESS in 37m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9c20bb9f0c824a45b40ce1090300b677 : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3af08929333f470a94534ba83a03b1e0 : SUCCESS in 1h 39m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8e0929562f847f88df4e4d6deb62605 : SUCCESS in 1h 28m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f250698804154edeb796893c16ede791 : SUCCESS in 1h 27m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/751f94548fbe4f1a967a11e97282c315 : SUCCESS in 32m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c62ad4468d145c8b7d4707324a20e97 : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9d412f894f34414acc197d1a4cf6b23 : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fee391694c5d494a82f3f848c5151922 : SUCCESS in 24m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6dc0d41722646c7b29621c4b6c7687e : SUCCESS in 26m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ae806b49a64428ab9a1913a75cb3ca : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5603a1beeff4bbb88600d1690d7dd20 : SUCCESS in 1h 07m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b1b2a3e79fb4e899003d3dff9e7112c : SUCCESS in 1h 16m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e38ec2a6c9e743d082eed6458d756bfc : SUCCESS in 1h 28m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66468203d964498da2899f4d7dd4f0f0 : SUCCESS in 1h 08m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9a2791cd10534eaab1ae2f3b73c14dcf : SUCCESS in 1h 19m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/0b21c7686a85473bb53fafe69b7c08c4 : FAILURE in 1h 00m 39s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0f177d7ff799411784f76d69baeddb26 : SUCCESS in 1h 11m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4c0e2a5e4464faebd80d1658f2556a3 : SUCCESS in 1h 24m 42s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e6346673e12a46afa5054e79d3786f85 : SUCCESS in 1h 33m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b43955c1249742d792added8a437d028 : FAILURE in 1h 22m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83080d8dc5994e44a1241a97fc0553da : SUCCESS in 1h 18m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d3213beb6df04147b17be908c4d0a7c2 : SUCCESS in 54m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15e2956a77534c9d94b38c7cdb48ded2 : SUCCESS in 2h 07m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b20a4523c3d4f8196ca8b9a7cccda6d : SUCCESS in 58m 50s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/819d8bee82ac4ef1a63de5ac97e988ef : SUCCESS in 1h 31m 55s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f5d3c936c2004794b8e4b0a3a4389363 : SUCCESS in 1h 16m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d083492914ad46d8b3989ea7ca31ba2b : FAILURE in 40m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23875b043f0b44e282980282baa2e560 : SUCCESS in 1h 24m 19s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/94d56c2db3204284820b86313e77195c : SUCCESS in 1h 38m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4e3be29bec04df2b621b7e9914b6949 : FAILURE in 2h 07m 07s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f4b696a40b694c24824f23ac8620a826 : SUCCESS in 2h 23m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/30351613cd6b43e7ad73f85b5592c1cc : TIMED_OUT in 3h 03m 05s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2757396afc114e6392d1dd8bc7ecab66 : FAILURE in 3h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1caca461aff33863112e73b6881c72e31b90eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 15:40:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6278606e4698cc940868284d8c6fb0b90e4a4b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:52:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d6e97ee14a5c4680a3316293b102c391 : SUCCESS in 57m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9a12705948ef43e38eb792c6f9f4a062 : SUCCESS in 1h 36m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0252112329434ffeabb7d923739d6531 : SUCCESS in 1h 13m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/15ed1c9194634bfd97b22a6dc26eff60 : SUCCESS in 1h 22m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5c5dd1ed5114b529dd6796704027332 : SUCCESS in 1h 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42b6bb47882940b2a10956c48b79ed80 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2171c76ff3c6407690367dc421175c4b : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fd176fde4974d2a82595c5057e9adbf : SUCCESS in 26m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a62fc9b95c4d089383b656107c4d43 : SUCCESS in 27m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72021758c5a946568e0dd909d44d00ab : SUCCESS in 8m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fcff122481c4d48b43850500772e25c : SUCCESS in 1h 11m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4f32ec23ea1498c87dd00a55c099230 : SUCCESS in 1h 13m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/897864c6720e4d9c9f8e07c0b8691e36 : SUCCESS in 1h 05m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4aa1bfdc60234e378957c8d169c81d4d : SUCCESS in 1h 16m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0219e4dd90944604b5054746a4cc5392 : SUCCESS in 1h 19m 05s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/50e2e0bd4880473cad690349ad5b0207 : SUCCESS in 1h 37m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2de10c911c14572b16239f8b303bd21 : SUCCESS in 1h 10m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/29ab098b0a9349caa94967ae2bd643da : SUCCESS in 1h 07m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/156c22aa11ac4d0a81fecc5085b3f671 : SUCCESS in 2h 10m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6904131bb58f4d94aa1ac9044ccb62cd : SUCCESS in 54m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a8ac716a9d9e40b4b7fb340fabde972d : SUCCESS in 1h 34m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e8e033307053ee087cb68dadb02735dd0e02d6df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:52:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b0e7273aa1d3b208a2c136ec7d00cf22c75c043b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:54:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e15199d515d4297a0e2a6ba7542e323 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4dca7fae87d9bea01e01f786465fdb4d8231dc69","revisions":{"d0f82a3fe5c8f83e9bc506ca60284965e4c117f5":{"kind":"REWORK","_number":1,"created":"2020-10-12 16:36:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/20/757620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/757620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/757620/1"}}},"commit":{"parents":[{"commit":"0b4d6fe4abc1aa0692e34d4e6f8f86d0d0a3723d","subject":"Merge \"Process ingress multicast traffic for 224.0.0.X separately\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b4d6fe4abc1aa0692e34d4e6f8f86d0d0a3723d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-06 11:20:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-12 16:31:38.000000000","tz":120},"subject":"Keepalived version check","message":"Keepalived version check\n\n[0] introduced keepalived_use_no_track config option which defaults to\nTrue.\nHere I add and extra runtime check to check keepalived version, and\nonly if that version is bigger than 2.0.0 add the no_track option to\nthe keepalived config file.\n\n[0] https://review.opendev.org/745641\n\nChange-Id: I16dce5b3faf62ce02535170a1b0e1ad22ee8ac30\nCloses-Bug: #1896506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0f82a3fe5c8f83e9bc506ca60284965e4c117f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0f82a3fe5c8f83e9bc506ca60284965e4c117f5"}]},"branch":"refs/heads/master"},"5504a40b8f60f2d2edd2c0c8ced882a30acb8f2d":{"kind":"REWORK","_number":2,"created":"2020-10-13 16:26:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/20/757620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/757620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/757620/2"}}},"commit":{"parents":[{"commit":"0b4d6fe4abc1aa0692e34d4e6f8f86d0d0a3723d","subject":"Merge \"Process ingress multicast traffic for 224.0.0.X separately\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b4d6fe4abc1aa0692e34d4e6f8f86d0d0a3723d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-06 11:20:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-13 16:22:55.000000000","tz":120},"subject":"Keepalived version check","message":"Keepalived version check\n\n[0] introduced keepalived_use_no_track config option which defaults to\nTrue.\nHere I add and extra runtime check to check keepalived version, and\nonly if that version is bigger than 2.0.0 add the no_track option to\nthe keepalived config file.\n\n[0] https://review.opendev.org/745641\n\nChange-Id: I16dce5b3faf62ce02535170a1b0e1ad22ee8ac30\nCloses-Bug: #1896506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5504a40b8f60f2d2edd2c0c8ced882a30acb8f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5504a40b8f60f2d2edd2c0c8ced882a30acb8f2d"}]},"branch":"refs/heads/master"},"4dca7fae87d9bea01e01f786465fdb4d8231dc69":{"kind":"REWORK","_number":3,"created":"2020-10-15 17:57:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/20/757620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/757620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/757620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/757620/3"}}},"commit":{"parents":[{"commit":"0b4d6fe4abc1aa0692e34d4e6f8f86d0d0a3723d","subject":"Merge \"Process ingress multicast traffic for 224.0.0.X separately\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b4d6fe4abc1aa0692e34d4e6f8f86d0d0a3723d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-06 11:20:06.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-15 17:56:49.000000000","tz":120},"subject":"Keepalived version check","message":"Keepalived version check\n\n[0] introduced keepalived_use_no_track config option which defaults to\nTrue.\nHere I add and extra runtime check to check keepalived version, and\nonly if that version is bigger than 2.0.0 add the no_track option to\nthe keepalived config file.\n\n[0] https://review.opendev.org/745641\n\nChange-Id: I16dce5b3faf62ce02535170a1b0e1ad22ee8ac30\nCloses-Bug: #1896506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dca7fae87d9bea01e01f786465fdb4d8231dc69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dca7fae87d9bea01e01f786465fdb4d8231dc69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
