)]}'
{"id":"openstack%2Fneutron~759345","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I9b611d62be5d0529e8b35e3d8280baa5be54bc2b","project":"openstack/neutron","branch":"stable/train","topic":"bug/1897637-stable/train","hashtags":[],"change_id":"I9b611d62be5d0529e8b35e3d8280baa5be54bc2b","subject":"ovs firewall: fix mac learning on the ingress rule table when ovs offload enabled","status":"MERGED","created":"2020-10-23 02:52:24.000000000","updated":"2020-11-01 11:34:44.000000000","submitted":"2020-11-01 11:33:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759345-1604230401412-bdd4afc4","meta_rev_id":"97e1f702c9472dc2fbe950747a71d423474c285d","_number":759345,"virtual_id_number":759345,"owner":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-11-01 06:56:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"date":"2020-10-23 03:27:46.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"date":"2020-11-01 06:01:12.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-01 11:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-01 06:52:03.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-10-23 09:57:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-30 08:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-30 13:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-10-30 13:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-10-25 08:22:03.000000000","_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 03:27:46.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-10-23 09:57:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-25 08:22:03.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"state":"REVIEWER"},{"updated":"2020-10-30 08:46:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-30 13:31:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-01 06:52:03.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-01 06:56:46.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-01 11:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74232e88b114e2083201db447f74dceeddc2894c","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-23 02:52:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eed64f9ef01e95ca6f253820490673c7890f872d","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-23 03:01:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8923442c283f1f888ad81c457b97522467dedbd6","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-10-23 03:27:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- devstack-networking-ovs-dpdk http://52.27.155.124/45/759345/2/check/devstack-networking-ovs-dpdk/7a08c0d/ : FAILURE in 16m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6164f6a43503a3655a32e3b2b463e2f484f9f302","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-23 04:59:57.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"32389a6e680ee54db3965c05509bc8011f2a5df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 05:12:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/578a9df696034c3a8a57bace692c6a97 : SUCCESS in 58m 01s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/80e03e37642a49ccbd15a36449d4dc0c : SUCCESS in 47m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ad27f82d10164d7f82425b3018b33abf : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/9acccce6645e43d484ec882adcb51c9f : SUCCESS in 1h 27m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/19cab937d03e4d508e7d1eff84592395 : SUCCESS in 1h 23m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f5222096134685b0e7b672c34c7c4a : SUCCESS in 29m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/234a975687364fa28cb5ccd50004619c : SUCCESS in 22m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73779aecae0b49c79d1650daf397fa6a : SUCCESS in 13m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16f9d682830a40c3b204eb8e7a0de0ce : SUCCESS in 23m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db7ba8bbc48d439c92c287047855d324 : SUCCESS in 26m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a02eff6255b4ab7b812a8c73431819f : SUCCESS in 23m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5e2019da424271b4fb4801b2986407 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/c76231ae65c34e31a2f985b78ee56392 : SUCCESS in 1h 00m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a032dbe8bccf44698d8c8710198909e6 : SUCCESS in 1h 29m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/648c263d50b04dfbbf67aefce6da45ca : SUCCESS in 10m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d4ae07fd0c64fccb9161fa127d2c480 : SUCCESS in 45m 09s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/1c6d3aa4a1324f59817ca48e2eb8b3b0 : SUCCESS in 41m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/179f58cb8f714528a998eb352074afa7 : SUCCESS in 1h 05m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8e029b9c210942be817d890c9e66ec3c : SUCCESS in 1h 50m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5e991013cd6c422f9a50d4474a054e31 : SUCCESS in 1h 15m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d87ffe5ec034677b3b532fa602d2937 : SUCCESS in 1h 09m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b39b188b795b4ae5a987b826fd33484e : SUCCESS in 1h 10m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f087a3048f24d33aa6b4f8eaf396ff0 : SUCCESS in 1h 17m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/937e160bc91a40cf8af67d766d47cda1 : SUCCESS in 1h 34m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3e0cb628d6b42b38054fe1d8e677dce : SUCCESS in 1h 56m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e0be4d3c4f545faaca5f7813cc3a28a : SUCCESS in 1h 08m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f67d435a3f3a42dba6e1572b9445d982 : SUCCESS in 57m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6730abe269a54b408dfa25b95979607d : SUCCESS in 1h 12m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7baed9fc7c1d4e84b9fd811ee25b4c4e : SUCCESS in 2h 04m 53s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/221b509820674955a2e7caebe01b7518 : SUCCESS in 1h 50m 23s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/f4c4c4fba1a1493891a108b00d69feb4 : FAILURE in 24m 56s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/56cb5029739e45d1acf14c968fc1054e : SUCCESS in 1h 19m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e829510866f48e5b61278219f24748c : SUCCESS in 1h 09m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d9cf47adaa0f4b7cb6b93037fbf2687e : FAILURE in 48m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6103b578d2274cfe84d6de871bda2269 : SUCCESS in 41m 04s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4503826d1a184f5999e6f7319f994006 : SUCCESS in 1h 04m 33s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6388cc338ebd4129a87fd40f83bd314d : SUCCESS in 1h 11m 09s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/465e32d93e584db0bbdb40ffbd33f945 : FAILURE in 1h 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d09bb3a97c6764464517b5a98a54f886418e0644","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-10-23 05:25:13.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/759345/2 : FAILURE in 2h 23m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1126ad02b8f1a7344a748d68d8ff216e3db86616","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-23 08:08:59.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"33efbb45a7219cc48accb2f3b872a9157f7ee72d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-23 09:57:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8d173ed0c879eff17b933f67c5a02fe2590a813d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-30 08:46:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"893b34ae81bfef3a4f31ee411918a68990d41093","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-30 13:31:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7436bec0e395d5b714edc33de0128e6949d2bed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:31:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08237bf1c19419d31f849ae6d8f3c20b6de0339c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:43:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a7aee4d99bf546d19131628f7574b4a5 : SUCCESS in 1h 03m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eac16cb583644f1a316e96086d410ff : SUCCESS in 26m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae51bf81b2f64414880affb9806c479c : SUCCESS in 10m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3b5c056893a42af8a9c609d4e317227 : SUCCESS in 23m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfd89b51b12a43d2b06ca5b0897e2604 : SUCCESS in 26m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db9db0fb08e24bfbb2fd424a2edd3bf1 : SUCCESS in 28m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082e71d5953e4fab82989128897e6bc1 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f054f75db1c04b5897c867582aab4ec8 : SUCCESS in 57m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b7277c903de492f914b435933a78203 : SUCCESS in 1h 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e31b78c735e04d92886bb6790f5c321c : SUCCESS in 11m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5003db61208d46fc92d18087e6ae4823 : SUCCESS in 41m 11s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b47ff7c022ba49df8ae933773b49ad28 : SUCCESS in 43m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/910a5317a56449e085b4dfb5bb1c9f3d : SUCCESS in 1h 07m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/656f3cfd62eb482788c49f5c67d123b1 : FAILURE in 45m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5b9c9222d6504f33b597c8179ad568cf : SUCCESS in 1h 12m 16s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ae76c9bbc214096a7e60bd03bf8967c : SUCCESS in 1h 16m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2849c264cb47468f99c9d3e89f40ea89 : SUCCESS in 1h 14m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/faaedff1a45e47df98c26d3d1914f61b : SUCCESS in 1h 11m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c3517037378e4d8ea8535117a9025e1e : SUCCESS in 1h 26m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aaa76bb1f8204d32ba0f5ca8ba94f15f : SUCCESS in 2h 06m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4ad4873ad2c04ad58cadb19b39553fa7 : SUCCESS in 59m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e071463f06e645d3ac919e8d3e12414f : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f4dd56dd1e6b3b9c9164aaf43387fe96e107e87a","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-11-01 06:01:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6712885260bb539a5ad248db597d7073a800cbc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-01 06:21:31.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a6c926904462fd6bc832689bec399c9feed2a0cc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-01 06:52:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/759345/2 : FAILURE in 50m 19s","accounts_in_message":[],"_revision_number":2},{"id":"88aba47dc584f8c30a501d9a1921ea6330b6d57b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-01 06:56:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/45/759345/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"65796edc397bb9cb105d1d74635b398299c2c66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 09:04:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d56336e34bfa4e91adf40b8836fdca56 : SUCCESS in 49m 40s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/009b826e7ac7496cb57dcf3e499aa56e : SUCCESS in 27m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/07f77891fd874ec48050092178ac12fd : SUCCESS in 1h 28m 27s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/73855565a63c45759a33eaae6558b510 : SUCCESS in 1h 41m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c6b703f49fde4176a8e1808a1a723c18 : SUCCESS in 1h 32m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a054cfe9fcaf42a6a0378b42af3b8cf9 : SUCCESS in 30m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c0307ef4b6645138ad7c5773b629537 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da96223a020840c2a8945f0d7b5e4d6c : SUCCESS in 13m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8af98006f8de49c699f1235830c4a12d : SUCCESS in 14m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcf20fff0fab4841bc59ac7993b148cf : SUCCESS in 21m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fd39db1e28549159c4f10857735a693 : SUCCESS in 27m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09328cd48e7c466199b5dc2331cc3776 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/7b841dd431684c58a654bb2e0da84e51 : SUCCESS in 52m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/61430c473e6f4799a433fd51354493fe : SUCCESS in 1h 06m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce4efa43ed94160aef61b9027d89348 : SUCCESS in 8m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d9ef07201a4e465ca3c859c9d8e967f2 : SUCCESS in 38m 29s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/cfaf7a9808fc4a68a3c6edaf934a0764 : SUCCESS in 37m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4c23425004974bbd9de35f816e67b941 : SUCCESS in 55m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f8c545109b794aee9aa96c067ef40b8b : SUCCESS in 1h 57m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6d2050a15f19460e95a4144611e109c3 : SUCCESS in 1h 14m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/36a1b5fe412d453e8dd718488a4613e1 : SUCCESS in 1h 00m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e7629ebbd074c8fb542b6bb2533a93e : SUCCESS in 1h 10m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d18a81746b7c4de4854103dbf62bfc7e : SUCCESS in 1h 00m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6b712b48dddc4af39c6d416e4674e921 : SUCCESS in 1h 26m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3bb6f0f6566e44c79d9262b24e1842cc : FAILURE in 1h 23m 40s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1aee0b08d73f48c091aa4f87196344da : SUCCESS in 1h 04m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ef03878e59d34e1aa1328ec7cc03a558 : SUCCESS in 59m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a66a8df5d8d0472d9860fca5e0534469 : SUCCESS in 1h 12m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a477cc43ab994050b3d51943bc7609e1 : SUCCESS in 2h 05m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c3d857c3e1a9422bab8747562ed1ee6f : SUCCESS in 1h 28m 47s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/8704b632c5be4586bba606607fb91d8c : FAILURE in 21m 34s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ed7586564804430d994bbc0b32eb8c57 : SUCCESS in 56m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4913d78bae5e44e788ec0080f4cf60fb : SUCCESS in 58m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f83496a52d564c539d3b3bb5472d18ec : FAILURE in 40m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7592c830e612432cb265dcdbd52b22e2 : SUCCESS in 57m 07s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ce271a17a664632a144bc4737679984 : SUCCESS in 1h 05m 14s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d32098eb60ad4eec8c229225a3f38431 : SUCCESS in 1h 11m 51s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e8f77f47466f42dcb30fbd9be06ebda2 : FAILURE in 1h 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c2a86e3b1d972a3dd9a320869a3886eb0c94cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 09:05:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f798cffbc8d89d387591063973bfd155170ad6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 11:33:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"26f5f374a706c62ef9d9715a58ea96ac6e08a0a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 11:33:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/e4746a345af141f891223123ccdd2f48 : SUCCESS in 48m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0958dd51a6c54988a7651e6053f993ee : SUCCESS in 21m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd3d32a770f4d5798f1de6630639e34 : SUCCESS in 11m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2ce6a29af9d4371bd160cf6e83a6b1d : SUCCESS in 22m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f3ec161227044d98d48d0c4dc279e81 : SUCCESS in 21m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edc059114a4c4676b80f2de845719ed3 : SUCCESS in 23m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c97888324654abfa33dd2b633e2e0d2 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/bbcf8020b1ae4d09838eb6d68ed072fd : SUCCESS in 45m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c39c83d57f534e0a84b3783040b210d4 : SUCCESS in 1h 01m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70cb4741462343f9872dd5861119cae0 : SUCCESS in 10m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1481677e2a3f4b89a9434d922d43fad0 : SUCCESS in 36m 47s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/0724406ad75a4697b5f3ff8735e16873 : SUCCESS in 42m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9abb3e447a6a4567b6e0e0107a20470b : SUCCESS in 58m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/efa5c3e012334f1cbdd03304a346eb4e : SUCCESS in 1h 10m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3ac4fbbaa1c740f7bb08eaeb3267ce04 : SUCCESS in 1h 00m 11s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e6bcdca74dd43ecbe83ab5b0ec98508 : SUCCESS in 1h 09m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf4b688e881445759df15cedd3e4f345 : SUCCESS in 58m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf5f57fb509940f3948ba146c6600c08 : SUCCESS in 1h 02m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91292064e49f4ddc9be492ba741c741e : SUCCESS in 51m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d35dfeb1614d4af1ac0f1448cb6f5f09 : SUCCESS in 1h 44m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e5db85b969e64397958f8bccd5c8bbbd : SUCCESS in 52m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3191dcbf980d437a86f228e7519bf07d : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":2},{"id":"de368959fc4460733e9e9518a88f8c104d48af7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 11:34:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0b7fdf951146a3bebd68fb22dbe224 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f37a4151e2eb49cebf190f2ed45c2ade : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbb949379c7acc5ec830c3a7d388aea59e1ab2c2","revisions":{"593b2709a29177a1a95c2c6078ea7fe86556363b":{"kind":"REWORK","_number":1,"created":"2020-10-23 02:52:24.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/45/759345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/759345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/759345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/759345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/759345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/759345/1"}}},"commit":{"parents":[{"commit":"8654eb2d183cf13699e50b8b967a6c52412602ac","subject":"Default dnsmasq --conf-file to /dev/null","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8654eb2d183cf13699e50b8b967a6c52412602ac"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2020-09-28 21:58:54.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2020-10-23 02:52:24.000000000","tz":0},"subject":"ovs firewall: fix mac learning on the ingress rule table when ovs offload enabled","message":"ovs firewall: fix mac learning on the ingress rule table when ovs offload enabled\n\nIn RULES_INGRESS_TABLE table 82 there is a rule for allow established and\nrelated connections. The current rule sends the packet directly to the dest\nport without doing a mac learning. This is causing ovs to age out the dest mac\nof the remote VM and causing the rule to be changed in flood rule. For the normal\ncase it fine as they try to avoid high cpu. ovs hardware offload reduce cpu usage\nby moving some of the packet processing to nic and flood rule is not offloaded,\ntherefore it prefre to use the NORMAL action to avoid the flood rule.\nWe also keep the same logic as today when using explicitly_egress_direct\u003dTrue\nwhich avoid NORMAL action in the entire pipeline.\n\nCloses-Bug: #1897637\n\nChange-Id: I9b611d62be5d0529e8b35e3d8280baa5be54bc2b\n(cherry picked from commit 8fc80b7e132031d18c787b5be582c146d262de74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/593b2709a29177a1a95c2c6078ea7fe86556363b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/593b2709a29177a1a95c2c6078ea7fe86556363b"}]},"branch":"refs/heads/stable/train"},"cbb949379c7acc5ec830c3a7d388aea59e1ab2c2":{"kind":"NO_CHANGE","_number":2,"created":"2020-10-23 03:01:12.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/45/759345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/759345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/759345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/759345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/759345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/759345/2"}}},"commit":{"parents":[{"commit":"8654eb2d183cf13699e50b8b967a6c52412602ac","subject":"Default dnsmasq --conf-file to /dev/null","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8654eb2d183cf13699e50b8b967a6c52412602ac"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2020-09-28 21:58:54.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2020-10-23 03:01:12.000000000","tz":0},"subject":"ovs firewall: fix mac learning on the ingress rule table when ovs offload enabled","message":"ovs firewall: fix mac learning on the ingress rule table when ovs offload enabled\n\nIn RULES_INGRESS_TABLE table 82 there is a rule for allow established and\nrelated connections. The current rule sends the packet directly to the dest\nport without doing a mac learning. This is causing ovs to age out the dest mac\nof the remote VM and causing the rule to be changed in flood rule. For the normal\ncase it fine as they try to avoid high cpu. ovs hardware offload reduce cpu usage\nby moving some of the packet processing to nic and flood rule is not offloaded,\ntherefore it prefre to use the NORMAL action to avoid the flood rule.\nWe also keep the same logic as today when using explicitly_egress_direct\u003dTrue\nwhich avoid NORMAL action in the entire pipeline.\n\nCloses-Bug: #1897637\n\nChange-Id: I9b611d62be5d0529e8b35e3d8280baa5be54bc2b\n(cherry picked from commit 8fc80b7e132031d18c787b5be582c146d262de74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb949379c7acc5ec830c3a7d388aea59e1ab2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb949379c7acc5ec830c3a7d388aea59e1ab2c2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
