)]}'
{"id":"openstack%2Fneutron~760047","triplet_id":"openstack%2Fneutron~master~If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d","project":"openstack/neutron","branch":"master","topic":"vnic_vdpa","hashtags":[],"change_id":"If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d","subject":"Add support for VNIC type vDPA to ML2/OVS and ML2/OVN","status":"MERGED","created":"2020-10-28 04:26:19.000000000","updated":"2021-02-17 15:59:17.000000000","submitted":"2021-02-17 15:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"760047-vnic_vdpa","meta_rev_id":"5b26a5580d10f33806846c379616e872d71858af","_number":760047,"virtual_id_number":760047,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-02-17 10:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 15:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-17 08:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-17 08:49:42.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-17 08:37:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 05:26:42.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 06:02:24.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-10-28 06:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 08:03:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 14:10:36.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-10-28 14:30:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-29 00:09:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 13:57:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-16 13:57:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"25cc6995d27313f46bf05842b55a699df6b592a4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-28 04:26:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9468e30036d2acae7a91fcda4bcc7fd368e16a18","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-28 05:25:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/47/760047/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/47/760047/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"df502d42edd08b945caa6b013d9e99eba65a5e1c","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-28 05:26:42.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/760047/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"13205fc06d12583389d9cfb7663a12799862285d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-10-28 06:02:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/760047/1 : FAILURE in 1h 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"85339cb1c3a2f1974909d371c80eb13e0028225f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 06:42:49.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4212bd750a6148c8be24fbce9498449b : FAILURE in 25m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7bdd730f284149b2a53cdfd6e3377e7e : FAILURE in 23m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/442f47b623cf4f05975a12b50418678e : FAILURE in 24m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dac32fc41fdb41a9adffa3333ee107cb : FAILURE in 31m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cfb6faebc1694a5b9372df29b295f72a : FAILURE in 55m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/21d08c2d25734e92a994ba9aea58a58b : FAILURE in 32m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/112b66872fee41869d670dbb0432ee5b : FAILURE in 17m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb545fb8915b45fd9a83b1add0696896 : FAILURE in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ad25f349fa427b89e421a0688aebf5 : FAILURE in 18m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/037fdd5a49064f139059574b451170c2 : FAILURE in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af906af62164cab9d6df7907912ca29 : FAILURE in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70ae6a0d0704ad8985651b3cb057f3c : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9a86f9363d94c4bae7fd7071a81b30b : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d40759857b4f430c8cf8112fc895c664 : FAILURE in 1h 26m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eadfe46edd3f47faa0acb32aa42d0f24 : FAILURE in 56m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8e276d36557a43a487bdcf68e3a0d402 : FAILURE in 32m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d887f79d70d24c5f98f758c80907ae7f : FAILURE in 59m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/294ca1f63330413181a27c15f2ad4f5a : FAILURE in 1h 22m 43s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/bdbe859685684eaa8f1918656520e850 : FAILURE in 1h 02m 01s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ad2ee2b95fb0435190180a3c14904e3a : FAILURE in 52m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6202a8f12cbf42ccb92c6d31f59e8de8 : FAILURE in 31m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dea68993c0364a87978d2047ac4006f3 : FAILURE in 27m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c1fdf7137c7c48e9ae8847980a448f5d : FAILURE in 36m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/474fbc45e39a4628bf24756e12d12cc6 : FAILURE in 32m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/92869b3cbc0642c9b13a059db5ff9a23 : FAILURE in 57m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8bb739a06af4444b9f4c32459e10de8 : FAILURE in 32m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5542054599f046ac833e18719eb2e9e4 : FAILURE in 23m 32s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2078d38c9d45417db311c346caf0ea04 : SUCCESS in 1h 33m 21s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e29e795733c24e33a307d7d18fd752c8 : FAILURE in 36m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4c23e3691b3743c7a1907394a8344d31 : FAILURE in 24m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9da6c7e840e6428cbd67748f70312bf7 : POST_FAILURE in 33m 11s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/310458d85d3d4feb83dc3e298f6288f7 : SUCCESS in 2h 10m 44s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ed8773a8e394da4a696be31a1ee4996 : FAILURE in 29m 01s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/92efd97488484ff4a71b4c6fadeb4bb8 : FAILURE in 2h 09m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c36a302d787846ee9ad810b58d4eb383 : FAILURE in 35m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/46548fd5028249d69346ec48fb91dc80 : FAILURE in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06e570eb10aecb919289aef2f5938beb153e0f60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-28 08:03:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/47/760047/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/47/760047/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"46fd7ec9aaca8b152e6d50f3c3b34c3ce17b9900","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-28 14:30:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"524f8a42e7ce60015ad58ad1fc4e19a90e423a01","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-29 00:09:53.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/47/760047/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0d11bdab168c4bdfd0f0d16dfc584d8362bd26da","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-17 15:19:25.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7987cd71a26d8e75a29b9ecab1df4a853c5de08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-12 20:12:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb13f46c25d82fe6b3934e61af9b70062e470e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-12 20:52:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b90bb6dd5965338d4a42646b11b295e0bff7d443","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-12 22:52:45.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/760047/3 : FAILURE in 1h 59m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bb812e407babdebf2b61688eace6aa98e688471a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-13 00:04:29.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/47/760047/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"ae1c15ae36dfeb8c7e55c0a5dcb57a33571d8ffb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-13 00:32:07.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/47/760047/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"60f5b8bfb849850a0956a5e0959c8dbd56429265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 07:28:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/94818bb2abb6429786824518ef450898 : SUCCESS in 53m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76e69dd61cf64be6b8dffec675b016f6 : SUCCESS in 34m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5951b9938a61440d892a9f6669b1317c : SUCCESS in 1h 48m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5110432b8c814553bdb268dc441e77d7 : SUCCESS in 1h 29m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e706dd7cb4c4762b8a098492b90503d : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/12d3a005eefc4d6c91af4c21def29032 : SUCCESS in 1h 57m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adf946a08cc04fc7b79ae0a9aed61db5 : SUCCESS in 33m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21a498ce58744f6bb74a41ab81d1c33c : SUCCESS in 35m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcbb1e3afc1d49ad8df705512be87b29 : SUCCESS in 12m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0724847f2ca845569839927b88dc470f : SUCCESS in 27m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f37407f1b8ca48c484f74a7f0edaa0bb : SUCCESS in 26m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f46d3f24511243f4bec831cc32be0130 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18b6211b12b406a9805fe0b8fd1a1dc : SUCCESS in 10m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6dbca8d34cd4303ae5ed0ba6dd84508 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dba0a9cc70444a6488b1cc46be574c56 : SUCCESS in 8m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bab723dcba942e1a6967abc1afa38ef : SUCCESS in 1h 10m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7fa1aa9db8404eb7b75be09f94b2e845 : SUCCESS in 1h 09m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fa0d6bf0de044852849b25b5119c1337 : SUCCESS in 1h 36m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee4efa1f075b412db507bc177200dd75 : SUCCESS in 1h 06m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/731ba418908d43c483ae5f3b80a10c5d : SUCCESS in 55m 47s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/fd575ff0b52942aa80ae57e390f7ad2b : FAILURE in 55m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0152f1a49a3c4c3ca800880d310365c4 : SUCCESS in 1h 17m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bda0bd878a094447882fd6462eb27ac2 : FAILURE in 1h 33m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c8f9fd5090324fd98f4572f61018f6b3 : SUCCESS in 1h 05m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/42ce8405ce674645973da9bf58e885f5 : SUCCESS in 2h 12m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e14d52c97bb7478fad2431dd0cfb2481 : SUCCESS in 50m 06s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/87d90cf9b0ee43f8a9af12169692abb0 : SUCCESS in 57m 35s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ba5d4617bec5472792c712e7372ac7ff : SUCCESS in 1h 12m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6a84a2c7c303435a898912cd23c7683b : SUCCESS in 34m 07s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/aaaf3ed2183a4de3b9c131d1abb9fece : FAILURE in 57m 08s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9751755539dc46b0b1b6b9854e32455c : SUCCESS in 1h 21m 57s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8d07e5e57f5c439ab32afddd4d69cbd8 : SUCCESS in 3h 25m 56s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c72e8ebbe4f94be98bb6f9aeeca6f598 : SUCCESS in 1h 16m 05s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/720e43591b9540e6b196d823dc83b6f7 : SUCCESS in 1h 40m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/240b73c58fef42e7895903acb6fc8369 : SUCCESS in 2h 00m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/70849af338224a0183e3d13fbe431363 : FAILURE in 1h 53m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1c0458df7ead4a088c6cd1580db4f82a : FAILURE in 1h 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"233bb3d96f126d5bf302812458186c6b0aeb691d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-13 07:46:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28ce0aa327804d8fbaf773a95e17edf99a919f3a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 09:15:23.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4ca4cc3c5f29bc486fa30073550fc31360962476","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 09:16:04.000000000","message":"Patch Set 3:\n\n-1 for the neutron-lib version, we should not revert the version we are pushing right now.","accounts_in_message":[],"_revision_number":3},{"id":"46aeff623682a13aa7b3c90cb507a186a48f1d48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 13:57:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5622555ec84012d344d9894c4f1980d088eeef6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 14:47:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa1684cd0dd54febba8a7328ea4bb517 : SUCCESS in 40m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf8e0369d287424fad4a189005845a01 : SUCCESS in 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93324be0ced5ebf4329215b51afde7b878d81d50","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-16 15:08:33.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/47/760047/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"c1e4bf9db2d9dd004380af36f45b21db4403f6dc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-16 15:48:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/760047/4 : SUCCESS in 1h 49m 51s","accounts_in_message":[],"_revision_number":4},{"id":"802c3d07d58625687b4099c27feac8146cb4b7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 16:23:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/edae1a7741df4218ab23774b1eb93a87 : SUCCESS in 52m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/be1c97e63f914423ab743b7a7c1ffadc : SUCCESS in 1h 31m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a342890fab144c2825e510c813b1ec9 : SUCCESS in 1h 33m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbd2a36fb645497e82b82703e54fbb8b : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ff85302044cd4ed0aa02252a6f8b8c87 : SUCCESS in 2h 09m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45c996880ede48889b112405d75987e7 : SUCCESS in 25m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67337ac8cde647dcbaf115736dc3fb16 : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f035314090eb462f91bb111c9964e94a : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f36fa25ef4e84b71ace4c936a58753d7 : SUCCESS in 19m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1714ce914a814080a8fc0474836d3042 : SUCCESS in 27m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa7746d3ce1a408e966ff2c6c29f17a4 : SUCCESS in 31m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c235aac1e34d0a92497d18c9064adb : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99a532fffa564b11ac738eba3992c8df : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e6cac2386cf453099496e4a5013f6d6 : SUCCESS in 50m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98337cfa6a004180a370e6dcb2560ab2 : SUCCESS in 1h 09m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8b9f3ed03251475ba6c8fe72001f0e48 : SUCCESS in 1h 28m 39s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4140ae10dfe04e5184ff75a378c243a2 : SUCCESS in 58m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e83fe3fe998d48a7b84049e6e27fae8f : SUCCESS in 59m 21s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b45aedcc0eaf49669adeda9cdc0a6e74 : SUCCESS in 1h 21m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e3b986042c84178a50ab2a4fef2ef93 : SUCCESS in 1h 38m 52s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/272cc92546044d4ba599b9bc1778468e : SUCCESS in 1h 38m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/926bcd5d6af94c8981789b05d56d55c0 : SUCCESS in 1h 01m 10s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a02820414462449f850ce28bd5c15b3e : SUCCESS in 1h 02m 30s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07d224c0f017448ba2ccadfa54c131d8 : SUCCESS in 1h 12m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9bc2587c50a64f68b24b7e323c7a0804 : SUCCESS in 34m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb5e1cc576b243daa1be734558ef943e : SUCCESS in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9fc339686d061263aaf5f023d9484dce71b188b5","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-17 06:55:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bab63f4f55ac32838879e89d59233c04c8b9e135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-17 08:37:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"13f9abcdbaf07a0fd816b2e86ad6f93ab95aa242","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-17 08:37:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)\n\nI just fixed to nitty nits so I think that Oleg\u0027s +2 is still ok.","accounts_in_message":[],"_revision_number":5},{"id":"ea7da1a1395d08647b1542a922d99635dd2bdbe2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-02-17 08:49:42.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"1f75d1b31328a9e330d20154e749c49a4e3e4ebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 09:29:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3a328b67457461cbb29ea1a7fd88358 : SUCCESS in 51m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d85c7342053e48eaa0ef78e23e83808f : SUCCESS in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70d48b855f7e7e29564b782ecd1ec575b38eebdc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-17 10:10:24.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/47/760047/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"c2ebccd149ecccde732c27be998aeaba175d3a99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 10:40:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d8e84cd960c4d269da62dad810d6c4c : SUCCESS in 55m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1f6a70875acb490bbe5f70f5d9fa9ee2 : SUCCESS in 1h 46m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/48ecdccede664b2bae6347ee8c281f0f : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92c27ef4440740db8ad8871cc91aa26d : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be6be6f4a7464078a9ce0269f1b8b864 : SUCCESS in 1h 58m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1f34c484c844234bc0d64fb546f9086 : SUCCESS in 38m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d0f524a4e83426f8b51562cac32d85b : SUCCESS in 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7e9e2737a84e059dd829207edd40a8 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08de7b28680543b7ae305aacfd4c3f87 : SUCCESS in 30m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beeb60acc86e4bc986207de7756a7526 : SUCCESS in 27m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1c6030e28be48658df7a4bcfd0f0319 : SUCCESS in 27m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ca09ee93b44f39a6dde2054fedbabc : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe103bcbbc194a6bb5b81b2be15b179a : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff4966ca4c534f09a7204ae54c8a1ebb : SUCCESS in 49m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61fb64d509324663a7f98a0fd9eafb3f : SUCCESS in 1h 09m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f0d8174b587b4e55906bb855e2134f75 : SUCCESS in 1h 31m 19s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4179fec942324ee7b3175755ff05bf4f : SUCCESS in 50m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/739932e432d6498e899bc709434c1734 : SUCCESS in 57m 05s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/182c892efe094c59900306545c4f5e31 : SUCCESS in 1h 25m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fde4ffb0c0aa4750948bee2aca7b53d6 : FAILURE in 1h 03m 24s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ea4a191128844aa95dbbcf758faa00a : SUCCESS in 1h 38m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3643bd95a4e47b7bbd8118ec5fd56da : SUCCESS in 1h 01m 55s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8a1d16c265e0438c8f7f378b48cd8e78 : SUCCESS in 1h 00m 56s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0f345ed1cf5842c59a564a401f9c87fe : SUCCESS in 1h 07m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/642d87910754422087f5505f566c8e91 : SUCCESS in 34m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f1f482564ae4c5f85aaf5f097fad8a8 : SUCCESS in 1h 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e068489e60fd7cf19c317fbb8c94c009211f180f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 10:41:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ac0318493d1b15cc9089c5862d8ba173e42de2e1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-17 10:59:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/760047/5 : SUCCESS in 2h 17m 59s","accounts_in_message":[],"_revision_number":5},{"id":"9fddd3bb4aaeceec051ab5d75116758a551965bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:44:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/76c6b04d24574d3694c44562f13799d9 : SUCCESS in 55m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cf91ada3576c41c09619ba2438c4252f : SUCCESS in 1h 35m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/13f4bc636faf4f71accd741ccbe87fea : SUCCESS in 1h 36m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1c51f8834d854161844808a9ceb90ddf : SUCCESS in 1h 36m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78bef5658a624f3d96aef5a3bf24a53f : SUCCESS in 1h 18m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8cc620619ec490998c499dcd29757c0 : SUCCESS in 32m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd36c82233e04abe9c522147b73febc3 : SUCCESS in 14m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44b5dfcb56084f078af610bccaa4ef75 : SUCCESS in 37m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de2b927e223d477685b241a672850bfb : SUCCESS in 30m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad6da4c4ce24880bef5e31380b9edb3 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a2735e67de54fc69891a547aa6ddf83 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62f408daabbb4bde9e4150ad0cebbebe : SUCCESS in 47m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c639b439f5024330861c5f3a7b2cc5b2 : SUCCESS in 1h 14m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ba6b32f2b2ce4cc8a05d2ed90f22b801 : SUCCESS in 1h 54m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2316bd35b0424e94bcb4f3e9e4727110 : SUCCESS in 1h 01m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/526f8e6d10fd4c85a72fb625b997f375 : SUCCESS in 1h 03m 28s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86eb2d08d8914b088bf88d8b370aeb1c : SUCCESS in 1h 47m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97f124000beb44e1a7ce52769cf1e8bf : SUCCESS in 1h 00m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a7edf4da82745a8bccf641ceadbce77 : SUCCESS in 1h 12m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/46d6f70b2e0d43fb8759088fa711d88f : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":5},{"id":"7334ff96cb2894707d2687038053f8cbec52e0e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:44:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5b26a5580d10f33806846c379616e872d71858af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:59:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0a591362324f2bb740dfcd6f26ed01 : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c850bda5ce0c4e19989e5d45a9454aeb : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ae22fe4495f90b69b5d753376d55e6150508ec5b","revisions":{"d12ec125e35a03aa1e8065ac560e59e33cd4486c":{"kind":"REWORK","_number":1,"created":"2020-10-28 04:26:19.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/47/760047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/760047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/760047/1"}}},"commit":{"parents":[{"commit":"6397a03ed861adfd3300351489547a9dfeb26db9","subject":"Merge \"ovs firewall: fix mac learning on the ingress rule table when ovs offload enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6397a03ed861adfd3300351489547a9dfeb26db9"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-28 04:20:41.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-28 04:26:16.000000000","tz":0},"subject":"[WIP] add support for vnic type vdpa to ml2/ovs and ml2/ovn","message":"[WIP] add support for vnic type vdpa to ml2/ovs and ml2/ovn\n\nThis change adds vnic type vdpa to the list of supported\nvnic types for the ovs and ovn mech drivers.\n\nthis change is lacking unit test and is incomplete as a new\nper agent config option is required to enable checkign\nfor switch dev support for vdpa interfaces. this is defered\nuntil after the ptg discussion to adress.\n\nDepends-on: https://review.opendev.org/#/c/760043/\nChange-Id: If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d12ec125e35a03aa1e8065ac560e59e33cd4486c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d12ec125e35a03aa1e8065ac560e59e33cd4486c"}]},"branch":"refs/heads/master"},"e77ecca5e939b0ce831a73b7944cefecb800b62f":{"kind":"REWORK","_number":2,"created":"2021-01-12 20:12:46.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/47/760047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/760047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/760047/2"}}},"commit":{"parents":[{"commit":"c68df5aeecb8d2dda3bae67217d8a04bf5cad241","subject":"Suppress deprecation warnings in policy enforer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c68df5aeecb8d2dda3bae67217d8a04bf5cad241"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-28 04:20:41.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-01-12 20:05:37.000000000","tz":0},"subject":"add support for vnic type vdpa to ml2/ovs and ml2/ovn","message":"add support for vnic type vdpa to ml2/ovs and ml2/ovn\n\nThis change adds vnic type vdpa to the list of supported\nvnic types for the ovs and ovn mech drivers.\n\nDepends-On: https://review.opendev.org/#/c/760043/\nChange-Id: If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e77ecca5e939b0ce831a73b7944cefecb800b62f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e77ecca5e939b0ce831a73b7944cefecb800b62f"}]},"branch":"refs/heads/master"},"e80e74e2c9530beccff7235b9a2314f406ef38f9":{"kind":"REWORK","_number":3,"created":"2021-01-12 20:52:10.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/47/760047/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/760047/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/760047/3"}}},"commit":{"parents":[{"commit":"c68df5aeecb8d2dda3bae67217d8a04bf5cad241","subject":"Suppress deprecation warnings in policy enforer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c68df5aeecb8d2dda3bae67217d8a04bf5cad241"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-28 04:20:41.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-01-12 20:51:53.000000000","tz":0},"subject":"add support for vnic type vdpa to ml2/ovs and ml2/ovn","message":"add support for vnic type vdpa to ml2/ovs and ml2/ovn\n\nThis change adds vnic type vdpa to the list of supported\nvnic types for the ovs and ovn mech drivers.\n\nDepends-On: https://review.opendev.org/#/c/760043/\nChange-Id: If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e80e74e2c9530beccff7235b9a2314f406ef38f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e80e74e2c9530beccff7235b9a2314f406ef38f9"}]},"branch":"refs/heads/master"},"69215778555d0c99f3884bcdf7c3efcfab4082e5":{"kind":"REWORK","_number":4,"created":"2021-02-16 13:57:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/760047/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/760047/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/760047/4"}}},"commit":{"parents":[{"commit":"a128ff67137985570d629592c6915768e626a82d","subject":"Merge \"Remove rootwrap execution (3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a128ff67137985570d629592c6915768e626a82d"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-28 04:20:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-02-16 13:53:53.000000000","tz":0},"subject":"Add support for VNIC type vDPA to ML2/OVS and ML2/OVN","message":"Add support for VNIC type vDPA to ML2/OVS and ML2/OVN\n\nThis change adds VNIC type vDPA (\"vdpa\") to the list of\nsupported VNIC types for the OVS and OVN mech drivers.\n\nDepends-On: https://review.opendev.org/#/c/760043/\nChange-Id: If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69215778555d0c99f3884bcdf7c3efcfab4082e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69215778555d0c99f3884bcdf7c3efcfab4082e5"}]},"branch":"refs/heads/master"},"ae22fe4495f90b69b5d753376d55e6150508ec5b":{"kind":"REWORK","_number":5,"created":"2021-02-17 08:37:15.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/47/760047/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/760047/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/760047/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/760047/5"}}},"commit":{"parents":[{"commit":"a128ff67137985570d629592c6915768e626a82d","subject":"Merge \"Remove rootwrap execution (3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a128ff67137985570d629592c6915768e626a82d"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-28 04:20:41.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-17 08:37:12.000000000","tz":0},"subject":"Add support for VNIC type vDPA to ML2/OVS and ML2/OVN","message":"Add support for VNIC type vDPA to ML2/OVS and ML2/OVN\n\nThis change adds VNIC type vDPA (\"vdpa\") to the list of\nsupported VNIC types for the OVS and OVN mech drivers.\n\nDepends-On: https://review.opendev.org/#/c/760043/\nChange-Id: If22aedc147f7e2256f8f8ad3bebb80b6bb2f6d3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae22fe4495f90b69b5d753376d55e6150508ec5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae22fe4495f90b69b5d753376d55e6150508ec5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
