)]}'
{"id":"openstack%2Fneutron~760372","triplet_id":"openstack%2Fneutron~master~Ifa545158779781d0ed907aceb5958f4de40767ce","project":"openstack/neutron","branch":"master","topic":"bug/1892200","hashtags":[],"change_id":"Ifa545158779781d0ed907aceb5958f4de40767ce","subject":"Adding VRRP fall and rise count cfg options","status":"ABANDONED","created":"2020-10-29 14:44:05.000000000","updated":"2020-11-23 14:53:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba763e3a27b650a8ae9586abdf2b2980b4c853ac","_number":760372,"virtual_id_number":760372,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"date":"2020-10-31 03:32:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-10-29 16:30:48.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-30 03:18:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-29 15:10:08.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"all":[{"value":-1,"date":"2020-11-03 13:17:59.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"date":"2020-10-29 14:45:33.000000000","_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 14:45:33.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"},{"updated":"2020-10-29 15:10:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-10-29 16:30:48.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-30 03:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-31 03:32:54.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01817b078a74f84a65ed04acc8b36f447f246638","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-10-29 14:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df67d79a8ce425cad7feba86872a02208989a2b8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-10-29 15:10:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/760372/1 : FAILURE in 25m 21s","accounts_in_message":[],"_revision_number":1},{"id":"87e1789e88541abbdc180be14353d4793337fa17","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-10-29 16:30:48.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/760372/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"dcd66351a08d7ef95d06696d2d9ff0f9a633ded7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 03:18:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad3b003c6bfc4ff0bf82bc4b457f3bea : SUCCESS in 56m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/29b07862459f4213baa3824d99331c65 : SUCCESS in 41m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cb02df272528438980c261921baa888d : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/25c73d8d8a6f40198565800dc19b8c08 : SUCCESS in 1h 24m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf1eef9a33aa40cd94b1faab67ebc9af : SUCCESS in 1h 35m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e6af9e259fdb472687203192f257fedd : SUCCESS in 1h 24m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/381c0a87d5ad4507b75c01ccc52493dc : SUCCESS in 36m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11a0f44c738e4094af778d2f764fc9cd : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88386227a27744e8a74054788ecc3a3a : FAILURE in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c5a475571b54e30a45b3dbdab5eaa27 : SUCCESS in 19m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18b9678f94374294ab68c246d2018855 : SUCCESS in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b092f695fea14ebfbd69438ce105cd1a : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92f67b1d050945df85998cc8cb11f410 : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0d0931ee16549dd88872234dea0d509 : FAILURE in 1h 11m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f5a008c2f854ca3a05f7bbfee8fe000 : SUCCESS in 1h 23m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/daa567a14f814781a8be0ddb3e701aae : SUCCESS in 1h 11m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7e002d8a4a546fe98c090163949bab4 : SUCCESS in 1h 18m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/766a7b1e0be24cf5bb520bf36de4d34c : SUCCESS in 1h 11m 15s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/6d89da941ab7403eae077e9ef6222a00 : FAILURE in 59m 20s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8342fb68c4024edfb568050c0f29c789 : SUCCESS in 1h 02m 34s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d074ae96dffe4a3e835f956a0dce3897 : SUCCESS in 1h 13m 37s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/275a8639ad14420b82a434311c42888f : SUCCESS in 1h 25m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d18b4a7f194494c92b3844ac83dd91a : FAILURE in 1h 22m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d08d9f5bd2e6480a9d6fd2d9b4803e09 : SUCCESS in 1h 41m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a1d54e7c482c4f158a4e4dd123cce89a : SUCCESS in 1h 13m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc72dd148a2f4fbfba8acb45f750d6f0 : SUCCESS in 2h 14m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a5cc1e5726143379ba695a83f3ceaa6 : SUCCESS in 1h 04m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/482ff3ebc2594e7a945b95edf063ed16 : SUCCESS in 1h 13m 44s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8c5bed326a8f477daf9d601eebf8d444 : SUCCESS in 1h 04m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b9bdc69b18584823b43a8e18d9ae7c64 : FAILURE in 35m 07s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d803bfd55d534b059ebfdbe00c21c9e1 : SUCCESS in 1h 18m 39s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/691235b554e3490190d55eec2c55fea8 : SUCCESS in 1h 42m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9db027838a134934ba29a931b5a239a9 : SUCCESS in 1h 43m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/032203b5354648f799e05eb738c2b8ae : FAILURE in 2h 13m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5ef99b77f0714a4bad13485f1cb1e1e2 : TIMED_OUT in 3h 03m 31s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d68b6146162b43709edf7c2808e653f4 : FAILURE in 2h 14m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/agent/linux/test_keepalived.py","accounts_in_message":[],"_revision_number":1},{"id":"f000156d5e66fa5c84de676fa728d8308b1e2418","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-10-31 03:32:54.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2020/Arista_Tempest_Testrefs/changes/72/760372/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b2f74dc8259e4c38e702f0c544f19e3d075c77e2","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-11-03 13:17:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ba763e3a27b650a8ae9586abdf2b2980b4c853ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-11-23 14:53:26.000000000","message":"Abandoned\n\nIn favor of generic templating in review #759886","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8771d0106299fc19c91731260226451b8312a932","revisions":{"8771d0106299fc19c91731260226451b8312a932":{"kind":"REWORK","_number":1,"created":"2020-10-29 14:44:05.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/72/760372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/760372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/760372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/760372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/760372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/760372/1"}}},"commit":{"parents":[{"commit":"9645f5eb28004ed06883d102e40a6f9bd777cad8","subject":"Converting keepalived config to jinja template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9645f5eb28004ed06883d102e40a6f9bd777cad8"}]}],"author":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-10-29 14:28:57.000000000","tz":-240},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-10-29 14:38:27.000000000","tz":-240},"subject":"Adding VRRP fall and rise count cfg options","message":"Adding VRRP fall and rise count cfg options\n\nChange-Id: Ifa545158779781d0ed907aceb5958f4de40767ce\nCloses-Bug: #1892200\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8771d0106299fc19c91731260226451b8312a932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8771d0106299fc19c91731260226451b8312a932"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
