)]}'
{"id":"openstack%2Fneutron~761391","triplet_id":"openstack%2Fneutron~stable%2Frocky~I70c5620e176bdb07346ea3875ca35e90dd6711f5","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1902775","hashtags":[],"change_id":"I70c5620e176bdb07346ea3875ca35e90dd6711f5","subject":"[stable/rocky] Drop rally job","status":"MERGED","created":"2020-11-04 10:42:41.000000000","updated":"2020-11-10 01:43:13.000000000","submitted":"2020-11-10 01:43:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761391-1604972593523-044f1fb9","meta_rev_id":"d3b59b228618af2166fc7309d27d9a2c4341a650","_number":761391,"virtual_id_number":761391,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 01:43:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-05 16:10:01.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2020-11-10 00:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-11-09 09:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2020-11-10 00:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-11-09 09:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 16:10:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-09 09:29:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-10 00:02:04.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-11-10 01:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b76964d96294940a8a5d37d4f1425033c6950db2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-04 10:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e510d776bc08571c9767a59a46e30518a714029b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-04 10:51:32.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/761391/1 : FAILURE in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b8516cb8ed27278f7b1f6c56575fa2af9148425c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 17:42:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d8f50965b09447438e97cc314a0d455f : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7530acd699b278085d79ad6b448f260ec3d7f56b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-04 19:55:19.000000000","message":"Patch Set 1:\n\nHmm, checkout works correctly, but then pip install in rally-openstack install latest:\nCollecting rally\u003e\u003d0.11.0 (from rally-openstack\u003d\u003d1.7.0)\nDownloading http://mirror-int.ord.rax.opendev.org/pypifiles/packages/0f/82/d57fa5f0fbc065267053426bc139eb37de63ed1d764f230f16ec856d8298/rally-3.2.0.tar.gz (3.9MB)\n\nAs test-requirements.txt only mentions rally \u003e\u003d3.1.0, and upper-constraints does not list rally :(\n\nSomehow only way to fix it seems to add a line to /opt/stack/requirements/upper-constraints.txt to block newer release - but this seems hackish.\nOr just drop rally from rocky (queens is fine as it uses rally, not rally-openstack)","accounts_in_message":[],"_revision_number":1},{"id":"2b575677cdf0db31f7a9f2ade31523dd325b56fb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-05 14:05:19.000000000","message":"Patch Set 1:\n\n\u003e Hmm, checkout works correctly, but then pip install in\n \u003e rally-openstack install latest:\n \u003e Collecting rally\u003e\u003d0.11.0 (from rally-openstack\u003d\u003d1.7.0)\n \u003e Downloading http://mirror-int.ord.rax.opendev.org/pypifiles/packages/0f/82/d57fa5f0fbc065267053426bc139eb37de63ed1d764f230f16ec856d8298/rally-3.2.0.tar.gz\n \u003e (3.9MB)\n \u003e \n \u003e As test-requirements.txt only mentions rally \u003e\u003d3.1.0, and\n \u003e upper-constraints does not list rally :(\n \u003e \n \u003e Somehow only way to fix it seems to add a line to /opt/stack/requirements/upper-constraints.txt\n \u003e to block newer release - but this seems hackish.\n \u003e Or just drop rally from rocky (queens is fine as it uses rally, not\n \u003e rally-openstack)\n\nAs it is old branch which is EM already I vote for dropping this job from the check and gate queue.","accounts_in_message":[],"_revision_number":1},{"id":"2284de0387acad719c2475bbc036bbbc7ac9c216","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-05 16:01:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b71aff0a6ba3ebdbc232eb77dabaa3f126a5e76","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-05 16:10:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/761391/2 : FAILURE in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a76231fbb4e8af650a939ac3677708dd4bc767ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 19:33:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/59ee40bb14704a81a7927980d014bf14 : SUCCESS in 43m 49s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/955418154f0843569b1186e2b9d26fd6 : FAILURE in 29m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/bd0dbd03a0924f09b29d360caed8cd74 : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/c3563e7991904c85b0a90b5d1ecefde5 : SUCCESS in 1h 17m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/8ac5c12bbafa48b3a89cedc69fe5e2ca : SUCCESS in 56m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b63160feacf4bad8ee02184480a3585 : SUCCESS in 12m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c86d4a7c3064d3a9dff6b028a476049 : SUCCESS in 22m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/91b07369bafb4eacb1200ca471333e22 : SUCCESS in 29m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f912ac79299a46589cf566de8e7af1b3 : SUCCESS in 7m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b8b5037ac40e4236af017169e97a6ba8 : SUCCESS in 1h 20m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/65955550006d4c23b8791aba14e2677b : SUCCESS in 45m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4138dce7b0184200ab874962ef10ea49 : SUCCESS in 1h 25m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/edd070d2d80f4358a2815ffca2d6b11d : SUCCESS in 49m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/41413976c5474aa782a1769c9969af1b : SUCCESS in 1h 23m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5fea224f00f9466ebbfcfb4881de99da : SUCCESS in 1h 29m 26s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b907a26571054e89bd8911bca9e4b881 : SUCCESS in 1h 27m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e4e423c61ba459898b52f42161f3088 : SUCCESS in 31m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b479c428ffc4ebeaf5460c3fa53e794 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"061115ad40f1785b4005a1c3647ffe356869a150","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-09 09:28:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c3272bd5c43c35f4a6c2b49f3d6266a828bcf7ef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-09 09:29:18.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"dd02bbd862a4e59eadf91c932b74d4c8bdf830ff","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-11-10 00:02:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8218511b0c7e4f9de7661d5a1d81a2fb77b5e4c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 00:02:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04ed181bdda55ffe8b8e93601bcbcff75080f398","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:43:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"232550f55c71cf770a98ac04d2729ba0280b5929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:43:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/ce59a5d2f3d848d4bbd7d89e9313f436 : SUCCESS in 46m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c83a845c87bd4a778f951d7a9c5e4469 : SUCCESS in 12m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76a2db97ef9b4c968ccde7c25a67bed4 : SUCCESS in 23m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/441b1523e3064a8daf0c7c30fca6dff1 : SUCCESS in 29m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f70866f01eb0411888612f4a95849c33 : SUCCESS in 6m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5cd10b3004c048b3a1aca2cd73ef997a : SUCCESS in 1h 25m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7d723cdf7f2c4ad7aaabbe3870a9876f : SUCCESS in 59m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3c6dcaac4484ddca412eb74d18be576 : SUCCESS in 1h 22m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e18c078483df462696b498a9ca3c175e : SUCCESS in 39m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ac8382d36c8d4ad090c63da8dc016264 : SUCCESS in 1h 34m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fa9dd1e87d8f424589562155f26acfe0 : SUCCESS in 1h 24m 07s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3879942603724c3ba6408648b4c5b2de : SUCCESS in 1h 31m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7e20552c9f142ab86325b45b1e0c6cc : SUCCESS in 29m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ea093777ab44141b199606e2fa5cb64 : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fefd6e15ba84c6f52bc896e1b4f82f9752a25f8","revisions":{"6304dc46ae630603e468f9cf2a7a214fae6355d4":{"kind":"REWORK","_number":1,"created":"2020-11-04 10:42:41.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/91/761391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/761391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/761391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/761391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/761391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/761391/1"}}},"commit":{"parents":[{"commit":"9532cac4ea4487bb95208e16118c23b2082273d0","subject":"Merge \"Do not fail deleting namespace if it does not exist\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9532cac4ea4487bb95208e16118c23b2082273d0"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-04 10:40:12.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-04 10:40:12.000000000","tz":60},"subject":"[DNM][stable/rocky] Pin rally version","message":"[DNM][stable/rocky] Pin rally version\n\nWe already had rally-openstack pinned, but not rally itself. Latest\nrelease does not support python 2 anymore, so use rally version from\nmatching rally-openstack upper constraints\n\nChange-Id: I70c5620e176bdb07346ea3875ca35e90dd6711f5\nCloses-Bug: #1902775\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6304dc46ae630603e468f9cf2a7a214fae6355d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6304dc46ae630603e468f9cf2a7a214fae6355d4"}]},"branch":"refs/heads/stable/rocky"},"8fefd6e15ba84c6f52bc896e1b4f82f9752a25f8":{"kind":"REWORK","_number":2,"created":"2020-11-05 16:01:40.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/91/761391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/761391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/761391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/761391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/761391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/761391/2"}}},"commit":{"parents":[{"commit":"9532cac4ea4487bb95208e16118c23b2082273d0","subject":"Merge \"Do not fail deleting namespace if it does not exist\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9532cac4ea4487bb95208e16118c23b2082273d0"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-04 10:40:12.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-05 15:56:18.000000000","tz":60},"subject":"[stable/rocky] Drop rally job","message":"[stable/rocky] Drop rally job\n\nWe already had rally-openstack pinned, but not rally itself (global\nupper-constraints do not limit rally, and pinned rally-openstack only\nspecifies a minimum version)\nWith rally itself not supporting python 2 anymore, this causes the\nneutron-rally-task to fail on rocky. As this is an older EM branch, drop\nthe job in this branch (newer branches still run rally, and queens also\nas it does not use rally-openstack)\n\nChange-Id: I70c5620e176bdb07346ea3875ca35e90dd6711f5\nCloses-Bug: #1902775\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fefd6e15ba84c6f52bc896e1b4f82f9752a25f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fefd6e15ba84c6f52bc896e1b4f82f9752a25f8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
