)]}'
{"id":"openstack%2Fneutron~762178","triplet_id":"openstack%2Fneutron~master~Iba773781f916221d598021d41656b584cf5e7dab","project":"openstack/neutron","branch":"master","topic":"handle-deadlocks-in-oslo-way","hashtags":[],"change_id":"Iba773781f916221d598021d41656b584cf5e7dab","subject":"tag: use native oslo.db retry","status":"ABANDONED","created":"2020-11-10 17:22:21.000000000","updated":"2020-11-13 14:41:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8c8ac4021c34d0a847246f83b9595b91f6e378f4","_number":762178,"virtual_id_number":762178,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-11-12 16:42:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-11-12 10:09:50.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-12 13:21:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"date":"2020-11-12 11:01:26.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-11-12 08:56:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 05:28:40.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-11-12 10:09:50.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 11:01:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-12 13:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 16:42:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3aadd176214d4674a20cfe4d6f9fbb28822c8106","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-10 17:22:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56b76457fe4c8a0cc26717e9a7c330b77c36551f","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-10 18:18:42.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/762178/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"35e6736e006b73a98b6362cc9197151a6ae1969a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-10 19:29:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762178/1 : SUCCESS in 2h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"13768251ac00c330de80bf69590b67a45fd80f1d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-10 22:54:55.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/78/762178/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/78/762178/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"852656d379cbc9797373bba2f59e8d9f3c3a54dc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-11-11 05:28:40.000000000","message":"Patch Set 1: Code-Review-1\n\nI left a comment on the RFE bug. I\u0027d like to understand which exact problem are we addressing here and whether we\u0027re safe from regressions before going forward.","accounts_in_message":[],"_revision_number":1},{"id":"b4562cd3f5695441d76ac29e0ec7bc3251777138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 07:36:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f5978bd09f4a43f4a3e41a3a78369ba8 : SUCCESS in 54m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/346761f42837405c8bd61b495b3a3913 : SUCCESS in 34m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/185c51047703470aaaddafe98e6d5b87 : FAILURE in 2h 27m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ec97321b253e4cdc9c1b0e015a9308d1 : SUCCESS in 1h 24m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/834edc80e189408090828b69d30c788a : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a7b36255e9254163b0323dfb22a9214b : FAILURE in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62df3cb46a73435c982ebed61d26f778 : SUCCESS in 34m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46cfe5cf12834084995fa766bf6af318 : SUCCESS in 25m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf4d17af0b642ba857740feb50f7935 : SUCCESS in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68c157dfb98c4a11a29d6776c334442d : SUCCESS in 30m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfab6bc86d0940e4907ee8d64a085da1 : SUCCESS in 23m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce231be240d42ff9339b162262ef71e : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/83a77c949828421dbf8af5c292435ed1 : FAILURE in 1h 04m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e8c93b2149e4818a8d060ae853eb3d9 : SUCCESS in 1h 17m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1ae4bafad5774af4bc35ec0240501dd9 : SUCCESS in 1h 22m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf262ebd4d994dc58c6176f6a92ce193 : SUCCESS in 1h 07m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bdf1b669840747698a0229bf3be8f6fc : SUCCESS in 1h 19m 10s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/b49a389c5ddc45198c2dbbcf884dd2a9 : FAILURE in 1h 14m 32s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e02003e530084b36a3593339a2b5b382 : SUCCESS in 1h 12m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/50f9ceafd2b14d7c8ad86a47aed40a6d : SUCCESS in 1h 13m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/54667264304040ccbbfe9315aecffb24 : SUCCESS in 1h 22m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b23c399b80a4c64911e48195177eae6 : FAILURE in 2h 25m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08097a838a8540e0a084f0f9a510595f : SUCCESS in 1h 15m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a6049c704c014b428e2eec23f3df16b5 : SUCCESS in 1h 14m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d16a8420085c47de9bc22b9e07c1896d : SUCCESS in 2h 07m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ae09761b6a94cd1a0740e0797677854 : SUCCESS in 1h 01m 23s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/89bd3ed591104a71b7d0746e082c2dc4 : SUCCESS in 1h 11m 39s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/df516bea2f224106893aba750d973fb6 : SUCCESS in 1h 11m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e7cec092adf24ee4af5643a676e58f71 : FAILURE in 42m 07s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3f55dfcbfdfb4432a7039261d4c6ebcd : SUCCESS in 1h 18m 20s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71b3edb50235434a85df02c9a7636f10 : SUCCESS in 1h 50m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf488e4335ca4c00a6f986ea5865747d : SUCCESS in 1h 50m 29s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/facae75ba7b64e6d897dd8daf00c9b6a : SUCCESS in 2h 47m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e8388268743f490ca265e04674a63a2d : TIMED_OUT in 3h 03m 52s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/16cd840f14744a88a940f760ac627348 : TIMED_OUT in 3h 05m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4cabf6ae81a80b74ba33cbd1da4c9a857fe2b04","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-12 08:56:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0580344bb2a93b8d55765a3ddb2d4107f4621c3a","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-12 10:09:50.000000000","message":"Patch Set 2:\n\n* sriov-test http://52.27.155.124/sriov/762178/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ba108a437ac3bdecd6785d82797a725d1320ca05","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-12 11:01:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/762178/2 : FAILURE in 2h 04m 02s","accounts_in_message":[],"_revision_number":2},{"id":"da9c32fdf8f7832f80ca4f7238b88260dbb85698","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-12 11:01:57.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/78/762178/2/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/78/762178/2/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"0810f484671259b5e395a64ccf67bfdda02b516f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 13:21:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a16948eaf69546f4be55686f0bdf8081 : SUCCESS in 47m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5aa573761d9949e4a55bd184a49be05c : SUCCESS in 30m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2bb2d038487f464a8cd9ed06b48b59f8 : SUCCESS in 1h 09m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c89add9ffd4b480b99649661401b9c2a : FAILURE in 1h 30m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e6b400f02274242a1f82faaf4d601ba : SUCCESS in 1h 25m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3a7172279b164e40aa46497409d14524 : SUCCESS in 1h 24m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f21c033fdb0f4936ba200b57a7021c57 : SUCCESS in 33m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc2358729a844628844d439565045f56 : SUCCESS in 26m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0127b393e38a47c5baa4dc69df091e02 : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6c371dc3fb248f39c0d336587498f7a : SUCCESS in 28m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5279a4cbd0743238b7f5ef63dc6de2a : SUCCESS in 25m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26a369fd33a437fa0b9c2491fe6c7c5 : SUCCESS in 8m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c3da56292ae4d01b1c8d648ade45377 : SUCCESS in 1h 05m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/295736f7ee2640bf9a16feb0daecd7f4 : SUCCESS in 1h 08m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/770b2955ecd640bbb87a9de5205176b5 : SUCCESS in 1h 34m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4722b4db6a8842a1be81f630c2c8acb5 : SUCCESS in 1h 10m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/15f3e81430f748a994eda37994d4f2a0 : SUCCESS in 1h 09m 42s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/91d597f41ed346c29ac902e28f055f9f : FAILURE in 57m 57s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/19a0bb78b2d74816a3c7afffc15eb1fe : SUCCESS in 1h 09m 07s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad1cce01bf7741af9e40262f01f551c5 : SUCCESS in 1h 09m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a6248b94d16545f485f372007f14c874 : SUCCESS in 1h 18m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e7585ef5c434210bfd9091f9c652e82 : FAILURE in 1h 18m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7a2ca90e73bf48388ca9687e56930df7 : SUCCESS in 1h 14m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/61d2d065eb1e4478a78a3e6d48fd6f51 : SUCCESS in 1h 14m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a568e1d0c2a4a22913a26198dfd3f95 : SUCCESS in 2h 20m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f31627cbd7647a09fafaa81ac71ba15 : SUCCESS in 58m 44s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/fc0cc42d22f34c4d9096c17ed21cc215 : SUCCESS in 1h 16m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1a92ded35c194bdb9ecfa10a51b332b0 : SUCCESS in 1h 16m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/05334ea625ff412281eeef6165a13f06 : SUCCESS in 40m 57s (non-voting)\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/f64503653c3d402f8581c4c4662f43f5 : FAILURE in 29m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c32e3b8a795d4ed092419d2e4a69587c : SUCCESS in 1h 23m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d995e2acc09743b993759407fe012c9a : SUCCESS in 1h 46m 00s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4930de58f45a4392a70cafc448a47e8f : SUCCESS in 1h 46m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4c24e970b1ff4751b4d5e70830304e05 : SUCCESS in 2h 47m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0328318f4e534882a949da094236b17e : TIMED_OUT in 3h 03m 05s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/dd94a743bc934dccb4c7ba6adef141a2 : FAILURE in 2h 47m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8461a85e9820736a783633df321c4872ff05f4f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-12 16:42:29.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/78/762178/2/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/78/762178/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"04613daf8f611220615e2679953b042419169bf1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-13 14:41:34.000000000","message":"Abandoned\n\nBased on last drivers\u0027 meeting discussion (http://eavesdrop.openstack.org/irclogs/%23openstack-meeting/latest.log.html#t2020-11-13T14:09:08 ) let\u0027s wait till new db engine facad migration is ready, and see how that affects issues like the one mentioned in related bug","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d065dfe1dfcbf21416fda169e56a6fc9af9140a","revisions":{"a1fc9790b59971b13e9e9d13e951fb076d82d30a":{"kind":"REWORK","_number":1,"created":"2020-11-10 17:22:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/78/762178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/762178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/762178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/762178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/762178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/762178/1"}}},"commit":{"parents":[{"commit":"e6dc0276a57a6d03904dbb474349fb9ed9675ef5","subject":"Merge \"Log exception generated in Controller.prepare_request_body\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6dc0276a57a6d03904dbb474349fb9ed9675ef5"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-11-10 17:04:56.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-11-10 17:20:04.000000000","tz":60},"subject":"tag: use native oslo.db retry","message":"tag: use native oslo.db retry\n\nChange the db retry  decorator to the oslo native one.\n\nChange-Id: Iba773781f916221d598021d41656b584cf5e7dab\nRelated-Bug: #1887523\nspec: https://blueprints.launchpad.net/neutron/+spec/handle-deadlocks-in-oslo-way\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1fc9790b59971b13e9e9d13e951fb076d82d30a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1fc9790b59971b13e9e9d13e951fb076d82d30a"}]},"branch":"refs/heads/master"},"4d065dfe1dfcbf21416fda169e56a6fc9af9140a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-12 08:56:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/78/762178/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/762178/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/762178/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/762178/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/762178/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/762178/2"}}},"commit":{"parents":[{"commit":"5b48b7ceff2fb7cecfa6b2afbf8874c371407f69","subject":"Add Loki testing for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b48b7ceff2fb7cecfa6b2afbf8874c371407f69"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-11-10 17:04:56.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-11-12 08:56:08.000000000","tz":0},"subject":"tag: use native oslo.db retry","message":"tag: use native oslo.db retry\n\nChange the db retry  decorator to the oslo native one.\n\nChange-Id: Iba773781f916221d598021d41656b584cf5e7dab\nRelated-Bug: #1887523\nspec: https://blueprints.launchpad.net/neutron/+spec/handle-deadlocks-in-oslo-way\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d065dfe1dfcbf21416fda169e56a6fc9af9140a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d065dfe1dfcbf21416fda169e56a6fc9af9140a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
