)]}'
{"id":"openstack%2Fneutron~762874","triplet_id":"openstack%2Fneutron~master~I40761b30563749251b9d74731bbe7a80a124da89","project":"openstack/neutron","branch":"master","topic":"bugfix/1722126","hashtags":[],"change_id":"I40761b30563749251b9d74731bbe7a80a124da89","subject":"Copy existing IPv6 leases to generated lease file","status":"MERGED","created":"2020-11-16 16:44:01.000000000","updated":"2021-06-29 02:17:22.000000000","submitted":"2021-06-29 02:14:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"762874-bugfix/1722126","meta_rev_id":"8ac35cddeabf3cd0ab4385a591391ad3ce91cd91","_number":762874,"virtual_id_number":762874,"owner":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 02:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-25 07:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-06-28 12:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-28 15:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":2,"date":"2021-06-25 14:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-28 15:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 17:44:43.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 18:05:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-16 20:20:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 22:19:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-17 05:19:15.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-11-17 11:41:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-14 13:35:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-06-08 19:21:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-08 22:18:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-21 08:06:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-25 07:37:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-06-28 12:06:17.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"ee3ca3138460659da4f410b1b407831ea074e6e3","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2020-11-16 16:44:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e0649cd2082053d2023e534ebc831a2b5a81005","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 17:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cde63898a5d275fbf847498a78bff28352aec9bf","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2020-11-16 17:11:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca07ba5f262dfbe92bd37cb33301d11b46de6342","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-11-16 17:44:43.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/762874/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b154ee26de6b6bb240aa27e72f671a186d19b2e2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-16 18:05:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/762874/1 : FAILURE in 1h 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a4a3b44cd99aaa8e9ac25d764204446a9cad5b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 20:20:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0c7c29592fc44207a411c5d7cd095df1 : SUCCESS in 1h 02m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c4e7c0adfd8d4d3684f78bd71cc5c89e : SUCCESS in 32m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5798de4cb0ff4327b1a958d8d9e0a1e8 : SUCCESS in 1h 53m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/288339c7e60d4583b26b31180dbfb7cd : SUCCESS in 1h 33m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34c7e46a5eb041b1a92d7e7ac7197d01 : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/51d6d856efc84c1cbd2dab8291132605 : SUCCESS in 1h 23m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e27bb9e4400494a86302193cff0efc2 : SUCCESS in 31m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fd1ff5fa22e443292438adbd268b715 : SUCCESS in 26m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed60d908d8f4ed785980d1e71994ce1 : SUCCESS in 12m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebb1dce846484b6d998ee8e685217257 : SUCCESS in 25m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3275e52de50a4b8caded8dbd59de1adc : SUCCESS in 26m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8eba45b77fb457c9f0e719528a3c9a4 : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49442d0e9bc14c9b872c9992d8194487 : SUCCESS in 1h 11m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f666fe0a91b42b5aff2e1fdeb4fba58 : TIMED_OUT in 2h 15m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/66afb35064e540b69cd3d6813b02d217 : SUCCESS in 1h 50m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4321bb92ba7a40ae94da8ce87fc7d4c5 : SUCCESS in 1h 07m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e41292aa9f2d4ce985f3d223136cd574 : SUCCESS in 1h 18m 18s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/44592edfc883425cb5dd672359175e18 : FAILURE in 52m 46s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3d4003564786426e8fdb19a25241c688 : SUCCESS in 1h 23m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef37f38ea4a040cf816acfd9dfb46ea1 : SUCCESS in 1h 19m 08s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f783fa135a664fd2a7b07bcfca3ae9e4 : SUCCESS in 1h 38m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/055ad9e501d54613b2ecc2dbc1480a85 : SUCCESS in 1h 25m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49155421a0484d3db7a6e759cff287b3 : SUCCESS in 1h 02m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/280bd2e5cd16445588b4323126b4499e : SUCCESS in 1h 07m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94761e7de6d049808e0ea8f27b53580c : SUCCESS in 2h 04m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2aa6c686811b4fb1bfdad375e6e3b21b : SUCCESS in 1h 01m 36s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ebd7c6c3163e4407a8fd79f0d05e1eff : SUCCESS in 1h 11m 29s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/56bba6ea079a485c86dd1c4b8cf3db4d : SUCCESS in 1h 38m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a1127ab89b12402dac1b7bf7e5be46b6 : SUCCESS in 49m 55s (non-voting)\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/7a313215b584436cb6c5c3245f27449c : FAILURE in 1h 33m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4838f7b5ea3b4c4b9f6a25b908f3ef79 : SUCCESS in 1h 22m 59s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0e49bb27f5ad453983786227a98969a4 : SUCCESS in 1h 55m 57s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/285b238343d3459d961d63d07f5a31c8 : SUCCESS in 2h 02m 13s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a6ec3afb88fd49fd9226b57c9799964e : FAILURE in 2h 00m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/66f4be7173c6453994846c99704b2262 : TIMED_OUT in 3h 05m 06s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bb439089cc2f4ab0b9ea8c7540529f4f : TIMED_OUT in 3h 04m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e0966f3d986e753d2fc19fdd4f19ec5b3dc79c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 22:19:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d43b5edec386e0c5c23f96c5e451db522aa8f4fa","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2020-11-17 07:51:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6df32f99b28810621731563af473d158b8cb156b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-17 11:41:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af8940845e939b55673ff4b0a47081e1e04f6a59","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2020-11-24 08:14:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5733acc855963b463d7311a39e9db5ad5af0f666","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:36:04.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"e5ad0efd18626fdcb4a51d30ee4cbe002625eec8","tag":"autogenerated:gerrit:restore","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-07 21:30:12.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b51745cace4491ca3ec115a5454c2f15480202cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-07 21:30:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16bcb9c3e5b67fba4570913c628e1df39b047872","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 22:07:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a1f6a2f461f4af096b389c0e09a0a6f : SUCCESS in 31m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/146409e2532747d5b6f371fcb4c07042 : SUCCESS in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"650abe93fbcbf1358abe110a87b4df27bd03727b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-07 23:26:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762874/2 : SUCCESS in 1h 54m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c590381bd6b43bc00d53f81359a854f62e12c170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 23:53:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/979ef83a307c49178947c9bad4db7a51 : SUCCESS in 2h 00m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72363a35d10348cda93623a7fc22133f : FAILURE in 2h 19m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e938565d88f441bba688265a59a31765 : SUCCESS in 1h 32m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d982015617514a0587ff498e6342f6bb : SUCCESS in 2h 12m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff178511f0a457c8396b222e8927de6 : SUCCESS in 40m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb60b176ffd04abaaf5b6c3edf70d1ae : SUCCESS in 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748716c0a7e74e1c86a150420d8f2e18 : SUCCESS in 17m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adb6e48c82734f6bbd02cc9f821897df : FAILURE in 32m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dac683ed1c34c3e82129d9982768f09 : SUCCESS in 26m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/800c4205c94e44ceae4e5445db789d31 : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08886080b9414efe8450d50e53bbf3f2 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1124b5770abf444ba89adc4105af5596 : SUCCESS in 53m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bb8f9cc40304848a8161df155832671 : FAILURE in 1h 17m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4e1143aaa1884939b9b073704472592a : SUCCESS in 1h 22m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87406978ac384683bcc7e72e2573be61 : SUCCESS in 1h 04m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/185b7640445f447fa843cd84b082a1bf : SUCCESS in 1h 14m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3fb0ead2766d47b0a19a6eae343c0c82 : SUCCESS in 1h 17m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/efd8634e7d0340df98cdb7d8ce751efe : FAILURE in 27m 47s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f6de9b0535344939f495f21ba8fccf0 : SUCCESS in 1h 45m 02s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dab02531b704737a0880c67e7df3f64 : SUCCESS in 1h 11m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9265da9ae2649d8bee3c83a7179439b : SUCCESS in 1h 13m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7c3850ff030149128fa1ee5a6ee08c50 : SUCCESS in 33m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4f2e6b2810a40979a251b4ad520746b : SUCCESS in 1h 30m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/210719ec4d82438cb0b7b41406cf9a62 : SUCCESS in 1h 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0730f7e4bfd015931e9c3bdb22727072b229ea3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-08 07:29:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b950d2fa3a9881e2b58ad8da890f58fe5284c767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 08:21:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86a4561790bf40da916826db3903fded : SUCCESS in 49m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/247fe1b7dbd64d81838e39b672f13cc5 : SUCCESS in 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf4925fca89408327fbbe6f04a42faeccf13a7a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-08 08:45:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)\n\nI think this is a good idea, some minor nits","accounts_in_message":[],"_revision_number":3},{"id":"8e7da7676e5aaeea2780faa7a8b2b9d7175b6941","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 09:22:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762874/3 : SUCCESS in 1h 42m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ea88594ffaecd02f6a0066805bfacdc51b36422a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 10:07:40.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/70d6b6654a684474a1718e87d9c9f7cd : SUCCESS in 2h 32m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a01bb211d7e14391a9a8ec3972e594d0 : SUCCESS in 2h 11m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ebdce3a5c0144c9a8c073551015f20f : SUCCESS in 1h 33m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4ef0551b53d4443b8d0d92c33e57516f : SUCCESS in 2h 15m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c103c71cf3034c9188f770fe07948b0b : SUCCESS in 38m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc7ebbc7f2fe41ed93553265932f1b93 : SUCCESS in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c95150a0764e2f83df0d6ce1dd1f98 : FAILURE in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d151aa37fad44d7a9aef6d9f0ed26a3b : SUCCESS in 23m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27f301056a8f4e4ab4a1529e32f9758d : SUCCESS in 29m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d3a674edb34a788ada0d7c4d58373c : SUCCESS in 28m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60bd2f42437646c9ac3faed13c389f7a : SUCCESS in 14m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7afb981239a741b68c6cde97236e0ff3 : SUCCESS in 45m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c73cf6421ca4c45bf94438b9789b3e4 : FAILURE in 1h 12m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/81a5bfc182f64dd1a6660b269e7a051d : SUCCESS in 1h 18m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/431d302cdc2d4e55a50032056a245825 : SUCCESS in 58m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/86f5aaea827f4e87b39ab9b82c62ac23 : SUCCESS in 1h 04m 48s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bb258bf5d2ac4efd8b785ef547c09be4 : SUCCESS in 1h 29m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88c32508bb3246dfa8b9fb6e25c6a9b6 : FAILURE in 32m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/643c38926e634822800d1037f42cb5ad : SUCCESS in 1h 46m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fe940a4c1004eb3a68c73d6971fbff7 : SUCCESS in 1h 01m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d587cbb575184ad0bb6017f7325eb183 : SUCCESS in 1h 02m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/35e98d4875354f16bbd928d368a2b9f3 : SUCCESS in 35m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b94d84e223064ca5bbb4afcfffa9cba2 : SUCCESS in 1h 40m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5bfb238df5254dd28dda551be49cc4ed : FAILURE in 1h 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"854e5f0cfac7372cb7badd15845e3e5190ab5a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-08 14:59:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ae9e418e28402ffd80b1eaa4d39446b1bcd29285","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:49:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bb67ff43acc491bb7e9911b2cf98c25 : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c69314d3063a44d9804a75647ebe207e : SUCCESS in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cca3d116a30690f21b1e8438b4705cb158779fc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-08 16:20:00.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2d8f6ae987934e2a781560ffc757954c6e54acb9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-08 16:30:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f888dbf6d307ae399677b7fa714684dbb627d51e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-08 17:14:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762874/4 : SUCCESS in 2h 11m 31s","accounts_in_message":[],"_revision_number":4},{"id":"d93a3e3bfaca19bfe2ce1ee44812eadd58f4afb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:18:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6f1a1707b441497a937be3b008987aab : SUCCESS in 1h 58m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ca9c74b56de04fb28a1661947d9d85c0 : SUCCESS in 2h 16m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a68e97a55f14f199394e6c1c19b98ee : SUCCESS in 1h 34m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3b58f330276146169bcc35b87f0b2d60 : SUCCESS in 2h 19m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/457d51453d73442ab95381a61e5aa27e : SUCCESS in 46m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e74dec54fbb84d09a5850b5727744cdf : SUCCESS in 38m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1589150b03946588336d567e456d82a : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88904b3fd2f24164bd28af64fe721de4 : SUCCESS in 28m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed83c166681b45b2a611cf295f3d8d21 : SUCCESS in 34m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0c9ef0341ab46a9a9a5640749ed9618 : SUCCESS in 26m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3a59c9b70d42728b36e2bc2f8a5c1f : SUCCESS in 12m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/57adce21726d4d5d85bee698609b5449 : SUCCESS in 48m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b43cbd0b995043799de9f9085ee18b98 : FAILURE in 1h 10m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7fac37360ccd4de7b5729c6f1dc9b8c6 : SUCCESS in 1h 14m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/195307bffdef4518a05a9564bc10c1f8 : SUCCESS in 1h 03m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/00c1e507545c40d6b5b109454b041ae2 : SUCCESS in 1h 02m 05s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/15f27f08958d459fb343a7fb1e889b6e : SUCCESS in 1h 19m 00s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e853f62d9f5b4abea7598179b13de503 : FAILURE in 30m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fca792386aa54e44aa21c1d04cc67d7e : SUCCESS in 1h 45m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/700d9dad4bbc4d4ab598575593177c04 : SUCCESS in 1h 03m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5ae4bac5d70b4583b2c0c9556beb7ca9 : SUCCESS in 1h 12m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/df129e99c62c4483b73f3c736c9e9240 : SUCCESS in 35m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ac0eb865166447cb9f94fab582a87da : SUCCESS in 1h 18m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f5d539f48c2749a6af51fb4bc47f6709 : POST_FAILURE in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8e0344c370e65dd4bff44e35fb86b9a3a0bf6cd8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 19:21:28.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/74/762874/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"35ab6c9b8d0ad088565951593bc00e178cf06f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-08 21:50:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a73c8a0c421ce05e1f37df31737291861752cada","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 22:18:43.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/74/762874/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"0d383bd1497134dbf804813e35b9427b0aab87ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 22:36:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c270d07f504949a79f0cbfaa6012bcb4 : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/035b64429cb84ce98508da8de8813c98 : SUCCESS in 42m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"172ceb39eac3f8157b50fbc755782821d66d6a79","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-09 00:06:19.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/762874/5 : FAILURE in 2h 14m 39s","accounts_in_message":[],"_revision_number":5},{"id":"7498feb72b89a0b32e966ef9e30cd2e1c54c6748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 00:27:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7d6c8d0e56134bd78d84f43993b8fc5f : SUCCESS in 2h 32m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f38eaab0a8104a598f38020f59aa6b40 : SUCCESS in 1h 46m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0827e44123454d6f83ae8b2a9bd126e4 : SUCCESS in 1h 08m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c76e6c36ecf84cc5b6aba5d907a5abd2 : SUCCESS in 1h 53m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fccf2726a9674612b5993f865b59cd86 : SUCCESS in 44m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5723e4d4125d4e338e0374a83d025c2c : SUCCESS in 32m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85831b03f62944e3ab75e7fb1d26dfcb : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9413e82b141e4f32a25d1ee1a691c1f2 : SUCCESS in 34m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb78871a5ac4f9ca8dd78cca8dba2af : SUCCESS in 29m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c905c7f6beca42f8a72a377a5bc6fb79 : SUCCESS in 30m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b15b4a3a1864bffa5a7acbeef5a6b25 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6eb5acfdc1eb438aae7ca00d94de0915 : SUCCESS in 46m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76eb042090c044618ba15945a7643ecc : FAILURE in 1h 21m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/15430c411cbf4e1c9454000a57622627 : SUCCESS in 1h 05m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db01d5632cd94ce4a623e53bc15f0c97 : SUCCESS in 1h 16m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d109cd72a37748ceb5af3ea3a023cd50 : SUCCESS in 58m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/457dbeba95e24fe0ad1460a3f88e2363 : SUCCESS in 1h 18m 57s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1711d189e53f463f98d2378548d226cf : FAILURE in 28m 46s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/364664e9f45e43d39c837418d17a49e2 : SUCCESS in 1h 35m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7fecf49fe3f45f1af55cf5623fc5e6b : SUCCESS in 1h 06m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/de2e741407d84771a5239d17ccc2ab9d : SUCCESS in 1h 05m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4a9a72a76b754c429724e170177892d0 : SUCCESS in 36m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2f999cfa367401a929439badae30a5f : SUCCESS in 1h 09m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bfac6b9c9d6842888626343dd7f2b9c7 : FAILURE in 1h 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7ea9c93e2c4a4d72496d1b44bddfaba0161efe74","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 03:05:40.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/74/762874/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"823fa8fb9fb267d9bf4ddb5c6960cc9802db4bdc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-09 03:55:44.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/74/762874/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"943c257408976cca157946f01a731007dde38b2e","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-09 08:58:10.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a8d53ed912cb2d37f80c191f66eecd3476388231","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-14 07:51:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4d020a4b0084430e36a8865aefc7346fb07b7ef3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 08:46:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70d97daace9c4fe287961af0c6e3e899 : SUCCESS in 47m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2be705e053243aa83987b8d697ed6bf : SUCCESS in 48m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c2d7e925a42d67be456d16b62dbb6794d2cd2a2c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-14 09:50:57.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/762874/5 : FAILURE in 1h 58m 47s","accounts_in_message":[],"_revision_number":5},{"id":"5511983572add40cf94ef968b022df25723bcc21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:59:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/463b53ca4c1646e28b49933ccddbe002 : SUCCESS in 1h 34m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9e4c478a837b4852b9df05890874f434 : SUCCESS in 2h 03m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06097a6273bb46f094593bafc14227f2 : SUCCESS in 1h 16m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ff29ce337814d908dc200c0a9a42837 : SUCCESS in 1h 46m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d4da11634d24b0ba1d75a89b84ed7d0 : SUCCESS in 36m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9690888fb4a41068190191512657c5f : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a9681bfa764fe5afe3553f569d0196 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d235c038e93442978aa6f92a7d59654d : SUCCESS in 29m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b6f94d52fb40349445b2471523a975 : SUCCESS in 27m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b0bf696f21248108db689c144c422fa : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8d9cf3b66a44139092229bcb07f4d2 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/32533528727a4ca893a001b2aa69bb81 : SUCCESS in 42m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd8a02788d7e49a388a7108df710a227 : SUCCESS in 1h 19m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/912908ae4048474fbcd5cfda5b9f7c2d : SUCCESS in 1h 27m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b568f12192044659b14bdd856eeb170 : SUCCESS in 1h 00m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f28b49a85e04e38a96547def2581abb : SUCCESS in 1h 01m 40s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/45737e68522c4111935c9ebd69ebd4c2 : SUCCESS in 1h 13m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/acd3440f74e94fb5911e386d38ba6b12 : SUCCESS in 1h 20m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/852114e465ab402c86d283d2bad5a843 : SUCCESS in 1h 34m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66531d82c9464d188d5f32353406e3ed : SUCCESS in 48m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ea001b1c850471e8f43dd234ae4bc5f : SUCCESS in 56m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6910c5a72cc1456b91226187a047e19a : SUCCESS in 31m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ad27e39b1e74052b6e859bde8b4ef71 : SUCCESS in 1h 18m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/18562aaa7f4c4d5f9f8dccf384841572 : SUCCESS in 1h 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8f3175033c080b5fa4209c76e76ba30a7a465c78","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-21 08:06:50.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb93f9d700fd36ba4cbc753d9986289701cbdcb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-21 12:57:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b98d99496959d8e3c86ff7648076d95230927f94","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-21 12:57:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a561bc4e6ba3a3427de1560f33dc1054d29f1ff7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 13:40:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3a8f07ec7404cb08352dfa71f2f0d8e : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/10c35983e1964bfa89c8948a8dd5055f : SUCCESS in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a095409661155ee7b1a6aac4223a3806925d2a32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 14:54:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/229a8977dc2a4205aaa64f6deb9ca4fa : SUCCESS in 1h 48m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f8ad457e6cb04f5d9cb266ccce630bf5 : SUCCESS in 1h 55m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a57530ab4e949a5b0e1e85df257a632 : SUCCESS in 1h 29m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ad79ac2d53df407ebf46eb6b5b7c130c : SUCCESS in 1h 30m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a9c9c76ad42417cb325cb54d784785b : SUCCESS in 31m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d14b8a512d9845ed88aab2fd586b6961 : SUCCESS in 35m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c9ff065f10c4f45b744443956725559 : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ffd4de97e6045be917ee4aa594b6fdb : SUCCESS in 22m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baf276c3833c4c3e86ec4cdfe272ba41 : SUCCESS in 21m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf5e52c4d85f4f2cb4c110ea13495491 : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73770224977d4009be70548534f02781 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26bcb07208034daea467f67ddf9c2377 : FAILURE in 47m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e9455d2e4b64ecbbf4742f5b6ca0cb8 : FAILURE in 1h 14m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d9b6c94113f0400e98c44412a0b17bbe : SUCCESS in 1h 08m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/622a29ff42e84aaf87dbc1ac595139ab : SUCCESS in 1h 02m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bcb0c7dc0af14c8392ab71e5e1ce073f : SUCCESS in 1h 07m 23s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/99b54852a95c45af8eb8fa94d95e7a21 : SUCCESS in 1h 08m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/df0f40b17b2142af8b3ca1a5a3fdd163 : SUCCESS in 1h 13m 33s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b69d0e1906f9475486b13ad874ff65c6 : SUCCESS in 1h 24m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e982e2a7f934581add229f0888942d3 : SUCCESS in 46m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f3e87b3b7a4a4148b942c93bdedc8b24 : SUCCESS in 1h 01m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4845f6979c754869a122f63ada4dbc9e : SUCCESS in 31m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43067544c6194816a4ebf632b8f2fd64 : SUCCESS in 1h 11m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/be4c06d160fc4967a710be719166c06e : SUCCESS in 1h 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"34aafdfe63b53c8725606b599e659ed40bf5969d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-21 15:12:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762874/6 : SUCCESS in 2h 14m 35s","accounts_in_message":[],"_revision_number":6},{"id":"ad094d6d5e6b3d9fffb832517ba4c844bb6a37d0","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-22 06:36:55.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9e189954d3ae2fbff50161bc549687f2f4ccc900","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 07:32:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a754b3afd97345dfad196b142fedffff : SUCCESS in 43m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2054e41105b0433e83abe2b9aac1f9d5 : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"df4169495b61ea22ad53cd617bb489dbadded2d2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 08:27:33.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/762874/6 : FAILURE in 1h 49m 49s","accounts_in_message":[],"_revision_number":6},{"id":"43daf2a9451741d671a7d3752abda3e142c0137d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 10:08:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7e549680f86a40e4ab3f3a0cf24fd474 : FAILURE in 1h 51m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/286cd3c9fbd24d1c9e8b54594955934e : SUCCESS in 2h 12m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e00b94660fec4f1990bda8f0b0d33881 : SUCCESS in 1h 30m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e26716fd6d974ea585a6623c524640ac : FAILURE in 1h 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53e09abca77e434dbf4a8670105af9d2 : SUCCESS in 36m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21ab7caa87704d67bcf0497c17cc1733 : SUCCESS in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c6ea1a4d524e4788367fec8a70e9ac : SUCCESS in 17m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b601bab980df42d3bf660a7db01df5d3 : SUCCESS in 27m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acb07f32bf584ce2a43062fc8a30ed22 : SUCCESS in 24m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b54bee55ac43480ab7a0db3d349fb2e3 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9cc382fd2a4b0bbf3553d111ecf8a2 : SUCCESS in 12m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/133f3fa7813e4925922e530709fd1c84 : SUCCESS in 53m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/917dfb85af394e1e88f6a091317220d7 : SUCCESS in 1h 31m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/92b845db920c4b2a893b22e34157f1af : SUCCESS in 1h 08m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23220e2ba6634fc48d71eee9dec783cf : SUCCESS in 56m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7cb6e9eb48b44692a5edf4f8ad89bff6 : SUCCESS in 53m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f23fb38bf0b64821aacaa3f5b2120772 : SUCCESS in 1h 06m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1313c19e2f134de392e88468a3c607d6 : SUCCESS in 1h 22m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97ad6b0a5c904c1db66c1adec49896e8 : SUCCESS in 1h 22m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b18206cdd2d0477aa373b74ad7b2a435 : SUCCESS in 47m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/49f4d6bd5a3b4622885d8362bf6291bf : SUCCESS in 1h 04m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1d0923337f0c4987bc9e620b66062a1f : SUCCESS in 33m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e254fdc8e77f470499b345e6f41806a7 : SUCCESS in 1h 11m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67240ab4d224432aa6396a5d176c4d7c : SUCCESS in 1h 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1e289de1a572b42f4296853be7a6f55aa44a1fca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-22 10:45:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c3982672c6914ad951e79c1057815e64fa7d8b3b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-24 20:23:36.000000000","message":"Patch Set 6:\n\nPlease, add https://review.opendev.org/c/openstack/neutron/+/795186 on this patch.","accounts_in_message":[],"_revision_number":6},{"id":"0fc10bcd56e61b77269882bcf7df85d260d6d128","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-24 20:49:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e9d6d36f5deb847229e53a7d88ea7c8f5bc3c059","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2021-06-24 20:50:54.000000000","message":"Patch Set 7:\n\n(1 comment)\n\n\u003e Patch Set 6:\n\u003e \n\u003e Please, add https://review.opendev.org/c/openstack/neutron/+/795186 on this patch.\n\nI merged this into this change. I also implemented the suggestion regarding netaddr.IPAddress. Please review again.","accounts_in_message":[],"_revision_number":7},{"id":"77d7c76f7f9ea743997aafcb85c360d11741ad14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:28:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daeed0bc80d6437ba19956a1c79c3c74 : SUCCESS in 31m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcf639b71e0f43a2a89a887b437b0ffe : SUCCESS in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"06aeb38fbc5c1d610975d8febbddcd1796b6e621","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-24 22:21:07.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762874/7 : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":7},{"id":"f72264697a4b1d7e666fa29c9489edc525c9ef6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 22:59:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3fc431ca69604cebbfc11dfecc8b1797 : SUCCESS in 2h 06m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5315dd20812345cd876bc821bf3967e6 : SUCCESS in 1h 41m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ba3f7f235354db7b722e44d95f9b98c : SUCCESS in 1h 15m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4a1ab1847a1048d4aca7c09372c7d681 : SUCCESS in 1h 58m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d09f1e7230f54626bd4766fc6fbeddcb : SUCCESS in 37m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03d98eb4b16347b9b418c1c2b3f982b8 : SUCCESS in 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd543ff359447dfb09167ebee787d62 : SUCCESS in 15m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0109e316dfe1456ebdb437bbf0005a5a : SUCCESS in 25m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7c2b58fa0ee4d15b33655722d1fe5e2 : SUCCESS in 25m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f572c52366cc41fabcb219492b72f2ab : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08eaa79f8c7841ba83c46180cf3ef663 : SUCCESS in 12m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/815cc25d1ba54ee397779d157a32ffee : SUCCESS in 52m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3db566773a444c80bbf90c70fad0f9e4 : SUCCESS in 1h 27m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f0652c74df6c44d7af2b75a3e028363e : SUCCESS in 1h 07m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e4a3ce45073423fad8a5971b4c8ae2d : SUCCESS in 1h 02m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/920628e7f0744ef69af0bc87c817cf18 : SUCCESS in 58m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/24f818b1af9b43ebb10543b4b6f5baef : SUCCESS in 1h 07m 34s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f119482192349c68bc9f913f3686fa4 : SUCCESS in 1h 19m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9fa042b608c451aa25870dba3c9a433 : SUCCESS in 1h 34m 42s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61804f8abfdb49afa03de84795f640ce : SUCCESS in 58m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d096ce39c8b34b81b69797775e94563f : SUCCESS in 57m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0fb487f96437446e852a6af761d84048 : SUCCESS in 28m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/342fed83d3e74d2b8df5f3853d2c88a8 : SUCCESS in 1h 14m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/45484736b6b7401b9de720a4873fb442 : FAILURE in 1h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe2f969fa9ba7e2e82504f461159dbcbe3400ea5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-25 07:37:52.000000000","message":"Patch Set 7: Code-Review+2\n\nLet\u0027s wait for Brian\u0027s (and other folks) review, but I think the code looks good.","accounts_in_message":[],"_revision_number":7},{"id":"9adbe3eabfb1e51c54073569cb878f679b6a3299","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-25 14:55:02.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"71a7973fbe46b0dafb118b5d85624ffc56797231","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-06-28 12:06:17.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"f86bfa68f7e494d7ca7e3ba9e83989c22910993a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-28 15:19:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33283968681441fd06e92ee8760cee300ad3d775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 15:20:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f5cb6b9eb2872368cd0d91255fdee48dc4e55f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 18:23:37.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3699ea709c6648ba98ad9c72a242fb86 : SUCCESS in 1h 57m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/46766287a8284389a0b65c8411300b5b : SUCCESS in 2h 43m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d8144a3ee00443d294e034c485304ac0 : SUCCESS in 1h 30m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d50d199f5b6410d97a09e2e9f8d906e : SUCCESS in 1h 39m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/383aaf722e15484298ebf88d3b9a1261 : SUCCESS in 44m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb26fd655734b818f45d90ec92ae96f : SUCCESS in 16m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2186c8f18f5f44b398758106b7141022 : SUCCESS in 28m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48e898a763e147eab4442606b449740f : SUCCESS in 32m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2df5647a4074eb182fb60fbe22da6f6 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdeba15313a04f72a85ccf2944040152 : FAILURE in 45m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe7e2f4316824536870f63dfbe045f3e : SUCCESS in 1h 17m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f7ce3be62f50493f8e2d8619180c8f75 : SUCCESS in 1h 04m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa45a957ce7c4d11acd6940e7add759d : SUCCESS in 54m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b06736ef985f49ba818b64941d480f23 : SUCCESS in 1h 42m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3cad1cb4d29432f8892f35c6dda282c : SUCCESS in 1h 04m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6f03730d18c246fea2f60746c246e67b : SUCCESS in 1h 00m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7e2dd237f4c34b828e62142e8bb070c5 : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":7},{"id":"a913216cfc62fd7f519c4a9ec261c25355282fd6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-28 18:48:32.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0c563610ddbd58c3aa3e94d8139d177dcaac7859","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 19:31:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1abe1d57962e441bac23dac6f630ed04 : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae29d3728c75484da043eb21fca7ab68 : SUCCESS in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"527344232058a5bb29de510fd4f0f8ab2e1a056f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-28 20:33:45.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/762874/7 : FAILURE in 1h 44m 34s","accounts_in_message":[],"_revision_number":7},{"id":"8af35ec61b9a0b48a5abd190d91a1b0ba817d16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 21:14:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d3793e951c8148a4a909db353e8029d0 : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/45adade4eb0f4220afb0e10763e7dc89 : SUCCESS in 2h 20m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f878a78673774c5d9b8546e8d18bcb8c : SUCCESS in 1h 39m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/26a4ee418409449e938208edd74f0ee1 : SUCCESS in 2h 15m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3533f8221de49c8924ea9e340325869 : SUCCESS in 38m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fda3762549de4bc0be07af6c17a9a874 : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6c212dc10c4d29996f679096d319f3 : SUCCESS in 16m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66e46b67c6d744b593348f5b03ca25af : SUCCESS in 30m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9259f7c4cc4443e3a88d0a913995f220 : SUCCESS in 26m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d878f82517b4824ac61c8d97b0ff3cb : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28459f22dccb41e9a9f5491988bb4b74 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f830065b1254662a52d8bad4d9b3cca : SUCCESS in 51m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/002e3a24aae445f7a28be5fb43e7f984 : POST_FAILURE in 48m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3a032c6cc2b64e32845c4bb686cf068e : SUCCESS in 1h 12m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4988a907ca6e4d3eb1df0de659378d0f : SUCCESS in 1h 11m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24a336e3ba424538b9aad1ca0821b371 : SUCCESS in 1h 10m 12s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/69b0784635ee439da01b0b8ffbbebcb1 : SUCCESS in 1h 14m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2e834be764f4c488b614145db26014c : FAILURE in 1h 04m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2fc1da27d41d479cbd95bb83fd5c834b : FAILURE in 1h 22m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/792225c7135e49a4a820d33c87b229e9 : SUCCESS in 56m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5fb3b69f65764b8e8881418e739c0825 : SUCCESS in 57m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a4ae782dd150452fb4e39347398eb52f : SUCCESS in 30m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3d4898b48a48460eba544c8c197b874f : SUCCESS in 1h 10m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/06822463f04e4592b0e793f21f518331 : SUCCESS in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4863190c248899e19041ffa620655935d3363895","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-28 21:19:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0b269bb73b04a60ed944635762194f19637b1ce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 22:07:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/811f527fa6654c33a243df91c9b3d0fb : SUCCESS in 45m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e026429a2122413cb5536b087f9634bb : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"605e9d34b14abd8da5dfb71991e0dd3eb99fcac0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-28 23:33:14.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/762874/7 : SUCCESS in 2h 13m 14s","accounts_in_message":[],"_revision_number":7},{"id":"a15e2db54051a259df65d8e8c9c2e8e25145b9fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 23:48:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d1f5617cc9a54bfe99d451dfc8d39d62 : SUCCESS in 1h 45m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c7b29caafa5943aeaf432fba5150cf38 : SUCCESS in 1h 37m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/63697ae3f6c84a268cdebb3bd3e598ed : SUCCESS in 1h 27m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6c344ace50a7402c9fceb789278d4f3e : SUCCESS in 2h 05m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/617539f7582c4d02b4588abb5e92348d : SUCCESS in 40m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc837047a7c8489dbd5296febe6d0b4e : SUCCESS in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b12c738a3fb40918122079bafe69ed3 : SUCCESS in 16m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eaef15b31b7415b881c448903fa3f91 : SUCCESS in 26m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83b3cfed823c4e1fa59e529f82098185 : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd9e9b98511246898e60b6a893c3474e : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b53e2061d9544edb041150392f55f40 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1cc4bfdbeef146868a799806fa1e121b : SUCCESS in 49m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd61c405539c4e4caecdec90b325201b : SUCCESS in 1h 23m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf84bd4530d1486190fc1f8b31d2a917 : SUCCESS in 1h 08m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3837ffe42bc461ea0486a4c987b6cf8 : SUCCESS in 1h 00m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60e1a20f4cfa42d7af9c4474cd94e4c4 : SUCCESS in 1h 03m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b54fac232f64654a7f5c96948be4efb : SUCCESS in 1h 13m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3599df25cbfa4109b5cbc456069c35a2 : FAILURE in 1h 08m 44s (non-voting)\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/99cd3d9421b146f8bbf4ba8a13ccabeb : SUCCESS in 1h 29m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/830ef64dd7fa44fab816093c7e05a17d : SUCCESS in 55m 27s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/56e5aa958e914ff4a09bc24448a77369 : SUCCESS in 56m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ef3e1d83600648fa8cd1d387e5a9c5b2 : SUCCESS in 34m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1780b7af4304b11a51b6b8e4a485446 : SUCCESS in 1h 10m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f5e90a3f9ccb47b6bde4e25e52806dae : SUCCESS in 1h 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9665c9787441d23f7b7cc2d9808b817605d16383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 23:49:03.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9a7c8f85394912e336fadcbb29fa48cc8f11c7f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:14:22.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a5a9f727da254d009bf57de77eaad81e : SUCCESS in 2h 20m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/424157edcdfe4998b81800fc772f446d : SUCCESS in 2h 06m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0ae618c740984f419c2ed5214ec55299 : SUCCESS in 1h 55m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c67ccde2b7b4d03aa2d1171d4573c12 : SUCCESS in 1h 33m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8865484b3c547ff9a98cb4fd590b7da : SUCCESS in 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a322594573f34df3ad622fab13d26fe0 : SUCCESS in 14m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d78cbc577e5744649499652c982a30b4 : SUCCESS in 24m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84ce64cafbc949e19f4fe4e4f803de40 : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c287640c60a545cb83d13733b760ffb8 : SUCCESS in 12m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/90894cf4cd87473397d26a4ec67692bd : SUCCESS in 49m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ebcdb1e7c1b4d2199f4c86d5f3d248c : SUCCESS in 1h 01m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/66ec044bba80460e9a7498a76233fd2e : SUCCESS in 1h 04m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a8d6f37a15d4e16918745f36934c8ce : SUCCESS in 1h 07m 07s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/28f9f6d75c6447f1afb4617ef5ac4ad1 : SUCCESS in 1h 24m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be8e303d39f5480a8f7da6926752aa53 : SUCCESS in 57m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/73b79086ce8d4145966a7c4e6d9c59ab : SUCCESS in 51m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f7873137e040d88ae2a9fbd596cc7a : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":7},{"id":"aaba6aa7e911aa968a93b5d575f2999bb7c1befa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:14:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"8ac35cddeabf3cd0ab4385a591391ad3ce91cd91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:17:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a9a1c4938c4ba2a49cea7f3f39f002 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6bc1c00d66ced82dd142739790222e8408684696","revisions":{"e92683d6436dac63d359e323988e0906d7a7766c":{"kind":"REWORK","_number":1,"created":"2020-11-16 16:44:01.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/1"}}},"commit":{"parents":[{"commit":"40c501dcb158b1ab14137c94c60697e19a73c170","subject":"Merge \"Add process name to AsyncProcess\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40c501dcb158b1ab14137c94c60697e19a73c170"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:43:47.000000000","tz":60},"subject":"Skip generating lease file if it already exists","message":"Skip generating lease file if it already exists\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are excluded\nwhen generating the lease file in case of DHCP agent restarts. This causes\nDHCPv6 clients to fail to RENEW their lease and to go through a full address\ndiscovery cycle with possible short connectivity disruption.\n\nGenerating the lease file is not neccessary if a lease file already\nexists as is the case if the agent was just restarted due to a\nconfiguration change or host reboot. This commit skips the lease file\ngeneration if it already exists.\n\nA lease file is still generated if an agent is migrated to a different\nhost.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases\nas they are present in the lease file.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e92683d6436dac63d359e323988e0906d7a7766c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e92683d6436dac63d359e323988e0906d7a7766c"}]},"branch":"refs/heads/master"},"b4ca6f52ed623d6095bb3c8c7fcc8eec841d7da3":{"kind":"REWORK","_number":2,"created":"2021-06-07 21:30:26.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/2"}}},"commit":{"parents":[{"commit":"6a7b52a2c89e6686fbb6504407c8f91c3c246807","subject":"Merge \"Improve AddressScope.get_bound_tenant_ids method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a7b52a2c89e6686fbb6504407c8f91c3c246807"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2021-06-07 20:50:42.000000000","tz":120},"subject":"Copy existing IPv6 leases to generated lease file","message":"Copy existing IPv6 leases to generated lease file\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are excluded\nwhen generating the lease file in case of DHCP agent restarts. This causes\nDHCPv6 clients to fail to RENEW their lease and to go through a full address\ndiscovery cycle with possible short connectivity disruption.\n\nThis commit copies the existing IPv6 leaes from an already existing\nlease file if present. While this does not allow for DHCP agent\nfailover, this is still better than just skipping the IPv6 addresses.\n\nA lease file without the IPv6 addresses is still generated if an agent is\nmigrated to a different host.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4ca6f52ed623d6095bb3c8c7fcc8eec841d7da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4ca6f52ed623d6095bb3c8c7fcc8eec841d7da3"}]},"branch":"refs/heads/master"},"47a0fa87280954ef857301bd27c8df00a5f914fd":{"kind":"REWORK","_number":3,"created":"2021-06-08 07:29:59.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/3"}}},"commit":{"parents":[{"commit":"6a7b52a2c89e6686fbb6504407c8f91c3c246807","subject":"Merge \"Improve AddressScope.get_bound_tenant_ids method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a7b52a2c89e6686fbb6504407c8f91c3c246807"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2021-06-08 07:29:46.000000000","tz":120},"subject":"Copy existing IPv6 leases to generated lease file","message":"Copy existing IPv6 leases to generated lease file\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are excluded\nwhen generating the lease file in case of DHCP agent restarts. This causes\nDHCPv6 clients to fail to RENEW their lease and to go through a full address\ndiscovery cycle with possible short connectivity disruption.\n\nThis commit copies the existing IPv6 leaes from an already existing\nlease file if present. While this does not allow for DHCP agent\nfailover, this is still better than just skipping the IPv6 addresses.\n\nA lease file without the IPv6 addresses is still generated if an agent is\nmigrated to a different host.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47a0fa87280954ef857301bd27c8df00a5f914fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47a0fa87280954ef857301bd27c8df00a5f914fd"}]},"branch":"refs/heads/master"},"0a2a68d093f079db6b9c8dd68fc38acc2234d2b6":{"kind":"REWORK","_number":4,"created":"2021-06-08 14:59:51.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/4"}}},"commit":{"parents":[{"commit":"6a7b52a2c89e6686fbb6504407c8f91c3c246807","subject":"Merge \"Improve AddressScope.get_bound_tenant_ids method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a7b52a2c89e6686fbb6504407c8f91c3c246807"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2021-06-08 14:49:59.000000000","tz":120},"subject":"Copy existing IPv6 leases to generated lease file","message":"Copy existing IPv6 leases to generated lease file\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are\nexcluded when generating the lease file in case of DHCP agent restarts.\nThis causes DHCPv6 clients to fail to RENEW their lease and to go\nthrough a full address discovery cycle with possible short connectivity\ndisruption.\n\nThis commit copies the existing IPv6 leaes from an already existing\nlease file if present. While this does not allow for DHCP agent\nfailover, this is still better than just skipping the IPv6 addresses.\n\nA lease file without the IPv6 addresses is still generated if an agent\nis migrated to a different host.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a2a68d093f079db6b9c8dd68fc38acc2234d2b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a2a68d093f079db6b9c8dd68fc38acc2234d2b6"}]},"branch":"refs/heads/master"},"9cc40d86deecfbf290070dd45a096ff050af4301":{"kind":"REWORK","_number":5,"created":"2021-06-08 21:50:58.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/5"}}},"commit":{"parents":[{"commit":"6a7b52a2c89e6686fbb6504407c8f91c3c246807","subject":"Merge \"Improve AddressScope.get_bound_tenant_ids method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a7b52a2c89e6686fbb6504407c8f91c3c246807"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2021-06-08 21:50:41.000000000","tz":120},"subject":"Copy existing IPv6 leases to generated lease file","message":"Copy existing IPv6 leases to generated lease file\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are\nexcluded when generating the lease file in case of DHCP agent restarts.\nThis causes DHCPv6 clients to fail to RENEW their lease and to go\nthrough a full address discovery cycle with possible short connectivity\ndisruption.\n\nThis commit copies the existing IPv6 leaes from an already existing\nlease file if present. While this does not allow for DHCP agent\nfailover, this is still better than just skipping the IPv6 addresses.\n\nA lease file without the IPv6 addresses is still generated if an agent\nis migrated to a different host.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9cc40d86deecfbf290070dd45a096ff050af4301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9cc40d86deecfbf290070dd45a096ff050af4301"}]},"branch":"refs/heads/master"},"cf20743b5612b60fcc30841a5d4f1a306c17ec8e":{"kind":"REWORK","_number":6,"created":"2021-06-21 12:57:21.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/6"}}},"commit":{"parents":[{"commit":"6a7b52a2c89e6686fbb6504407c8f91c3c246807","subject":"Merge \"Improve AddressScope.get_bound_tenant_ids method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a7b52a2c89e6686fbb6504407c8f91c3c246807"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2021-06-21 12:57:09.000000000","tz":120},"subject":"Copy existing IPv6 leases to generated lease file","message":"Copy existing IPv6 leases to generated lease file\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are\nexcluded when generating the lease file in case of DHCP agent restarts.\nThis causes DHCPv6 clients to fail to RENEW their lease and to go\nthrough a full address discovery cycle with possible short connectivity\ndisruption.\n\nThis commit copies the existing IPv6 leaes from an already existing\nlease file if present. While this does not allow for DHCP agent\nfailover, this is still better than just skipping the IPv6 addresses.\n\nA lease file without the IPv6 addresses is still generated if an agent\nis migrated to a different host.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf20743b5612b60fcc30841a5d4f1a306c17ec8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf20743b5612b60fcc30841a5d4f1a306c17ec8e"}]},"branch":"refs/heads/master"},"6bc1c00d66ced82dd142739790222e8408684696":{"kind":"REWORK","_number":7,"created":"2021-06-24 20:49:29.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/74/762874/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/762874/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/762874/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/762874/7"}}},"commit":{"parents":[{"commit":"e12ba06e8bc959b35ccbd338e0e359a4224d4c9d","subject":"Merge \"use payloads for PORT BEFORE_UPDATE events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e12ba06e8bc959b35ccbd338e0e359a4224d4c9d"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2020-11-16 16:41:18.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2021-06-24 20:42:09.000000000","tz":120},"subject":"Copy existing IPv6 leases to generated lease file","message":"Copy existing IPv6 leases to generated lease file\n\nBecause the DHCP agent does not know the IAID (identity association\nidentifier) of assigned IPv6 addresses it\u0027s not possible to generate the\nlease file including IPv6 leases. Because of this IPv6 addresses are\nexcluded when generating the lease file in case of DHCP agent restarts.\nThis causes DHCPv6 clients to fail to RENEW their lease and to go\nthrough a full address discovery cycle with possible short connectivity\ndisruption.\n\nThis commit copies the existing IPv6 leaes from an already existing\nlease file if present. While this does not allow for DHCP agent\nfailover, this is still better than just skipping the IPv6 addresses.\n\nA lease file without the IPv6 addresses is still generated if an agent\nis migrated to a different host.\n\nThis commit complements the fix implemented in\nIb1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e which just skips the IPv6\nleases as otherwise the lease file would be invalid and all leases would\nbe lost. It does not change the behavior for still valid IPv4 leases.\n\nWith this issue fixed an additional fix is required to not loose DHCPv6\nleases when the agent restarts dnsmasq. Currently the DHCP agent\nregenerates all configuration files on restart. This means that DHCPv6\nleases are lost as they can\u0027t be regenerated. This changes the agent to\nonly delete the config files if the agent\u0027s ports are also removed.\n\nCloses-Bug: #1722126\nRelated-Change: Ib1b2f284ab81f1c4af7b08b5257b45a3f6e79c3e\n\nChange-Id: I40761b30563749251b9d74731bbe7a80a124da89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bc1c00d66ced82dd142739790222e8408684696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bc1c00d66ced82dd142739790222e8408684696"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
