)]}'
{"id":"openstack%2Fneutron~763349","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ib0d3a501f88c149baea7d715c7cfe5811bc85e4f","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1892846-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Ib0d3a501f88c149baea7d715c7cfe5811bc85e4f","subject":"Fix migration from the HA to non-HA routers","status":"MERGED","created":"2020-11-19 10:54:15.000000000","updated":"2021-01-14 16:28:42.000000000","submitted":"2021-01-14 16:28:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763349-bug/1892846-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","meta_rev_id":"3a40b60598ba30bad49df241c68a3918ff9444ec","_number":763349,"virtual_id_number":763349,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 16:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-19 10:56:43.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2020-11-30 05:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-11-27 08:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-01-04 20:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-01-14 11:07:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 10:56:43.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-19 17:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-27 08:46:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-30 05:50:38.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-01-04 20:20:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"13750f76e0c0868e33dae561febf313fcec1af8a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-19 10:54:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"805a2f014ee298194bc642bdaeb8f7a3aa4e0e45","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-19 10:56:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763349/1 : FAILURE in 25s","accounts_in_message":[],"_revision_number":1},{"id":"763fc1bc8959916f55fadc3c7d6922a7e3e824ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:03:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/af3962fb7c4b4d4fa8567fe933dfa34e : SUCCESS in 37m 25s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/c00080dbe5f742639eed996b581e7b00 : SUCCESS in 56m 05s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/39303c88d4224f9c9cdff7da1966d68a : SUCCESS in 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873347a1f0f54c05bd8e320bc9eb4790 : SUCCESS in 11m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7b271eae27a4764ba9865b56d7d5fd6 : SUCCESS in 18m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb6366f8b69b49dfa2d72ba27c474c60 : SUCCESS in 26m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/851f57d7281f4ee694eeb640e2944d17 : SUCCESS in 7m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed00caa1a85e422c9556eb8928418961 : SUCCESS in 1h 25m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/136eeeac607f449c9663f1dd72a6a213 : SUCCESS in 55m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad046b006a79489db75022ad86aa66a4 : SUCCESS in 1h 21m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c1d5b99d5564456fa2eed0eeda56028d : SUCCESS in 39m 58s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/ad5827e6371843748c43e4f5ea452484 : SUCCESS in 1h 43m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b78019b8fff34334aed5c66719dd3baa : SUCCESS in 1h 25m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b5fffdc887074db9bbc18cded1221050 : SUCCESS in 1h 30m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27e55a81af104c349864ee1b55921c37 : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":1},{"id":"78b0121d69d20a3d228fca55795c2b9748ffcf28","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-27 08:46:09.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport up to queens, nice!","accounts_in_message":[],"_revision_number":1},{"id":"e1a172b8fe3d8037b17957e2f50aa9a2fbeb66a3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-30 05:50:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8e6619ccb1141491bfea38ba969fdb0fef4f5d30","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-04 20:20:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61c4dabf79f9de5e7d26978c2c68573d7acff62a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 11:07:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"169cb72953f43805263a3e145e42c489597da7dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 11:09:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cff703f2fa8f168a10da27d5e37420d6d6818c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:28:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/a78504bdea524c20b790b20400dac677 : SUCCESS in 42m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283e4bda2eac42379dc5819c0421a711 : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e183a62187f4988ab4f636617244b96 : SUCCESS in 22m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88fcd92db98a448aa0a3fbbeae391de5 : SUCCESS in 16m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1c8b8bf1ba774241808190c7ef61ba02 : SUCCESS in 7m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d25eeffed7534ff0b2dbe6ec6ee7ee2f : SUCCESS in 1h 11m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bfe79604259541ce877ec782e7ac1852 : SUCCESS in 38m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd52f93112254431b9cf8e8a3f2dc39e : SUCCESS in 1h 24m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/33b35d623562475aa87f50bfd44414e5 : SUCCESS in 35m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2481efe1c17d49b0ae3cc7b6ae1fc263 : SUCCESS in 1h 22m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/45403389ffa141ffbb01a1b85642b385 : SUCCESS in 1h 24m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1182ccbee795404ba1863be42b043366 : SUCCESS in 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3a40b60598ba30bad49df241c68a3918ff9444ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:28:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02ad1b7909628f435515eca043c5b849d937176c","revisions":{"02ad1b7909628f435515eca043c5b849d937176c":{"kind":"REWORK","_number":1,"created":"2020-11-19 10:54:15.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/763349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/763349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/763349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/763349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/763349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/763349/1"}}},"commit":{"parents":[{"commit":"0dce8dc53ea95da5603c3b3b9c7f5c37577e0ba4","subject":"Merge \"Revert \"[Security] fix allowed-address-pair 0.0.0.0/0 issue\"\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dce8dc53ea95da5603c3b3b9c7f5c37577e0ba4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-25 12:44:26.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-19 10:54:15.000000000","tz":0},"subject":"Fix migration from the HA to non-HA routers","message":"Fix migration from the HA to non-HA routers\n\nIn case if during switching HA router to be down, there will be any\nfailure, router_info will be stored in L3 agent\u0027s cache as HaRouter.\nIn case when next update on the router is migration to non-HA router\nthis is wrong class and it causes other issues, e.g. with\nremove_vip_by_ip_address() which is correct only for HA routers.\n\nThis patch fixes that issue by adding check of the router\u0027s ha and\ndistributed flags and update local cache with new router_info class\nin case if at least one of those flags don\u0027t match.\n\nChange-Id: Ib0d3a501f88c149baea7d715c7cfe5811bc85e4f\nCloses-Bug: #1892846\n(cherry picked from commit 489e0ead7297e3b17ca2bf8c4bea9701ad14a939)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02ad1b7909628f435515eca043c5b849d937176c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02ad1b7909628f435515eca043c5b849d937176c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
