)]}'
{"id":"openstack%2Fneutron~763744","triplet_id":"openstack%2Fneutron~master~I0cca37b626a94a05fb983a8528c22a660e89e673","project":"openstack/neutron","branch":"master","topic":"remove_abstract_property","hashtags":[],"change_id":"I0cca37b626a94a05fb983a8528c22a660e89e673","subject":"Remove \"abc.abstractproperty\"","status":"MERGED","created":"2020-11-23 08:03:32.000000000","updated":"2020-11-24 13:45:25.000000000","submitted":"2020-11-24 13:43:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763744-remove_abstract_property","meta_rev_id":"8c6b60b179bb6b76fd4e3cfcd13e6b4f8ab33bc7","_number":763744,"virtual_id_number":763744,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-11-23 11:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-24 13:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-24 10:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-11-24 11:09:49.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-11-24 11:09:49.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 10:32:01.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-23 11:12:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 14:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 10:19:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-24 11:02:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-24 11:02:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-11-24 11:02:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"575a71c6616258f4c0deb4aa47b301a46d0d119e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-23 08:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ff4df11b0a77e2b4908a7fafb46c876408a5471","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-23 10:32:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/44/763744/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/44/763744/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d25137202081cc0a0fa5c6329a3be8aebe3f9553","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-23 11:12:31.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/44/763744/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c44e4fc09a6810d79da236de0c0305f8efcf42b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:10:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f959939f1614470aa205e541fd3699cb : SUCCESS in 48m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04bd626a8fd0406e91170ebc3157f821 : SUCCESS in 35m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7309e9ed9b0a4111a2a279589eae0c7a : SUCCESS in 1h 38m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/69e8a132f57946e197f9353aec20169e : SUCCESS in 1h 37m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c3afeef0aebb41dbacb8e4fb6bc78676 : SUCCESS in 1h 29m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/696be346feda43f280381adf8e9dd581 : SUCCESS in 1h 28m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1c7a6fe782d4cf9a4dead468a425965 : SUCCESS in 30m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fdb0c76376a486fab288f907c0f8171 : SUCCESS in 24m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fdf0ff9dc2410dbb9ff3f873084a9d : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f70db83c64234cac90b541dd7d43f5a5 : SUCCESS in 24m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b8872b8b620498196b81f8a43858111 : SUCCESS in 25m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/599fd24ceebd4c2396a60bae7ec107d0 : SUCCESS in 8m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec03288c2f5641e0aaa57f533ec7bac4 : SUCCESS in 1h 11m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c030a4fb6ab0484997dc00d7537d1242 : SUCCESS in 1h 11m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3f785d6287194db7a6015d1b0972c14c : SUCCESS in 1h 57m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b30105e3e46f45ec9470b44d6bd8071e : SUCCESS in 1h 00m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f611ac9a2f44bbda084848cfc9e5dce : SUCCESS in 1h 09m 45s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/2f096b7c21554f7eb248fbc0fa2c3c44 : FAILURE in 55m 51s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/044302d79f3b4f36a0a8c475f88cd205 : SUCCESS in 1h 11m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1072fc37160d45cd8476256dc5f5b724 : SUCCESS in 1h 26m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/150c9bc3cad34467ae8fb707d538a574 : SUCCESS in 1h 31m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/105cbf6338124d9fac60932182ad8ce7 : FAILURE in 1h 26m 16s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bc774d53153e4b9a884fe96b0b4d604a : SUCCESS in 1h 14m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5e0ee682f9504c4e984a7a673c55454b : SUCCESS in 1h 02m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/285f7de60ffc4519b246c7ee1524c302 : SUCCESS in 2h 11m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54d9dad4ae8c448ea820080ba60d328e : SUCCESS in 1h 09m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/027a2258dc0442c091249273c266a86a : SUCCESS in 1h 18m 22s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8acb36a309a84982bd6f48ded39c2647 : SUCCESS in 51m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0be07e954afe4dccaf09cd63e6bf2a1f : SUCCESS in 46m 18s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/687f4d41ac6e453cbfd94a2e38404b96 : FAILURE in 1h 02m 42s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d6cd7a063165404ba483e35d9a353587 : SUCCESS in 1h 17m 39s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a0eaef907ebf40aebac2f6a346c0a144 : SUCCESS in 3h 37m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/820e73c476fd49569c9900bbac8daddd : SUCCESS in 1h 27m 20s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac3cd1fdee604655983e3458b775ac3b : SUCCESS in 1h 44m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2c7680636c5747bcaf206e1d6d9a3115 : SUCCESS in 2h 07m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2d49327089a846d7af61f42f643744bf : TIMED_OUT in 3h 02m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7c473f3cb3154f2ab35637eda9ab68a4 : TIMED_OUT in 3h 03m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3bad890202346c75a536cd8cf6a5039e04ba0fc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-24 10:19:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4823965f39b44a25eee122f3149580599d61d56a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-11-24 11:09:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed019cbea3614960f417364eb88ce6b29009acbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 11:10:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a6a8cad5115deaf3c1c27722c44f8056320fe08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 13:42:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0dd7ef853a1e422ca8e2594e78056cc4 : SUCCESS in 58m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7be2ec54fa86490cb059a09e1d95c7c8 : SUCCESS in 1h 29m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/07a50a4e7e604143986e5a46e6a2b631 : SUCCESS in 1h 22m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5ceb31698a974ee9aa2c15b449fce507 : SUCCESS in 1h 39m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87e433b0ccba429bbf3bc69cb7772677 : SUCCESS in 1h 39m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c3bf7f9078f464bbbde70101dfbc5fe : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fe8afe07804388bfc99d3212ec3130 : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4a663a25ff4a339868fee2529e8ade : SUCCESS in 27m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38a0f442a61c43b783f4033e868037ea : SUCCESS in 23m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd46b8a3aa94ab1ae52d2637ed48939 : SUCCESS in 17m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3760c66b7b0549c7a12fb91331681620 : SUCCESS in 1h 13m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b4a990e909842c58d0cc6d0176aaf57 : SUCCESS in 1h 20m 52s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1afd977afd143f3b61223e2ed2e1226 : SUCCESS in 1h 24m 56s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/93721d75f1d749238462b4d52fad1651 : SUCCESS in 1h 40m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b5ecc0b92c7475ebe6db39f4c1f78dd : SUCCESS in 1h 01m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/293fa49bc3b140cb88c04dd5df4f8b16 : SUCCESS in 55m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21e1fba93ed64325ba218de49ed9fc7e : SUCCESS in 2h 24m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a37fb6f63ee401c87a34ca732a18164 : SUCCESS in 1h 11m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6bea2989a324499ea53886cf2420dc41 : SUCCESS in 1h 32m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/db4f9b1c175e4b9ea62494a47e75042b : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":1},{"id":"24413c158b68b5567bf64423af66b6911324c9f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 13:43:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c6b60b179bb6b76fd4e3cfcd13e6b4f8ab33bc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 13:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16b7ef952aa4dd3a185bc4455e033ef : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"790fff1d24db32d961a4020ad0f3c0a52de3ab34","revisions":{"790fff1d24db32d961a4020ad0f3c0a52de3ab34":{"kind":"REWORK","_number":1,"created":"2020-11-23 08:03:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/44/763744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/763744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/763744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/763744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/763744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/763744/1"}}},"commit":{"parents":[{"commit":"144c70f896405f8d92721ef31148672cb5d72673","subject":"Merge \"Set metering iptables chain not found LOG level to WARNING\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/144c70f896405f8d92721ef31148672cb5d72673"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 15:59:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 16:02:07.000000000","tz":0},"subject":"Remove \"abc.abstractproperty\"","message":"Remove \"abc.abstractproperty\"\n\n\"@abc.abstractproperty\" is deprecated since 3.3. Now it\u0027s possible\nto use \"@property\" on top of \"@abstractmethod\".\n\nChange-Id: I0cca37b626a94a05fb983a8528c22a660e89e673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/790fff1d24db32d961a4020ad0f3c0a52de3ab34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/790fff1d24db32d961a4020ad0f3c0a52de3ab34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
