)]}'
{"id":"openstack%2Fneutron~763745","triplet_id":"openstack%2Fneutron~master~I7e39f40b325a6974a46ed34641cec5226c9e5a3f","project":"openstack/neutron","branch":"master","topic":"bug/1904871","hashtags":[],"change_id":"I7e39f40b325a6974a46ed34641cec5226c9e5a3f","subject":"[OVN] Ensure metadata checksum","status":"MERGED","created":"2020-11-23 08:04:51.000000000","updated":"2020-12-12 05:12:04.000000000","submitted":"2020-12-12 05:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763745-bug/1904871","meta_rev_id":"c3bc21611020af29cba9958a9f05ef3f081cfe02","_number":763745,"virtual_id_number":763745,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-12-11 02:45:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-12 05:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-11 10:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-11 09:42:31.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-11 10:14:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 11:02:24.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 14:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 13:12:11.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-11-30 10:15:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-30 22:33:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2020-12-01 10:13:37.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-12-01 15:49:06.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2020-12-10 14:57:17.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4e52f86b361badcea4f2817c8f7aad2c482690eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-23 08:04:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3117ccea7d5acbd23cec893be9a4850d61a4a863","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-23 11:02:24.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"03403df9e085f07ffe7740ff6e4c1c5363c8ebb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:05:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d3d455fb0cdc4a518e2c0d5d3ac61a93 : SUCCESS in 57m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df480e0594d449aab4b58580314eeb7e : SUCCESS in 33m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0afe000d36c548f090c81772bed84a08 : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/272f465b4039408494d9b1bed028b419 : SUCCESS in 1h 35m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/842b8500bb5541da84e0b70ad398a7a7 : SUCCESS in 1h 47m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/40d7e8d265604ff9b9d155c162017db2 : SUCCESS in 1h 20m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21ca27c610814e95b8c94ffb01c01d2b : FAILURE in 34m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd86efab9ab24a28912a262a28ef5086 : FAILURE in 23m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45f42f5ae35437e8ed5a64072cda611 : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fba8f27b13b64ffaad4bb6c7db25a0f2 : FAILURE in 24m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fd5ed9c400e4de6a53c70c0d69fc186 : FAILURE in 21m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2949f73ab854bc9a4ca054df1622afd : SUCCESS in 8m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/901c80073f5e43fd9b966347daee26af : TIMED_OUT in 2h 15m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06e5b04290c34f20811a5743c86fbc1c : SUCCESS in 1h 12m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b1a515814c2d4d439d2eff3e048ebc30 : SUCCESS in 1h 52m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/726bb5aed0074c20ba4c670b54bfb0e5 : SUCCESS in 57m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ef1c9df22944541aeaf8fbeaccc4c44 : SUCCESS in 59m 43s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/da0594f6412a4e559795a101713291bf : FAILURE in 50m 04s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3687278455434e87b72c92ab82b6115c : SUCCESS in 1h 16m 37s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/28ea3fbcadeb4af8bab291f0f81b8c6e : SUCCESS in 1h 13m 07s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e886279b728e4053a02178825a6c1f3f : SUCCESS in 1h 32m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2b504cdc25e4490b876c246550c3f6f : FAILURE in 1h 13m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bdb40086fe44d209f41444eb28ea185 : SUCCESS in 1h 07m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5a7d49c5596e42fe93dec65f6a6d68bd : SUCCESS in 1h 16m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e8b6ee9fe3e47f29cb26f5ceacf85d7 : SUCCESS in 2h 07m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afde1f47d91f464a9701cf066fb6c06e : SUCCESS in 1h 00m 21s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c4e02cfcb5ef4723806c48a5a23e6938 : SUCCESS in 1h 18m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/997d705570b44aa8af3857c668bee8e5 : SUCCESS in 1h 21m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6063561ef71a495aabcf5c8d3cff1b15 : SUCCESS in 42m 49s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/808b0fda910f44f6add46e449fcb969c : FAILURE in 1h 11m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/db76a41507dd4d858e0bf2b4a2bdb715 : SUCCESS in 1h 22m 28s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5ef33d1ca1984673baf99a93b959aebe : SUCCESS in 3h 28m 14s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0953bdaa673a4cd2a91905fd232adaf8 : SUCCESS in 1h 20m 49s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/97545688669a4dc78c64d85caa2d91f3 : SUCCESS in 1h 40m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/852218bab52e4a91b3cdce6851ada9c6 : SUCCESS in 2h 05m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/864bb7eafc69414ab9927630b6781e25 : TIMED_OUT in 3h 03m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2edd5aeec3cf40f38e8b4e4f77e6c676 : TIMED_OUT in 3h 05m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a00e48200d6160a8dbaea55401a701012791e942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-23 16:31:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb220944c7cc35d1f349d008d6185a20cc31dab4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-23 22:11:57.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cb5f2306bb1ae3f2e967f9c638a10be57e2909c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 23:46:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4773d8ea814546dfa60cffd7d8506d47 : SUCCESS in 51m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d77714875c8b44eabd1a406ae7cc95c4 : SUCCESS in 38m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d0e496c8f14b49859840e16aab141677 : SUCCESS in 1h 32m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/669dff2bbc8f4aa99245b74ebc575bfe : SUCCESS in 1h 38m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18f82fb2085647d18ba28febb44a3c09 : SUCCESS in 1h 14m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/137250672c03449ca636aa5254abe171 : SUCCESS in 1h 20m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d1e40df5f614e5693718092f14957a1 : SUCCESS in 37m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05657c0b6698451ca5d1910e878bef92 : SUCCESS in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc10ad62bf8459f8dcdbf4927065242 : SUCCESS in 11m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18515f68e8eb4bc79119a312cd99788a : SUCCESS in 26m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06378ee6fe194c3aa2b4acd7938f0207 : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236a77ae99364a1e9e4dfe56a7bb10b3 : SUCCESS in 8m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c46edfc1d684cc2ab726c414bca5848 : FAILURE in 1h 09m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/46a2c4839db345e3aa6c117e22811138 : SUCCESS in 1h 19m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ffb686ff481c4351b3d5f868b4939b67 : SUCCESS in 1h 40m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1553723d69d648649fa12bb4e7f849cc : SUCCESS in 54m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8019d8dd678747b8ad842fb2db3b9e6c : SUCCESS in 1h 12m 56s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/22f3f10bd2794af2aaa395a45708f29b : FAILURE in 1h 04m 27s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/897f8e8b5f8e4c9a902a0c76395e0e92 : SUCCESS in 1h 24m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c919705ea2fb41c486026c14273f435b : SUCCESS in 1h 51m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fc5c387a353844fcab20a5ccc42f9d2c : FAILURE in 1h 33m 40s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8c970d38f61d4501b168a92889140c8b : SUCCESS in 1h 20m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ac79fd068614010a6da75adb3cb775f : SUCCESS in 2h 06m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e50c17904f44360b01e11ac901ba450 : SUCCESS in 55m 20s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/41e266705213468eb9eb01cbe784cb2a : SUCCESS in 1h 10m 35s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9049241bba4948ada6f665164f6f7023 : SUCCESS in 1h 15m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e305f6758c574658b34f3049baa59e50 : SUCCESS in 44m 23s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b43cda400e1e4f988175f2f9bda622df : FAILURE in 1h 13m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/69bf8e4107b94bf4a98fc123d5bb01bd : SUCCESS in 1h 17m 09s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/721f52505a8e4e8ab46b88096c4115e6 : SUCCESS in 3h 58m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/682a6a592197486589532f66222fcd25 : SUCCESS in 1h 20m 16s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38b0b1a953d54577b5dde537fe52be80 : SUCCESS in 1h 45m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c9bac1f14f284808941dc0925a516ef5 : SUCCESS in 2h 06m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a6a316a077ce4baabbe457aed3b6370f : TIMED_OUT in 3h 03m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6082f3978b6e45c9a6e391cb2efa48a9 : TIMED_OUT in 3h 04m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"154a106b99f3f07104f64100a63ebd68a068ac5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-24 11:13:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"827f59f107f7dc00fda479766f1150768bbaf4bc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-24 13:12:11.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763745/3 : FAILURE in 1h 55m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9794a7ff40f7449ff51750871ec08bbf6a6d839e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 13:23:35.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"df4973f2195d64ab30b0bdc26d33ac91d2606e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:08:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9074572495814b5ebf85ed99200efc91 : SUCCESS in 47m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1cdc80775494adca2b780491f9f99c2 : SUCCESS in 30m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/819f640bb76d4de7b1f19fd243f039aa : SUCCESS in 1h 37m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e5befd459b554acc9a2879a57a2eda88 : SUCCESS in 1h 49m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1fe8638f6a74729bc4c2f561a814373 : SUCCESS in 1h 30m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0a6c124e5d954932a912704d21852bb8 : SUCCESS in 1h 26m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abb01712f938456587e58159018ffb4d : SUCCESS in 31m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e9f4aac61d74e76a207f40a43a3c543 : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d1b3c23a37455f96b87480d2b83fc9 : SUCCESS in 10m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15be665313d248f9a41d9278db1d0977 : SUCCESS in 25m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9b9961dae5242a1a56fc889e936252f : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6471fa06f1ec4dc49087e958be782f93 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/978e1c567a534155b4ce8065a583c914 : FAILURE in 1h 13m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a561d137479d4653bc8b8ca73b19dd7d : SUCCESS in 1h 15m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a301fac8ccf4eb4ba763c3a8083954f : SUCCESS in 1h 50m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c2c5307d72b4d7c8b2cf4f2c46d6498 : SUCCESS in 1h 07m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/437b2201ac944bc89c29463f954b201f : SUCCESS in 59m 51s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/e25172f66f4a4506ad4e5d6000211597 : FAILURE in 56m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2ba9445c7034f06b868b0e28c83a0b3 : SUCCESS in 1h 13m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5b605d92394c445b989089ebd746b47f : SUCCESS in 1h 42m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a7c2ffe722c8472cb6421611a5787f36 : FAILURE in 1h 42m 03s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b2be4677240f453bb0340bdf667e0f49 : SUCCESS in 1h 18m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3a4f35dd30e4a11ad6ea4adb7af3b53 : SUCCESS in 2h 13m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6ae69dfc8c5475293864e5cb22fdc08 : SUCCESS in 1h 08m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/20da3ccbe3e747d2878221e7fc71dccc : SUCCESS in 1h 15m 08s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a2bc615b2e184df4921539cbda10af81 : SUCCESS in 1h 13m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/871254dbdffb4af094f8a5870bf731a3 : SUCCESS in 35m 21s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/5fe7a7e6aec04199a0acf7cab5cf69e9 : FAILURE in 1h 04m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3a498ba96c8e4e6094d902f3290d3804 : SUCCESS in 1h 06m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7d8fad67ddb047f3bef6e7ea2e59a6fb : SUCCESS in 3h 32m 26s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/67ac6c3481b34c6dbee6e9a9a28ed4ab : SUCCESS in 1h 20m 40s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55921f178c6d4b279f16172b6d5f275b : SUCCESS in 1h 41m 05s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/da76b31fca524034ac0fb7b31b5816d5 : SUCCESS in 2h 03m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2c70da14c447489d84dce0f3f35565b1 : TIMED_OUT in 3h 03m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/112457a7f151456fa877076121aacbc6 : FAILURE in 2h 47m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0c7fa8e58d98982bb6ade61789046c785843887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-24 17:31:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c19c346741e1a98ca9419f69f4bf7cfc726ab74","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-24 19:45:33.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ccd3d802920f139d63b157dd15586c57a7a6a1e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 00:58:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/633d01efa9f0413ea3728af6cb8f94b5 : SUCCESS in 51m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c170de886c20458ca8fc50f64e5c83ec : SUCCESS in 39m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cdcbf066b06f4ea1a43468e6f574626f : TIMED_OUT in 2h 51m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/03f2cd1168a44564b315e68edaa832ab : SUCCESS in 1h 31m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac660d0d9686498a8dbb5bfbcdb340ef : SUCCESS in 1h 10m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/66bdb2f9089e487cb74baf4ef121680f : SUCCESS in 1h 38m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d46336b5eab0499591936ab98b9c57f1 : SUCCESS in 37m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6175accafe854625a43b5af0a2b51270 : SUCCESS in 26m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcd3dc0211b4dbebe03c11ed3d28366 : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b884febb55249deac215d0e450e597f : SUCCESS in 26m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0608ce978c2d4e96b852086cfaa5ed36 : SUCCESS in 27m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e177a5befb4458a902978a7b8210e93 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc5bdab3725449b7b4f6b9d8ccaea13a : FAILURE in 1h 09m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f81f5094303140a8a9d5e3c47deea000 : SUCCESS in 1h 13m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6d9a2624056b4c26b40763a08de41520 : SUCCESS in 1h 29m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/438939fe85bb4c1daff0094446969cec : SUCCESS in 1h 12m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4028bcc2d9c546088f21215ee11b076a : SUCCESS in 1h 09m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/9d09cde51866499c84b61d725c6ef9f4 : FAILURE in 1h 06m 42s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d298525813e54410a47c0d8e8202cd61 : SUCCESS in 1h 31m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6cbbd2a5a429458684352254f75aa67e : SUCCESS in 1h 44m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f1004ba99eb4da6852f03918fda306e : SUCCESS in 1h 49m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7cd559eca71a405ab40e97ca93e5e8ff : SUCCESS in 56m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e2b0c01a95846c882bb046f0cedc900 : SUCCESS in 2h 06m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22b6dc406a0843a8b28bd7d476fc1784 : SUCCESS in 1h 01m 14s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/40bfb71ae9ee4dcbbd7c5f80a1da0c9b : SUCCESS in 2h 03m 51s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/baa5751fdf414b0f88c1e9fcb4638ba7 : SUCCESS in 1h 15m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4bce67bf8e5449d49ef6d20c71db6dc1 : SUCCESS in 39m 46s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b072857de1d4462b8f81ac2d6242d2ed : FAILURE in 1h 12m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/88d2ca0c43464a6e8120b602da98f299 : SUCCESS in 1h 19m 48s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9a97a9cdb1e649dd96584b7a6b3e0496 : SUCCESS in 3h 28m 56s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d0cde276050748d7b9fa63f19ffa4787 : SUCCESS in 1h 19m 44s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1636dbd05b9f40749f063707817593f5 : SUCCESS in 1h 44m 11s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/63182ae74fe64f048a8656762de93304 : SUCCESS in 2h 00m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/51e3252715574c369b65e7dd4a054186 : TIMED_OUT in 3h 03m 00s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/96b419d44daa4cc59989641f9542542f : TIMED_OUT in 3h 05m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"efcf50e96803496894b2a44dd4c0d3a7fc966e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-25 09:12:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1daca90faafd77425afd01aeb787b17041381f0a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-25 10:50:17.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"55468e49e20097470e890e2b9aae2d810d872c47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-25 11:03:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e516d0de8a92663b4de304e5771fb5b7bf7af6dd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-25 14:08:33.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"1c1305f770fb0af982dbae383310990cc479ec62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 19:37:16.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb099ed71cba4991869845bdd3b7dfa6 : SUCCESS in 54m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be13179ae3a44cf8bd3902241347e0bf : SUCCESS in 37m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a5f598dda68046febe752d625a9602af : SUCCESS in 1h 54m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ffe1a8b2e4d641389bf2ecceb2be8bbc : SUCCESS in 59m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57651c80481b4a24a1128f998280fa19 : SUCCESS in 1h 08m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2d89d493774145828ffc0e59119ee3fb : SUCCESS in 1h 24m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/152d695c253048cfb7da7b91a965047c : SUCCESS in 31m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20cae81bb17241c6bc7b1ee6f1df4ce2 : SUCCESS in 27m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff312bf8f13a480a8be6e2c078af84e0 : FAILURE in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34760048af9418f887ad98ee9286309 : SUCCESS in 25m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c6d835eb2bc4141a43c6847de882a83 : SUCCESS in 24m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243c2f32fa0041fa9012f7f4e348d802 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e41e04e98bd470f879d2f89cd1c16a3 : SUCCESS in 1h 08m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ea353bc13a4456cb4fc31dbfd65d0b6 : SUCCESS in 1h 18m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e6310b3c429640878210b71e1557ae6e : SUCCESS in 1h 57m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee54051e994342df931674c0946acc82 : SUCCESS in 1h 13m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/066bb23dfdc6413396cd35f6a976fac9 : SUCCESS in 1h 03m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7f61795a03474454a219c9e979f664a4 : FAILURE in 1h 03m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1278aa8dae6145b5b2c7aea1d68f8410 : SUCCESS in 1h 15m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/00273ce9a7f3430b8e837ec792370ff9 : SUCCESS in 1h 39m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85a2222f8ec042f88c68c9fe7ef5bf93 : SUCCESS in 1h 44m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1978939806b24664b8fc8d6927c63d33 : SUCCESS in 1h 03m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a570c4b7b74443bf84ae9698a1a8460f : SUCCESS in 2h 07m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0306e7c750e34982bde2aea3069347f5 : SUCCESS in 1h 30m 33s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/97b0756fa4e54976b06049bf14ded5a6 : SUCCESS in 1h 11m 21s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/038a7b6f2b12408f84dc4ce5ba4233df : SUCCESS in 1h 34m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9e6fea5c1c284b30bf0f73803b45d39a : SUCCESS in 44m 45s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/ebacbaf36c9a41c5a821c9c360579e5f : FAILURE in 2h 22m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a2371f2e6d8b4035a537426656b59d8c : SUCCESS in 1h 29m 21s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/020e9839821e47fa9235ad1ba77efdf1 : FAILURE in 1h 20m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/490d506bd6d3419187604b54647e84dc : SUCCESS in 1h 40m 51s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/88bdd3b0ea1641b2af4158c83e88b367 : TIMED_OUT in 3h 03m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e1fc495f10a440b6ac511cc619dc922a : FAILURE in 2h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0130b6f299b9ceff77f8659198ad7de9d746f1d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-28 07:59:55.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e429e403f8728bd2ef608665768bccbea09e17db","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-28 08:14:47.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763745/7 : FAILURE in 14m 01s","accounts_in_message":[],"_revision_number":7},{"id":"cadb09fb1d96138cc165b72bb38deb1f220b0def","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-11-28 09:43:20.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2020/Arista_Tempest_Testrefs/changes/45/763745/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"9ede9d0424074f2646fa3c4125f5e374637717a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 14:55:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1f55d364f813455c8b6f9824f3255bec : SUCCESS in 48m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54d05e042a5642ada12446e17c59e749 : FAILURE in 32m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8aa517e4013a41bdbdbee9ed176adf66 : SUCCESS in 1h 45m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/de1bb1c29e3248059a999e76ac557290 : SUCCESS in 1h 43m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ccdf188826b04812872990c6c1bb9de9 : SUCCESS in 1h 47m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cf1901b5cd7845a9940b592b2816a002 : SUCCESS in 1h 41m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c765f964fc8b4f86b9c061e3210e171b : SUCCESS in 32m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8ca2db1b3142a28ad7569725308862 : SUCCESS in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b167eb417c947dca4490e0319bdd454 : SUCCESS in 10m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/937930e024f34073ad8273d3ec4b29c3 : SUCCESS in 33m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b424aa89c04c0e9def56cc663c0453 : SUCCESS in 25m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd99a6a6b88142739390da9294cca9d0 : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e4d43046384415fb5963e8d335ec336 : FAILURE in 1h 12m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4799c12f7dac4ee880c54ec52d83650c : SUCCESS in 1h 18m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3842a02fcdcc4ef88b3463192d09aba3 : SUCCESS in 1h 38m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62b490c717764423a0330c6c17061f3c : SUCCESS in 1h 03m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7b64c0e6c244447dac36c8721b81561b : SUCCESS in 1h 12m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/cb56647861db4fabb7975e248dd718d3 : FAILURE in 56m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d598edd3e5544cb0994d1a58150a21df : SUCCESS in 1h 11m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/623f9cc429a94e479ed0a0d3e162803f : SUCCESS in 1h 34m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb37ae5efc714bbf87d37a00658c2f07 : SUCCESS in 1h 43m 46s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/67cd905765ab4766916038e78fa02ff6 : SUCCESS in 1h 13m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f77a8a7604a488283656a4bf095bba9 : SUCCESS in 2h 00m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab748a0fe51e407fa85d2ece56a51113 : SUCCESS in 49m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a4c111be9f564baeb79441d1c9a3bd8e : SUCCESS in 1h 20m 27s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/637d23669b1e483ab5984dd6fc9707d5 : SUCCESS in 1h 16m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6caae32bbdde41ac94f250712000dbb2 : SUCCESS in 41m 30s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/08f5c33be6804b359455a1d7ad326c8f : FAILURE in 1h 09m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/89d46eb1f69f4919bab16c3ba70e16af : SUCCESS in 1h 21m 13s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d2a309004f294f7aa51a53c45879d783 : SUCCESS in 6h 53m 23s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/adac0999a12a49739d1ccab480fbe25f : SUCCESS in 1h 05m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99b348b80d7b4c26b9d4c2c891408ee1 : SUCCESS in 1h 38m 45s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/31a5e47b44a342b28eaa8879e3e75617 : SUCCESS in 1h 33m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5ee3f9e922734db8ae2f2e5e3bfede15 : TIMED_OUT in 3h 04m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/33657217be424284a09830b19a5bce2c : FAILURE in 1h 49m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c63c579f5160243843f7645985a1c18781f0a512","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-28 22:56:38.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"23feb788a028f8092196d090547f472abbf95a49","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-11-28 23:11:12.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763745/7 : FAILURE in 14m 01s","accounts_in_message":[],"_revision_number":7},{"id":"0e7d81d19eebaea4a4ff889192b43b386df834f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 02:01:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b3521e69763746138398b0f7b65bc35b : SUCCESS in 44m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f94a4ce11e004d0b9dbd0493f4297657 : SUCCESS in 28m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1b72aa2e511d4f78af37f5fde8252b03 : SUCCESS in 1h 29m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36295692848048419e4091866d3475b3 : SUCCESS in 1h 29m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9c6c05e20ac470db65708b57e2bc25b : SUCCESS in 1h 29m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/44551db2624247f0a398740e57b196b7 : SUCCESS in 1h 52m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f9e7bb502ab4b3096c435948e01bb8d : SUCCESS in 34m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b76d100ca50460298ff66c464743371 : SUCCESS in 24m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e907c113af431bba7ea815188a8211 : SUCCESS in 11m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a99a6716c35043409a53d7c8532a8c30 : SUCCESS in 27m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e09240d872094eb0bebb4ac5b2b58c11 : SUCCESS in 24m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936f8da44b6e408bafb72eb501bf7b6b : SUCCESS in 7m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7bde1f997e0944f1a9696eb4b2c46e30 : SUCCESS in 1h 15m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0762e38417564919b77ba2e090a7e737 : SUCCESS in 1h 15m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b2b012b032654369aa5cb7562178a531 : SUCCESS in 1h 27m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b27989c8e6f24fd084b11044db131ce1 : SUCCESS in 1h 03m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82fe939961954042958b062313e3f8e0 : SUCCESS in 52m 45s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/c80319e0342b4ff48a0231338abe4117 : FAILURE in 57m 10s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/118b3ec3ed334e9da93101b330b2b9dc : SUCCESS in 1h 20m 44s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b06ba5c8a8c242b0ade8a3cffedd53b3 : SUCCESS in 1h 23m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ebf97f36224a498087556c9ddc3074e6 : FAILURE in 1h 21m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bd1cab5cc31245189df007d8025feb1f : SUCCESS in 1h 10m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e71ecf8e7c94f0ab38a9b911362eef7 : SUCCESS in 2h 11m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46cb01f6b9c1417aa615e8e8134c764c : SUCCESS in 59m 36s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2ae03b8e5fdd48c695fdb489f26613a1 : SUCCESS in 1h 05m 17s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e0574146994d4bcdab10065d24582b13 : SUCCESS in 1h 19m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9e6667d4bdde4f0688d6048cccb33b4f : SUCCESS in 43m 10s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d4a36d81b4db49049c44425f8d7f3492 : FAILURE in 1h 16m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0a34382e76bf4f689ca16f4b79d8467a : SUCCESS in 1h 22m 31s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0b7f08317a8b40b9a5871f283c79758a : SUCCESS in 2h 40m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fdc8fbeb1706444ebd3c7b372b4188c1 : SUCCESS in 1h 07m 01s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb206e2d228042108d0dcde7bbfef238 : SUCCESS in 1h 44m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5c3fe9011f7e4f9a8ff7aa62c6c68d23 : FAILURE in 1h 39m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fa068864761648de86ea9d79151e172c : TIMED_OUT in 3h 02m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1dd5f006a3fb4e7c9ffd9fe27cf70bad : FAILURE in 1h 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f28e6de5badc697fc5ac84254e34a252b29922cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-30 10:15:15.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fc0792458fc675eedc64a25d71fae3e7a28a8e21","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-30 22:33:27.000000000","message":"Patch Set 7:\n\nI was just remembering this change where we removed the TCP checksum code:\n\nhttps://review.opendev.org/c/openstack/neutron/+/654645\n\nAre things different for OVN where this won\u0027t cause an issue?","accounts_in_message":[],"_revision_number":7},{"id":"aa792d80ef0fa3159397248d7375864a5cc325bb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-01 09:46:01.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e I was just remembering this change where we removed the TCP checksum code:\n\u003e \n\u003e https://review.opendev.org/c/openstack/neutron/+/654645\n\u003e \n\u003e Are things different for OVN where this won\u0027t cause an issue?\n\nVery good reference. In this case is not the hypervisor but the VM who is dropping the packets, because OVS-DPDK does not do checksum calculations for veth interface [1]. I think the problem is slightly different.\n\nIn this patch, unlike in [1], instead of disabling the checksum offload depending on the OVS datapath, we fulfill the checksum for all metadata requests because of the minimal impact in the system (metadata traffic volume is trivial). This is easiest and works for any deployment.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/738485","accounts_in_message":[],"_revision_number":7},{"id":"9bc38478a6b37c1e37fbd214c8d5e8424737f7cd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-01 09:46:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"922d0bd54522e7e41f43a34998f2e51e6060f4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-01 09:47:20.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"9c9a931c1b0158e0761bfc0acee20f64f85b0877","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-12-01 10:13:37.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d6f766c53182cd90e173da9ed61637ebfc9abcd3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-01 11:04:34.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763745/8 : FAILURE in 44m 11s","accounts_in_message":[],"_revision_number":8},{"id":"2536c78c743c7f5c3189b52b47d2ed7c912117fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 14:47:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c7824e207efb40b3a61a893d6884a3ef : SUCCESS in 49m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e6dcd4cd6b64a80b767cf998d584cbe : SUCCESS in 31m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/33cf4d97f19c4f64bd9b85b011359bbc : SUCCESS in 1h 54m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6dcc25bee87049a78d3e8e0584a1fcc2 : SUCCESS in 1h 24m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f29ba886a6254ffdb99be9d209b9b8a7 : SUCCESS in 1h 15m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d499fecf4453485f945bfd22e82451e9 : SUCCESS in 1h 26m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25cc6dfb80af4166b2ca4a3c4155f7d7 : SUCCESS in 28m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49d8c29b16314f78bc6429c3bcfb82d5 : SUCCESS in 25m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7834107bf60841ad99b184e002c0df68 : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb39b1b70514e9b8e6d125888296fe4 : SUCCESS in 26m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4748330e76d14f00a4a735408f91ac02 : SUCCESS in 21m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41624b31d3d4e4aaacc78a9ebba9da7 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d7cde44c314403c93754d73d7cebb57 : SUCCESS in 1h 10m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a40ac6cdc20a4f57860901e103334fbb : SUCCESS in 1h 11m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2b4a2aae7f1a4678a80acb4b5a6071da : FAILURE in 2h 26m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae96d27586114c37b3b9177659222980 : SUCCESS in 1h 00m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/530fc5c922f54a8fafbb3942a3043ec6 : SUCCESS in 1h 03m 07s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a343430cf5664d5380bbe4abd35af067 : FAILURE in 1h 03m 41s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7dcb8627fb9f4cc988de76204681b571 : SUCCESS in 1h 29m 15s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e1ba5ac99d6b42a98ced46ef9948cbdf : FAILURE in 2h 48m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c715ed8ce3b24556bfbfdf6414f38875 : SUCCESS in 1h 46m 36s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/20507ab425e8429391de70f308fa42d2 : SUCCESS in 1h 04m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f57a5758a0424156ab936977a29544c7 : SUCCESS in 2h 01m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc950df3c99480f872dd53134ddf1ab : SUCCESS in 59m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c32ddc9769f84a72a31afa97a9a02baf : SUCCESS in 1h 16m 58s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25749e8ad51b44c98ff53f14e2bf2489 : SUCCESS in 1h 13m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f49d8534c9d4450eb8fa4c7a8322e1b8 : SUCCESS in 45m 21s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b294557642b14e11bfd1b43937a860ee : FAILURE in 1h 42m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bb3f870c472f48b3b68def129c7c60a7 : SUCCESS in 1h 19m 33s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/46ec0277772c4b4e98cd9f8f44d33200 : SUCCESS in 3h 28m 15s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/268a5717fd9248199142f2e3ac695d32 : SUCCESS in 1h 29m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2c197958a5434dbb98a689e4009da505 : SUCCESS in 1h 51m 04s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/02cdb703a19642458ffd84500f5ac611 : FAILURE in 2h 03m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ecd461e5ede04485b9e426ffc3c8f2cf : TIMED_OUT in 3h 03m 40s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/31408675b1ae4d6cab471c6edb6ae0d6 : FAILURE in 1h 52m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4eab1e0fa0f61fc1688a54044cc82121ab5d61ec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-01 15:05:30.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c38aff041452dc9d522cb1c6743050188667ed0c","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-12-01 15:27:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4ad67c526db488a5a81dcd8f0ae58b7704c3616e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-12-01 15:49:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dc46aa8d3894736431a24ba268f8bac1054511f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-02 20:49:19.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/45/763745/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"6e3e51554c012276c4756dd7bd505bd8686559a3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-04 08:35:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"878be8d650308937e8d9e5771094b37ceeb405c1","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-12-04 09:04:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f1e6d9532d142fb9518cf0c3265533593e781fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-04 09:11:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2d6bff624761719c3d7ec86ac1434f58b4c40f39","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-04 11:20:36.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763745/9 : FAILURE in 2h 08m 09s","accounts_in_message":[],"_revision_number":9},{"id":"e8fa08246ac6bcb1561007f37e3059e4e9e7ad27","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-04 12:59:13.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/45/763745/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d4a41e4ab1a45dbc5e25fb7deda0e32566501724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 18:06:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4256d43dca3b4bc7883bf969e1eb831f : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1a642460f744fa48325a8178963998d : SUCCESS in 38m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df46cd13c1dd490fbc9e65545be5e9b0 : SUCCESS in 1h 38m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/16bf90bf847f49fbb7e46b831da3ba4b : SUCCESS in 1h 34m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1d1ff15276c420d93caa8e96c2c9494 : SUCCESS in 1h 07m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e445ef39fd5747dc8311635e90294409 : SUCCESS in 1h 54m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d34e56b565f4dfcb2867057ab3d8e61 : SUCCESS in 40m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceb57cd0e3a64ea4bdc96fa924e0f98f : SUCCESS in 26m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ebd4a022ca74c1caa3b398f52b09bf6 : SUCCESS in 13m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c7f5fcbdb4c4311bb8659f3923b262c : SUCCESS in 30m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4644cd50c6648b2ad9ff846aeeeb233 : SUCCESS in 27m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fbccb80f1442dd981ed05efc8a0375 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cb6646d1b674210bb30cd889488669c : FAILURE in 1h 06m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f8397b40f18406ebfa0a0d0bb0ae9a1 : TIMED_OUT in 2h 14m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9667b1121d8d4261be425b751e251bb1 : SUCCESS in 1h 58m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0a867a9ba8649b693a7f7826ca93aea : SUCCESS in 1h 40m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5805e5e89e514844882efb4ade331986 : SUCCESS in 1h 44m 20s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/818783739f8a477e8373302be8dd1170 : FAILURE in 50m 42s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/282a987e5dbc4f618132aab16acda146 : SUCCESS in 1h 06m 01s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9074082fed0344aa91fda8d82f57c4e4 : SUCCESS in 1h 11m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e3027814454c4df8994a5cf85dec4ae3 : SUCCESS in 1h 28m 51s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c1b6aa9d8a542c4be14e585b3bd7c07 : SUCCESS in 1h 20m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dffbaefadc864c8d9c69e62cf1bf6e38 : SUCCESS in 2h 13m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9a34f8f55d74bbd823879bcf3f550fb : SUCCESS in 1h 04m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6597126b2abf4f46b2a2c75a3dfbc46d : SUCCESS in 1h 11m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/129372807ed248d1a1d0c1d461e22c5c : SUCCESS in 1h 03m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1d1626108f46c98e861c65b6099aeb : SUCCESS in 40m 18s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/856e24198ea9438eafbee39b9dac30dc : FAILURE in 1h 01m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4fd1f539d47442cc8983d6865037ff06 : SUCCESS in 1h 22m 56s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/61ee396fdb7f4fe7b38f90febd1f57cf : SUCCESS in 4h 13m 39s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/24411cf5374048f3942e4d03b66418ff : FAILURE in 1h 25m 23s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b90ef8762056480480c09330bed4f548 : SUCCESS in 1h 41m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c742cd31741c4672b07ca3cd50798322 : FAILURE in 2h 04m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/822ba235f78343009a96ad4fd124f5fb : FAILURE in 2h 57m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a83bd75fd1fe467080e0efb48311c65a : FAILURE in 1h 50m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4baef03e24254b0e934ce37f6f5de7465e9757db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-10 11:59:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9c8deb4594929cb10f25eb3c86265a6a9d2f289f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 13:55:18.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/763745/10 : SUCCESS in 1h 55m 23s","accounts_in_message":[],"_revision_number":10},{"id":"d94a066f62a2f4c61764ad1322ae7a963fab476f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-10 14:57:09.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"ce71986245b3431ca43815ed25ea5feac28315bb","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-12-10 14:57:17.000000000","message":"Patch Set 10:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":10},{"id":"cccca945c86fd46ea4761678288cb0919b8b53bc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 16:48:01.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/763745/11 : SUCCESS in 1h 49m 42s","accounts_in_message":[],"_revision_number":11},{"id":"4b293ef8fd43f7a700f39d4a1c4bc42997138087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-10 17:18:50.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"6fd5c15b04a00bc779cee85d2e6917b0f41a2976","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-12-10 17:19:35.000000000","message":"Patch Set 11:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":11},{"id":"a45cb6e1fcde75f614a1f582d166ffa4b451cc77","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-10 18:30:14.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/45/763745/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"746606eb1b63bdbcd5a04bd4d5f56d1183de816a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 19:12:01.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/763745/12 : SUCCESS in 1h 51m 35s","accounts_in_message":[],"_revision_number":12},{"id":"490ba1e43902013efa31c2a9e4ab8fcc04b4ef8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:31:51.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/343d1f127eab47ddbcf33382a64ea800 : SUCCESS in 49m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7882a5a7214946269810570f83b222fa : SUCCESS in 34m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c6d749027ea14eff9f3c306d92dcd74c : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bc1382d076604869a33a5e2e2dd19831 : SUCCESS in 1h 49m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fc73dda20fd42a5829ee4b5004a8d8e : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/84eb1b085d0a480ab9dfd7ba62e131d2 : SUCCESS in 1h 41m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c220336f6dc44c4ea96129b0ef96d453 : SUCCESS in 41m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d118b8c22b25448793725678295cbb5a : SUCCESS in 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f967bedda64d478f86f59eb49d62a8c1 : SUCCESS in 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d00f0c0e18d4d2f9a2ef6615ed444c5 : SUCCESS in 16m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03865015dd914e82b293e59195ad5c59 : SUCCESS in 22m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b600855ff9124723a799f374b19cc636 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b2362c048fe43f8a21cf09c5a3814b4 : SUCCESS in 1h 14m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/720cb6a815aa4b7eb15d5e7a3591e6d3 : SUCCESS in 1h 18m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/450922ad670049788211e1763e491ada : SUCCESS in 1h 35m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86c3781cdcab4db6b74f6e44540d9b3c : SUCCESS in 53m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9268622acc594082a2b13697da4a7901 : SUCCESS in 49m 49s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/97dc829be6894f5d85de9f06ad0e993b : FAILURE in 1h 00m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/116e24c2a9c34f138554e066f7959f0c : SUCCESS in 1h 21m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/407130c846974419ac03225659e8a493 : POST_FAILURE in 1h 01m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9534bf7098624ed19b410e5fccc8e912 : SUCCESS in 1h 15m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78f7afcd7d624fd2923cb8c7c79a3d02 : SUCCESS in 2h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/036074a9f8924cd49f3671100d8112b1 : ERROR Failed to update project None in 4s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6db81bb8a8664a208d0d8ba4b0f0019d : SUCCESS in 1h 20m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ad8ab84ead474961b058d691f9d88784 : SUCCESS in 1h 03m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5240bdc602f84fe1aef7aeb385129b7c : SUCCESS in 36m 37s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/70d7b7cd644b49bf99fd3d994fdcd77f : FAILURE in 1h 01m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0fc79af7521648dfa10140805755de38 : SUCCESS in 1h 12m 56s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a3fa4198aa1d4c68a29577939c5bb694 : SUCCESS in 4h 05m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9e6a3b601bb4b459cfdf2cb4b2d0318 : SUCCESS in 1h 35m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/95c624f849864ee484fe80e3a4266e3b : SUCCESS in 1h 39m 15s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b640cc5ced95479e9261a5fbaa5c9043 : FAILURE in 2h 10m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/54b33df03f6348bc98000d236736de6b : FAILURE in 2h 45m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/76f5968ed8f4426d88531e69d2eda1a1 : FAILURE in 2h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"752ab013f4ee578e18cf84203e5a4cf458afddeb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-10 21:55:35.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/45/763745/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"bc98db30fbd071bce689989ef6431e13819a36a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-10 22:05:39.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"f6728d25a6ddd43e6416154bee5022ad69155e7e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 23:53:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/763745/12 : SUCCESS in 1h 47m 04s","accounts_in_message":[],"_revision_number":12},{"id":"ee0c3aacc75938a8399d2221ef99a1fc24d490ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 02:01:06.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ee43bc2356484c2f81db0bb85a0ffae4 : SUCCESS in 1h 00m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b48caadb52c448fcb7323d603cbae62f : SUCCESS in 46m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1f87a6e996df46b69dff8902f5c1bddc : SUCCESS in 1h 50m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/99c5a6393afe49919093b8b042daf98e : SUCCESS in 1h 37m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9c200230e6834753bf2c895bbc0b2b0e : SUCCESS in 1h 26m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/60f9cd1cfd02495d99cb35740995ef30 : SUCCESS in 1h 26m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7afbde857128452ebee72f193c205008 : SUCCESS in 37m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2161443ad254b3aba1327b967c41dfc : SUCCESS in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2c24286fad4cf4a870f14f87b4541b : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad86be0f37584318956bf0f2d07fbf4e : SUCCESS in 26m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fba1111b3994a57bb6b0a5bf6692571 : SUCCESS in 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7387c034a01148ae85295bda63d9f7a8 : SUCCESS in 8m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d305b1c882c4a88a3ee8e45a3f06b50 : SUCCESS in 1h 14m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1061010faca4d86b37b1b23b38a4887 : SUCCESS in 1h 20m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4635fc95f08b479d800c0d2d99e88e4c : SUCCESS in 1h 41m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09cb5a0d428e40b7acfde042a16adce5 : SUCCESS in 1h 10m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/114489a51a61472fb3664cde485925e8 : SUCCESS in 1h 08m 15s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/05019ed3cb1d498686b3451225a4ba58 : FAILURE in 53m 44s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ad34207b430409eaaa44ed537030ebf : POST_FAILURE in 1h 15m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/84e593bc845f4378be569880c205c9f4 : FAILURE in 1h 28m 03s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8037da4a81f546cb8da00b175ff0fba5 : SUCCESS in 1h 13m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/729f15568fed4dc89112472cb2b6db94 : SUCCESS in 2h 04m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a781844ca124c23bb094d2573b284de : SUCCESS in 1h 07m 29s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2a631fef261548b99dc68d43487081f2 : SUCCESS in 1h 04m 33s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4fe2026e7ac6465ab07cb7d501466c46 : SUCCESS in 1h 13m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6867375226d34b308f00d4bcb8f76e74 : SUCCESS in 40m 20s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/de268e7b158e4424929a16f5c0724117 : FAILURE in 1h 08m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/081db7690bc34bcb9a32057e8bd69bda : SUCCESS in 1h 21m 24s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4c1bcc2765ea4c5f9e5820bf330cdb58 : SUCCESS in 3h 46m 52s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/78ed7a44c3dc4dfcb5539da8a6525749 : SUCCESS in 1h 29m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c233f70211f245c6ad539e05dda384b4 : SUCCESS in 1h 32m 28s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5a7083862f1743769fb9899f319d94ee : FAILURE in 2h 04m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0efca97521824a05b1c14f61dbdc4d15 : FAILURE in 2h 52m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/af8807197f2e4115ae71c9311b72278c : FAILURE in 2h 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4df9645d61491fffb366b5702ff8f1837fc418b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-11 02:45:26.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/45/763745/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"3b45814c5ef7acb8af31f19bbe5f6baa74a09d81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-11 08:09:22.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"281cad21caba4d6501a9e3f7923f8bdc510447d1","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-12-11 09:42:31.000000000","message":"Patch Set 12: Code-Review+2\n\nThanks Rodolfo!!","accounts_in_message":[],"_revision_number":12},{"id":"955e681bc72d8716c2e0995bccd9a3eb27624a1b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-11 10:07:14.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/763745/12 : FAILURE in 1h 57m 06s","accounts_in_message":[],"_revision_number":12},{"id":"29c7b945e25a7c5cf723a42c5bbf3ef1d499ae3a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-11 10:14:59.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c4169c2f78d4e71378fab82c6fe50e503a1f8174","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-11 12:09:42.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/763745/12 : SUCCESS in 1h 54m 05s","accounts_in_message":[],"_revision_number":12},{"id":"d94a51e4be39f29bcd9a07f314fbdebe4b34ac53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:12:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b4cdf5f2bd345d2a0fec91cddb96e29 : SUCCESS in 1h 13m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ff2b23c70624bf6b956253ae4913fdc : SUCCESS in 34m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cc9ac227cf604909a4705ee8c6a9903e : SUCCESS in 1h 11m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/879348b188dd417786a22340c40d8a2e : SUCCESS in 1h 26m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d30a92e00ef74412b342b262fdd5076e : SUCCESS in 1h 27m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/178aaebc3d7b4055a16b2b6b09432a70 : SUCCESS in 1h 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c232b66074f44a758f1cced8a5a2e0d3 : SUCCESS in 34m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ba30fa9d1b941c2941413003ae5120c : SUCCESS in 31m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ca64abea02409abfb71195e6f917ca : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a99707c52334aaf8b00d9081607c4a4 : SUCCESS in 28m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acbe36dfc1994b6eb670da3d122313b7 : SUCCESS in 29m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf10443dbd1450eb0be74c593f080c6 : SUCCESS in 11m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8b31ce0841d470f9aa19d46e912f320 : SUCCESS in 1h 11m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3ac7a5f6f4b4cbbaa43762323dd3433 : SUCCESS in 1h 11m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ada92ba005024eb48697c47f2d5febaf : SUCCESS in 1h 40m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f403dede99e42e8b7069d1e605c9cae : SUCCESS in 1h 04m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f4121d36bcb342cc92ad624c256e0bc4 : SUCCESS in 58m 38s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/d96c146c53d84e518122b898557ea844 : FAILURE in 55m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5b39fad95ca84a18bda0c5045655d227 : SUCCESS in 1h 26m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/60dd53c25b5347a6a2a73b9183e7b8c5 : FAILURE in 1h 33m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/056f4dbd0f5d408484025511be49423f : SUCCESS in 1h 12m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56c3452eba7a46ffb875ae4c850608fe : SUCCESS in 2h 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d3b356656a049ef848c6c52d8b7aa2d : SUCCESS in 1h 00m 21s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4e0e4d88db614702a1b6664ee5cf116f : SUCCESS in 1h 07m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d22ce7e2c50542ec98e07f76b7c7f44a : SUCCESS in 1h 10m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7c174e98f6bc40e28c7345423185dc74 : SUCCESS in 39m 24s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b97bace626ce4646b511a370c2122231 : FAILURE in 1h 13m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/91dafe12ae304dbfb75fb3ed78063d55 : SUCCESS in 1h 22m 42s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ec7dda9c4dfb4a5a9aca02dcdb24c8c8 : SUCCESS in 3h 54m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03e99a17ba2c4b5c804a490e11803fb0 : SUCCESS in 1h 35m 35s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f5cab70cd064d80b74d86ee0e87a403 : SUCCESS in 1h 39m 01s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9993605096f4415883a7a58c6bf1b716 : FAILURE in 2h 10m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e3b153d284c54f40bc4b787c5821b966 : FAILURE in 2h 51m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/74040a11805b4c56a62be6fda03866d0 : FAILURE in 1h 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"29e2cba2b682fdec7ad44726f1d4df8b62b62511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:13:27.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"7a569496d35f87a2f916f776b7f89ae1a2c91b94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 14:42:34.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e7f373a92f064f05ab072e59fd9df874 : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/77589071644849c48dfd607e7863532b : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/759bf4cc172a4a8a880cf5a395353a1a : SUCCESS in 2h 03m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/68be4e6b98324d7597bd7a54c47e2dfe : SUCCESS in 1h 22m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b73a2d25d307415199bbeecc90702edc : SUCCESS in 2h 15m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04dbb8c49ed7435eb349d65566ea6639 : SUCCESS in 42m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0952cb30175e4471a5b911fa632e8d37 : SUCCESS in 16m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65e542eb8cf74efaa7ef450d90931d9e : SUCCESS in 38m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37cebdc38c30464ea0c440957d724d10 : SUCCESS in 27m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82cf01ba14f4807967ac9f1e03583f2 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7abaef743fd948e9949786dcadb261ef : SUCCESS in 1h 13m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b53ad914b1d1471ca7b772e6926fe284 : SUCCESS in 1h 11m 37s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4b6de7f5a1f94f7f8ba176225eb785cf : SUCCESS in 1h 29m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2915d0af1d0d494284526121cef88256 : SUCCESS in 1h 09m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/606827c83808451e88c8cf1839f98d48 : SUCCESS in 1h 04m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f62b591382e14d20b65a936ec4a17662 : FAILURE in 2h 15m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c4989d388da437ca015397c609f3f97 : SUCCESS in 1h 04m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4a403d9a89564ef59c8056372bae6633 : SUCCESS in 1h 12m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/41c86b73f25f44bea98740c4d95b2d82 : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":12},{"id":"7c4b5936808c738e3bd5bfe431bff9e780eaa5a2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-11 21:10:59.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"cf03f7def5598b99d1bb378b787fcbcfd66118eb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-11 23:16:04.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/763745/12 : SUCCESS in 2h 04m 28s","accounts_in_message":[],"_revision_number":12},{"id":"a6340531d379ff7bd7817884677703e11002c736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 01:53:51.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6d3fc7112d824d18bdd5688d50c28c73 : SUCCESS in 54m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/84c33e1bbd6d47709eb0b3dff567a17e : SUCCESS in 39m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3f62add1226d4fd99ec1bb82974fd5cf : SUCCESS in 1h 43m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d478a0d01dde48ab92b82861e1333bad : SUCCESS in 1h 42m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1962a38e16fe42efb299c3c7fceba97f : SUCCESS in 1h 37m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e6bbf128a5254d5e84b006e7efa55620 : SUCCESS in 1h 19m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cde9d3893ad546b099f90b01f3953d09 : SUCCESS in 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01c745ea1af846739d3168aae6a76d45 : SUCCESS in 29m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4592720caf41eebc136b37614f3a6e : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d53fc431eb41a48244c1061ae34827 : SUCCESS in 24m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e96c785941f4a19829fef8b06daee90 : SUCCESS in 24m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d546b8001c744a51aaf0af7004b14ecc : SUCCESS in 24m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fccb76fd1cc49efbf2f3f649882bc07 : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cce91f7b528b4439b17e4469fcc2d0f8 : SUCCESS in 1h 07m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c3ad500a69a4ebca5a36824532f2fbe : SUCCESS in 1h 19m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/df641b793333495d9db5aa7da7370ea7 : SUCCESS in 1h 49m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0aea4fb847194779a679bd5c6d9ac852 : SUCCESS in 1h 09m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e15f0b5622e04a788088690d445d197d : SUCCESS in 1h 11m 08s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/044099490a5844fabf9154f63b3a5f46 : FAILURE in 57m 37s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/48fee495a89b4f6486a2d9e4be9e297e : SUCCESS in 1h 21m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/66dcfcf9b671437a9afff076009fd658 : FAILURE in 1h 40m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8526aa9d20c54085a0c688a659ccb60a : SUCCESS in 1h 18m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26e7d8916e994fd1a80f84c576f15bbe : SUCCESS in 2h 28m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eca4e42da03c4ec5b9908f45c5732f41 : SUCCESS in 1h 02m 50s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/054f51f411f04f07b0d5e241933f3e27 : SUCCESS in 1h 09m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0454f68bf4134948a8bd7fe95774e8e5 : SUCCESS in 1h 09m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f676d1a8531641b3a9a8123f5de84fc9 : SUCCESS in 41m 34s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/cac94e2fa8184cbfa36e2a5c65da34a9 : FAILURE in 50m 09s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/87c104c4c6214f3181d5ed9bfa820070 : SUCCESS in 1h 19m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c2ccbc616e0048e3bb10331caeb2901e : SUCCESS in 4h 40m 34s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8369afb2e9eb4fdb82fb6e126d545588 : SUCCESS in 1h 19m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/121ecd032e1a40ce8fafc42cba5b728a : SUCCESS in 1h 05m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6137785d5c6344b384263c99b3fa741e : SUCCESS in 2h 10m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3e2c55a425c74da28b5a2087b1313188 : FAILURE in 2h 55m 32s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d2f3df6e9c6f4892b87868509b27dfd2 : FAILURE in 2h 09m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e694428d3bb0d1b305df8e368d18f0275f53586b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 01:54:14.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e52614ebb769aa077d295234af0ccff75a1c5203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 05:10:42.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/95bbfe071556469aad4785b587174d31 : SUCCESS in 56m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d8938418d882456f9d9c5894fb188165 : SUCCESS in 1h 16m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/478885439c444afb9a2f2976d91451f6 : SUCCESS in 1h 15m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ca705ba5c92243fa9c61ab0b57058af8 : SUCCESS in 1h 33m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34049cb177dc420585ef40ab8233140c : SUCCESS in 1h 31m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f167ad36d23461db14d8b58db42249c : SUCCESS in 29m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712c3ecbb0fb47d299f1bda8d402c0a0 : SUCCESS in 10m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f18ee68b3eb64be49a1ca7b17cfef26f : SUCCESS in 26m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7f7f541431746879aece511e47d8062 : SUCCESS in 23m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a55f0886394ef68899cb5ae0c189dc : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da7d7c7dd2fa48bd9b606f7ce6995998 : SUCCESS in 1h 01m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe427804d49d4d76a9adbcc9c26dc6f1 : SUCCESS in 1h 05m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8db84649ab1b48f08130ca67602e2bb0 : SUCCESS in 1h 18m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edf136e4fdbb4030b3eabe69d80142e5 : SUCCESS in 52m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f596d78eb38f45faade3805a6ec48319 : SUCCESS in 59m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f06b51b69ed47f29f884732e86d9f58 : SUCCESS in 1h 54m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ba931032b61455b9e189cb63c3fa621 : SUCCESS in 1h 03m 05s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/80bdffc2ce854ccea68609a6f2f2b43b : SUCCESS in 1h 09m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e6ddd3579dd244d09e97f8aebbd67317 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":12},{"id":"fad7ab9c99ad1dfe6c8ab00c9d922c3f49c0629d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 05:10:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"c3bc21611020af29cba9958a9f05ef3f081cfe02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 05:12:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057f7f63fd7d497c9486d9170ccc355a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"26a30a3b78bfaf6d3f8ce6271b1a7415440c602c","revisions":{"648ef7bd782ec462d5d6c806eefd9dfe2aaf4533":{"kind":"REWORK","_number":1,"created":"2020-11-23 08:04:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/1"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 15:03:59.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/648ef7bd782ec462d5d6c806eefd9dfe2aaf4533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/648ef7bd782ec462d5d6c806eefd9dfe2aaf4533"}]},"branch":"refs/heads/master"},"15c4da19cecec14dae63daf6a87ddee25ecf8f48":{"kind":"REWORK","_number":2,"created":"2020-11-23 16:31:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/2"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-23 16:31:24.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15c4da19cecec14dae63daf6a87ddee25ecf8f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15c4da19cecec14dae63daf6a87ddee25ecf8f48"}]},"branch":"refs/heads/master"},"7e9ea4f7e2cd4529f671129374f77247b7f72afb":{"kind":"REWORK","_number":3,"created":"2020-11-24 11:13:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/3"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-24 11:12:51.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e9ea4f7e2cd4529f671129374f77247b7f72afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e9ea4f7e2cd4529f671129374f77247b7f72afb"}]},"branch":"refs/heads/master"},"3f99f3fc8b503ce8a691cbbbf312368093f73306":{"kind":"REWORK","_number":4,"created":"2020-11-24 17:31:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/4"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-24 17:30:20.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f99f3fc8b503ce8a691cbbbf312368093f73306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f99f3fc8b503ce8a691cbbbf312368093f73306"}]},"branch":"refs/heads/master"},"6ee5ae5e450bf72309ff26972e03835ca3f4ab45":{"kind":"REWORK","_number":5,"created":"2020-11-25 09:12:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/5"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-25 09:12:00.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ee5ae5e450bf72309ff26972e03835ca3f4ab45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ee5ae5e450bf72309ff26972e03835ca3f4ab45"}]},"branch":"refs/heads/master"},"70c2f7b5bb93f526e9085f5c5dd4e0a6bc0eb796":{"kind":"REWORK","_number":6,"created":"2020-11-25 11:03:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/6"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-11-25 11:03:26.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70c2f7b5bb93f526e9085f5c5dd4e0a6bc0eb796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70c2f7b5bb93f526e9085f5c5dd4e0a6bc0eb796"}]},"branch":"refs/heads/master"},"3216bd40eb99517ee650f9cfb85c1deca2b4d5d2":{"kind":"REWORK","_number":7,"created":"2020-11-28 07:59:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/7"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-11-28 07:59:55.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added in minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3216bd40eb99517ee650f9cfb85c1deca2b4d5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3216bd40eb99517ee650f9cfb85c1deca2b4d5d2"}]},"branch":"refs/heads/master"},"30ef59a320f3a89fe31803ceed5075a0cdc8b225":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-12-01 09:47:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/8"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-12-01 09:47:12.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added is minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30ef59a320f3a89fe31803ceed5075a0cdc8b225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30ef59a320f3a89fe31803ceed5075a0cdc8b225"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3eecb48036aee1f76140c4ffef74640d2929bbf5":{"kind":"REWORK","_number":9,"created":"2020-12-04 09:11:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/9"}}},"commit":{"parents":[{"commit":"c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8","subject":"Merge \"Import packaging.version explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2e39c2ce92e5bde86c63c8e4a89616060bfd6b8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-04 09:11:01.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added is minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3eecb48036aee1f76140c4ffef74640d2929bbf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3eecb48036aee1f76140c4ffef74640d2929bbf5"}]},"branch":"refs/heads/master"},"9508b034b664c80f64c523963cde539d644c515f":{"kind":"REWORK","_number":10,"created":"2020-12-10 11:59:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/10"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-10 11:58:44.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added is minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9508b034b664c80f64c523963cde539d644c515f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9508b034b664c80f64c523963cde539d644c515f"}]},"branch":"refs/heads/master"},"dc5a692a02cb4ce55c9b2d0ba8867572aebdf2cf":{"kind":"REWORK","_number":11,"created":"2020-12-10 14:57:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/11"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-12-10 14:57:05.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added is minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc5a692a02cb4ce55c9b2d0ba8867572aebdf2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc5a692a02cb4ce55c9b2d0ba8867572aebdf2cf"}]},"branch":"refs/heads/master"},"26a30a3b78bfaf6d3f8ce6271b1a7415440c602c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-12-10 17:18:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/763745/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/763745/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/763745/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/763745/12"}}},"commit":{"parents":[{"commit":"ff33049eb674caf75763ec95d3537c9dafeaccf3","subject":"Merge \"Remove neutron-tempest-with-uwsgi job from CI queues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff33049eb674caf75763ec95d3537c9dafeaccf3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-11-20 14:37:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2020-12-10 17:18:50.000000000","tz":0},"subject":"[OVN] Ensure metadata checksum","message":"[OVN] Ensure metadata checksum\n\nEnsure the TCP traffic leaving the OVN metadata namespace has\nthe checksum correctly populated. This is necessary when the\nOVS datapath is \"netdev\".\n\nBecause the overhead added is minimal and only applies to the\nmetadata traffic inside the metadata namespace, this rule is\nalways set.\n\nChange-Id: I7e39f40b325a6974a46ed34641cec5226c9e5a3f\nCloses-Bug: #1904871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26a30a3b78bfaf6d3f8ce6271b1a7415440c602c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26a30a3b78bfaf6d3f8ce6271b1a7415440c602c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
