)]}'
{"id":"openstack%2Fneutron~764367","triplet_id":"openstack%2Fneutron~master~I2539420411e8fb2f54a5da9d9047171fd37bfb11","project":"openstack/neutron","branch":"master","topic":"new-debugging-guides","hashtags":[],"change_id":"I2539420411e8fb2f54a5da9d9047171fd37bfb11","subject":"[Docs] Guide about running and debugging fullstack tests","status":"MERGED","created":"2020-11-26 16:17:47.000000000","updated":"2020-12-04 01:26:18.000000000","submitted":"2020-12-04 01:24:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":75,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"764367-new-debugging-guides","meta_rev_id":"8b47de95c3557f7ee22e7ee3cf00c68b7aa98caf","_number":764367,"virtual_id_number":764367,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 01:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2020-12-01 21:17:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-12-03 12:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-12-01 22:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-12-03 12:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 17:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 16:08:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2020-12-01 05:51:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2020-12-01 10:17:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-01 14:22:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-12-03 12:31:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"51a3cea636c51a0f72fc69da0ace1967c54f7cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-11-26 16:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"419cb8a45133025b5d4f555913c3e91cdd72f44c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 17:47:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed7d4f5787042eca996addbf1ad61c9 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b24726c3c74ee6b9d4c3631ed76d75 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"98ec3aec247a092e1cb509db63e669a38af5f422","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-30 16:08:08.000000000","message":"Patch Set 1:\n\n(37 comments)\n\nSorry, guess I had a lot of comments 😞\n\nOn question though - why move this out of TESTING.rst as that\u0027s where all the other info is on running the other tests?","accounts_in_message":[],"_revision_number":1},{"id":"8e5d045c3a14d79a15011cb7ce1a5c048167c147","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-01 05:51:00.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks Slaweq, and Brian","accounts_in_message":[],"_revision_number":1},{"id":"96a824524ddf04d9c26b9a18ff0edf4d3fc7ebd3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-01 05:52:18.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e (37 comments)\n\u003e \n\u003e Sorry, guess I had a lot of comments 😞\n\u003e \n\u003e On question though - why move this out of TESTING.rst as that\u0027s where all the other info is on running the other tests?\n\nDuring the PTG we discussed that some documentation more specific on debugging issues in gate and locally would be helpful for newcomers mostly.","accounts_in_message":[],"_revision_number":1},{"id":"9a8fa5032b143f9cf13d56ae10f57ede8b8204e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 09:30:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbf703d8c0ca7f75e86a616cde15f492e8a0a3c4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 09:34:12.000000000","message":"Patch Set 1:\n\n(37 comments)\n\n\u003e Patch Set 1:\n\u003e \n\u003e (37 comments)\n\u003e \n\u003e Sorry, guess I had a lot of comments 😞\n\u003e \n\u003e On question though - why move this out of TESTING.rst as that\u0027s where all the other info is on running the other tests?\n\nI moved it there because I think that with all this new guide this overall TESTING.rst document may be a bit too long maybe. So my idea was to left there only some brief info about fullstack testing and link to more detailed guide.\nIf You think that it should all be in TESTING.rst document, I can also move it there.","accounts_in_message":[],"_revision_number":1},{"id":"a7dfcfd230cf23047cf227566684f5d6649c56b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 09:49:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107b43bb7c3c469bb204df18f1813ff3 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63d2b61b5dd400698087069a906ee1e : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d602c5abe84ef7949c8914273bb4fc52c5258c6a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-01 10:17:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/764367/2 : FAILURE in 43m 21s","accounts_in_message":[],"_revision_number":2},{"id":"38ef61d35a723f41c1a69532fc6f9ad1401b4d8d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-01 14:21:50.000000000","message":"Patch Set 2:\n\nThanks for the update Slawek, and I\u0027m fine moving things, I just wasn\u0027t at the PTG meeting where this was discussed.\n\nI was going to just push a quick update to a few more things I noticed, minor update.","accounts_in_message":[],"_revision_number":2},{"id":"38bfb2354d82c2b72c2aa6a171ccf39b4e10061a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-01 14:22:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a5f3f11ff47056d61a9d017edbb4972a8c969fe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-01 15:14:44.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/764367/3 : FAILURE in 36m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a5f9dd481326c3f875c34309cbc116d2f13e6591","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 21:17:18.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Thanks for the update Slawek, and I\u0027m fine moving things, I just wasn\u0027t at the PTG meeting where this was discussed.\n\u003e \n\u003e I was going to just push a quick update to a few more things I noticed, minor update.\n\nThx a lot Brian :)","accounts_in_message":[],"_revision_number":3},{"id":"229f78d85677b8ecd35402ac37e42e202744bf1d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 21:17:35.000000000","message":"Patch Set 3: Code-Review+1\n\n+1 for changes which Brian did :)","accounts_in_message":[],"_revision_number":3},{"id":"757bfef3daad628602a1173f10d7274777b63d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:32:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb0473c4a424eb4bff30dc1bb86cbf6 : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7139b00f7b3746c5b7ca2bb52077ba50 : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":3},{"id":"d420ecbdf67beea681209653fba3af4cb718ddf8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-01 22:33:48.000000000","message":"Patch Set 3: Code-Review+2\n\n+2 since I only fixed a few typos","accounts_in_message":[],"_revision_number":3},{"id":"7276683d03fe0fdb1abd853475bcdc5b6284682b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-03 12:31:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI think it will be helpful for newcomers.\nThe 1 for tempest is on its way, but I progress slowly with it.","accounts_in_message":[],"_revision_number":3},{"id":"40ec04455e4505368290b380be8a7d77df7dfe7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 12:32:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a238761be9798557c8b79ffb52ea75ed62164296","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:14:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5f0c3935a54ea38dcf1e81936e874d : RETRY_LIMIT in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b729a29f8ff4ec0887e5efc83822b1d : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":3},{"id":"e0534a3bf9f3f6ec8ec9fc1de6f64d3521a176a8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-03 17:11:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"78f5b713fb8b64c24e18a58989a78f05411f7c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 00:56:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9128033a6546f8ad45320bf4b1c636 : SUCCESS in 17m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60e39817da04679aa62ea3735fa9884 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":3},{"id":"1983f3623843d0eed06d149c6d3be945696ad1f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 00:57:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c4cdd9aa6fff217fcced6d6ef7e99cfece0dd2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 01:24:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14719325456c4170b33189215b8b4e7d : SUCCESS in 20m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25344aeebe754d90a5e88d970927d31b : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e36c75a4a6e0abbf89e44d15188efe0f8f55a9e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 01:24:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8b47de95c3557f7ee22e7ee3cf00c68b7aa98caf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 01:26:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755cf46fa62e4690b350be03c62dc734 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"94f752bdee85183fcf9f17d41af64d19678dcae5","revisions":{"115ba84e9df8306bb91a09891d5d6da08811b930":{"kind":"REWORK","_number":1,"created":"2020-11-26 16:17:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/67/764367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/764367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/764367/1"}}},"commit":{"parents":[{"commit":"e59a7c9aca56a9b5edc70d74ca55b50737bd66d6","subject":"Merge \"Adds a unique index to quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e59a7c9aca56a9b5edc70d74ca55b50737bd66d6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-26 16:15:33.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-26 16:15:33.000000000","tz":60},"subject":"[Docs] Guide about running and debugging fullstack tests","message":"[Docs] Guide about running and debugging fullstack tests\n\nThis patch moves detailed description about fullstack tests from\nthe general TESTING.rst document to the fullstack guide which is in\ncontributor/testing/fullstack.rst\n\nIt also adds new sections about running and debugging fullstack tests\nlocally and how to investigate failures which happens in the fullstack\njobs in the gate.\n\nChange-Id: I2539420411e8fb2f54a5da9d9047171fd37bfb11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/115ba84e9df8306bb91a09891d5d6da08811b930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/115ba84e9df8306bb91a09891d5d6da08811b930"}]},"branch":"refs/heads/master"},"3dce29d7198ab409488213cc6daf9c2de51f5cc2":{"kind":"REWORK","_number":2,"created":"2020-12-01 09:30:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/67/764367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/764367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/764367/2"}}},"commit":{"parents":[{"commit":"e59a7c9aca56a9b5edc70d74ca55b50737bd66d6","subject":"Merge \"Adds a unique index to quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e59a7c9aca56a9b5edc70d74ca55b50737bd66d6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-26 16:15:33.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-01 09:30:17.000000000","tz":60},"subject":"[Docs] Guide about running and debugging fullstack tests","message":"[Docs] Guide about running and debugging fullstack tests\n\nThis patch moves detailed description about fullstack tests from\nthe general TESTING.rst document to the fullstack guide which is in\ncontributor/testing/fullstack.rst\n\nIt also adds new sections about running and debugging fullstack tests\nlocally and how to investigate failures which happens in the fullstack\njobs in the gate.\n\nChange-Id: I2539420411e8fb2f54a5da9d9047171fd37bfb11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dce29d7198ab409488213cc6daf9c2de51f5cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dce29d7198ab409488213cc6daf9c2de51f5cc2"}]},"branch":"refs/heads/master"},"94f752bdee85183fcf9f17d41af64d19678dcae5":{"kind":"REWORK","_number":3,"created":"2020-12-01 14:22:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/67/764367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/764367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/764367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/764367/3"}}},"commit":{"parents":[{"commit":"e59a7c9aca56a9b5edc70d74ca55b50737bd66d6","subject":"Merge \"Adds a unique index to quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e59a7c9aca56a9b5edc70d74ca55b50737bd66d6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-26 16:15:33.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-01 14:21:42.000000000","tz":-300},"subject":"[Docs] Guide about running and debugging fullstack tests","message":"[Docs] Guide about running and debugging fullstack tests\n\nThis patch moves detailed description about fullstack tests from\nthe general TESTING.rst document to the fullstack guide which is in\ncontributor/testing/fullstack.rst\n\nIt also adds new sections about running and debugging fullstack tests\nlocally and how to investigate failures which happens in the fullstack\njobs in the gate.\n\nChange-Id: I2539420411e8fb2f54a5da9d9047171fd37bfb11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94f752bdee85183fcf9f17d41af64d19678dcae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94f752bdee85183fcf9f17d41af64d19678dcae5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
