)]}'
{"id":"openstack%2Fneutron~765286","triplet_id":"openstack%2Fneutron~stable%2Fstein~Iae86705f1d30c89dc5482261d852b45787bd8782","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1905568-stable/victoria-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"Iae86705f1d30c89dc5482261d852b45787bd8782","subject":"Fix calling of add_tunnel_port method from sanity checks module","status":"MERGED","created":"2020-12-03 08:35:33.000000000","updated":"2021-01-15 02:40:57.000000000","submitted":"2021-01-15 02:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765286-bug/1905568-stable/victoria-stable/ussuri-stable/train-stable/stein","cherry_pick_of_change":765285,"cherry_pick_of_patch_set":1,"meta_rev_id":"c43c859cc4c41029c6995013feed5cfead766e4d","_number":765286,"virtual_id_number":765286,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 02:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-01-13 18:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-12-09 13:56:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-12-15 22:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-13 18:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 08:45:18.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-03 17:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 20:48:18.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-09 13:56:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-12-15 22:38:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-13 18:16:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3a816a549fbd261a771fccc4a3dcfcf6e543b847","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-03 08:35:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"7032c27c0da065f2d110e756c0b03bc37d185b60","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-03 08:45:18.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765286/1 : FAILURE in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f0e0fedcd9e019692446d0995eb17474828d5428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:30:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/4862cc396bc242c98c997e83754775a6 : SUCCESS in 1h 02m 49s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f581255d29564be4824a052add7862cf : SUCCESS in 35m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/816f2a1e48134184b97086e613000502 : SUCCESS in 1h 48m 38s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/87575f58e7b940cb94c94d5e1b33f681 : SUCCESS in 1h 31m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/bdf024e5e58e4792aac1de9812f7f0ab : SUCCESS in 1h 06m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d28d4b15818547f2864e534ba0f56c6d : SUCCESS in 20m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25093bed442f4601abae86cd23cc1d8c : SUCCESS in 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7899b5f00883405282584fc608b03494 : SUCCESS in 11m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7258b134e384b12a77b8ae389efcf0b : SUCCESS in 20m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/03173411cc574e9ab763071eb6101ca8 : SUCCESS in 34m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10a93b7004d247f8a1796f53ed6c5895 : SUCCESS in 21m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a7db1b8209a4a36a735ea55dcb99446 : SUCCESS in 22m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1687b76c49b44976815c4377936adc59 : SUCCESS in 5m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f27db3eb63704733a73ac6aec07dac74 : SUCCESS in 1h 27m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30ab1feb3d5c45acaeee8d29ef6a74dc : SUCCESS in 1h 00m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c282867eae694adbaef8b2e9f952d596 : SUCCESS in 1h 25m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2404c9a424314505915ed2fde0acdd61 : SUCCESS in 39m 18s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/201c29155fa547c3bde4c389f7ef0d95 : SUCCESS in 43m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fd82cfe4d24745df93818bb4ffd1f574 : SUCCESS in 1h 06m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f10b0f9912524d9b98598c5758f9de6a : SUCCESS in 1h 46m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cd1f346311384ace901df342e1a62f25 : SUCCESS in 1h 16m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4b16c769754a4e7e854829da4965a5af : SUCCESS in 1h 22m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/49625dbd06fb4ef4a466c40d92dc6d82 : SUCCESS in 1h 17m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a98d7ba426b84108ba400e595c4e502f : SUCCESS in 1h 16m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2053a98957564410b5ee76bb37987068 : SUCCESS in 1h 27m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/93da557292cc4fccbc624199776a18f9 : FAILURE in 1h 40m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cc506e800784fe796c219b77e039b58 : SUCCESS in 1h 18m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b335b5b55e7d4980935fd87768e53139 : SUCCESS in 54m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48c14a27e0b241a396c6f82fe368f2cc : SUCCESS in 2h 10m 48s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7adaa7061bb2428aa46f370b8a0badfa : FAILURE in 1h 50m 03s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/05b3bb28fb1147899a7ed326b1f1249b : FAILURE in 54m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ed08bbe46a2949f193354ea861a401b1 : FAILURE in 1h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcb8cbd992cda3635519c7d8d477a2a396e9d4b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-03 20:48:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"137c938a2314d0b770f7b4d71a867f3b36157e3e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-04 15:52:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb2adee0a6195924c107bff78a57fb2d25172bfc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-04 16:18:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765286/1 : FAILURE in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c014e9340e0eb8891e0268200074300f2531bdc4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-04 22:51:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6398e317790a489375226e8bd27e590d55b5f446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 04:05:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a1e5787da5dd4c5dadda6d0fc29304f2 : SUCCESS in 39m 33s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c5471af6d89a455c92ed5ec2a906665a : SUCCESS in 26m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/da30d80950844dd4950e981a307ca127 : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/574ec7ded895472d81a86c87ebad2345 : SUCCESS in 1h 40m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/b96b224685104dc8a43d547e024f3324 : SUCCESS in 1h 44m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e5980b957ca4905990bb592b297e7ab : SUCCESS in 19m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/656c0538f8974e8cb469867605c78ea1 : SUCCESS in 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e0b36d0a3742b5918a3262a69fcd77 : SUCCESS in 10m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e51794261590443e9a4d0969bac0b59c : SUCCESS in 22m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fecfb0bcdaf94bc68c7594a25501621e : SUCCESS in 15m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4307a34909242f3853d151ef1106cf4 : SUCCESS in 23m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/819673321cd34b15b3213980da6505b6 : SUCCESS in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c350d4149264477ebd8e417bdb732e85 : SUCCESS in 4m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/443136d75eb14ac3843fecaace850a10 : SUCCESS in 1h 35m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2cd03336ca604a2bb1cd01f455566cd6 : SUCCESS in 50m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8a64b43ee224fd68bdc6ff2b41f02bf : SUCCESS in 1h 31m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0630e7362c334c828c1e28e977afe3c5 : SUCCESS in 39m 57s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7aea81b9fdc74574b6fdfdb7102422dc : SUCCESS in 42m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/38d0ec92e3b84a379250636ff6cad2b6 : SUCCESS in 1h 05m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/82c87feaaef64b7bbf5d14c2cba6f75e : SUCCESS in 1h 33m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1668184bf5404ab2ada67f412128d95f : SUCCESS in 1h 30m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/310e3817b8e94b3784fae6273577598b : SUCCESS in 1h 25m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72b0d179be394b7e91bfe9277c6b85dd : SUCCESS in 1h 10m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c6f09aa7d954e45825e6fbf5b850d01 : SUCCESS in 1h 08m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/480a6bdcb5f14eaf9caf7611b101dabc : SUCCESS in 1h 32m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06549b0fdf1248bf896070357ea69240 : FAILURE in 1h 34m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/352a9eb247134bb7b8fbaa8802fa99b5 : SUCCESS in 1h 32m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d69b308dd2ba4d7e99414f34893eff8c : SUCCESS in 51m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb996d7a52af42bc9e6bcfcc75ea674d : SUCCESS in 2h 02m 21s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9afe38c5883e4eb5b508a0abf3ac4551 : SUCCESS in 1h 31m 59s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4b091171b4464a97a5247908a38e27df : SUCCESS in 56m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a34fa0a385834e3ba61847c1c086fc80 : FAILURE in 52m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93dff04d3b167943b0499664e9563697102d0089","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-09 13:56:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e77bc5137796b32971b4ea0b2165d8b065867449","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-15 22:38:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ef94c41f4028ba834dc72ba842b4590a1b4f3bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 18:16:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0f2aa8bf7a91540038e5a35dc30694e5a673c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:16:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a02fe110a0cce455bca008faf0b5eb6ad6c5821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:24:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/347898d7196c425aaa32519c04bf50bc : SUCCESS in 45m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec65721a098c4d31b6788f70851221e9 : FAILURE in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b940986378804cbeb6f7cadd5ecba79a : SUCCESS in 14m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10f323fa5c6544008272513bcc10d954 : SUCCESS in 23m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a77d6141e8143a69af7f3db2d55e70b : SUCCESS in 25m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cca2c6c22c404be0a3a0390078adde8c : SUCCESS in 22m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11be57db02ad4bb6838bb334db0cf479 : SUCCESS in 30m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9585d04c56a44d16a730a5a2ab36fd7b : SUCCESS in 8m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/19bbc2d26be241bdb60fd58ea28fd625 : SUCCESS in 1h 27m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/36d29d68580e4407addb245deb083733 : SUCCESS in 59m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12a7e50081ce42b1b592cca57a58d6e2 : SUCCESS in 1h 22m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/539d3d47f3464bf5a3c08a1952dd540d : SUCCESS in 46m 16s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c64b2e989f9446a89ee3edf071cabb49 : SUCCESS in 43m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0ca34a3c3556427486bb98feca245f80 : SUCCESS in 1h 05m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6d5d245f61a54b1cb43c47d010259c8e : SUCCESS in 1h 26m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cfdefd6deccd4f0bb4bfed5992cddc0e : SUCCESS in 1h 23m 31s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4dd458ee957a4404be600712730c28dd : SUCCESS in 1h 22m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/139d4c6216944a1fb3b83b1064a1fc44 : SUCCESS in 1h 05m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66f0d0f1202a4b1ea1cf1afe5739e7cc : SUCCESS in 1h 03m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be3949c5639c4c3db5d0c4a1527f56d7 : SUCCESS in 1h 58m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1a26185acced45c9b4ad5c141c2c1067 : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"dfc2216773659bd49172d0745184bc43bdfd3cf2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 22:08:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e9b29f4676515399bd3f8cce4845f7339f30fa0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-13 22:14:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8f64465961988fcd110e5b551e2bb8aad2f7eab5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 22:16:52.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765286/1 : FAILURE in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2b12dc95bc4b495fe40b58db32c3c5c287167816","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-13 22:20:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"067029a4174f13546efc79d61be5caa45ac83c33","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-14 11:02:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"933cfec678177925a6caca672e0045bcfe574926","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-14 11:07:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9054dae645ae890d0bb8e973cfb25a0782c391ef","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-14 11:10:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765286/1 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4d55946c0a492a0b9d5a4190c07f2406ffc5c2ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-14 11:18:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/86/765286/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f3d63aca27420af7af6f292404436df12287f86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/693dd49847184a31b618286c0ec0c80b : SUCCESS in 48m 38s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/2fb8682aa8314e10bde6361214c68dff : SUCCESS in 1h 02m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/1070fd09b2d447c6859632b304ef4c9c : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/cf2ef9ea79c9489a9fbe6ce4d8e10b4b : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/3a35c461f37f4c728e96ad23724f200b : SUCCESS in 1h 22m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0407a9355bf4e1b8dc54166aeb3dc2a : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3181856302a54cd495b746eb679be24c : SUCCESS in 11m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4902a2ef077c4f139336155084e7ccdb : SUCCESS in 22m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0bc7e439f5b446689cb3eddf2233d996 : SUCCESS in 23m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e8c6cde44c4d2ca979b453a8f52531 : SUCCESS in 20m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6545b4c84bf0449a9822bbeb269f8760 : SUCCESS in 28m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4524c8d2f0b4437f8936a5054037fbe1 : SUCCESS in 5m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/40827846d81c4816ad2a86b215b9b149 : SUCCESS in 1h 59m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/45e549045817471b84050c028a68bba9 : SUCCESS in 49m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a259bb1124884c1f8e54f260183a809c : SUCCESS in 1h 19m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ca31dd2112cb4b2ab42446f933087dab : SUCCESS in 37m 42s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e427114cc44849ed815f638e2686397e : SUCCESS in 40m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5ca5e20c32a4a6e9dbeaa0cbd142e78 : SUCCESS in 1h 05m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f978dcd3c9ac4eb484d4bc7837a5a591 : SUCCESS in 1h 34m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1e39211e316848efa79b4c9282a073d1 : SUCCESS in 1h 25m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bef70c77088642d7a8f5c92d97544a98 : SUCCESS in 1h 29m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81a26d7dcf1b4f72a6ffb9281e4a0e75 : SUCCESS in 52m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1283cc19988948c4bd5d7e9482bd2c17 : SUCCESS in 1h 03m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c458ebb0eb014793b9b83b3ad5cbe7f0 : SUCCESS in 2h 32m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d99469a38f2044e1a7c161f08b7aeff6 : SUCCESS in 2h 17m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d31c5630cf6d452e9a3d22aa84cc8ee2 : SUCCESS in 1h 23m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ab52b60715574b45ba0d7b7f7e8f1608 : SUCCESS in 46m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f45d6d217bef4855a272731d5ce38923 : SUCCESS in 2h 00m 14s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2cf27e15854c40e4abebc79e612d7870 : SUCCESS in 1h 17m 56s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a083335806264188b0ed3461206e9a4c : SUCCESS in 57m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/85bfe3537fdc462c9c1c64a1fe465827 : FAILURE in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3825fe55a6dc995665f739fb7e0fdca222df1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:12:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79289bfc8abbff01b64c44a186c67330b72ad08e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:38:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/0771f45b547e4f5283c258bef40b8d70 : SUCCESS in 45m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96beac42154e49399ff1429e8783ec81 : SUCCESS in 12m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e1faa9bc78741a3bc11d26ee69b1230 : SUCCESS in 23m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1a5d340f974b4b3c92b6746e85188775 : SUCCESS in 23m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e7296196aa745acb7203d1f4525e24b : SUCCESS in 20m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ecfd1f6ca524622814dbc9a09a013b9 : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0c09d01c83437fbc1214e9009458ee : SUCCESS in 5m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/073d1de94d8942418345dc1316bbef2b : SUCCESS in 1h 25m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7904fe16eccd4f4c87b01e4f8812b573 : SUCCESS in 50m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81dcf71643034d8d843c5d9efd4134e1 : SUCCESS in 1h 14m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74f9368386c749a4b80ea544ba8788a0 : SUCCESS in 44m 52s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b6450530a64b49fb8867d4764a795154 : SUCCESS in 41m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8cbe4efe764a4aff94b96712e6da8218 : SUCCESS in 1h 03m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/de71f01663fb4bad84e01b017f9a9687 : SUCCESS in 1h 24m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5a62f1847a8840409a1a2c7b4ad7be52 : SUCCESS in 1h 24m 39s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/691d9eeb7bcd492b81ba088577089b54 : SUCCESS in 1h 24m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/222d972173de49efb3fafc9cd9fec8fc : SUCCESS in 1h 04m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1a999c3591434d62b54a6c85dee57d29 : SUCCESS in 1h 06m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a65f911c0ec4be7adb8623e949f03ef : SUCCESS in 2h 03m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/19b4b7ef90c54542a3a0c3ee65b16960 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":1},{"id":"766e1c7f3c0eb3a5dfe729dd145014fe16727e21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:39:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c43c859cc4c41029c6995013feed5cfead766e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:40:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13025a0d9f5144df8c50de1916bdf50c : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5655f46262ffff63cbda44fbbd41bccdbb9fda67","revisions":{"5655f46262ffff63cbda44fbbd41bccdbb9fda67":{"kind":"REWORK","_number":1,"created":"2020-12-03 08:35:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/765286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/765286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/765286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/765286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/765286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/765286/1"}}},"commit":{"parents":[{"commit":"10990578ad63b1ed0c6d2dcf9a0159d104af5ef6","subject":"Merge \"Rehome api tests for propagate_uplink_status\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10990578ad63b1ed0c6d2dcf9a0159d104af5ef6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-25 13:59:06.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-03 08:35:33.000000000","tz":0},"subject":"Fix calling of add_tunnel_port method from sanity checks module","message":"Fix calling of add_tunnel_port method from sanity checks module\n\nSanity checks functions which are checking if vxlan and geneve tunnels\nare available in openvswitch are now passing all mandatory parameters\nto the ovs_lib.OVSBridge.add_tunnel_port method.\nPreviously port_name was missing.\n\nCloses-Bug: #1905568\nChange-Id: Iae86705f1d30c89dc5482261d852b45787bd8782\n(cherry picked from commit ab6c59b57e732def62e3817a80d081b8392d669a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5655f46262ffff63cbda44fbbd41bccdbb9fda67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5655f46262ffff63cbda44fbbd41bccdbb9fda67"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
