)]}'
{"id":"openstack%2Fneutron~765287","triplet_id":"openstack%2Fneutron~stable%2Frocky~Iae86705f1d30c89dc5482261d852b45787bd8782","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1905568-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"Iae86705f1d30c89dc5482261d852b45787bd8782","subject":"Fix calling of add_tunnel_port method from sanity checks module","status":"MERGED","created":"2020-12-03 08:36:05.000000000","updated":"2021-01-16 12:13:20.000000000","submitted":"2021-01-16 12:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765287-bug/1905568-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky","cherry_pick_of_change":765286,"cherry_pick_of_patch_set":1,"meta_rev_id":"d6af3c7ee2b853141f7caf92798ae781d0499a37","_number":765287,"virtual_id_number":765287,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-16 12:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-01-13 18:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-12-16 08:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-12-09 13:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-12-15 22:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-13 18:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 08:46:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-03 17:09:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 20:48:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-09 13:57:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-12-15 22:38:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-12-16 08:32:22.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-01-13 18:17:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"5a7a2c1127e47c4d350d79389bd0fe442094f5aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-03 08:36:05.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"63212cc4467245a83d2f30e7a8bb8246cf558a61","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-03 08:46:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 9m 14s","accounts_in_message":[],"_revision_number":1},{"id":"43fce2f6108340e0026aeeddb120e07152c39255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:09:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/c22978ed433c427099c2a019e5c313ca : SUCCESS in 54m 41s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/cf43175022184e8e85a483f02f653845 : SUCCESS in 34m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/d477e504aa07405192b58981abd0b347 : SUCCESS in 1h 17m 20s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/64eccbc445164b589d098c60c2efe402 : SUCCESS in 1h 04m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/e0bd49c6120949e58e2f2a5771ed27e9 : SUCCESS in 1h 07m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825ceac6a29045e6bb403fed347b5e0f : SUCCESS in 13m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ef2369be63b46a78c351e8c7fa5aeaf : SUCCESS in 29m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/433a42f1103b43258c61e721b44662c5 : SUCCESS in 26m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4b05caa0f46f4e328bd8d613a3316750 : SUCCESS in 6m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/29c008b0424a43749fb22ecaa8e2172a : SUCCESS in 1h 31m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3c520dc502504a3d933211871f4b12c2 : SUCCESS in 51m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a1055562a56403ebb43f27391e5523b : SUCCESS in 1h 25m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eb0a6e8e93184a9186fc4c27b6d8bb01 : FAILURE in 37m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/12d0dac753aa484e88c443f22107ce9f : FAILURE in 28m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7c0a785995a94bebbf96fac4b425f323 : SUCCESS in 1h 24m 28s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7999ad83230241fab72921c5cdd30f46 : SUCCESS in 1h 33m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d82f1ec0e72c496389954424837c8e13 : SUCCESS in 29m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b165a8e9c2cc4b518741d80a773710f0 : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":1},{"id":"119927426a0e14b372f3c63a67803fb2ac242def","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-03 20:48:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5e6f03acd18fe576fee2b5c20bf418dd9ec4ed08","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-04 15:52:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eeb02b4dd8f790bf53a67c6ff14d92616058eda0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-04 16:21:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 8m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c59029899d2c14444ab5a82590a8afdc04ad2db8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-04 22:55:26.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9430bf4a8331cefe07e0bcdfee54322817268522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 03:50:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/231ba271f2154012b5e838ee06481134 : SUCCESS in 50m 24s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/c95a507055c847bf830e18c80e4e76c9 : SUCCESS in 31m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/a23c00b7b674409da26e17ffc9f1f903 : SUCCESS in 1h 38m 22s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d57e6bd6ee044c19b801d8a820730d85 : SUCCESS in 1h 14m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/499e98d499b0429295e0596b8256a66e : SUCCESS in 1h 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8346c525c72c43c4934cf688cc956922 : SUCCESS in 12m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a52fed0df9374246958d9e0358cdd8de : SUCCESS in 26m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f7206732232c4e75b39382bf52d28759 : SUCCESS in 35m 27s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a700829ecfa04ed08cd6dc6fe2e2b680 : SUCCESS in 7m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1e396ed06764069874e00703c024731 : SUCCESS in 1h 10m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dcaa5bd7272c45e68cddd3e4c66c286d : SUCCESS in 59m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b3392f7e6f248269fe1f9eda4b4fd63 : SUCCESS in 1h 19m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c567badedc14d29a403f70eb4ce2a49 : SUCCESS in 41m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dd46b8db3260435c82d720383cd44465 : SUCCESS in 1h 33m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/80334e75a03e49c181f5a95199ba2a5d : SUCCESS in 1h 10m 56s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2566dda31cdf4ae78ed622ca33e9c47d : SUCCESS in 1h 34m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e432e2a897cf4e58ad239ef4c3274448 : SUCCESS in 40m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/177ac7852c4d45b6a6c864406f241197 : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5457d50528ca0c7f90ba55276ce9166e9b7d4ae0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-09 13:57:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2535ca18f188fc652f00cf1df87a3220bf305f1a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-15 22:38:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd65d06fe75ebdb0aa6c28ce8361621854753504","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-16 08:32:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f33e8f46beb08bff352b713d57d631741f718fa5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-16 09:53:33.000000000","message":"Patch Set 1:\n\nrecheck checking pip resolver","accounts_in_message":[],"_revision_number":1},{"id":"b134e266c8638149a62afd8222f8ee3c09f1d068","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-16 10:40:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8e825c181db3226df00235eb1282c3ebab170fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:40:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/a5161885f4f94f2bb9e390522b90c348 : SUCCESS in 47m 36s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/75afd053313649e687e35b9b018e6f1b : SUCCESS in 33m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/ffdffe2abba54ee39f41a8da469a4d27 : SUCCESS in 1h 14m 46s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/2600c0d54489490ca98e1c5abac2ed95 : SUCCESS in 1h 08m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/0bb851d590274836a3850fee050aad06 : SUCCESS in 1h 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d1c7471b03434e9298ab42f1d3842a : SUCCESS in 10m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0b18bcb240d4dc995cafce727be53e8 : SUCCESS in 26m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d58f9e850d1e48ba909455755e199a85 : SUCCESS in 24m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/08010827d40f459487687c44856ccec6 : SUCCESS in 6m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/651cad2cf5fa4486938745dcbccf7f13 : SUCCESS in 1h 33m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7885c42434fa4eebbf721ee75ffc9b0b : SUCCESS in 58m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af1003a9228344449b7e908ca625a90a : SUCCESS in 1h 20m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/11235e95ff9b49e6a895472693504093 : SUCCESS in 52m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fe81a80eb64245858e2f921a9d323c52 : SUCCESS in 1h 21m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9d3f65866a55404fbf82f5d52cfc8ec9 : SUCCESS in 1h 08m 39s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d83d3281bb846ee83c36fcf4e9ff632 : SUCCESS in 1h 21m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3fc838ad4324f4293aba66a3b298a56 : FAILURE in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d78a437c693432ab05bf88d22631888 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8b944ac7f9e88b1900a1297116ba9ab728c325db","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-18 15:26:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2acf6458d2c8958c3e3bb6d0a287567fd2235fc8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-18 15:38:31.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"18a4c128679176e53885f51e4356d84eb9ac613f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-18 18:04:35.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9580ac8b571b815ddbf01213d6f50fd8cdd85bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:29:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/ab2e7f3be6dd4bba8b40f267e4d2cb30 : SUCCESS in 42m 57s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/c9b72d4cfb3f4e9fb852ae32e2892d18 : SUCCESS in 26m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f1f4a7522912424a8076f228153a96cc : SUCCESS in 1h 34m 02s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/73fcfe589d6f4cafb61b8f35b334e88a : SUCCESS in 1h 13m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/37242c2fc21c4815a10a62c409cfbdce : SUCCESS in 1h 15m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e24908af284f5d8dd1a8b2fde13d16 : SUCCESS in 12m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07c2f0302b4643e58554ec14458107aa : SUCCESS in 28m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb116417227242f6a9569e2888178be5 : SUCCESS in 29m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/85e06555c678478f95e57adae62eaddc : SUCCESS in 6m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e6f27a0dd2b412fa59c4c2beca53e2d : SUCCESS in 1h 23m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3d707d0e36644da180be1041c19d1349 : SUCCESS in 1h 05m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/299d780f680547479b074a9fa0c584c5 : SUCCESS in 57m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/669b8b816c5b477ebf8264259dac6727 : SUCCESS in 37m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cae3cad48b7548939011d6962d43e739 : SUCCESS in 1h 25m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cb92fa518f6649509ee5021f9e3bad8c : SUCCESS in 1h 27m 59s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e87ac2c8da347349d5e5b41cc80d8bc : SUCCESS in 1h 24m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ff2430845304e838b17457e0a6a34f8 : FAILURE in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f67ac90f61614174a3bcadc7bdcdd57d : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e3097ec8faa19c1739178005212f1df38da6498e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-11 11:10:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a9cf5187e6851c661d686b65e808a95dc5eea9c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-11 11:20:16.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d012ec95d4589b35c96f46af693efb8aa2a294f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-11 12:59:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"087634a4f0d2735c949ed705bd1f3b98438f58d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 13:58:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e69a5898fbd24b9db09bfbd0045567a0 : SUCCESS in 45m 33s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/1767e77cc352480abe805b4de3c6e1b2 : SUCCESS in 31m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/a74de3ab4643477a8da6f478495bd720 : SUCCESS in 1h 02m 43s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/1b7a67f906284a6198d47def5ea472e7 : SUCCESS in 1h 19m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/ae9785bdf9804a46adc3c8c393df2103 : SUCCESS in 55m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4e85cbf7b64ebc84885664974e680d : SUCCESS in 11m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4a40781883a47b7ad7e619d77211291 : SUCCESS in 22m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7ec35a78f5d4ea9b214cb7c8fa7efbd : SUCCESS in 26m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/67ef4a70a70d4e3ab4d609e8220220d3 : SUCCESS in 5m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/28780a7e448c4213adf34e37fab25ca1 : SUCCESS in 1h 24m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b2768a492df040e7a0637cc522f9eea4 : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/449332a751ac449a823f8491b0cdf7f8 : SUCCESS in 1h 24m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/92521c3da6ea4e8cb83b23161a158d1a : SUCCESS in 41m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b9a14629f0e64252be11d0cb4921a782 : SUCCESS in 1h 36m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/23022645912a4d30a03f0e45568f940d : SUCCESS in 1h 19m 22s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/129958be108549aebaaf89bd964bca28 : SUCCESS in 1h 22m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/161f4ba1c6204027b707b457d9a3833e : FAILURE in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74c64bc0a866450ab65d49a89d07f097 : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"69e259b6c9f2e8ef5de3210acbe66f4a1157745a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-13 15:27:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15e08b41d02971d1af79b65f0b7ca6b9d1e939e9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-13 15:36:47.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4bb6d169f6ead8572d8c5c1b2554534d1da3500c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-13 15:39:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ef4a11a125c82251d2d834d9f053fa6972da3b2a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 18:17:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4940a95482cc2f33606e7de6117032d2688a4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 04:31:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/feb2fe9b2fea46eea53af4f201f56cfc : SUCCESS in 47m 33s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/351508d45ebc4ef0b8e5b9c438ac9d13 : SUCCESS in 24m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/c673b3fea48e44068ac2ac52607244af : SUCCESS in 1h 04m 28s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/32e6e933db71441f8e5d5ac97479a5c2 : SUCCESS in 1h 24m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/1d0f39046c2649479fbd09deda1249d6 : SUCCESS in 1h 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5830a09c83e245208b31ae1e796eb370 : SUCCESS in 12m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2675a7ab5be4bcda1da761147b7090b : SUCCESS in 37m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dd482cd49c924f6698eda11b6cca6151 : SUCCESS in 34m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/055ce24a54674456b6ed6fac0b183f3d : SUCCESS in 10m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7435b5e70aba49f69548cd9d7d017720 : SUCCESS in 1h 21m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7d07810ab8cd420f879aea56a4596307 : SUCCESS in 1h 13m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73af5bb47c754e00801297e507f7ea89 : SUCCESS in 1h 34m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5904b3704a94ba092b11211adb306ef : SUCCESS in 58m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cffa2a84e7e747b79808124cfb52f708 : SUCCESS in 1h 42m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0ddaff78a81b4c0393c11b74fadaf8e5 : SUCCESS in 1h 39m 38s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f762570bbac34e1dbe5e53ebb1048184 : SUCCESS in 1h 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c9b81dbb3ac470290ded0bf987dd8b9 : FAILURE in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1d85e00ea3846d192be7dad1c373caf : SUCCESS in 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9d0541715919d69f40fa6b304567adf9e9736c17","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-14 11:02:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18781be982cea86ac56b78b02a01767e14fe7222","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-14 11:10:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f952e809659a770e34069025d86c3299a5e8a0a4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-14 11:10:57.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0122b32af839b7e8a1533fd99b9f271f3791b3b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-14 11:21:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c80fb1fa5a74bcca489ae39ba8eb33f165ebc1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 05:11:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e4a9526642a24669929a3e1fb5f23d59 : SUCCESS in 52m 14s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/911aeb74e9a74cdf95f20952bb146934 : SUCCESS in 35m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f246cafad0e942408f645d8fae0acf2f : SUCCESS in 1h 42m 36s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/1e67d23c2bba4a2ea9c25b7b5dbfece8 : SUCCESS in 1h 38m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/9465e9f909ee415f8a58decd7ac9d54a : SUCCESS in 1h 02m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b13e783c30246d6ac8d356f46b0f1d0 : SUCCESS in 12m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a589aaed3be1487aa2d82385958e1304 : SUCCESS in 24m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bf9b4264f4e648cbbe60c0a17798074e : SUCCESS in 28m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/624483e1cac44a3aa8e060d9bfb27e13 : SUCCESS in 7m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/226f8792c0dc4274ac83b819b884af2f : SUCCESS in 1h 36m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/299f7b5d0edd48b2831f671a6d499447 : SUCCESS in 52m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cf70a267e0644a385e8098bbda4cc75 : SUCCESS in 1h 46m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9921ad47f59d4dd1b39556e4d569b300 : SUCCESS in 40m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/39c866b725c546dca6f5411c3d8f4127 : SUCCESS in 1h 20m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/41ec4bdd749c4a0ab4eb7941b26e6e26 : SUCCESS in 1h 37m 33s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a592d08fba2a43fcab7c39f1502427c6 : SUCCESS in 1h 19m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10f12e34db56413d907799f5c7a27969 : FAILURE in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa6695d87d8046e282e8d5e3b8695160 : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7849b5acaada9ea3c48727326772c1bcfda16d9b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-16 06:56:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa5732d20258757c6f9357652086d0943c552962","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-16 09:03:28.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765287/1 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"19aa599ea3c2d1934285e36d476fa3078a6938cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-16 09:37:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/87/765287/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a5daf697b08fc9876697de6e726154a3af4f4d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 09:50:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/9ed0f2b1e7694487b6a658ea41c22da5 : SUCCESS in 50m 56s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/4b830bd5ddb447e0a12fe1949dd60a73 : SUCCESS in 25m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/b29cb125fafb497eb3fd8cbfcf8e4cd1 : SUCCESS in 59m 04s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/41e9a2886d974b409de8317ae2eb637d : SUCCESS in 1h 09m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/675d5611557a4fe68c580edf0ae38073 : SUCCESS in 55m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f849758cca8243dda0aef232875a1948 : SUCCESS in 12m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3960573e02a42ac9f61804f64698af0 : SUCCESS in 24m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1ea513fd0e54426a86dcd46ed9306fb2 : SUCCESS in 26m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a341c3dafde74f09ac3531cfeeb36117 : SUCCESS in 7m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f782046c9f7d42ec8ba34a61606f7103 : SUCCESS in 1h 17m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9bbdd7895b024343951f6014bfcf0872 : SUCCESS in 46m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8af9df113f6f4b919e7b50301ddc35ad : SUCCESS in 1h 20m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/52119fc1b1d3449aaaa64741e09cc174 : SUCCESS in 40m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f17428b2b7824775af3aa850249dac89 : SUCCESS in 1h 23m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/96199d285dfe47fa827e10b273c22501 : SUCCESS in 1h 19m 35s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f010f6456a64b40aa14d17018df1956 : SUCCESS in 1h 25m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1941d6210ef4c2a900881d7538fbeff : SUCCESS in 29m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0a31f76c9a0f7be694079d3cb545b9c9dda5050a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 09:50:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ceca92a9429c075ffcd44fd905f7412d76ffbc19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 12:13:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/82a26a9e18f84ce3a04c9da1374abf58 : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd034ef4533a4a588307dd7f688fbd95 : SUCCESS in 10m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c35eb6c5fcd4fc78b5659277d4065cb : SUCCESS in 22m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94d24d0a18b84c7b8e3fc99820e100ec : SUCCESS in 24m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ed47dcaf2bfc4d5ab350426d51943573 : SUCCESS in 5m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d468afc71f2d4efaae71b91043fae85c : SUCCESS in 1h 10m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6159f1e9b3344de9bfe6328b6b9a6767 : SUCCESS in 41m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33d74dea0ed04128a12a21a8d5c2d91f : SUCCESS in 1h 09m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c3b6776337884c99afe504711640cc95 : SUCCESS in 34m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d24fdf22497c413e80e986a736817fe6 : SUCCESS in 1h 11m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8152a2ef25724476b2f7af90132100ca : SUCCESS in 1h 24m 58s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12b71d6d5c104df3aabe26b11f1b392f : SUCCESS in 1h 07m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f486377b1ba44259b629de1f3419fc84 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d6af3c7ee2b853141f7caf92798ae781d0499a37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 12:13:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5e3b919a1f19f286b4e5ca7181fa242d154f59f","revisions":{"e5e3b919a1f19f286b4e5ca7181fa242d154f59f":{"kind":"REWORK","_number":1,"created":"2020-12-03 08:36:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/87/765287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/765287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/765287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/765287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/765287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/765287/1"}}},"commit":{"parents":[{"commit":"4bd90a00ce22d432bbdb138f3be08dfe052adc6a","subject":"Merge \"Revert \"[Security] fix allowed-address-pair 0.0.0.0/0 issue\"\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bd90a00ce22d432bbdb138f3be08dfe052adc6a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-25 13:59:06.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-03 08:36:05.000000000","tz":0},"subject":"Fix calling of add_tunnel_port method from sanity checks module","message":"Fix calling of add_tunnel_port method from sanity checks module\n\nSanity checks functions which are checking if vxlan and geneve tunnels\nare available in openvswitch are now passing all mandatory parameters\nto the ovs_lib.OVSBridge.add_tunnel_port method.\nPreviously port_name was missing.\n\nCloses-Bug: #1905568\nChange-Id: Iae86705f1d30c89dc5482261d852b45787bd8782\n(cherry picked from commit ab6c59b57e732def62e3817a80d081b8392d669a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5e3b919a1f19f286b4e5ca7181fa242d154f59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e5e3b919a1f19f286b4e5ca7181fa242d154f59f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
