)]}'
{"id":"openstack%2Fneutron~765288","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Iae86705f1d30c89dc5482261d852b45787bd8782","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1905568-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Iae86705f1d30c89dc5482261d852b45787bd8782","subject":"Fix calling of add_tunnel_port method from sanity checks module","status":"MERGED","created":"2020-12-03 08:36:39.000000000","updated":"2021-01-13 20:09:56.000000000","submitted":"2021-01-13 20:09:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765288-bug/1905568-stable/victoria-stable/ussuri-stable/train-stable/stein-stable/rocky-stable/queens","cherry_pick_of_change":765287,"cherry_pick_of_patch_set":1,"meta_rev_id":"7fd30f4e69b4edda4a4b21e223b161f0c33c9873","_number":765288,"virtual_id_number":765288,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 20:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-01-13 18:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-12-16 08:32:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-12-09 13:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-12-15 22:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-13 18:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 08:47:00.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-03 17:26:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 20:53:30.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-09 13:57:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-12-15 22:38:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-12-16 08:32:32.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-01-13 18:16:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b58dde7e8c7e55c51561ff7eed457cd9aaf474f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-03 08:36:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"fe3b2eef0fc14998456d3eb93852906c1e179a27","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-03 08:47:00.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765288/1 : FAILURE in 9m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dee6a6be298c8a942e75b9460cc288adb76a33e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:26:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/729e779c1f5f4029a9b167826951805e : FAILURE in 23m 49s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/27508267dadb4d4282f654f958398a89 : SUCCESS in 1h 06m 57s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/042fd731882c4ecda32a9571bc3c06d3 : SUCCESS in 1h 04m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e28191834140939964baf51ba27b86 : SUCCESS in 10m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53ae45be4a234b0b9be539ac600027df : SUCCESS in 23m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46ecfec0589145fbbff31ece1d422a1d : SUCCESS in 25m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9648025bcb3f4c059a9b32a8fb138bca : SUCCESS in 5m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/20a3d6467fd7473d8ee23367af3adffa : SUCCESS in 1h 17m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b0b06f4634a54e2bb8370cb4d944c830 : SUCCESS in 53m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a811e0a8d0bc4e78849ad3e1aac16f3e : SUCCESS in 1h 32m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5f6cc5312c7c444d8730147391cd1031 : SUCCESS in 43m 19s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/1ee49bd154324f0699ab3bf9c0cff67c : SUCCESS in 1h 49m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c0b4490892345519f949e74a13a85bd : SUCCESS in 1h 26m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6110af2e048e4b1da01585bfbbbbc49e : SUCCESS in 1h 19m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdcfcef37f4b455c86e8bdcbfe88e4fb : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0bef8f436bb737261453cc3db887411d3c0649c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-03 20:53:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4503040afc4a2721479699b82b1584770b55b91d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-04 15:52:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1cd07d41422c82421fc3129c0f0ba230019f609b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-04 16:26:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765288/1 : FAILURE in 8m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2ff2086791c6b93403678c7e30c180fa53e8a949","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-04 22:59:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d5c0639f717c6e93cc33ef958f9c48502d6b85f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 04:09:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/97a9cf1c6be740589caf9935020c01a4 : SUCCESS in 45m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a3fb4106e25b4eb39649154fdff8a824 : SUCCESS in 58m 06s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/a099129abd2544d3b5cdd3159ac4e59d : SUCCESS in 58m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0863c1c1e4474ec7aa36fd80581c9927 : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e902553fccd14cc297fd501d605adcf6 : SUCCESS in 22m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba6484e2410e47baac30ce3be29e842d : SUCCESS in 25m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8d3e52c223d04fdeb124f9ea20f4bae7 : SUCCESS in 6m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e86d77bb73074386a3cde91f7f6eafab : SUCCESS in 1h 20m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8d7a566b61894433a5b1da860e2f4f97 : SUCCESS in 56m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b5bde3301c1421bb4808a36a21ba772 : SUCCESS in 1h 18m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/85f49405dd5245a1a8ca90fa68826775 : SUCCESS in 42m 41s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/5a5fb7b864c340959310281316479f13 : SUCCESS in 1h 49m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ca6d8d647240411eb069168bab2ec56e : SUCCESS in 1h 13m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6e74b867fbe445d58eb1674e94299191 : SUCCESS in 1h 06m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03eaa49d6d7f4b369e36d79f4271126d : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e6512cea85e6805a2f1736c0a6dd510c6276242d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-09 13:57:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da636d6723b72de31d0321ef8c00e215fa3c718c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-09 17:02:48.000000000","message":"Patch Set 1:\n\nrecheck seeing if queens has pip resolver issue","accounts_in_message":[],"_revision_number":1},{"id":"d1e1f45e006fcc89aca750962944ddb3d88fb9af","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-09 17:12:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765288/1 : FAILURE in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8ff9129e98b17ac7d9c01a5d9f9a0cc4b75de41a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-09 18:48:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"01aa2f31be1e115ccd31ea59edc4711de3de0a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 20:33:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/09ff111aa37f4e76ab4d33e4c3083c81 : SUCCESS in 51m 45s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/cdca258a27b1456f8e54b4d2b39a00ea : SUCCESS in 1h 07m 50s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/a0462a747cbb4c45b9c7def780ac27e1 : SUCCESS in 58m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c065311fed3741a08eadc941d05ce08b : SUCCESS in 12m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e9c07da0ecf4a8b90c2e3ee073afbff : SUCCESS in 21m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4266290863434ad09628828b17bd540b : SUCCESS in 27m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f333e3f7095f433fa553b11c7f7e554e : SUCCESS in 7m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/081c5e0950304d1da89dbe718afcbd47 : FAILURE in 17m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58d6bf267793486fa0fe9b5f29e0391f : FAILURE in 21m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e9552ec66df4c22baca1c3679ba2c11 : SUCCESS in 1h 32m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1a87d9f09c6844aea75f364a47b3122b : SUCCESS in 43m 14s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/b27dc894ba694597ba245596abb17910 : FAILURE in 16m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1abce298f5c0477580d17b4df6b42318 : FAILURE in 13m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d87ab1ec403047a7bc8b96fa9d2b5891 : FAILURE in 20m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ad8632c48bb4dacbd47066c5cd958c8 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ca0884f55dcc349efee28ab5a400fe83dc2712db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-15 22:38:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0100ac4adebb9a9f51d55927534770070ef2a23c","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-16 08:32:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a28e865f926921ef2fef9a6986e205e06297f0c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-18 15:26:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ccebeb914663aad6d1529117859a4a9feb5b29f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-18 15:46:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765288/1 : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3a876ca967567d1763a40d744d242e06244e8117","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-18 18:07:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/88/765288/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4d80384d55432bfc0c7701f5413ada3bf05d5dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:43:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d8f85daf20d540e5be2acc708ea480e9 : SUCCESS in 43m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/79c0a7692a734240b51ad17f8aba93ef : SUCCESS in 48m 58s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/7a186529f907499e948d17fffbf46155 : SUCCESS in 58m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c144b0c29944e391d0d5acf6e3c350 : SUCCESS in 17m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/233c5035780b43ffa36bc5ebcf8a728e : SUCCESS in 18m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/48d45eeb1c2c4371be33ffafe5cfef89 : SUCCESS in 18m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8330ec1fb17d4b2c87d280b9c2d82c9a : SUCCESS in 6m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/052e7382b26f4a80ad1a1772e8e44bff : SUCCESS in 1h 07m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/82f7fe066b3249658707cd07641e2795 : SUCCESS in 45m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f709703e1264e30a4d8f1108f82339b : SUCCESS in 1h 18m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c798ef9621a944f69b6f9d9278dbcd0e : SUCCESS in 32m 21s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/afe47b20b9b04ea1aca22db0f627c68d : SUCCESS in 1h 45m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5e6c98a9dd94c5e954622ee735a4fd0 : SUCCESS in 1h 11m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a4b16766b2fb4c9aa3b9506bf4eb6ba4 : SUCCESS in 1h 08m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fae0e48200154fa7941a023b7988450f : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"11d0b21ac5c0800ac39a86ca4747225e88404bfe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-13 18:16:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52475b919d3c9035eb6e72987091518c3eceb798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:17:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"899f42070e7d2038f6ecc024c9aa41497cb44fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:09:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/4fa1a3c0a3b245b4a7db90dcb29a2cdb : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df75402485a043718642ff99fb0c5fdf : SUCCESS in 7m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f22382548d774eb6bf4e5ce326694eab : SUCCESS in 21m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0977854f7c724f798e550aa56ba6c6cb : SUCCESS in 31m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5e7222b8e4f6465286244299b38ff08b : SUCCESS in 5m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6f3a761bcbc74bdebd39df929c6103a6 : SUCCESS in 1h 18m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2d80f82df00948f5bbebd3c7347f1ace : SUCCESS in 1h 02m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1f1250889ad4ea48273aedc61a8982b : SUCCESS in 1h 33m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d2352653dfa54a0393291a17a601d84c : SUCCESS in 49m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e818a0e42d3142b49f464edc45d24add : SUCCESS in 1h 39m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5af9d765ae754e90818f3b2426246308 : SUCCESS in 1h 26m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffba5ac07f3a4ef48ce575bc1190c380 : SUCCESS in 30m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7fd30f4e69b4edda4a4b21e223b161f0c33c9873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:09:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a05e657d3b116e978fa415c73892ec9f8f5aaba","revisions":{"4a05e657d3b116e978fa415c73892ec9f8f5aaba":{"kind":"REWORK","_number":1,"created":"2020-12-03 08:36:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/88/765288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/765288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/765288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/765288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/765288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/765288/1"}}},"commit":{"parents":[{"commit":"0dce8dc53ea95da5603c3b3b9c7f5c37577e0ba4","subject":"Merge \"Revert \"[Security] fix allowed-address-pair 0.0.0.0/0 issue\"\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dce8dc53ea95da5603c3b3b9c7f5c37577e0ba4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-11-25 13:59:06.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-03 08:36:39.000000000","tz":0},"subject":"Fix calling of add_tunnel_port method from sanity checks module","message":"Fix calling of add_tunnel_port method from sanity checks module\n\nSanity checks functions which are checking if vxlan and geneve tunnels\nare available in openvswitch are now passing all mandatory parameters\nto the ovs_lib.OVSBridge.add_tunnel_port method.\nPreviously port_name was missing.\n\nCloses-Bug: #1905568\nChange-Id: Iae86705f1d30c89dc5482261d852b45787bd8782\n(cherry picked from commit ab6c59b57e732def62e3817a80d081b8392d669a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a05e657d3b116e978fa415c73892ec9f8f5aaba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a05e657d3b116e978fa415c73892ec9f8f5aaba"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
