)]}'
{"id":"openstack%2Fneutron~765343","triplet_id":"openstack%2Fneutron~master~I109c73b7ad5f1157f40b8340723084f8c51272b2","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I109c73b7ad5f1157f40b8340723084f8c51272b2","subject":"Implement secure RBAC for auto_allocated_topology API","status":"MERGED","created":"2020-12-03 15:45:32.000000000","updated":"2021-01-22 03:30:18.000000000","submitted":"2021-01-22 03:28:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765343-secure-rbac","meta_rev_id":"f9f160b968e31d74e9a491087b9fee921441b75f","_number":765343,"virtual_id_number":765343,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-12-15 16:32:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 03:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-12-16 10:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-22 01:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-22 01:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 15:48:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-12-03 15:48:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-12-03 15:48:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-12-03 15:48:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-12-04 00:23:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 03:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 16:23:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-16 11:02:36.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-22 01:09:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"6179f0d1138db64c294d6aca332fad59077c4c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-03 15:45:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2991195ccccb9999e39dd1039d06acedf44dfc9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-04 00:23:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/43/765343/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a91d4388d27dbb09401088637e3137d8db52c980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 03:46:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad8887ba3ba348a19b5cf34cf12a479d : SUCCESS in 54m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c23ab3416199416fa0c70dd00a9bea1f : SUCCESS in 33m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0629f5a49fc847fdb5a1bddf741ad9b2 : SUCCESS in 2h 10m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/809687b6ebc644cd93ff62d19fdd26ba : SUCCESS in 1h 39m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b578b7691f8d423e9c6acd67b187e17b : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be470862606143a39376e9d2a49b73fb : SUCCESS in 1h 31m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecd4c33669e24d0e8b15275cb90c84f5 : SUCCESS in 39m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03578fafdd6a461eb1c98c1dcb12831c : SUCCESS in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735ae3dfc3db443cb35aaf83d13906bf : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b88bc4045a7a43cfa8568d6bc43c1101 : SUCCESS in 27m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49c1e190909b45198bcfc044baaee0a4 : SUCCESS in 27m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ee586913384ed0b38e79a6cce990ec : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5098961955e48019d9e98246b6a119b : SUCCESS in 1h 15m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/53eede8be93042c6a8c1954e568a42b1 : SUCCESS in 1h 07m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dce0dcae97fc4bc7858b286e95e4a56f : SUCCESS in 1h 37m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/96d7916ee0e04720a7556b639fc47205 : SUCCESS in 1h 03m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ce739eeee714ead86c09981d35c9e72 : SUCCESS in 1h 01m 20s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/af42b6986e7f4cc5bd3160d128a8e87e : FAILURE in 1h 28m 28s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae401444928b41a08348601dd15055e8 : SUCCESS in 1h 13m 14s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e56f138882c94a28b00fd63794232068 : SUCCESS in 1h 30m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb7b53a5712d471e93181c3c272f4d3c : SUCCESS in 1h 30m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/adbd52e857cb46d09944a3993d7f3bdc : SUCCESS in 2h 08m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d798004e89be4c2b832da45d484012de : SUCCESS in 2h 48m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ab158fd0be4651b5a66b9d39232273 : SUCCESS in 1h 08m 11s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/44a5d9e2e5db4c78aa6c3761a4311f9c : SUCCESS in 1h 10m 51s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/efee4671bfa04a7794e8c02cd594d78c : SUCCESS in 1h 08m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8105729f145449b5be268ee2bc492e5b : SUCCESS in 39m 20s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/4223bd6fc96d4bafb325e17fb08707e8 : FAILURE in 1h 09m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6d55fb5f811d4eb0a4e05cc2d25db2ad : SUCCESS in 1h 10m 58s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e821e7c8431a40ef8160a8fe23cdeff0 : SUCCESS in 3h 27m 37s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6c5071d1a7c14a898ea672742b091d10 : SUCCESS in 1h 12m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1451aa5074af4c62ac4b8156d88198a0 : SUCCESS in 1h 33m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8ed6dad5cab6433c9f016825b510e1c5 : FAILURE in 1h 42m 19s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cc8299b349f546f3beeb56243c895566 : FAILURE in 2h 58m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7274affcef2548459f1b31277d5f6e3d : FAILURE in 1h 55m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a7a917010db0fef33c9d0b6858169876e3be562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-15 13:36:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b58f2aaf591ca03f628bb23a997ba4e75668ce54","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 16:23:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765343/2 : SUCCESS in 2h 10m 11s","accounts_in_message":[],"_revision_number":2},{"id":"57e23f0caf39a8e3e4e34959d8b82e4632c44daa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-15 16:32:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/43/765343/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5f4beee7c663d23febc869672b0325935972f763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 23:23:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d9fe7129e2a4beabeccfc910793c4ce : SUCCESS in 54m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f27150dce1fb4de19d2a0a7c3c9ab3c8 : SUCCESS in 41m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e37557f67ba54459a17d0b4d82c18c41 : SUCCESS in 1h 38m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3f3ae3fa66a442588a614b8ef765ef04 : SUCCESS in 1h 54m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b73999b437a45e9a24b75ec5beec585 : SUCCESS in 1h 35m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/355d54255aad4db0a9e6152caae76c6e : SUCCESS in 1h 29m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0781c906c73649248bee4e60e3105c2b : SUCCESS in 33m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e418de050227416c8fe6b1c49cd90bfa : SUCCESS in 40m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816019ec6c374a9a82d2dd80d9e60ed2 : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf44e9c109448fd86dad38691a030de : SUCCESS in 27m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69dabe7e86794a36a8d53a03ffe6a0ed : SUCCESS in 23m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d1bbfe4f3d94bbeb2661038238b028b : SUCCESS in 29m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9f4b6fe49b458aa6addbd6740dcaa3 : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/347cb8a6a0904b6a8e9b60e4100a78e5 : TIMED_OUT in 2h 15m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad9fa16ceee548669600094873608221 : SUCCESS in 1h 10m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/29386486638f4e0c8c8d89bf70f49246 : SUCCESS in 1h 42m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d98d1045dfb4f1a83f40a4540f9b10d : FAILURE in 49m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45971b3c40a34ba6b531c0c2ef86e00f : SUCCESS in 1h 08m 52s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/94f2888a0cc844278d3825e9e95d8997 : FAILURE in 57m 12s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c21cf2396d2040fba8fd6af7bd525739 : SUCCESS in 1h 10m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/71f3474642a84f228f12b14e7c98e3c2 : FAILURE in 1h 43m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/308187733a9246008c6876443fc5d994 : SUCCESS in 1h 14m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8b7cc07eb124748b395fe26ac47525c : FAILURE in 47m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cd6f168cdf948a8aa9fcda38a014dbc : SUCCESS in 1h 01m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9ac8275e84be4c09a2d256aa28d57ac2 : SUCCESS in 1h 07m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9e8e5b01ed1c45c99f5d0aad23eb38ca : SUCCESS in 1h 05m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/10132ef594ae400796e983aff83968bf : SUCCESS in 33m 11s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/839ccaddc1fd45579442585842eaf0e8 : FAILURE in 1h 03m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4c4ec62424a945808cbab3c9f22243d9 : FAILURE in 39m 04s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a9e7bb59569347be8a62bc5daa9c4fdf : FAILURE in 45m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56fe4ae6a83644a78149501884e022d2 : SUCCESS in 1h 41m 18s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e8d520af6f7645fdb4742dc19b28e9ac : FAILURE in 2h 56m 17s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ab54ba9022ab4f259f2f6069de11a0fe : FAILURE in 2h 09m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0deeebdffa2f7a4124ff98cccb20d97b27466ebb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-16 10:52:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08d15a7174a5352afcfb83c7ce93bc58d8ca2e30","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-16 10:53:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e30be23e3d3b9a0dd97364ad4eef849ec4967994","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-16 11:02:36.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/43/765343/2/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/43/765343/2/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"f9970445405331ac7f9d600cfb5c63c044d9a559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 01:51:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7323a43a7a6e4580bc3b23de2d38f254 : SUCCESS in 1h 00m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa7d19f54de84eb4abcdb250c9866488 : SUCCESS in 33m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/094e0563a27b456fa22c8fd015f83f01 : SUCCESS in 1h 36m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7177b690f5a94ced93d9a3546b29a6df : SUCCESS in 1h 22m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de63a29ab7044daca70fd2e47df8ba50 : SUCCESS in 1h 14m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9210a3b5282b46b2a61566c17530c835 : SUCCESS in 1h 23m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c78ecd8238435cbc12b77806ab19a6 : SUCCESS in 32m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f14e8a5d08345aca74e0c2cb13d71f9 : SUCCESS in 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee6a76b40d743a5aeaa2834ddce10d4 : SUCCESS in 12m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490a5752811f429b83ff1920d338cf18 : SUCCESS in 25m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/148ecd6327104bbebb5fadaa0b3df658 : SUCCESS in 29m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d613d80ec25d4894adfd9baf670d9822 : SUCCESS in 29m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8468a6a8dac84fd8a3d2af5e2b70dd32 : SUCCESS in 8m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfe05c8c9b5c480e8458a94fd8ef3d50 : SUCCESS in 1h 13m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3a4f2a9fdff49ba85d59c2edeec47dc : SUCCESS in 1h 15m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/26b1c6bcde314822aac04be3ca5d0c57 : SUCCESS in 1h 41m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86bfc336147948609731c5cd78ac0b68 : SUCCESS in 59m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/95c148eb255c487aa805028f85b408d6 : SUCCESS in 1h 04m 18s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/f369e6821f6541a5b65a725122f81e74 : FAILURE in 1h 04m 47s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5bf23a222d0941f8be575373522ee8c4 : SUCCESS in 1h 20m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a349d29a6d0545dbb763329f96a62f1e : FAILURE in 1h 26m 27s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/39f5edcb08d744f9865b68738b0a44ed : SUCCESS in 1h 16m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/717ed8dc04de418e8645f89849d01bcb : SUCCESS in 2h 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b2ca58e26734291a3671e2c26407218 : SUCCESS in 1h 01m 27s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a6213473fd51472a8a58bf21d9b0bb63 : SUCCESS in 1h 10m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/034b621570434f649f19964429ecfeb3 : SUCCESS in 1h 08m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e513b63ae23740848083e033268d9128 : SUCCESS in 37m 20s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d41e3040cc1140b5abe321748599d9c0 : FAILURE in 1h 11m 27s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d71aa88f29c44315ba01078d6f26dc21 : SUCCESS in 1h 22m 15s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2bdd8827e2bb44cc98d685d509805fd8 : SUCCESS in 4h 27m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6902f33ef24a430aa986130576de18f9 : SUCCESS in 1h 27m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c676c8ea552040cdaea57928ef745a95 : SUCCESS in 1h 37m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fdd855e8e2a44a16a941d76d2130c943 : SUCCESS in 2h 16m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bd77dba1e95e4153a54002a3f86e4a16 : FAILURE in 2h 41m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/08c42f1b42f24752a29cc4a049d92e21 : FAILURE in 1h 46m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b96b9eb58eb710e70118c663a02d7809c3b726d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-22 01:09:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2268f920b1ab446d2bb08643c0ec8d71aa78521b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:10:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2758b575b8341353d7a52186ce56882ac3a9c85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 03:28:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/271b54a0d8454728b035a75d882c6d94 : SUCCESS in 46m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/63c3f9c38c6045019c3cf524468f0510 : SUCCESS in 1h 32m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1471e6c3aedb47079050b644cba4a88f : SUCCESS in 1h 15m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e236d7868e624d2096e390a01f4ca9df : SUCCESS in 1h 43m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be5ed116fa934d87a7dd7321819ec122 : SUCCESS in 1h 17m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a6bdd872af043fda1c5e1b704dabbfc : SUCCESS in 29m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03720a27080944e284a7f59996bd21b9 : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38b998c2d2ed4c28aabc0ad753d8ef97 : SUCCESS in 28m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a93d13f8c4ce4572ba5efd7ff34f838c : SUCCESS in 26m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054d8328b03843d3ae9c4cfa3e73cfe3 : SUCCESS in 11m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f0295c61b1f4ff697b454d4d1553a8f : SUCCESS in 1h 09m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f90f084282714507b39756bed2708c31 : SUCCESS in 1h 08m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc984e61f30b467cb14ca7fc5e25c414 : SUCCESS in 1h 18m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17fc6b34c1b646ceacd38833ec333631 : SUCCESS in 1h 01m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/967c15486ea745d6a9940424ffbda1b4 : SUCCESS in 1h 05m 52s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dddf5db2aa904f42ad27c8395a7cf736 : SUCCESS in 1h 34m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a092863b3a894ff49db1090a9d3f7992 : SUCCESS in 1h 12m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ebbe3abae0ed476fab4af4b3c50d08c8 : SUCCESS in 1h 09m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/af0172adc65b4dfd9bea34d24a57a521 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":2},{"id":"c1cef5acb15035173c6d9d61cbfa93ea04167666","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 03:28:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f9f160b968e31d74e9a491087b9fee921441b75f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 03:30:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1028ee66c4b49b49becdd6db873c6f5 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05ab8c591f487aec84b68832393cd2fdca68ef38","revisions":{"fbf4d35a098fe8a06f6542d724d65f4c2aba7064":{"kind":"REWORK","_number":1,"created":"2020-12-03 15:45:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/765343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/765343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/765343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/765343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/765343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/765343/1"}}},"commit":{"parents":[{"commit":"17b46e8525533ca453bb32c31e158aa6a485c4e6","subject":"Add common system and project policy check strings to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17b46e8525533ca453bb32c31e158aa6a485c4e6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-03 15:43:10.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-03 15:43:10.000000000","tz":60},"subject":"Implement secure RBAC for auto_allocated_topology API","message":"Implement secure RBAC for auto_allocated_topology API\n\nThis commit updates the policies for auto_allocated_topology\nAPI to understand scope checking and account for a read-only\nrole.\n\nThis is part of a broader series of changes across OpenStack to\nprovide a consistent RBAC experience and improve security.\n\nPartially-Implements blueprint: secure-rbac-roles\n\nChange-Id: I109c73b7ad5f1157f40b8340723084f8c51272b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbf4d35a098fe8a06f6542d724d65f4c2aba7064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbf4d35a098fe8a06f6542d724d65f4c2aba7064"}]},"branch":"refs/heads/master"},"05ab8c591f487aec84b68832393cd2fdca68ef38":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-15 13:36:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/765343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/765343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/765343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/765343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/765343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/765343/2"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-03 15:43:10.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-15 13:36:14.000000000","tz":0},"subject":"Implement secure RBAC for auto_allocated_topology API","message":"Implement secure RBAC for auto_allocated_topology API\n\nThis commit updates the policies for auto_allocated_topology\nAPI to understand scope checking and account for a read-only\nrole.\n\nThis is part of a broader series of changes across OpenStack to\nprovide a consistent RBAC experience and improve security.\n\nPartially-Implements blueprint: secure-rbac-roles\n\nChange-Id: I109c73b7ad5f1157f40b8340723084f8c51272b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05ab8c591f487aec84b68832393cd2fdca68ef38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05ab8c591f487aec84b68832393cd2fdca68ef38"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
