)]}'
{"id":"openstack%2Fneutron~765394","triplet_id":"openstack%2Fneutron~master~I0c2062619a72184e6681530de6c074fb3c23635f","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I0c2062619a72184e6681530de6c074fb3c23635f","subject":"Implement secure RBAC for security group API","status":"MERGED","created":"2020-12-03 23:04:21.000000000","updated":"2021-01-25 14:14:05.000000000","submitted":"2021-01-25 14:12:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"765394-secure-rbac","meta_rev_id":"aff49fe60ff9e960a86451a79e2ebc6d62d44637","_number":765394,"virtual_id_number":765394,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2020-12-17 19:55:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 14:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-21 16:19:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-01-22 15:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-25 11:45:28.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-25 11:45:28.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 07:23:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 07:57:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 10:00:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-08 10:15:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 18:45:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-16 09:31:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2020-12-17 19:55:19.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 15:22:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-01-25 11:45:28.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"f55f29ef171251571a8bc90fd788d743aea468c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-03 23:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84f5375ae259eb72992ff26f9d5c1ec195e0ce89","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-04 07:23:31.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/94/765394/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0736802a0bd113465cc82a5bf185a8f28b536954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:57:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d0c862f115547db95b279979cbb7ae6 : SUCCESS in 51m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9321fb6aee4844de865eef9ce894b544 : SUCCESS in 30m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0418d1bbc84741c9bec5260237be7c62 : TIMED_OUT in 2h 52m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/76598931ae7c41799626f38b46f4ebc9 : SUCCESS in 1h 41m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e912e12e6a27457fbf5a2e205d08384a : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2ca68aaa33e04ba2b80507489730fb0e : SUCCESS in 1h 18m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83d0e007c1314245af26903a0417169c : SUCCESS in 38m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/438aed3619bc4530ae10492aa429b531 : FAILURE in 46m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870b72d3c641473c9698dc00e42a276c : SUCCESS in 10m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5fcf2ea68804910a646375fbbe0e2dc : TIMED_OUT in 1h 02m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11268956ac124e1ab0cec1d52cc0d2df : SUCCESS in 28m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e6ff874c7941929a4f95bfe4b2d99a : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8beb129e8efd4bdfb117a476831d63b8 : SUCCESS in 1h 16m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ac4379fa1f342b8b25b3b3445e7de32 : SUCCESS in 1h 15m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e1614ce142fe4a85a44f41a59d024c18 : SUCCESS in 1h 38m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c3edc76624a4abb88984b8322fcfede : SUCCESS in 1h 08m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/47019b9a9d6d4eaabb2a93c14a5fa7c2 : SUCCESS in 1h 41m 14s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/57ae077782d24b83aabb4caa36b18cfe : FAILURE in 51m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d162356258c4a0dad55e2ac6a4cd5ab : SUCCESS in 1h 17m 05s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0746e2c89a6494998a74ff193ba69be : SUCCESS in 1h 26m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46dc58c527984a1d9b7fe9284401d791 : SUCCESS in 1h 34m 56s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/31819f8a34254c638e667a8ca078e80a : SUCCESS in 1h 19m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c801bd6c96cd422f9a9fccc1e7ddf865 : SUCCESS in 2h 16m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c9a65b900f44e8b9533923df5645522 : SUCCESS in 1h 05m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7dca9a212a85421d9b3c1448c096d7df : SUCCESS in 1h 09m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/19a78e15294d410a9bd4397f5468b95f : SUCCESS in 58m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/12e969195ea04a19bcdd1e1110ee7f64 : SUCCESS in 39m 10s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/9dfaa192dc284ef09a1091c3f941bd24 : FAILURE in 1h 06m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0470b3b957e24d29ab8d46da2531d686 : SUCCESS in 1h 23m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4471f4cf10746cfb4f56b404f7d0225 : SUCCESS in 3h 45m 02s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7a2f44e98d434a5ba04c0be48ed31c34 : SUCCESS in 1h 34m 24s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d70f4865c054379ba5eed7ef7cce416 : SUCCESS in 1h 50m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8233adfbac72407ebe1d4eadbbeeee7d : FAILURE in 2h 09m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/27e6185d9d7b45d2b2c03a6aad8e8d9d : FAILURE in 2h 52m 43s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1a1565bfbf8946dcaf11e9372c5cf163 : FAILURE in 1h 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d18f2d1ae68fa06000e884d2ebf9abb2c3c49d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-08 10:00:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e484fd09f9e5387cfaeb3ce658ee5850604ffc70","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-08 10:01:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1c8cc3c3df7956956cc67ed4b9e524aa974ea46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-08 10:15:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/94/765394/2/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"6136e4f427080c25d0e403901567d0decccc6f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 15:50:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0d46c189e6304aabaed6043227432bc6 : FAILURE in 30m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efc7e6ba3f17468986adc4f371bbd193 : FAILURE in 28m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/59541b9aa49f48f794c4a3efb3a298d5 : FAILURE in 32m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/457827ff323040a49e9887d4c55138be : FAILURE in 37m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ecb6d0e744f4f93b8efc17f26cf275d : FAILURE in 36m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a0616094e4ba4458b5e164dc42683a24 : FAILURE in 35m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e170e019c7d44e3685431e65c98fddec : FAILURE in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6454b83f82b54787b77cfb91750eb44a : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c86668734eae4b5bbf3f3a85df34ebb2 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da757f688dfc4c59b9afe9411a9220cb : FAILURE in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4f4f58d3a7848ad9717bebc74ca1194 : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88aa249d216a4afaab9cc5864afd6f79 : FAILURE in 5m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa61375d315149088064fcc484c88c67 : FAILURE in 21m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5e5b7126c674dd984f9093fcd29c9c9 : FAILURE in 16m 03s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/43651278b32b456ba329ab48af914290 : FAILURE in 24m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/afbf27e5631d4c71ba35d249b285e356 : FAILURE in 52m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce800737149a456a8ca018627ef8fea4 : FAILURE in 53m 49s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/6418b254f0cd495d93b859b1ef080c6b : FAILURE in 1h 05m 08s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0c6ec9fcd214f658d8cb2a28cd6ccaf : FAILURE in 33m 20s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1632a430c28f44a9a4cdb6b85a17c8d1 : FAILURE in 50m 36s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5636cfb6b83a42048722319eb04da891 : FAILURE in 51m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3dbf1e63ee1b41359cf1fdcfd028acc7 : FAILURE in 46m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7960cedb76b94ca8b8c825058255d5bb : FAILURE in 54m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5543b82297c468cae7389547a870dbd : FAILURE in 38m 56s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bdcb6e6e78174331b7e6e438bbfc4acb : FAILURE in 36m 24s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ddd9c9704cd342278d407a53021e85e3 : FAILURE in 42m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/75caff62a68e420682e186f3c5aff16e : FAILURE in 34m 41s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/03a55e22b4954d92a71e45369dd8a303 : FAILURE in 38m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f5eb3d0f541c4109a2134a7ba15c0dad : FAILURE in 1h 20m 07s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/12d1c11a5db84e1e926c23ee2c79f6d7 : SUCCESS in 4h 08m 21s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/63bd8579582e4de88a826c3bd06e3ca3 : FAILURE in 1h 32m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c4dd71260b046e2b7860963434d5fc8 : FAILURE in 27m 21s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c408f07feada45048cf91edf5a27eed5 : FAILURE in 2h 10m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/805c54a9c9bc4d938f42c2675f5aefac : FAILURE in 45m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a59c9a85ab18410690c334b051193959 : FAILURE in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"818b1405f5c158508d68eafd21e777382e8dcfba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-09 01:31:40.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/94/765394/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"9f6fe173651cf95f925271f438b6fd18182b72fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-15 16:14:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f271984eaf055355627378c6cebcbc8a6f6d2a92","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 18:45:16.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765394/3 : FAILURE in 19m 49s","accounts_in_message":[],"_revision_number":3},{"id":"fcc17113971e1781ab3eeab6f469f3867b74f63e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-16 04:16:52.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/94/765394/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"367be093ab3e6ecf8ef95af395142fdb0c662bce","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-16 09:31:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"de7163e881670659a45fb792e6ee493602963bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:15:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/984927c0988b4ee79f3208afc37baafa : SUCCESS in 1h 13m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2979abe6fac8498b8d7db4472cb56d71 : SUCCESS in 38m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/706d82b9ffd1465dba874782f4cfcac9 : SUCCESS in 1h 51m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3add837fa5f042ac808cdc35e6c7c0e6 : SUCCESS in 1h 53m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c0734b44e234a41a0df22793574bce8 : SUCCESS in 2h 19m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d40bb34b69e142dcb917b1e3e6fe7519 : SUCCESS in 1h 46m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4908e1930ce442a96df2bf08217c601 : SUCCESS in 36m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0e63ac2b5ba4aa3862f3e6a90a67e57 : SUCCESS in 31m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36456f7191448b29854e3bad79e2388 : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65776fb02a2f4bc8a3111100e692a91f : SUCCESS in 26m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abf570c6a4014e92aaa8e62e80b9d77b : SUCCESS in 24m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/027b042975af43fb91c8f4ce51229e7a : SUCCESS in 28m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe79489904d43198f0e6ba43a6f3032 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a58ce6ffa234a7ba63e046aafdb3ed7 : SUCCESS in 1h 17m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a56f0b0c3c44909b9879668cbdb4c7d : SUCCESS in 1h 24m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d0540a61aee45f6867c23b3b030170f : SUCCESS in 2h 04m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2ad82b47ba844f183ef4dc0f2b33a0c : SUCCESS in 1h 01m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/39cf0921733048b2bb85d95885d7b841 : SUCCESS in 1h 02m 19s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/e87316ef08714e1fb6171181eac2ffe8 : FAILURE in 1h 06m 14s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a683fde7e6a64de492b81a2515514f52 : SUCCESS in 1h 21m 14s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b60a38a2039c42d1ac24393f9d4f1151 : FAILURE in 1h 24m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7e421a47a4eb479dad0449f0b44d5875 : SUCCESS in 1h 10m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e0774c3fa034a51acc19f6eeed03ca0 : SUCCESS in 2h 05m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acedf42a234c468da8cfbe15641c379d : SUCCESS in 1h 02m 09s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/11c0b9bad9174391a3120851812ad592 : SUCCESS in 57m 40s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b3c9ef391c694d35b8ce6ce0dcef6f1a : SUCCESS in 1h 07m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/233d03356ab242c1a8bc66a2f8f2dcc8 : SUCCESS in 34m 26s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d9cdc7518bac4a98b65e6a15afa94f6f : FAILURE in 1h 03m 11s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/404931719d0342098f9cb23f025f4b2d : SUCCESS in 1h 23m 12s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/76718d368ced403790a21621659a9408 : SUCCESS in 3h 10m 41s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/11158323ecbd4c68ab8a8e7b71bf6fe9 : SUCCESS in 1h 23m 30s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b6dd002f1eb4b1895f5162568893cc5 : SUCCESS in 1h 37m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/233044686db548e9bc47e557adcbbc58 : FAILURE in 1h 48m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/45e2019ece76415b98616284537b9856 : FAILURE in 3h 05m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/39d03556713c4552a0a8f02219cf415c : FAILURE in 1h 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4592300095df080cde0e6b2a9ea6dcb39277a583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-17 12:08:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5bb0f4f8f659e1b9e5e8c0f25c1a1b5bc8876c8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-17 12:08:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ff796fadb4dd9588112cbf993e3cf99d966c22b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-17 14:36:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765394/4 : SUCCESS in 1h 48m 58s","accounts_in_message":[],"_revision_number":4},{"id":"fa3b458b4422de3cdfe77d9f9a30012e919f0841","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-17 19:55:19.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/94/765394/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6225dd313b217baf758c6b3a48799b6ac5b5ccad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 21:31:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37f30fa7b0ad4f9f825476604b06ec48 : SUCCESS in 43m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/898fd903c36b4faf993452f3f6a1ec3c : SUCCESS in 28m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/07fb0ae1faa242bc8e03f743a60a65b5 : SUCCESS in 1h 44m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/147f789192524307951dbc8695e57806 : SUCCESS in 1h 31m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b38a599bc644fa2b183e76103bebe46 : SUCCESS in 1h 19m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/39e55914503e4dda88c259432f87eb26 : SUCCESS in 1h 33m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79c01efaccbb4103b2d5d1dca88d5ecf : SUCCESS in 32m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8722ad96b134de9a57422471e72b39c : SUCCESS in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c05c6f753f48be8fd84d40451ebd84 : SUCCESS in 12m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1b3e7444eb435398a7833ed4064eb5 : SUCCESS in 25m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b918eb84b4a44821ad6f316e999e3b29 : SUCCESS in 20m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6106dd34806845c2a22412e84a09b6dd : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29044ba7d844482acb089fc27b9b6dc : SUCCESS in 19m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/66173a1d84824c0b8f8caed467d5c201 : SUCCESS in 1h 20m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/22bb3cde70cd4cae9d442624a37be566 : SUCCESS in 1h 14m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/859b5c50184d489bb863f1caee9c467e : SUCCESS in 1h 33m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e60966d12d874181a57e7e5aff9de253 : SUCCESS in 1h 33m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/014f614abba54fe6aedde719b05e206b : SUCCESS in 59m 30s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/77e199d80db34d5e9253152f4671512c : FAILURE in 49m 30s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/246c67b2564041ad8dba8c8a57b58031 : SUCCESS in 1h 46m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/def7bbcfcaf246a391313137e088205b : FAILURE in 1h 13m 04s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6d012d0574e34bb6af8679cdffcd604f : SUCCESS in 1h 22m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50d9202243464b458de5e0bf08051941 : SUCCESS in 2h 12m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09a44e78849641cf960043f183f327dc : SUCCESS in 59m 13s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c549e0495f7d483e9b42a30a65ef6227 : SUCCESS in 1h 14m 11s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/57834ce17a8141ff955529569f1850f1 : SUCCESS in 1h 09m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/26b7b0e2cc4c4d2f9b9c845a2b3310cf : SUCCESS in 31m 46s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/163b4586b8814cf8974ef9b42a13dc56 : FAILURE in 1h 06m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f7847d1ae4764c909c9ef55882ccb2d9 : SUCCESS in 1h 15m 51s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c7e59e87722b4826a54c733cee77d3e7 : SUCCESS in 3h 23m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e95633007bc0419eadadd35b89b4ebb7 : SUCCESS in 1h 25m 17s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56c4788008f94bfe8d4f8be4aa269166 : SUCCESS in 1h 38m 04s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dc7abcf18db4406ea3f5b450b3d683c3 : SUCCESS in 2h 05m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb63ce46b2474fafa7cba2a0e709d01d : TIMED_OUT in 3h 03m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/55b2860ee40843219c27f71aa62b1ab2 : FAILURE in 1h 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"520a797e2ffc921018fffc081278732f16b72e5c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-21 16:19:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e3fec1c4ece588f612ac4250c61e71c7a2ac4193","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-22 15:22:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"822163a33bb72245c9e8a8ea34c020ecd768de6a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-25 11:45:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"86322b348a49c284c2f6a121e25ed3e39a11c4e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 11:46:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7191e65f096bcad69c2eeecc5e9fb5c4961df090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:12:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8d1bc1a5c5624b01a7c274af1da770e2 : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/258b24b9102a460c9946325a7700ca25 : SUCCESS in 1h 52m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/10d67a8bf31b4819813203028ce4f270 : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1d4b792bbd8f4ca19291d628fa67609e : SUCCESS in 2h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b034dc61e89f4e63b8d58f2a3aab7c51 : SUCCESS in 1h 25m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6375dda162cf4216aa93da38fb171763 : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f206b1b96504a02b14d8584846d2bfc : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fef7e256f524289bbf0a2ae424896a2 : SUCCESS in 26m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/153c065da1c94db8999eb6742a386a0a : SUCCESS in 23m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658a08a55f3f49bbbbd4ac42720b99d5 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a7620a3924f4040b4c84e67f66bbd49 : SUCCESS in 1h 10m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b42fab7142c844c9b9d0af335f6d38f8 : SUCCESS in 1h 13m 27s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/16383b5c88c946059ed495532142ae75 : SUCCESS in 1h 21m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae17c392c9f34e888bfe793c0ac2bb2d : SUCCESS in 1h 02m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a36ea301712f441cbfcf67d58287bdd3 : SUCCESS in 1h 12m 39s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c87c406860624223aa2c27a81dde7777 : SUCCESS in 1h 18m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/077ce21d16234156bb822b4fce27f06b : SUCCESS in 1h 13m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9f998363f1c46819d23ed6563d2c35c : SUCCESS in 1h 07m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c5816cb4f6a64afe9a7b2ade486496b6 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":4},{"id":"b4011599b1f28e9603f5f2e0c70eaf83af50f314","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:12:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"aff49fe60ff9e960a86451a79e2ebc6d62d44637","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 14:14:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdf96502949462cb1ae096efe9ae667 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e302cba68a13a3b92180c37ac9331f9855e921f1","revisions":{"386f2fa0bcd320f25e34b131df2c3cdddc856be5":{"kind":"REWORK","_number":1,"created":"2020-12-03 23:04:21.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/94/765394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/765394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/765394/1"}}},"commit":{"parents":[{"commit":"efefd73d4228c27b40b07bbe68e504a8e9378503","subject":"Implement secure RBAC for the segment API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efefd73d4228c27b40b07bbe68e504a8e9378503"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 21:39:46.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 21:42:13.000000000","tz":0},"subject":"Implement secure RBAC for security group API","message":"Implement secure RBAC for security group API\n\nThis commit updates the security group policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I0c2062619a72184e6681530de6c074fb3c23635f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/386f2fa0bcd320f25e34b131df2c3cdddc856be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/386f2fa0bcd320f25e34b131df2c3cdddc856be5"}]},"branch":"refs/heads/master"},"5e982d4e218159edee16bd7c42240f583ac86c14":{"kind":"REWORK","_number":2,"created":"2020-12-08 10:00:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/765394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/765394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/765394/2"}}},"commit":{"parents":[{"commit":"8fe0e3cba8645720e334902ea924d2b95b482db9","subject":"Implement secure RBAC for the segment API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8fe0e3cba8645720e334902ea924d2b95b482db9"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 21:39:46.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-08 09:59:33.000000000","tz":60},"subject":"Implement secure RBAC for security group API","message":"Implement secure RBAC for security group API\n\nThis commit updates the security group policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I0c2062619a72184e6681530de6c074fb3c23635f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e982d4e218159edee16bd7c42240f583ac86c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e982d4e218159edee16bd7c42240f583ac86c14"}]},"branch":"refs/heads/master"},"7b2b52c62d19040871f0baff6c45b7b699528192":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-15 16:14:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/765394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/765394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/765394/3"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 21:39:46.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-15 16:14:36.000000000","tz":0},"subject":"Implement secure RBAC for security group API","message":"Implement secure RBAC for security group API\n\nThis commit updates the security group policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I0c2062619a72184e6681530de6c074fb3c23635f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b2b52c62d19040871f0baff6c45b7b699528192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b2b52c62d19040871f0baff6c45b7b699528192"}]},"branch":"refs/heads/master","description":"Rebase"},"e302cba68a13a3b92180c37ac9331f9855e921f1":{"kind":"REWORK","_number":4,"created":"2020-12-17 12:08:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/765394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/765394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/765394/4"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 21:39:46.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-17 12:07:53.000000000","tz":60},"subject":"Implement secure RBAC for security group API","message":"Implement secure RBAC for security group API\n\nThis commit updates the security group policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I0c2062619a72184e6681530de6c074fb3c23635f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e302cba68a13a3b92180c37ac9331f9855e921f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e302cba68a13a3b92180c37ac9331f9855e921f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
