)]}'
{"id":"openstack%2Fneutron~765416","triplet_id":"openstack%2Fneutron~master~I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45","subject":"Implement secure RBAC for RBAC policy API","status":"MERGED","created":"2020-12-03 23:04:21.000000000","updated":"2021-03-08 13:53:02.000000000","submitted":"2021-03-08 13:45:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765416-secure-rbac","meta_rev_id":"383b8926a29df47b1a901c1c59213d194a99ed95","_number":765416,"virtual_id_number":765416,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-02-19 14:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 13:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-08 08:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-05 16:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-08 08:36:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 07:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 08:09:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 13:47:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-15 16:12:48.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-25 22:50:45.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 14:51:13.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 16:18:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-08 08:36:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"dbf4a34a7b4459222b63b6ee6adbd681a98fef68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-03 23:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b9089945c88cb987a9cf522c053e8f0cf371626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:36:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e4e32c346c245cbbc071cfcae88bf47 : SUCCESS in 1h 07m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/506c2d4fe4d346c3a258102a96a724ee : SUCCESS in 46m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0725856555894a2ba5f13a759051e08c : SUCCESS in 1h 28m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cb8a3ba1f3aa45ad9ca72ca0479ab4e4 : SUCCESS in 1h 54m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62aa09fe4b8e420c95a739b5aa0e6740 : SUCCESS in 1h 48m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6a085a98c2b84571a53ceac6f7d37da4 : SUCCESS in 1h 17m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b06f41a57a446b83a4186c38713053 : SUCCESS in 40m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/562a5e65d9d142a6b1621bb6912a99e5 : FAILURE in 53m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aaf26ebd6b74a9f83de0879f2a19a47 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4da25e7bc6d47aa8a008c432cb893be : FAILURE in 50m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc7ccea645014114ac50396dd4e6af9d : SUCCESS in 56m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa903ee7b6834f4388646516e1f62efe : SUCCESS in 8m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21ec7c72c6a74102b10b8d70250303a6 : SUCCESS in 1h 08m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebfb1913845b40d89edd6d7d049291bc : SUCCESS in 1h 17m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2d27aee183b84898b4401e5745d279ab : SUCCESS in 1h 47m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9795ebefe3e1454a814ce67c5e446e08 : SUCCESS in 57m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/441306b1513a4eba9bf36aa2ed8f254c : SUCCESS in 59m 30s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/64155f30dafb45738e19860a07ef4d3f : FAILURE in 1h 44m 27s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/45c3f43a19bc4564a2cf85938056a859 : SUCCESS in 1h 20m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/08ad1a16c9774bb79e64f803ff0b10ac : SUCCESS in 1h 21m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ac56eccf70d44d5993f9e7ce31c156b4 : SUCCESS in 1h 37m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/65d3b7a9474540e2b14ceb9fedb20bb6 : SUCCESS in 1h 16m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4905056512b3412690ee859e0ac5d913 : SUCCESS in 2h 11m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d3b57911c8943ddb1642b2d8c266b9f : SUCCESS in 1h 16m 52s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e63ad51f9bb74451a6fe39d94bf50541 : SUCCESS in 1h 06m 53s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a3b25183b8d4b78841e55ab34256bc5 : SUCCESS in 1h 16m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/207441e875a444179d61f8600a9069f4 : SUCCESS in 44m 25s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/45ff38f34d91452292760358668e3768 : FAILURE in 1h 09m 45s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/67b46597d1934b0aa2eaa2ea413bc67c : SUCCESS in 1h 49m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e01405d5f5d640f9af8eed8e71ee5467 : SUCCESS in 3h 15m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/186f3d7ed74f4f5397f5e8798f3d9a23 : SUCCESS in 1h 12m 29s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c772e7480674e0ba68ced403f849cb2 : SUCCESS in 1h 51m 31s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/141be116451144198ce3bed17045e1a8 : FAILURE in 2h 07m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/74be82ac66814d1d844bdb861cc77f51 : TIMED_OUT in 3h 03m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/04bf644b9943474d90e796f111eac9e4 : FAILURE in 1h 49m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65cef296302f8402dd2add2b5c5fa34948a55152","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-04 08:09:00.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/16/765416/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"be64d48d7438f113c9b25331e22d3d21649e0690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-15 13:47:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c0685684965cd8f071dbb55d7c7be528270a3c6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 16:12:48.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765416/2 : FAILURE in 53m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d99885795ccdd5e004b98a1f7114e593ed433187","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-15 20:14:03.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/16/765416/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3b05e6629a96824080ad94d8cdc67f983c86f149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 23:57:12.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/28c7fe1102df44fc842c6154e97f7eb0 : FAILURE in 58m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7651aace77fa4331a9e0ba12f639fd0d : FAILURE in 37m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3f5f5845762a4ebb96aff58a6b2528f9 : FAILURE in 34m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68c071ab7f0d4ef487873ed46b92ecc1 : FAILURE in 46m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d407b5fe68494205aee76526aba98be6 : FAILURE in 36m 58s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/760a046e792a4b5e8fbabd7ee9527b2c : FAILURE in 37m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f755a7d9ea424aaa8d8cd9f3e924e340 : FAILURE in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78977e7848d34bfbbfbb0f034af8ba3f : FAILURE in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c5e2ca5187469caa1bf150cf6f389b : FAILURE in 10m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3076615a617464485393b3b576724cb : FAILURE in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67bde4d467444b05b2e66d5270b125f5 : FAILURE in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed46ce5597f34d72a57e7059b1cfeb82 : FAILURE in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af81f42b714461bb85bc684b5855ec8 : FAILURE in 5m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3100cfa3e4fb4f459b82c50ff9973b9d : FAILURE in 25m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a45a35f1a04443749914a8b865543463 : TIMED_OUT in 2h 14m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d893a944d4a149d7924990619c0a79da : FAILURE in 25m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61e3d7a5c4bb4fd095a3115b9643f209 : FAILURE in 54m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4f786028c1114680a61bb65f30ee6de6 : FAILURE in 1h 04m 24s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7e0995d311d34618a9accbe93d32125e : FAILURE in 58m 44s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c7c7f146d23e4ec8a500fccc39d58c9f : FAILURE in 57m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/49e0fe20ba57426a89642f41a5c328e9 : FAILURE in 49m 43s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a514b4b72ad9474d96c272ce7e0cb7cd : FAILURE in 48m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b7b834dba444e3fada69016600a4706 : FAILURE in 55m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d965f53bfbdd43bd9dddacba2644cc49 : FAILURE in 43m 03s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d44629621aa3429191a4124715ba2530 : FAILURE in 1h 00m 11s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e898430447504c8490c526a171b6e44c : FAILURE in 36m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bcd4600425194598bf2d42a4a8948e3a : FAILURE in 38m 41s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/3753a80a08704d8ea88d8eb2c64f4bac : FAILURE in 44m 41s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e51fa5b3eea84f7bb0ff8cc52d1628da : FAILURE in 1h 17m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f316cc1496bf44bfa471b96ada3b40a0 : FAILURE in 39m 01s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79979baba44e46989674ef5a8438e1c3 : FAILURE in 35m 34s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6a4994f85d1842b08bb29ab3354f1543 : FAILURE in 58m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/16f57f068fc44e8da766a5193f356d50 : FAILURE in 59m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f713062835bbddb094ffd75d95d62f47af24fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-17 11:53:30.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e28a1a19db3c8fe45e7ef7ff04aea5eb1bb49648","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-17 13:23:21.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765416/3 : FAILURE in 53m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2b3a58c9ba1d9d3f9cae4f849012940b0fc44492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 19:04:48.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b3f089e342f495881f9ec0972e93089 : FAILURE in 38m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c951d3ba175d4b9e9b2a9a3d48a81dd3 : FAILURE in 27m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0fe11b2ee4484a06a6b497743caa6200 : FAILURE in 28m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/430b3f6a7e384db08a48ae69c9e9bfdd : FAILURE in 33m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/006d17f6e5d244dfa702ca680df10ed6 : FAILURE in 30m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/386b38354e5a41adad1bee27b2f2bf69 : FAILURE in 30m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa35df8eda794d56ac12b9f73bcbaf92 : FAILURE in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ea472b2dab24a258430f83fccb69c4f : FAILURE in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f2d55df94849c1b9f3e55bdc9a7b06 : FAILURE in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/060b4f95362c4ad9be2f2f17ad3b1afd : FAILURE in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d54a1b1e8b2a4fcfba1b8d713fc5c31e : FAILURE in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/415d5a9ab0c44f06bcbb1a57cbd5a586 : FAILURE in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872ef0be9c5646d3a53717085ed31bac : FAILURE in 5m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/852a8846b9694559933055ccfc2cc988 : FAILURE in 16m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a15e0caa740142978ca24557873c8550 : SUCCESS in 1h 16m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3e0bc9b03b924c889152ed7a11bd2a39 : FAILURE in 29m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab24236ddb694d21a8f93a6b6dee134c : FAILURE in 55m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/117e8df8ef0b43078235948f047441db : FAILURE in 1h 04m 29s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/46697ee2eff84fb7a596527853c38025 : FAILURE in 51m 41s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/daa0a413996e48e99cf41ef6bc705114 : FAILURE in 42m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/36530369d7f34036a053bf3acaf7b914 : FAILURE in 47m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/255da239fbec4309a1b71187933e1145 : FAILURE in 38m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87941ff7481d41cb8a3460555391bbd1 : FAILURE in 53m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/943339adb12f40f286cfd9616bf2365e : FAILURE in 40m 02s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/96025316e8684f15866dc0e20168514f : FAILURE in 36m 56s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/576a7baaa50940abbfadbd0113cee7c8 : FAILURE in 36m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/77a49a232ab848079dea0f4aeeb501ff : FAILURE in 29m 11s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/b59efa7f35994324aed1f05b8c24ca85 : FAILURE in 40m 55s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bdc3a4f5c5cf41fcb0a2592136221b6f : FAILURE in 1h 12m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c025c839b35848eaa45b4e81f0004ea7 : FAILURE in 29m 22s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b567283c72d1440fbcfc3344751a7b61 : FAILURE in 33m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d0f411e6aebd4926a176cb72a4cf40e3 : FAILURE in 1h 19m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5574386234004af1aa3178e225eb21e0 : FAILURE in 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40a7f5f659bf16ff16e74926e06daf9aa7030e43","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-17 19:48:47.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/16/765416/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"de18710293c6869fe78e12c993879aab6c5212bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-21 16:31:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bcf5ebb7c3d39bd32f967ebfbe7fb27d1ccb6046","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:51:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ed49c82a09f41f2b47e05584b228713 : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d899c202be6845fa8e41b978d65f0ce2 : TIMED_OUT in 1h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f273e2ec6ae7e6f7013b66ace34c9ac97fd75c6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-21 21:13:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765416/4 : SUCCESS in 1h 48m 57s","accounts_in_message":[],"_revision_number":4},{"id":"46474bd293322c0681979a1d83c3d18655687940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:32:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74bcc270d8184808896bc69c3fc316e6 : SUCCESS in 49m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdea70bbeb4c44bbb63a9ea95013e460 : SUCCESS in 34m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ed5c57b832bf431b83238f9a2a28e7a1 : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/71a6fa05326c476e9f0ec8b875f84b26 : SUCCESS in 1h 27m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2f10254035c14ab98381118c71ec2971 : SUCCESS in 1h 14m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/445633ed6a6145c69f1c15c3f814eeda : SUCCESS in 2h 43m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea76ffe45f444522b0c8f2d15f8fd861 : SUCCESS in 33m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67979cd7ef244a6b904939a7e4942814 : SUCCESS in 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8867ceb5568b4c0fa53fdb8863b1c26c : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3bf18f07d514cdfa2be1682532d1d60 : SUCCESS in 29m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2781d21ae5941398875f9ad0ef9a0e1 : SUCCESS in 27m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/527c84ce236e48ed9679506d8efe681f : SUCCESS in 31m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e42c249cd044eeb4d35efc05303173 : SUCCESS in 8m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/477cc44592cc45d48040cbe0492fb273 : FAILURE in 1h 09m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cfea3c32b76433b8630cde857b14dba : SUCCESS in 1h 16m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c3e8dd5cec334f499f6ceb46fe44fbfd : SUCCESS in 1h 41m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0e9b1d1141b40188f2f04b3b97128e1 : SUCCESS in 54m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c41e23fe67c47268fe1c0ea8d8057da : SUCCESS in 53m 19s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f86dc4e43ce347589f04dfc3109795cf : SUCCESS in 1h 21m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8e418ca761d43bd814cf75ad3181164 : FAILURE in 1h 17m 12s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b00452612f446da8d74e58f8f94f846 : SUCCESS in 1h 46m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3172d9596db40218613ada703dc7535 : SUCCESS in 59m 37s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/79def347dd644a7897c2e046857f7220 : SUCCESS in 1h 05m 10s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3ed3655b0f7b46de8f8b3c5a6d4a2b77 : SUCCESS in 1h 06m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/952045051ab54d8ab74e707df3256c0a : SUCCESS in 34m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d8fb12b7735e4a43a72b96244587681f : SUCCESS in 1h 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4cee66398002d733fccd054af6f18b44924e839e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-22 08:34:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"abb11dc623cd77f5eaf16ba67bc9ce9d1faac6c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 09:47:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93a8cc4c7b554fde91c67e4fb9393594 : SUCCESS in 59m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1bb8443ae8b544bea185eb48839c7537 : TIMED_OUT in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4545690fa6103f55707bf4b10a2dc82e9453eeb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-22 10:31:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765416/4 : SUCCESS in 1h 55m 41s","accounts_in_message":[],"_revision_number":4},{"id":"78078d1ac626aea36935040e362abc0db5e2fbaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:51:10.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b2031a16991498fa27e54ac3e13220f : SUCCESS in 40m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76f88160cba64d4cb85a3b42bcd6febb : SUCCESS in 44m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b9d478f106d84d48aa0153fe21a06134 : SUCCESS in 1h 24m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a41a35c39aae4cbb8d50ded2e8e7a362 : SUCCESS in 1h 47m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f16de97223e7478c9c5994ee57a0e70d : SUCCESS in 1h 17m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/473e8dfe1b854d0eb7884d73180e8c04 : RETRY_LIMIT in 19m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bdb72c09c5c4547bc53a4534c998c90 : SUCCESS in 30m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04c98c2e1de14bf1802fad7ccb85a90c : SUCCESS in 23m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b826050c99b437aaeb62ddc7bf8c528 : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aad9289fbf549048b5d768e3b5718a4 : SUCCESS in 33m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43e3f13a01d0432787f237fc84e5a3ff : SUCCESS in 23m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f95306d2d2d5484a8497460453279cf0 : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a74697608a4ebe8989694e760ff720 : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e79e3143bc24af99ae6309c2d009a25 : SUCCESS in 1h 05m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc41676f3aba46188e19a176eaaafe3b : SUCCESS in 1h 14m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3479137b12c14be08d6ae2cba673f353 : SUCCESS in 1h 33m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8b24c7c07df4e01843ea0edcd82f34b : SUCCESS in 1h 20m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de74d2b577e8472eb2a948063b2fab5d : SUCCESS in 1h 07m 28s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/379cc858c40c48bdb532901adb137d19 : SUCCESS in 1h 17m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4608ff36cd674da6a7156c52d708e671 : SUCCESS in 1h 32m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc5ec034fd3d4770bf227b43c6a86926 : SUCCESS in 1h 40m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6db7f2c985e64a5d93e1bd563a856082 : SUCCESS in 1h 20m 11s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4c414a97cbbf41b6a0a33061d5cef94c : SUCCESS in 1h 03m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1faf230d19044f87998bc73002cf4965 : SUCCESS in 1h 05m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ace68ce159c947538c5ece33d06f5339 : SUCCESS in 36m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43c3eba29c7e44cf906752c8e80825fc : SUCCESS in 1h 48m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b0d90a0bfb5d8b9b2c27430ade6798f826cf980","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-25 17:53:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9aa22819d116e1ea97250584b94d0d06e95d9cec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 19:40:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02b93ac3a8b356cbd99239e6c82486b7bee3291e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-25 20:04:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765416/4 : SUCCESS in 2h 08m 43s","accounts_in_message":[],"_revision_number":4},{"id":"13b9469efa00bdc4c6d2a607f2f61fca0c9f9c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:38:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/48b2b1e8ab364addae31172cb2d667f2 : SUCCESS in 53m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/012c5f05498540478ec7b02fd5d1d33c : SUCCESS in 27m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/044aa575fd5f47d7a347ba239890b9a5 : SUCCESS in 1h 20m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3dc410c96f6345bcaa1a0a7657d5264c : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6ec8c7b69c64c8fa3f0a1bdc5fe8273 : SUCCESS in 2h 13m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/904c54bfe1d54399882926dcb19e8d01 : SUCCESS in 1h 50m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e08a53ae37534f3ab233f71fccc2816f : SUCCESS in 33m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fc6d34ed82a4374a978b83d206dad2a : SUCCESS in 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336bbbc8628745719d2d047499da4922 : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af4b8660c5b47df9d1479855cb304c4 : SUCCESS in 26m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6af940d43b040e7ae51fa2874891a0e : SUCCESS in 29m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72eac411b2cd476c88f89ca7c05b7eb9 : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9842a67d8e64483a827ee14bd500246 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/058be1a6f89444bcbd46a8859e278426 : SUCCESS in 1h 09m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc6d631ecf864047aab8574f7778346d : SUCCESS in 1h 12m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e310a02ba1954c1788e966ec2e790407 : SUCCESS in 1h 37m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2d20c1d94d64b3aa560e6d5a6ff7d72 : SUCCESS in 58m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/948ab5adc54b4a90b2fe81a6431e2636 : SUCCESS in 1h 02m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4ad4a9979f094ddfb05a04da72437078 : SUCCESS in 1h 16m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a07b7dda38ad4ac0a10e7a3ae03c9900 : SUCCESS in 1h 30m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0255fca0258d4a47b6e181d5535cb6da : SUCCESS in 1h 42m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/712b64387bdd44cdada3fc51fa7f72e6 : SUCCESS in 1h 08m 40s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ff20ed1cee5e42f0a6bb0e82b71168ba : SUCCESS in 1h 01m 06s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4e2f97622f6a47fb8ecc087abc334b85 : SUCCESS in 1h 11m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/644247ddeed5434fbb6cb16e7d25c1c4 : SUCCESS in 36m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5059e97a54a147ab99f8a7e836842564 : SUCCESS in 1h 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f52e238c5c3c44ef0036e54d059483872008b1a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-25 22:50:45.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/16/765416/4/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"58ba13faa2deeec0cbaebd6cbf5f6f2507f0bddb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 15:49:20.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/16/765416/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"41fd66503324e28bb0200b72ea5e1107cf878ae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 12:29:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f5aa3552e2a49e8e75fb987b68150b2af7b2d723","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 14:35:45.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765416/5 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":5},{"id":"1baa723fc966d26b172f03809e7d2da41472c2e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:02:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdebbe0bed23441b8f88aa615a0997d0 : TIMED_OUT in 1h 01m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be5dea7771a243f394d62b44856597ce : SUCCESS in 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"450053921ed1113e8f1e4eb372d3ad2c862ea886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:12:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/462cf513ac4546ed97756a0e8f2afa01 : SUCCESS in 55m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3e496cec9f9f4e9882b495856ac4ccfe : SUCCESS in 1h 28m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c6bd7af822c2497db2157ddfdf02f77c : SUCCESS in 1h 40m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0951de7da234b1db93481796508ef29 : SUCCESS in 1h 31m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1293d0c4a0114449912013b66d242c2f : SUCCESS in 2h 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39f84de6908949f6ad5d2b9203b55218 : SUCCESS in 45m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca3a12a9edc84c69ad06ae196aa85cda : SUCCESS in 33m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07afddb903e246b99d3b1eaaf4ba3d70 : SUCCESS in 14m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83bcb2e998864bacbb6768c6a4fdffdd : SUCCESS in 28m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68a057b9de7d44388af53c8785eb649e : SUCCESS in 30m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4823fccf168a4bc4866263f141297951 : SUCCESS in 27m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04eef0ce5e7c4c0e81c2bf80c4efa6a2 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af687a57dc09467da96f732bc3b85f9b : SUCCESS in 42m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71235a4c8fc54c9aa1fdc3fc6197f556 : SUCCESS in 1h 06m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a1c3c9810f484848a3a5ef5cb85661ee : SUCCESS in 1h 29m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d177f45342d0473dadf96fd532b45f22 : SUCCESS in 59m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dcd64f6b8e994a878402c6c4945ca39a : SUCCESS in 55m 59s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8fbfa3539b7a4aa9a34bafdf6cd26b8e : SUCCESS in 1h 39m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b0878ff038a84bdc9a9f76279765a6cd : FAILURE in 1h 21m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7784f2875ae24f6ca0937fe0098362b2 : SUCCESS in 1h 38m 22s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb9ca7669aea4a178f556544547061f3 : SUCCESS in 58m 08s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/169b8ed426f841a696c1a78ef79f1078 : SUCCESS in 56m 01s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1882f3b1e8a64005a6db1aec0f03d225 : SUCCESS in 1h 08m 46s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e9fc6766eb324c4fb4eba8df1c10ae59 : SUCCESS in 33m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d27f3cbff6e14d808eb16ef17a549107 : SUCCESS in 1h 32m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d0f7593e16374de094ca603fefab518c : SUCCESS in 1h 27m 07s","accounts_in_message":[],"_revision_number":5},{"id":"8bb060fc2863dc08a92aefbe6649f50562604c29","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-19 14:51:13.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/16/765416/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"fc8bf716b6aa0a6186685d8277a6916af0c17257","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-05 16:18:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0f8efb3d53fd8209646e58d823ed93ae09c830b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-08 08:36:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d9a068a6fbdf1d9470d699e2d06b7b8e36e9db97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 08:36:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ca6d080b7e8cdd35236b1324797abbc03c4afb53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 13:45:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4a45bd263a9747ba89454bd29c8862a8 : SUCCESS in 53m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/70507bac853f421baafb25cf0d7bca19 : SUCCESS in 1h 51m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6bb57e9e3dcb4db1aac19002a99aaaec : SUCCESS in 1h 28m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f99ac66e992347a1a526b4af516fdb89 : SUCCESS in 1h 56m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7c7b95af941d4b35a947ac7782b93351 : SUCCESS in 1h 38m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98357def5bf84f5b98146b2dc1c05bc2 : SUCCESS in 31m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11762175ac54819a203f34c3c1f8c3f : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a39bbbb15e0241fc8dc94a059fe26da2 : SUCCESS in 46m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/445d4f48b9214c93b388f20c126ca4f4 : SUCCESS in 34m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74caee3f9bf04ddf8eaa09d6e39869ff : SUCCESS in 10m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2fd804d9f594c35becc021651e632f4 : SUCCESS in 52m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c668d73dd4ea427191a648578b45d521 : SUCCESS in 1h 24m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c2bb1a81d024047b8e9800a0fe0e6ff : SUCCESS in 1h 29m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fd43db566854d72945a3766769e1410 : SUCCESS in 1h 04m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2692cfaa0dfc4d759d11374599af0fb1 : SUCCESS in 1h 01m 00s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4ab06e64cb04c1e9f415f6fb0fdc906 : SUCCESS in 1h 36m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd7a9874901b48e8b1bb6698e1e78f9a : SUCCESS in 51m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/edfad345145c4ae699444c891e65b3e9 : SUCCESS in 57m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b43a73508fb646c4bcf4ff0e76727563 : SUCCESS in 29m 04s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/55bc6511fc6c4673b7e6b3c38529538c : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":5},{"id":"03feb8d3732e2fad4ff73aacd7fd9d35f9f4bad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 13:45:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"383b8926a29df47b1a901c1c59213d194a99ed95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 13:53:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d956990aeb1d4ad583e82c76da79c9f1 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2638c8a561af6cd495c1ba9abfa4dff58f5a9e48","revisions":{"61cf7e66df713c3262404d125752f7aa23046136":{"kind":"REWORK","_number":1,"created":"2020-12-03 23:04:21.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/16/765416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/765416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/765416/1"}}},"commit":{"parents":[{"commit":"b821b6f66e086e8c1ad18a63695fce6f0faf05c9","subject":"Implement secure RBAC for the router API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b821b6f66e086e8c1ad18a63695fce6f0faf05c9"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 22:27:31.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 22:27:31.000000000","tz":0},"subject":"Implement secure RBAC for RBAC policy API","message":"Implement secure RBAC for RBAC policy API\n\nThis commit updates the RBAC resource policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61cf7e66df713c3262404d125752f7aa23046136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61cf7e66df713c3262404d125752f7aa23046136"}]},"branch":"refs/heads/master"},"59f4d515ad48664cf9f642eea35f325bab8eea1b":{"kind":"REWORK","_number":2,"created":"2020-12-15 13:47:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/765416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/765416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/765416/2"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 22:27:31.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-15 13:46:44.000000000","tz":60},"subject":"Implement secure RBAC for RBAC policy API","message":"Implement secure RBAC for RBAC policy API\n\nThis commit updates the RBAC resource policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59f4d515ad48664cf9f642eea35f325bab8eea1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59f4d515ad48664cf9f642eea35f325bab8eea1b"}]},"branch":"refs/heads/master"},"4147965afba5d507bbb375b9d5f00f0b2d65a344":{"kind":"REWORK","_number":3,"created":"2020-12-17 11:53:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/765416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/765416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/765416/3"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 22:27:31.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-17 11:53:13.000000000","tz":0},"subject":"Implement secure RBAC for RBAC policy API","message":"Implement secure RBAC for RBAC policy API\n\nThis commit updates the RBAC resource policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4147965afba5d507bbb375b9d5f00f0b2d65a344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4147965afba5d507bbb375b9d5f00f0b2d65a344"}]},"branch":"refs/heads/master"},"e1b85af555e808b4a5a6443c8f010fe8a3cf9095":{"kind":"REWORK","_number":4,"created":"2021-01-21 16:31:10.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/765416/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/765416/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/765416/4"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 22:27:31.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-21 16:30:40.000000000","tz":60},"subject":"Implement secure RBAC for RBAC policy API","message":"Implement secure RBAC for RBAC policy API\n\nThis commit updates the RBAC resource policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1b85af555e808b4a5a6443c8f010fe8a3cf9095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1b85af555e808b4a5a6443c8f010fe8a3cf9095"}]},"branch":"refs/heads/master"},"2638c8a561af6cd495c1ba9abfa4dff58f5a9e48":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-18 12:29:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/765416/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/765416/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/765416/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/765416/5"}}},"commit":{"parents":[{"commit":"90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13","subject":"Merge \"Bump neutron-lib to 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-03 22:27:31.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 12:29:17.000000000","tz":0},"subject":"Implement secure RBAC for RBAC policy API","message":"Implement secure RBAC for RBAC policy API\n\nThis commit updates the RBAC resource policies to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I3eb8c0969e1db867fcb2f148c2feb6f69fc75a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2638c8a561af6cd495c1ba9abfa4dff58f5a9e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2638c8a561af6cd495c1ba9abfa4dff58f5a9e48"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
