)]}'
{"id":"openstack%2Fneutron~765628","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I8c940ac8f7632c69607dea7220146ef59d55ed56","project":"openstack/neutron","branch":"stable/ussuri","topic":"fix_ovsdb_probe_interval-stable/ussuri","hashtags":[],"change_id":"I8c940ac8f7632c69607dea7220146ef59d55ed56","subject":"Ensure ovsdb_probe_interval set before connect()","status":"MERGED","created":"2020-12-04 21:53:46.000000000","updated":"2021-01-05 12:08:26.000000000","submitted":"2021-01-05 12:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765628-fix_ovsdb_probe_interval-stable/ussuri","meta_rev_id":"174706a0597859b9a4e35e120ece036b4e4942a3","_number":765628,"virtual_id_number":765628,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 12:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-01-05 08:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-12-09 10:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-01-05 08:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-05 08:11:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-09 10:04:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-05 08:51:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"af4632ba45ad9ede0f560e5d03a456d24c11fded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-04 21:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bde56f6b115604ba7e2862883500175c6df91019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 08:11:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/8e0c7f0ea493422d99bcfeb62feaaf98 : SUCCESS in 56m 04s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/57600fe0834043558b0d9f106509553a : SUCCESS in 31m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/95539c620be045d1807e332c8e3c97f1 : SUCCESS in 1h 41m 03s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/79ec0a5e453148ea8fda6e8d12e1d5d7 : SUCCESS in 1h 42m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/05247bae60cf45fc8b03c06d7f287be9 : SUCCESS in 1h 43m 00s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/9f2b994a04744c829c33eb20d7222dfa : SUCCESS in 1h 54m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0d4fdf049a3422eb180d23113e70b06 : SUCCESS in 31m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e058a239f7a4cecad750ef9a8abf922 : SUCCESS in 30m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f20e248cc94ef2a9b603f13533ee6f : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/440fa5f91c6e44819dc6f3d58e6be011 : SUCCESS in 24m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c52bf3a033814f1cba7726fca529c654 : SUCCESS in 32m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26cd253b4a0346b18d0f135a016cc7af : SUCCESS in 27m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907a4e8370ed492da390a0f5b5e9e54c : SUCCESS in 12m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/79dc7f48ba064455888c55c3385c3b89 : SUCCESS in 58m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/06967400d80145859c94a584b7f3e363 : SUCCESS in 1h 14m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d9d691828e2488d8576866c1d4866b6 : SUCCESS in 1h 15m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1ed9b306ecb84b8183781868d4235a5a : SUCCESS in 2h 01m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b740109b03b1411482f268d9d79be36f : SUCCESS in 1h 24m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b5440688758404f8f6c764420e02279 : SUCCESS in 1h 13m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dd18aeec84124db6ab96ae9cea136c3a : SUCCESS in 1h 19m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ce389424640488f8f23c779d1da257b : SUCCESS in 1h 20m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/acfcd70d63b9401989d3c78d106c1561 : SUCCESS in 1h 16m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b9c79fddf6b540bdacf1c04e415f2342 : FAILURE in 1h 20m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/260be974ac1b43eb868cab30ff73cab6 : FAILURE in 1h 39m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ff26b44d0f245708eabd10b37c6a646 : SUCCESS in 1h 19m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/42308eb8b1ea4d25a68670907eb1076a : SUCCESS in 1h 05m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c7cb992212f4dad95cd1964922a40b4 : SUCCESS in 2h 06m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8633d5a9d04f4e03ba6ac679df42c7ac : SUCCESS in 1h 13m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3095d2fe1ef0427b90b98f17e7267c7a : SUCCESS in 1h 12m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a2feaea6d87c458e8cad6b5bd058009d : SUCCESS in 37m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b11c930de20e45b7b9433aa1b1a3d4e8 : SUCCESS in 1h 18m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f25b774ca120491c8716ccf1d4df65d2 : SUCCESS in 52m 59s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2a0dadae69da4d27ae13b1325d7039e6 : SUCCESS in 3h 43m 45s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53d73b06d38f44d28414d7331eb0fb2d : SUCCESS in 1h 22m 08s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e1222c44ae1b477ba32c0d46ebbf7285 : SUCCESS in 1h 54m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/08dde7266d4245e1b0116f24712bcc8a : SUCCESS in 2h 12m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e84f8ce432d945aa9a91489de2f67a1d : FAILURE in 1h 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d40351e4090ac475daf07e8ad7f74624c45fedcc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-09 10:04:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d345780cc18e5ed38a95f0aa08fe6e6ee02e94eb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-05 08:51:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7ba1a1940affffb37b0f77090b1ffce46d3b1dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 08:54:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dee6308f681e9642c8b7ec25f9bc6cd3c1cd881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:06:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/7dd52f1427e8423e9d7c52cd4028967e : SUCCESS in 50m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d22baa7eeab4e93ac798e982f3413ff : SUCCESS in 28m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc9585124744d46844b3e9aca174729 : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e819ecf31dee4535aed69f6d3cb547ab : SUCCESS in 26m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68198392a93b4eab9b2b1c2c496fc163 : SUCCESS in 28m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0745738dd5431688dc884a2234dce1 : SUCCESS in 8m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8198d5c0cdd04f9eb8b2003f545c6737 : SUCCESS in 54m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4e9333768677438d936e055b46d58e31 : SUCCESS in 1h 17m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df34505954274b07bdd7e5b120a49e8e : SUCCESS in 1h 26m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c25905d36bca49bc9ef64a75f5b0b001 : SUCCESS in 1h 28m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa6dac2942cf49ce80fdf2b2e36f3abd : SUCCESS in 1h 05m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f85d9fd243b414793bb0b91ad598268 : SUCCESS in 1h 11m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/95ec14f821ef43c692c6852c2e356c00 : SUCCESS in 1h 21m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb56fb0059694138a2b9b302fb518bb8 : SUCCESS in 2h 02m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c979cf9ad2a6422796aeb809c1555d6b : SUCCESS in 1h 04m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ace21320d4ed4b81bf1dc9f7a2d5fada : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5e865e8f87d81470b401a3488b7832d2511f8ee7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:06:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"174706a0597859b9a4e35e120ece036b4e4942a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f24be27721642759a017507b7599b94 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65d3f79ce655ace0c8dacab2196fb019ea9d7f08","revisions":{"65d3f79ce655ace0c8dacab2196fb019ea9d7f08":{"kind":"REWORK","_number":1,"created":"2020-12-04 21:53:46.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/28/765628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/765628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/765628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/765628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/765628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/765628/1"}}},"commit":{"parents":[{"commit":"ac2276c39643048be5dd874374d8d6f6a1fd329b","subject":"Merge \"Rehome api tests for propagate_uplink_status\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac2276c39643048be5dd874374d8d6f6a1fd329b"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-11-25 19:55:32.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-04 21:43:59.000000000","tz":0},"subject":"Ensure ovsdb_probe_interval set before connect()","message":"Ensure ovsdb_probe_interval set before connect()\n\nSetting the ovsdb_probe_interval after Connection.start() is\ncalled means that the probe interval is not changed from\npython-ovs\u0027s default of 5s until after the initial copy of the\ndatabase is retrieved. On busy systems, this can time out and\ncause infinite reconnects.\n\nThis patch passes the probe_interval argument to the ovs.db.Idl\nclass so that it can be set as part of creating the jsonrpc\nSession.\n\nSome unit tests were removed and replaced with a functional test\nwhich ensures not just that set_probe_interval is called, but that\nthe value is actually set before the connection is established.\n\nConflicts:\n  neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovsdb_monitor.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_impl_idl_ovn.py\n\nCloses-bug: #1905611\nChange-Id: I8c940ac8f7632c69607dea7220146ef59d55ed56\n(cherry picked from commit 5783e952887b2fe5e8aaa32b3077e8dcdefa7252)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65d3f79ce655ace0c8dacab2196fb019ea9d7f08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65d3f79ce655ace0c8dacab2196fb019ea9d7f08"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
