)]}'
{"id":"openstack%2Fneutron~765840","triplet_id":"openstack%2Fneutron~master~Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b","project":"openstack/neutron","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b","subject":"Implement secure RBAC for the port API","status":"MERGED","created":"2020-12-07 18:03:49.000000000","updated":"2021-03-11 03:21:37.000000000","submitted":"2021-03-08 19:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"765840-secure-rbac","meta_rev_id":"e56d62423f89b7acf7e708e6319945028e3f6303","_number":765840,"virtual_id_number":765840,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 19:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-03-08 08:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-03-08 07:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-03-08 08:59:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 21:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 12:54:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 15:54:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-15 17:11:17.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-04 12:42:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 08:16:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-03-08 07:51:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-08 08:59:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"0cb7c5c8463d62e7ad51d310454553f953a31140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-07 18:03:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d18956150f61a513808697b6df6d7d130dab381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 21:39:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/756ba89e1ce343ada4fd72db92e26c79 : FAILURE in 1h 04m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a0aceb9bc604e9984d3d845a2105b27 : SUCCESS in 43m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a5451c432b04fb1a39da23e6e7ae99f : SUCCESS in 1h 12m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/359d3a8c95374991a3e850d6de33e4d5 : SUCCESS in 1h 35m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f24e626ba6140ae98af56eac4550281 : SUCCESS in 1h 38m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a4a771e57aca4ff3bd64b46b34234334 : SUCCESS in 1h 58m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c979c1cd67924edabf8a7f286a3353e1 : FAILURE in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d56f2f66c1c42228b7f672c2af058da : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b46ad503da494b853e09a8993fc77f : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6390414e059405cab2feac223ed9861 : FAILURE in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2211cce372b94d5f8e33308bb34085dd : FAILURE in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49504f4888714a05bc28b476ccd24ac1 : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e52b09e9cbc8499080b5028521af91e8 : FAILURE in 14m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/866a7ebc0ac5417db20c17ce72188aef : FAILURE in 20m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7bb2b13cd100418f8cec628478cc0856 : FAILURE in 2h 05m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d59592a137504aed817320c0ef49185c : SUCCESS in 54m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d78e5a2604f4feda6669f39239e2c4e : SUCCESS in 1h 10m 55s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/f2bad8f1b31141678205de141707c2c2 : FAILURE in 1h 02m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa85ed69e2a74f78951af2a52c03be99 : SUCCESS in 1h 04m 15s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82a94ef5130d41c2a04671e2559c36b6 : SUCCESS in 1h 24m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d703211391ca4aff88cb62cda26ba501 : SUCCESS in 1h 30m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/575d619e1edb4ba397209d2ff19f0510 : SUCCESS in 1h 16m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/65dc4f58c89c43a48aef441961fb133d : SUCCESS in 2h 16m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c97eef5c8e79466e93812e39f5644db1 : SUCCESS in 1h 05m 49s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5e68de3ecc7b4189bf2c1a7e40ca241a : SUCCESS in 1h 02m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cff28cab213f4139871625d1b8f45470 : SUCCESS in 1h 08m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/75548aeb122f484aa010030b17280d10 : SUCCESS in 42m 58s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/0bdc3ef424e54ae8a2a3da48a630b2be : FAILURE in 1h 05m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b6d6fb6e3cd84a639e1e4f7d9816b86f : SUCCESS in 1h 25m 19s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/85b7c58dbc3f4f40aeaf4834bfa20d34 : SUCCESS in 3h 23m 54s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee0fe6590e2b4610bad6b73fb9f91df4 : SUCCESS in 1h 22m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1ef3fddf9ba14275a35fd908cde16af5 : FAILURE in 2h 02m 07s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a70122a0993144f18f9cfacd991276f5 : FAILURE in 1h 54m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a5880d9651d74b41aa5383ce5d2f002c : FAILURE in 2h 49m 53s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/657202b0e96540e9ad608a9d7af1006b : FAILURE in 2h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a7df779f04c566eba454472f81455ead6c48c44","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-08 12:54:55.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/765840/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a40c18b743854eb95e8855618f67930106cbb4c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-15 15:54:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a02bb1c6025e3cdfdcef5651a9cc4e0c7a800fd1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 17:11:17.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765840/2 : FAILURE in 52m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3dd780cffeb790ca3084f51079f3cf3547647567","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-15 22:42:13.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/765840/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5b984955df887d73e380e86d0080ba9f2b71a1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 01:46:34.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4074d70a55a24caeb7420ce4cb3ac793 : FAILURE in 33m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa0b22d362084be1a355f21e81a069fd : FAILURE in 38m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5775ec0a09574b4787933433fa4088ce : FAILURE in 34m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/363e9d0bba7b4e8cac00a37d1d93da9e : FAILURE in 33m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/942686538ef74170abed10e09b471818 : FAILURE in 29m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/21a5ea7e678d4c498013a65f8c8a92b9 : FAILURE in 31m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0defad1d67634ababfe13e58d5aa7454 : FAILURE in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b066fa3d1a940a4825c2fd552ce1ef8 : FAILURE in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff9bbf7febc46c28ee0fe3f69a5c774 : FAILURE in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef148e96ff0540de81c7401b5482a1ea : FAILURE in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4be07ebf21d4900be4fa5b7b1c6cc8a : FAILURE in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fbc72e628714cd7a5c35c18d2e3d48e : FAILURE in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299d38e5405d4646b1d4439c1727c0f6 : FAILURE in 6m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/587d7ecef69c4c9d988d774c339bfd4f : FAILURE in 24m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/45ec977317b745d68d54a364105b1f2c : SUCCESS in 1h 12m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/219bae37c72c4af484aa481905dc4ca3 : FAILURE in 26m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79465d7285464c1186c7a8ecbd598c9c : FAILURE in 57m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a313e1f3f27b4b0e80e4888c83b56ba8 : FAILURE in 1h 16m 59s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/3dba21f07c834c9aae7945a9286c5a99 : FAILURE in 56m 18s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5098ed6c3a574ceeb96a455eb4781387 : FAILURE in 43m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c9cf4f790554cc4891336de21250df6 : FAILURE in 57m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d0a04977f9614451852372829beb4e5a : FAILURE in 50m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e50b53930e64230a9369da3e36e3bc3 : FAILURE in 1h 04m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bde8638f57c41bf932c9737b2570c81 : FAILURE in 28m 49s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bc49794aad6241529e5681b9233bba49 : FAILURE in 37m 58s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/98c9fcca8ebe49e2962875f3e7993d11 : FAILURE in 39m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9e82c26d5176477da721c5d2e6958094 : FAILURE in 31m 38s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/0ce3eaaaefd844be9fd60ca01b4e6f11 : FAILURE in 49m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/69e2c62e79b7433696949d628d149fb4 : FAILURE in 1h 14m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e052fdb776174e3b8c659ed4b6673469 : FAILURE in 36m 06s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c542c5757ba48c487b8a92f910ab845 : FAILURE in 30m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3ffc015131cc407ab517b4f893a49acc : FAILURE in 46m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c01523e1122943aba12a2bc78730097d : FAILURE in 48m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d4a35ba348502b97ce25c276efebd5f7d74ba14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-17 13:04:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce0d620c9a2fb50c677cbd1736898d152516e8f9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-17 14:32:51.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765840/3 : FAILURE in 55m 21s","accounts_in_message":[],"_revision_number":3},{"id":"441594a929bbdde741beca86500becf698a92581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 21:03:40.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74dd61ea23184e628810ec2d011bcc12 : FAILURE in 30m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9319c0962d546aa8d0d95bf632b0a41 : FAILURE in 30m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dc89ae2c38a8446c92ac75ee0adf4901 : FAILURE in 35m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/be185af0d20846ffaa62be03e59a0056 : FAILURE in 27m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/868e5ea83bc94b28a0f0eb370492aa7c : FAILURE in 37m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b41a81293a684c74925aabaed8611151 : FAILURE in 31m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1836e0754e1c4b358b6c9cde3a4782e5 : FAILURE in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa846f524c1a4d6e8a9b05f1b5bb72a8 : FAILURE in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0b6710b6ed4e158b9f562ce0fcc3e1 : FAILURE in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f253453d411d494b817ff43fdc13e1d9 : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/961b84f4e8484cd3830414a3656af316 : FAILURE in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44ccba8069414523a315e3546331fc0e : FAILURE in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4d9c14c54b4372af497527fa49c14d : FAILURE in 5m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1203d08d31924e1f94a7d028eb4c90a3 : FAILURE in 22m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69eeccbaf51947feb4a0a6d763c110d9 : SUCCESS in 1h 01m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7c76a5b7d4624a728ddd44ec1d4c04dd : FAILURE in 26m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e67dc9b8ee5c410c8ab03a850477654b : FAILURE in 58m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/595c3d00fa36446a80a61694fef29301 : FAILURE in 58m 01s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/1a6d23a76ee7451ea1fb3c4a4c3fdbe8 : FAILURE in 1h 02m 00s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f623c9142a314088b44102a85999a1a6 : FAILURE in 55m 32s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4fa1fe2f3f794f7bbefeeacefbec0a80 : FAILURE in 47m 13s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/95b8699644b54e2ca7b4cb7072f91d0b : FAILURE in 39m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3bfacdd3de8342ca9af4fc354e2c864a : FAILURE in 47m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/721a058e188343379fb390f790da53a2 : FAILURE in 29m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e9d363632414456a9737d8f34fafecba : FAILURE in 40m 26s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c3f2d43b3f074f62bbd2807e9740e52e : FAILURE in 36m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ea0faf9cfc4d40809ce154da96d77fe4 : FAILURE in 30m 02s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/471d6baea0fc42b8b0d642dd358b23c1 : FAILURE in 37m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bdecf03c30c34ad4be4bc286ab37cf6f : FAILURE in 1h 10m 53s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ed218e01d8094feba8020d94ce2c1b9b : FAILURE in 31m 14s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90a52c074e474634aaf155548ddca08e : FAILURE in 25m 58s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/735dcd0fdc894b2abae21c7da326bfd3 : FAILURE in 53m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/935693aa823a48b6ae13af9bc9e4240b : FAILURE in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"992189de94f4b0845e09f9e656dd7be154f99766","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-18 00:19:18.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/765840/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b3721d9a287a771de5c1d50f74cccc472702fd46","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-04 11:44:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"739230e8c41045dc587ba9dd6d7f01da4d511541","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-04 12:38:59.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765840/3 : FAILURE in 54m 10s","accounts_in_message":[],"_revision_number":3},{"id":"2cc85b8f933631b3936d4a75ab1cf854a330c6ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 12:42:06.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/765840/3/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/765840/3/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"88d523ce56f375a95b16ecd04283a8cbc71a894a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 13:07:46.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/89b35c3696d642fd83684f2e2f1733df : FAILURE in 29m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c2b05280a9941e99fcaf7e26da77dd6 : FAILURE in 27m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c062be95312940d9b77ae59323b5baf1 : FAILURE in 26m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8b6b047d2c814698856a257ba0b44a98 : FAILURE in 26m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c9c7f01c025402bb882b5e9a6228e9c : FAILURE in 30m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/08c4d85c63184d09a065e01d5408e495 : FAILURE in 30m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c272045e17d24786a891f0e2aee0c6a2 : FAILURE in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/070aa67535d04b8187fd40ebe95b66d9 : FAILURE in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea4f4e5641e4613833344a840766fb2 : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7229df3638ea4f949567482c1635da51 : FAILURE in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/518c03b522904039a4c0c4fa15a0aed9 : FAILURE in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf94b2598d3b49a5bfd297cb7232a5eb : FAILURE in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e93313561743b68d4c76d0d6233fee : FAILURE in 5m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d23887cfcff4ac5a799b9654d6cab61 : FAILURE in 21m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7455454c0fef47168e5b8f1fbbcfd290 : SUCCESS in 1h 08m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/39e8e3b4a54249b2ab4f2b8f143025b7 : FAILURE in 27m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1aaa3dd982f94170ad824aa4fc24fe54 : FAILURE in 45m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8e8c05d0d3444ed39e631850aad46448 : FAILURE in 55m 18s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/73d17a4f8dd642eab79e5fb39a7b2bf8 : FAILURE in 48m 49s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6d8fa72e40bf416182962f0e8283cb01 : FAILURE in 41m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f46ed334edf14ce9bfde9fadf27f00f7 : FAILURE in 51m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d265666f1aa04102b7fac961d6c485f7 : FAILURE in 38m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2f24c90da8f40468e3685aab2df0113 : FAILURE in 45m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3955e1235b04095aadc12f0cf2b9e5a : FAILURE in 35m 14s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/51c93d67426d41b4b5aa568b95138b76 : FAILURE in 37m 03s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0fb755f74df94001aa105dddb106aaaa : FAILURE in 37m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5e1d51f806a248e29a22f711d0ba3f14 : FAILURE in 31m 39s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/0a46ec550f6341d6bb6a6c4edc91a9d2 : FAILURE in 32m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/90e93f910b774bd4ae2396f21a0c4de0 : FAILURE in 1h 18m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0b571797b6634065a22452ae9a8fec1f : FAILURE in 24m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10c33bf05731494cbc7ca92c628aa8be : FAILURE in 25m 01s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb0991232af34a69991e048cf3dac61e : FAILURE in 45m 32s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6a13c3b21d6c46e1bf7bc8ab50b7d094 : FAILURE in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6d7efb84288e00ff06865af80a0c903daff2ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-21 16:14:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"26feb9a149e3b994c285a5dc7a9332195c21b2ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:26:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9140d69f1061404591b996f68cdeae94 : TIMED_OUT in 1h 01m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bfc531f2ca84a13ad802f0381c2cadb : TIMED_OUT in 1h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d748ef0f6054c369c36c778d5e6fb9424cbfaffe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-21 18:51:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765840/4 : SUCCESS in 1h 51m 46s","accounts_in_message":[],"_revision_number":4},{"id":"52c509b41b597679c7e54da1ef690f7d418e919d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 23:14:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a52b0d3e31b24f82a682bffcf8eb1a43 : FAILURE in 56m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3355cffc98446b69cd0a48aab0351a0 : SUCCESS in 28m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/431767b656374cc0b2745700a6a93be6 : SUCCESS in 1h 23m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1829812feef44818b25f3ab3dd1bed7c : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a6fafb8985c47678ee1a223b533bf07 : SUCCESS in 1h 41m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6e4a6971fb2a4569beb314808db6d96d : SUCCESS in 1h 37m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a06dafc4fe8439c9c0c2cfedbc37ea8 : SUCCESS in 33m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ed32e3ac09e44d7a2b1be67f29c199a : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1ff8912b734d019af78d27b79d3f9e : SUCCESS in 14m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d232dd099f4b178004d6651c684588 : SUCCESS in 18m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d236d966185344c380314c1872d4f741 : SUCCESS in 23m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96f1ac45077f465795dbc5ac21385099 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b93883c25c4183974ebe934cd3ecde : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2bcd5aebf804aa6baf27ff47c395a43 : SUCCESS in 1h 11m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8b66d17cbf743f7929933047060d10b : SUCCESS in 1h 13m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/30f919a10bc0414db5935244ba0407af : SUCCESS in 1h 07m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3548aebc1e774ec5b090f229aa333ce7 : SUCCESS in 1h 00m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/71d22e43ee034d0ba520b6b57624178b : SUCCESS in 56m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/96863d2b5f204bccb133dd6fbefe05ae : SUCCESS in 1h 16m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2e89e129f8b4ba7b7221209eb4cf28a : SUCCESS in 1h 39m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9646c66345064f4e98403297568e9dc9 : SUCCESS in 1h 33m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94b0ed7fa7574595bc6c3d65e5df043f : SUCCESS in 1h 05m 34s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d48fd71f3bc641d9aa03a29e0bbbfcb1 : SUCCESS in 1h 11m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce796242c62840d0bf24361da1e362b8 : SUCCESS in 1h 21m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b0dcbc57caeb43a5921daed9c51abf5b : SUCCESS in 32m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a69d6430142d472caa33e3d8a68ba5b0 : SUCCESS in 1h 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4cbd9cc461170517ca9e1b90ff7eace5e43840ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-22 09:56:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"afb06218743e3980bdd7dd4b742a445f0bea7ab8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:46:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e40e6f25b58b5a56de709df4cd9b32bc709b78a1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-22 12:28:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765840/5 : SUCCESS in 1h 55m 03s","accounts_in_message":[],"_revision_number":5},{"id":"04ad8a7b2b20ab0ee2ceed419e96e656839da087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 13:01:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/64b6e2894b984bbea4f833a82cf858e7 : RETRY_LIMIT in 18m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3feb84c292b4b02a33eaa89656a35ef : SUCCESS in 30m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7bf15ff1d73f45c0a4d7e7e368ff58d4 : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5408b433da424547956de497f10d0d8d : SUCCESS in 1h 43m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/edd9c1e256744143ab6ac981ec9d4f6a : SUCCESS in 1h 21m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c5916d1720ef4ca780c733ee5037847f : SUCCESS in 1h 59m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69dc3642642d4de392876a7a66cbaffd : SUCCESS in 39m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbbcec82cfe04891a9a0cc17d6d2004c : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a1adb026e14c2a99b424589dfa84d8 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c1c47f6ff514ad88a82dce7f8fd4239 : SUCCESS in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2f3aa15eee943f1963a1c5f3f5ea842 : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11ece6bccd6b4b8fae363779ba88769f : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8144b44bf28e4cc3be33293b4b2a2050 : SUCCESS in 8m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd67ec0a507d493f976cfea313bc3af4 : SUCCESS in 1h 25m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a40cc605c614950930f76ab775205cd : SUCCESS in 1h 28m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/50674cd06ba2422c9109fc09b16a88cd : SUCCESS in 1h 35m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/83475fa3ec544ccf8f0b871b9cf85023 : SUCCESS in 1h 06m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1db88d5da3674005b560f39cf55a8c59 : SUCCESS in 1h 05m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/304bb537f28846e5aef92b1d39a110f2 : SUCCESS in 1h 07m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9d115ec6b436418ab667a85bb5d3d186 : RETRY_LIMIT in 1h 09m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c486966f1fc34b04814634e6b2a5d8ce : RETRY_LIMIT in 21m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35a4625431b14a88a93940018957c304 : SUCCESS in 1h 07m 17s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/7ecb69713b864ef597d6d6641210503b : SUCCESS in 1h 00m 57s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ff73f6e337a48c6b2b8da2efdfa08a3 : SUCCESS in 1h 10m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/839cd2b9dfe94c258906aceef9093f06 : SUCCESS in 44m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/89b47c6031ee4bcc87d94e073b0964c6 : FAILURE in 2h 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7dba3894488b9ba8cc4af2788591fcd90bff7434","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-25 17:53:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bb24dfd48ec1ac7a80139340bf6970e95ee0f525","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:54:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eb52aefae34a25d748ce8c190f1b70fff5fff28c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-25 20:05:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765840/5 : SUCCESS in 2h 10m 45s","accounts_in_message":[],"_revision_number":5},{"id":"5cef567af42ee60501254ae805c3ce5355416bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:10:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/521aaea893d3468da5734b9ee14b274f : FAILURE in 48m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef1843edaf4743a29fed8937ba4efbf7 : SUCCESS in 31m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9a1aa858501348bc9ab136e6d7e7f2ec : SUCCESS in 1h 53m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a27ee0fa48814e72a8f7b84007e5c668 : FAILURE in 1h 51m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b1b718eaf36d4a0ea66729ca6c0ea2a9 : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/577f3952a86d4a6a92e3d95b8681d6ef : SUCCESS in 2h 23m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c286aa677d4475fa4a9ef675d01ef39 : SUCCESS in 38m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af2d1ca10e7f422099066547b83f979f : SUCCESS in 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d5b81d7d2744eda1f2297214747e02 : SUCCESS in 12m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28a2d865951f4409995d6f52eae210de : SUCCESS in 49m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dd574fe64d9461fae4cdae33c2a9207 : SUCCESS in 31m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b95cf77609b84fd9b2d314252589e75b : SUCCESS in 29m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea72bf5367e145c5a666c2d1cf01181e : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cb5e84bbef446daa47f2ea7c51c3f6d : SUCCESS in 1h 12m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6442b43bad1447c6b3517e26582a34b1 : SUCCESS in 1h 10m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b79d5f18ab9946b388f90acb16653423 : FAILURE in 1h 46m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a69529462e134291a91dd56e30033bc5 : SUCCESS in 1h 05m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5fd1a3c1e9ed485b87ae1b9875aad272 : SUCCESS in 58m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c7f24ad028ff4aeaa3876a97a19e93d4 : SUCCESS in 1h 09m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8c025c978ba849d7a21494e3b54b63a9 : FAILURE in 1h 14m 32s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8c18123ef4d40338f75783259a37871 : SUCCESS in 1h 32m 59s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97a85c02d76243c193c0a8aca53bcd5e : SUCCESS in 1h 05m 49s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e95fb8eb243c414fbf4d99111ea4873e : SUCCESS in 1h 10m 50s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d4b44a2b355146a5854b3ace16a54b8c : SUCCESS in 1h 08m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8561df5dea0f400789524c1b2d70ac34 : SUCCESS in 33m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/274ef39c14c24680a1c8555f8d1d0fcf : FAILURE in 2h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4388b9aeda97f221161762a02efeac163cbdf2b8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-25 22:44:02.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"97c19b69639b861a38e689f50386723941ced50b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 15:24:19.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/40/765840/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d02895dd6ab683263c32bf80ea3ffb08098a9c5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-26 16:19:36.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/40/765840/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2b7399c605509e954c1bc441331e57e00cbc87e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-27 22:24:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"87d9fee45e2873a7ffe9a243acd002f2d7fff19d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 23:10:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e48fdeffecb485bb7b4d20e543b5fce : SUCCESS in 34m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e5c7e64f0dc42cc832f32c9861da0a2 : SUCCESS in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a206260265d82790c699ad743e23bda7bfa27df","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 00:22:26.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765840/6 : FAILURE in 1h 56m 13s","accounts_in_message":[],"_revision_number":6},{"id":"0c6ea248e16528906e4042c9473c088440cde5a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 02:28:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/75b51e3839f44769a682eda58bfe71d0 : SUCCESS in 51m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b3cab296a994a2b87419fffccb92494 : SUCCESS in 35m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6f9acdbbf7cd4a1cb8dcc5602ac4202c : SUCCESS in 1h 26m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4066fda0433c4bb0a30f794fdabe2433 : SUCCESS in 2h 02m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/873340a7b55b4039bc76f296030e9fe1 : SUCCESS in 1h 52m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2c69816d7eb941058b435fac8a9eb068 : SUCCESS in 1h 46m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/445b38b409904b918d5d1411758a47c5 : SUCCESS in 36m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/673797e87b714c81a45de73c88843710 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7eb2098e004538ab0dbb785ba80d15 : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d32cb159dd74494856c72890282ebd2 : SUCCESS in 28m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab02cb027fd4784a409a475796046f2 : SUCCESS in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb28b180ca5246ebbd42e77a25ee726d : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7748c88ae8374452948767ae95f0992e : SUCCESS in 8m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/653e24b425e74e35a541e8a6757b0496 : SUCCESS in 1h 08m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9d9250fbb3a45ceae7f0e0d52243d16 : SUCCESS in 1h 08m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fdaf80c4a63547f8b05941a53a010217 : FAILURE in 1h 44m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/653b75703eb9417ca2bb6826f2aa49e1 : SUCCESS in 58m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d580a7fff718440aaecfd39e809c14c5 : SUCCESS in 1h 01m 35s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c046128d26f4b248eb7007f7db1393e : SUCCESS in 1h 23m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/66116df9e0804377a9d877db6f410dcf : SUCCESS in 1h 35m 41s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67cede93b3f0467a986a17e4c09ab8a0 : SUCCESS in 1h 37m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25d697b668c04eb3857b5d34a94c365c : SUCCESS in 1h 42m 02s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/89c0e077b1254597b0d2c6daba67177d : SUCCESS in 1h 02m 43s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cfee5bd714fb42aeb7572dfe71ddcf38 : SUCCESS in 1h 15m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3495b5bac14e4c12a95de94556a69586 : SUCCESS in 39m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22aa0986287a42c8b50dab022b2f4695 : TIMED_OUT in 2h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e2f94d3cfe636c5d1016676c6741bb39fda4b60f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-28 10:33:23.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/765840/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/765840/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"c7e443f55c995ed4bd1f687d85848f5cd931556a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-28 11:14:02.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3b8459574dbe90ccd0fd92795d0492d8fb767d46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:07:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da3b05e553074f87bd30103408552a55 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57030bd3ebd747bb8725a990168ea15f : SUCCESS in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4c4bf317c7ce234973ab45cbebdfb4f0c0b9d078","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 13:21:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765840/6 : SUCCESS in 2h 06m 21s","accounts_in_message":[],"_revision_number":6},{"id":"3e594895ec5cf60cf1e2b15961ec337c93ba346f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:03:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/429fd6107c484adc8de072b6d6b0c749 : SUCCESS in 46m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/33b638aa39844d579a3654fc6df0473a : SUCCESS in 29m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/25774262a4024529afce41feb84ea7a8 : SUCCESS in 1h 20m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/913fdbca99ff40cbb0177335e7f549a8 : SUCCESS in 1h 27m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00caba90213c43a6830e98d1f4e8b347 : SUCCESS in 1h 42m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ea4695c2da7b465e8979626c28ce5fed : SUCCESS in 1h 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/883072d3578d47a6a5fea36151b5104a : SUCCESS in 22m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72eeaef0e4b7428480d832f6ed6acd27 : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f15832d5eea4fea899c1108727ab91b : SUCCESS in 11m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/489d239b140b4d92acac7161f2b2ef8e : SUCCESS in 18m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e9a2c8ef49749659714e559b74955ee : SUCCESS in 18m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5b16ffa95804b58b5419264aa065b72 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a36d811616354bfea3d3fcb15473f84c : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d335141186924152813e967a8db03133 : SUCCESS in 1h 15m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd559e1de62e44358956a75f6806972a : SUCCESS in 1h 05m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4ccffa0018634e8aa9207972eda91f4c : FAILURE in 1h 46m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c5c3162cbed4e9fa5fa181af4f35a29 : SUCCESS in 58m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b2be6e36b77846de820124f6584e81f2 : SUCCESS in 1h 04m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/766c5859191b41a6a8e33114116a0f8e : SUCCESS in 1h 49m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb9a1ca9ff544a25be4bf8aa955f0c91 : SUCCESS in 1h 36m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c020cb7f5e2c42c3847da5a0fec4c340 : SUCCESS in 1h 49m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39cd1f3488524c6996313679ca24745c : SUCCESS in 1h 09m 03s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a761a7cb6e82412f8257eff47fa2c93e : SUCCESS in 1h 06m 50s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/142e32f74ae94221a5cfc0d0c4cf88b0 : SUCCESS in 1h 06m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a6e73ac7bb01426c929b2366c43497a8 : SUCCESS in 34m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54146331a05640c6b44d10a0ce9163e0 : FAILURE in 1h 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6dd4437fb1e15e38aaf0872c75dec819d297837c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-28 21:00:22.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/765840/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/765840/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"e2b533eb8986ab656d613e18876a918cd3d52e9e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-02 20:43:05.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/40/765840/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"6b689b964e293cf5b1ab8cabc831ba767ea06091","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-18 08:16:51.000000000","message":"Patch Set 6:\n\nrecheck let\u0027s see how rally goes with https://review.opendev.org/c/openstack/neutron/+/773422 merged","accounts_in_message":[],"_revision_number":6},{"id":"4e14b042efa411b112bffb988877a4124590d6cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:12:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2622672386a4d26a27ebe8cdfb01d58 : SUCCESS in 45m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec664791651c4fddb5358cf8724e5e60 : SUCCESS in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"325c5c5149b91c380d14b9054210751234c3aa6c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 10:03:23.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765840/6 : SUCCESS in 1h 43m 12s","accounts_in_message":[],"_revision_number":6},{"id":"c9ada0e53cec51e40002c1259acbabafaf7c6354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:04:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a7f777e6e9ad4db38b105282f6bd5c8e : SUCCESS in 1h 11m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bfe5da5d80d746c4ba53e7eca1673e71 : SUCCESS in 1h 34m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/588a62603ef147fcb3791b38beade621 : SUCCESS in 1h 43m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e782c9fde334ae6b4bf8d013e6be633 : SUCCESS in 1h 18m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4954a095e9c1431c9380bd76cfae04c5 : SUCCESS in 1h 43m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4d5b8685d28452f83f78500901cb0e7 : SUCCESS in 40m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/397f2668011d41c3b97c00d4ca538bfa : SUCCESS in 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d3a00b0cac47f8b2afb9b1ed16b82e : SUCCESS in 12m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe6aedafbba24364ab0db84f1bcdd98d : SUCCESS in 19m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d91b555bdc364f3e956a390502dad748 : SUCCESS in 24m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f663f838aa7425690985c2580543f42 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d8612ce96a477ca97662d2a9968ca3 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/abfb05f7c542410cbbfa8a02ce2f3ed5 : SUCCESS in 36m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/707d0ded77954c01991f89c4cf4ec668 : SUCCESS in 1h 07m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e8edbc9ed3484e458c99d4816302a4da : SUCCESS in 1h 04m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8865a54ecbc64363964204984131a07b : SUCCESS in 1h 00m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/503c9545d7d3430f887c1b727d399ed8 : SUCCESS in 55m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/97a36f5237964f4388ff800db736dcd5 : SUCCESS in 1h 12m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/421c7a603b214a7589c32efa6742c41e : SUCCESS in 1h 38m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/930ae2edc04e44c19149b8bc56e2df27 : SUCCESS in 1h 37m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffcaf03ab3904413b2e18ac5a7b7b239 : SUCCESS in 58m 24s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bae8ac5f5380424aacad2691c13d1dbe : SUCCESS in 1h 01m 44s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a365a641f3644a93a74ebd123325c770 : SUCCESS in 1h 10m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f29e7562b82d4f38b7d98968e24a3633 : SUCCESS in 34m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a3a48cf378c5492bb4ec86397d43f1e5 : SUCCESS in 1h 25m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5b894789a8ed42c3a5880c1069e2e6fa : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":6},{"id":"1152b57becdcf493f66b11061ac0b72afd7b4a99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 11:07:59.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/765840/6/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/765840/6/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"5f969d7cf2d57151e0f9a92d2634ea887131b356","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-18 12:27:59.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cc604cbc601cd506da1f528645f065a3ccc541a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 14:54:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f18b3851695d41bc85b6296cf515e447 : FAILURE in 59m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"199936bc3287d9873ab9070bb333f9357d1e191c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-18 16:13:44.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765840/7 : FAILURE in 2h 18m 09s","accounts_in_message":[],"_revision_number":7},{"id":"b2455746761a926694f7bf11ae1d4c55bc5a49fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-18 19:55:23.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/765840/7/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/765840/7/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"3f457d93d7d90e37ae0335981bc8f6153e451606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:04:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12cce9272743495e8381d6a296ad3078 : SUCCESS in 49m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/14e5bfb165f041328ad4b19e97dc2fde : FAILURE in 2h 03m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/17732b0222c44e81ad9b87e53abfd97d : SUCCESS in 1h 42m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/638b6ee945874da8982fad74094f3029 : SUCCESS in 1h 46m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c54deb1fcb7e473689b84ea1bf1efa62 : SUCCESS in 1h 29m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eedccf54ba7e40af899a8c298a0d23f7 : SUCCESS in 38m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50c60862a9454a158f2a142dd4e48f05 : SUCCESS in 30m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b026fa41ff934fb6ad6a04c78080c21e : SUCCESS in 15m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cf656d99f8d44d29000d32ac110056e : SUCCESS in 32m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/090dd033c2c348ba8462c6d208a5a68d : SUCCESS in 31m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d8a9c684a154ce199bdc35fa6ae103d : SUCCESS in 31m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d6b557163f4115b956b839fa705b5e : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8305b609ed264811a96b5b402fa86ee1 : SUCCESS in 46m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a668f911950f401ca1be89421bd44a25 : SUCCESS in 1h 07m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c64626af961e499fabbebee5a7065034 : SUCCESS in 1h 33m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a99cb4fe4c1c41eeb906cdaa449d12c9 : SUCCESS in 56m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/55da71a0057c4b4c8666d0d4f8d36d7b : SUCCESS in 1h 01m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1372a6ca3ac846a68e31018d7f78d32f : SUCCESS in 1h 51m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e3249363406045329bbd110214997e74 : SUCCESS in 1h 36m 22s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6e040b6970a4f6fb7a363d967d7008e : SUCCESS in 1h 36m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a897e9f7a1dd4befafcac2b01bdde13c : SUCCESS in 58m 27s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b5ef4a2acc8844c69921a0e5a4b37628 : SUCCESS in 1h 08m 28s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/363dfe05db194ed38c7572f876eead10 : SUCCESS in 1h 18m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6d3021a57d4c4fe7a1c61321cc205a0e : SUCCESS in 34m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c8aea09a67d4d46acf81cdf2745161e : SUCCESS in 1h 39m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e0bc06180e514830a59aedf55123cabf : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":7},{"id":"524236305b62f872d57beb50b8654dd5ec308cbe","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-18 22:36:22.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"a0581d71c6af3d08f8b0e047a64c0903cb4cc232","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:21:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e84b1940d8984ffab0b789c1d0c1c339 : SUCCESS in 37m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/926b0af0bb104dddbffc1230099e9f9a : SUCCESS in 39m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3a8d1c867838928cfa3d6959a04122e61fe9b4ca","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-19 00:26:25.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765840/7 : SUCCESS in 1h 49m 29s","accounts_in_message":[],"_revision_number":7},{"id":"a5a6a5c479a2bbfad14e6c141f1ceed146bce5a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 02:03:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/139618782164499a9ece9ecb1005138d : SUCCESS in 46m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3d91223148524b9eb04d30a5c8bf7028 : SUCCESS in 1h 59m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fbaa7ad0e05c4e5ebdd4a46dfe3199e5 : SUCCESS in 1h 37m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/142bfac7560d486184a564b6faaecec6 : SUCCESS in 1h 01m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4e2e3988a0ec4b8f9fb94b42513bbffd : SUCCESS in 1h 51m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fcac633b6c94c258de48f52965d2a27 : SUCCESS in 38m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4910babdbe5450b861b750bddf0c3c9 : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fd64401da44df68dbf2cab34556ecd : SUCCESS in 12m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39f4697877dc4d24bb1e51c7e9c346b0 : SUCCESS in 34m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88149cc1faa142af93be66ec8e2418b4 : SUCCESS in 32m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/507290964d2a485e9325742e8604117a : SUCCESS in 27m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad20592ab024dc585adf7649ff26fab : SUCCESS in 10m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9c8656bb98c4633954ac04849082310 : SUCCESS in 49m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/55d0f2d8587d4232937e0c1e9d14b2b9 : SUCCESS in 1h 15m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ff752a7daebd4c85bf98a8b4309fa245 : SUCCESS in 2h 00m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc2cf7e828954dd2bcbe688a2430e505 : SUCCESS in 1h 01m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1e09c8eb506b4a988ee8b4b2d0a6dc6a : SUCCESS in 1h 03m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d9c0d4383546418cb22908b7ffe6f5fc : SUCCESS in 1h 16m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15dbc62794c94556bbd86e7b872f7276 : SUCCESS in 1h 31m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d71269737f44dbf85d6ea09fcf6994f : SUCCESS in 1h 37m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acd64bd15e0b4bc8821f26adae4be6cd : SUCCESS in 1h 05m 52s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bf144aef7fa84ea0a599c3f600c6bb8f : SUCCESS in 59m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cca478015aa14a4b9fa00f6a9f5b8c7c : SUCCESS in 1h 01m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4add4e62455e4f9ea9ddce5eb4d94ed1 : SUCCESS in 29m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d99e931582ee42958902deb1513a1be1 : SUCCESS in 1h 24m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8ed8790694a5403695bff743b0b43dfe : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":7},{"id":"d40151bb09e26975cbddc664abfe57fa01d27578","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-19 02:29:12.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/40/765840/7/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/40/765840/7/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"7abfff269da921abf5373e3804c7d5ccb4dfd612","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-19 08:31:23.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"52079d5ee62e2c54ced0ecb548e156ea568ef877","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-02-19 08:42:43.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2021/Arista_Tempest_Testrefs/changes/40/765840/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e0d1746b2e0fe34c8782e99e301594f67e80e195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 14:05:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"47c6c14236b203977d9da5381d214e65bbe75237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:24:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ee44a8adb9e42b49eee2948c498a676 : TIMED_OUT in 1h 02m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67030570555441449ceb4d0d68ade5b1 : TIMED_OUT in 1h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"47d05092ec3aa8ad376cc79082a2f79384276399","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 17:21:12.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"0ed83ed3fcbee8c99d0ea3b67a6e73037872232f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:52:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5a0af1479c234b16841684c21c24fc13 : FAILURE in 48m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cf5184b2c98c41e18efa21d12ee33140 : SUCCESS in 1h 35m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5df79de8bd6d46f9b65a3f5845d014f2 : SUCCESS in 1h 59m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7ff20ce78974f16a7f143e89f532d1f : SUCCESS in 1h 46m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d31cb4f4de014fbdbc367bef70a08e2f : SUCCESS in 2h 11m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ced42f61d6a94216b5657e8e106a3219 : FAILURE in 30m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11014299d408494ea8cd773a9fd5c534 : FAILURE in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/403fbab5bf53428b910f19320d2f225b : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc7065d942f74fabb519178de82d495f : FAILURE in 33m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6380e8d7585547778bfff55e9d4d3271 : FAILURE in 37m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d539285304a74779b606a3e686d819e1 : TIMED_OUT in 42m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e39a115088f4d8cbc83965b806850ed : SUCCESS in 11m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cba7f0324ebd49a1b78a78ae76db8d68 : SUCCESS in 50m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e9fb8f3751742ceba8e070e4fd1f1fe : SUCCESS in 1h 23m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/869b01f52a1a45c8b1041f9ca925767d : SUCCESS in 1h 26m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e9767772fb343bda2168e1ffbdb01b5 : SUCCESS in 59m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/086cdde5581e4d51919595ffdd8cb005 : SUCCESS in 1h 02m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0c1504e905f34440a4df4da66d75f3e6 : SUCCESS in 1h 15m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/36c675191ac94c4082b22d6371b04f10 : SUCCESS in 1h 43m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e67085db37a430d93c9e52a98371f23 : SUCCESS in 1h 34m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3ac67d1fa2e4f779d4d8ed9e48fef54 : SUCCESS in 52m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f2a085e962b34d3183e3a0261a760123 : SUCCESS in 1h 08m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e1f10592731547909d609c3e6a394392 : SUCCESS in 38m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a51dbac0c71443fc9a066eb9e43fa6a4 : SUCCESS in 1h 33m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/89ba7c13fbbb46da91ef8292721ab7c0 : SUCCESS in 2h 05m 07s","accounts_in_message":[],"_revision_number":8},{"id":"e98d86471bed6ae039cce05c0a02f73f83e5ba27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-06 16:43:53.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"2d5863b59324dbedca6c8d3ce933f703ff7e4e49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 17:27:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59cef8127ee54394bce634fd7e93d7af : SUCCESS in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"abbd778de1d1e6263ad52f37c60a8f601b28aaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 19:03:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3c688d757a004772acbafd39a981cc13 : SUCCESS in 50m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e928d6bfa7d24859b6c640fbc012f3fd : SUCCESS in 2h 14m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/07fb46b192184c0f8d2eb7d446aba0b4 : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b434d1c0cb31407bbc70efe610e5e496 : SUCCESS in 1h 30m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/50c74df1132b49919698ec071dd1a2e4 : SUCCESS in 2h 06m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dcd962f70ac4092aadf029571b20877 : SUCCESS in 44m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae0900124950452e9c91495a96b168a3 : SUCCESS in 36m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa4e85128964582aa8891201bf911d3 : SUCCESS in 12m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ada13ce77d4dbf84e5b8da54e2ecac : SUCCESS in 34m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7b80b7ecd2a478697149c022105ca45 : SUCCESS in 31m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86a3f75269314306867d9e4777d0b791 : SUCCESS in 33m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fadcb3211f74920bb2c622608001294 : SUCCESS in 12m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/719cdc0d3d9c4af9aaa9bf2162a69925 : SUCCESS in 55m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3ff481dca4843e19e383cca9c4b59d0 : SUCCESS in 1h 25m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/06a947967e0944219583049e359abc94 : SUCCESS in 1h 30m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db71c7d10a7b4ba5b2f953ad2b350098 : SUCCESS in 1h 05m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f674de79b9947a3b1baa895dda06f6e : SUCCESS in 59m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ec3df0dbfe78423ebed41ac94101abde : SUCCESS in 1h 15m 28s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c538f7f463947cb94fee886ca5178dd : SUCCESS in 1h 47m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/225079c82d604451981d42917b097771 : SUCCESS in 1h 38m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7aeabab74ae46f2aa6a91705095dbd1 : SUCCESS in 1h 17m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4e41bbd302ab4136bf015f7c0dc2df31 : SUCCESS in 1h 17m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f90b49b27b404768816a60da7fe3fba0 : SUCCESS in 35m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/95587ecdd4514da2bd1132a7c16b9b8c : SUCCESS in 1h 32m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6831f3f00cc941e3a738197c855f6f3d : SUCCESS in 1h 24m 30s","accounts_in_message":[],"_revision_number":9},{"id":"8b17e54d513e8abbec53cc39e35265a9a357965d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-08 07:51:28.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"19edfd032e3001b0feb1d03ffb86002d50d66956","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-08 08:59:29.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7c3f7a3ca3854ec811ddfa79411f3549bf34fe7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:00:34.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"2b360b4d39e63ab9118f6e9a317d32dfe90cf1a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 13:46:28.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/32aa1a45155b4540b32b4581c393ce58 : SUCCESS in 1h 09m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/99154cdf7a844bbeb4a3c6b4283122d4 : SUCCESS in 1h 42m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5725644280a544bf8171fa42dbb08974 : SUCCESS in 1h 46m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9b6be09fc89b46da90da572038663284 : SUCCESS in 1h 43m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0fbc604bd5514454b754c71b4f12cc9c : SUCCESS in 1h 16m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99612f95edc34679a5435031c37eb4b3 : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b87cc0a70a463dbae85d9d24484046 : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b709cc497c14b5eb81d3f9f411ebd52 : SUCCESS in 42m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23429e507f844d9185e6e582285284c6 : SUCCESS in 22m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b12f5f8708749a89eef6bdd63ea2902 : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49d9bf17c3e44ec1ac9d99b5e5ed317c : FAILURE in 41m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c44f01b5fd5e4d0fb3d9d85bd33af76c : SUCCESS in 1h 26m 51s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9051927ed4d948858d6336e70d87cde0 : SUCCESS in 1h 05m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d1a3b976f654936a71545130f7af402 : SUCCESS in 1h 04m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ddb4dc0bdde646ba87d720497db2bf28 : SUCCESS in 58m 11s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85a7c897baa94f91a7446b6d016dbe21 : SUCCESS in 1h 43m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa8f534258394d07a504598af576ad0e : SUCCESS in 56m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d6b40d7af570479f88cfe92cb7317722 : SUCCESS in 1h 06m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/62eb8ceeafe84fcbb8ce7ccf1abeb598 : SUCCESS in 34m 34s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/20f9dde080964c04bcde97c6729800e2 : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":9},{"id":"82b23d5c15ac00bef928c062441e3aa87a0eb3cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 13:48:00.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c4a158f0a30de95a45fb0bf5d7123946a2db9c68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:06:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62805d28669f4184aa42095771fc22ee : TIMED_OUT in 1h 01m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2210939276a146f8a2472e9251c0a7d4 : TIMED_OUT in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d80f3dc075cfcc6269c765f91614b9810ad1095d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:34:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/35ca10c88aa04e1eb7fc2456dd80b81a : SUCCESS in 48m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/982bf49e360e4b1e9b77ef900771a687 : SUCCESS in 1h 19m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/49dcbfee01774ec7acaea6ad32df4278 : SUCCESS in 1h 20m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db6e9af230e84efd92072a73a1da607f : SUCCESS in 1h 35m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9330fdff851044e7a6150e3e46063ba2 : SUCCESS in 1h 38m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f55fe8ad49484a11a728dcf9aaa0ce91 : SUCCESS in 32m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb84f36df124217930d0c993a47a5d8 : SUCCESS in 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7143a27cf8974fbb9b865191ae7287cc : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e18d7773dd254af3a23348d347c8adc0 : SUCCESS in 37m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c31ca1d50ea44dfaa42b62af84829713 : SUCCESS in 35m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08bb41a4341d4f6d972d9f08ab9447e0 : SUCCESS in 35m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3801bd9e56364ee58c31bf1bd0f208eb : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9d4c14b89c048518e74f7cdc4f639e9 : SUCCESS in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d4e8204aaaf4220a6398d1886ce36c2 : SUCCESS in 1h 20m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/03475c0f71b449d6aa6009934f988c87 : SUCCESS in 1h 32m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/07e92042679d4b4195218f6883ba0fb2 : SUCCESS in 1h 04m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f2fab95e6ec4702a324eca2f8e6edc0 : SUCCESS in 1h 08m 50s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/69be65345d0941579b7fefdeb0fa842d : SUCCESS in 1h 18m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5b5153a523b45acb4684db9bf10fd32 : FAILURE in 1h 32m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da3fb3c5a4ca494caeba808cd7bf118c : SUCCESS in 1h 37m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/541e6bea88ef4181b04e98eefb6021ea : SUCCESS in 58m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/771089c38d3a4653be7672e1dfaa9faf : SUCCESS in 1h 10m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a772728cdea411c91917999074e4d34 : SUCCESS in 38m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ed461ec927640e9ba028cd738f81911 : SUCCESS in 1h 27m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/24733d6e1be44bbcb366dd3e5fcab315 : SUCCESS in 1h 27m 23s","accounts_in_message":[],"_revision_number":9},{"id":"d9d524fed09e8fc027f3484048f7e191874ec92e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:35:10.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"8e2230914279f03162819b3b9fe3b79e7f13bd9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 19:28:23.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8fbfe7c94dc4c8c9acd636d86ba95d0 : SUCCESS in 1h 12m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e09a55ccb47444489ef02bd5de4d3365 : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dccd40445c1545c3b286c6c0b2102ff1 : SUCCESS in 1h 30m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4214e928a7e7407aa22cd6422d6e0a6a : SUCCESS in 1h 55m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9dddee6d14c3434c837bb24187f20de0 : SUCCESS in 1h 27m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77eb3275a06640bbbd49c8e39eacafdd : SUCCESS in 37m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14cff97df8f04eb699440349bba6ecde : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eaed8463b274645bd518a29a7b9c598 : SUCCESS in 36m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38dc3fbaf5a94726adda6b97b343046a : SUCCESS in 34m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc9ca52d13848678391a967a75fda09 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3469fcb642ad47cdbdbe6c1bc7f4f2b1 : SUCCESS in 44m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e11b6f039d740528d0e410a353dbcc4 : SUCCESS in 1h 15m 13s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a1be989ccb8496e9e6df783d368d896 : SUCCESS in 1h 07m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff3833a27b1d42ecb52de3e589df9ca0 : SUCCESS in 58m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/91ba805ec95445f6b01779c517b8cb86 : SUCCESS in 1h 00m 10s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfb9ec85dc5645aa987dd5611f627b9e : SUCCESS in 1h 51m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/897db3e1b09f40399d61d834d68c9576 : SUCCESS in 1h 00m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/803a5c7ac9e54ca29d3a1d0e47307a2c : SUCCESS in 1h 07m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4b4bcc33afe84a99b91e36ba34c6da5b : SUCCESS in 43m 45s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c3c3ab18aa754b8f96358046601c9b45 : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":9},{"id":"b462e59e5cb1ab8b2920f52278d34eaf20456090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 19:28:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"1373fab7eccfcb079b0194524fcf4f74f49f03f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 19:31:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614d85bd6a644960bfd47300fca4c0bd : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":9},{"id":"62b6860be20e5442144c5dae010c6e7b40c22f66","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-10 22:54:19.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/40/765840/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"e56d62423f89b7acf7e708e6319945028e3f6303","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-03-11 03:21:37.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2021/Arista_Tempest_Testrefs/changes/40/765840/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e6ce2b450a4e4b5cead0792d206bef314628b41e","revisions":{"508ac117bff1fd6e80230cf33cf47eac7a3eae68":{"kind":"REWORK","_number":1,"created":"2020-12-07 18:03:49.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/40/765840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/1"}}},"commit":{"parents":[{"commit":"84f356e4b1c37648e56b866fcfdcb3320ade9a41","subject":"Implement secure RBAC for QoS API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84f356e4b1c37648e56b866fcfdcb3320ade9a41"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/508ac117bff1fd6e80230cf33cf47eac7a3eae68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/508ac117bff1fd6e80230cf33cf47eac7a3eae68"}]},"branch":"refs/heads/master"},"705306dfea4a480e5edecf52e046b023b102bd08":{"kind":"REWORK","_number":2,"created":"2020-12-15 15:54:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/2"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-15 15:54:11.000000000","tz":60},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/705306dfea4a480e5edecf52e046b023b102bd08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/705306dfea4a480e5edecf52e046b023b102bd08"}]},"branch":"refs/heads/master"},"a14eb4397ba00e730c8079af70637a161dc4fb90":{"kind":"REWORK","_number":3,"created":"2020-12-17 13:04:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/3"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-17 13:03:29.000000000","tz":60},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a14eb4397ba00e730c8079af70637a161dc4fb90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a14eb4397ba00e730c8079af70637a161dc4fb90"}]},"branch":"refs/heads/master"},"0b7f70aa0d27854ea6bafc8c162897c95418950f":{"kind":"REWORK","_number":4,"created":"2021-01-21 16:14:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/4"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-21 16:13:45.000000000","tz":60},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b7f70aa0d27854ea6bafc8c162897c95418950f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b7f70aa0d27854ea6bafc8c162897c95418950f"}]},"branch":"refs/heads/master"},"29a3a1c4235e77d16ce745bf633f5ddd2400e3fd":{"kind":"REWORK","_number":5,"created":"2021-01-22 09:56:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/5"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-22 09:56:33.000000000","tz":60},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29a3a1c4235e77d16ce745bf633f5ddd2400e3fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29a3a1c4235e77d16ce745bf633f5ddd2400e3fd"}]},"branch":"refs/heads/master"},"a62ef93ed03c6d79564586f19c5437f423615826":{"kind":"REWORK","_number":6,"created":"2021-01-27 22:24:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/6"}}},"commit":{"parents":[{"commit":"6a8fa65302b45f32958e7fc2b73614715780b997","subject":"Merge \"Disable dns-integration API extension if it\u0027s not enabled in ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a8fa65302b45f32958e7fc2b73614715780b997"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-27 22:23:57.000000000","tz":60},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a62ef93ed03c6d79564586f19c5437f423615826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a62ef93ed03c6d79564586f19c5437f423615826"}]},"branch":"refs/heads/master"},"efaaaee2b7a4c8c0da65b426b5edfa2e03277f26":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-18 12:27:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/7"}}},"commit":{"parents":[{"commit":"90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13","subject":"Merge \"Bump neutron-lib to 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90309cf6e2f3ed5ae6d5f4cca3c5351c2ac67a13"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-18 12:27:59.000000000","tz":0},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efaaaee2b7a4c8c0da65b426b5edfa2e03277f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efaaaee2b7a4c8c0da65b426b5edfa2e03277f26"}]},"branch":"refs/heads/master","description":"Rebase"},"a0bc8cfe5fd005868bd82913cbc051d690c57095":{"kind":"REWORK","_number":8,"created":"2021-03-05 14:05:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/8"}}},"commit":{"parents":[{"commit":"cb64e3a19fdddb3eac593114a482c9dd69be68d5","subject":"Merge \"Add new policy rules NET_OWNER and PARENT_OWNER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb64e3a19fdddb3eac593114a482c9dd69be68d5"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-05 14:04:16.000000000","tz":60},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0bc8cfe5fd005868bd82913cbc051d690c57095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0bc8cfe5fd005868bd82913cbc051d690c57095"}]},"branch":"refs/heads/master"},"e6ce2b450a4e4b5cead0792d206bef314628b41e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-06 16:43:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/765840/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/765840/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/765840/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/765840/9"}}},"commit":{"parents":[{"commit":"067241b11b0e48513d326fa970acd2742bbb0bdb","subject":"Implement new policy rules NET_OWNER and PARENT_OWNER","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/067241b11b0e48513d326fa970acd2742bbb0bdb"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-07 15:51:25.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-06 16:43:53.000000000","tz":0},"subject":"Implement secure RBAC for the port API","message":"Implement secure RBAC for the port API\n\nThis commit updates the port policies to understand scope checking and account\nfor a read-only role. This is part of a broader series of changes across\nOpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: Iaa1d1024a5a9f1883ef3c8b073c685f17a099a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6ce2b450a4e4b5cead0792d206bef314628b41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6ce2b450a4e4b5cead0792d206bef314628b41e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
