)]}'
{"id":"openstack%2Fneutron~765874","triplet_id":"openstack%2Fneutron~master~Ia198d45f49bddda549a0e70a3374b8339f88887b","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"Ia198d45f49bddda549a0e70a3374b8339f88887b","subject":"Rely on worker count for HashRing caching","status":"MERGED","created":"2020-12-07 21:01:11.000000000","updated":"2021-01-05 01:17:20.000000000","submitted":"2021-01-05 01:15:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"765874","meta_rev_id":"39aa372ef6f13c9ab4814462e4d9b02b97ae1e23","_number":765874,"virtual_id_number":765874,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2020-12-21 16:00:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 01:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-12-24 15:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2021-01-04 21:07:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-12-29 07:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2020-12-30 13:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-12-22 10:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-04 14:21:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-01-04 14:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 21:05:35.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-12-07 21:05:35.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-12-07 21:05:35.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-12-07 22:46:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-08 01:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 21:56:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-09 22:57:56.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 11:19:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2020-12-15 17:32:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-12-17 09:13:07.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-12-22 10:29:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-29 07:41:19.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-12-30 13:03:20.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"}],"messages":[{"id":"529501c99fc2163fa024360aee67994289eee820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-07 21:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a799b412343f1d9e7b21b645e0b76f7a4013e5c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-07 22:46:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/1 : SUCCESS in 1h 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"63486d73347fe699f255cc3e84e43477d012e7dc","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-07 23:08:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75268d54d863642b8aed01895b315902413e60eb","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-07 23:25:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bae7bd92c8f5b4663a7fc6091d46feccc91a50f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 01:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1cefaeec16e64e83885953119e44b4a5 : SUCCESS in 54m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aea3088e3f444aa3be80e4a3e6cde287 : SUCCESS in 34m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2d4c33fc012544bda18585d7d464184b : POST_FAILURE in 2h 14m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f6b33a960ba84fb097bfe0800e47ff20 : SUCCESS in 1h 45m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11bd2010153348c995ec61cec2ab67ec : SUCCESS in 1h 27m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d8fb27b667b7479c9583d9947446387c : SUCCESS in 1h 44m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf74fb2338a340529386edf6b44d2555 : FAILURE in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60b978124e764737a2b1f3a5b78b2c7f : FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3ca06015e943d4bfc60d07ab37a84a : FAILURE in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b8620b9261455eb88adf2727a64651 : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee1586bed8f94706901230d8fef9b56f : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d9ef84fe574638bc94c151fe71c9ad : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/edec63a505eb40a59134b487cbe151a7 : FAILURE in 20m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67482bdf84ce4ddc93235a30094e36c0 : FAILURE in 25m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6661a0d432ba4d8c916994cf8cb66a3c : SUCCESS in 1h 32m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a97791d8b9924f1aaf0a034a38cf347a : SUCCESS in 1h 10m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/23dce4300f4f4c9aadf7141df233fb98 : SUCCESS in 1h 08m 25s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7e516eb5b71040fc966360bf77c607f4 : FAILURE in 1h 02m 52s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a117873b44674e10a57eaa16061369a4 : SUCCESS in 1h 08m 57s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f83aba66cb874a70891e02d398a9deee : SUCCESS in 1h 27m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/01f3bd2b667e47649b10546bdf6fefb2 : SUCCESS in 1h 30m 19s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/32d293e54f45436591af1df5e4118afb : SUCCESS in 2h 56m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7344053e1059416980b1b5f743f69eae : SUCCESS in 2h 03m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66bf2b3cbd1548d3bc7dfc89df360453 : SUCCESS in 1h 06m 54s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/434590db7d8c4a8eb35c7d9438c4a92c : SUCCESS in 1h 06m 06s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/680eb2e774194bc59becffecdd1019fd : SUCCESS in 46m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/11d42934f6a7409cb16c98a420e73565 : SUCCESS in 31m 08s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/e2acdc6709b748b281556daa19e25f21 : FAILURE in 1h 16m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2d0e2b02f3d343908cd9e693da531e0a : SUCCESS in 1h 23m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0a1559e58d30459e83c77bada62a0b31 : SUCCESS in 3h 39m 38s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c66b8e3af1534c43831e17218ee07d39 : SUCCESS in 1h 59m 07s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0cec231f025645a89a5f605cb41baaa0 : SUCCESS in 1h 41m 29s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/eec3b8c0ca75477a98f18c36802c8ab8 : FAILURE in 2h 11m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6df3767c4f144af9ad0c7e1976e8cdc1 : TIMED_OUT in 3h 03m 58s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0a140c01241d4e508ae76bcfbca475b9 : FAILURE in 1h 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"128e704b85f7e46dfd6ade96aa5bff0da522c136","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-08 01:37:32.000000000","message":"Patch Set 1:\n\nlooks like lots of pip package version failures?","accounts_in_message":[],"_revision_number":1},{"id":"20132d75643d33d32e110ef3b734180a379ad2ff","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-12-08 09:52:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nrecheck\n\n(+1 waiting for CI)","accounts_in_message":[],"_revision_number":1},{"id":"69b71e55ebef70c7f5fe3021b348a4b502804a7d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-12-08 10:28:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c003c1f56c780fa48951cc298e8baea282c6139a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-08 11:01:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47e56ac634c80a69c7fb0aee53096350e07ce807","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-08 11:04:05.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI took the liberty of doing the mods Kuba was mentioning. Hope that is ok.","accounts_in_message":[],"_revision_number":2},{"id":"827d91842df02a95e25cf9a7c4b178b6a9220a4a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-08 13:11:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/2 : FAILURE in 2h 08m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e6ed8ad267b3b38964ef69ef4cbec3bd42f4d42e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-12-08 14:26:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3f5d36db44a4768be131d1b6511923849bd0c72","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-08 14:53:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c5541f1d3f4efa45c10f897009f6b095e36ebf5c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-12-08 15:45:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97d9327b95baf60dbe26a8194d426993faf7487b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 16:00:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b54740eb388244b788628ab255d500b1 : SUCCESS in 1h 03m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b7d94c6359c41f19195c02f521d3478 : SUCCESS in 32m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/56e46b27ec434a788ab378c9dcb380a7 : SUCCESS in 1h 34m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/12bdb9c111e3478ba5d5da3154052f5c : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95099c1e064740a1b1dfecb14bf2db22 : SUCCESS in 1h 27m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be19ae9046b14345925d76069f5ddc69 : SUCCESS in 1h 31m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da17ffff00074f9eab3b9eaee8f8f526 : FAILURE in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddf4603f0c7b48d78acf521319bc286c : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523ac6edee2a4a8ca892bb69de03c477 : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8646495d40cb4b498f95629848a152ad : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c4af4a0145c4b5c84aba7b58f5e18fe : FAILURE in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912fbabae33a44c1826eb017cb2918d2 : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17958f6f83c8407b8c5fba58499235d7 : FAILURE in 12m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f2f4ec1b8744f599ae0fa68d52b348a : FAILURE in 19m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b70cae9974104aefbac5bf3ce6260ae0 : SUCCESS in 1h 39m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b451135157941ed9d3c18991c76b052 : SUCCESS in 57m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/792b88a805414080b5d9ec098c7dcc69 : SUCCESS in 59m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/84d870704cd545ba8ddd168eb737f14e : FAILURE in 1h 01m 42s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2eca5d43eada4abfabc2cf3fce9cbc8b : SUCCESS in 47m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9dd6df212af34683b552bd9e1dacd3e7 : FAILURE in 19m 41s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3c924b7a140477599755d117d05c58d : SUCCESS in 1h 31m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e98800877ce14383bdad72dbe0214e1d : SUCCESS in 1h 15m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/966e4c33893b480abc49157f03b142bb : SUCCESS in 2h 21m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91e1fb4fbfd849268b1e8ed33dd00012 : FAILURE in 16m 23s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b23b2c515cac4a02a225d040c858097c : FAILURE in 15m 54s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/79c0c7ac6a7b4749a2c77f4c5a0e1cdf : SUCCESS in 1h 11m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a495e1cad943495bbe2990fbf7d47ee1 : SUCCESS in 34m 56s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/da5b8bafb60740a58b1ceb112ccd2f59 : FAILURE in 1h 02m 31s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/90aed44ba9dc4948876ad79448169fea : POST_FAILURE in 11m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/83a05896f79a4db580517fa4b519dcb8 : SUCCESS in 3h 53m 46s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/489a5aae7d5a48699390b12269c2f862 : SUCCESS in 1h 36m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a48e0e17fa32411aafff84a971183f82 : SUCCESS in 1h 05m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dae67e2d11124cac9cd63b58a21249c1 : FAILURE in 2h 02m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/feb5d4d6ef7e4f1bb06b87f39e2accbd : FAILURE in 2h 51m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ad7b87b89a3048efa4739d4c2795ab8c : FAILURE in 1h 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b4008350ce3706d325dec2f946af7bdcd59468fc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-08 21:56:35.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"18d4652a63e846a0cd261191386f4bb5cb317536","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-09 03:05:23.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5ba5fe30c81df3a9cd761ebcd2ea05041aceaf7d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-12-09 10:34:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f6b38614b341afb396ecffa8216715a0b8cce70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-09 15:39:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4d96b8f467c28dcb4a8d583995fb743888596584","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-09 17:14:15.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"93b4a76c4252b881818470232b9f1bc8dd58f2a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 21:00:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a37f4f1172a347beb26d6dd03a3014c7 : SUCCESS in 46m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a925ef034f34e66a5dfa341a7c5f920 : SUCCESS in 25m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a93aaac5c6db45dab64ab66f9ae78827 : SUCCESS in 1h 32m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6e062b7a141479db857503ffb53c417 : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4792cb65ba81479498f6a6436925b31c : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a3846bb1bd4d4703af487d6a41233bd4 : SUCCESS in 1h 43m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aec01d43f55a487fa35d94751db467fe : FAILURE in 29m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe77727081b64125b86dafa2ed707d17 : FAILURE in 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9664b1cc268745d98512ccf0fc034be5 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3460da255214f479c6d68bf6508ed56 : FAILURE in 34m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45571b58f5664e4a898bb694cba1238e : FAILURE in 29m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2732aca1ad446c592285ac3114808d1 : ERROR Failed to update project None in 13m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9261f4477d04c86a7442eec81f6f47c : ERROR Failed to update project None in 11m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/43a3d9819e2e491c8cd63cce0e140c2b : ERROR Failed to update project None in 11m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8bf6d8954cd049f99b60e5974770cbdf : SUCCESS in 1h 32m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/75ce90bff1c4450d906b0c6139f342af : SUCCESS in 1h 11m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/afa01d57357b4e86927ace15359faf52 : SUCCESS in 1h 13m 00s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/04e141560f954cbc814a7a7d853380e0 : FAILURE in 1h 00m 57s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbc589f047384f71b155ad850bee0686 : SUCCESS in 1h 22m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0d85f296cb2844f3977c0912d100baa2 : SUCCESS in 1h 19m 30s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ff70d2304dc6481492dabf8e120fe38f : ERROR Failed to update project None in 8m 37s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de5afc6a900c46d69897d79154b320ca : ERROR Failed to update project None in 8m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a0a07e2915e4a12a1af56e1e95e8369 : ERROR Failed to update project None in 1m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64486c4d85e24b918af47eb88979133a : ERROR Failed to update project None in 8m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/661933a9605b4a04b626603a2bed0e89 : SUCCESS in 1h 05m 11s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a5d2671eb6fe4f13bd4ca640f3439493 : SUCCESS in 1h 21m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/584b52aa54d14b4588a0a2a0c10ba2aa : SUCCESS in 38m 48s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/6596acdf6929445d9edb656d28c8dd8d : FAILURE in 1h 15m 55s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/93dbc1041b0b4129874cea188ce4e829 : SUCCESS in 1h 19m 09s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cde193b56e7d485480949e04eb5ede67 : ERROR Failed to update project None in 5m 22s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37b2ab40c54744dc84a796df331ace27 : SUCCESS in 1h 35m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/af927cf2c6334f7a93128b56da35391c : FAILURE in 2h 22m 31s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0cb20078dcfa4a938bad3265be658ff2 : FAILURE in 1h 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49c0329027eacd89b3f791287482e92bd77c91e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-09 22:14:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"024bcf4db0b572af13d1429694e16dd008aa4450","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-09 22:43:25.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"74a36dea4e4bdf2cbe172b2a88b7d4a1763e09bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-09 22:57:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bf42881fa24ad3a39e5774921205df4890864688","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-12-09 22:57:56.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"d04167fe027e4bafa5b47236b849ce99b1e8a661","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-09 23:10:55.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"55473e25d8f065d12e779dc5c07457000c86175c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 01:10:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/5 : SUCCESS in 1h 57m 58s","accounts_in_message":[],"_revision_number":5},{"id":"4bf6ef44b9ae16a7a4de34debf9a1f12f774f447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:00:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/32a88c055741443d9321f1d169e62c94 : ERROR Failed to update project None in 1h 09m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9be22e353c241ce9bee1d5c134c3806 : ERROR Failed to update project None in 1h 49m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0578589bf5e74c868f677a71fdc6cfb4 : ERROR Failed to update project None in 2h 29m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/de3104db49e94bd291ab4a28101722e2 : ERROR Failed to update project None in 1h 47m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b441d42caef04317852c978c6e70c9cf : SUCCESS in 1h 17m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/96dc479121014084a6402b1536a99f7f : SUCCESS in 1h 50m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/754a8ef3b7b74df29afe68bed197874d : SUCCESS in 32m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4464a3d79d342b5b5c024cf0fbeeefc : SUCCESS in 33m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2807294272854730804c44cae2dd346a : SUCCESS in 15m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79057aaa312e4e1e9bac88218d01f2b0 : SUCCESS in 26m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1d69e92942b4bb68b563b180a453c2a : SUCCESS in 27m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2273d1926e143fca3aa48fa494a524c : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb4c2bbd6b1545a980224efe3057b6c6 : SUCCESS in 1h 15m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d88878c9455744f1a7dc97d82aa5fc9c : SUCCESS in 1h 03m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a55350b7b7fe4499a7961c3c08f25f30 : SUCCESS in 1h 52m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebd441b3a87545d38372aec1d279d32b : ERROR Failed to update project None in 1h 44m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/62eec203ac6d4b58afd42a9794e40d62 : ERROR Failed to update project None in 2h 10m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a7eab9edbd4a447e8c5393cd2ef095b5 : ERROR Failed to update project None in 45m 18s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fc4a95fca384826a33250d961a1e8d4 : SUCCESS in 1h 08m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e87ee3ca42174764923cda15aa1c4751 : ERROR Failed to update project None in 1h 08m 37s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/703bb630d9fc4e84aa30acb8b42a2984 : ERROR Failed to update project None in 2h 10m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c99261fe8a09407093eac960994eec51 : ERROR Failed to update project None in 1h 09m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc09741dbe674521b7f237376001a881 : ERROR Failed to update project None in 2h 29m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1b3cdb05d90478e9e1092e26d0cddce : SUCCESS in 1h 08m 02s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0be13d8e64cc4795a45ede3e21753fd0 : ERROR Failed to update project None in 1h 44m 02s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/59ba453c23f34c3cb49c8c4c498ccdb7 : SUCCESS in 1h 02m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c6fcb9aa3a2444649f541b0e14c56df8 : ERROR Failed to update project None in 1h 08m 37s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/fd1f46f45f5645d9adb1d2a1903482c5 : ERROR Failed to update project None in 47m 07s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7d892c165b7f4b64bcbf9431eebc47a3 : SUCCESS in 1h 07m 33s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/95f3dbfb383a4e4ebe2b70625c4220c4 : RETRY_LIMIT in 2h 20m 04s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/062ebe1d948b45d8953a932bd5b27999 : FAILURE in 32m 28s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/424f8fa5c76c454389dde68923360701 : SUCCESS in 1h 11m 39s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/18140e23f30447298b9406c31ab80605 : FAILURE in 1h 12m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6710461e8a1046a6a7caf5f0a9bc19db : ERROR Failed to update project None in 45m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/cd314f7dfee541ec8d07b8d21d851769 : ERROR Failed to update project None in 47m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3831709037413aabf4e5ebc54e49a62fc866d879","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-10 13:37:37.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3e4e6889b46a9edf059003c752a0a159857dc0e6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-10 14:30:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e1187280b4a16d5f70f8dd056318b090b966aaa","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-10 14:31:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e835714c3066d22a6d01b629d4640ab63da4d1ad","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-12-10 14:33:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a3b0d5033668fa4379ff7a90ca17adeaad0d663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-10 15:24:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1eaa71c77c3f72804106983b66f35711535f35d3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-12-10 15:26:10.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7189c41eb418224e049a7d384698a19c96a17454","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-12-10 15:41:04.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks Terry, waiting on CI before +2\u0027ing","accounts_in_message":[],"_revision_number":6},{"id":"4297b171c02deb6567373f5504f0de11e9b0bd52","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-10 15:43:03.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6b7c6108d3c3f0df8c56f0f3927003e8ca5224be","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 15:55:02.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/6 : FAILURE in 29m 54s","accounts_in_message":[],"_revision_number":6},{"id":"84a92344af9f4612a759df5b2a981c28438129e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:13:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c4f129e7b0e541d999b674240e3e3d4f : SUCCESS in 49m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f8060fa0aa74a1d915839354b5e58d4 : SUCCESS in 32m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6e0062cc18c84d93a2924f450e12d683 : ERROR Failed to update project None in 3s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bfb8164679354081b3747adaf52ebed3 : ERROR Failed to update project None in 2s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/939cdb54468a4f738fc1ef71a400cfd6 : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d87cb2b48e214d8aa260565f14fbe2d8 : ERROR Failed to update project None in 3s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e488bcc2ebe943e79eaa629f62653f29 : SUCCESS in 39m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad795acf36844295b27fa47d6613d245 : SUCCESS in 32m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872cb29a18854eb489d25348d9c30d2f : SUCCESS in 15m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a603af3b883449168240e365b583ed93 : SUCCESS in 29m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13c2adfcf43c419e9caac9691c65dab4 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5526c839344b67b795e30e4a33155c : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6125c231985848c8ab94fbda39812332 : SUCCESS in 1h 17m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67a2d1abd75c4099aab1c8db200bdfd0 : SUCCESS in 1h 14m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5ffd01f8b9b94639b4255e648beb94e2 : SUCCESS in 1h 26m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edb22cc808784147bcefc33163cd8142 : ERROR Failed to update project None in 2s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dae72524cb93477e9638dc0e0d0dd532 : SUCCESS in 1h 08m 21s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/96df7fa941f04ee2a0902ecbdff6549d : FAILURE in 1h 06m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d5084832706a4438b4372bece0a41731 : SUCCESS in 1h 33m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8a05203242440238789e81cf2ea47bb : FAILURE in 1h 31m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/99874b7769e54ac886863d5bf6eb1be8 : SUCCESS in 1h 18m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/923f28f521ee40ac957e0e7a1bf95242 : SUCCESS in 2h 22m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf64d9af55cc40b1822da62154ca03c2 : SUCCESS in 1h 00m 04s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6cb05f0ec96141389188f08c43fb0a41 : SUCCESS in 1h 11m 32s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1cb132f9e9244e71b64543519d7d085e : SUCCESS in 1h 13m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dc478b92315a4428b90ca37896aa7e56 : SUCCESS in 42m 32s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/e5dccf9180a64af1829d58de89e7de35 : FAILURE in 57m 17s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d626e40156c0406f86defc94048d98e1 : SUCCESS in 1h 18m 24s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/593f2d7dc6e24ef492f2a4ae58e8cddb : SUCCESS in 3h 41m 49s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b07cdb3825b64c4cbaa39189e47ee486 : FAILURE in 41m 01s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4a41010bbc548e681f25193550ce382 : ERROR Failed to update project None in 3s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a1ab4705944e42da9373ae8bca1cbbf5 : FAILURE in 2h 14m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c45ac6510b4f4d91a5a266a8522823a3 : FAILURE in 2h 56m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/47b32eb62d8a4024b944aa41e138d6b7 : FAILURE in 1h 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f37475847e55d71db7e8c628f465c0545aaa7685","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-10 22:26:55.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"7539c3f6ca56b14a238646bcfbc153580de79607","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-14 10:20:33.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0821fcd08c547fc648c724f30cccc8dddd7fdad7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-14 12:15:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/6 : SUCCESS in 1h 54m 30s","accounts_in_message":[],"_revision_number":6},{"id":"fcba9d63230c895aa6453d5a62759d9c70a9bd31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 14:55:31.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/87c556be38124ddd90d38a9094651ea6 : SUCCESS in 52m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1dd64dd08cde401ebc5e292043e5acf0 : SUCCESS in 49m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d7fb0c511b1d4c68ba8a68d89aabf78d : SUCCESS in 1h 48m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ef84a00c9a84dd2a261fde05524d60c : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78cfc974c6ba40ea90354db0e17e2dc0 : SUCCESS in 1h 30m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/83d25d0a167e4c8bb2fbf2a881fc3bcf : SUCCESS in 1h 18m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72990b9987c440ebfd2ea2c6b3b2a2c : SUCCESS in 31m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a23c2ffc3284bf98a6bfed29cc94dc0 : SUCCESS in 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349b993f036341b1b6cb9507d42df2f0 : SUCCESS in 12m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c315a2f1e71a4c9cbc2e71812b031add : SUCCESS in 27m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3785097c2147499b89f3ac5a60b1168e : SUCCESS in 26m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26a090c6e9eb4493b805a6a96d68f8c9 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9741897465a74a17b17db09b6595dbdf : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62971a391d2a42e5af07289d4fbedc26 : SUCCESS in 1h 09m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/341c42e12442493f8e2fa985ce73dc63 : SUCCESS in 1h 17m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/05ca599f6bdc4f888fc8e9df85c633ee : SUCCESS in 1h 27m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3055f6256d1a413394b8da9207cd5d2e : SUCCESS in 54m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66cbeeeca3f54601aaa44c286cdcf562 : SUCCESS in 59m 08s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/068df752d0064fb4a6f69d9fb80cf701 : FAILURE in 46m 48s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a772dff9549f4d5e83491a053d13246e : SUCCESS in 1h 22m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f5211653c404e619cd1c82380fefbbd : FAILURE in 1h 36m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/75b2aea2dca64f06a4ac9a957afb840a : SUCCESS in 1h 17m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9628ad00e7841baada850855dc5f859 : FAILURE in 2h 00m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90ebde17338946b1b98670fb02d3c496 : SUCCESS in 1h 16m 56s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b1f28a63d4774483a89b18996b5997a6 : SUCCESS in 1h 00m 41s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/196d0f0b43b64e21a96e3bf99c24369a : SUCCESS in 1h 22m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/edb0dda8e05d43f49c22078758df640b : SUCCESS in 32m 21s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/4417e069798c41b79e3b4127d9820e07 : FAILURE in 1h 04m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/afe0a6d52c5644fca8b4ed4877168ed8 : SUCCESS in 1h 19m 40s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/922d10db197b43cbaad7d9cbd2169865 : SUCCESS in 3h 20m 10s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fc0ae9fb1b394078aeac704af6775f90 : SUCCESS in 1h 14m 28s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e6a2b136e1554aa086692b446b03d217 : SUCCESS in 1h 37m 48s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c631ca7223a14da5ba5d604efcf0d5be : SUCCESS in 2h 02m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6f75d94193f74618acea67421550b7f3 : FAILURE in 3h 04m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1b6cab58f3134b21a41df03a94c61645 : POST_FAILURE in 1h 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ffdef9f91532841c09b12e3c8e8cb7baae36b458","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-14 14:58:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"515843e2ad103b91e9d8e3c04d2fe2280a95040c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-14 18:01:04.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/6 : FAILURE in 3h 01m 30s","accounts_in_message":[],"_revision_number":6},{"id":"0704a60b51c591b1262adbb8e34ddebec10d4f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 19:55:31.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/53939b96e3474ad2bbff23932e75cff3 : SUCCESS in 59m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0830262e6d3431592f708a128bc63d5 : SUCCESS in 34m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d7475b85388b4c17969290d1cc53700e : SUCCESS in 1h 58m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/63580eea858f47b78508099f35df015d : SUCCESS in 2h 06m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/752adbdf01454a4aa884daa853aebe37 : TIMED_OUT in 2h 51m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/39e161e83bd04a709cbf4799e9f3b39d : SUCCESS in 1h 48m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/939d2a65ff43431fb4f4ec5ad36ba09d : SUCCESS in 32m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bbd505c5c014d84b85cec14ab827b59 : SUCCESS in 31m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d4e7dce23e2433f9558b0aaed08d343 : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62d04e7fb70340d28920e1f8c5a6c730 : SUCCESS in 26m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eb7731024d349518124acdbec411685 : SUCCESS in 26m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aefe072c75e4300bed03a4b4965c5cf : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f11668bff043e68265afaf9a6751db : SUCCESS in 8m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e116038ad54d44e09fb005d99579e112 : SUCCESS in 1h 09m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7137356e7814b2ca1fedace57e6380f : SUCCESS in 1h 28m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/97f4514da57045c1bb8c4b496e72e568 : SUCCESS in 1h 29m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68dbc06ab3364f07917e8e433279da84 : SUCCESS in 1h 04m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd57397646ff4b0598a9160fa1976316 : SUCCESS in 52m 44s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/b0fae0ff3bd64f39bd3395cde9735fbe : FAILURE in 47m 57s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/70b1bf48e3ec4a33b482fabab8142876 : SUCCESS in 1h 17m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/650606ce24ee460c861e73437735810d : FAILURE in 1h 51m 28s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3696d3100d504e6285d269c531a9a24f : SUCCESS in 1h 26m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b87fa4fd9ccd45f08a8febd76f5388a9 : SUCCESS in 2h 01m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20515bb4e83f414fb3df671b8b6cd89f : SUCCESS in 1h 31m 38s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/411c7342c3e84401be3078ba4a03a1dd : SUCCESS in 1h 20m 30s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d59ff52400d04cbe96e638f53f90c00a : SUCCESS in 1h 12m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8c60882d314c4b2aa88d60d30dac3d12 : SUCCESS in 37m 45s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/a24794df201c46219ec5a5f4f920424d : FAILURE in 58m 26s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/870c20a351cc45aea14ba49288f8a5f8 : SUCCESS in 1h 20m 35s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/552ac03591c040c5aa2a67ab66cd4738 : SUCCESS in 3h 29m 44s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6441595595ba4250be4e115ce26370d3 : SUCCESS in 1h 13m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06e1fcfe04054029863788bbf916b3c2 : SUCCESS in 1h 34m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/df19752035ca483e8923614fd2bbbe02 : SUCCESS in 1h 55m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/846d5eb8fb46448290fd092658654440 : FAILURE in 2h 58m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c97980228ed144fcb28dd15f0b6447c7 : FAILURE in 2h 07m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b6eeb8bb351233f42b3db7f83b175f8810eba1d3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-14 20:10:16.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5914a7df6a30de1b533d2f8e9bf709dc840084e7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-14 22:13:31.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/6 : FAILURE in 2h 02m 38s","accounts_in_message":[],"_revision_number":6},{"id":"afd8a1f7f94366efb333ff9c421d67d61e7c1714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 01:01:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e07d4b321406484bb9faece9d908959f : SUCCESS in 57m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c95f021c5cf24992b4fcec3dacd12067 : SUCCESS in 39m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9d20b64fb79e41c3a226f60a2c4d2bf5 : SUCCESS in 1h 25m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8c427cff0c3c4990995a21c94228fbfb : SUCCESS in 1h 57m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f30006b233804544a89f2ab404360a01 : SUCCESS in 1h 24m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb9e8187560a4d828f3241f4ce15fc44 : SUCCESS in 1h 26m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46ed823ca13840d98e38d129518b8b5c : SUCCESS in 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e929581f216a4f04b54c65df00baf860 : SUCCESS in 30m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0b8746e2b54e14b700d818f4e3716d : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71d9b97f400f4052b0d2316d63403c6a : SUCCESS in 17m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d632e4d7154245da9d1cc2187e520cad : SUCCESS in 16m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f30a268f7c314c119c824cf478d2257d : SUCCESS in 30m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adab78f4adf249158ac22a438f941975 : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/215ca11bca40402b866f1f726b6fc790 : SUCCESS in 1h 14m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa95e48bb42341858255b56d5e135709 : FAILURE in 1h 37m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/024a67613c4b475489ce9950326dbc5d : SUCCESS in 1h 37m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7912a6d2624442629b3f7b99bfd43b02 : SUCCESS in 1h 02m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89b378b7492e487aab2c3e73c95a4e23 : SUCCESS in 59m 29s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/1843e5fdb7f34e67a6011cedae5e3af9 : FAILURE in 48m 36s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/822b399ce1f34cff8447b2727aa34d82 : FAILURE in 1h 11m 09s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d0793489a3fb4ca9a1c627495ee00147 : FAILURE in 1h 22m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/54d993a22cd941d6ace46b3ebcaf63c5 : SUCCESS in 1h 14m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a684dc3382dd427195615eda3c9e573b : SUCCESS in 1h 58m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1048bdb26d9f43e38d2adf31dac3d6b2 : SUCCESS in 1h 03m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b503255e14c946f1b7c982b04cca9371 : SUCCESS in 1h 01m 19s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/84d0e26447f9478e8317dbd501077222 : SUCCESS in 1h 05m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c4b646eaf7304b6f84672416284619b6 : SUCCESS in 34m 50s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/a03178e606094bb382244f8607047d27 : FAILURE in 1h 02m 07s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/90c5f329aaa44b9483979d54124abb56 : SUCCESS in 1h 16m 09s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1afd32fcd955435b81e77e118eeddab8 : SUCCESS in 4h 44m 36s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0683779ac8f14fe5bbf4724f04e3700f : FAILURE in 2h 51m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/78a991c0a0334b92811b488ade30234e : SUCCESS in 1h 36m 06s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/af0bec2203ba4e4788274dc1d4d06000 : SUCCESS in 2h 00m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0ea8390258c5460cad6b063a67a24d27 : FAILURE in 2h 41m 40s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/174e223aacf3448a9a87f6c6355aeef7 : FAILURE in 1h 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"79ecc65bfeca2f05846c6c3baaa0be01fea6f4a5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-15 11:19:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8bcf417849edf15d418b2b42413a01cd1fe71b4d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 14:12:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/6 : SUCCESS in 2h 09m 43s","accounts_in_message":[],"_revision_number":6},{"id":"0c6cf67b43a2f5911f23e2a7292909700fefa580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 18:27:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/39257e6efdab4c5da04d482e6cb0e7ef : SUCCESS in 49m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a97160867e7545369f0d52ae1975aae4 : SUCCESS in 33m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fab036951e894738bafcb7262fb14110 : SUCCESS in 2h 00m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d427fcc15134bbd96d12954b693b8c7 : SUCCESS in 1h 20m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20dc9120d3d842308e8f2c72f1f2b331 : SUCCESS in 1h 21m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/64ca8dcdbc37456bb4cd6dcb1634c277 : SUCCESS in 1h 06m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25e277a2ba1b4113969e7d3481061f22 : SUCCESS in 23m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7968d13ea2c0453c8aa4ad2968bba557 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10a6a6596a80483488e73ca8bc59075e : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e44753c60a584687af088ae4f7581859 : SUCCESS in 27m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4e6733b64ee42198df0e3c1869ae2d6 : SUCCESS in 35m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c88c9292695843b590b6ac1e751c6faf : SUCCESS in 35m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a7c4ffe89745f180ed1c8075b0afb7 : SUCCESS in 8m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e20c509bf9c4c79b36ee08d49d22911 : SUCCESS in 1h 15m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a514c4463074357a868401707f45dbc : SUCCESS in 56m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7c7ab1e98fe34626874e335cfd368ff1 : SUCCESS in 1h 39m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61eb7dd3e9dc4abaa6650f5920904638 : SUCCESS in 1h 02m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/29c8c3755ced45aab2bc9814f3dfb233 : SUCCESS in 1h 07m 17s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8afcf91f2dc740d7b8eb3cb775ece412 : FAILURE in 1h 00m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/755647c6d27b44b9b7289a37dc575b14 : SUCCESS in 1h 21m 59s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea215425f3a6421da75c0dc9522e15ee : FAILURE in 1h 39m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3551cbbe69ad4e37afbb3000f0dfd57c : SUCCESS in 1h 11m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e1ff30678d948538045a88a1d851131 : SUCCESS in 2h 13m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3683a8a265e45cd8b4167489c3c0f6b : FAILURE in 1h 02m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1972d80391f847338bb2ccd69d83d9dd : SUCCESS in 1h 01m 20s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ec0d24ee4e0445b9a05e44ec137b8008 : SUCCESS in 1h 10m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/447eb5ef13b64028a3e13b059f256857 : SUCCESS in 39m 41s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/c7aa7c02fdf9455f867bae524e53f9c4 : FAILURE in 1h 07m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ea4d92c2042f4eee87de188a1647be56 : SUCCESS in 1h 22m 31s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cd9138a3b646451eac9c43519b5cc72f : SUCCESS in 4h 33m 28s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0e66e4ac432143b3952a981e6a254853 : SUCCESS in 1h 30m 26s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05c2304a6ccd4cfeb1d0eebb1f6cafe1 : SUCCESS in 1h 34m 38s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f12f9044654945039124b4f887af3aff : SUCCESS in 2h 29m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/456cfbfde562409f973ed14641d3a49c : TIMED_OUT in 3h 03m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b8436446742a447fa41d78e8e07fdb99 : FAILURE in 1h 44m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"27d460e45c5803419c6b4e33273496652f90233d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-15 19:41:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c78ac51af4b6c2e70cc66b7813264418ac1049ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-15 20:19:29.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"5554559b24f1a8fffb47d874e6675e3098d8f426","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-15 22:13:08.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/7 : FAILURE in 1h 52m 30s","accounts_in_message":[],"_revision_number":7},{"id":"93ae3991b5f3231f39e28f2009938fb1d5597597","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-16 04:46:38.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"f8ea7bf193c731ee1f4873e818e3b612059a6a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:30:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0559ece3798242a1907d08675e7b3b76 : SUCCESS in 47m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a48c11747b304ebca24ad221fb873060 : SUCCESS in 30m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4c2f2226182a4f0d9aba55cef5788bef : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/78b99596b8e34e9581c2a2977b2cfae3 : SUCCESS in 2h 07m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0514d0b64f824252889ee9ab070930c6 : FAILURE in 1h 26m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0d9220bee8ba4208a89c34214e32a791 : SUCCESS in 1h 52m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ed2ed9f16f748debd2770219192edea : SUCCESS in 30m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/072becd3cf1a4beb96834c0e9445f4cd : SUCCESS in 35m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c227a755b0e4cc09db7aa162fd92f5e : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8d2378c46e54e3d9f41ccbf2dac29b1 : SUCCESS in 25m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1ee23e6b2349a29337cb11f88ab5e1 : SUCCESS in 23m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97ad3654e765447a888e29722fcdec7d : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ab8d6a3b484b7c885d8931d4e38aef : SUCCESS in 8m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f238731b44e44dafab40b7bf7420ab10 : SUCCESS in 1h 17m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e26f4f9ae4841049c432c7c100da44f : SUCCESS in 1h 07m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d325aca7f51e4bb4a1f0ccdd34b517b8 : SUCCESS in 1h 42m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8613e6c19b374882beec392fd292fa03 : SUCCESS in 57m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b848c9c0b79f4d9b8bedfa2970a89423 : SUCCESS in 1h 02m 18s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a471efda915543df80b9d90c9eb0d96b : FAILURE in 57m 52s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9ec9ce9dcba846d983c8f2b5687ed680 : SUCCESS in 1h 15m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7725c512ea054608b3a4f60ff3750b44 : FAILURE in 1h 36m 32s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4b2474cf6e8348df804855593240f428 : SUCCESS in 1h 07m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4a838fcc95b4326adcf9dbd013ed766 : FAILURE in 2h 01m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f90821c4221e481090619f6995719c8d : SUCCESS in 1h 14m 53s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/84545e3bf9874c46b9e2ff9b59867654 : SUCCESS in 1h 07m 39s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce69e2c9cfb64241a393e3804edfea5a : SUCCESS in 46m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e9e00bf0fac6488b8f17257372fe9469 : SUCCESS in 45m 24s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/ce7f4bf7c26b4680b3540fe8e22a54b2 : FAILURE in 1h 04m 22s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b584ffc1478042d68d452d81451d6411 : SUCCESS in 1h 27m 49s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8290205342f64c0a98f2e6129ed8544c : SUCCESS in 4h 01m 33s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b30c2a3d296d4c29bf1573fd5598843a : SUCCESS in 1h 52m 35s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a5645711eb3c4468a2c50fffec912e3d : SUCCESS in 1h 34m 05s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1468efe631d042968d425b7e4e04c56e : SUCCESS in 2h 20m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/96eba444be51467ebd49a94a5d06be68 : FAILURE in 2h 43m 41s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fa575f4de69c4bdd8087219ebdc5697e : FAILURE in 1h 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"efb96d64649d871e8ece35fd36ae097a2befc838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-16 18:39:17.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"21789ed160c67b74c523f2942cc5e7afeb89d18c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-16 19:58:55.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"434a0af038e27d30778b8b0513c53fc00de5d1a2","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2020-12-16 19:58:59.000000000","message":"Patch Set 8:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":8},{"id":"aba835e77c7130e7f66537074947070f28bbbc30","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-16 20:12:52.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"29ad9daf3fe6198e26503d2f761688325d7f8bda","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-16 21:20:43.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"388e3029012f0a1a3cad4f10d6d351078c45c962","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-16 21:47:17.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/9 : SUCCESS in 1h 46m 59s","accounts_in_message":[],"_revision_number":9},{"id":"41229e226adca9aa8e94634bf161949e4ebd9869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 06:41:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/369c95eff92b4da18ba81e60b3cbf9e4 : SUCCESS in 49m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1c5d9a66ddd43b781cfbfb487b0e215 : SUCCESS in 27m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c783cbb1d8d14ff6abb22765a03c4379 : SUCCESS in 1h 47m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5c7d5b10f6d64b77be6dc72cdc582bc3 : SUCCESS in 1h 39m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e165fa653ab04f48bedce7babc33117f : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/af1bbceb104843c7a290e83fc767a497 : SUCCESS in 1h 33m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e017c9ac19b14ae1970c96aca99c17c2 : SUCCESS in 32m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f7f2765ca1c40d789ba910bc6064a38 : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e14e4d9ae374d64add4733fccd6cea5 : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af0d74bbbcb348c3ab472d1423462c92 : SUCCESS in 26m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c90eb4250ed145a497f228fd9fe2b1a3 : SUCCESS in 26m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa8f9bb791b543a3a6044b2e120555b2 : TIMED_OUT in 40m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69b11cfb4dc40e68508359f9878ac7f : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98a73341fb3a4dee8779ea3f7bf55f6c : SUCCESS in 1h 13m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fad436d2d2c44e8abcd947c356723447 : FAILURE in 1h 03m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0d1dbf1364444b389b7639ac6a7b6c5a : SUCCESS in 1h 31m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d329e0ea4c294287be8ff90e95ee00b0 : SUCCESS in 57m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/62f4c9e43ed24775b4b16febc0ed479f : SUCCESS in 1h 09m 39s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/2fe4131b128a4f2db45559ffb4d58a6a : FAILURE in 1h 03m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6173e71ef2dd47cbad3b8eaa1b2f6cf2 : SUCCESS in 1h 10m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4668d5230c3b412a8d20fb5d323fdcb9 : FAILURE in 1h 44m 25s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e253338b99e049a585692ff619a5d4dc : SUCCESS in 1h 09m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4065fac6d8c46ce8ac3e10b3493a7e3 : SUCCESS in 2h 13m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27768938fc634312b22d0a5ca6327abb : SUCCESS in 1h 14m 30s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d3ba551b35ec42f0a59696f191df611f : SUCCESS in 1h 07m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/62b1c21abf6243e49241d99ca8813f4c : SUCCESS in 1h 07m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/93eec8029af840b7b21102dc0700a68a : SUCCESS in 44m 54s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/59c5ebb3449d4d4baed6131197bb4164 : FAILURE in 1h 02m 33s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/052aaec30333411c984381848ae9058d : SUCCESS in 1h 25m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8a3ea4733b2d4931a36b62b51def77ce : SUCCESS in 3h 50m 33s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/04f9831e069342a0ba8741a814f89224 : SUCCESS in 1h 35m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6f8a48a81f7f48efb00cbe264e023235 : FAILURE in 52m 44s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/531ed60a538245f7a2ce04fbedcffb20 : SUCCESS in 2h 07m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cfdab48039bf44b9a3a3eb099aa84e5d : FAILURE in 2h 50m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/31560cb58de94dacab8df91c2db10952 : FAILURE in 1h 45m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1684d7043cde056f6d281dbf0f2e939ca77cc459","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-17 09:02:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4a84a66a6cbe79b308ed7a66c10770f1652c81d0","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-12-17 09:13:07.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dde923c17ddb960dee5a0e8197135a48952983d3","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-12-17 09:13:33.000000000","message":"Patch Set 9: -Code-Review","accounts_in_message":[],"_revision_number":9},{"id":"95a3aa7b1f726c0d67dd4d613151507ace063eeb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-17 09:23:29.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"849349c995b5c78b96a942583ca095afc5d012a0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-17 16:04:42.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ed8b70bc8a327f8537c86cfcbcb2cf7151ccb496","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-17 18:05:18.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/9 : SUCCESS in 1h 57m 49s","accounts_in_message":[],"_revision_number":9},{"id":"9f3375a5678fa12ec8a14fffb6fa86f8458d8e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:17:14.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b3205bbecb74a43b99d9ca1a3f360ce : SUCCESS in 52m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f99026bfad44444392da1cd85cf3ea2a : SUCCESS in 34m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8bee759d5b7a4c378ac5c4ceec737967 : SUCCESS in 1h 49m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ce2afc108096497cb41dd6743241f19f : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb3b9060e03d446984bf8475c9403306 : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c1fdccdddaf94618864efdee9990ff5a : SUCCESS in 1h 24m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/195c2fd18ca24d799b9ef6a9d852d0f5 : SUCCESS in 32m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36b308c480284d68a0940f54b4ceb465 : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c3210bb6a94d7593f8263beb1ab50b : SUCCESS in 12m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a53fec5bd4394419ad1f70c76a173451 : SUCCESS in 27m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e18c5ae190654aeab271640a88103257 : SUCCESS in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee4e4a7fd7a74873ac9370210830be46 : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0faeeb9de04ec69b980056fe2f8b1c : SUCCESS in 9m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b6f0c0516984d2fbb1f72568ae8e3ef : TIMED_OUT in 2h 16m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4550bd3c67c343bc8453da95b560ebc2 : SUCCESS in 1h 17m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5b10698e44e464d8d78513feae3c3dc : SUCCESS in 1h 41m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa70c052ac4543e8bbfa2b8ea80b6398 : SUCCESS in 1h 04m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52232c9bf1634c0da9601277f7e7d5b1 : SUCCESS in 1h 09m 31s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7c6b088ffa7f4226a50fc2bed47ce4a6 : FAILURE in 53m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3b4c1640ffb14743a5ee429573997567 : SUCCESS in 1h 12m 47s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa64ffdf40ec4e1fab7cd4143558c38e : FAILURE in 1h 20m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/878ff88b2b7d489ab9a4af1d6af75014 : SUCCESS in 1h 08m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e04ca498062409fb6605c2657579ae1 : SUCCESS in 2h 02m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e6efbd0723a4f279123087b79188f3c : SUCCESS in 1h 02m 46s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a8c1ba5522d240fca542e7e7b214eab3 : SUCCESS in 1h 12m 30s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e31653da5d4743b4acaa5346eb83b815 : SUCCESS in 1h 02m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5c81b114d9f64d5588fed24acfbe865f : SUCCESS in 51m 49s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/387a685274f54781971937cea55a87e1 : FAILURE in 58m 47s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5fffff6472cc479683f73dcbecad3a19 : SUCCESS in 1h 24m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8231f4c00e234e4890fe352fb0ea6855 : SUCCESS in 3h 29m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/176297a401d346ba84346fd365afd581 : SUCCESS in 1h 07m 42s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21f6b5bf979d46fab518b2d62ae95856 : SUCCESS in 1h 38m 53s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ff7fe32b59b54697b61a188ee92c519d : SUCCESS in 1h 43m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b35bbef77da24aefa272efcb32e4c7b7 : FAILURE in 2h 18m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e935707201594971ab109e5f4cb6dee8 : FAILURE in 1h 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1527cc04e358e6903e87d1c748bcbc0bb99e98f2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-18 09:57:52.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e0fd1b608a9e75a328deb9701d3a2bbc146f784f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-18 11:46:30.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/9 : SUCCESS in 1h 47m 56s","accounts_in_message":[],"_revision_number":9},{"id":"fe6929c1deeeb59dedd70ecd26e6aba9b97a8c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:27:11.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b17057113e234b3d8368ab8b34577a87 : SUCCESS in 47m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a922b3c83db440b8828323c82905ec3d : SUCCESS in 34m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c506bd01965c43978131aa29168fb078 : SUCCESS in 1h 54m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/45e5ebd0313e455eb1d7793c343a1d21 : SUCCESS in 1h 26m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28c968e198824f8882e6f102366cc010 : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/37a778e5f9e04a5eb69866d2a1510da3 : SUCCESS in 1h 23m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/448cab588c9c49d9aada0faed8a5044d : SUCCESS in 31m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/533131e267304f31b40d4e70e4ad7cd7 : SUCCESS in 32m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b7c49246764ed4b3061d30de13d3b0 : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7de3a064432345edb1eaf1ef7014c1a5 : SUCCESS in 37m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae62b656289497e89ef098d0f3f974a : SUCCESS in 37m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30b72e2d7dc04f0fb44e9094d598c41d : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66df88c878984faca9a37244dae82787 : SUCCESS in 8m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c250db0545340e79157105e150d12b7 : SUCCESS in 1h 11m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d2fdaf4a3844cd5895ee7088784f2b5 : SUCCESS in 1h 11m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b600630b61da4814b12db519ae3d6586 : SUCCESS in 1h 31m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2af6789d74bd490984b92d36d55cf244 : SUCCESS in 1h 09m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7471c1444c646ca90f2fac94f68eaf0 : FAILURE in 56m 55s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/df753f9fe7524eab956674243040d5aa : FAILURE in 55m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/91c08082cc61429a99ebd4198cb893bc : SUCCESS in 1h 14m 49s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6a9dceff113a4e389528cdd54b01f89b : FAILURE in 1h 31m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/54f2747fcfdd4352b153cc0d990647f3 : SUCCESS in 1h 10m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85ae5d0202434458a9862e53182cc0f8 : SUCCESS in 2h 06m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/207bbd4b1d0e4311ae347290f19f2693 : SUCCESS in 57m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/1fe6af98755b472da5f7cb0bce2e1850 : FAILURE in 2h 52m 05s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f4dfc14406884c3da213d0f0ed69788b : SUCCESS in 1h 12m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/17afb28b05f74dfaaa477b6de32d3072 : SUCCESS in 40m 09s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/72b62a604c5444a6a33f7b357ca4377a : TIMED_OUT in 3h 10m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d999bbbafcfb476a842c640d6c390bd1 : TIMED_OUT in 2h 37m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/47457d85fe96455591ffac6051ec9fcb : SUCCESS in 3h 36m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/824812014a264524bc1c7c1e9dd4c7eb : SUCCESS in 1h 12m 42s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3ece3f7c72aa43d1a1c6a5f953ff6271 : SUCCESS in 1h 31m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/41426041375d48eb8dd9c46f110add71 : SUCCESS in 2h 05m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e10d15230f94dc09c3ba5f667715c77 : FAILURE in 2h 53m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/23664019e3f44bad819af36b06185979 : FAILURE in 1h 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7e0507ab5b688b0a7bfba07a716875ef1734fc30","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-18 14:49:16.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"9ccf9ee9481058dece975e750ddd4592c9a079e4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-18 16:38:58.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/9 : SUCCESS in 1h 49m 06s","accounts_in_message":[],"_revision_number":9},{"id":"0e5ac280abbca8df5780e38b31bba859dc2f0e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 18:41:17.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/94b6b368a53b4bc8a23bb204d9364364 : SUCCESS in 53m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5558b8f037324b10898108c8b0f8ebb3 : SUCCESS in 32m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4d1451f6d8f042bf8f0a6b4910a9f63f : SUCCESS in 1h 46m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c8d71cca9bbc4cedbb7e4aa1b73c8830 : SUCCESS in 1h 44m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/39e7edaf3d51451d9abd703402bd2bfc : SUCCESS in 1h 24m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ce24af81757e48c6a1a152de285c99ed : SUCCESS in 1h 34m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7219be1f3fd744be9deb4cf155df48db : SUCCESS in 27m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7a97ed57cb14c2ba2384c31322dcc72 : SUCCESS in 33m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c73b356cde4c72ac6d18149b3c083e : SUCCESS in 13m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3539bdd15e23441c86fdab0e04e58b6a : SUCCESS in 27m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5153063365e04afeabac4583e24df763 : SUCCESS in 26m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e48965533534c32b931760474f488dd : SUCCESS in 28m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432b8deb68ce4a54b3ba10977e44b7ce : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06909f4d1355451a998a251aa9b3c1bc : TIMED_OUT in 2h 15m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/03eaf74f3a4342dda8ca42ea2108be4e : SUCCESS in 1h 14m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/651a3042ae294dddb88851f74749edd4 : SUCCESS in 1h 34m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32d6df023eee470b8f1fb710edb43fc8 : SUCCESS in 1h 06m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d63caed57b9e424ba603a1baf85faf49 : SUCCESS in 59m 05s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/f8bc713fa1a142f88546b7ed07b65aab : FAILURE in 57m 04s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d2e11335f92c4e2bbd08c7cb182a04ab : SUCCESS in 1h 13m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a4e9f7847994ef7b4d15f574743bb7b : FAILURE in 1h 12m 19s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9baaa46029024fc19f7a3c56b008c430 : SUCCESS in 1h 18m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4219a77aff1495086954a1ef5e8388c : SUCCESS in 2h 06m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bf2071279034693ab225074ff764b33 : SUCCESS in 59m 01s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f0b8b14c8a4d42d5a74768430bb75f1e : SUCCESS in 1h 05m 33s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3339bbefb59e4fb1a78bf7f528f8dace : SUCCESS in 1h 14m 53s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c0b735605bee47d1a7a571811a18291e : SUCCESS in 41m 09s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/c6c07f72ffc4455c98b0c2b38c1a7330 : FAILURE in 1h 04m 39s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6180df01542e4e55b332ecc91f9f4308 : SUCCESS in 1h 16m 04s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/614a730da38a4f00b6f6e560c0ea14f3 : SUCCESS in 3h 15m 12s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3fdbf8f62c4e4bc89c13a012dbc7f300 : SUCCESS in 1h 27m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7279234aff5f400f91b3e557f414e9c6 : SUCCESS in 1h 27m 32s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fe8245b3ae58407f84c6024bf55c0f75 : SUCCESS in 1h 54m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/69041421d967499d946579b9dd076d76 : FAILURE in 2h 57m 06s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/394fcb37f88248ab8d36b4e165afb40c : FAILURE in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fc3cf6b5df7f1c52e061db6325836e52ccd72546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-12-18 22:06:38.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"fccf3f8b7d9f9a4639ea231f579282d6ae5bcf81","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-18 23:45:45.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"71f98007e4b7a830e4c142525a551075cdf08847","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-18 23:49:54.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/10 : FAILURE in 1h 42m 25s","accounts_in_message":[],"_revision_number":10},{"id":"280a2e626119d4e4abd2b68dc64344c0ef21bcf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 03:25:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/973be101e96548dbb9aeb7a91c267a8b : SUCCESS in 57m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4858e2d5c9f4455d8b7319e03db6c932 : SUCCESS in 40m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/146df5f65c1a4d8ba1d5555bed757dff : SUCCESS in 1h 40m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7181087255e4278b5a0eeb098d91e87 : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/70aa29cced6e4da8aa7a6a18744eec66 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1443e26d950341aabd76f6632d4c0cf6 : SUCCESS in 1h 24m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/556f3b486f7f4127b8a829227779b6e6 : SUCCESS in 31m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0d151af30b849e09dc37285ecf392a3 : SUCCESS in 42m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da0ec39c8674961ac9e865acc4ce913 : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53b87cec4da34d6b87b4448da0f67718 : SUCCESS in 37m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7150680016847558170a6543ae4e2fa : SUCCESS in 27m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4a00b7f365741a6b1671bfd68953d80 : SUCCESS in 25m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c11a6fc945447eba7b49a49cfaf0d0d : SUCCESS in 17m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59f145715b7b4af5903dc15e03fc1700 : SUCCESS in 1h 13m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80582b7529404cea831db79feff6b83a : SUCCESS in 1h 30m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fd8e10d867a54224ba975862a4da7bfc : SUCCESS in 1h 37m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ed905e89c1345ea9859292cb22847ec : SUCCESS in 1h 00m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f9d1d990ac44cc3a8009395cc635ca9 : SUCCESS in 1h 04m 19s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/1c7cdf9f58cd438a8bda3ce5047d8ec9 : FAILURE in 52m 53s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/95f5efb153b8462483930337316066d1 : SUCCESS in 1h 29m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/094a26f792e84a9c84a74643e2d733f9 : FAILURE in 1h 08m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9a5d92602a8d411580e4964d8c9238a8 : SUCCESS in 1h 05m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c9f6ff1dc214a8f9ea6932c50c879ed : SUCCESS in 2h 00m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae1c5b22766d405d8363cc8ffef4a86e : SUCCESS in 56m 31s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/4be007ba4a3447bca78f40dc7843ff65 : SUCCESS in 1h 07m 14s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d9918828f0e340998f7211f83ca2bb43 : SUCCESS in 1h 03m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ff9695373a1147bfb2069b9b17cd4792 : SUCCESS in 41m 10s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/9e48615948e64c619117a6ddb10b2cac : FAILURE in 1h 01m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/61177e296ca44ecbba3fc31617fd6ee2 : SUCCESS in 1h 16m 55s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7e020fa2aa56451a9d5163c1d8bb70ce : SUCCESS in 5h 02m 48s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/134ee515d0a3469db0bd53ad92b0cae8 : SUCCESS in 1h 57m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91889904a0454aee967b550ed5e4af2b : SUCCESS in 1h 34m 54s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/90d71adb8978416fa57701e0e3d3d226 : SUCCESS in 2h 01m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3a5da1af4bfb440e863b4f0d5b87c426 : FAILURE in 2h 41m 31s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/12620836aae3494ba6bf524e35873125 : FAILURE in 1h 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"31fc5664bebf662a19e2f9aa6476ac5d96bbd55a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-21 11:50:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7445d0c3d91e07a29e5061e22ba2079c45d8a57b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-12-21 14:37:57.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"ab5f4c66e7d8dd6916660fb6ec90f2668c9f6352","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-21 16:00:59.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/74/765874/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"6b3e3e9b61d4248015ea56eaa27499eaf57713ae","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-21 16:27:48.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765874/11 : FAILURE in 1h 49m 00s","accounts_in_message":[],"_revision_number":11},{"id":"7f23e719e806af904fa8b30139d4e3290a7bc6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:17:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/faa06b91c0cb470b854e03655c566d03 : SUCCESS in 1h 12m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/679006d53996494b8725c4cab8eeb559 : FAILURE in 29m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fec9a59a94ca428fad89dbb9dabdf675 : SUCCESS in 1h 50m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/015432f78dad4a63a5e9779ec579b7e9 : SUCCESS in 1h 20m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aea79241654a4fa89ba44cb045174ef1 : SUCCESS in 1h 40m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1d2ac4d169f9495ea7f63b315c9ba2b1 : SUCCESS in 1h 54m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f027fde91bf408c81c3d93b47dfe765 : SUCCESS in 38m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d93815a9b1f4ac2a75fbc2d1b200753 : SUCCESS in 31m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c312bf650a194f738260c41bd1c12e21 : SUCCESS in 10m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc778e2908c4a47ada2cdf9d8e8e72a : SUCCESS in 32m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61e7d8cb125640ef9f66fe034cdfbc0a : SUCCESS in 31m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09dfc4afe44645d8874a7edde929f2c6 : SUCCESS in 30m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb81457fba4437bb56c140feb3ee1ac : SUCCESS in 26m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2257914d8cc4b128bd79ad8de94f599 : SUCCESS in 1h 28m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/22b5bc03066342ceb1d336756cb3cd26 : SUCCESS in 1h 32m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f13417a2f69843a8b51108493ae30d36 : SUCCESS in 2h 32m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d55002719014b7d83cddf00edc6fcdf : SUCCESS in 1h 21m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60b0e0c1a6cb4321abfd371f90c5974e : SUCCESS in 1h 10m 22s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/6c3098b1dcbe4e75bd42b08cbaed32b1 : FAILURE in 58m 08s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/39d9dfe5b6b2470fab769cb41453202c : SUCCESS in 1h 13m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b1616e57a944f77a6689939d8e51c59 : FAILURE in 1h 19m 00s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/aecd1837688d45febc64e1c80ddc37bf : SUCCESS in 1h 14m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/869b456cf1bb45cdaeac4b095e52cb1d : SUCCESS in 2h 22m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bc367e33bbc4352961d1528d5287c38 : SUCCESS in 56m 37s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/642567d75261439ab4ef2c7eccaee966 : SUCCESS in 1h 00m 12s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/74253679ec534d479810b855c6079923 : SUCCESS in 1h 10m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d893b2ee1a6748c79b6fbd9488a03839 : SUCCESS in 33m 28s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/194137c1baf34365b3729f11dc8ac75a : FAILURE in 1h 02m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0aa91d2a9fa141479bf3b26faf805f40 : SUCCESS in 1h 22m 53s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ff4efb8bc13449c09eea6a38057e1ab6 : SUCCESS in 3h 34m 42s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a48d262a534f43e18c4697ed563a07f6 : SUCCESS in 1h 17m 39s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/45469776e7f646429f482a1a13fc03f4 : SUCCESS in 1h 35m 09s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bddebaa9e74345099ccbabe98ccec7dd : SUCCESS in 2h 05m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3027c210246d4ead91050a50792b2001 : FAILURE in 2h 52m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/08730ad3644d488c96ddf3ba58b0b48f : FAILURE in 1h 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d201b64bba05e18124d39f0c59456bbf3cef3b1b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-22 10:29:11.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"17bc687c88ee59b5d43a6c343f07925a478b64a8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-22 10:30:24.000000000","message":"Patch Set 11:\n\nLets check with https://review.opendev.org/c/openstack/neutron/+/766920 if it really helps for that issue with errors 500 in functional tests","accounts_in_message":[],"_revision_number":11},{"id":"f53ef6363c9e27897ebde6c5d6ba000fc43699a6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-24 15:09:59.000000000","message":"Patch Set 11: Code-Review+2\n\n\u003e Patch Set 11:\n\u003e \n\u003e Lets check with https://review.opendev.org/c/openstack/neutron/+/766920 if it really helps for that issue with errors 500 in functional tests\n\nAccording to 766920 CI execution, seems to work. I\u0027ll re-check again this patch.","accounts_in_message":[],"_revision_number":11},{"id":"567c8927e9c1a641aa5e96e518a9802a6d691b06","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-29 07:41:19.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"2fa70ec8f884b95dc35b7ece111cb6131f94c4cb","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-12-30 13:03:20.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"35c5fc036edf3d273455c91e38fdbcaea4de2752","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-04 14:21:08.000000000","message":"Patch Set 11: Workflow+1\n\nSeems like it works fine and fixes error 500 errors in functional tests :)","accounts_in_message":[],"_revision_number":11},{"id":"637f741f1aaa936376216dfe34570975160573f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:21:26.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"94c7d4e2ede9423255dfd8dc3c226f9a8c6f3426","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-04 14:21:28.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"f428a29c70dd85a1898aea7ec6d3de9b21c713c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:44:14.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3f3722bf3bd840d4abb7e98a323b6d89 : SUCCESS in 38m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/14f7fb99e0214ff4b6d824dde2511e0d : SUCCESS in 2h 07m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7610cc6f4867471190b7983149bf88b2 : SUCCESS in 2h 02m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7cccc696241e446c90cfbe1fb4c9998f : SUCCESS in 1h 45m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be5389b4cadf4216b73df4d6786c6b64 : SUCCESS in 1h 26m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/745528c9f69b4fa0ac306da46780f638 : SUCCESS in 40m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab354bb3d714892a550c9353fecaf14 : SUCCESS in 13m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe946e3a8d54441c99a7586e763cd97e : SUCCESS in 32m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c760c1b6fff4305a9126f840d529f7d : SUCCESS in 36m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f120d837fe6348d18faf1056a24262d1 : SUCCESS in 8m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cef9ebf23714615a6335d814af6d197 : FAILURE in 1h 12m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af493e8b943a41279db4153a441afea7 : SUCCESS in 1h 13m 06s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0c9adf73a99441008b9948f1abf5b20b : SUCCESS in 1h 21m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f32f0343c4ed4640a5e6af353d962020 : SUCCESS in 52m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f34c909606941eeb8433c941bf40ce5 : SUCCESS in 1h 00m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0957ca497984082be60064a011823eb : SUCCESS in 2h 13m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d2d971f93f64d2c9c74265bcdd761c1 : SUCCESS in 59m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/507e29681c1b4d47ba249028ed8c928e : SUCCESS in 1h 10m 37s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9139aaf49f94419c9144eec9d936c5bc : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":11},{"id":"63e59d70200d1440d4d3b7041ca57db45964ad06","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-04 21:07:07.000000000","message":"Patch Set 11: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"b8f06976f6dae57fd46de2e0b3536bf71fc78c8d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2021-01-04 21:07:33.000000000","message":"Removed Code-Review+1 by Flavio Fernandes \u003cflavio@flaviof.com\u003e\n","accounts_in_message":[],"_revision_number":11},{"id":"5680869cf03c793c1005f91d950d6f52afad5e80","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-04 22:57:18.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765874/11 : SUCCESS in 1h 48m 36s","accounts_in_message":[],"_revision_number":11},{"id":"c6d94a8185ca489efab8b348a2df713ddf365d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 23:06:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/edcce3d6af0f4030a2a9ef3fd6ffde38 : SUCCESS in 50m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6750af21a7f34478b43b78adf1fb1fc2 : SUCCESS in 32m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2d0914029566436599dfb43f5a1bae40 : SUCCESS in 1h 37m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/be568fde51f1483a9cc7d3150702320a : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20b726480124444fb92902d9b5718753 : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e8d5803e16be4539bed87ce0201506c1 : SUCCESS in 1h 15m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83a72b547bd849f8b75fcfd5cf504f8e : SUCCESS in 46m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbe8774caff0424289217303b684d2db : SUCCESS in 53m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452b22108d7c431b820e47d3b1b7e615 : SUCCESS in 15m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54d019bd124644bcade2e00f755f0857 : SUCCESS in 37m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/783e7469677c4e58a157c98a2473ddad : SUCCESS in 32m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f46e6e7e4b14d158adf4084dd28b1ae : TIMED_OUT in 42m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2100555dc701402eb50b9b64e82b7ca4 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a728a8de020c4576b9be8fca81b6ed24 : SUCCESS in 1h 14m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4942ad9638a4c56a20531a64b559238 : SUCCESS in 1h 12m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e5c03a6ff5cd417ebb35bbe8dc984ff6 : SUCCESS in 1h 33m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c4c999f9d1a4126b6074a34436208b9 : SUCCESS in 56m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/94342c46bc984cc48906c148956a1416 : SUCCESS in 1h 01m 32s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a1b5a64e569742d09adc83e4599a27b1 : FAILURE in 1h 01m 02s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8e1c1065a2eb453ab324706c825a13f5 : SUCCESS in 1h 37m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1eb68eb410214d319ede12da3ef14145 : FAILURE in 1h 10m 41s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/576ab9e7d24d4f61a4bbdcb4b71d7f96 : SUCCESS in 1h 13m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/846037b88f944452970deb4f06a59da6 : SUCCESS in 1h 56m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/814e1252987d4a928e9b6da3600229aa : SUCCESS in 59m 36s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/52f2dc56a50f435fafec8b2cd48a1b62 : SUCCESS in 1h 13m 08s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ab75a130db1843568d6a969abd873fe9 : SUCCESS in 1h 18m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/76a210d29a4449b1a99b5d7bbf565a8f : SUCCESS in 41m 13s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/6a05aa6ae8a74c88aa6cbaac0a623be8 : FAILURE in 1h 03m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/963aa2de6e864c34bc019da73c41f9c4 : SUCCESS in 1h 35m 39s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/80c00f5ffee54a3b8a99cf5f35e875f3 : FAILURE in 32m 35s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/717b6efeed6e41ec9a61681ad5bf47e2 : SUCCESS in 1h 42m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6a22322ceeba4c03b7446b881cf3bc27 : FAILURE in 1h 34m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ad03bc22dee5432ebc20a601d4b421e6 : FAILURE in 1h 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"190fb47fe493d19d1cab2b440b0b155cfb9de241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 23:06:49.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"80dc3b206e92a3a9dbeb282b22c28a58c3224620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 01:15:50.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4f61dabcd357434f986d79843ec3427c : SUCCESS in 44m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bf4c7f66480f4adca39de1fe10e1a2fa : SUCCESS in 1h 16m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/da20a7bc91ba4612878655fb6aca9158 : SUCCESS in 1h 22m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a06e01aef8a44fd5970decbbae147d1f : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d41792545314abf9b83e23b2041f4e2 : SUCCESS in 1h 34m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8750fce3a14b0584ece09d2687b292 : SUCCESS in 48m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20268dc54615407b92b0194b71743fa5 : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc60133b50dd4ecc971ee43a3b2278ae : SUCCESS in 29m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d34b42ae35804685b374b7560450c45a : SUCCESS in 32m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa34e6e880344a519bb17a661f4be296 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c92d9f95b994fd59c118908021254c2 : SUCCESS in 1h 08m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/edbde7b878d24a1abef07081ca170fd3 : SUCCESS in 1h 06m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0a1d900939a24327a681f9782f069557 : SUCCESS in 1h 15m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d17884fda8ff40fe923b500f321913d2 : SUCCESS in 1h 03m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f6efb15f5e6b4102a05687fe082c1647 : SUCCESS in 53m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdbb5a06bd8a45f49cb2479ed7479e93 : SUCCESS in 2h 07m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5eaf5825974f9db397206c9b0da486 : SUCCESS in 1h 05m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b1352f7fd40c4257abaacd63e03e239d : SUCCESS in 58m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac3814a69c44a7180cb26e52b4c6f9a : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":11},{"id":"9bc89362383ca8b7cde75065a657b1e1f560e889","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 01:15:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"39aa372ef6f13c9ab4814462e4d9b02b97ae1e23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 01:17:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7b9e7ad018468e88ea4297997c0c4f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"c4007b0833111a25d24f597161d39ee9ccd37189","revisions":{"39dac639eea2f45c32e10cea40f88d32287b3cd0":{"kind":"REWORK","_number":1,"created":"2020-12-07 21:01:11.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/765874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/1"}}},"commit":{"parents":[{"commit":"b5ea767bc335f69645a604607569515fa3cf4d85","subject":"Merge \"Switch to new engine facade for ovn_db_sync module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5ea767bc335f69645a604607569515fa3cf4d85"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 21:00:56.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39dac639eea2f45c32e10cea40f88d32287b3cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39dac639eea2f45c32e10cea40f88d32287b3cd0"}]},"branch":"refs/heads/master"},"5f4e0d8b76d7573e36c245f5569371ccf1a75c5d":{"kind":"REWORK","_number":2,"created":"2020-12-08 11:01:42.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/765874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/2"}}},"commit":{"parents":[{"commit":"b5ea767bc335f69645a604607569515fa3cf4d85","subject":"Merge \"Switch to new engine facade for ovn_db_sync module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5ea767bc335f69645a604607569515fa3cf4d85"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-08 11:01:30.000000000","tz":-300},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f4e0d8b76d7573e36c245f5569371ccf1a75c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f4e0d8b76d7573e36c245f5569371ccf1a75c5d"}]},"branch":"refs/heads/master"},"2893a5914a0d5438f60f318f1f2b2d4aeb4355e3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-09 15:39:52.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/765874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/3"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-09 15:39:52.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2893a5914a0d5438f60f318f1f2b2d4aeb4355e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2893a5914a0d5438f60f318f1f2b2d4aeb4355e3"}]},"branch":"refs/heads/master","description":"Rebase"},"8d7c60d1ff49ca06e98971565eeb4ed6bcbbc77e":{"kind":"REWORK","_number":4,"created":"2020-12-09 22:14:41.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/765874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/4"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-09 22:13:58.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d7c60d1ff49ca06e98971565eeb4ed6bcbbc77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d7c60d1ff49ca06e98971565eeb4ed6bcbbc77e"}]},"branch":"refs/heads/master"},"93fd1fd37629858c399926e48317ee42f83ebffa":{"kind":"REWORK","_number":5,"created":"2020-12-09 22:57:51.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/765874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/5"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-09 22:57:43.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93fd1fd37629858c399926e48317ee42f83ebffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93fd1fd37629858c399926e48317ee42f83ebffa"}]},"branch":"refs/heads/master"},"4f9822207f4df143f61ff7f6b28485960bad5ed9":{"kind":"REWORK","_number":6,"created":"2020-12-10 15:24:20.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/765874/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/6"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-10 15:23:07.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f9822207f4df143f61ff7f6b28485960bad5ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f9822207f4df143f61ff7f6b28485960bad5ed9"}]},"branch":"refs/heads/master"},"f2cc615c19e47823327b2c4afeda6f7c95234a41":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-12-15 20:19:29.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/765874/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/7"}}},"commit":{"parents":[{"commit":"a5c513ef7f42702b727c012cc9a9f2bcd11b4923","subject":"Merge \"[OVN] ovn-metadata-agent: Retry registering Chassis at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5c513ef7f42702b727c012cc9a9f2bcd11b4923"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-12-15 20:19:29.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2cc615c19e47823327b2c4afeda6f7c95234a41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2cc615c19e47823327b2c4afeda6f7c95234a41"}]},"branch":"refs/heads/master","description":"Rebase"},"f8d1f8ddfc1b0c10247a9f1ab9cdb594a41fb5f8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-12-16 18:39:17.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/765874/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/8"}}},"commit":{"parents":[{"commit":"0a852dab998c1d8f782770263e094d0e7206f3ac","subject":"Merge \"Handle router HA port concurrently deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a852dab998c1d8f782770263e094d0e7206f3ac"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-12-16 18:39:17.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8d1f8ddfc1b0c10247a9f1ab9cdb594a41fb5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8d1f8ddfc1b0c10247a9f1ab9cdb594a41fb5f8"}]},"branch":"refs/heads/master","description":"Rebase"},"796f902689520e312852d8ccf4221eda1a6a022b":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-12-16 19:58:55.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/765874/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/9"}}},"commit":{"parents":[{"commit":"1994fbb8f909178dd8dfc45efd880429b5257e4c","subject":"Merge \"Switch to new engine facade for IpamPluggableBackend and Ipam driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1994fbb8f909178dd8dfc45efd880429b5257e4c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-12-16 19:58:55.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/796f902689520e312852d8ccf4221eda1a6a022b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/796f902689520e312852d8ccf4221eda1a6a022b"}]},"branch":"refs/heads/master","description":"Rebase"},"e2dd53a796bb2cac7d5d4dbc5d57d8b56086c89d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-12-18 22:06:38.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/74/765874/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/10"}}},"commit":{"parents":[{"commit":"74e4cd31f1f68153bad62897fe3944054568ff3f","subject":"Merge \"Move linters dependencies to tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74e4cd31f1f68153bad62897fe3944054568ff3f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-18 22:06:38.000000000","tz":0},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nRelated-Bug: #1894117\nCloses-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2dd53a796bb2cac7d5d4dbc5d57d8b56086c89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2dd53a796bb2cac7d5d4dbc5d57d8b56086c89d"}]},"branch":"refs/heads/master","description":"Rebase"},"c4007b0833111a25d24f597161d39ee9ccd37189":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-12-21 14:37:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/74/765874/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/765874/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/765874/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/765874/11"}}},"commit":{"parents":[{"commit":"74e4cd31f1f68153bad62897fe3944054568ff3f","subject":"Merge \"Move linters dependencies to tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74e4cd31f1f68153bad62897fe3944054568ff3f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-12-07 20:46:53.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-12-21 14:32:15.000000000","tz":-300},"subject":"Rely on worker count for HashRing caching","message":"Rely on worker count for HashRing caching\n\nThe current code looks at a hash ring node\u0027s created_at/updated_at\nfields and tries to determine whether the node has been updated\nbased on whether updated_at - created_at \u003e 1 second (due to the\nmethod that initially fills them being different by microseconds).\nUnfortunately, due to the notify() method being called which calls\nthe hash ring node\u0027s touch_node(), a node can be updated in under\na second, meaning we will prevent caching for much longer than\nwe intend.\n\nWhen using sqlite in-memory db, this continually re-creating the\nHash Ring objects for every event that is processed is exposing an\nissue where rows that should be in the db just *aren\u0027t*.\n\nThis patch instead limits the hash ring nodes to api workers and\nprevents caching only until the number of nodes \u003d\u003d number of api\nworkers on the host. The switch from spawning hash ring nodes\nwhere !is_maintenance to is_api_worker is primarily because it\nseems to be difficult to get a list of *all* workers from which to\nsubtract the maintenance worker so that _wait_startup_before_caching\ncan wait for that specific number of workers. In practice, this\nmeans that RpcWorker and ServiceWorker workers would not process\nHashRing events.\n\nA note on bug 1903008: While this change will greatly reduce the\nlikelihood of this issue taking place, we still have some work to\ndo in order to fully understand why it rubs the database backend\nin the wrong way. Thus, we will make this change \u0027related to\u0027\ninstead of closing the bug.\n\nRelated-Bug: #1894117\nRelated-Bug: #1903008\nChange-Id: Ia198d45f49bddda549a0e70a3374b8339f88887b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4007b0833111a25d24f597161d39ee9ccd37189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4007b0833111a25d24f597161d39ee9ccd37189"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
