)]}'
{"id":"openstack%2Fneutron~765994","triplet_id":"openstack%2Fneutron~master~I99e33e8c9a5146dcae3ac9d98aae0284d4d80f20","project":"openstack/neutron","branch":"master","topic":"add_destroy_patch_ports_script","hashtags":[],"change_id":"I99e33e8c9a5146dcae3ac9d98aae0284d4d80f20","subject":"Add script which can clean patch ports between bridges","status":"MERGED","created":"2020-12-08 13:03:40.000000000","updated":"2021-01-04 15:23:44.000000000","submitted":"2021-01-04 15:21:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765994-add_destroy_patch_ports_script","meta_rev_id":"668c3181ecf2b5ffe69e881333490dce1c49b310","_number":765994,"virtual_id_number":765994,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-12-10 13:18:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-04 15:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-12-18 15:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-11 12:32:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-12-18 16:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-01-04 12:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-01-04 12:50:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 13:05:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-12-08 13:05:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-12-08 13:05:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2020-12-08 13:05:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-12-08 13:28:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-08 16:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-09 03:35:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 13:20:21.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-12-18 16:46:46.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-04 12:50:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"4d5e2a4e4415cf201b2cfdff209f63f0231b5fd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-08 13:03:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eacddd0289382c3cc4f49bb48ba7f8a854217dc5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-08 13:28:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765994/1 : FAILURE in 24m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b58b8a72560b3e0e3041949ee980a6a759fe80fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 16:26:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/665e537a75e34b7ab251853e4ab476e7 : SUCCESS in 52m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1bb8524a0c64d3abdbf22d89c8c2602 : SUCCESS in 41m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9ab616d154554e2a9e2ffcee00d89eac : SUCCESS in 1h 35m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aa93a3fc0e3642c3ab560c1b97873c2e : SUCCESS in 1h 43m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/835f42c959f94d8baca0b3f537f617eb : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0d898413e7af4755ad7a573a8b4e7786 : SUCCESS in 1h 26m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9568624b6f6f4cdaa26895d4ffbe6746 : FAILURE in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaa24f2c39334dec8ab88d14621755b3 : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1bc2391cbf4b288e349bf4fbf786c3 : FAILURE in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49cfd8486299432da80b187179738ebd : FAILURE in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7b8f6667ad40948aecc81a10a25907 : FAILURE in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b116be652bf24944849665a1e3f5de86 : SUCCESS in 8m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dac784f0728a4bfaa72aa14a60bce86a : FAILURE in 17m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1c9456f0f414be1bfb1892e22a30b2e : FAILURE in 21m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/af97880bb90b48788b61a9161128009e : SUCCESS in 1h 42m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb37db3b348040b7b0f4146c3aacbcd2 : SUCCESS in 1h 02m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce626747792141d3aa6ec9b851c08a09 : SUCCESS in 1h 12m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/900ce89a429c4e99800bcdcae4d1b916 : FAILURE in 1h 03m 16s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7de88c212c2a4bc59ca2c427e9195224 : SUCCESS in 1h 28m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d3ee6d454cf3427493ce8d80da79f226 : SUCCESS in 1h 30m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55dd07299e244acfb8a4f9b8fceb3a11 : FAILURE in 1h 15m 54s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fb4025cbfe914d05aa958c09c1f5752e : SUCCESS in 1h 07m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06f877b46cb74b859835ec99bd7fc420 : SUCCESS in 2h 10m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d752d0647f5943e08c5567fb71beceb6 : SUCCESS in 1h 03m 59s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b11c0b0a519f4b62ad778648ca675d32 : SUCCESS in 1h 14m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/92f596513ca44a9087912f2e6202642c : SUCCESS in 1h 14m 35s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/72f9adbf863d4803a9370ab727964f0e : SUCCESS in 48m 26s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/c75c2400cbb8417791e376542274d403 : FAILURE in 1h 14m 19s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c22dd3c8d0384ab2927b67ab2ee5607f : SUCCESS in 1h 20m 06s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ef4b002620c44108ae8e2bd11a4ad0d5 : FAILURE in 38m 58s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53547b7fb3754e9fa346daa23116e9af : SUCCESS in 1h 32m 32s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4546a9d6a618400b90ded42f1dc1bb8a : TIMED_OUT in 3h 03m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4f32bb36d879463583478647ad94fc79 : FAILURE in 2h 09m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0b2107ee21948ecd12552027d3a7911d6ef6b18","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-09 03:35:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/94/765994/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4d1d70a3ccb5e308e9ed5fe4e96b71dc9c15a0c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 10:08:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"27b826056c90d7ee5c029a6de194500657a9a1fd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 10:16:41.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765994/2 : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":2},{"id":"672701fa1d7ee7be25025d36348d6a30843db360","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-10 10:26:15.000000000","message":"Patch Set 2:\n\nIs there a LP bug for this?","accounts_in_message":[],"_revision_number":2},{"id":"236dd6ef4b389d2c171d659306409af33a5f2370","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-10 13:18:32.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/94/765994/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"80a77a81ad3649093e8383b72a27471de9e031ea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 14:26:43.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Is there a LP bug for this?\n\nNope. It\u0027s not really a bug but \"improvement\" for us. This script was originally hosted in the RDO repository:  https://review.rdoproject.org/r/gitweb?p\u003dopenstack/neutron-distgit.git;a\u003dtree but it\u0027s very hard to maintain it there. So I think it will be better to have it here and keep in sync with all other neutron code.","accounts_in_message":[],"_revision_number":2},{"id":"b5e851fbb9c5a53135525699bb424624fb444e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 14:41:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2e5f144afbf346de9e4d28ae8dc97577 : SUCCESS in 56m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8fe8bbab14a9483497bfcd5eaae1fa16 : SUCCESS in 30m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1d4b74cfa09b42678169186e8603f4cc : SUCCESS in 1h 57m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/67206acda9834d829eafebfab12bf2c4 : SUCCESS in 1h 37m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8ef8fb9083b94c8da4bc9b57e95785cb : ERROR Failed to update project None in 4s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/29d1722b022c44c3867f813482eac0e3 : SUCCESS in 1h 42m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c8bc48c0ae149cdb89d3f4aec174e78 : SUCCESS in 46m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7443fc9e85b74bf1b101ac508b3543d0 : SUCCESS in 35m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e8643d4c804e73ab7d14e92594a740 : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6897d5455bc243dfa9e8eebb223317ff : SUCCESS in 24m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42ad4d9ae25349f2b1c557fbe5294a99 : SUCCESS in 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af9793e2c5b24a66a1c434035167e550 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa3043d9ec2849fcaed7ed7c45d963c5 : SUCCESS in 1h 16m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aaeb281379ca4d34a23346d4fcf91fb0 : SUCCESS in 1h 17m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cc4cccfbacd24b36bef4cc811f808913 : ERROR Failed to update project None in 2s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bb0b18afd76498aa6e0c673b5f2de04 : SUCCESS in 1h 07m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8a5cbcad81e40c7970838d1f58098b9 : SUCCESS in 1h 16m 41s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/05ee6c7270c6426a9799517056ad5960 : FAILURE in 1h 00m 02s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0d398a0b1b5a49d18fbeb583e71863f6 : SUCCESS in 1h 26m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e2d8ee7ce0945b591fc18ab0f18fbd7 : FAILURE in 1h 43m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cff6a90fcdc54de599a5b754e6db9290 : ERROR Failed to update project None in 3s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15b35dba4c0e459cb444c3aca28e1d04 : SUCCESS in 2h 13m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1254149e96a4769b01c3ae2a4a911dd : SUCCESS in 1h 10m 03s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c6dea76184a44dfe96daad5a4b0841a7 : SUCCESS in 1h 08m 00s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b6fa772dfa704f73a75350fc69a73c82 : SUCCESS in 1h 12m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a59c38ea4da34d10a241aa0dd766ab60 : SUCCESS in 43m 05s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/f1b4030ff5ab49809689a360836d6655 : ERROR Failed to update project None in 2s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a08342029ce24452808d277993573baf : SUCCESS in 1h 19m 05s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/44fc37ade71b4b50875b94cc117fe5af : FAILURE in 33m 42s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce8719f10b224eed9ac032bdd65a8a98 : ERROR Failed to update project None in 2s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/caa17c91d43e40a490cf96906cc6a768 : FAILURE in 2h 33m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e34abf3653e241579c5a68f9a46fa45a : FAILURE in 1h 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f737d87064c1f30d6e657c6fa55849111a5d4ba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 14:50:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"17434ee2776edca2b193fe2bfda05f0086e1a67e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-10 14:51:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765994/2 : FAILURE in 41s","accounts_in_message":[],"_revision_number":2},{"id":"a2ee448d94d61a7adf5d7d488d81acb9c0f76e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 18:28:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/50b299154c77410a9c2c99e91b129648 : SUCCESS in 53m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/75018d53c32d46aaab5e26c6bf084318 : SUCCESS in 35m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9c13bf3d561145a08160e5f635e133de : SUCCESS in 1h 08m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a942f55f73641f4b4a1f8cdbace8e08 : SUCCESS in 1h 08m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d3e59948dc764e53a5185768860ecd38 : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/31a9ef1db9ae46059391a683415538c9 : SUCCESS in 1h 18m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e71d34e67134f3ba7e9e2d64afa57ba : SUCCESS in 34m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/960a20e2319043fbaf8795f1468699d3 : SUCCESS in 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d5641fc9064c0887f081c2fa36fc3c : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/439094d7d5834fea8e906c0a386a16c6 : SUCCESS in 30m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88fe5fa78ee24d5f8a663efa55bca987 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76113adbc68a464c95a8787b6f2d47d4 : SUCCESS in 8m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d1773bb963645ab944b5c585a9c2da2 : FAILURE in 1h 18m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/70357419331f43c695a6672d570f4f7c : SUCCESS in 1h 10m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b686f61024414662ae91404595f1ee9f : SUCCESS in 1h 33m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/615869b63a524a0fb6f7289a54006216 : SUCCESS in 1h 02m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/91af08a861504e66b27d21a17828524a : SUCCESS in 1h 09m 21s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/5efa23bb9cfd4e7a88555227172a380f : FAILURE in 50m 15s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c24555499e9c4a68bb131d63c20ce749 : SUCCESS in 1h 25m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38432d5b109e419b9c5939999438f608 : FAILURE in 1h 51m 55s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f8f4c960aaf445ac8c39644a8635b57c : SUCCESS in 1h 28m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9abcdde3fd484451a2ce887ab5a614db : SUCCESS in 2h 27m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bee5aaff58e46759c50312c6110f3cf : SUCCESS in 1h 00m 35s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/bf1b4d01a84f40d6b5b1df2dd21d2408 : SUCCESS in 1h 05m 11s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2eec18b0b9ad4c41b1f8e21b4a30da46 : SUCCESS in 1h 09m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b571915418c945a3b1b8e4dbd9133958 : SUCCESS in 43m 33s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/22c3f4897f1f424197d4a21999d1f72f : FAILURE in 1h 08m 59s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b97fb6b36051493ebc8c07f1c1043930 : SUCCESS in 1h 26m 49s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/14bee7d0642f42f69941feed2d988fa9 : FAILURE in 41m 07s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/18909fdeac9e4256a7b643516a036a97 : SUCCESS in 1h 44m 09s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5d779fa2289e4c10b5196ed9d9a53570 : FAILURE in 2h 55m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ccf45fcf90d344babd39079ab46d5aad : FAILURE in 2h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de37763e5e361eba8c0537cf7c65253891fa5bbf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-11 09:08:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7201b7be8c32a35b1a878d9abec02df7ea2404b9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-11 09:10:06.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765994/2 : FAILURE in 39s","accounts_in_message":[],"_revision_number":2},{"id":"1d17ea39e66a736612e30f6a88abe631877e9df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:18:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62ae2d2e49754c61b2dcdd4629a27d20 : SUCCESS in 42m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1796aed621645fabb44439f54302494 : SUCCESS in 28m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b0d92b00d32541d490b6e35b6f10e6d3 : SUCCESS in 1h 31m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0cbcc340e3b5474b90593f13128c89df : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fddd27260a68485eb118d95d3af74456 : SUCCESS in 1h 37m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a1319cba32764158a865a2a7f0614567 : SUCCESS in 1h 32m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cb7ec0e54294fcebce7f4324f686353 : SUCCESS in 36m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98a22e5a7eb9436b9faf662343b83198 : SUCCESS in 32m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89395302168c451ca3e1e587ca8c3c32 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60c4be943f664cf6ba786b4aeea57f61 : SUCCESS in 27m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4ba985d6ca04bef90eb0f56c8e54f98 : SUCCESS in 30m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d8e080059d46ab91a016f71dfc4edf : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fbc9149d1404907bde1056a401d5e6f : FAILURE in 1h 10m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92fd78a3149d47fb9185a9f4c25215ba : SUCCESS in 1h 15m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c23891c41c5a4b1890b4b0107a7a3f8d : SUCCESS in 1h 23m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a2d4644f5444e7e8d15644c2bb2ae7d : SUCCESS in 54m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9db2c4dcfee545289674be723a207151 : SUCCESS in 1h 03m 20s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/b47082b86a45464abb52d102f1edb5ea : FAILURE in 52m 08s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a802d32b294e4c52a6468ca95c784619 : SUCCESS in 1h 12m 35s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9afa4589d30a472bbdc2adeff3ef3976 : FAILURE in 1h 30m 46s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8f6a15387d5c45309bde362fb0a5e8c8 : SUCCESS in 1h 12m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/662b2d572da1430f909de170cc94486b : SUCCESS in 2h 09m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f091968f8cb4301ac1262b875a67f41 : SUCCESS in 1h 07m 10s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9ccfd52f66ea4f798c4f6c4cfea1392c : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/26f56a96c10e471e950c9c0e5e09e344 : SUCCESS in 1h 12m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf835847b334d3589682109dd230e4b : SUCCESS in 41m 37s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/078870dc633347e1acfc18f10cb1dd05 : FAILURE in 1h 12m 53s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c896b19151bd4016824a922b386f225d : SUCCESS in 1h 22m 18s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/142935ba7a214d259eabcc0d1fc255bb : FAILURE in 37m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ecc79be2d6c742cca332ca6960bd1104 : SUCCESS in 1h 32m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4b4303f5841147f29de1e2034025fc9e : TIMED_OUT in 3h 03m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7bf2992267ee493c8df7257db95fcdcb : FAILURE in 1h 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5ea831d7eaad093d8ad7af0ce4decc121369003","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-11 12:32:40.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e6b36f2055e8cec1223292c56ac8d0c2932c4d58","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-11 14:34:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/765994/2 : SUCCESS in 2h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"2b36ca48d59bb1222da8dff5d724f157646fe5b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:33:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a23b6b8921eb4503b3a2f5b465479c56 : SUCCESS in 52m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/809ba497582c4f10baa1f3b8b1c49a4c : SUCCESS in 31m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c9b5045675604d578efdcc55ebe0698e : SUCCESS in 1h 28m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5205d96fb8c24435878a9afcdabe81bf : SUCCESS in 1h 38m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ebfb7fa0f9634ba6879687f03c43318b : SUCCESS in 1h 48m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/296faceb6e8846d0a49f4076a9f94e14 : SUCCESS in 1h 25m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2d4a424f514b56a2704fff16b037a1 : SUCCESS in 40m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ecd76b9e5ae4783aa10c74ac8b33a78 : SUCCESS in 38m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6727801ef44a61bb48f00b761bdf45 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2202faf54da40fe971a4dd07cb73573 : SUCCESS in 26m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41300c4a5b804546a987d86b4522d23a : SUCCESS in 28m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed53913e94346f2babfcc5653dc4420 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b1cc86f15474d7c9709329810dd9f1a : FAILURE in 1h 06m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98dcead5df19465881ce208a0c12f62c : SUCCESS in 1h 13m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/185812f87ebb461581281173c103df9d : SUCCESS in 1h 33m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc9f2a94839040ecae4544ac2a1d5bd3 : SUCCESS in 59m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f2547ddddcae4297867f8f69f2ded3ac : SUCCESS in 1h 05m 46s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/2d55cc91cdc3493abe2d6085c96f0f08 : FAILURE in 1h 05m 40s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6a12898f34a0406aa09f101175e9f06a : SUCCESS in 1h 15m 56s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/81eaae1d648f4f7f8a2f0760ab1c21c2 : FAILURE in 1h 41m 30s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9afcb17682ea43b09ee1900115e85ded : SUCCESS in 1h 09m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5d4893d3a4248f2a89388478b836671 : SUCCESS in 2h 08m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d682b8019a84bf39369ceb788605f2a : SUCCESS in 1h 00m 21s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/64edfc71091f4b569152f430c917d7b2 : SUCCESS in 1h 09m 15s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5c291ce3f97a49b1b62df662970e803a : SUCCESS in 1h 08m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fa72cc0a0f7c4a7087ef4ccea0f3d308 : SUCCESS in 38m 50s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/4b300209de4b441a8888340c8b90cd3b : FAILURE in 28m 34s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1f83c63d0d8643128d1efb0ad4206a9f : SUCCESS in 1h 21m 52s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/976cf1dd3c714f2ea00576c860b57746 : FAILURE in 37m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/66b4f161a3bd41d089703b737c65825f : SUCCESS in 1h 33m 40s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ef1889ea7a194e5db6fc35e942a9d229 : FAILURE in 2h 51m 03s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3cae464ca0b14a5db0116e0493a62e30 : FAILURE in 2h 07m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0e9685e93e6be406194cce29f8f61e58507298a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-11 16:27:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"41c436968e62178c17370de5f7ed67a6769296a0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-11 18:19:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/765994/2 : FAILURE in 1h 51m 16s","accounts_in_message":[],"_revision_number":2},{"id":"196cf92ba13c7018073e8768952367ce0d209e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 22:16:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b325b2a9d0634ec187be6a0c88071770 : SUCCESS in 45m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2421de141d914347bd51502d717b1ae2 : SUCCESS in 37m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cfa95550e5c447198f798ac259b23a62 : SUCCESS in 1h 51m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/76d1b2209f7b4ae98c02682e0bd86957 : SUCCESS in 1h 46m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87d978ac6c464be5ac57564709dce9c9 : SUCCESS in 1h 31m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e3758e9751a7484bb0948511547fc723 : SUCCESS in 1h 27m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42a0c0a7761d48789386355c80f97e1c : SUCCESS in 41m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bbf9f0a232843e88684009bb4049bc3 : SUCCESS in 35m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7f92a6036e4df89885eaa4a85269b3 : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a0de8fee51b4e8daf20ae74f41bdd63 : SUCCESS in 32m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/535bb15c85e642b88c48804642c73829 : SUCCESS in 24m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cadc147ef2844a7a587fee9fadcea14 : SUCCESS in 40m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75db502387bb457c83cb30e92ed90dbc : SUCCESS in 8m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7871bf2bbc084ec0a9adde874b07f02a : SUCCESS in 1h 06m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96223e61fcbd41d3b0fa92b4288f70e4 : SUCCESS in 1h 21m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3d16ab2da57d41a29c797f904cc16fcd : SUCCESS in 1h 34m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da1306aba8554b1d82b8e27eb30976e6 : SUCCESS in 49m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3cc31b03f6ce428590cf102fa1e40ca6 : SUCCESS in 1h 13m 29s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/b92c3ba9a6594c2bbffe85d04e4798a3 : FAILURE in 49m 16s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3343fd32acaa4692b554f546ab73c412 : SUCCESS in 1h 16m 39s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a100de18a4c24f169b1ce5596a456dd4 : FAILURE in 1h 14m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bfbd45b9eb5d4f0abb06422ac37de754 : SUCCESS in 1h 18m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f73786e1836b4ff1943c34e5763ea95a : SUCCESS in 2h 07m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd7a97cca0a146b6845a7a2bf33997b4 : SUCCESS in 54m 24s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/b5688317918e498faa9806ebb89083ff : SUCCESS in 1h 06m 08s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d873b338b7aa4dbc89dbf82fac13f8b7 : SUCCESS in 1h 03m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4ef911a317b040b1bd20fb90d6b60029 : SUCCESS in 36m 25s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/464144f47b37407aad17969565c32c75 : FAILURE in 1h 06m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/156b21fd467b4b5987c2e82f5571f6d1 : SUCCESS in 1h 27m 42s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8abd0956f78e459ebebe4aed5f46b9ca : FAILURE in 37m 14s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38b2ed45c4554f6e80654e9d77801de9 : SUCCESS in 1h 28m 29s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/96e837cc68b340f28b70d931ba81db51 : FAILURE in 2h 44m 03s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c016e70a908d49fa957d765a1f3bc70c : FAILURE in 2h 00m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9790abd1c11be85260c375f943fc293bfdfb3b8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-18 15:55:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9284e5a0b5641c2e674db0eb632cdb27f28c0b10","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-18 16:46:46.000000000","message":"Patch Set 2: Code-Review+1\n\nI remember that script well 😊","accounts_in_message":[],"_revision_number":2},{"id":"bc9cd5a163955bbd0080f58b380ab5e6354ff8aa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-04 12:50:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70124ef8e6c28e35916bdd2bde253bbeb32121f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 12:51:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"835a21fbe00cc5a43c25cdbaad470f965845e72e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:21:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a46fff8342464628b1343eba235b6198 : SUCCESS in 1h 16m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ecaf74654a9d48d794abc5b3fe93dec3 : SUCCESS in 2h 27m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/42e7dd94512543ada3643666f126493e : SUCCESS in 1h 49m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1bdbbee8223d41a8b6bcef9b11f0dc70 : SUCCESS in 1h 58m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/acae29d2d530454ab6933b3d6b290577 : SUCCESS in 59m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d87a691378f4e7f9fa4da3bc33de98b : SUCCESS in 41m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79481c03ff0c45499767e82ba4978014 : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e99bbfe23da47ffaf4fdd71d56d74c7 : SUCCESS in 26m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df0c88d74d749d5a2f2d39922d913de : SUCCESS in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831af18918664f2eb7b8a1d866a8b91e : SUCCESS in 7m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/816d3d26947c4f1d9d122fb8ec68077c : SUCCESS in 1h 04m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12787d475dd64f6ba6e0b8637ad7fb6c : SUCCESS in 1h 06m 01s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fc9ffec5f2574143bd13c653f9773a93 : SUCCESS in 2h 00m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c929da6f09d4423b16556155ad8d08b : SUCCESS in 57m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6cfc05282efb41e591951faf872acc39 : SUCCESS in 46m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebe0e75cd2a74328b3355b773196e83a : SUCCESS in 2h 14m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a9913c7cb144823b181b849e9bec0f6 : SUCCESS in 54m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a2d7d125f48d4623bc7abe2d538748a8 : SUCCESS in 46m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/01fa0d29eaf84092ad046b812c29c9bf : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9a2548b39c9281d896b60325ab4c8b9f814b7f44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:21:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"668c3181ecf2b5ffe69e881333490dce1c49b310","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:23:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9825bf467cf94fb69fac71a08c78435e : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86f6ea347ac5ae2b966c34843bd2532fd63608b2","revisions":{"edabc68e4ae91e8214a248fead60e0a7014c8926":{"kind":"REWORK","_number":1,"created":"2020-12-08 13:03:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/765994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/765994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/765994/1"}}},"commit":{"parents":[{"commit":"b5ea767bc335f69645a604607569515fa3cf4d85","subject":"Merge \"Switch to new engine facade for ovn_db_sync module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5ea767bc335f69645a604607569515fa3cf4d85"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-08 12:55:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-08 12:55:19.000000000","tz":60},"subject":"Add script which can clean patch ports between bridges","message":"Add script which can clean patch ports between bridges\n\nOnce provider bridges are brought up, they should not send any traffic\nto the integration bridge until ovs-agent configures the bridges.\nWhen node is rebooted, patch ports to br-int are preserved because they\nare stored in ovsdb.\n\nThis patch adds script which can removes such patch ports. This script can\nbe used e.g. in some systemd unit files or ifcfg scripts and be run\nduring the node boot process.\n\nCo-authored-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\nCo-authored-by: Yatin Karel \u003cykarel@redhat.com\u003e\nCo-authored-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I99e33e8c9a5146dcae3ac9d98aae0284d4d80f20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edabc68e4ae91e8214a248fead60e0a7014c8926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edabc68e4ae91e8214a248fead60e0a7014c8926"}]},"branch":"refs/heads/master"},"86f6ea347ac5ae2b966c34843bd2532fd63608b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-10 10:08:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/765994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/765994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/765994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/765994/2"}}},"commit":{"parents":[{"commit":"8d6c301301e33697fd4eb4d6de53f680b909a04d","subject":"Update requirements for recent pip failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d6c301301e33697fd4eb4d6de53f680b909a04d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-08 12:55:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-10 10:08:02.000000000","tz":0},"subject":"Add script which can clean patch ports between bridges","message":"Add script which can clean patch ports between bridges\n\nOnce provider bridges are brought up, they should not send any traffic\nto the integration bridge until ovs-agent configures the bridges.\nWhen node is rebooted, patch ports to br-int are preserved because they\nare stored in ovsdb.\n\nThis patch adds script which can removes such patch ports. This script can\nbe used e.g. in some systemd unit files or ifcfg scripts and be run\nduring the node boot process.\n\nCo-authored-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\nCo-authored-by: Yatin Karel \u003cykarel@redhat.com\u003e\nCo-authored-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I99e33e8c9a5146dcae3ac9d98aae0284d4d80f20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86f6ea347ac5ae2b966c34843bd2532fd63608b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86f6ea347ac5ae2b966c34843bd2532fd63608b2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
