)]}'
{"id":"openstack%2Fneutron~768052","triplet_id":"openstack%2Fneutron~master~I10c64c318cf04692c30eff6cb718450ea174caf0","project":"openstack/neutron","branch":"master","topic":"osken_app_refactor","hashtags":[],"change_id":"I10c64c318cf04692c30eff6cb718450ea174caf0","subject":"Refactor the Neutron OSKenApp class","status":"MERGED","created":"2020-12-21 09:37:08.000000000","updated":"2021-01-05 10:57:01.000000000","submitted":"2021-01-05 10:55:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"768052-osken_app_refactor","meta_rev_id":"8d7da55c57259dc2a3321d6dda3bb476ee054299","_number":768052,"virtual_id_number":768052,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-12-21 10:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 10:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-12-21 14:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-21 11:12:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-01-04 13:51:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-01-05 08:03:17.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-01-05 08:03:17.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 09:47:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-21 10:53:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-12-21 10:55:48.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-21 11:12:00.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-21 11:42:19.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-21 12:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 13:51:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-01-05 08:03:17.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"441c0e4c12459fa6f49df9e50c38b7d700505ac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-12-21 09:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97497117c927032ac240783931e5e9d4880cf89f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-12-21 09:46:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"21061c0ab0769b5a2c1739504b2ad56eeaf43101","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 09:47:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/52/768052/1/SRIOV-neutron-direct : ABORTED\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/52/768052/1/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"1ff22ff1851446a51eb6da3c7ee857bec9bfb51f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-21 10:50:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/52/768052/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/52/768052/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"26035d654a98bfeb490e34b756208c07af7d1551","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-21 10:55:48.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/52/768052/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8c4cd117e2802397a4a30305971db0eeb63725fb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-21 11:12:00.000000000","message":"Patch Set 2: Code-Review+1\n\nThx for that 😊","accounts_in_message":[],"_revision_number":2},{"id":"9b14ec9f55c268a399e28e874c5c17f9206de40b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-21 11:14:56.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/52/768052/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"8d873264e7a6e3bb72434fa7e2927fda5e0d876b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-21 11:42:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/768052/2 : SUCCESS in 1h 53m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0ddbd9ab31c8cc2f47dc33132cb4b2d3e5e7d02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 12:55:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/936cff89a69d4902bc9b8c8313200c4b : SUCCESS in 47m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e1089020d23457897e7bb727b009965 : SUCCESS in 28m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b3af4c6cbfba447989189f6e5de9f038 : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3d01c114438f406ea402f37c1c5abd4e : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca483e34718645c190c5df8a3efa66a4 : SUCCESS in 1h 43m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7a0cffe0f55c4faa9e53f7b26deb6e57 : SUCCESS in 1h 23m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb78447d6c7541b4a7b9327611eaa600 : SUCCESS in 34m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79417e94c499488988b4dd5288072760 : SUCCESS in 33m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8b767e57d244b5b938ea8355b1ba99 : SUCCESS in 12m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32620c875e984b8bad17d0402351bf6e : SUCCESS in 29m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a386b26c2fa0419aa434bff804de651a : SUCCESS in 27m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e8e042309b24cc2aca85663e8f15923 : SUCCESS in 25m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1483fa6e4d824f8aa67a44de6930d7f0 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bb5cee858b542ffaaf482812a7933c9 : SUCCESS in 1h 16m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f425192c988482eb0533ee91c64e827 : SUCCESS in 1h 14m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dc3bca1391554016a9106426d34f747b : SUCCESS in 1h 44m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c4d954b163c4b0c8024299e2c1992e8 : SUCCESS in 1h 02m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/425ae696a77d41108652520cb0f0f4ed : SUCCESS in 1h 01m 23s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/93374acc1f224e05a896328b5b48275e : FAILURE in 55m 33s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/88b3a2d235014ca2aecd9cc32d865137 : SUCCESS in 1h 44m 20s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39d2117bbcff40f6814509534ed1c32a : FAILURE in 1h 32m 59s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0fdd1fe1e7b14e1aa2c9f0b0ec2d6cfe : SUCCESS in 1h 13m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b748db09cbe44278b91719ff4f283888 : SUCCESS in 2h 08m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a288b34e01514818909d24b135bb77fc : SUCCESS in 59m 41s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e30db9a70291493fb2d9865f9778b315 : SUCCESS in 1h 02m 24s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/301b6367a0e8478a90ea8f8cc70addf8 : SUCCESS in 1h 04m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0de4b38118024a4a9b2f87680919f0fb : SUCCESS in 40m 35s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/a7de601834c5402eaa07a4c176319997 : FAILURE in 1h 26m 52s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8c16686654ea481188c8df25bb515760 : SUCCESS in 1h 13m 36s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e13fde24420d463f818d83911603829f : SUCCESS in 2h 56m 53s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bd32b09ddd544a68921c3b7a107d462d : SUCCESS in 1h 23m 20s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2afaf73135ec497cb2cc2b6122833b78 : SUCCESS in 1h 49m 47s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4921305051314c0fad70619899125513 : SUCCESS in 1h 47m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7bbb429710a84959b54ef303fe0ef9d0 : FAILURE in 3h 01m 41s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8af448a173864293ad30015f25d24b3a : FAILURE in 2h 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6e7f78273c83a9e0d838d149f76fd7a524cfbc5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-21 14:02:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6086638cda7a6c463c8c6e4d8381f4a26dda5234","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-04 13:51:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97a94e5dd683efe6d5836e37c11fdc7d1acc50c9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-05 08:03:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7fabe8993e30123c328bd390d305b4663ab2be17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 08:03:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"006e5c5a6dbd1ca5b3b72da587ed59d5a5f239f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 10:55:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6c8d5ed4da54a139ac3992b609f7ae7 : SUCCESS in 49m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f3d14cb6f2634d6cac7f829d0a38347b : SUCCESS in 1h 36m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/35d618d0a4dc41cfb6ad3abf83e8d38c : SUCCESS in 1h 46m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d76fbbc6f2bb4ec1bcc3de28132a473c : SUCCESS in 1h 33m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd70f26e271744359293b4c439dbf84a : SUCCESS in 1h 47m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1973cada6eca410cab96ab51c05cefe7 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86f3175d4dd472badcf6252e0577932 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f611cabafbc843d1b06bb947b0192450 : SUCCESS in 31m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520e388c4b4743a886da1ade9bc76a88 : SUCCESS in 31m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/554b413c651c468fab7a37dca06a1e9a : SUCCESS in 11m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fbfbd98f4174dab8a377008daf824c3 : SUCCESS in 1h 09m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9a7f716e5f94718903d8da5a1c7f6cc : SUCCESS in 1h 20m 12s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3ba3dc3133d34d498aef6bde031b423a : SUCCESS in 1h 17m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76aa2935d7ad4cd2ad2d4dbbbf196516 : SUCCESS in 1h 03m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d6f5bc7aeb84035a193081a9b9d33b3 : SUCCESS in 58m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12bc22140fe54a7d89521a231c20d904 : SUCCESS in 1h 55m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14d59431244e4081980912cc1d688f45 : SUCCESS in 59m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e72cb96c9b6048e399776b8c5cdc3da7 : SUCCESS in 1h 00m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/89db7dfb75d9465d8dcf6db4abbecdc9 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":2},{"id":"78839e54aaaf60b0471361c26388f451ce8cf6d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 10:55:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8d7da55c57259dc2a3321d6dda3bb476ee054299","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 10:57:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b56a3fa3324e0bb3cf05151c5881ae : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f158922b67cd5a3085a06818dd56b3f115a1b98","revisions":{"83aeb28423aba083438feaa307feb6aa41c605f9":{"kind":"REWORK","_number":1,"created":"2020-12-21 09:37:08.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/52/768052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/768052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/768052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/768052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/768052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/768052/1"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-12-21 09:31:48.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-12-21 09:35:26.000000000","tz":480},"subject":"Refactor the Neutron OSKenApp class","message":"Refactor the Neutron OSKenApp class\n\nMove the duplicated code to a base class BaseNeutronAgentOSKenApp.\n\nRelated-Bug: #1774459\nChange-Id: I10c64c318cf04692c30eff6cb718450ea174caf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83aeb28423aba083438feaa307feb6aa41c605f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83aeb28423aba083438feaa307feb6aa41c605f9"}]},"branch":"refs/heads/master"},"9f158922b67cd5a3085a06818dd56b3f115a1b98":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-21 09:46:49.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/52/768052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/768052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/768052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/768052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/768052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/768052/2"}}},"commit":{"parents":[{"commit":"a37173c876e9bbca39eb125ac15ca03dbbcef1f3","subject":"Merge \"ovn: Add a gap about e/w traffic fragmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a37173c876e9bbca39eb125ac15ca03dbbcef1f3"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-12-21 09:31:48.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2020-12-21 09:45:55.000000000","tz":480},"subject":"Refactor the Neutron OSKenApp class","message":"Refactor the Neutron OSKenApp class\n\nMove the duplicated code to a base class BaseNeutronAgentOSKenApp.\n\nAuthored-By: Swaminathan Vasudevan \u003cSVasudevan@suse.com\u003e\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nRelated-Bug: #1774459\nChange-Id: I10c64c318cf04692c30eff6cb718450ea174caf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f158922b67cd5a3085a06818dd56b3f115a1b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f158922b67cd5a3085a06818dd56b3f115a1b98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
