)]}'
{"id":"openstack%2Fneutron~768240","triplet_id":"openstack%2Fneutron~master~Idbf2ffb4ef403f343973bde3b3599f9cd3aaddf2","project":"openstack/neutron","branch":"master","topic":"task/39975","hashtags":[],"change_id":"Idbf2ffb4ef403f343973bde3b3599f9cd3aaddf2","subject":"Change the logic of \"catch_exceptions\"","status":"MERGED","created":"2020-12-22 17:12:14.000000000","updated":"2021-01-04 13:58:06.000000000","submitted":"2021-01-04 13:54:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768240-task/39975","meta_rev_id":"f7eb1e367ef715e71bf32eafda8aacd307aa7408","_number":768240,"virtual_id_number":768240,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2020-12-24 19:35:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-04 13:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":1,"date":"2020-12-29 07:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2020-12-30 13:03:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-04 11:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-12-25 06:35:16.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-04 11:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 18:45:44.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-22 19:22:25.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-22 21:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 05:51:10.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2020-12-29 07:40:56.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-12-30 13:03:38.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2021-01-04 11:50:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"56124be3adf7c5270776551587101cce78b3a0aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-22 17:12:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36356a2b6f0cde01ca97640cf1b8ac770a548bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-22 17:35:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17e133cbbf75e2360e67720ec0b18d7aca929d64","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-22 18:45:44.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/768240/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0d0e1952612b684ec48116a9a487ea8257e17154","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-22 18:57:30.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/768240/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f8c0d9780736868f4d3d2312b75195581e5e6f5d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-22 19:22:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/768240/2 : SUCCESS in 1h 46m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0a7795c095f0d43ba7614e216dc7e534d40d0e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 21:04:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2a845fbbc9de47c3952e5b2f780bff6b : SUCCESS in 46m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e63685fe0e942d6862ca96144759875 : SUCCESS in 28m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b669b49d0a5405eb53b9d59bc165a4c : SUCCESS in 1h 20m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/35c71cb9aa6b45e6a99bc13e75b8af21 : SUCCESS in 1h 26m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/43ff6ba1d2a64ce99e15631644b29a2b : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/77a0c3738ff74155923f7e8f2e277c3a : FAILURE in 1h 30m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8339f6813fe04dacb8137ebbc4064208 : FAILURE in 44m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81506e60d58b40fdb36ac59d6fdc3afc : FAILURE in 44m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cb13d162d2f4822ab911c6a0951c4c1 : SUCCESS in 16m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6304edcab9c943d88e1b3618aa3323e4 : FAILURE in 39m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c9bce5f54de410a999684db08010a0f : FAILURE in 37m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7566f520e13240d5a3c7e61df1a2b6f9 : FAILURE in 31m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085bad253b1b43ce91d027ebd5d3a945 : SUCCESS in 15m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4000aee4eef34115bd4c01024aaa22ba : SUCCESS in 1h 03m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/015ada74dca4478bb99a79da1bacfee7 : SUCCESS in 1h 13m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/70c7797783074709bc18f0c3dea7289c : SUCCESS in 1h 52m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5271de5ce87848a58ca6e496c178eb9e : SUCCESS in 1h 17m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aef9205ab64d49fba4893752d50bf86b : SUCCESS in 1h 05m 37s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8f6d1aae4e16464b8a861f3e9a1d7b07 : FAILURE in 1h 06m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/46bbc13e9bfd4caaba08c6e91c0678e4 : SUCCESS in 1h 14m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1053ce770dc740949982b853bf3765f2 : FAILURE in 1h 24m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ff5c511b7e08482192666d86c725a3e1 : SUCCESS in 1h 27m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53632db78a7e4b5e9c2b6cb9f2c7fd1a : SUCCESS in 2h 17m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22077c73943b463e83770bd27e48682c : SUCCESS in 1h 09m 23s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e9431f5cc67146a28134bf9c792e0c1f : SUCCESS in 1h 07m 31s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3a95e72e87a54178b8dcad8861f65ed1 : SUCCESS in 1h 16m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/084ec7fdd6784121beca7b6d69d62b28 : SUCCESS in 45m 11s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/8495c5a7a8b345548ae729af63ed418f : FAILURE in 1h 04m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b0c245ec9c204594991f0a3f8b5420b8 : SUCCESS in 1h 13m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0787b54af62e4bbbb9a1516fdff8ad4c : SUCCESS in 3h 24m 30s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/64b9eb2fa03d471b9d8d7d8c1dcc65f5 : SUCCESS in 1h 16m 03s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a744a0435b7c4159a89c2b69540604d3 : SUCCESS in 1h 41m 57s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0a43249e9204441784ecaf3d8c44d3e7 : SUCCESS in 1h 58m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c1f2b2c60abc40f190c5d898321634b8 : FAILURE in 2h 44m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/aa045d6a8cb44ea6a05aa911de392320 : FAILURE in 2h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0507a976efe3aa0fe38202d3c985fb98f29e378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-23 14:30:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b2e97690102e815f974a94012d3a5ce4809a782f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-23 16:10:37.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/768240/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"323772bda962c1eb30d31d6ddb4c3cd062fbacc6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-23 17:05:18.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/768240/3 : FAILURE in 2h 33m 35s","accounts_in_message":[],"_revision_number":3},{"id":"8eeff26c9b398ebf2d8f054fbc99e202fd1f9cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:13:51.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/282375c1a9da441599826f99dc11f236 : SUCCESS in 1h 02m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/868bf41d0eec493ba6fbd1f1d7005dd5 : SUCCESS in 33m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7caebb6562ff4e2bb02d9f2b2b4b85e7 : SUCCESS in 1h 46m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4b25a84247904923acc461f107646ce5 : SUCCESS in 1h 27m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/652253f417474f2c899e465434d68c1d : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cd2a12ec49044cffaa82241128eaf4c9 : SUCCESS in 1h 43m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4f0403bc56641b1850dc622284223a9 : SUCCESS in 32m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79079743825c4a098ad95df3013c09f0 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f9c5ea24d446f1a7f79c98d5aa20e0 : FAILURE in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26c5d74b7a5547689e772709bfa8e1df : SUCCESS in 26m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acad7689a2b646828e45662f4a1adb33 : SUCCESS in 23m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e50d7751ee0844e6985b568c03d6d97c : SUCCESS in 28m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbe8dce9f4741119bd7e570c28aae49 : SUCCESS in 8m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9add3e08c2d481b9ecfc5de30c1c20f : SUCCESS in 1h 11m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1d91d066b6e48339acf9b827fbfe6e9 : SUCCESS in 1h 10m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c1fe50bd384e464ebcd1525b439f5083 : SUCCESS in 1h 37m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b28467dcc3cf4e5da056ee0e9ce122d0 : SUCCESS in 55m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/516b772f3a29453496fe69938ab183d4 : SUCCESS in 1h 49m 17s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/e80b4f9395e549f49864bd53f5545405 : FAILURE in 54m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ce14d07ba5ea46e5998451c7d6ab1103 : SUCCESS in 1h 20m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/678ae32fc293426f9a8db6d6875f09cc : FAILURE in 1h 22m 11s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/af0bb9a2388d42cc91779a67fa3356a7 : SUCCESS in 1h 18m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71b786b066db40a1a09bd21d78f8465a : SUCCESS in 2h 07m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7763d3251b3c4539adfe929b9bbafb00 : SUCCESS in 1h 04m 00s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/d21f350780c4413ab0e9dddb4b58a987 : SUCCESS in 1h 03m 29s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/995a06e847da4b99ba6f3b3c1a734a68 : SUCCESS in 1h 07m 38s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1d560caf1fc14df8936afa64d53f373d : SUCCESS in 39m 47s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d32526dc2a9d43d7bcabfa35389fe946 : FAILURE in 1h 02m 53s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4202d6f107314d819ee44667b3d0ffcd : SUCCESS in 1h 22m 30s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5d0baecc03a44eca87b8ee064cd41d23 : SUCCESS in 3h 36m 40s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/475be70b42c249d0b2e6392afa4ea1b4 : SUCCESS in 1h 18m 32s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/857665a059a648da88b81e12b6d2aa71 : SUCCESS in 1h 30m 16s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3cd6f3f435dd4d2887ab07b1d8c48aa5 : SUCCESS in 2h 00m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f9529806f61547809793ea122f4f4411 : TIMED_OUT in 3h 04m 14s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f7c7a944b45a4df2be485624b3fc33fa : FAILURE in 1h 51m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a88012cacec7f38afed6e702275e5631b6061363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-24 16:27:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b590023d65aab9b00cc2294cbc07aa41522267ea","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-24 18:21:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/768240/4 : SUCCESS in 1h 52m 59s","accounts_in_message":[],"_revision_number":4},{"id":"ce6e54f5162e7547763770e2b288d0c70f7a187a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-12-24 19:35:44.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2020/Arista_Tempest_Testrefs/changes/40/768240/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"fcc5385ceed1f2ccd7d10d1b95a86df8f405aa77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 19:44:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f944271b165d4ff282d1070f3bdf996d : SUCCESS in 37m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9225f4467bca4ef0a75a7ecbdc1d0f75 : SUCCESS in 35m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/885df1a475414f7f98b185465c4914ad : SUCCESS in 1h 34m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a3823e919db4739a2f71e7da7a7501b : SUCCESS in 1h 16m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b55d0c5d905443e0b2ae97a31043c46a : SUCCESS in 1h 10m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e7ef145616ce4554a9f72d69d0869eb1 : SUCCESS in 1h 26m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a5e0ff8a63141db9417ee0d47925bd2 : SUCCESS in 37m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b187ebd466749b9915991dc73febac0 : SUCCESS in 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b60a2a8d38a4aeab93d93c789e21754 : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94f86b3b0f3e4a25a62c14c1004496ae : SUCCESS in 16m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e8dae33672f42cc9299fab96d805d4d : SUCCESS in 27m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d00ed354e4e445697cfc401afb89fee : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a207ee36604060bcac7fed16fbae80 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/944a3715ad014143a5da562779424c64 : SUCCESS in 1h 10m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7128292379fb4e3e87995ecb80bb7e7a : SUCCESS in 1h 07m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/49cb32aa6f3e4e589a180ed68134e477 : SUCCESS in 1h 06m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ae47e6b4137431a8c992860aa077999 : SUCCESS in 56m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f17954dbc75040d093be0be321a5f062 : SUCCESS in 1h 00m 28s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/5b618dc0536740b0bb03e07813a4882a : FAILURE in 1h 10m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c1f375e4ba6f4a0bbe0c5391ab740ade : SUCCESS in 1h 13m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e9cec96f7e4478ba82c96271668a6c6 : FAILURE in 1h 36m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5594ca09a56549c5a809320ab58c0d25 : SUCCESS in 1h 08m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/afa865388e2b4bbd9a84abb4764f21df : SUCCESS in 2h 09m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43291dd0ecb24991b03ef4166d3cdc07 : SUCCESS in 59m 48s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/27a30375bcc94ef19f7e6adcb28535db : SUCCESS in 59m 09s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8ee559f78ba04e409e2588f461e499fd : SUCCESS in 1h 04m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/051b5572f6434257b20411c5386e14c9 : SUCCESS in 40m 31s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/24da0ce8512e4d8db3a51dcc3985a94b : FAILURE in 22m 18s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8d3ec1c932e141c7919b11d6bac04185 : SUCCESS in 1h 23m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/103af369c6d7472991a053838c17b546 : SUCCESS in 3h 13m 25s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/90fe8323cba5455f81fa86c80dac5821 : SUCCESS in 1h 25m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb9fbced379c444e900a4e54960aa94a : TIMED_OUT in 2h 13m 52s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d7257adb2b68434eb2293e1e77c8c574 : SUCCESS in 2h 00m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9c289275085148a587a8afe764bb1354 : FAILURE in 2h 24m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d67898a4b075410a9271d60fd3e0c94e : FAILURE in 1h 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e67460b1559f4cb81d2c1f233793c8d742411308","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-12-25 06:35:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"03bb072ce2ec21353dbdcd6bb16c257478105d0f","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-29 07:40:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e19cb4f7e63476ad6b793f2e48e765c663bde55e","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-12-30 13:03:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c57c721f07e11bd3cbd279a3e2029ef4df0592b1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-04 11:50:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f86c8b6e1b899df05d7ee2729a90394b7d40000d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 11:50:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"91ba27bbf83a6493e51644bbdf41546b90d7a8e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 13:54:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8d9d310288354f68a4c29b56f42161a3 : SUCCESS in 1h 23m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/154ef436118b4aa288aeb4f0ea63f26c : SUCCESS in 2h 01m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d85a978dec024667a29455b8b61316a7 : SUCCESS in 1h 31m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/03317b25a4e74318be43bd3831fc7ac9 : SUCCESS in 1h 02m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c44ac53bdc9401a971129209431359c : SUCCESS in 1h 46m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92eccfdbaace4919a5b40135a5f97270 : SUCCESS in 47m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e201fbaf06454fbd16cf6684585a6b : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b497078216e84dd783fd206ac86b35b2 : SUCCESS in 42m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96127d0bdeb84534a51ffbdfd52650f2 : SUCCESS in 40m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acaabb6e5ab5486985a5f28e3363fc41 : SUCCESS in 8m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9147a14e7a23407997d01720b5cfa563 : SUCCESS in 1h 05m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e214960c953f489597151e176ca0a652 : SUCCESS in 1h 51m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2475669d85bc4bcb9ad2871f04096de1 : SUCCESS in 1h 19m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f98d0a1d19894418a671e2da555eecfb : SUCCESS in 1h 00m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/021cd329ae1446a880885162e7faab10 : SUCCESS in 59m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca4423f0aea94f818d51f411eaed7060 : SUCCESS in 2h 00m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5591f26c04394b6bade41b36ffebb347 : SUCCESS in 59m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5d2811d7cb524006b643ef89cbe1ded3 : SUCCESS in 54m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/471f29ba75584fd895496211f49f0bcd : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":4},{"id":"bacdc1952506a7b62ae7d0d2849bf62471688e34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 13:54:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f7eb1e367ef715e71bf32eafda8aacd307aa7408","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 13:58:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6d9f488d03492fa27e6faee8ce5a2d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9e725ed72ed09c6815fc7c4c901c70bf2bbab24","revisions":{"fd8baad88a065e6f49eea0e98746c38241c39e21":{"kind":"REWORK","_number":1,"created":"2020-12-22 17:12:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/768240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/768240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/768240/1"}}},"commit":{"parents":[{"commit":"607f15c1ac1169710f01ba955a75e849a15c57ae","subject":"Merge \"Remove duplicate licese text from the br_tun file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607f15c1ac1169710f01ba955a75e849a15c57ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-22 17:09:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-22 17:11:41.000000000","tz":0},"subject":"Change the logic of \"catch_exceptions\"","message":"Change the logic of \"catch_exceptions\"\n\nChanged the logic of decorator \"catch_exceptions\" to provide a\nPythonic output: if succeeds, returns True; if fails, returns False.\n\nChange-Id: Idbf2ffb4ef403f343973bde3b3599f9cd3aaddf2\nStory: #2007686\nTask: #39975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd8baad88a065e6f49eea0e98746c38241c39e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd8baad88a065e6f49eea0e98746c38241c39e21"}]},"branch":"refs/heads/master"},"4837621d40549f306d99e1e0b84270e892a0bb63":{"kind":"REWORK","_number":2,"created":"2020-12-22 17:35:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/768240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/768240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/768240/2"}}},"commit":{"parents":[{"commit":"607f15c1ac1169710f01ba955a75e849a15c57ae","subject":"Merge \"Remove duplicate licese text from the br_tun file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607f15c1ac1169710f01ba955a75e849a15c57ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-22 17:09:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-22 17:34:41.000000000","tz":0},"subject":"Change the logic of \"catch_exceptions\"","message":"Change the logic of \"catch_exceptions\"\n\nChanged the logic of decorator \"catch_exceptions\" to provide a\nPythonic output: if succeeds, returns True; if fails, returns False.\n\nChange-Id: Idbf2ffb4ef403f343973bde3b3599f9cd3aaddf2\nStory: #2007686\nTask: #39975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4837621d40549f306d99e1e0b84270e892a0bb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4837621d40549f306d99e1e0b84270e892a0bb63"}]},"branch":"refs/heads/master"},"ae91533fd46673e7c894d430a0e2b60114a273b4":{"kind":"REWORK","_number":3,"created":"2020-12-23 14:30:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/768240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/768240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/768240/3"}}},"commit":{"parents":[{"commit":"607f15c1ac1169710f01ba955a75e849a15c57ae","subject":"Merge \"Remove duplicate licese text from the br_tun file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607f15c1ac1169710f01ba955a75e849a15c57ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-22 17:09:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-23 12:37:01.000000000","tz":0},"subject":"Change the logic of \"catch_exceptions\"","message":"Change the logic of \"catch_exceptions\"\n\nChanged the logic of decorator \"catch_exceptions\" to provide a\nPythonic output: if succeeds, returns True; if fails, returns False.\n\nChange-Id: Idbf2ffb4ef403f343973bde3b3599f9cd3aaddf2\nStory: #2007686\nTask: #39975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae91533fd46673e7c894d430a0e2b60114a273b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae91533fd46673e7c894d430a0e2b60114a273b4"}]},"branch":"refs/heads/master"},"b9e725ed72ed09c6815fc7c4c901c70bf2bbab24":{"kind":"REWORK","_number":4,"created":"2020-12-24 16:27:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/768240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/768240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/768240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/768240/4"}}},"commit":{"parents":[{"commit":"607f15c1ac1169710f01ba955a75e849a15c57ae","subject":"Merge \"Remove duplicate licese text from the br_tun file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607f15c1ac1169710f01ba955a75e849a15c57ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-22 17:09:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-24 16:27:13.000000000","tz":0},"subject":"Change the logic of \"catch_exceptions\"","message":"Change the logic of \"catch_exceptions\"\n\nChanged the logic of decorator \"catch_exceptions\" to provide a\nPythonic output: if succeeds, returns True; if fails, returns False.\n\nChange-Id: Idbf2ffb4ef403f343973bde3b3599f9cd3aaddf2\nStory: #2007686\nTask: #39975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9e725ed72ed09c6815fc7c4c901c70bf2bbab24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9e725ed72ed09c6815fc7c4c901c70bf2bbab24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
