)]}'
{"id":"openstack%2Fneutron~768459","triplet_id":"openstack%2Fneutron~master~Ifafb327f5b2e2a4b9e46e07cf896e29f9f24f8a4","project":"openstack/neutron","branch":"master","topic":"l3_qos_doc","hashtags":[],"change_id":"Ifafb327f5b2e2a4b9e46e07cf896e29f9f24f8a4","subject":"[L3] Document QoS support in Neutron","status":"MERGED","created":"2020-12-24 10:02:13.000000000","updated":"2021-01-07 13:26:57.000000000","submitted":"2021-01-07 13:24:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768459-l3_qos_doc","meta_rev_id":"c52d5dc53474f2ef10b84b05d33ae59b1b7f0c4e","_number":768459,"virtual_id_number":768459,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-07 13:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":2,"date":"2021-01-07 11:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-01-04 16:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2021-01-07 11:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-24 10:16:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-24 11:51:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2020-12-28 15:19:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-12-29 07:40:43.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-12-30 13:03:44.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2021-01-04 11:56:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ba545e59bd18b9ceb351da41de1023527017209a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-12-24 10:02:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"140d3150baf17f026530ff73d41572bfbdd27bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 10:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc344480c2764c86bbc2d4af8f853a2e : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8648681d1b4c0785cb994a798e83d9 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"41262844f13007bc6deaed105770c66fa9d668c3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-12-24 11:51:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/768459/1 : SUCCESS in 1h 48m 47s","accounts_in_message":[],"_revision_number":1},{"id":"824bf36b61039ecdf9cd45c0a5ddab3230c3f924","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-28 15:19:04.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for clarification","accounts_in_message":[],"_revision_number":1},{"id":"4da70ba6e5effeb6ec8516fb777899789ceefa69","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-29 07:40:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6333d8093bd0f726dafd98caea23c2c6de19a513","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-12-30 13:03:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"10862d175fc5835981d428052a0349bf0dc79895","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-04 11:56:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nOne small nit to address. Other than that LGTM for me","accounts_in_message":[],"_revision_number":1},{"id":"ee39bbf4d410d3dd5be2a0612a68c01d745f4d2e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-04 12:08:28.000000000","message":"Patch Set 1: -Code-Review\n\nThanks Slaweq","accounts_in_message":[],"_revision_number":1},{"id":"26c17ee644b7e048ceb686ab1ff1c9655292b4e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-04 14:27:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64b66111ec45863e1d7c349295af1f4a01780ca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-04 14:27:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c0503bc17585c1653202626ee6d2aa69d16e6f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:41:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c218ed10cd0641bdad5ffdeb356a4a6f : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b35e73c15a4fd1a95903134d9fe0b5 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ef4f119b2da88f0435408ff5da4f146858526d94","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-04 16:18:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/768459/2 : SUCCESS in 1h 50m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1f1eb94876eb52b573429c6a3cab09d1b54d9749","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-04 16:35:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6e6e64c801ec677a286bda1c0d9c9e64d90faa5b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-07 11:09:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f74c7f28732d4b6a9a079a1bb34fd44820abd761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 11:09:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5da8bbc5a7161eb06bc5de3eef70fec1b3de3420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 13:24:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46cc29f54984ba880629316024961d5 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a88a2378f64745ae13e27693394903 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":2},{"id":"8c51a427b8ec694e20bc9ede395c27dafa3b462d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 13:24:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c52d5dc53474f2ef10b84b05d33ae59b1b7f0c4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 13:26:57.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3be6f0655e4b549a46df350ade37da : FAILURE in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"74490e09ffcce6d5ef3fa7936dc60cf8c5db845d","revisions":{"e681507bcc25aded55866ede30e03e9b60a06a71":{"kind":"REWORK","_number":1,"created":"2020-12-24 10:02:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/59/768459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/768459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/768459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/768459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/768459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/768459/1"}}},"commit":{"parents":[{"commit":"607f15c1ac1169710f01ba955a75e849a15c57ae","subject":"Merge \"Remove duplicate licese text from the br_tun file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607f15c1ac1169710f01ba955a75e849a15c57ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-24 09:45:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-24 10:01:42.000000000","tz":0},"subject":"[L3] Document QoS support in Neutron","message":"[L3] Document QoS support in Neutron\n\nDocument the existing QoS extensions provided (for L3 router and\nOVN L3) and the QoS rules supported (bandwidth limit).\n\nChange-Id: Ifafb327f5b2e2a4b9e46e07cf896e29f9f24f8a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e681507bcc25aded55866ede30e03e9b60a06a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e681507bcc25aded55866ede30e03e9b60a06a71"}]},"branch":"refs/heads/master"},"74490e09ffcce6d5ef3fa7936dc60cf8c5db845d":{"kind":"REWORK","_number":2,"created":"2021-01-04 14:27:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/59/768459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/768459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/768459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/768459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/768459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/768459/2"}}},"commit":{"parents":[{"commit":"607f15c1ac1169710f01ba955a75e849a15c57ae","subject":"Merge \"Remove duplicate licese text from the br_tun file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607f15c1ac1169710f01ba955a75e849a15c57ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-12-24 09:45:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-01-04 14:27:37.000000000","tz":0},"subject":"[L3] Document QoS support in Neutron","message":"[L3] Document QoS support in Neutron\n\nDocument the existing QoS extensions provided (for L3 router and\nOVN L3) and the QoS rules supported (bandwidth limit).\n\nChange-Id: Ifafb327f5b2e2a4b9e46e07cf896e29f9f24f8a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74490e09ffcce6d5ef3fa7936dc60cf8c5db845d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74490e09ffcce6d5ef3fa7936dc60cf8c5db845d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
