)]}'
{"id":"openstack%2Fneutron~769896","triplet_id":"openstack%2Fneutron~master~Ie4d7f0b85fe8997ea695fb633202c730a31a9b63","project":"openstack/neutron","branch":"master","topic":"fix-ovn-extensions-description","hashtags":[],"change_id":"Ie4d7f0b85fe8997ea695fb633202c730a31a9b63","subject":"Update path of the ovn extensions doc","status":"MERGED","created":"2021-01-08 09:24:28.000000000","updated":"2021-01-12 00:17:35.000000000","submitted":"2021-01-12 00:15:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769896-fix-ovn-extensions-description","meta_rev_id":"81152e1088dbbc75a26ebf846e72b220788f15f2","_number":769896,"virtual_id_number":769896,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-01-08 12:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-12 00:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-11 21:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-01-09 14:20:24.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-11 21:54:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 11:14:28.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-08 12:03:07.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-08 12:52:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 13:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-09 14:20:24.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-11 21:54:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e4fc5f574eaca5c83bc088a85732d0b1ef707836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-08 09:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4de7b5391cdff8550eecb18489995d12a6b9a25b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-08 11:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/769896/1 : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3f2deedf2622d96bf6329743822e62a229e6d337","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-08 12:03:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/769896/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/769896/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a23732243da8d322a04956f3f4380e5a197aef86","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-08 12:52:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/96/769896/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"856b5bec42debc9810c00a79cba988f51bef4ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:22:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3ef5471d86634eba92a51553853d065d : SUCCESS in 43m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ccf22e1f03b44628a0a56e416424118d : SUCCESS in 34m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6dc6b458b2444dfb5a0f7436b45bb0f : SUCCESS in 1h 52m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/61e7b4fa56ee4beb84f93689c551e1ce : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b021a6f8429947cbb2b74e9681f4412e : FAILURE in 1h 27m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/95ecabf9d4884519ac28f8a6e254215d : SUCCESS in 1h 35m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/347b69ed2fa34c7aa5c1f0f94fb5cd99 : SUCCESS in 34m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f84c796da5640829a16d115c885dcbc : SUCCESS in 38m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8617c29867464eb64638dac36e0b8c : SUCCESS in 19m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be9178f23f0b40d89d9e40e1922bbef1 : SUCCESS in 31m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/745f5beaf796406bb0ca72e01cff3728 : SUCCESS in 31m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22ae0b0eebfe482a80483dfc61bc68d2 : SUCCESS in 40m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05598cb3e9b3477b9dd2e54dcf64b037 : SUCCESS in 10m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/798035512e3645b98c223d0d99c9dee2 : SUCCESS in 1h 06m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c227b9bc6cd54d1686dd787da1832360 : SUCCESS in 1h 08m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/65e5ca62464e4e588ee093d9f46af750 : SUCCESS in 1h 33m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de79c168eba54ba8a4572581039646e9 : SUCCESS in 57m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b68a161af2ba455aa0158cc963f7652a : SUCCESS in 1h 02m 51s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/4b4d56b957d843348e0ad2d45fc4e841 : FAILURE in 51m 59s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/69b9e707c5fa4af281a014af348ca452 : SUCCESS in 1h 40m 22s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/909beb2fbc97465288d5ab65768f4139 : FAILURE in 2h 28m 03s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/02c8bd36023a4d93bc42a51ce1ba4ad0 : SUCCESS in 1h 11m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2aa52c6b8f64c0cba81dd2bc1a700f8 : SUCCESS in 2h 02m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cc62f61acaf434f87869fb3f65c7e5e : SUCCESS in 1h 09m 05s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2238fc18e66e45b38b1d30c4dbdc6299 : SUCCESS in 1h 00m 16s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ca71590504b44dc5838e2d39c2067fb1 : SUCCESS in 1h 08m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1bd68960321146cc85e275d8cf00a315 : SUCCESS in 33m 44s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/aff69757618e433d9a433e1fc83afefb : FAILURE in 1h 07m 26s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b7a4de62f57c46f08807e054b84446b9 : SUCCESS in 1h 20m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b3f1eef10cc0431683b9746c3eee935c : SUCCESS in 3h 29m 35s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e799d4383da844f2bf5d725e3be1dc48 : SUCCESS in 1h 16m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/03ee0d43f2464d08b42bd87f5c80dd1b : SUCCESS in 1h 31m 11s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c78ff76b4b5d4697bcfed3ee3c55ae4d : SUCCESS in 2h 08m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/47fb69c0a1804269a8998ffcae562f6c : FAILURE in 1h 43m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9d9b00579d5b438ba099069cec494613 : FAILURE in 1h 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3378c389dd46fb22bfd7208bad912f14315c9d5","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-09 14:20:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fe8835c7d057b09d8fc162aabee87d5ef2e4f84","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-09 14:20:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"779cda49d2bd80c7dfabce63f368d37a7dc50971","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-09 15:31:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/96/769896/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/96/769896/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6dd3e49fd77f35ccbea04ca31c027c049d469056","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-09 16:05:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/769896/1 : SUCCESS in 1h 44m 17s","accounts_in_message":[],"_revision_number":1},{"id":"429439ad597f21206f88a9f936af3ea6b9620ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 17:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a0ad9782504c476db0e58b4efac86718 : SUCCESS in 40m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7aa75e4ee97b4634bf2b56e09707467e : SUCCESS in 33m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0cf4ee3c67a54bb88eeb994718c25b28 : SUCCESS in 1h 40m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d8ccc90a75054acba3b28e3dea43da7c : SUCCESS in 1h 54m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/68b59cb2a65c42f180c29583b9ce8e3b : SUCCESS in 1h 45m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/59d451e8383f41c0b5bb8d3d19ad0f90 : SUCCESS in 1h 53m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75898e347fff439e87a17beee69a0842 : SUCCESS in 34m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/153da62621ab481880d4b48eb875892e : SUCCESS in 37m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885a7942fc0549028b5f158fd78f274a : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb4d34c16184e718bf7bc951ef503a1 : SUCCESS in 31m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bace330336a74d66b4e68b4766fe4810 : SUCCESS in 33m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/686ddd27c47b4989804f10e4d1863dc6 : SUCCESS in 29m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82fb2a06de96401cbfe09b1a02abbe40 : SUCCESS in 8m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3486f50f5f624e0c89d74b139afe833f : SUCCESS in 1h 08m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b10145912dd94654895eb4e2e92555e8 : SUCCESS in 1h 21m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/943b37f38639478c939e75e9ba830922 : SUCCESS in 1h 31m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00b2c3baaf51453e8469094ac0bafe2a : SUCCESS in 1h 01m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/273f2d5233524ca3b926ca2e1e250842 : SUCCESS in 58m 54s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/044656559b37423084f77c53d05c4199 : FAILURE in 57m 26s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eaa577a8c3904fc092a981c88f8d9ad8 : SUCCESS in 1h 17m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/79e638c217d94a1b9cd0c480191b9e6d : FAILURE in 1h 14m 01s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/72c14ed7b7ec419c8704e9a9d0432449 : SUCCESS in 1h 11m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/758b1c8779e24c5a98f20a2231933683 : SUCCESS in 1h 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b66a7a909b54199911604a16d33a544 : SUCCESS in 1h 05m 22s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8c2b9ddde51d4583a4c51c68d1c4ef72 : SUCCESS in 1h 02m 21s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82604667d0bf434e8f7dad20593c49b9 : SUCCESS in 54m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/32d058e7b04845618347e07390e94f0c : SUCCESS in 38m 14s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/d2ce3b5c972d4670a3ae91ab25952089 : FAILURE in 1h 06m 28s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/099d66ba2bb4439b914c264716d53308 : SUCCESS in 1h 13m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6ad75bc8d7d24dad9f825adf01c4da3e : SUCCESS in 3h 31m 38s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee97a957a18f4b2daf7b03a68becf7c4 : SUCCESS in 1h 29m 12s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/480b4f97d9d4487fbc59b45fe4d66b15 : SUCCESS in 1h 47m 00s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b9ec0a61443241d587cf04d6ca18826c : SUCCESS in 2h 01m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/aa6c7ff3deb94edf978118ee6e74d496 : FAILURE in 1h 39m 51s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3fed9088f55242dcb47a79e6acf67a60 : FAILURE in 1h 41m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd81d0e79dbad58c681860a3927e986e84ccbcec","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-11 21:54:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6cd7b3ee28c3ced63c562f044decc9999fd914b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 21:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cceaeca635e59c9aa10af14b309d1b90784c434c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 00:15:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6677aba7e88b4826b6f6a5488c78bee6 : SUCCESS in 48m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aecf75e5f1e94dbc81c0060e3d754357 : SUCCESS in 1h 38m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e50e0e17ba2b464f873aeaf8c1389690 : SUCCESS in 1h 36m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5f001d7c4d044a0b87670a5b6519b1c0 : SUCCESS in 1h 39m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1ea45268f3141b5bbb14fe63d3f1dd1 : SUCCESS in 1h 31m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcb1a6c96c6f4d15afcba7e4b20646b7 : SUCCESS in 48m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81b1b100d6e423d917f41990be3b911 : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d7157d95cc14a358d9bc70acc0cad9d : SUCCESS in 24m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/737fdff6fbe94b8dbe525a7c605363fc : SUCCESS in 34m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b170d807b54c403aba7a3e39a1bee23a : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7feb4ae584a54eb2b02bb428ec05d4af : SUCCESS in 1h 08m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/075781bf79174e29a433bca32c7a69a6 : SUCCESS in 1h 10m 02s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a24f30e95f1340c886937c88ea2b482a : SUCCESS in 1h 24m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b1afee8161847ae8f308ce6403e257f : SUCCESS in 1h 09m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8d5316d614454ccb987795bf73331c58 : SUCCESS in 58m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90795f482031456bad91cd2e27f52792 : SUCCESS in 2h 08m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2074a1f6a0d4725a8850c650d3bbcd3 : SUCCESS in 1h 00m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bef417170ef74cccbe65a82bd2d3dc45 : SUCCESS in 1h 19m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e9267939fd3e4c2b99bd1cbfae6259a0 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c1b3920c91ec3e033e701466726ade238249d84a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 00:15:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81152e1088dbbc75a26ebf846e72b220788f15f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 00:17:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cbaf5101ad46b993e5adbb0d81d008 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a97b20580a4e284f7226bf1cc67586caf3d22fc8","revisions":{"a97b20580a4e284f7226bf1cc67586caf3d22fc8":{"kind":"REWORK","_number":1,"created":"2021-01-08 09:24:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/96/769896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/769896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/769896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/769896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/769896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/769896/1"}}},"commit":{"parents":[{"commit":"588be19298447c477bd86636431437e3bcfe69c4","subject":"Merge \"[L3] Document QoS support in Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/588be19298447c477bd86636431437e3bcfe69c4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-08 09:23:10.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-08 09:23:10.000000000","tz":60},"subject":"Update path of the ovn extensions doc","message":"Update path of the ovn extensions doc\n\nIt was changed after migration of the ovn to neutron repo but\npath in the comment wasn\u0027t changed.\n\nTrivialFix\n\nChange-Id: Ie4d7f0b85fe8997ea695fb633202c730a31a9b63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a97b20580a4e284f7226bf1cc67586caf3d22fc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a97b20580a4e284f7226bf1cc67586caf3d22fc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
