)]}'
{"id":"openstack%2Fneutron~769975","triplet_id":"openstack%2Fneutron~master~I989b6b1903e3296d20e4ef6a7770aa7b78f58b46","project":"openstack/neutron","branch":"master","topic":"bug/1910717","hashtags":[],"change_id":"I989b6b1903e3296d20e4ef6a7770aa7b78f58b46","subject":"Randomize port name in \"BaseOVSTestCase\"","status":"MERGED","created":"2021-01-08 18:08:26.000000000","updated":"2021-01-12 12:27:59.000000000","submitted":"2021-01-12 12:25:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769975-bug/1910717","meta_rev_id":"bc98fceb59731893d188271132ee8468bd5ca272","_number":769975,"virtual_id_number":769975,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-01-10 18:03:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-12 12:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-12 09:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-12 07:17:55.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-12 09:47:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 19:07:46.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-08 19:30:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 20:31:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-09 01:10:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-09 14:33:38.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-01-12 07:17:55.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-01-12 09:47:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"92f73a50b29517a40a39022e15d943c00eba7e6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-08 18:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17a183df8581eb80cfc41dd43e02dae98657daff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-08 19:30:53.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/75/769975/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"131c3c72661e9552bab554db6a1df9c617d7d915","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-08 20:31:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/769975/1 : FAILURE in 2h 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4da084536496889af5a1b28037b57c7a7c402299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 01:10:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12ceba87cc6a40329718a0568d0496d5 : SUCCESS in 51m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98cad18df6a14b50b1d2884a261447e4 : SUCCESS in 49m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/21339b56f50c4140abcb1f78849980fc : SUCCESS in 1h 36m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/27f0207e6a5a483a88bff7f277f7d76e : SUCCESS in 1h 46m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ad2b44a859384de49bb9e909bb6ffcc7 : SUCCESS in 1h 19m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ba724d8ff0954023a60027a3385efc8e : SUCCESS in 1h 37m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35333fb446cd4c9fb2071fbe1246d380 : SUCCESS in 34m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f71a84e5fcaf46e08c115063fb32267c : SUCCESS in 54m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ececcdffa884df0a9cdd31764c711cf : SUCCESS in 15m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e036993491e74a71a9828c637d911b42 : SUCCESS in 44m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0edc68d64d2640cea01c9a488550140d : SUCCESS in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34bbd89c19b3424cb4816bf4981b5e99 : SUCCESS in 38m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4491738502f4582bf6a82884a829b7c : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0338413c7a68467693884fb06117eb43 : SUCCESS in 1h 08m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/025f54ac067244eb8ca016f904900979 : FAILURE in 1h 55m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/887f15bff0fd4e0b9cf3581c843fc6fd : SUCCESS in 1h 43m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b370042c8a9441ca947be76c5c9929d1 : SUCCESS in 1h 02m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b0a12fa88cf420b96ba7eabf0fd37aa : SUCCESS in 1h 04m 19s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/d9b0f3b2a55847b9a0d5da739bdcc49f : FAILURE in 1h 01m 26s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9fadc53759de4ee3bece308b17031aa3 : SUCCESS in 1h 27m 24s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4ee98a8fe13e403aa0c1b4b9acb8336c : FAILURE in 1h 16m 12s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cdc01c5c46144e88ae02ddd30b1a0b61 : SUCCESS in 1h 14m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44f18d8592d34a49844286d3b4559d45 : SUCCESS in 2h 14m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33a580b2cb7d481cb054607d86cd60e9 : SUCCESS in 58m 58s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/8f189617d58b40e8a5152075e9f5bdba : SUCCESS in 53m 25s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/736a9936f9f449299667e9accec85abf : SUCCESS in 59m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a68d617235548749b1bfe54521b782d : SUCCESS in 42m 30s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/82b99eeace47426188e37e6ccb61bea7 : FAILURE in 1h 08m 50s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9e473c39ba9d494db62920511a448895 : SUCCESS in 1h 23m 40s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/271b23b7f7bb4644a44b39e2fcc37e6c : SUCCESS in 3h 35m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/637ae353bffb4f20aeb6a9d51194324a : SUCCESS in 1h 17m 13s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/55a9e22494374fcbb2fec2196fc294db : SUCCESS in 1h 43m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8b35e5b0bc0b4e7080de70ac5c9c7f9e : SUCCESS in 2h 05m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7403eb7c126f481ebf0584307870b5f5 : FAILURE in 1h 43m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0ad9cde28c8b4b9ba03b704c189c5adf : FAILURE in 1h 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5a52ff222b64072f135ebf77049ebb4a72967f9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-09 14:33:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d58699a5b3d1d5175fd7c8a43bc0ce28e74ad13","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-10 16:40:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67ff43a7b3895bb5100ed41632e668ba34efded4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-10 16:40:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2c32d53df3cc73947cae815c4e6eee03dfe9b521","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-10 16:55:37.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/769975/2 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b246426b3dd369b51658b56c6ef1bf4b064a0410","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-01-10 18:03:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2021/Arista_Tempest_Testrefs/changes/75/769975/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"19b41370ef05e02d93841fc5c00044b9f873a152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 19:45:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3b1735b1762b44bdaf5f0899acb49aea : SUCCESS in 47m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10ff95430c56445b89e90104e7d10d27 : SUCCESS in 29m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/875d56e7f92b441ca2af435f4fe3eac3 : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/14a3c3185f2945e6b148c40f49abe467 : SUCCESS in 1h 36m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/931552e012404588be7c77f49e742770 : SUCCESS in 1h 23m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7ed75192b059434c89d6056032a9f709 : SUCCESS in 1h 23m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87feb6905bb74d6ab4a2a3263daf4abe : SUCCESS in 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/732fbe7a6bef459cb63ca12082eba0c8 : SUCCESS in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca745581b2034bee9d308052e1466dd4 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a632fb43eb4449939baca2a91c4136 : SUCCESS in 29m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ad233358754d6da28b3977a66909b5 : SUCCESS in 37m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68616a3005d24b16be5c7883814e3296 : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8c9ed493cd459e8a8fbdc30aca2852 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b3d736ca4d74589b86ab150f2070220 : SUCCESS in 1h 12m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/44f29cb002124c9c9691fb4a621c9f6b : FAILURE in 1h 03m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f45bf04217b445b38ef7a5fdd3fba0ad : SUCCESS in 1h 27m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2c4222bb4e64988884a81692b4db06d : SUCCESS in 1h 05m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58b6f6ac908b433a8d927f9d28118487 : SUCCESS in 53m 50s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/72807be9bb5445108cce9973acd4c852 : FAILURE in 52m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ed046b31362040018946a168ca17dceb : SUCCESS in 1h 19m 07s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ff521c9d3a784ae89ed61d61ff95be30 : FAILURE in 1h 23m 41s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/31cb065cf5c44d828000821959f59d08 : SUCCESS in 1h 20m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72a18c77e8bc40568ab7e17c951bfc90 : SUCCESS in 1h 54m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91cd1e320a264ac29f8fc38a50a90bc9 : SUCCESS in 57m 17s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9781eb7a384d4da087257b892650d742 : SUCCESS in 59m 06s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1250dadc9c844755b99a858d29b70766 : SUCCESS in 1h 05m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2adb11c980c045d2b8e7940c7cf5df51 : SUCCESS in 32m 28s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/7482fd8f2c664f4391db7ad69605cca7 : FAILURE in 1h 02m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ad9f2a46eeef4d5ca5cd452ad40c52f4 : SUCCESS in 1h 17m 14s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2f1de48ec8bb4c15bc7d61ff4f915f44 : SUCCESS in 3h 00m 33s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/743a570c6fdb4d66a1da30f82a55cddb : SUCCESS in 1h 04m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14bddd3ec98c46469c5cad7cd512df65 : SUCCESS in 1h 42m 27s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0cf24fd16d4246eb803b433c935c78dc : SUCCESS in 1h 43m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/379c0d6c1494448a8aba6042e4d6c8ba : FAILURE in 1h 48m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7072438a2fc04be5a9c5f1cc5d1449dd : FAILURE in 1h 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"877b064967723742bfca3a3a2b2547f08394e469","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-11 17:02:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0918e0343be28198deffb44060d37f2137574c07","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-11 18:52:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/769975/2 : SUCCESS in 1h 50m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9e91ca427c28133f002bcf6d023c6df9745b024e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 22:46:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad626b4a1bb14632a94f2cca60ab48da : SUCCESS in 1h 06m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec06ce3a9c6a4794a9f76c5d52a5a0ac : SUCCESS in 33m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/94f853105d78451ea70edeb082b3bd1f : SUCCESS in 1h 34m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/28a53303d2e0452f9e2a841b2c108d7f : SUCCESS in 1h 14m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/135409291e0f4e22a22b71de70ed3d74 : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cbaffd9d97be45e09744ce554c270ec2 : SUCCESS in 2h 08m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54d9d21af4c543f6a5d193a1d7d46226 : SUCCESS in 48m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a5fcda0cba5435ca4cbcb953ab53413 : SUCCESS in 42m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc55814a27a415cad5b23abbb36d332 : SUCCESS in 12m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ce139993cc4af598ac4cc396f4368b : SUCCESS in 35m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362cfafd008148a9ac45aa68af62ab3f : SUCCESS in 36m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81435371ddec4749a09e9df14ff48f02 : SUCCESS in 35m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090d0871d6a64ad0a6dca742005ddb49 : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5331d0a035f949b5b83473f252d0794b : SUCCESS in 1h 16m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/faa7954990ad4c5ba144b201f8801c58 : SUCCESS in 1h 16m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2c22c69a03a14800bb6354771c4b8057 : SUCCESS in 1h 44m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d360cdc1fb5244b79614c50651664e1f : SUCCESS in 1h 00m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f82cc215fb247c08c1a9f9fa420a401 : SUCCESS in 57m 41s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/8f6cde67309d400d8d26a8d053fc8935 : FAILURE in 1h 04m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f22933e50c64f8abc6716df0c4c9a60 : SUCCESS in 1h 03m 53s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1acf05f79874fd4bfdee6d1a8eacc49 : FAILURE in 1h 34m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/974ddfcb37e14cc5a43e55ed49f77470 : SUCCESS in 1h 15m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a08b32e95f44cd9bf8bc1a02b4f3d11 : SUCCESS in 2h 04m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34519676b99e4402a61120fb6367c49e : SUCCESS in 1h 04m 40s\n- neutron-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/c0dcacd55fcf4edb95ba49e6d2b2d208 : SUCCESS in 1h 09m 54s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6b14ff225fe34f3fa9b333fbfb6dea80 : SUCCESS in 1h 11m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f66bf4cfbab04c028bd14545a1c5c35e : SUCCESS in 41m 20s\n- neutron-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/9cd911b9604c41aa9736020229882cf5 : FAILURE in 1h 12m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c4fd025d0d914270b62de5d28b324952 : SUCCESS in 1h 17m 01s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d793d2c0939343209119aeeabaf55f70 : SUCCESS in 3h 23m 51s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7727804c0293405ea4adcacf279064fe : SUCCESS in 1h 17m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/63974c0daad74ed8939a6212029cfe1e : SUCCESS in 1h 43m 00s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0eba60c5db8646ffa1b552cdf627f77a : SUCCESS in 2h 04m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/31f450fdf23945eebf196a3bb170f503 : FAILURE in 1h 52m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c7e7ac3be82e4091adb9b241bfc06e05 : FAILURE in 1h 48m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4cd2ab9e9e2b472a885b34efcbe1ec616b418c7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-01-12 07:17:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"07bc917c00fe9337bea8210e54ec7dcfc3f0400f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-12 09:47:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64a079f7ee57136713b9c76b49479af2ccfb6ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:47:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8a2449645c12d96082a964f067c8264c1fe6e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 12:25:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/28f1f6f104d14c6e8d2c6ad2afb871b2 : SUCCESS in 1h 04m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ee6530216a4c44deb3f3ed2defd41d81 : SUCCESS in 1h 34m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7439d5448d4493781371b771508de21 : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6112d07b44e24751885e3402cad256da : SUCCESS in 2h 30m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d1d62fd5f8243ee99f46d701141c1eb : SUCCESS in 1h 18m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0542c556e6440a598ad9fe8dedc7f6c : SUCCESS in 44m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52182f66fab645899add98016b939996 : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1547db2b43d3443fa7f56fcb38bdd329 : SUCCESS in 35m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b6eee6293954d2aa7f005b82d353570 : SUCCESS in 36m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4942aed86cd46a7ad940ddebe1276f9 : SUCCESS in 9m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2fc1d0d81a64bbeb8398e8d62c4f75f : SUCCESS in 1h 13m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aeeaeef8bae44b87a4aaf1afcbeae587 : SUCCESS in 1h 12m 09s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3d86327bb72946ca8f6a41e432e2ab26 : SUCCESS in 1h 28m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9267ef5367714c549b70f6a9a2c1c174 : SUCCESS in 1h 37m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ea3056f259dc400f93f87649c42375bf : SUCCESS in 1h 16m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/feb3c4eb66bb41479d4f0e9635513095 : SUCCESS in 2h 30m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50ffb7e54af648c2b93e25b20a82dcd0 : SUCCESS in 1h 06m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3c0d7c91d267433bb533649a279a03d8 : SUCCESS in 1h 11m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/12bd5b8a89d14134b28f1e1deee4bb92 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f219d7354336130508f24ec82844fc411cb46356","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 12:25:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bc98fceb59731893d188271132ee8468bd5ca272","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 12:27:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd0d50368b14377b0a48813c0725066 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82ddc5ee1ecfd92c998787cf62ac2a099ee08a61","revisions":{"69f2f0032a2ec5382010c2e29d3067b4f024a59b":{"kind":"REWORK","_number":1,"created":"2021-01-08 18:08:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/75/769975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/769975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/769975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/769975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/769975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/769975/1"}}},"commit":{"parents":[{"commit":"80d56920f77603fb89549fe5340bf15fe7b3d915","subject":"Merge \"[Docs] Guide about running and debugging tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80d56920f77603fb89549fe5340bf15fe7b3d915"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-08 18:06:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-08 18:06:27.000000000","tz":0},"subject":"Randomize port name in \"BaseOVSTestCase\"","message":"Randomize port name in \"BaseOVSTestCase\"\n\nRandomize the port name creation in \"BaseOVSTestCase\". Also check the\nexecution error when an OVS port is created.\n\nChange-Id: I989b6b1903e3296d20e4ef6a7770aa7b78f58b46\nCloses-Bug: #1910717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69f2f0032a2ec5382010c2e29d3067b4f024a59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69f2f0032a2ec5382010c2e29d3067b4f024a59b"}]},"branch":"refs/heads/master"},"82ddc5ee1ecfd92c998787cf62ac2a099ee08a61":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-10 16:40:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/75/769975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/769975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/769975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/769975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/769975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/769975/2"}}},"commit":{"parents":[{"commit":"80d56920f77603fb89549fe5340bf15fe7b3d915","subject":"Merge \"[Docs] Guide about running and debugging tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80d56920f77603fb89549fe5340bf15fe7b3d915"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-01-08 18:06:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-01-10 16:40:53.000000000","tz":0},"subject":"Randomize port name in \"BaseOVSTestCase\"","message":"Randomize port name in \"BaseOVSTestCase\"\n\nRandomize the port name creation in \"BaseOVSTestCase\". Also check the\nexecution error when an OVS port is created.\n\nChange-Id: I989b6b1903e3296d20e4ef6a7770aa7b78f58b46\nRelated-Bug: #1910717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82ddc5ee1ecfd92c998787cf62ac2a099ee08a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82ddc5ee1ecfd92c998787cf62ac2a099ee08a61"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
