)]}'
{"id":"openstack%2Fneutron~770033","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Id5d8ac09a38c656619f88a6f87b8f384fe4c55a8","project":"openstack/neutron","branch":"stable/train","topic":"bug/1903433-stable/train","hashtags":[],"change_id":"Id5d8ac09a38c656619f88a6f87b8f384fe4c55a8","subject":"Improve the CIDRs overlap check method for router add interface","status":"MERGED","created":"2021-01-11 09:02:04.000000000","updated":"2021-01-15 16:02:35.000000000","submitted":"2021-01-15 15:57:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770033-bug/1903433-stable/train","cherry_pick_of_change":761829,"cherry_pick_of_patch_set":5,"meta_rev_id":"0937101285550480a788d9fcc3a62427d8a87400","_number":770033,"virtual_id_number":770033,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 15:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-01-15 11:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-01-12 10:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-01-15 11:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 09:16:47.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-11 11:15:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 10:09:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-15 11:28:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"4d644098a41cdc36cff718622b95684bdc58fa6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2021-01-11 09:02:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd88eefba985d3ad0f0de6d282f47a6edc7b1106","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-11 09:16:47.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770033/1 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"21cba030b228b554e6fd6907827764009b5cf6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:15:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/6364d149a4724dac915263d9c018e5ce : SUCCESS in 46m 20s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/4b64c53e89764f2fa0c914ab92bc151f : SUCCESS in 28m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ae8f228aad554d15914e374f293ca87a : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/665a2bab5c2e4ec8bd9f5470e8e8a756 : SUCCESS in 1h 31m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/3198451596224298971fa2bef0bedbbe : SUCCESS in 1h 17m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31a7aede3f7c4291bc6a9e0e2fc054ce : SUCCESS in 27m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d80e8ac66d744f083b8c89334f8fd64 : SUCCESS in 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3edb42759bed4ddb9ddc5d8e4d504f89 : SUCCESS in 14m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d3e6c5c360440568604855e3133deaf : SUCCESS in 22m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de888334c7d94ff09b4bbda1b577a5fe : SUCCESS in 21m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e13a84e557a42f3a632780a6fd2c86e : SUCCESS in 22m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80da6c60094847ea8194d73c345811e8 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c9df448737ea4c6880128ab36ac80ed7 : SUCCESS in 57m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93080300d1b64c4193e48758846dfd17 : SUCCESS in 1h 08m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c78d7cada80b42a1b83697ad7f84308e : SUCCESS in 40m 22s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/54f0a3b4eaa14ff4a8cba31cd582acc1 : SUCCESS in 40m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8a9b675789d84f9eb431c9fba564dad2 : SUCCESS in 58m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6e34f33314bf41b7bc5e2033f23d5faf : SUCCESS in 1h 56m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0e9c2b9c91fe418d885af16ed55800ce : SUCCESS in 51m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/08a85aed743b4af3b2feda9ad8c1a3b5 : SUCCESS in 1h 08m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba719bcbc72245e6a6ebfc71ba7bd071 : SUCCESS in 1h 09m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d9088493be64e35a6ac453395402489 : SUCCESS in 1h 11m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9ed182120057423ab749e88039ca3ca0 : SUCCESS in 1h 42m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bc0bcb4ae9b945f2a88fc0343fe08ae2 : FAILURE in 1h 52m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06c1a056cd8243f79625df7afc69091f : SUCCESS in 1h 13m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c91b539fa5744609b5cdd86d465bbf66 : SUCCESS in 51m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/90027630048044fa9aa2ff95a52c57d0 : SUCCESS in 1h 07m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e72eac54abf04ea4975a0997fb1b9781 : SUCCESS in 1h 55m 16s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/385cf5ce288f4b49aa6a7ba8c38abb43 : FAILURE in 16m 35s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/069f9c01874d47b7a7237ef7c49528a6 : SUCCESS in 59m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00cc1cb1e5db4aa2bf08a42395b36cdd : SUCCESS in 55m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/530d4e75f976476fb288734a32d60103 : FAILURE in 1h 20m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac80ba6975a447db82ca6621217ac6e2 : SUCCESS in 40m 35s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d29163117c894d52b5650f54a91abc2e : SUCCESS in 54m 31s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/474683bdcb43407cb73371520116b745 : SUCCESS in 1h 06m 34s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ec225327fa5477298450a9faeffa267 : FAILURE in 1h 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19ecc7ddad297ce07acc4be3ed968fae84296e24","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-12 10:09:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a58cf2f9d89ec3c3b8fce12a037404f8695ba67","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-15 11:28:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23f7cb05fc84237e7ca2bdcf259cf0c6244cd822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 11:29:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18cb63f634177c4c0b13b25e272f587a5f3d6ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 15:57:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d586fd0393274d17a57f52e05ad383a8 : SUCCESS in 47m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bcbb337573840fd9a0bfaf2c52eba5d : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b8524bd6ee4ccabd7344ebe9712aba : SUCCESS in 11m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ebd3492c3554a2bbdbb23efdb20e02d : SUCCESS in 15m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cb82f5289054139acc512338def8f3d : SUCCESS in 23m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3734805281548ddadabf9e340e91eac : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6bf39d66a0a4e43ac1637b50eba0e83 : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/f9d7019161784e02ad2c4fb841786cb3 : SUCCESS in 1h 14m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8fc017e0ee424a05bd657faeaa8ab830 : SUCCESS in 1h 45m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/86cbfc7c9f43492a88666b349f4ea3e5 : SUCCESS in 51m 04s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b28bd354d4424298a0b0d08624d20866 : SUCCESS in 44m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/72ce3f0344b445cba9be9b1a538eba8d : SUCCESS in 1h 28m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/91eb1ab5b67d42d788d4be621c025f63 : SUCCESS in 1h 13m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/045edddbd58f48a286303505b452e831 : SUCCESS in 1h 17m 04s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d437bff86b14ad1b899855c8462878d : SUCCESS in 1h 15m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bdcba773cf74f3289988bf5c805a80b : SUCCESS in 1h 03m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0f7587ab9e3411899c49b12fc86409d : SUCCESS in 1h 06m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fad5a65f022d46cfba5153f4b6429c5a : SUCCESS in 50m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f56fc3b7e27a476db6ec13c1acfbe2e2 : SUCCESS in 2h 16m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dbff64d3ac214d429d77d5df6a0c4f61 : SUCCESS in 59m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/363093b3a3044c949d0e8edf6765bc7b : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":1},{"id":"efef315febab017ddd9976086b486533f5502a08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 15:57:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0937101285550480a788d9fcc3a62427d8a87400","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 16:02:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714b03cbc9d24d81aeb06644c66854f0 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf4d045ac5583ecca291f38da745c62fd1d5d688","revisions":{"cf4d045ac5583ecca291f38da745c62fd1d5d688":{"kind":"REWORK","_number":1,"created":"2021-01-11 09:02:04.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/33/770033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/770033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/770033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/770033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/770033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/770033/1"}}},"commit":{"parents":[{"commit":"c9416b10136ac4c6ba8943431b6b645f8baecfe6","subject":"[train] Fix tests with new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9416b10136ac4c6ba8943431b6b645f8baecfe6"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2020-11-08 07:06:10.000000000","tz":0},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2021-01-11 09:02:04.000000000","tz":0},"subject":"Improve the CIDRs overlap check method for router add interface","message":"Improve the CIDRs overlap check method for router add interface\n\nIf an external network have multiple subnets, the router whose\nexternal gateway at the network will have all routes about those\nsubnets. So, when a internal subnet was added to the router, all\nsubnets of the the network which the router\u0027s external gateway\nbelong to need to be joined to check whether the CIDRs is\noverlapped.\n\nAlso, this patch revert the patch [1]\n[1] https://review.opendev.org/#/c/473356/\n\nChange-Id: Id5d8ac09a38c656619f88a6f87b8f384fe4c55a8\nCloses-Bug: #1903433\nDepends-On: https://review.opendev.org/763621\nDepends-On: https://review.opendev.org/763626\n(cherry picked from commit 8ebd54d1bcb6d4437cf9501f1c5993177747fac2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf4d045ac5583ecca291f38da745c62fd1d5d688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf4d045ac5583ecca291f38da745c62fd1d5d688"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
