)]}'
{"id":"openstack%2Fneutron~770110","triplet_id":"openstack%2Fneutron~stable%2Ftrain~If19803ab6db144e2d17d6805d379c1c76d8fa343","project":"openstack/neutron","branch":"stable/train","topic":"bug/1906366-stable/train","hashtags":[],"change_id":"If19803ab6db144e2d17d6805d379c1c76d8fa343","subject":"Limit usage of resources in the fullstack tests job","status":"MERGED","created":"2021-01-11 11:42:58.000000000","updated":"2021-02-04 16:21:23.000000000","submitted":"2021-02-04 16:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770110-bug/1906366-stable/train","meta_rev_id":"013ade1a0ac84be61f14faf4c059ad08158a3c51","_number":770110,"virtual_id_number":770110,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-04 16:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-01-28 09:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-01-18 15:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-28 09:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 13:47:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-01-11 16:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 14:41:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-28 09:58:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"18999c426286b2dc8e3518528698f516bfcbde99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-11 11:42:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e2c3c1bb53d82254eaf9140ee18b4e768d410af","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-11 13:47:46.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770110/1 : FAILURE in 2h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4c854bedabb229d0919d2a5cc1dcb879d9805352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:50:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/88d450ac2de24280b851f890180b79a6 : SUCCESS in 46m 20s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/b4b76d339be5473ea26c6e6680f7f4c8 : SUCCESS in 34m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/fd3ad35d960e43d09c25a401501b182e : SUCCESS in 1h 59m 25s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/27a5a96c86f247978a2a9357d20f28a6 : SUCCESS in 1h 28m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/0a87a9e88dd7467e8eb46d4cb574f71f : SUCCESS in 1h 26m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ba95f7e6aba4e878f2f99ad30a68a67 : SUCCESS in 23m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/187d20adf156468a89e140b3d0ca5aa1 : SUCCESS in 23m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd51a46e9894813a531c9671cca34dd : SUCCESS in 12m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cf120cff39d41cdb57ee94070039a21 : SUCCESS in 23m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/915c7ae175bf4f0b87667f572693ef47 : SUCCESS in 23m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d081fa4b8cae419a89e8702d793d3290 : SUCCESS in 29m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5c9ebfe26d4252823d32a6dcaced63 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/a7c3c29a2294472e8414ac5f8852b7c3 : SUCCESS in 1h 16m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a9d07440f37040ed99dfbae268361aa3 : SUCCESS in 1h 24m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd7c296affa2451d8f15fd3957589ba1 : SUCCESS in 9m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8438cd067a26436d94434d30f36f851b : SUCCESS in 42m 52s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/b386682410e94935879ce262e0749fdf : SUCCESS in 42m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d3a89b11b43743beb43843c8004b574a : SUCCESS in 1h 11m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6e8222d945c842699de49bf313cab76c : SUCCESS in 2h 12m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8d6644611240488b8b4a642f9220fb1d : SUCCESS in 1h 20m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7f32e5b91e248f390a538bdaffa30c6 : SUCCESS in 1h 03m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e7d3f89f3851420d9a2ba91242cd6345 : SUCCESS in 1h 12m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a0bcec5cb48466c82629c063ebba524 : SUCCESS in 1h 41m 48s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/029053a414e0433788360e27ab7cabf2 : FAILURE in 2h 09m 16s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e873ad045fd403c90a614bbbd9b09b8 : SUCCESS in 1h 19m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f1c28d6bdfc94e7f8b94773d25c3eae7 : SUCCESS in 52m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e0f5666000944d90a7b8e19a3abc042a : SUCCESS in 59m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/920cbffbf85442308e03e49dd10d16e7 : SUCCESS in 1h 50m 30s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2d8139afb54d49d3981415fb2cb89f5b : FAILURE in 17m 55s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/42d8fb0c6512462789ced8b66593bf44 : SUCCESS in 58m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/119ca57108864f588ffe9c4f375875f6 : SUCCESS in 57m 26s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cbf5eb414eb241c89ac06c82c8541df0 : FAILURE in 1h 28m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e94c40179a6946b9a22ae0f4d331b46f : SUCCESS in 52m 25s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b412c6f7021d4f7baa65bd9e66a707f7 : FAILURE in 1h 33m 16s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0c2eb6e2f584218b13194abcffb1da9 : SUCCESS in 1h 06m 36s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71d5e893cfbe4d73a05604b6aca2bfb5 : FAILURE in 1h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7dc22127b6cfbe346fcd90be08153b19277545a0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 14:41:31.000000000","message":"Patch Set 1: Code-Review-1\n\nOverall looks good to go, one question: should we disable etcd3 in job defintion in .zuul.yaml as it is in newer branches in https://review.opendev.org/c/openstack/neutron/+/764907/3/zuul.d/base.yaml#27 ? Or it does not exist there?","accounts_in_message":[],"_revision_number":1},{"id":"ff2d369b953180a5b0adc1345a8de38d51d1d1d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-18 13:54:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e0d880cc33d6b6a9285d891e3e1ba898354fe37","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-18 13:54:38.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Overall looks good to go, one question: should we disable etcd3 in job defintion in .zuul.yaml as it is in newer branches in https://review.opendev.org/c/openstack/neutron/+/764907/3/zuul.d/base.yaml#27 ? Or it does not exist there?\n\nGood catch. Done now :)","accounts_in_message":[],"_revision_number":1},{"id":"2a5ce46270c01d1444a11423aa51b597dcd7ae9a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-18 15:55:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b09cf5508ca616139e6ca43ef3d5c7909667dcb8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-18 15:57:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770110/2 : FAILURE in 2h 02m 05s","accounts_in_message":[],"_revision_number":2},{"id":"63885acf8e992f95b3414bc1f85927eb8357bb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 22:33:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d8f60b9d21774fad937fe3133fd6de13 : SUCCESS in 45m 18s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/30d7af32298e40deac01fd68bffadedb : SUCCESS in 36m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/4ff8aa53ead048e1872703b213bd3603 : SUCCESS in 1h 39m 19s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/608fed4f76e54ab7b8251fc9d8958335 : SUCCESS in 1h 21m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/4748f87f9fa54ee79cc63ff5c7add35f : SUCCESS in 1h 15m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d267ba48e4943c4b4e9a69d26237903 : SUCCESS in 27m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5487c812053d4d8486d7ac93f42b3878 : SUCCESS in 12m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99c52a43a92746d1a45da8c59545e965 : SUCCESS in 27m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68dc9d3768624061a78e05d0512f381f : SUCCESS in 22m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54985f983126457f99902dab16a1faa8 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd75fa3fec0d408e9df93a5fcee51ef9 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/df3fe8e75a624c019fd9cba0b9bbd974 : SUCCESS in 1h 08m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2424d388ab564783bed877433ce7a673 : SUCCESS in 1h 09m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/404f94d4f73c463ea486a954eda4b066 : SUCCESS in 9m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b70b6b19a6b4ec58ea9d67e74a11056 : SUCCESS in 45m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/6322089cc98d46f69ab04a4f6a724c8e : SUCCESS in 40m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/be3fc0658cbf454090fe002673cccf69 : SUCCESS in 1h 14m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d5067265c738462c948dc85fe6b00d3b : SUCCESS in 1h 43m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bd9b11b6b7e740b7b68dee109c9270ae : SUCCESS in 1h 11m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/72a68056c0cc4e68a48279d256d63298 : SUCCESS in 1h 17m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e39f84b1ad8345cf9c4005220a3f6953 : SUCCESS in 1h 06m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd31617354924f6d8a3e52fe4c12f7d2 : SUCCESS in 1h 06m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5b27d27a1f224d8da9523ad783358cac : FAILURE in 1h 28m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50b3f3b4ccd24a16920d1280722ba7a0 : FAILURE in 1h 39m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ded51192d8ca47caab4f33a11f53ae86 : SUCCESS in 1h 16m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2a2135d701944d4e8b648a0715ea904e : SUCCESS in 1h 03m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3074120e866a45b8b96cc34f7f01b0a4 : SUCCESS in 1h 20m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/274bc468a8664dd2aa4cfce4eab87803 : SUCCESS in 1h 59m 04s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4351cb65d0cf49eea4d7f0010cf9d66f : FAILURE in 10m 33s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/43a42c62730245478824561644499667 : SUCCESS in 50m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2348a029a79643de99847f66156a497b : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c65ddfd8f58f43c698aac9f7b05262f9 : SUCCESS in 1h 20m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6ef3a9efd3543e8a24435745d2a48fd : SUCCESS in 36m 01s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c3ec6630b72408ab798e863c84fbf1b : SUCCESS in 1h 06m 14s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f213ecea9d54b058d06ea975831bc6d : SUCCESS in 1h 12m 51s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/24d6a6e1a2a24987bae2c44343bc8eac : SUCCESS in 3h 10m 51s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0652eb89eb2e442d96299d5ec51325ac : FAILURE in 1h 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b7482856e0f079cfebe5ec0372704afb799324c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-28 09:58:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"649c866d1e77df702814ddc3dff3db59bfb5db68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 09:59:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"88a06d3b62ccf087cfe83098c6ec01cd8e908f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:18:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ab504c6d74e6418eb7b25d2a9f44cce4 : SUCCESS in 45m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0accf0c16ba4b63ad0abcbc3bb7954e : SUCCESS in 13m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5433522a7cf4460a8277e11a64c1672 : SUCCESS in 24m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0f81654a8bf45a3b53a99e0e6987a8f : SUCCESS in 22m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c6f124346914e2e8fab960f3681fcb4 : SUCCESS in 18m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6145dba892c242988511f2e24d5dc5ca : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/6dc1565c004749aa99a01ac27d1b5d9c : SUCCESS in 51m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/65a7b67dac964aafab0451acd62aac2b : SUCCESS in 50m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dea4f46fa34a4407a34025e6536f15be : SUCCESS in 10m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/413946198fab4b9d97b7f9e16b763455 : FAILURE in 1h 52m 57s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/279f5ba5c8434f6882c707d337a8afde : SUCCESS in 37m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2aa2b9614c2b4349a351bbaf7d35ef60 : SUCCESS in 1h 04m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f6c2d8f241ba4e8faccfa46783959656 : SUCCESS in 2h 00m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1d4724a4d7d240c6a2b3be2ca3f7d446 : SUCCESS in 1h 22m 19s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/af017f144d3241b4ac8485b2ad24d2dc : SUCCESS in 1h 10m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/05d9b40c03ff4a51bf922465ebb81093 : SUCCESS in 1h 08m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1994763aa01a4262853d489ce9288965 : SUCCESS in 1h 08m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/df0755a8807e4f6e91c5404180d95d1f : SUCCESS in 55m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a53668a3a6374459b5309cf4b3342aa8 : SUCCESS in 2h 03m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a5e5512060b48c7b4d55a7a9091b06e : SUCCESS in 1h 04m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be47c0dabc6f4b8cb0e15e9525767722 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":2},{"id":"514fbf4b985b71d202585a88b9ee1dcc39131019","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-28 19:39:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c4187a4c2235515f1835cc5a2c4050672efa6cc1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-01-28 19:47:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770110/2 : FAILURE in 7m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3014f826b0f5e41184cf50f68abb922b3e251b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 02:24:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/1fc69325b5124e11acb28dd6f26eb632 : SUCCESS in 54m 31s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/bb2cfd4df968476ba5d45cfa9d44dd63 : SUCCESS in 29m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/850cfce816f743de90a7d4424187bb99 : SUCCESS in 1h 29m 05s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/d8b0d15641ee454594cd717c8c1b7e67 : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f74ce52e40a94484a74fc27c8651fd26 : FAILURE in 2h 42m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e17a46827ab14d6c8539e69244798c39 : SUCCESS in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6565b3baf07d4799924e02e38167d72b : SUCCESS in 15m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7896bfc28c2b460397455edd66dce758 : SUCCESS in 24m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45d9dc1b3de94f09b01d17dde9abd54d : SUCCESS in 28m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c717f3959c184140871e28831e7f66b3 : SUCCESS in 27m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b434133d384f427da295a905de1fcbbb : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/4ed770410b1f4f9e99b7a262ee9070d2 : SUCCESS in 1h 10m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/da6d1dedfafd4c63a58bdb974adcc14c : SUCCESS in 1h 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c3007fc50a4d90adf82eb90f4bd76a : SUCCESS in 8m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5417deec3f5749a38b0f0173662a864a : SUCCESS in 36m 45s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/13bc2bdd724c4a9683f90bf6bcd1989f : SUCCESS in 45m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ec3b3b121b1647fca5888b6f0cb039f9 : SUCCESS in 1h 04m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9ab40bad340b4ccc852906a1072d48ce : SUCCESS in 1h 37m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ab31aab9955e4f6c9f0a2d117e1b7593 : SUCCESS in 1h 19m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/657e32c0de16455781bc63d71ac55e51 : SUCCESS in 1h 13m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34dcc3877e7c493facb5cb0870c6993a : SUCCESS in 1h 07m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/54fb98041e8b45dabb6a7f1888ba3e01 : SUCCESS in 1h 23m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/043448a38d5e47eaa7d9ffe05519e919 : SUCCESS in 1h 45m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/231f78d051064ffaa38e9a0165b46673 : FAILURE in 1h 51m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c4bdb54fc011492caf0ba9dc908d4816 : SUCCESS in 1h 05m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b782af424cc84ce7a79c069d39a45325 : SUCCESS in 54m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6efb59567f374d0b98cf485c88a6e35e : SUCCESS in 1h 10m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d374ee978a6429391cb7215232a0b20 : SUCCESS in 1h 54m 52s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ee31317ccf62470995825f9ec514c7a5 : SUCCESS in 1h 56m 27s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ca769014c4034892a4ae496b7dc46daf : FAILURE in 24m 39s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ff5633f3e5894c02af18efcf1dfe7a6f : SUCCESS in 58m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bae46c5db044318a9eb76aed29efb49 : SUCCESS in 56m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f27c451e3f0245af810667c45476a6e8 : SUCCESS in 1h 25m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fd32f14984c420f8d3cbd422190a265 : SUCCESS in 43m 21s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a892633a3ff5401b82ad7a51817efe7d : SUCCESS in 1h 12m 21s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4e86e9a2820452dba8947832074b01f : SUCCESS in 1h 11m 16s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/9c10c5c5a53c4461a1890cdf7b8dc18b : SUCCESS in 2h 33m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7c5ef1814e0e428c81472ceee3499457 : FAILURE in 57m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24c880c068e431afb80b919e9474628436be47b3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-03 09:01:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5a12c6e511dac88f4ec3ddd3c7d2358bf5852b4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-03 09:20:57.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770110/2 : FAILURE in 7m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4b7987c84154b383cccc096ee630ca2097a146d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 12:49:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/3d497e35340040ee9456030202e5d2d7 : SUCCESS in 42m 27s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/1741f2690fa446a7991cee81cb5e4e55 : SUCCESS in 31m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/a39507292ef548e88b9fc4961e1e2296 : SUCCESS in 1h 52m 50s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/e9f33a1de4a84b57bef67b9850009dc8 : SUCCESS in 1h 17m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/7000f1a625894f338ebfb74a4784c064 : SUCCESS in 1h 24m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d18478a3aaf4d1f8e24ebb6038e0c50 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3c3a83946840c7891870a472613836 : SUCCESS in 12m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8f00fa383fb42df8d1ba741485d2524 : SUCCESS in 16m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba84e5cb980a4b9abc90f67f83ab3a82 : SUCCESS in 23m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8bf3b0dfd404de2873612b75888906c : SUCCESS in 23m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3275f03b36214d889d8c9a94147d5b65 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/659ce8d705b84060a48ca3c0859182a5 : SUCCESS in 1h 03m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2c8a445a23a45b38301ac66950e4714 : SUCCESS in 1h 05m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26aed39591424e43a651c15768dd557a : SUCCESS in 9m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6f692e23c7234c078e6e46f0b4a6a24a : FAILURE in 45m 03s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/4b16d3233f214737801a7065285cd348 : SUCCESS in 45m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9cb13e8b907d437899137aa7c2da0e53 : SUCCESS in 1h 12m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4dcb83b6fc854b4fa6fa2e55e05c0bf4 : SUCCESS in 1h 15m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/28e99d94f650489d840dfafe1a77e0a4 : SUCCESS in 1h 11m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f1da35d8b3904b00b566394d5fe09d40 : SUCCESS in 1h 09m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62db1c69c185461a82dc934974ea158e : SUCCESS in 1h 00m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/75207155d91b43e4b9d4160ab337a832 : SUCCESS in 1h 01m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2a1422dec0ca413e9afa648928439f2d : SUCCESS in 1h 31m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/daa59e081db04404b2a6c41f7ab5e87f : FAILURE in 1h 51m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/122aa8c67cdb40559b60c80140c7680f : SUCCESS in 1h 19m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4b3ae10597654d978c846535c21f7d0e : SUCCESS in 1h 00m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3644b4110bac4e2cacf5ff8a7542a2d2 : SUCCESS in 1h 04m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/869181a569b947628d7fb0ced7f025ec : SUCCESS in 2h 02m 24s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a120f0f2bb414f29a4ee2b04290f1b8e : SUCCESS in 2h 06m 30s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/8d9c879602804ba48e94a1e94f128cdc : FAILURE in 14m 34s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ad8d45b1d3154df2ba1376c09491b2e7 : SUCCESS in 47m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1faa705e78774faa86d9e45df0f1c5b8 : FAILURE in 58m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ecc5d24516a34f05b5c1080b43d170db : SUCCESS in 1h 20m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5e5b5e0dad04c058850756e820eb24d : SUCCESS in 43m 38s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/458d365edc9846b49bfd4679443c541e : SUCCESS in 1h 11m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2270ca36cb544f9b23e9b417b7c856d : SUCCESS in 1h 18m 19s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/6b08ea76a19a448093beb864ad3ce335 : SUCCESS in 2h 23m 43s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/173bd953323c4477be0f91e0fa3f5f4f : FAILURE in 59m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08aea84b9cda5bffd068b89a5d1ca1e3cfa1758d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-03 16:38:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3bc180978612723888431bfa1386d4118b863866","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-03 16:47:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770110/2 : FAILURE in 7m 53s","accounts_in_message":[],"_revision_number":2},{"id":"a4440a949a7dedadfc8e8e368d69db6f3d8532de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 23:18:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c0ba0ec429cb41a19498a28df99452b9 : FAILURE in 18m 53s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/3433c4be2f984176af065d5962ea10f3 : SUCCESS in 30m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/f1b1fc52f4fd41bcb40ea93a718bc5ca : SUCCESS in 1h 37m 05s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/45c0081516514a7e8aa1d37a1d0821a6 : SUCCESS in 1h 57m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/5bb7cccbb24b4696a73fa568714352e4 : SUCCESS in 1h 21m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d5560d3bdc445fc8f9f5c6534fc2cc3 : SUCCESS in 29m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f42ce983f214db7b3a08e2c61eb831a : SUCCESS in 12m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38bc9bd6227f4471b7044587d9d6567a : SUCCESS in 23m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f443bcbfa98f4e08a3f327ef26ec6b76 : SUCCESS in 28m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86258e59aa9c4848abc430074c1eebf0 : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/841aa74669734326a59a33a03605dddb : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/cf22b19558a44bde92e00ca774b40d66 : SUCCESS in 53m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc0acd78407c4b0394642efb0084dea9 : SUCCESS in 1h 07m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/654c4baeddc04e9c9a0c99eb13484167 : SUCCESS in 10m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a837baadb13f46d881312f173ab02975 : SUCCESS in 48m 52s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e9489a46cb354ebdb90de9198edd205c : SUCCESS in 44m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f44a9e8e507f4fb9a1af6fd7de55bdcb : SUCCESS in 1h 19m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/277264bf40db446484f580934ba32128 : FAILURE in 2h 06m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/612c547008334555a46d4a7f95a340cf : SUCCESS in 1h 16m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2d45a1996afa4d71b5fe493e4d07b854 : SUCCESS in 1h 17m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da756513ab1b415a88e5371a9e7d115a : FAILURE in 36m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b0e4c45227a149f9985e3b3064e1ee78 : SUCCESS in 1h 14m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/13d6f573b88b4f3f848ac5e66db468cf : FAILURE in 24m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ced51ef5972942ceae4a2801d7f6e76a : FAILURE in 27m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c1767400c0d424c9bfaa7d4895961df : SUCCESS in 1h 22m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e63471293904207a5553ff65b9ce6ee : SUCCESS in 1h 00m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2099bdb78297423587f56cdfdad377c6 : SUCCESS in 1h 11m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4c0aecdcb0a1463d8d59e7ee907e574b : FAILURE in 22m 18s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9e6681cb2d1d4144ab00306ff857e80d : FAILURE in 20m 31s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/01adf340bdd9493e8548c56d92d7dca0 : FAILURE in 15m 23s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04f3d2e2c0834c09bcd95c90d9ca9957 : SUCCESS in 56m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/477e7445869943a597e4a69e17e9c520 : SUCCESS in 1h 07m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ade7b2cfe8f846ccbf34aa7c8a34a774 : FAILURE in 33m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5eb1333973a2402595c99dec68b37713 : SUCCESS in 52m 45s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b3c3473c18847018b2511c6f74bff3f : SUCCESS in 1h 13m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5fcc76c4985a4f90a868c360fce62bc0 : SUCCESS in 1h 18m 32s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/fa1ee6ec0e4a406f8cc58da04e3b38e6 : SUCCESS in 2h 48m 44s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9740dfdd09f441c7a6af32366104b9f9 : FAILURE in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"930b70af9fc92b4b6d323efd1d2bf5c5b5429fba","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-04 07:53:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d6ddefb068403ec3fb8611aa82a408acb12b8781","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-02-04 08:02:35.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/770110/2 : FAILURE in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ff35540a87001a5df8f82dfc263b5f71d87c0848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:50:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/8d1a2a90ecae4ebfb3ad77946b1cf0c1 : SUCCESS in 51m 11s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/0507f14095ec4c0ab44b9639e53462e8 : SUCCESS in 37m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/e8fffca032944e178591bb1e729316a1 : SUCCESS in 1h 55m 02s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/b2a76756daae4e41bf40f033cf866f5c : SUCCESS in 1h 56m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/96a99e1fc3844c4c968c1fa04498ab87 : SUCCESS in 1h 18m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/854a3bfe483347f89114ddf0be548290 : SUCCESS in 41m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78291d1dcf2541fdbf1abe64b9cc3a01 : SUCCESS in 13m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce809581c5914f808a11f0708eaf317a : SUCCESS in 36m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/586b0e34539748ffa0e3ddaa83460206 : SUCCESS in 23m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/333e50822fe04b159747948b830610fc : SUCCESS in 38m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42c5a9ee26444f69766316c7362dbbe : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/e8680307fbbf42d79471330dbb03315d : SUCCESS in 55m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e0702aa8cfed4cf9acb2ebab9e1662f5 : SUCCESS in 1h 04m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7407661193c241388a9f9e7369e3adb8 : SUCCESS in 9m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae0721c760654805857c1da072ab76a1 : SUCCESS in 40m 45s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/fab79e8a2daf4b62879ad944f1db6d66 : SUCCESS in 41m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/062dbdfec9814c5899946427e018fc72 : SUCCESS in 1h 08m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3fc9afef8855432f8d533c3ec824405a : SUCCESS in 1h 49m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/af97cd75e3fd4275a2d8a86e75bc894f : SUCCESS in 1h 04m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/67c3e5c3464e4ee29faf45b9c89aaa81 : SUCCESS in 1h 06m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a37217c8029e4b93bb61b7394d271e53 : SUCCESS in 1h 06m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/780a69f149a440c39a23f6bcd2f1a86e : SUCCESS in 1h 21m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c380207a6c0944c1a9c2c09cd4b10551 : SUCCESS in 1h 42m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ac4ae917bcf045579792cf289ad60204 : FAILURE in 2h 03m 57s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4641cea25fe24a68a1282213a0dcac52 : SUCCESS in 1h 09m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b4aa94c0e024bbab1227fa7644e4a8d : SUCCESS in 53m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c77d1e33fb7f4de49055ed85445069a2 : SUCCESS in 1h 19m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/681a431ce27d40dbb7ab8a81201fe6cb : SUCCESS in 1h 58m 00s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/470984c81e2b4c7aa6e74180133c1da7 : SUCCESS in 2h 05m 59s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/765b81a2d0db442ba3dba65cedbebe9c : FAILURE in 18m 03s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1be15fcfcbec4b299580b743224072d8 : SUCCESS in 57m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39bc886a9e8e43528d24a525ebec01f0 : SUCCESS in 52m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f6003f9a3b8a4f6597c0a44f4308bffd : SUCCESS in 1h 19m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae55143f779f43d0af67872196c1ee52 : SUCCESS in 1h 03m 18s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4bf6370a6ff4dca8f1c3c74570e2014 : SUCCESS in 1h 13m 23s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/49d31a650cad4d099341c2519b536748 : SUCCESS in 1h 14m 23s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/176675e2f2a14aafb25524a3660781c1 : SUCCESS in 2h 50m 22s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bf9f3144203e4cddae0a326d99c3a957 : FAILURE in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5577b8d188e8cf2d476b22ad144c4e1d1b2964cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:50:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7a1457bf55e26fd12a7c08e0670c87c9a2d52e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:13:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/9f1a2898b3f34fc6ab5e961832960d57 : SUCCESS in 50m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663794909652481fa7bbf97873a4c5c7 : SUCCESS in 11m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88fe18a00dc7465c816ed25bd0028d66 : SUCCESS in 27m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3caeef09b6a8489e88fa9aa5ba2efe59 : SUCCESS in 22m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c49e7477d0744e29214efad841e3eb7 : SUCCESS in 27m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b987100cce91473fa8cfd15ef9972741 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/0e2511c3f2fb4a1b97fabe2e9dfb9417 : SUCCESS in 1h 03m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8314dc49a75847db864150d69fdbb83b : SUCCESS in 1h 03m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dfaf60ad26b4e199d3cde73030c1ae0 : SUCCESS in 7m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d7bd39b670c046bda2a818940a75158a : SUCCESS in 43m 33s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/97270ba331fd44e9ae887b25d41bd35a : SUCCESS in 39m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a4e28d6a12364d5f98b4ba8d827e2678 : SUCCESS in 1h 46m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/951e4ed2228d429ea3ea8a779c2fb314 : SUCCESS in 1h 15m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ba9e0a1d29147018ce7e052fbc6c6b1 : SUCCESS in 1h 16m 07s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9702fc32384c41bea9be07cf190b6ea8 : SUCCESS in 1h 08m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7aef4ee5f38464e90fcc22322429b91 : SUCCESS in 1h 00m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/95db8d66b90740adb08bee01de22b81d : SUCCESS in 1h 08m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e6e8132a07b6408788b23e1c1a216b83 : SUCCESS in 52m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5402f2ce22f4864a8471974471420ae : SUCCESS in 1h 43m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/92a2cc5532604fe3a5ab367c0a235804 : SUCCESS in 1h 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b59a70ae0d0b44579e5da615e8382832 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":2},{"id":"50b24240bbc030144198c87a7472380a3572db5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:13:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"013ade1a0ac84be61f14faf4c059ad08158a3c51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:21:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7026204f98394b0dac103fa431e88929 : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce106e3d9749467ba33542039a1b7b73 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d1c96e36e398143440ca6d75d40108360b86f31","revisions":{"6ce29016d37094230dde2db8ed446df5856ea4f1":{"kind":"REWORK","_number":1,"created":"2021-01-11 11:42:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/10/770110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/770110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/770110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/770110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/770110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/770110/1"}}},"commit":{"parents":[{"commit":"c9416b10136ac4c6ba8943431b6b645f8baecfe6","subject":"[train] Fix tests with new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9416b10136ac4c6ba8943431b6b645f8baecfe6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-01 08:18:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-11 11:42:14.000000000","tz":60},"subject":"Limit usage of resources in the fullstack tests job","message":"Limit usage of resources in the fullstack tests job\n\nTo limit usage of resources (mostly memory) in the fullstack job and\nto avoid oom killer to kill e.g. mysqld service, this patch:\n\n* Makes number of API workers changeable by tests, as a parameter to\n  EnvironmentDescription and defaults its value to 1. As neutron server\n  is spawned separately for each test and is used only to process just\n  few API requests during that single test so this should be still\n  enough there, and where more API workers are needed (like some dhcp HA\n  tests: TestDhcpAgentHARaceCondition) it can be changed,\n* reduces number of test run workers from 4 to 3 - job will run slower\n  but hopefully more stable,\n* in the functional and fullstack tests job definition disable etcd3\n  service - this will not save us a lot of memory but still it\u0027s not\n  needed at all so why to run it there.\n\nConflicts:\n    neutron/tests/fullstack/resources/environment.py\n    neutron/tests/fullstack/test_dhcp_agent.py\n    zuul.d/base.yaml\n\nChange-Id: If19803ab6db144e2d17d6805d379c1c76d8fa343\nCloses-Bug: #1906366\n(cherry picked from commit f63dc09206a41c85edc92117c3e9cf3a3d49b820)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ce29016d37094230dde2db8ed446df5856ea4f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ce29016d37094230dde2db8ed446df5856ea4f1"}]},"branch":"refs/heads/stable/train"},"6d1c96e36e398143440ca6d75d40108360b86f31":{"kind":"REWORK","_number":2,"created":"2021-01-18 13:54:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/10/770110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/770110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/770110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/770110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/770110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/770110/2"}}},"commit":{"parents":[{"commit":"c9416b10136ac4c6ba8943431b6b645f8baecfe6","subject":"[train] Fix tests with new pip resolver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9416b10136ac4c6ba8943431b6b645f8baecfe6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-12-01 08:18:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-18 13:53:46.000000000","tz":60},"subject":"Limit usage of resources in the fullstack tests job","message":"Limit usage of resources in the fullstack tests job\n\nTo limit usage of resources (mostly memory) in the fullstack job and\nto avoid oom killer to kill e.g. mysqld service, this patch:\n\n* Makes number of API workers changeable by tests, as a parameter to\n  EnvironmentDescription and defaults its value to 1. As neutron server\n  is spawned separately for each test and is used only to process just\n  few API requests during that single test so this should be still\n  enough there, and where more API workers are needed (like some dhcp HA\n  tests: TestDhcpAgentHARaceCondition) it can be changed,\n* reduces number of test run workers from 4 to 3 - job will run slower\n  but hopefully more stable,\n* in the functional and fullstack tests job definition disable etcd3\n  service - this will not save us a lot of memory but still it\u0027s not\n  needed at all so why to run it there.\n\nConflicts:\n    neutron/tests/fullstack/resources/environment.py\n    neutron/tests/fullstack/test_dhcp_agent.py\n    zuul.d/base.yaml\n\nChange-Id: If19803ab6db144e2d17d6805d379c1c76d8fa343\nCloses-Bug: #1906366\n(cherry picked from commit f63dc09206a41c85edc92117c3e9cf3a3d49b820)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d1c96e36e398143440ca6d75d40108360b86f31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d1c96e36e398143440ca6d75d40108360b86f31"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
